)]}'
{"id":"openstack%2Fneutron~778784","triplet_id":"openstack%2Fneutron~master~Ic9e64aa4ecdc3d56f00c26204ad931b810db7599","project":"openstack/neutron","branch":"master","topic":"bug/1917811","hashtags":[],"change_id":"Ic9e64aa4ecdc3d56f00c26204ad931b810db7599","subject":"Schedule networks to new segments if needed","status":"MERGED","created":"2021-03-04 22:20:40.000000000","updated":"2021-03-16 14:51:51.000000000","submitted":"2021-03-16 14:45:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778784-bug/1917811","meta_rev_id":"4d72d761875e7931db6f6797ec5ea879cb065398","_number":778784,"virtual_id_number":778784,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-11 00:24:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-16 14:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-15 15:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-11 07:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-16 05:03:15.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-16 05:03:15.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 23:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 23:20:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 23:37:14.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 00:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 08:13:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 08:32:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-08 11:32:21.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2021-03-08 11:32:21.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-09 06:16:23.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-03-11 07:19:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-16 05:03:15.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"bb63344624e4d6787bd8844fd46770aef5250bd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-04 22:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bc30ee4223ffacf5bf59412629cdc7a657e61a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:06:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cebc58830b92475ca12c019475095bd5 : FAILURE in 36m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a0cc94488824b02b8e5dae337aae844 : FAILURE in 38m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"502ce9cc7ad34a75c6358090eb7020dfcdf74033","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 23:20:45.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"0a370519579ad94d7967bdd16c21913eecd4de31","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-04 23:37:14.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/84/778784/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e6a3a2dbc6ce08b8880ffb3f1bdc7c20f7502d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:38:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/15f42d1fb2de4ae19047ce32be49663f : SUCCESS in 51m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/00486ac077504c80b8c2082fbbbc34b0 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4214cd19fe0c4fcdaef424c7584ac14d : SUCCESS in 1h 26m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cf90745893942ffb449dad07dd5b47e : SUCCESS in 1h 03m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2f27b7b60f4464cb8369148e6b49f8d : SUCCESS in 1h 23m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/628baa1626774d1c85e454338337a8b1 : FAILURE in 47m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5164b509ad004ef7be3f69ce759cd85e : FAILURE in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b031a9e36212482ab3defdd5b541b68b : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/590b37068af747ef96e21d63a78b9ead : FAILURE in 26m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc00b3393c484b528dd4f5aecbf7f58a : FAILURE in 29m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/684e93ec63c745d38fb1110bd4bba5cc : FAILURE in 31m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7055d06001bb48b0935c3265b192e0e2 : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/32e6ae11607f475784cf1d12635d8cf5 : SUCCESS in 52m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d25ce04240454a328df2edc8154362d4 : FAILURE in 1h 09m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e449d4a00b4643858908a090310658a2 : SUCCESS in 2h 10m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/287346efdbf5465d937d2c85e4f37f68 : SUCCESS in 1h 07m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/04833833a68f40e0a26a60ec395297e1 : SUCCESS in 1h 07m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e9e9565fdf734061b004326f819bce25 : SUCCESS in 1h 18m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de782b53bb7848e38fda66c813b3b0f1 : SUCCESS in 1h 37m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b15eff7971294064a32cdf393085bab8 : SUCCESS in 1h 59m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5101367fe69495fb9cf082131456b3b : SUCCESS in 1h 06m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/485dccfc6f3f48c4a192af283522c144 : SUCCESS in 1h 18m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/66cc2991c3e541d4b679f939bcd2f81b : SUCCESS in 34m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/52a8d1f83cf945c6970eae57939ee0f5 : SUCCESS in 1h 37m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43640a4296d24dbbadf8fce96edcc06e : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"58a5507b2037e7f7dc2b18dd12fc8bef7c2b4b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 07:05:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13b8f328e44df37ba7c8b6ad8c204c98f74707aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 07:59:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db5d1b057218437c9d0ec7b5257f194c : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9e191b0710142e7b5b47a07d8ad142e : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83df623bca3789a22194a5768b50df708841961f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 08:13:10.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/84/778784/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"31ab97bbacec9dcd6ea0b5c6dcec3530a85377f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-05 08:32:22.000000000","message":"Patch Set 2: Code-Review+1\n\nCode looks good but I\u0027m still testing it.","accounts_in_message":[],"_revision_number":2},{"id":"4dbef5a420c3fec52391e79f23ebeff9b5d4373f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:19:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f7d2ee351754a4386e75b774ef90da9 : SUCCESS in 1h 01m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4f88f6a9dcb048f092ea1cb709ac6064 : SUCCESS in 1h 46m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0e8405b1ee0444c7a96f028a11974af3 : SUCCESS in 1h 38m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/040b34fc5a5b4d26a65062957a9f8466 : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/33ad2ed85c5f4738820265a07351d7b0 : SUCCESS in 2h 09m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75b4fe6fa3c44eb2b6cc3b5123c96aed : SUCCESS in 46m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/021e42f5b7db4b56802e87496ad2d70d : SUCCESS in 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c5b93210e24bd680e8d04ec76ca5ac : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d71820219c74f1fa55c3c5278c0d2ba : SUCCESS in 40m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857b891525094bf1aba444ee31ac25a3 : SUCCESS in 31m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d223a822f44921ba60b68ba9b80304 : SUCCESS in 32m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a133fa2841ed4fb59071ddf497d67051 : SUCCESS in 10m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d5db614f87c4b6498eb817b1af70d32 : SUCCESS in 54m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/520d6c9ccc544206b7c1617c4e3d6a9a : FAILURE in 1h 11m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/826faf238a3c444fabdadf6c4857ffd3 : SUCCESS in 1h 30m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bfd1477c0e54f56af019eece8972dd4 : SUCCESS in 1h 00m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e149e4fdb3f04b8194b37c6edf9f905f : FAILURE in 1h 05m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/46b79329cd3d48479312e55d38a9ebfa : SUCCESS in 1h 28m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ede0241a95874fc49d0db66592a8c877 : SUCCESS in 1h 34m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc1b1fc4f9384e4fa9f229f28fed26c9 : SUCCESS in 1h 59m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4379c5a9b34fa7a9e6267c5042d0c0 : SUCCESS in 1h 10m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2ed0ac55d17749cba279df71bbd175f6 : SUCCESS in 1h 06m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7886516cf6f848a1b4d86584d6138eea : SUCCESS in 32m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bf8ac3fcf4d469bbe0af6ca3e7d4d7a : SUCCESS in 1h 49m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a7ca4cc8639c46a5be4d0b765b04d1a6 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ada22c0534f65af962114afb4833e2ff9a060020","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 09:39:26.000000000","message":"Patch Set 2: Workflow-1\n\nI need to investigate fullstack failures as they seems to be maybe related to that change","accounts_in_message":[],"_revision_number":2},{"id":"28187218b52ead08bb37ac87fbf87d7cee2f24cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 10:29:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"08427935980e61d04bfac89c28f3e5c111938f35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:53:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1354061b9a58f8fa2b7037b3589f7c2c90866a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:42:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/781d27d71a5848a79a6cd0b8ed43ce4c : SUCCESS in 48m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/55f6c1baf36041739be8ddc4f7f05f9c : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3abd64cd83154f569275fb051bfc8223 : SUCCESS in 1h 58m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5bd52f8f52244c4a94083a204daf2aa4 : SUCCESS in 1h 31m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ed0f8d4227fd49dfb405cd25f5d8f81c : SUCCESS in 1h 40m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dcea45d8479485c880c25d96922ae23 : FAILURE in 41m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e98f18f06d4f4378a65d472905732aa1 : FAILURE in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fbfa43b1ca4202b88efaefbc09e210 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a9f4f41e1d94093ab165a7706fc1fec : FAILURE in 30m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a799244f4ca4010bd82c5793b2b602a : FAILURE in 34m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eea8b7b0a0040fdac005d8cec2cb341 : FAILURE in 34m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c60e7940e3849a5a149f36b7e13c833 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f68b2fad362048ddb9422155d221dae0 : SUCCESS in 49m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91755ff4fe24481a8aa5391c1e819016 : SUCCESS in 1h 25m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/262af72d4c2f4d86b75bae6e754b3e01 : SUCCESS in 1h 28m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a42e9437f7c41f498587d6df5ed920c : SUCCESS in 1h 01m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0203d445d0940d291536b863c338a0c : SUCCESS in 1h 00m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/be24bacd1c6c48999f8fdf8110509fce : SUCCESS in 1h 20m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e47a7e9103b4df2993ae015d36af5cf : SUCCESS in 1h 23m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b542a39e8704d1eb6206e5853042698 : SUCCESS in 1h 45m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9026276ee75c4c4386aa37e406e64129 : SUCCESS in 1h 03m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bc0867ed3c6e4020870f3d5463a9a585 : SUCCESS in 1h 11m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bec627c59e2c44dfa9ea9e4ebf2ebdd6 : SUCCESS in 34m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d84783b452e943c09f88a4f59f8f27d3 : SUCCESS in 1h 10m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/009810eb27844ff090cad63740324ff4 : SUCCESS in 1h 34m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ae94282b25d13ecf61cad936c1dc19a61a9f11b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 16:05:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"345906e0fcf457f478426ba39a62df3400bd41da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:05:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/700eceb1aee3468baf4603f13e03304d : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61855c11571b4fd68293689a25132a32 : SUCCESS in 49m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"71e447f330a7b48f24cb4fd330a8c865d093707e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 17:50:49.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"f671796873cba51a67b40a05cd03b64566fd8db7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 19:21:03.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/84/778784/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"cf1d3614ee9eed6bbcc4386e009efae1f4cd1f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 23:45:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b3448f499d3442d8a672df9967d0942c : SUCCESS in 1h 15m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8762ff10197743009c012b241f0905df : SUCCESS in 1h 21m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a3e16fff4a1b4f2e89add5881158695b : SUCCESS in 1h 42m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a91b10cac3f42909581cbcecbd3334f : POST_FAILURE in 1h 31m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d2df20adc59e4549ba7518d36093e7ab : SUCCESS in 2h 19m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecf5ae1b7edb42ee995f75edb378f046 : SUCCESS in 39m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e13b159b4874ac8a620217a7c0acfcb : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c5c53edb50414797ccc1d34d55ac90 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5c58b1583f421fa465359dcd86506f : SUCCESS in 44m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe19660fbf34c6886dccc130b9c451c : SUCCESS in 21m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/532e2542217641edbf907e7a741163f8 : SUCCESS in 35m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f8751e926f4bb197e09ca08f42f4a3 : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ce4fd00fe1c4764b778f4e847ff6a4e : SUCCESS in 51m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6d707a576a8414cba52eff86b0959ed : SUCCESS in 1h 33m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ddaf6d61a5c54210b3d09ec7d657f2b6 : SUCCESS in 1h 57m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84a542fea975451387e6c5fdde7e0ff6 : SUCCESS in 1h 05m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2bfb119ded484892ab570ba77300c5c6 : SUCCESS in 1h 11m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7df441c963e44455ba59d2f5ccf0081b : SUCCESS in 1h 15m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/18d4b05da01d4da4a82b32c2916c85b4 : SUCCESS in 1h 50m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e0cbdb314634e4484be05dca04ca057 : SUCCESS in 2h 15m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c24ddbbd3015499fbc232353c5a3bace : SUCCESS in 1h 05m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/162f3b7b844b4373bd2c072d896c7446 : SUCCESS in 1h 21m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3560ef7fd0584d3587162e33f6c0bda2 : SUCCESS in 45m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/08db380a83b44dbb87254e01067b09ed : SUCCESS in 1h 25m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/05e31b4e191d49eea0133a08c1282abf : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":4},{"id":"395c357072a2ec521b16cfe1c4a01e819bf7b8ba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 18:13:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4072b092a645fc29146b7dde933e9f75b7bffe03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 18:58:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a58c58f61784f1cab623d9393136782 : SUCCESS in 39m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7bca77d3d89a4772adfa2b2eb203ca4c : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a941710b641ff42eb7bce632b19e359fd80606d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 20:29:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6a9aae54449540d1a2b02c4c41ffe715 : SUCCESS in 38m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c70d7ea05d4e4b2996ca0fa3c0c728c0 : SUCCESS in 1h 10m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ee96042e48c4dca889f81c92f60ad28 : SUCCESS in 1h 36m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b90b99da0374297a6098dcffd3d5800 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/24cd218cb2134774a2edc84896fde5a1 : SUCCESS in 2h 02m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74b5e4b19c904718b9b5042f16b838c3 : SUCCESS in 26m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb08d94b8a2a4c96bca0d06bd52bab75 : SUCCESS in 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d881d2b5f84246dc8868eda30090e1de : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0d67027a46d4a9ba63d03d9026efefd : SUCCESS in 37m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11919e8a6dc422396f744934df84158 : SUCCESS in 21m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57e3cc257f0540e2bde356ab644f18d8 : SUCCESS in 36m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c60f02447548b4914831a3db08c7c0 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/07c627b3a9f7461685f8ef0c7babb8d4 : SUCCESS in 58m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60ae857fd496462f9baddd58d357db80 : TIMED_OUT in 2h 14m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9b4247a438454286855c9bb116273208 : SUCCESS in 1h 34m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc241fd4e7e74c9086eb80ab768abb22 : SUCCESS in 59m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d01089ee08147829fabe95947cb9d80 : SUCCESS in 52m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3866877487a9485798c953bf4e8035d3 : SUCCESS in 1h 03m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e301bcec60947a7a8f4a876eea35cf7 : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c39b42d91364a288d6245e900211d0f : SUCCESS in 1h 42m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d82e8c3a462d4ce0913eaa47e4367202 : SUCCESS in 1h 12m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/91cb0a373f804301866e013e74013707 : SUCCESS in 1h 23m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8833a2e4442448f99e73a31c4df0237e : SUCCESS in 39m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/321b27a124024edeab7568ee6698b925 : SUCCESS in 1h 37m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/90e9462f98014acf8a103f7e1412b964 : SUCCESS in 1h 28m 09s","accounts_in_message":[],"_revision_number":4},{"id":"294dd0eddd589a6671e0c130809646f78963a01b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 21:26:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b5042930fcf45101ea039063f596217b32fd835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 22:12:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b55380eea4549568befea28815dd658 : SUCCESS in 42m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7f93d6a1e974f10aec2e0baf1334ad3 : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59fabb4c20c8039d0a935875cdb0781c64dceafe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 23:40:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b9b48af75dd4714b01e4326cbd284e1 : SUCCESS in 1h 06m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/65ee2b682f9e4a0eb3149ccf500aa018 : SUCCESS in 1h 44m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/075702df278f484391f8bdf5dc2ea219 : SUCCESS in 1h 39m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dcf3bd7d6c664114a5798a7e8176a648 : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f76cfa947ae24fc8adb9888838754048 : SUCCESS in 2h 06m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6305ce5f0d324b87852391e4867d790d : SUCCESS in 49m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2fdb45acbd14bab845792d63fb329db : SUCCESS in 33m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1310e8ddfe874d9cbda9fd6f747d09f1 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467192bc43e54b509f5cb7c93ecf0491 : SUCCESS in 21m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44a37083430048829d2436c242da6a51 : SUCCESS in 21m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee5c3e5fdd94a42b2a42b2344e58cab : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96364a244c6464caaebbc52628d417f : SUCCESS in 11m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7ad950110df471db0be2b7be5136a11 : FAILURE in 44m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f54f93ed6270487a8b12991e568c4169 : SUCCESS in 1h 13m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ecdb423383f34c2f8f028ceba083fd98 : SUCCESS in 1h 02m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f618a8f37a5e4c12946b11bef69f22bc : SUCCESS in 47m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9a1d77ab67040b58ff69b0904b1702d : SUCCESS in 1h 00m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f68ead5b8ab463d865335f98dd70ebe : SUCCESS in 1h 06m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8b4d8f78d4a4e968b71410ce45c00de : FAILURE in 57m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66c19dded951498da9e95b1a381b7e7a : SUCCESS in 1h 41m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5da5a4c5c4b640a0b441c436f23a69ed : SUCCESS in 1h 03m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/17154645e6744b9c8b8af4f60c3a9d03 : SUCCESS in 1h 02m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1efbe2a68849b58fab5edec22acb2b : SUCCESS in 38m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed53e79828494e5e95b6bf6a636b90b5 : TIMED_OUT in 2h 12m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f6b5ea48c954965bb32a3d77d8f1025 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e27c3ed56071f96b6870c735a2f556c85cf07f8c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-07 06:56:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"24461da8a434b188f4c6aef7a45179d5e4eb37cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 08:01:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7629e19ad2004d658dd67ebde1d05822 : SUCCESS in 59m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51b7daf1823f444b9707fb17c587e32a : SUCCESS in 1h 00m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0927182e7774844d1bb7b27c021d4951840887e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 09:08:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c02fa3f654d04e688f582cf8e1efac3a : SUCCESS in 48m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2331f480c1d947719bb5a2a3b81e9382 : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ad48906beb9434a9ec78eb74c97af92 : SUCCESS in 1h 36m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7d26f872c1a47fc889383ff0ef15c7e : SUCCESS in 1h 37m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e168489c9abd4eb5b4f8a606fbb06bb1 : SUCCESS in 2h 01m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d74edb4deeff4b26a9ce6e72360f98ed : SUCCESS in 48m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96bce3173c384bb6b4d0ebae838786bf : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c705bee2d944fefbdd981d64e84784f : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daefb64c81e9426098176a77378dbccd : SUCCESS in 35m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80e53e8afe1842fe8a4b016b4a6e0882 : SUCCESS in 33m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f5c902431234868a016b6981937938f : SUCCESS in 37m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ed6cbd8ad740bfbba71d112500725e : SUCCESS in 7m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e67d6a601bcb4c34a5b0c1beaaba1296 : SUCCESS in 48m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35317cc4f89340aebcc5faac0bd99a91 : SUCCESS in 1h 44m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/96a3a8497a0a4238a1ecba9ca838f881 : SUCCESS in 1h 42m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/28dce1d16d224fc2b6d42008c2b130ce : SUCCESS in 1h 31m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/63ec0109a6594a5ca5e20cc0d5bf6673 : SUCCESS in 1h 06m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/828f4c134734463b997c2584c3cb77d0 : SUCCESS in 1h 17m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39d46b7575c3413a9c238fd498bdf1ff : FAILURE in 2h 07m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/722b0916827346c797741c2880f32100 : SUCCESS in 1h 33m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6aa7a8d672bd48a59377c95f2d2f4c06 : SUCCESS in 57m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb4d8132e30f47b8b810efc91a249afe : SUCCESS in 1h 38m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4f7fd78ce2a942efb7aab6f8780c27d0 : SUCCESS in 34m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2016a5548ee0496ead186ae3a6cac23a : SUCCESS in 1h 56m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7f67beec75e04031b83af1dfddb27045 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":4},{"id":"de6ee664899fa4315f0dcc2194e6786050d5be9a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-09 06:16:23.000000000","message":"Patch Set 4:\n\nrecheck multiple ssh timeouts due to no DHCP lease in dvr-ha job","accounts_in_message":[],"_revision_number":4},{"id":"97ff9c1f7c845cf01f1301df9ded34f93699bd21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 07:25:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8c06dea91434d9498d8164e2e54bbf5 : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ddd1dc6800c4da590591e989b030f10 : TIMED_OUT in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"06e37a3596e5ba1ab3ca6fdfb1ec281240e71f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:45:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/86c7f614a0d64c88b96b6904f14221c0 : SUCCESS in 52m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fdf4059d10004390ab3b06baeb64eacd : SUCCESS in 1h 45m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/86424941eb8d4d3e85d78f984bb8bb70 : SUCCESS in 2h 01m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a1f8dc63be3e4f56b6fd613e17bbd5ba : SUCCESS in 1h 58m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e0e3465db00b45b791ff1a804c0be1c9 : SUCCESS in 2h 06m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32b1a844079b4975bb9e088936909adf : SUCCESS in 44m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/270afcec0c5a47cd80cd12bedd43a01a : SUCCESS in 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff91c4f837564fd29b239123a6f7600a : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcd995fe2c3749eea06fed72eebd97b0 : SUCCESS in 34m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/207fa3f2749c4d14a0824ddeca6faf35 : SUCCESS in 38m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd71b815a3eb4ef6b2a280f606ce693a : SUCCESS in 37m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb56d3d834d422c841b99d29d27af40 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b33b8b4b8324e39a9da131d31e12e71 : SUCCESS in 49m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1566d7773f5644768308986e29079c60 : SUCCESS in 1h 27m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/622dae646508438481fab62c5f34ed42 : SUCCESS in 1h 40m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d385ddef03bb4c1e9a050f95fe1bc42f : SUCCESS in 1h 04m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4547b4e91a87420d95e714cab2a5a60d : SUCCESS in 1h 08m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d722fdfc07354a1ab65a11563ef342e2 : SUCCESS in 1h 22m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d773f8ff4ad475da764865424ddc642 : FAILURE in 1h 13m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9eaf443dc8c240c18c22eb5ed3f160b6 : SUCCESS in 1h 11m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/121fcb84ffa54dc4bc23dcb6cee6bc8a : SUCCESS in 1h 07m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef95107fdf7c426a88dd3c35335a7c53 : SUCCESS in 1h 18m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ac3dfddfb23b4a5488689906ee70f372 : SUCCESS in 29m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a85a36abeb874b8eb2caecfb537101fe : SUCCESS in 2h 00m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6fd4f7baa41455194a6d7b15100bb80 : SUCCESS in 1h 44m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9527e29116f3eb4a4766302b0bb119fc546dce0e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 00:24:56.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/84/778784/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"bcd07e651c7997b37e54237228932dd446e089ae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-11 07:19:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"09e2985e0234aaceff6e57cda56d373fb6286f55","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-15 15:44:14.000000000","message":"Patch Set 4: Code-Review+2\n\nFinally I had some time to test it. I created a network with two segments. When I start the second node, I see how the scheduler adds a new agent.","accounts_in_message":[],"_revision_number":4},{"id":"d75ae1ba6bb974d89ce219326b63658c282390c0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-16 05:03:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ebe442a19b0b924a67705d5e421ef72253be3c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 05:03:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e3f1bc5a5985bcb842201356843bc80bf455d241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 07:13:32.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d808c37efd3549129cc0c798a4d21645 : SUCCESS in 51m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9e466635b53241d7b599a6b325f6ed5a : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ce0b3eb6cd64b4fb54239af9e60a48a : SUCCESS in 1h 48m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cbc6cb7a0f904ba4afd5eb24a3f8b006 : SUCCESS in 2h 06m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/132eda82cd5c46b497bff9908f64f4a9 : SUCCESS in 1h 39m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dbf356b60694766bdefbee120d3cbda : SUCCESS in 39m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b726f7beae47f7a9eff3f419860470 : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70ecd2b0135a4e3aaa607be820ded5ab : SUCCESS in 37m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c28e915ffdd4209b8de868b4725b21e : TIMED_OUT in 41m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e79a4c5b8884253888ae10c8a718de6 : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af09cc38ad9c4129ba3aa74f140effef : SUCCESS in 49m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/46f8900afe4f4a24852552031cf72903 : SUCCESS in 1h 23m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7a9d8569ad6f473e986adafadf9228d9 : SUCCESS in 1h 30m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2c083b423b540d58ed6375a34edade5 : SUCCESS in 1h 04m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89994690533349a5acde39d35f1b28cf : SUCCESS in 1h 06m 45s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c890f65cd89f4c80b8964139c100933a : SUCCESS in 1h 30m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e89aa9ab04884b3b831132aa99f56841 : SUCCESS in 1h 09m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/958c793cba724dcb8f8230c4d4524faf : SUCCESS in 1h 17m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/183c4de9d8a34371b1d0827e01aad11d : SUCCESS in 33m 38s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6b2d22eb13e4c308e3e3f12c462e923 : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":4},{"id":"a8bd4baeec41d8e951dff9c2811465fce83308b4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 07:16:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d599f0a120fabec9a8f22364bb0bbef108cee0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 07:17:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98a9832f137d244c72c847dbf2818b7c673203b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:38:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f6c131945a1745a08607cca133d6013e : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/def4109a52094031b3a3074a93174d0f : SUCCESS in 1h 56m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7758ff1c80bd4bb0bcf6cccb37483485 : SUCCESS in 1h 13m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4f5e63b8e414d6eaca0aeebfbb363eb : SUCCESS in 1h 52m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9da1148d6fbb42ba969d34e76beef489 : SUCCESS in 1h 23m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4260bdd910764db5a635c06a0ed2cc36 : SUCCESS in 35m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/158d3d98bb7440a8bce9ae9494d89698 : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b906d33d4c94bac8668fc728468b967 : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab78905a7224671b3b3744bcd63874f : SUCCESS in 24m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/735b09d508204ba399bd8c808585841e : SUCCESS in 38m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52b8b85600f04bdfa94ddc01928f54bd : SUCCESS in 33m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b671c7a8802841db8705464d3661b90e : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/769e39bb926e4bc59eaf41a596808467 : SUCCESS in 44m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c89741649a84124b28e4876f1724b8f : TIMED_OUT in 2h 16m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a01978b90cbd4d84b69a82c05cdeb1a1 : FAILURE in 1h 40m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d718d5506fe446358d7e9ecb94b2bce2 : SUCCESS in 1h 06m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c5ca784de8404a439995540eb7e9df34 : SUCCESS in 1h 17m 08s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/12125b870eff4d569db516f727252117 : SUCCESS in 1h 08m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d71ef4ac2f44445f97daafeee60e6ec8 : FAILURE in 1h 34m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b07baf7dc43f441ea742c5639fdfecf2 : SUCCESS in 1h 15m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d581bc9ccbd40d58209aa366ba0d169 : SUCCESS in 1h 00m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b010720d4bc14a329b94af13f1305990 : SUCCESS in 53m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/791ec3f51b1c45ea86e5e0199633f779 : SUCCESS in 31m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/71e540ef67eb42c7a5e6c3ac892a79c6 : SUCCESS in 1h 05m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c3fc18a1e894de6b01ef90b26791331 : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":4},{"id":"31b0cf6341d5d6476d969e421f1325e043059da7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 10:10:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bff29180a95affa90a38a5cfced7da7dc4523737","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:01:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc83dff919ac4c7eb173dacb9fb65f5d : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75a0820e00f14418a26722cc2c4f8753 : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aae54b93e811ac2bc30d28d92f14d30efd700938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:02:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c0f8e8b1164436096765617a78c127b : SUCCESS in 54m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d2674934533c4d27a7cc93b0fd53cd0d : SUCCESS in 1h 45m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4104ce72064345d9a32a67fa374774eb : SUCCESS in 1h 35m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95543b206afd4d5da51a691a6b830e5f : SUCCESS in 1h 26m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8ce2a9152cde4c298a42aeca123287ff : SUCCESS in 1h 51m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/664779810c424617be8db075bc2eaa4a : SUCCESS in 42m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0b47148b0e4f7db9a8f6df6a7310bf : SUCCESS in 33m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc12a919ae934606a48db6b18d81e244 : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/619d9a8adafc49f198eafc353cf9b992 : SUCCESS in 33m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e8ee738c8614448bad80545547ea288 : SUCCESS in 32m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da12d23392d477fb50602d94c643dc3 : SUCCESS in 34m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7881904bc134263aff3a40d4430a052 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15ce57e339d142c696a8653b3140a3a4 : SUCCESS in 52m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c5c2373a75b431c836f291bfb6f2b8f : SUCCESS in 1h 22m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a981e0a30c24600919949d9206da35b : SUCCESS in 1h 31m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34993467ba3d4e0d9d4cc21b0ef6bde3 : SUCCESS in 1h 03m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c4d2ceb236d4d1d9a74820df7eaac08 : SUCCESS in 1h 05m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/184248194bf742b58aeb50b2c7a4b950 : SUCCESS in 1h 26m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6602c984084e4a3c9b8270de858409c1 : FAILURE in 1h 22m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fed603f84c4140e78c73491e21c67021 : SUCCESS in 1h 36m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1215bc208664cba98b36298c1c74ef3 : SUCCESS in 43m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6d6cfd712983448893d4e38cc7f924c9 : SUCCESS in 1h 44m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c914514f873f4a05b24b16272ce0ec06 : SUCCESS in 34m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/57e60c1f3222458c9698e3ed10c63807 : SUCCESS in 1h 38m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8b39db3d19cd447d907beba688657c58 : SUCCESS in 1h 41m 05s","accounts_in_message":[],"_revision_number":4},{"id":"d5772a611b06727b47b9f5db1ab142bee69e8ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:03:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ff8beb9ce68e8ae0db76860bafc0df2e7f368472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:45:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af48283b1522446e853307ddca911da0 : SUCCESS in 59m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/07c66c2bb615481db0dfe8dd2fb7ccab : SUCCESS in 1h 38m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e120425aab4842e49f3a052c181f9e4f : SUCCESS in 1h 45m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/93ad346c6d9a47e0938b43702df8b134 : SUCCESS in 2h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e37809c6a2a478fb74e09001631f1e1 : SUCCESS in 1h 19m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d325556ba804178afba54e4d165c1d3 : SUCCESS in 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bacfd033924f658b254964ba26c75e : SUCCESS in 15m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3767be8173ce494a95afa0e00f836cef : SUCCESS in 36m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40542a21e5b45d39c6f363913c7cc52 : SUCCESS in 31m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533b211d0a94452d82fa764aa888a6c3 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbc73a049a4246d3b9f45f5b0398efe8 : SUCCESS in 59m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/537be15465ba44a38263eabf50698af9 : SUCCESS in 1h 34m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a3ac7fd6cc84bb796fcc429972cae0d : SUCCESS in 1h 19m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a6e302763c14ef4859de07cfb779183 : SUCCESS in 1h 07m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4a143b7930f48928b9ab10c4d5be424 : SUCCESS in 1h 08m 15s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08c03b1b56b741ab8792a0df377a072b : SUCCESS in 1h 44m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f03ef5d8c2e40e2b51f351bd46678c2 : SUCCESS in 1h 01m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e220b35d61e64402a809f7cf2e9ae2ea : SUCCESS in 1h 09m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/296f990cc3334c8796b987edf35bdb72 : SUCCESS in 26m 41s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8701b237b261411ba12d00f4aacf5227 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6c49da8d5580fd97e87136196155c5a2b9ea3727","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:45:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4d72d761875e7931db6f6797ec5ea879cb065398","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:51:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f8932159284f26861c4b31368cf8da : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5c931f29132c3c5c01127f089f7d5909d2a19e4d","revisions":{"c794cdb57e55afd336f6608850e6a5db768e8126":{"kind":"REWORK","_number":1,"created":"2021-03-04 22:20:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/778784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/778784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/778784/1"}}},"commit":{"parents":[{"commit":"3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5","subject":"Merge \"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:18:16.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:20:34.000000000","tz":60},"subject":"Schedule networks to new segments if needed","message":"Schedule networks to new segments if needed\n\nIn case when subnet in segment X was created before hosts from\nthat segment was mapped in the Neutron DB, network wasn\u0027t scheduled\nto the DHCP agents in that new segment.\nNow DHCP scheduler is triggered in such case.\n\nCloses-bug: #1917811\nChange-Id: Ic9e64aa4ecdc3d56f00c26204ad931b810db7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c794cdb57e55afd336f6608850e6a5db768e8126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c794cdb57e55afd336f6608850e6a5db768e8126"}]},"branch":"refs/heads/master"},"a4ada858c9eaacba5fe8566906aaeb4c3e363621":{"kind":"REWORK","_number":2,"created":"2021-03-05 07:05:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/778784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/778784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/778784/2"}}},"commit":{"parents":[{"commit":"3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5","subject":"Merge \"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:18:16.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 07:04:50.000000000","tz":60},"subject":"Schedule networks to new segments if needed","message":"Schedule networks to new segments if needed\n\nIn case when subnet in segment X was created before hosts from\nthat segment was mapped in the Neutron DB, network wasn\u0027t scheduled\nto the DHCP agents in that new segment.\nNow DHCP scheduler is triggered in such case.\n\nCloses-bug: #1917811\nChange-Id: Ic9e64aa4ecdc3d56f00c26204ad931b810db7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4ada858c9eaacba5fe8566906aaeb4c3e363621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4ada858c9eaacba5fe8566906aaeb4c3e363621"}]},"branch":"refs/heads/master"},"b01c3c1a8ec0c9684a2ac36a00aee9e7a226263e":{"kind":"REWORK","_number":3,"created":"2021-03-05 10:29:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/778784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/778784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/778784/3"}}},"commit":{"parents":[{"commit":"3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5","subject":"Merge \"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:18:16.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 10:29:01.000000000","tz":60},"subject":"Schedule networks to new segments if needed","message":"Schedule networks to new segments if needed\n\nIn case when subnet in segment X was created before hosts from\nthat segment was mapped in the Neutron DB, network wasn\u0027t scheduled\nto the DHCP agents in that new segment.\nNow DHCP scheduler is triggered in such case.\n\nCloses-bug: #1917811\nChange-Id: Ic9e64aa4ecdc3d56f00c26204ad931b810db7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b01c3c1a8ec0c9684a2ac36a00aee9e7a226263e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b01c3c1a8ec0c9684a2ac36a00aee9e7a226263e"}]},"branch":"refs/heads/master"},"5c931f29132c3c5c01127f089f7d5909d2a19e4d":{"kind":"REWORK","_number":4,"created":"2021-03-05 16:05:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/778784/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/778784/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778784/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/778784/4"}}},"commit":{"parents":[{"commit":"3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5","subject":"Merge \"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ce1ca690cd18bd1f2c54bfa41a1af84ebdc0ab5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:18:16.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 16:03:49.000000000","tz":0},"subject":"Schedule networks to new segments if needed","message":"Schedule networks to new segments if needed\n\nIn case when subnet in segment X was created before hosts from\nthat segment was mapped in the Neutron DB, network wasn\u0027t scheduled\nto the DHCP agents in that new segment.\nNow DHCP scheduler is triggered in such case.\n\nCloses-bug: #1917811\nChange-Id: Ic9e64aa4ecdc3d56f00c26204ad931b810db7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c931f29132c3c5c01127f089f7d5909d2a19e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c931f29132c3c5c01127f089f7d5909d2a19e4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
