)]}'
{"id":"openstack%2Fneutron~778929","triplet_id":"openstack%2Fneutron~master~Ic3ddfbfd0df59ffd10576f02730202c8b740ed26","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ic3ddfbfd0df59ffd10576f02730202c8b740ed26","subject":"Add new policy rule SG_OWNER","status":"MERGED","created":"2021-03-05 14:11:01.000000000","updated":"2021-03-11 13:54:58.000000000","submitted":"2021-03-09 18:47:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778929-secure-rbac","meta_rev_id":"f9b7eaae0c6ebe5b14086bc78462a53dfdcd1115","_number":778929,"virtual_id_number":778929,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-11 13:54:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 18:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-09 09:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-09 09:53:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-09 07:37:35.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-09 09:53:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-09 09:53:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 14:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 17:14:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 19:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 08:18:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-08 08:30:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-09 07:37:35.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-10 23:19:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"830dbf190470afb0e2c543d49bf134c83b2f4839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 14:11:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63db155fbc9853539f6be865b91fe1acebc9df41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:19:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c1b91efe58b0527bff521683a16257c244e225e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 17:14:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"3d345082aa9496ef52549c224318d2fff1b113cf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 17:22:32.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"7521d82105842fde199a5fa3ead869fdf6773ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:43:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db774493230740c69395b5eebe95926c : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e438307ad48a4ee29dae40dead3f2806 : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2fbee6c13ebd44a29eb47727ca7f74d7 : SUCCESS in 2h 00m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cf833dde1e4a46979db21beace7c6e8b : SUCCESS in 1h 42m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a850c54f6cb84da59f882df41c3d965a : SUCCESS in 2h 02m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de90f0fd76f04aaaa4c8dead4f2709d4 : SUCCESS in 39m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aed671aef954eb29b83d6184c41308a : SUCCESS in 34m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0872dd5f73bf42149bb46e17f02d8e3c : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/113e87095f2b460ab0653a7a75c7d1d5 : SUCCESS in 38m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afe32d64c2e14b1999720d7274ec93ec : SUCCESS in 34m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef6a73b22c44222ba188d62426fc9cd : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5e426ac5a741988d271d0e607f04d5 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa18d08ff40a478b8ea2039b757b9198 : FAILURE in 17m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0d22176ee4b4ab1b5e0f8397edbe072 : SUCCESS in 1h 31m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c70d0d33b12245fb8fb93a623b483868 : FAILURE in 20m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78cb8ccb188840e29339e58774a61dd0 : SUCCESS in 1h 08m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2c603a85fe524bbbbe574c9c8f7bfe3e : SUCCESS in 1h 08m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2510649d6d484d2e88c4a16910a39d80 : SUCCESS in 1h 33m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/33dfbbd252424efdaefe34c9b296db12 : SUCCESS in 1h 30m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b920a87d9f4246278a4bf31a623dac48 : SUCCESS in 1h 53m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dac1bb3d9944a0589abfaf7908d06a2 : SUCCESS in 1h 04m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/86dc054f3c2f49a39853a4e898bab1a5 : SUCCESS in 1h 26m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6cfeb263974def8b2b158bdf145084 : SUCCESS in 34m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aacb08a113fb4212a7121c0e5086ef7f : SUCCESS in 1h 29m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0a4cf447d4ba44119c37f3d4cffd4529 : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"991c6a56ab364ec97108deccc31ec30f8f621a46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 23:18:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6b8dcaf081e562a9dd4f00892015b6596f97e1b0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 08:27:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82fe316f6174a37ec0bf96e1ade850aec852f07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 09:40:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12c8572e27284d71b04557a6b10719ca : TIMED_OUT in 1h 01m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7e9694a275c4bfeb346008165465adc : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecd4109dae05e3d6575b65cf30898cf4c1cdae1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 10:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a817d306edd4d21a0a73ee5650c6ef8 : SUCCESS in 50m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e01aa7a8a1774fe5baf8129e158d5626 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/65394e519466499c9ca91ee6425c5ac8 : SUCCESS in 1h 24m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2bd3f646d8f84b4081be6dad7f899ab7 : SUCCESS in 1h 33m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1917a0214bb24424851d5a8372570bad : SUCCESS in 2h 17m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa6a5cc1770a41abae880e65c658024d : SUCCESS in 47m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29a0e8c93b5a49098aac37ef7c6e5ece : SUCCESS in 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731b9b2a7bc14eeba26961621adae600 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726bed1aae1249059bae5d2b0a1299d6 : SUCCESS in 41m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bb24f5c243481f8700a087a2083737 : SUCCESS in 39m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ede4e990fec4436a99da40a007ae574a : SUCCESS in 32m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e31f90f95f402ba959f4ba108df48e : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecc6573019704526b18ace9500409b73 : SUCCESS in 51m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/146768950aee462b9945db8f5b488027 : SUCCESS in 1h 22m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a3fade9ee7db423e9df732becb6fbc2d : SUCCESS in 1h 38m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ce281f30c0644ea91733e86ee280da6 : SUCCESS in 1h 02m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/709ed742cd694625b268d1e40b7d872e : SUCCESS in 1h 03m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8dc30fb352134a519230338da80c65df : SUCCESS in 1h 06m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/031c243520f24f8396de6bb604e477c4 : FAILURE in 1h 25m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5431ffd41a624ac9890de0cde8deb9d9 : SUCCESS in 1h 26m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04b498d41aa442d7a9c21a59773a27af : SUCCESS in 1h 13m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/167004814cc94d06985fe68f0353e703 : SUCCESS in 1h 00m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f815f364034f4759b4fa5b24c6541b28 : SUCCESS in 35m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e6dda287098414d8b145de7ecd5cf31 : SUCCESS in 1h 21m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/07365db257a0422f8535c995f8cb0c56 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c3fab9e72df5a2b8bc9a40445446e1f6d98a6cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 16:47:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a056028925a60a424e4af7a4480ccbcad55a112c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 16:49:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67b03b443f469f4bf868220d4cc30b32059180b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 19:12:36.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0389c4c1c8874940a6e9220476ea8263 : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0db955586e7247c8bcc5a166d9d861a3 : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/34d55c4179034d5a944d12de268830a9 : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b14a44e10ac94be8b4da8f7e979fb8f0 : SUCCESS in 1h 18m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8009a20d956c474b8ca98f0546a6f6b5 : SUCCESS in 2h 20m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ecf8c0f73b4d5e8f45c8c07564499b : SUCCESS in 31m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/850d5a9ffe494382903e0dc2a484f30f : SUCCESS in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308ee21ad93e432f9505eecdb92ebee4 : FAILURE in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ff6d65552042bb889ce20a48db3257 : SUCCESS in 30m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4056ade1183b44cc94329612f68300cb : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d5529b02c64840b0bc9625b4b36dd5 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9423e3eb4286432f8de03d72a447ead4 : SUCCESS in 8m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e88a13f92d0413c89f4415d18f41950 : SUCCESS in 51m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ca9437f93ae483c90d999b0dda6e031 : SUCCESS in 1h 20m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/449bb001711a4f9d9e310a5f2334cd49 : SUCCESS in 1h 40m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a70f0e55a545454d8c3d6cc93c33290d : SUCCESS in 55m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6243e8370b3460fa04e18cedd70c881 : SUCCESS in 1h 08m 25s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/292493c0375747e2b5cc6ae6f523b571 : SUCCESS in 1h 18m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/624441cd87214af1ab8bd9539dbb49e0 : SUCCESS in 1h 27m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c36dd30f660f4775b64e6f28591afaaf : SUCCESS in 1h 35m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/775d0755410940e6ad50af2684f7eed2 : SUCCESS in 1h 06m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/57b3b12d4c164f34bbe002e00311865f : SUCCESS in 1h 26m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f7e80637bd48a785afa2c03eabea89 : SUCCESS in 32m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c028dbd2127c4188b36a417313d632ff : SUCCESS in 1h 37m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/92b236a1dfcf4631b9debc32ac8e8df6 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ad5881f50958e37be4ff99c73e4d7ac0cbeed6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 19:52:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8c80d0d2fb9a3a0516e0379836638b42956b5edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 20:38:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bf159fd7e954305934cc70e95ab20f6 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/764d3541d75b4a62a63d4b4e9dfcfbb5 : SUCCESS in 42m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f199c4fa5e4f7515140841c9a214a9efa5837bea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-06 21:54:19.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ffd5d3b99d43d62707408bfae8c0687162adc10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 21:59:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0b84cddf65f4351b1c9495e426bcdad : SUCCESS in 38m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e95fdccaecb64c1daa433811a4c971fe : SUCCESS in 1h 35m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/08d39fc05ed24244b8db39d970366088 : SUCCESS in 1h 51m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8394eee8430b4846b1d566d055e1b4c0 : SUCCESS in 1h 28m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7587990340434ef9aefc847837ff65bf : SUCCESS in 2h 00m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f917286a93422bbe1b56348d9d0063 : SUCCESS in 37m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab6414cddbd04483b16b492fc952ef0b : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f50e2c5ec649d5b34f9c6843f6543b : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984338bca36a4939926feae2e957e2b0 : SUCCESS in 35m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b95923f20384ba79122051d880ddb70 : SUCCESS in 33m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50fda2d796bc4d84a5a986f35ecd6de8 : SUCCESS in 32m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f1b7e0a9484a5caf2b0c8cccf39f82 : SUCCESS in 8m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/512ed5d853de4579b6cc1c472325493a : SUCCESS in 50m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1a848e3be14d0987f96c2ca1614091 : SUCCESS in 1h 39m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a787421f355c41b2b8b2054e67f11327 : SUCCESS in 1h 29m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/783a85decff5437dbec79068f034ca29 : SUCCESS in 1h 08m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbd098eef9e24e5785a7da2f4da3d32a : SUCCESS in 1h 10m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b075cd0d23aa418ebe9fbdefef4b8187 : SUCCESS in 1h 16m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aaf233f4e6d54a04b151afe81fb8c475 : SUCCESS in 1h 28m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa16c892b0d143dba824ed77bcc7a919 : SUCCESS in 1h 50m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9fd6f7958684284ad2101626c1deb7d : SUCCESS in 1h 13m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ceec8aad1a064f97a90e71b070063442 : SUCCESS in 1h 25m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d389f367ed0e4000b0bb11772f7da05b : SUCCESS in 30m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ef638eb62ae4f59a527b94761cd95ab : SUCCESS in 1h 39m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5349a98e25de43c0b5c0d5cf9e22a505 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2d7a9c1d3ed0fcc2cebc11bb6f55d60a39af2661","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-08 08:18:36.000000000","message":"Patch Set 3: Code-Review+2\n\nYeah, it makes the code readable","accounts_in_message":[],"_revision_number":3},{"id":"f5fdb912a13c5180a577892a6e3e0a437e822868","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-08 08:30:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"431484e75a9ec0e910186225131a11e41c88818e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:43:42.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"9db9524bf078c49c715323e8c267989171188c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 15:04:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c541bdde73909b962850d5decbc0a51b77855a4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 16:04:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7323d5c54e4e4e83b775ccf4597bd97e : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/863b001a05174334a2ab195a6c0fed9d : SUCCESS in 46m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2bcff7eb98bdc573d0a25c25b1b6deb17def4d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 18:46:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5c9af1fe00df433ab23e8db15674cdf5 : SUCCESS in 47m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d8ec89b0a0434ec8bdfbf66bd88509dc : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a1bee3d22f148f7a660d6e008a6de22 : SUCCESS in 1h 48m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86e07205a1204a1b901cae7b2535fa71 : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5e169ec430641fb83c00efba7c020e7 : SUCCESS in 1h 40m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50ebc95ed99e4034b5a0d7b3a214c201 : SUCCESS in 46m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59039a0170e34bb0882d62498b0ef021 : SUCCESS in 47m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d80da79198145f6914d4f296285d099 : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/762d3a35ee3a4736bb2d651c02ade8d7 : SUCCESS in 39m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a87f2cf0d244cb58b594bc07ecc0975 : SUCCESS in 36m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53bb727e30fb4fb69e9451918d78ef1f : SUCCESS in 40m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc38cab60e374a32b582e41eda09dcc2 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3b60345bb6148e98321bcc7a83cedf5 : SUCCESS in 55m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/44ca7ff563724b5eb9fc8500b5255afe : TIMED_OUT in 2h 15m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5b9180f5b5142cba4cf22405701c495 : FAILURE in 1h 35m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7109cfad7431401b8aba991054fe9d55 : SUCCESS in 52m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/38edd0762c5a47ee98ecf97c786c7210 : SUCCESS in 1h 03m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/72f90de595874826b76ffd46b77c7e1b : SUCCESS in 1h 21m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a66540a910714e7688d8d94bedab5074 : SUCCESS in 1h 31m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c252030a93314772baad2f753f3e47ed : SUCCESS in 2h 01m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb0e2ebd9aa744029ec821834d79748e : SUCCESS in 1h 08m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b4ba6fc04d8c4280b1b79bc066528180 : SUCCESS in 1h 27m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d587eacbc55a4f0ba87d974bedeeb3a2 : SUCCESS in 41m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e95c022356934463a3d78efc3826d0a7 : SUCCESS in 1h 51m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6658c995fe54537ad81a4d78842aa66 : SUCCESS in 1h 42m 56s","accounts_in_message":[],"_revision_number":4},{"id":"918d51d927abebf99fb25b8b878b7ff308c6cba4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-08 19:44:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d04fc032f6c68c626b91d68737b08ec22a2e63b2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-09 07:37:35.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b2e89ca5164dcd5d0d655cc54899d15c814cbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:32:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3626eb28b2f342aab105429b7d9d4218 : SUCCESS in 38m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a70c5c51d0d409682f7603cad3cde71 : SUCCESS in 49m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6701e4135ee84d2e422aa6aa2d50d56d9aab6bcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 09:34:09.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e17e1239ef08516310cdcaa1318b4639343f47b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 09:51:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e726488a8e34731b7a09f449fbe4e2f : SUCCESS in 53m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c94d7d9a5681401ca5c6126db47a9e87 : SUCCESS in 2h 08m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fac24ffb82324e10bc1e5380f9891ba8 : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f33befb52604b2cad377b68594b7706 : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4abe917158d14da196585169de3273c6 : SUCCESS in 1h 36m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09a3d42407da4c75957ae49d250cc1e5 : SUCCESS in 41m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8a37f86a0f148fc8899f34de099d351 : SUCCESS in 41m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63cc5ea831b54e1f9e65bc8b740552f0 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e79bda6d795948b1a3a1015450e78a3c : SUCCESS in 47m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab05e0e17b44bc9ba8dfc3c1cfb45a9 : SUCCESS in 29m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ee96465485465486ceecca4eea9312 : TIMED_OUT in 41m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9edcc188cb6b4b2abcef2028557a0cc5 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b015b0b3517649dbb8c8f785bc96fd43 : FAILURE in 51m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca2b70dbc2e748deb821cfab85257be2 : SUCCESS in 1h 24m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/62b228e22077475ab73abc00cee0987c : SUCCESS in 1h 27m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e90431ddeb63449faca8d373b393cc0d : SUCCESS in 1h 03m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1cefc8476ec94b058ee43a31f976ae2a : SUCCESS in 1h 12m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15b6a42a49064e478a6cfd88da01caf0 : SUCCESS in 1h 15m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3291eafb356b41759beeb8dfc14e656b : SUCCESS in 1h 23m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef791f24ef504388b67d40ad949db694 : SUCCESS in 1h 42m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a594cad2c8b44738bc1017e727fcc583 : SUCCESS in 1h 02m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/54741182d2c44037803ba26e67502dbb : SUCCESS in 1h 08m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5127ec25d4fb4b9e91691cb3b5b26d87 : SUCCESS in 42m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/60250211ad35402f98cf31c1265d0a28 : SUCCESS in 1h 45m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f9bdb5eae20e48bab402211df78b123e : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":4},{"id":"775aadd70afac89d4b0e6b9751294eef3eae2586","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 09:52:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f1c77c68d6097fc18639d5fb2d2bfbf709bb782b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 09:53:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"02f9f9e0694531c5f54a75d3b53ae02a4fa13ee2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 09:53:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"63909e9dc2f65b72a9e4ac8b0f8161d1ac1865ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:24:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f21bf31506f18031adfcf6a0434bd7a3f32a1c65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 11:02:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/778929/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"60403f1ded029ba86799934a72b3bdf0c0d3872d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:58:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e0f00cc3568842bab6247bb16f689816 : SUCCESS in 45m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f303a10241214f33930c5e9a3f67995a : SUCCESS in 1h 39m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4c1dbf6a409c47f8b8150cb940e6ceb7 : SUCCESS in 1h 39m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce6289cc4d8142b7aa620537f2907f5b : SUCCESS in 1h 32m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a5103858e2ec46afb8f2bdf06ead1d63 : SUCCESS in 2h 20m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e2e059438f4f58bb5c1e3e93932369 : SUCCESS in 53m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7d7316de8ed486ab290054fe01ab4dc : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7737c759ea4048279e9797804b040a2f : SUCCESS in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28ac74ab576f411888b683f63ab0f61f : SUCCESS in 33m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98aa69a3de2141039388799bb094c09e : SUCCESS in 42m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16d93d71c08a462d810dcafad9b43a4f : TIMED_OUT in 41m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f846489a27c452c9fd0d81583a8cc33 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b70d317ec6da4aba9e5566aeb2e421d0 : SUCCESS in 46m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/557990a9b1034b8c8ed2f7e5ab498c58 : SUCCESS in 1h 21m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/939325e40f8047369896311908cc665d : SUCCESS in 1h 32m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0d073d65bdf4162bfa853eee01e9820 : SUCCESS in 52m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01fd255ccdca47b19e7fcd17d4b5f78b : SUCCESS in 57m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7558d09d89b4acfb9424b8f23b01615 : SUCCESS in 1h 17m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/192d40d820354ac48624acec41524e4e : SUCCESS in 1h 33m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da3f7773c2ee48b5a37917bf4d48a849 : SUCCESS in 1h 37m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02989760b42845c48e839e5647ab3463 : SUCCESS in 1h 07m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7b9a7777a61b477d83c84e2d497fc81b : SUCCESS in 1h 08m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e18a43b53c4cf59d816878d546a57c : SUCCESS in 37m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de2ac555d5614bbcabfa16ceb60c8267 : SUCCESS in 1h 46m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/849f2c4aea194f25a10521ef8cc91450 : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":4},{"id":"b8458cf28d70a5022e663986bb84855c6b7317a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:59:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"acfd08ba8ed94c7dee7fe11b09acac38bc64716b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:47:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/add2eba7acc94680ae53c916fb7b6650 : SUCCESS in 57m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/91f433e688e844f0b82b591cf951cec1 : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/52075358207941e3817d62ccef122b67 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cd43d3dc9ced4fbe82119c5e63878a5a : SUCCESS in 2h 18m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88516c07dee44a5d9315b9c87af317c0 : SUCCESS in 1h 20m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/752d854f82f34e4597b1faacf23d95c9 : SUCCESS in 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082b2eb36c064d6dafb634bc9615cdc7 : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9846413fe9874235a2528d977972b707 : SUCCESS in 36m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c39fba4dd014dba9d60cd627cc9b991 : SUCCESS in 33m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57b556b2f494e1c89ad66d76115a5e1 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de44188ad16e47d9ab94e5d5b136933d : SUCCESS in 48m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c27783f16dc544f8b33c2f47818727a6 : SUCCESS in 1h 22m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d5bc779dcaeb46c1a0117e805a9b001a : SUCCESS in 1h 20m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7994c17c83204652943dacaf91dd44a3 : SUCCESS in 57m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/162d06c8b09840ed89758674b45c2174 : SUCCESS in 57m 37s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87c32d80165f4c62a7d75126d4a5886e : SUCCESS in 1h 39m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/103baec1a6054f23b792fec460ca55aa : SUCCESS in 58m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c898c4f6adc9479d8848fee4db1431e1 : SUCCESS in 1h 06m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6bfda7f1bd1d403cb6b826e1c9030ea9 : SUCCESS in 36m 29s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e7944eaca0af4541951d78ff6ffb9af2 : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":4},{"id":"ec3c42b6567afa5b7b1292ab9a8f388cb7acf1fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:47:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f763361549e5d827045355c1351896862e76cde0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:55:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d781035e01476da8a6565debb64d27 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c387324351a7f57cbf66a6d942cd12c45b58d0e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-10 23:19:33.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/29/778929/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c4e01bf60a73292ea1e5563ea2560839f0985ddb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 05:07:41.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/29/778929/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5a7caa00fa57e82494b380c12855ffa314ac67ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 06:37:56.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/29/778929/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f9b7eaae0c6ebe5b14086bc78462a53dfdcd1115","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 13:54:58.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/29/778929/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"90edcf64419c708424c26f31a68e5c9c39bb7993","revisions":{"34b93e77f155403128cbcaabf9f2b4369c684070":{"kind":"REWORK","_number":1,"created":"2021-03-05 14:11:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/778929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/778929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/778929/1"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:08:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:08:19.000000000","tz":60},"subject":"Add new policy rule SG_OWNER","message":"Add new policy rule SG_OWNER\n\nDuring the migration to new, secure-rbac policies we already proposed\nrules NET_OWNER and PARENT_OWNER to extract \"ADMIN\" from the old,\ndeprecated rules.\nWe missed to do the same for SG_OWNER so this patch adds it and also\nswitch new SG rules to use that new rule instead of old one.\n\nChange-Id: Ic3ddfbfd0df59ffd10576f02730202c8b740ed26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34b93e77f155403128cbcaabf9f2b4369c684070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34b93e77f155403128cbcaabf9f2b4369c684070"}]},"branch":"refs/heads/master"},"e5201d07038f694cb13f18b3dcb56bcac719b2c5":{"kind":"REWORK","_number":2,"created":"2021-03-06 16:47:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/778929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/778929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/778929/2"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:08:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-06 16:46:50.000000000","tz":60},"subject":"Add new policy rule SG_OWNER","message":"Add new policy rule SG_OWNER\n\nDuring the migration to new, secure-rbac policies we already proposed\nrules NET_OWNER and PARENT_OWNER to extract \"ADMIN\" from the old,\ndeprecated rules.\nWe missed to do the same for SG_OWNER so this patch adds it and also\nswitch new SG rules to use that new rule instead of old one.\n\nChange-Id: Ic3ddfbfd0df59ffd10576f02730202c8b740ed26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5201d07038f694cb13f18b3dcb56bcac719b2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5201d07038f694cb13f18b3dcb56bcac719b2c5"}]},"branch":"refs/heads/master"},"a3fb6b5856fe235398ed5f3e9f4c0ca7993aac19":{"kind":"REWORK","_number":3,"created":"2021-03-06 19:52:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/778929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/778929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/778929/3"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:08:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-06 19:52:23.000000000","tz":0},"subject":"Add new policy rule SG_OWNER","message":"Add new policy rule SG_OWNER\n\nDuring the migration to new, secure-rbac policies we already proposed\nrules NET_OWNER and PARENT_OWNER to extract \"ADMIN\" from the old,\ndeprecated rules.\nWe missed to do the same for SG_OWNER so this patch adds it and also\nswitch new SG rules to use that new rule instead of old one.\n\nChange-Id: Ic3ddfbfd0df59ffd10576f02730202c8b740ed26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3fb6b5856fe235398ed5f3e9f4c0ca7993aac19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3fb6b5856fe235398ed5f3e9f4c0ca7993aac19"}]},"branch":"refs/heads/master"},"90edcf64419c708424c26f31a68e5c9c39bb7993":{"kind":"REWORK","_number":4,"created":"2021-03-08 15:04:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/778929/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/778929/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/778929/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/778929/4"}}},"commit":{"parents":[{"commit":"37f245c406768fafd9a048fa67b5ea68e7d77990","subject":"Merge \"Implement secure RBAC for RBAC policy API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37f245c406768fafd9a048fa67b5ea68e7d77990"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:08:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-08 15:03:25.000000000","tz":60},"subject":"Add new policy rule SG_OWNER","message":"Add new policy rule SG_OWNER\n\nDuring the migration to new, secure-rbac policies we already proposed\nrules NET_OWNER and PARENT_OWNER to extract \"ADMIN\" from the old,\ndeprecated rules.\nWe missed to do the same for SG_OWNER so this patch adds it and also\nswitch new SG rules to use that new rule instead of old one.\n\nChange-Id: Ic3ddfbfd0df59ffd10576f02730202c8b740ed26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90edcf64419c708424c26f31a68e5c9c39bb7993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90edcf64419c708424c26f31a68e5c9c39bb7993"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
