)]}'
{"id":"openstack%2Fneutron~779079","triplet_id":"openstack%2Fneutron~master~Ie0d30a444720c8a0fec261a153898b19aab4ee62","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ie0d30a444720c8a0fec261a153898b19aab4ee62","subject":"Implement new policy rules NET_OWNER and PARENT_OWNER","status":"MERGED","created":"2021-03-06 16:43:08.000000000","updated":"2021-03-11 03:33:22.000000000","submitted":"2021-03-08 10:12:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779079-secure-rbac","meta_rev_id":"37b66c5a4b9454839a56b1c23f266fa1a0998d95","_number":779079,"virtual_id_number":779079,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-11 03:33:22.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 10:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-08 08:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-08 07:33:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-08 08:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-06 17:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-06 18:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 07:33:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-08 08:17:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-11 03:33:22.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bac857e6b3c6a911f4e62601d9cbd4271322dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 16:43:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"694650515792f2b37144ee2c12c99275e70116d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 17:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abfe43e5cdbb454388a7ca8ef4fa36b0 : SUCCESS in 52m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bc1af8e93970ed93fa9bb3e5ba64e408997e185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 18:51:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/efceaf8ec698418da136912ac52591ea : SUCCESS in 50m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc566e7d52f04bdc8a0fa7efe1a24c9d : SUCCESS in 1h 27m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0bc4647b0b934a3c813406b62b722bc2 : SUCCESS in 1h 45m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a150ed79acc944acbe8b2df86746f7e0 : SUCCESS in 1h 49m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2efd388dfd694498843256ccd3c11095 : SUCCESS in 2h 05m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36bd1947a1d24481ac02e8e3e9df7051 : SUCCESS in 34m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee0b4c417d884a51adacff9b9c419f02 : SUCCESS in 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c227f4b3ed8e45f791341c9ed8de1d51 : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c2151e834c3493ab33068282aded452 : SUCCESS in 30m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5427f8a0ae444f038d1bcd24cfc992e5 : SUCCESS in 33m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca1794e7ac284fbdadd1135d63198da0 : SUCCESS in 28m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8155cf7094c04204b5a1d9d4eb131869 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60e58b969ecd40b3865bdffce6d7b653 : SUCCESS in 48m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af91f4084d084fe99483b9910b2dad94 : SUCCESS in 1h 15m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d2c052c360ad4b08882b32d822892778 : SUCCESS in 1h 24m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a9c74c2dd78491a99f210893cea8aba : SUCCESS in 53m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/844500a60abb4a3ab67108159f96d0c6 : SUCCESS in 1h 04m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c56dfa379b0642b69541699d635b99e1 : SUCCESS in 1h 23m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb2b73be9d114831ad8cec81d8443a3d : SUCCESS in 1h 29m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f526b2248e02413f9b82bb395b6d2bf9 : SUCCESS in 1h 48m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d5d3810451843d4bd33b53bcce743b5 : SUCCESS in 1h 06m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1e7af152a1f54d8f90d69e31534d31be : SUCCESS in 1h 08m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd07204140647ffa6f5a094389682b8 : SUCCESS in 38m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0b02a3d891440ed9828197b541135db : SUCCESS in 1h 26m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7bbfe875bc0e4816ae3edc5dbce27e5a : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"12a1710b4ba466b2e4501380499e33c07a1d3046","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-08 07:33:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b4815baa84525bacc476e19ca350c4ca21095d4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-08 08:17:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0c0c16a15ffc8d0c026818796a73725ea3cde7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 08:17:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93e4343994b502215f9f1a8c2d536f024fb7f97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 10:12:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7fd7fd1f6c2140098f11ff555a40927a : SUCCESS in 58m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/12ca3bc54b4d473f82d8c470a552570e : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/902a6382c2a64bcfb96efdde6257338b : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6393f01d0bce4f2dbeef83d99aca3f3c : SUCCESS in 1h 49m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/715688ec4a6d455aab5247763e4a1858 : SUCCESS in 1h 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0780776a29b48349d6c9262af0ade30 : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f01eb11f615401d881e6b9bc7c94c9e : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5788a2cf03ac4d218fc237854d77dd12 : SUCCESS in 33m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb5d01797f947abae6992374314343e : SUCCESS in 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0bf5202007b4328bafb30e100cdb090 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f612644fa96044348b4611e6285b7067 : SUCCESS in 45m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a475dc335b04ef2a8c6668cf4d1726d : SUCCESS in 1h 20m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/725103ecf7304ba3b39a0a2c5dfc47d4 : SUCCESS in 1h 19m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc987814d9414379ac8f043732f45830 : SUCCESS in 1h 01m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d027574f1c10464aa319174c931be977 : SUCCESS in 1h 01m 22s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78f92eaf85f744239bb3c5dee92e4be0 : SUCCESS in 1h 28m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d5a1de120304b25b150aef10e5870b4 : SUCCESS in 57m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c53bd59437148e8b2c1b9f38c1655ac : SUCCESS in 1h 07m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/58ee456ae44547d0bd5e19fe7df8e241 : SUCCESS in 32m 15s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/25ffc93280fb4807a7f904209e33e97d : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":1},{"id":"697fa25be3c185ebbe47a66b1734c1e0022a14c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 10:12:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"306075d85667456c901ef217aa3624d8382d741c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 10:15:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ca6ba8a639422f89727b389de65372 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1},{"id":"37b66c5a4b9454839a56b1c23f266fa1a0998d95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 03:33:22.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/79/779079/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"067241b11b0e48513d326fa970acd2742bbb0bdb","revisions":{"067241b11b0e48513d326fa970acd2742bbb0bdb":{"kind":"REWORK","_number":1,"created":"2021-03-06 16:43:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/779079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/779079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/779079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/779079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/779079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/779079/1"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-06 16:40:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-06 16:40:39.000000000","tz":60},"subject":"Implement new policy rules NET_OWNER and PARENT_OWNER","message":"Implement new policy rules NET_OWNER and PARENT_OWNER\n\nIn commit [1] new policy rules NET_OWNER and PARENT_OWNER were\nintroduced but I forgot define what those rules actually means.\nNow definition of those rules is actually added.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/778391\n\nChange-Id: Ie0d30a444720c8a0fec261a153898b19aab4ee62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/067241b11b0e48513d326fa970acd2742bbb0bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/067241b11b0e48513d326fa970acd2742bbb0bdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
