)]}'
{"id":"openstack%2Fneutron~779149","triplet_id":"openstack%2Fneutron~master~I276319e7efcbe18aa9bdf1ace2572b786e6321bd","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I276319e7efcbe18aa9bdf1ace2572b786e6321bd","subject":"Add constant for bagpipe flooding table","status":"MERGED","created":"2021-03-08 13:22:59.000000000","updated":"2021-03-13 12:53:44.000000000","submitted":"2021-03-13 12:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779149","meta_rev_id":"3fe92cae478a4f55dc1966950c7639dcaf0523fa","_number":779149,"virtual_id_number":779149,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-11 10:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 12:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-09 10:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-09 11:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-09 10:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-09 11:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 14:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 15:36:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 16:22:55.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 10:09:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-09 10:10:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-09 10:10:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-11 10:19:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06d2fd0ebb9184cf110c82b51b71ceb2a4dc562d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:22:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c3edf9d733431c4adc01438d22d135ab3e31752","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:25:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ec2eec70b25b9e882c01120e92fcc8eb140e7913","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:25:12.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0fc6a54860e59cf8a8be4f5d121cf89f2bee146b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 14:18:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6ee7f95f1e04fd4aa7588fb6dc41407 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7f43eeb089e4afc99b143f20e5d1f9f : SUCCESS in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80c97fa79ac7b784d9b00e66c60b54b65d7f4a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:36:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dbe5144ef0474d4bb127006b5ee74125 : SUCCESS in 55m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c3d26de433a7476e9ed2b78917354e37 : SUCCESS in 1h 39m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea4e7f89911d42f7b6862b0a0f5f36b6 : SUCCESS in 1h 38m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e667c9accf0a454ca058befe638aebac : SUCCESS in 36m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21fe4791e1264e128bd41020e5feb5fa : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c05bb0193e43a6af8fc288371a9071 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf5a87ee5dd4c2cbe23caf156f0b685 : SUCCESS in 38m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/118230d533df4bcca4c70d7a6e9da46a : SUCCESS in 34m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f1635a8fe4c40d8bb19ec61f1524946 : SUCCESS in 36m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0a2268b4e64229a103ce1652b16c31 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6769b4a37c504991a249b88d8c666322 : SUCCESS in 54m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80ed8d2ebd1342819af5452c4cf477b9 : SUCCESS in 1h 26m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dd9373af5c4f45f1b9eaac93db5f3e17 : SUCCESS in 1h 51m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdcec65a14b844bbb86fdb32144e9ae1 : SUCCESS in 1h 15m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eef3635459c441f18ffd50ee495e955c : SUCCESS in 58m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7e1537e6f7d4be39c1427beb04736a6 : SUCCESS in 1h 13m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6bab7a336d5942e9b4bce82ffd6cbbd1 : SUCCESS in 1h 19m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/234feedc7ea24da98071f698057cbd9f : SUCCESS in 1h 22m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7791939f89f748259ee9dd6d5162c659 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2cfbd3bf336185b0d9b62acde42169a67dc85b03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-08 16:22:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"dfad981e94359a8a191a2ec3f2a4624e63ebcb46","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 10:09:46.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"16b70d229913fa4a934c8b510d89fbbb27dbaf95","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 10:40:05.000000000","message":"Patch Set 2: Code-Review+2\n\nIt\u0027s ok to have this constant here to make it visible and avoid clashes with other projects or features.\n\nBTW, in that case we can consider to move all of them to n-lib, but this is another discussion.","accounts_in_message":[],"_revision_number":2},{"id":"84ee97c7f6065124c5a0f7e225a03b296e34c974","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 11:29:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e Patch Set 2: Code-Review+2\n\u003e \n\u003e It\u0027s ok to have this constant here to make it visible and avoid clashes with other projects or features.\n\u003e \n\u003e BTW, in that case we can consider to move all of them to n-lib, but this is another discussion.\n\n+1 for moving all of that to the neutron-lib in next cycle :)","accounts_in_message":[],"_revision_number":2},{"id":"80a95cada63fffb776b48f26aa9bddafe5c3ec93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 11:29:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c672692635cc5695031d1043d5d78e2275d2ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 13:23:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2d3a89f79f1d4630ae00392ece455417 : SUCCESS in 49m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11eadb4c2a0a452ebe85d4db92db29b4 : SUCCESS in 1h 47m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdb2f53e765043489ba635c0134d1cb7 : SUCCESS in 1h 17m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38a64882462a45549ba9044f09b198d9 : SUCCESS in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68790b7cd824b74982c36c0926bf170 : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0830eb65f7a40019b9e147fdf487f4e : SUCCESS in 37m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/951e2a6aa96341279738f753febfa7f3 : SUCCESS in 35m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b32a1779d1b4e9bb7f229ce075a8bf6 : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9ad0d600e52417cb47a0b2e58335b3f : FAILURE in 47m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/967cd41fb9b84f7f8a289364559429aa : SUCCESS in 1h 24m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/acb10b8580e541ab90fa1eeb087f3fcb : SUCCESS in 1h 19m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32a6803e6f354d20b6a360699c98b989 : SUCCESS in 59m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd12d101771441a48ea31de2b347d896 : SUCCESS in 58m 57s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5d1dc41c2464b4788cedd46a946d5b9 : SUCCESS in 1h 35m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e107eeb36f2448a6b90c37fc0bf232c9 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ca1a8003b4a82bd8b83e1221388db02181c69463","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-09 14:52:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"72f97c235fbd285d03f32b047d5bce45b51a6e33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 15:49:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a369f5c819434203b22f285759a5d2b3 : SUCCESS in 50m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f685f3b972f4c659f0ab5d4d52a292e : SUCCESS in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f74dfbfd6bcfdff74d3f098e17617223494a839b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 16:39:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a947593ece677530238ad7b8c32eaa108fcd4a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:10:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/858fdd533f624619be4ca77cf34e4619 : SUCCESS in 1h 04m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/afb0be84588e423fb66575c550258df2 : SUCCESS in 1h 44m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae8dd3846adb4e04b940a7be6484a5c2 : SUCCESS in 1h 44m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdb6f547156246fa88d17128c339710d : SUCCESS in 48m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83631869e238430da2cc7902b4441ea6 : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343a2b391e494a02a1b0a4f64c316c0e : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e41f558807a4d67aedeed10f889a6e3 : SUCCESS in 38m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9bac2ee3579426593ac477c38efd75a : SUCCESS in 41m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9b016e2b9ac4410941a17bfb74c0ada : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dc657eb0e84eaf9eff691c85e35300 : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c3f2941ef34441ab1b0a30d7750d32c : SUCCESS in 1h 01m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0dbe182f50e4eeeb982da54afb62dff : TIMED_OUT in 2h 15m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0ad5738234f842c8a167bb351949025e : SUCCESS in 1h 50m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e55851a7b02460cad5b9e9d6b0f2524 : SUCCESS in 1h 03m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de4f15fd6fed455883ba0ddea00d5292 : SUCCESS in 1h 00m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b40cea5c16c442db9abe9baccd9269b6 : SUCCESS in 1h 25m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/127c8fbfbf1a4a1b8f69db13aa4acec3 : SUCCESS in 1h 32m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fb3c0976cf84c799533c94e5251b100 : SUCCESS in 2h 02m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3725ba8769c4b4680807d9754e862b5 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9fffe2e373828855f3425abbe0afc300cba9faa0","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-10 09:11:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68aeb77ef9e0c4cb45fb501433c0f7cb3b00fe26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 09:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb62a57c610b497984382a388cd14cae : SUCCESS in 46m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/226fea933c6b47578b6013ac4761ee36 : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"301356037e681b55dbc9d0221f8f347ba3140c2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-10 10:25:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"13cce50d5554d089838a0158eb6de50af6bef560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:22:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f88dbff053424e699ed96e13474242e2 : SUCCESS in 51m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11ad767b082443da8070dc5d15d5ab67 : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83cadda394134cfd86979eb6395b3122 : SUCCESS in 1h 36m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58543d00159044669a21110ab9471736 : SUCCESS in 45m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14bc31984087439fae5a6d098ab410cc : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba68c3310a1d49b39dc86768e729cb56 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46f6268a454417f866395ca54b7a6fe : SUCCESS in 36m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16b311f0023e45099a62f200238c985a : SUCCESS in 33m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a77c1dad55cd49a9a383c02af190dbf0 : SUCCESS in 35m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94283aa7d5f48dfbd4981a603738f01 : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b79a911306404f8a88009b2df71c35ee : SUCCESS in 48m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dda34085d96145e8b6a6f70d70c64555 : SUCCESS in 1h 27m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b67f0c18e9ef48e3b86d6340b2002df9 : SUCCESS in 2h 04m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c490a7b69fe4c15a04dde27b6a958d8 : SUCCESS in 1h 10m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61478535496c4bd184cf2445334b8171 : SUCCESS in 1h 18m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/16cacc933c8d44189925acd34df796f1 : SUCCESS in 1h 23m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a97bea6d7db2432b92e159f7fd38450b : FAILURE in 1h 26m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6225745c29644af4a6c34f850e9f2ce9 : SUCCESS in 1h 42m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84c7a243f5eb429fbe5133a8bcf94c2e : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":2},{"id":"21a28bbb6f91f2029701a6e0fb00cacc46a59f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:23:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9c58156245b32a8eb7ce419fdab966ecce5888a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:50:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e2e64713cab844ababb643e18ccddc28 : SUCCESS in 52m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/87f41fe1f2d44cf59ecb7706e1e6a492 : SUCCESS in 1h 29m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37227f27c766442faedcfd01a428ae2a : SUCCESS in 1h 23m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9aaec6876ff46239d447d8051396227 : SUCCESS in 35m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b623392ebb0417797fc3d4f7c533ced : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e835d2adc714b5ab824752be2ac9ff7 : SUCCESS in 53m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee701f13b688455087b75796bc34d471 : SUCCESS in 42m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b8c3ef5ae24f5bb6ea7c6581f0ab30 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccc391185c6f4f80b8c8ab2f70d46269 : FAILURE in 1h 00m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bee142607bd469d8abbad64538b1ac6 : TIMED_OUT in 2h 18m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4fe3bb7851224103a0a1076582ca71a5 : SUCCESS in 1h 18m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/389b00a05be94c82b7850b522d40278e : SUCCESS in 1h 07m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a8f4ea97e64a40ffa5036e31ee64b319 : FAILURE in 46m 04s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d52826d0749f40f9ab5444ae17eba5a8 : SUCCESS in 1h 39m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93e4d6ea9b004b4d866d69e80c267013 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ef1b9773b0753262213d77c655faa79f870a26ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 10:19:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/49/779149/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c478a2cc1dad9605effb10b59dcd1f74037934ac","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-11 16:24:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"97d8afaf8b045810b24a63eeeeb5425681d6dbff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:27:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3632725d898f4e03843c1694cf65757a : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7bf3226cadd04217999f76bf6a642c2e : SUCCESS in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"222e700d4415d6c8a06cd0f81b479a504c7aa9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:18:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f2bf23edee840b6a2700c81c31d1ec1 : SUCCESS in 50m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ded1822a48e450a8f34f131898fe370 : SUCCESS in 1h 52m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2fe95f851b14d70b95ad7ae85713c90 : SUCCESS in 1h 33m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a5aefe674734228b48eb862d0f0306e : SUCCESS in 41m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b52226008e94c0ab189b48ff25cc8a7 : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443aaba38d4443dbba7edfdc0ced4b02 : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/494fdf659f3c47c5a620cb6e2d909055 : SUCCESS in 36m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/187cd979a4b24346b878bb0e7fcca38f : SUCCESS in 44m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a10182009c460da6c88750bff06bff : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef0a6c9b24040b9b0bfa7824cac4848 : SUCCESS in 11m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/415817528dba466a9ef948e0b2b7e8b2 : FAILURE in 45m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8386a2eab1cf47ae9b563336d506966d : SUCCESS in 1h 23m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/819e65f72db8482689595b6741b6635c : SUCCESS in 1h 31m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d57462d621be4112a48a6d73f8aef7e7 : SUCCESS in 1h 03m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/049cff2fc58d4e61b1019956f3773b32 : SUCCESS in 1h 14m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f32273f8f49f471b98229b64df3f1971 : SUCCESS in 1h 27m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f819983e90d40ee84b31d63a4c4698b : SUCCESS in 1h 27m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2453a35d944a4817a15fa9ee423ad6bd : SUCCESS in 2h 00m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edcf1a59d1f847dd91b31e08061c83e3 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"773b38ea4674c0185f17af56ddd3e8f8a8eae040","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-11 21:28:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e0eec40e836c6daea5aad14589b7292367f024bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-13 08:57:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad5447b0cc4b723cf2ae8634fb4bb60b8499b167","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:45:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/826481f134cb4cb3acc77c78e0d313c9 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"888bf824fa22dc56ff16016bf6671acba5c57fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 10:50:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3217f37fff59465f968875248593ec1e : SUCCESS in 56m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a2ee8945011421795e4f240b20aa1a0 : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aa1ce9541b694c9098aa26dd2a0d61a3 : SUCCESS in 1h 26m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf9d38aac5464d58908974e3380ae011 : SUCCESS in 24m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2edad9ed8d4f4c658aa0c0cca633bb13 : SUCCESS in 36m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e96aeae50bb4448816bad1be93ff1a9 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd28b0bee6b341a78e97636a257db5e9 : SUCCESS in 23m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2645b521c6346ebac246fbf59286034 : SUCCESS in 23m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/776cbceb27b94ced85f4c2341638fe72 : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1364ffca924703b4229ad18d9edf7c : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b61840bb81f64b29812b91fdf24f6c71 : SUCCESS in 50m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b03802c11f544b3aae44a11e794b336 : SUCCESS in 1h 24m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7f29aa3a789644b193c3524b5a763a1e : SUCCESS in 1h 42m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a01baa49b3ba4eb5a66188b18c05fbf3 : SUCCESS in 53m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b3e427ab1894d9c9b28c288f01139f1 : SUCCESS in 1h 05m 48s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f0a92e1c04f4fdcb8398724901af69f : SUCCESS in 1h 37m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a7bb99830d94796b7cf76efafe8462e : SUCCESS in 1h 35m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cc7a7d6f20845fda0a3deaf619fc6cf : SUCCESS in 1h 43m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15f5dd2e110c4481811ad62c8634eaf9 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4b0b374637b797d5c72e878408a9db5c93be1af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 10:50:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87e9b0ca5626ad72f9f43808cfe35bc59a6a8e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 12:49:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eddf624b34054b829eb2b40de22825b5 : SUCCESS in 44m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0ccc2a9e9dc4460e91ab45b162dee7ea : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2051a57845854a8da498a9a49cfb8ad7 : SUCCESS in 1h 19m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa1f7756a174400fafc342da9a35c9aa : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bb131bd2ae4f699cdc534020a6d303 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f212b409944e4c54b8291f7cefaa2c20 : SUCCESS in 29m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4562c9b86744d38832c1c768401a092 : SUCCESS in 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f767300d8c4838824688355a466f82 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bb027069cf44e1cb2829a256ceb4ce7 : SUCCESS in 45m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e357940e5fde4d4d8fb6217ff254c9b0 : SUCCESS in 1h 20m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/066a75b1532c449f99b6b03dda353999 : SUCCESS in 1h 15m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8da33b2e065a4686a03b981108431da0 : SUCCESS in 1h 10m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82c2e7c0c69945188d452bcf371a3b5c : SUCCESS in 1h 01m 59s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41ace19faff94edaa89b47824a59e73d : SUCCESS in 1h 33m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e422ed23462448848999371ceb96d36c : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ed0560287ec22c33fc942c5e4b3dad5433e0ccf0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 12:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b5c31babc07f0a21bd5b7a7fff60989c5b265f90","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-13 12:50:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/49/779149/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3fe92cae478a4f55dc1966950c7639dcaf0523fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 12:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60700dc6856d4416891d4efdb46c5a9d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02c8c037f82f810074c6f552eb697054563b6731","revisions":{"1d8ff60d262b8ec2021a252d9a753adb1b46cea7":{"kind":"REWORK","_number":1,"created":"2021-03-08 13:22:59.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/49/779149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/779149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/779149/1"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:22:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:22:59.000000000","tz":0},"subject":"Add constant for bagpipe flooding table","message":"Add constant for bagpipe flooding table\n\nThe patch adds constant for bagpipe table used to flood traffic\ninto EVPN tunnel and standard tunnels to gateway nodes.\n\nChange-Id: I276319e7efcbe18aa9bdf1ace2572b786e6321bd\nRelated-Bug: #1915522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8ff60d262b8ec2021a252d9a753adb1b46cea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8ff60d262b8ec2021a252d9a753adb1b46cea7"}]},"branch":"refs/heads/master"},"02c8c037f82f810074c6f552eb697054563b6731":{"kind":"REWORK","_number":2,"created":"2021-03-08 13:25:02.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/49/779149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/779149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/779149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/779149/2"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:22:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:24:58.000000000","tz":0},"subject":"Add constant for bagpipe flooding table","message":"Add constant for bagpipe flooding table\n\nThe patch adds constant for bagpipe table used to flood traffic\ninto EVPN tunnel and standard tunnels to gateway nodes.\n\nChange-Id: I276319e7efcbe18aa9bdf1ace2572b786e6321bd\nRelated-Bug: #1915522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c8c037f82f810074c6f552eb697054563b6731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c8c037f82f810074c6f552eb697054563b6731"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
