)]}'
{"id":"openstack%2Fneutron~779258","triplet_id":"openstack%2Fneutron~master~I99a60b9af94b8208b5818b035e189822981bb269","project":"openstack/neutron","branch":"master","topic":"bug/1918108-mcast_flood_reports","hashtags":[],"change_id":"I99a60b9af94b8208b5818b035e189822981bb269","subject":"[OVN] Set mcast_flood_reports on LSPs","status":"MERGED","created":"2021-03-08 13:38:47.000000000","updated":"2021-03-12 03:56:14.000000000","submitted":"2021-03-10 10:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779258-bug/1918108-mcast_flood_reports","meta_rev_id":"d882b540bcd83fc062eb87a2ae43b4f4ac9979ee","_number":779258,"virtual_id_number":779258,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-12 03:56:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 10:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-03-10 08:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2021-03-09 21:24:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},{"value":2,"date":"2021-03-09 10:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-03-10 08:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 13:43:31.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-03-08 14:09:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-08 14:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 15:00:24.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},"state":"REVIEWER"},{"updated":"2021-03-08 18:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 10:36:54.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-03-10 08:45:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-11 11:05:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dd21b465b7362e8c15809836b5aaf36d8ba0f63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 13:38:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77bac76ccce0caeba56f4380e636bff7bfe4d94a","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-03-08 13:43:31.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me. Thanks Lucas 😊\nMaybe worth a release note?","accounts_in_message":[],"_revision_number":1},{"id":"b66db67c9258107b29751db39f338855c48f124e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 13:53:46.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e Looks good to me. Thanks Lucas 😊\n\u003e Maybe worth a release note?\n\nGood point, I will add a reno for this","accounts_in_message":[],"_revision_number":1},{"id":"9221ac1f2b264661d58913cc8b7b153b6141da18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 13:57:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8fd8791bbbdbea236363ccdf7c6c1e9ba162160","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 14:21:15.000000000","message":"Patch Set 2: Workflow-1\n\nWaiting on the bug reporter feedback","accounts_in_message":[],"_revision_number":2},{"id":"d9686aeefbee8e3d5cc44a10b25046912a4fb3b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 14:52:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17bfba3110243fb9dc85bf49c360901 : SUCCESS in 45m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b124de20646647dbb3c8c040e57e4c90 : SUCCESS in 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9eef182f657ae23598bf09da190a0b8fe57da9ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 15:00:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1504e1de76d2a602757260ab24a9daf701aff4ba","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 15:00:39.000000000","message":"Patch Set 3: Workflow-1\n\nWaiting on Pedro\u0027s initial review","accounts_in_message":[],"_revision_number":3},{"id":"c6395dd7a6a99c68de30f799fddd9f52cfe6412f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-08 15:01:48.000000000","message":"Patch Set 3: -Workflow\n\nMaybe we do not need the workflow -1 since we want to attract as many reviews as possible","accounts_in_message":[],"_revision_number":3},{"id":"903cc41cf7f0017063dfc87edbabe3dc75eaf1ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 16:05:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c35dd48c9fb64b4a80f4711ab0a689da : SUCCESS in 59m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aba5c4f3f7724a5e8b7ef8a34530cb84 : SUCCESS in 54m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1cb20d4b530eabf634f1fc2521e5b14e4836a0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 18:22:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ab575f079cfa4520bccb96234ad313b3 : SUCCESS in 56m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7388c1eaa2741dfa17029276cbf87c7 : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ee0ade450b0f499693bc081ee290b169 : FAILURE in 1h 34m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff1045e85b784312875aff091a31cdc9 : SUCCESS in 1h 05m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d05562e885394ac6aca38c4db44a2c16 : SUCCESS in 2h 06m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/836f93bfa06f4fdca93fe9c520f4dcd5 : SUCCESS in 38m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ad1971d1af341f3b74b409cad1a00a3 : SUCCESS in 38m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c64b9293934435940c5548361f2d06 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20532527d3d44a60a6444dc8ff5189e1 : SUCCESS in 30m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4bda3f1ff39466184b2002aa761d024 : SUCCESS in 29m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b9ffbf01a5b4293bbd0a141c1f073bd : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a75d0ff77914ec8854c92697faee65f : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b553a85c3e5345b7a3b11573af40fbc3 : SUCCESS in 11m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2bfc17eea9943198b828348c78ec4bc : FAILURE in 46m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7530d55701bc41ab95d46b85e5744788 : SUCCESS in 1h 25m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/65ab5fd2daa3432096f82121be14e165 : SUCCESS in 1h 12m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb7976a5a68f41b7b0a9fc0d64120c26 : SUCCESS in 51m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac2e0797ea0347b7ac7416942c2bf0d0 : SUCCESS in 1h 16m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a5c83d2af64345c08e7893b04323ce45 : SUCCESS in 1h 23m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/54e7a4285cfc41638cd5870d8816b64a : SUCCESS in 1h 24m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71efac52cfc14eaab03d48be46ae06f2 : SUCCESS in 1h 36m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87f5df9ab08946efac25535bb65dd9c9 : SUCCESS in 1h 03m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/629d9752e4be436cbe059f8be8f9b7a3 : SUCCESS in 1h 09m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/031e7b6630e34456824ea9c17989abc5 : SUCCESS in 34m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d667a2039e5244089074f25dca6e175f : SUCCESS in 1h 39m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ca113c46d254f398ead78eb71d7f8cf : SUCCESS in 1h 29m 11s","accounts_in_message":[],"_revision_number":3},{"id":"40b166d2204561e9f1eb2ce1bd8a2f9766b94d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-09 09:15:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6c0aa3fb812c7705dc01ca126fd218fc09477d4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:12:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91185c2876b747fba33015c9e2c05bfb : SUCCESS in 49m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5186c78d9cd4418ea94fc578f38f8f14 : SUCCESS in 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"049e05d285fc3ead0a872fa5fd92bbc0c23c1391","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-03-09 10:36:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nGood job! I thought we\u0027d need an RFE originally to fix this and I thought about a much more complicated solution. This is perfect.","accounts_in_message":[],"_revision_number":4},{"id":"3b0b39d4af95a8bd706a209d5d6f3d2a1ce2ef85","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-09 10:47:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d277115242ced2991e98e8ce6dbff03d57954527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:32:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c6caaa08fc54bc6b6116d0ac790798d : SUCCESS in 46m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d862c731633e4194839d9438c1b0d75e : SUCCESS in 1h 27m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/910f21d4c1504094875188ea265cdf62 : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e96a16f2b8dd47e4b422d754e799bef4 : SUCCESS in 2h 14m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/76c667706d714338866949a731f7e965 : SUCCESS in 1h 44m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9cec9347d2549d28a288ddd67236407 : SUCCESS in 54m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e3d0a50fd96457da5535a6b37d1385b : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb69746aa2046428e071781fde0c36c : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbfcebf6613d4b28955521b3faee134d : SUCCESS in 35m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb9df24f7474776bd8145f7671b659c : SUCCESS in 34m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/164b4cad36e542a4b7548bc4b25b7bb4 : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c358b400ff49d791d98fc1115ffb74 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d27b7f12e1ee4b4cacb93267c8ff7a23 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a11f20baac29438db46e9ce761f26b90 : SUCCESS in 52m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1869b64b3d2e474199fde4ef8fc97edd : SUCCESS in 1h 27m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/38f85d48d72c4d0fbef579dd7b8cf685 : SUCCESS in 2h 13m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a65497950e8448c82b97716fdf46a05 : SUCCESS in 54m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c46fd018fcf8481faf83be5047e011bf : SUCCESS in 1h 05m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b82744bc99346ab8a9a4dce48a378c6 : SUCCESS in 1h 30m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/784ec0a3aa4b4af3a030dcf2e1437eb9 : FAILURE in 1h 24m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc0dbf43f4124bfda83b5bacb214aa02 : SUCCESS in 1h 41m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e088dbde5faf4df3954e76cc60165e65 : SUCCESS in 1h 13m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5274452239034be1ad174e69ca42639a : SUCCESS in 1h 21m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/21f417a57bf641389faa366c70a96c93 : SUCCESS in 39m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a789fdd8f50f4f90831f794aa838ddbb : SUCCESS in 2h 09m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4ca64e0c866446a3ac5fdadb926cf757 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":4},{"id":"2775ea0ecaea71876c4ca905199838963d567071","author":{"_account_id":28510,"name":"Pedro Guimarães","email":"pedro.guimaraes@canonical.com","username":"pguimaraes"},"date":"2021-03-09 21:24:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b9d5df1afcb470042b5b839697defb2a03a1f2db","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 08:45:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"731e523e7d49c4370338022b1cf5f22ca272bce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 08:46:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4cd29cfa110aaa66692099573ff8af13a2b94663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:55:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a587fff5e36d4ec3a8c908e60ab545eb : SUCCESS in 45m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/44def882a0f04d59b2c66be30e237ae5 : SUCCESS in 1h 44m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ef0b0333769a4f468a20c193cf784823 : SUCCESS in 2h 03m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/26c1366f2f074bc38025d7af564116a6 : SUCCESS in 1h 40m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1044cdf7c19a4577a375681180d3cdde : SUCCESS in 1h 45m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70de3236433c4900a23eef96dda779fb : SUCCESS in 41m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/616305eeb5064f5ba51e6082c9dfe8aa : SUCCESS in 12m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54ce45ca76ca49ccb8951d7bfca53d37 : SUCCESS in 37m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e233253bda74a2fbb2a77f2bd6adc9b : SUCCESS in 35m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664cc61958bb4e6cbf173413139c75cc : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c680f3b00f57431695c97d389914de2b : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20cd31ee3706402a915a52c6f835b728 : SUCCESS in 47m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd5f76ffb03f46de936f3cafeaef9804 : SUCCESS in 1h 34m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c52cc7a0d8843a68d3d0914f154b44d : SUCCESS in 1h 24m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2791e2742e743f183fe6cfc3070d78c : SUCCESS in 1h 05m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/13819bcef0d847d5a0b839886bded773 : SUCCESS in 1h 10m 24s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/052bfdbc63e24b71af075fbc9f37160c : SUCCESS in 1h 43m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54019d8545d94143b0ea8a11209ebca0 : SUCCESS in 1h 10m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ccb64d29e5bc412481fb94e71fb5efd5 : SUCCESS in 1h 06m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4af62a6103de4807bdf5bef9bd674419 : SUCCESS in 29m 59s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/edcb0989593443b8a1fdfa7e871a79e6 : SUCCESS in 1h 23m 25s","accounts_in_message":[],"_revision_number":4},{"id":"0536b64a8662f179f3d6dbeeee0f66e880586a8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:55:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d28b5e4fa5da7c1f2cbd3d0e0d5e5c0117665341","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:57:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c619478e1e942cf82fdaed00a33cd62 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/195cc9cc617142eaafad1a028e3f70ce : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4},{"id":"18396c70b83a51605b99776d45930f1799d0651b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 11:05:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/58/779258/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4bed43f2c33abda014ac3c1d67956ec55dbd5d84","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 11:46:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/58/779258/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a7518e75c98f65f7268f7bc482f2e93ba1e417d3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 13:05:42.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/58/779258/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"d882b540bcd83fc062eb87a2ae43b4f4ac9979ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 03:56:14.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/58/779258/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b04d64b90f2822222f96ddd726d379a2f19b539e","revisions":{"f937311f70a785024459facc35b16aabde5d2091":{"kind":"REWORK","_number":1,"created":"2021-03-08 13:38:47.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/58/779258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/779258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/779258/1"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 11:44:24.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 13:24:44.000000000","tz":0},"subject":"[OVN] Set mcast_flood_reports on LSPs","message":"[OVN] Set mcast_flood_reports on LSPs\n\nThis patch enables the \"mcast_flood_reports\" option in the Logical Switch\nPorts in the OVN driver. Without this option, the ovn-controller will\nconsume the IGMP queries and won\u0027t send it to the LSP ports, meaning\nthat external IGMP queries will never arrive to the VMs.\n\nIn talks to the core OVN team, it was suggested [0] to enable the\n\"mcast_flood_reports\" option by default in the OVN driver (at least until\nfixed in core OVN) as a workaround to this problem. And, to avoid having\nto update all ports (which can be many) based on the igmp_snooping_enable\nconfiguration option, we are always setting \"mcast_flood_reports\" to\n\"true\" in the LSPs. This won\u0027t cause any harm (also confirmed by core\nOVN developers [0]) since it will be ignored if multicast snoop is\ndisabled.\n\n[0] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1933990#c3\n\nCloses-Bug: #1918108\nChange-Id: I99a60b9af94b8208b5818b035e189822981bb269\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f937311f70a785024459facc35b16aabde5d2091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f937311f70a785024459facc35b16aabde5d2091"}]},"branch":"refs/heads/master"},"079bbc9f65a14ee3b52fd922798a80609424119c":{"kind":"REWORK","_number":2,"created":"2021-03-08 13:57:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/58/779258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/779258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/779258/2"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 11:44:24.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 13:57:39.000000000","tz":0},"subject":"[OVN] Set mcast_flood_reports on LSPs","message":"[OVN] Set mcast_flood_reports on LSPs\n\nThis patch enables the \"mcast_flood_reports\" option in the Logical Switch\nPorts in the OVN driver. Without this option, the ovn-controller will\nconsume the IGMP queries and won\u0027t send it to the LSP ports, meaning\nthat external IGMP queries will never arrive to the VMs.\n\nIn talks to the core OVN team, it was suggested [0] to enable the\n\"mcast_flood_reports\" option by default in the OVN driver (at least until\nfixed in core OVN) as a workaround to this problem. And, to avoid having\nto update all ports (which can be many) based on the igmp_snooping_enable\nconfiguration option, we are always setting \"mcast_flood_reports\" to\n\"true\" in the LSPs. This won\u0027t cause any harm (also confirmed by core\nOVN developers [0]) since it will be ignored if multicast snoop is\ndisabled.\n\n[0] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1933990#c3\n\nCloses-Bug: #1918108\nChange-Id: I99a60b9af94b8208b5818b035e189822981bb269\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/079bbc9f65a14ee3b52fd922798a80609424119c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/079bbc9f65a14ee3b52fd922798a80609424119c"}]},"branch":"refs/heads/master"},"18d391539d2a6cd6ad8a2778a2ea120042285b4b":{"kind":"REWORK","_number":3,"created":"2021-03-08 15:00:02.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/58/779258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/779258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/779258/3"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 11:44:24.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 14:59:53.000000000","tz":0},"subject":"[OVN] Set mcast_flood_reports on LSPs","message":"[OVN] Set mcast_flood_reports on LSPs\n\nThis patch enables the \"mcast_flood_reports\" and \"mcast_flood\" (on provnet\nports only) options in the Logical Switch Ports in the OVN driver. Without\nthese options, the ovn-controller will consume the IGMP queries and won\u0027t\nsend it to the LSP ports, meaning that external IGMP queries will never\narrive to the VMs.\n\nIn talks to the core OVN team, it was suggested [0] to enable the\n\"mcast_flood_reports\" option by default in the OVN driver (at least until\nfixed in core OVN) as a workaround to this problem. And, to avoid having\nto update all ports (which can be many) based on the igmp_snooping_enable\nconfiguration option, we are always setting \"mcast_flood_reports\" to\n\"true\" in the LSPs. This won\u0027t cause any harm (also confirmed by core\nOVN developers [0]) since it will be ignored if multicast snoop is\ndisabled.\n\n[0] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1933990#c3\n\nCloses-Bug: #1918108\nChange-Id: I99a60b9af94b8208b5818b035e189822981bb269\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18d391539d2a6cd6ad8a2778a2ea120042285b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18d391539d2a6cd6ad8a2778a2ea120042285b4b"}]},"branch":"refs/heads/master"},"b04d64b90f2822222f96ddd726d379a2f19b539e":{"kind":"REWORK","_number":4,"created":"2021-03-09 09:15:03.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/58/779258/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/779258/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/779258/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/779258/4"}}},"commit":{"parents":[{"commit":"b7c4a11158786431c262cfcc2fc4bc46ab6bacd2","subject":"Merge \"Implement new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c4a11158786431c262cfcc2fc4bc46ab6bacd2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 11:44:24.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-09 09:14:46.000000000","tz":0},"subject":"[OVN] Set mcast_flood_reports on LSPs","message":"[OVN] Set mcast_flood_reports on LSPs\n\nThis patch enables the \"mcast_flood_reports\" and \"mcast_flood\" (on provnet\nports only) options in the Logical Switch Ports in the OVN driver. Without\nthese options, the ovn-controller will consume the IGMP queries and won\u0027t\nsend it to the LSP ports, meaning that external IGMP queries will never\narrive to the VMs.\n\nIn talks to the core OVN team, it was suggested [0] to enable the\n\"mcast_flood_reports\" option by default in the OVN driver (at least until\nfixed in core OVN) as a workaround to this problem. And, to avoid having\nto update all ports (which can be many) based on the igmp_snooping_enable\nconfiguration option, we are always setting \"mcast_flood_reports\" to\n\"true\" in the LSPs. This won\u0027t cause any harm (also confirmed by core\nOVN developers [0]) since it will be ignored if multicast snoop is\ndisabled.\n\n[0] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1933990#c3\n\nCloses-Bug: #1918108\nChange-Id: I99a60b9af94b8208b5818b035e189822981bb269\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b04d64b90f2822222f96ddd726d379a2f19b539e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b04d64b90f2822222f96ddd726d379a2f19b539e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
