)]}'
{"id":"openstack%2Fneutron~779448","triplet_id":"openstack%2Fneutron~stable%2Frocky~I688a6e6d0ac42c00d87571484f726e0eae091675","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1916022","hashtags":[],"change_id":"I688a6e6d0ac42c00d87571484f726e0eae091675","subject":"[WIP] Add log file for neutron-keepalived-state-change","status":"ABANDONED","created":"2021-03-09 09:46:49.000000000","updated":"2021-03-11 15:14:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f6b994815abf3392494e5f54102a2a7dc7cdd3c","_number":779448,"virtual_id_number":779448,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-10 20:23:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 09:46:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-03-09 09:59:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 12:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d1146fbd6d4b3a0011eed0b36d367ae49fab2f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 09:46:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e88e4916f1875c00f33914d6412cd8e6fda4a630","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 09:59:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/779448/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/779448/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"212175c1983c6e4a835f3c92c3e75eee13870db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:06:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f78d598e05af431b9c1475cf5eb0aefe : SUCCESS in 1h 22m 21s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/ba7c5adfacd246abba6460814d135540 : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/74bf9daaf5cc4ef39e80770324dab2fd : SUCCESS in 1h 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76fe53f69b5418884dc8cf6e8a6a2e4 : SUCCESS in 12m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a8e569e59e14ee0aa4c0cd9ed51876d : SUCCESS in 41m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ace641300804fb18281d9ab5507028d : SUCCESS in 30m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0125c6801b624865bdfcd70ea01c68a6 : SUCCESS in 6m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e6f1c11ad2864599b0d4bce0d7c62050 : FAILURE in 1h 06m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ed7b705ff0ed4f9f95a4755f9010f21b : FAILURE in 46m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8b574a84001431ca41f11623c29afe9 : SUCCESS in 1h 26m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/503b86283ed241b6a4c09e00f8c40ee4 : SUCCESS in 14m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdcfb57e4182425a855364dfe79347b9 : SUCCESS in 47m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c90e1852002243808bb7f382e5c61ab4 : SUCCESS in 1h 24m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a9c12cab804f4ad1bfb8efafad9aa35b : SUCCESS in 1h 15m 30s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/84160b0c1e6c49108dd4956d1a3af56a : SUCCESS in 1h 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78be35f240474814b37852da6902c1bd : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"83817f513f528f6d9e590f3ee8fd5a3ff1553cf3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 14:27:02.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"5de736ef508611dad20dda59f089555dfceda2ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 15:37:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/779448/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/779448/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"69e7608edc0d949e5454a15c6f9968dae904773d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 14:35:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3deac8bbc0a7a5dc5a1d6a0237e088b781cbf3eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-10 14:39:21.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/779448/2/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/779448/2/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"f1969423ba1bb0519dd4c0fb994358fddbe1d62a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-10 17:00:48.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"84ee7e81ee8825eb98edeb23ba75980213c1045c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-10 17:05:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/779448/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/779448/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"771fab10e6ce53c1f702668fd4483e3662c9d310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 20:23:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/97516cacee2347b8b56777ea33eb2779 : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/5abfe863b43c4d1a8176eeb63717a501 : SUCCESS in 1h 38m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/db3b5cd00d2a4eb090f137b0f85cb9bd : FAILURE in 1h 04m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4376c14d4944e288f52a0ec5faf2bc4 : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a8cfb58e69c43b6bfe47e332cb26c18 : SUCCESS in 18m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91e1eb4e9c424ff28b4a7ba53b200e16 : SUCCESS in 22m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d8353282a9954011a5d17188b6119acd : SUCCESS in 6m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e4ab0fa4f11428e80d9975a155337ee : SUCCESS in 1h 29m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b76e8aa0fde44ce98715ab3e132ba412 : SUCCESS in 56m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e7c2ee5cc4646f7a157cf6ee3dfb640 : SUCCESS in 1h 28m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e510188ad32f4a6cbf3591f41b5c3019 : SUCCESS in 10m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f3d276f051b94c84899463d0c979394b : SUCCESS in 33m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/915660aa1f9048c18c248d601693c48d : SUCCESS in 1h 26m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ff8ef89226a4d72b69b0bd3c897febf : SUCCESS in 1h 54m 29s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26fa429791544104950d0833f4043f49 : SUCCESS in 58m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de7bd6b668d04e76b02bd970de275e21 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2f6b994815abf3392494e5f54102a2a7dc7cdd3c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-11 15:14:58.000000000","message":"Abandoned\n\nNot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6a036d57e5326000002ccac43daeac446b7f311","revisions":{"afb7b48e5377c01aa7503490340a506547fcd714":{"kind":"REWORK","_number":1,"created":"2021-03-09 09:46:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/48/779448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/779448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/779448/1"}}},"commit":{"parents":[{"commit":"e81aa71c5f647176269d9ac4ac287de0eecea373","subject":"Merge \"Add some wait time between stopping and starting again ovsdb monitor\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e81aa71c5f647176269d9ac4ac287de0eecea373"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-03-28 16:07:11.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 09:42:03.000000000","tz":0},"subject":"[WIP] Add log file for neutron-keepalived-state-change","message":"[WIP] Add log file for neutron-keepalived-state-change\n\nneutron-keepalived-state-change may not start but have no method\nto find out why. This patch adds the log file for it.\n\nChange-Id: I688a6e6d0ac42c00d87571484f726e0eae091675\nRelated-Bug: #1822155\n(cherry picked from commit ccf76c36bb508a8cfd4e0f3a24fc1de7a77dc307)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afb7b48e5377c01aa7503490340a506547fcd714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afb7b48e5377c01aa7503490340a506547fcd714"}]},"branch":"refs/heads/stable/rocky"},"f6a036d57e5326000002ccac43daeac446b7f311":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-10 14:35:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/48/779448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/779448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/779448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/779448/2"}}},"commit":{"parents":[{"commit":"6f884614be34375b751a57c5772ea115e047298e","subject":"[Functional tests] Set proper user name in the CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f884614be34375b751a57c5772ea115e047298e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-03-28 16:07:11.000000000","tz":480},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 14:35:04.000000000","tz":0},"subject":"[WIP] Add log file for neutron-keepalived-state-change","message":"[WIP] Add log file for neutron-keepalived-state-change\n\nneutron-keepalived-state-change may not start but have no method\nto find out why. This patch adds the log file for it.\n\nChange-Id: I688a6e6d0ac42c00d87571484f726e0eae091675\nRelated-Bug: #1822155\n(cherry picked from commit ccf76c36bb508a8cfd4e0f3a24fc1de7a77dc307)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6a036d57e5326000002ccac43daeac446b7f311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6a036d57e5326000002ccac43daeac446b7f311"}]},"branch":"refs/heads/stable/rocky","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
