)]}'
{"id":"openstack%2Fneutron~779463","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I12b5e672b01f64fa118fe50d3add09ac2ae419c5","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1916761-stable/victoria","hashtags":[],"change_id":"I12b5e672b01f64fa118fe50d3add09ac2ae419c5","subject":"Revert \"DVR: Remove control plane arp updates for DVR\"","status":"MERGED","created":"2021-03-09 13:07:17.000000000","updated":"2021-03-10 12:30:37.000000000","submitted":"2021-03-10 12:28:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779463-bug/1916761-stable/victoria","cherry_pick_of_change":777903,"cherry_pick_of_patch_set":2,"meta_rev_id":"a343757d6bc9a1eb359e3436d2595219869cb6df","_number":779463,"virtual_id_number":779463,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 12:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-03-10 09:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-10 09:11:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-10 02:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-10 08:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-03-10 09:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 13:07:17.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-03-09 16:54:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 02:19:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-10 08:50:01.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-10 09:11:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-10 09:32:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"93657cefb7ff08d5bbe7d3f71ce4250ac1a4ba72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2021-03-09 13:07:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d41cca8f1efb01c790e7c05d37ee375199e7ffdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 16:54:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/59a1411509e84569a7b4952f3d924b14 : SUCCESS in 44m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/ee78a191df07490e869f48445cd9b3a9 : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/6713fb9284cb442d9eee54a06d29c122 : SUCCESS in 1h 03m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/5de90587357c42398839a7020fefc562 : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/e46e642478da48b9ad7c4054fecb8322 : SUCCESS in 1h 19m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d4ac1164e4f4fa0916501efe0b10df7 : SUCCESS in 42m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/747fba6f437845c9a133fea6abf0116d : SUCCESS in 12m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4acbe628134462fa1791b0fdc1c96fb : SUCCESS in 25m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd703c7469db4e2bae497d49723d3119 : SUCCESS in 25m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f87395dcca4545832adb859a8a509d : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f374b45ce614fa2b83a7e0fc660d0c7 : SUCCESS in 1h 16m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/852120acd3874877bfae23fc741bdc16 : SUCCESS in 1h 03m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c2c220847f8c4990a0fef0b82c8434cd : SUCCESS in 1h 28m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b33cf0dea35422a9acd049cf38700ef : SUCCESS in 1h 02m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/817b7082429c44e09823b79571099b42 : SUCCESS in 1h 06m 48s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/03ae48b0747b48ce875795fceddfec14 : SUCCESS in 1h 05m 57s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4f593bbfd7a74a8caab3f5b254678587 : SUCCESS in 1h 16m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b27a3c66554442e991de6c536578c16 : SUCCESS in 1h 22m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e7576f71313c4d63a30e639cdbae968a : SUCCESS in 1h 34m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/606c33b983c746f59556a09c9b29d16c : SUCCESS in 1h 34m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d29a38f7003044aa875221fc573dcc00 : SUCCESS in 1h 09m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ac8dfa8c80734939bb46ee1a3e7227e4 : SUCCESS in 1h 12m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1686be6067c94350962432b4726bf091 : SUCCESS in 1h 56m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ebef9b8cf5f48d392b9351019534996 : SUCCESS in 1h 04m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e972f3956abc40ad89f5d86ded7f039b : SUCCESS in 1h 20m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62fca99854ce4c2391e05b1936e86a66 : FAILURE in 40m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/10a60a9991f04191a690515b244767f5 : SUCCESS in 1h 32m 08s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/77be13dc99454a1da6905a871fe6aaae : SUCCESS in 3h 41m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1c890bf686e9457eb4eb3243bf8c6cf8 : SUCCESS in 1h 21m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c955b4556a38433a9efed0cfeaa68d76 : SUCCESS in 2h 07m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c98412566b5440b8a9edd606d1f12dd8 : SUCCESS in 2h 07m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cb785d926dd748818e0fe3822ef22f6f : FAILURE in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08f342dd0a3ab1c11ce1ce6114a2c1f2612087c1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-10 02:19:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"036b5e4722fe919467b6c4a5f7d2ad9ed7f6276e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-10 08:50:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1332ef0dfdf0736235ddabfbf1390d032b1a4fd9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 09:11:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"67c61427296f55a0bc6ca69d954ece47994c91c1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 09:32:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, and from LP description sounds good to have in stable branches","accounts_in_message":[],"_revision_number":1},{"id":"a42c89be2838198525d8b228099cde459642b9c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 09:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f70e18db97e64d052fa8a92d235bce2deac68529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/54961f1d94ef4ad196286f440a38c600 : SUCCESS in 46m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb77f0b18a2e42cb85c03e2f8b2675c1 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7ae7205b99a427c9e2509086ca580cb : SUCCESS in 25m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b157987851454592b035fe78247419c3 : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5c272565bd483c8d3c14a599d89e70 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ee76f901d0f44468c35839b97b822d5 : SUCCESS in 1h 11m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/924e88ce7d9a4f87af58bb8717dd99f7 : SUCCESS in 1h 51m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2598177764a54323a2f193bd4012968e : SUCCESS in 2h 06m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6781a3e4489b4dcf8d3a968bdb87bfd0 : SUCCESS in 57m 41s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4ebacbbfad4405d873d5199a62776c6 : SUCCESS in 2h 02m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/815d813b3f2641bead7120b458db0564 : SUCCESS in 1h 21m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df7fdc02f3694ebc9286b0f029d02293 : SUCCESS in 1h 07m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a9d1df7d0b74acd9c39c79c57dd225e : SUCCESS in 1h 00m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b855aa6db6a4e3bab5b95d1a80bcb14 : SUCCESS in 2h 01m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b85f6f951da946e39d94b23a929839da : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ee678342027429e835aee4e04216a1a : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"086dc395d359f6cafc6811d711b3895575659c06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:28:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a343757d6bc9a1eb359e3436d2595219869cb6df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:30:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c324f483d0d54c6b8ce1a1a4615fefda : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dac4539c10c3d68779f0337d423e2ea40e3078cc","revisions":{"dac4539c10c3d68779f0337d423e2ea40e3078cc":{"kind":"REWORK","_number":1,"created":"2021-03-09 13:07:17.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/63/779463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/779463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/779463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/779463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/779463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/779463/1"}}},"commit":{"parents":[{"commit":"a3057f0967a99ec4599098d7c9338bd272d0b9c6","subject":"Merge \"[L3][HA] Retry when setting HA router GW status.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3057f0967a99ec4599098d7c9338bd272d0b9c6"}]}],"author":{"name":"liuyulong","email":"i@liuyulong.me","date":"2021-03-01 02:39:08.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2021-03-09 13:07:17.000000000","tz":0},"subject":"Revert \"DVR: Remove control plane arp updates for DVR\"","message":"Revert \"DVR: Remove control plane arp updates for DVR\"\n\nThis reverts commit b3a42cddc5c8716091d2b403831fb406d9dfcc75.\n\nReason for revert: bug/1916761\n\nConflicts:\n   neutron/db/l3_dvr_db.py\n\nChange-Id: I12b5e672b01f64fa118fe50d3add09ac2ae419c5\n(cherry picked from commit e6ddea1aa0c17fd16e746fe49696a675894f7209)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dac4539c10c3d68779f0337d423e2ea40e3078cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dac4539c10c3d68779f0337d423e2ea40e3078cc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
