)]}'
{"id":"openstack%2Fneutron~779659","triplet_id":"openstack%2Fneutron~stable%2Fstein~I97ada62484023b9833ed12afd68eb4c8d337fd1f","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1896734","hashtags":[],"change_id":"I97ada62484023b9833ed12afd68eb4c8d337fd1f","subject":"Improve \"get_devices_with_ip\" performance","status":"ABANDONED","created":"2021-03-10 08:00:33.000000000","updated":"2021-04-14 13:27:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7bd438cf12237d1e3a5191fe9408835a57ac7a4","_number":779659,"virtual_id_number":779659,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-10 10:34:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 08:00:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-10 10:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57444e9014d72fb521379fb4bd84907d709152a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 08:00:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc49094d8a60e6d889c47f2cebd926da492dfbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:34:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/23eac0ca465345d79bc798edd9973d09 : FAILURE in 24m 49s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d7dda69fbadf45f089d371f3182789e5 : FAILURE in 27m 49s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/91481fff500949eea452e9fdf5c9b882 : FAILURE in 32m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e893d1e42dbb4af3bdd7c302d340d1f9 : FAILURE in 27m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93014c69b73c48dea55ba047972bb304 : SUCCESS in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4114b72005d74d6980bf6b661af717ca : FAILURE in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e04f8246160440a815156fb1f37bfd3 : SUCCESS in 26m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f60c716c720046b1ba15a1253e346f4e : SUCCESS in 22m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0774d36e04154d888232c048226fbccb : SUCCESS in 27m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95d3e4ebac384afb95d4719210b76a73 : SUCCESS in 24m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ebb595387644148264dba9433506e3 : SUCCESS in 5m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/96f0659164f04dd0b0f05b114ee62274 : FAILURE in 27m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b5efa8d8aebc4332a78a6d1419b93d90 : FAILURE in 48m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80f1d6fd38414d11b9b1e4438917de47 : FAILURE in 28m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ac66f5759374389bb4054271513d34b : FAILURE in 33m 47s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c1fb07f2b4984d839144095535bdb9c7 : FAILURE in 29m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/df90070fa8a74d238983084be7b531fd : TIMED_OUT in 2h 14m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/050ddbac7d3f4c30b893f8982084ad5d : FAILURE in 29m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cadf133f63db49fb8e3a1b24f50fe471 : FAILURE in 27m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3ab0013c44e1404ba6c3ae0a9b5d2b6f : FAILURE in 32m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/471601b623ee4cac8b924b2a511b64d5 : FAILURE in 1h 01m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd02ebb454894b95ae459de5a0240a9d : FAILURE in 1h 02m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ea761bb0706a44ecac0eb50074efe8ee : FAILURE in 34m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/405d45dc7da1402ba6d260ec4da51282 : FAILURE in 33m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e4dce434b363484bbd7c44c2dbcc57e7 : FAILURE in 30m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dc2657206f464b78b3669a0254cf6125 : FAILURE in 35m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfedd777af774884b2da6fa1f5164f89 : FAILURE in 32m 39s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6c0e978a9c6841f099a5db9da9acfdf9 : FAILURE in 2h 26m 42s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/03d8db75124f43a7aa15f0cd150c70d9 : FAILURE in 51m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8f61d4a1341744699cab2b91ec5a64bb : POST_FAILURE in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7bd438cf12237d1e3a5191fe9408835a57ac7a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:27:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0bc31e63f957fc23762d9eb0c663c627e3b7c1f","revisions":{"b0bc31e63f957fc23762d9eb0c663c627e3b7c1f":{"kind":"REWORK","_number":1,"created":"2021-03-10 08:00:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/779659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/779659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/779659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/779659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/779659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/779659/1"}}},"commit":{"parents":[{"commit":"ed43b511f65513f8d60eb91f27c22cdacaddace6","subject":"Merge \"Fix deletion of rfp interfaces when router is re-enabled\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed43b511f65513f8d60eb91f27c22cdacaddace6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-24 14:46:10.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 07:59:35.000000000","tz":60},"subject":"Improve \"get_devices_with_ip\" performance","message":"Improve \"get_devices_with_ip\" performance\n\nAs reported in LP#1896734, there is a limit in the size of information\nthat can be transmitted in one single message between an application\nand the privsep daemon. The read socket buffer is limited in size;\na message exceeding this size will generate an exception.\n\nIn order to limit the amount of information to be sent, this patch\nimproves the performance of \"get_devices_with_ip\". In the previous\nimplementation, the whole list of network devices from a namespace\nwas retrieved. In some environments, the list of devices could be\nso big that the list returned by \"privileged.get_link_devices\" can\nexceed the read buffer size (as reported in the LP bug when the\nOVS agent tries to retrieve the list of IP addresses in the system).\n\nNow the function calls \"privileged.get_ip_addresses\", that returns\na much smaller list. This patch is also reducing the number of system\ncalls to just one; the previous implementation was retrieving first\nthe devices link information list (that method was returning a much\nbigger blob) and then, per device, retrieving the IP address\ninformation.\n\nConflicts:\n    neutron/agent/linux/ip_lib.py\n    neutron/tests/functional/agent/linux/test_ip_lib.py\n    neutron/tests/unit/agent/linux/test_ip_lib.py\n\nChange-Id: I97ada62484023b9833ed12afd68eb4c8d337fd1f\nRelated-Bug: #1896734\n(cherry picked from commit a6cdf273c06a3bba713d7b8d0ef3a9b71b0ffc57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0bc31e63f957fc23762d9eb0c663c627e3b7c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0bc31e63f957fc23762d9eb0c663c627e3b7c1f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
