)]}'
{"id":"openstack%2Fneutron~779706","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I99a60b9af94b8208b5818b035e189822981bb269","project":"openstack/neutron","branch":"stable/victoria","hashtags":[],"change_id":"I99a60b9af94b8208b5818b035e189822981bb269","subject":"[OVN] Set mcast_flood_reports on LSPs","status":"MERGED","created":"2021-03-10 10:34:03.000000000","updated":"2021-03-11 18:32:28.000000000","submitted":"2021-03-11 18:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779706","meta_rev_id":"49ab19605b436f859704118b2defc684ce7fffca","_number":779706,"virtual_id_number":779706,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 18:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-03-11 14:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-10 19:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-03-11 14:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 15:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 19:48:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-11 14:57:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"884b7ce8a982fc7ae5585cc33899804a7cbda305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-10 10:34:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb14ffbe9b31b35779e1df7de3cfec0763f19409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:25:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/24536d33429d4e869bfafe3e4f571296 : SUCCESS in 45m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/7595dd9b4e754f96bfe2890f94c20421 : SUCCESS in 59m 41s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/011d826ddbf44c37bcb5272cb3b4d8b2 : SUCCESS in 1h 21m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/bc3c5656ecd94fa7ae27093af396bb21 : SUCCESS in 1h 18m 40s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/78cae7cf55784dd3b9a918f49d449502 : SUCCESS in 1h 28m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65f2c3a27e124c94b87a67eb58a8c07f : SUCCESS in 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9082edba05bc43bab087d57e5d253807 : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78c4be5478a648059775ba16b944ed2f : SUCCESS in 32m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d20180ef318f426485057407a8701284 : SUCCESS in 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc6040d4e4f48bf8b241173fea87927 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df7bd47333524a0aa216abb732d1ad3c : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/33dd926c0372400fb7bc4ef9a6e7b719 : SUCCESS in 1h 14m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d909279027254f3fa78a2b9830e6cb0b : SUCCESS in 1h 14m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9263a82133a94561a993c7bd377ac923 : SUCCESS in 1h 35m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbc08aa7967f4c91a6ad4f0176d9aa6a : SUCCESS in 1h 05m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48897880a91c40dd84ca40ed7b20bd6e : SUCCESS in 1h 05m 56s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7b62dac039874607b02aca834a627567 : SUCCESS in 1h 29m 08s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/926f5fb7ba3e48fe9ca299ba215a3f95 : SUCCESS in 1h 14m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/33ac529d11cb4156b7182edf123d7019 : SUCCESS in 1h 32m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/37e62158a19e4684a5c5b946bea8c6f1 : SUCCESS in 1h 28m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ef1b6a90bbe4e9fae7e8933f3e5cd50 : SUCCESS in 2h 08m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f15cb3a353fd4a6091a46bfea1385c2b : SUCCESS in 1h 20m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/aa1b08888b9e40b29c32b3be1dee904c : SUCCESS in 1h 17m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bb8981eb58f42b1885b9c70c0448126 : SUCCESS in 2h 14m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cbb47f444144e30afb86b3571b405e1 : SUCCESS in 1h 02m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de51fd56223a46f99db9593b0c717bce : SUCCESS in 1h 20m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4c1e713c256f4649842c767116d071de : FAILURE in 41m 38s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5dc211b9fc1b4e8eacd5f7ba4961a00e : SUCCESS in 1h 23m 15s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1f22d952706e4a8eb136ff37b6597ea2 : SUCCESS in 3h 14m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eea2f466e8984eb5ad045323dd6c647f : SUCCESS in 1h 23m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/886057bff8d549f3ad907f09e1a45243 : SUCCESS in 1h 56m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c535458f2aa94424a818a36ed93e347d : SUCCESS in 1h 55m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6af5cb93ffa84b28a1fad1ca0c743874 : FAILURE in 42m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43bb86b50a64087579ef5fba34872856af64af7c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-10 19:48:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54c43bf82f398adc2f8a82086fa4765758912ee6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-11 14:57:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da2d3f28b3bb980cf79450ade61f9d83ecbf1ac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 14:57:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73af2a4ec1acaaa63421436d13209e22b9d86408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:26:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/20f3eac683104ab1bdacce53c745a90b : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c7aab7969e440ca27ce25d9baa646b : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62616b5d91094e1583aba3f07f1d085d : SUCCESS in 25m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765653b451f844d2a29b0f4691b47348 : SUCCESS in 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e35b06723234c3ea6facab4e09ed4e2 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67e3c75f007f42fda95a1cfef548157f : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1782bd59d56b4301bf5a50f2b604722c : SUCCESS in 1h 20m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4c01ef028ae4a20ba99cc8b1531e052 : SUCCESS in 1h 19m 43s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/35cf0d79227245abaf399287db366971 : SUCCESS in 1h 28m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/63b075f101084c9c8a8f63b1653d29dd : SUCCESS in 1h 17m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0247c2f1ebdb4c9f86b362ae88a77979 : SUCCESS in 1h 20m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/21fb1c457a654c88ad152d29995dbe23 : SUCCESS in 1h 40m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/430fd02b7a9142d8bea494b5270627cf : SUCCESS in 1h 05m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/74725e0bbcf04af8859f57b36c05bf3e : SUCCESS in 1h 11m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca868ff2b04e4b4383c5eaf2d9eb0faa : SUCCESS in 2h 10m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2fce43e1f4144c196995304842f81e4 : SUCCESS in 50m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2379ab69052741d3a159aae591b518de : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8cb909daff8a91131ced0a6ad3efb80ea16d0719","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:26:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49ab19605b436f859704118b2defc684ce7fffca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:32:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7310ba67873410ba705ed936b8108ca : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64be638291fb4de9b6fc7a0e47e222fa : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44c367d98520cbc0cef60ce914c304918dce660b","revisions":{"44c367d98520cbc0cef60ce914c304918dce660b":{"kind":"REWORK","_number":1,"created":"2021-03-10 10:34:03.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/06/779706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/779706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/779706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/779706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/779706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/779706/1"}}},"commit":{"parents":[{"commit":"a3057f0967a99ec4599098d7c9338bd272d0b9c6","subject":"Merge \"[L3][HA] Retry when setting HA router GW status.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3057f0967a99ec4599098d7c9338bd272d0b9c6"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-08 11:44:24.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-10 10:33:58.000000000","tz":0},"subject":"[OVN] Set mcast_flood_reports on LSPs","message":"[OVN] Set mcast_flood_reports on LSPs\n\nThis patch enables the \"mcast_flood_reports\" and \"mcast_flood\" (on provnet\nports only) options in the Logical Switch Ports in the OVN driver. Without\nthese options, the ovn-controller will consume the IGMP queries and won\u0027t\nsend it to the LSP ports, meaning that external IGMP queries will never\narrive to the VMs.\n\nIn talks to the core OVN team, it was suggested [0] to enable the\n\"mcast_flood_reports\" option by default in the OVN driver (at least until\nfixed in core OVN) as a workaround to this problem. And, to avoid having\nto update all ports (which can be many) based on the igmp_snooping_enable\nconfiguration option, we are always setting \"mcast_flood_reports\" to\n\"true\" in the LSPs. This won\u0027t cause any harm (also confirmed by core\nOVN developers [0]) since it will be ignored if multicast snoop is\ndisabled.\n\n[0] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1933990#c3\n\nCloses-Bug: #1918108\nChange-Id: I99a60b9af94b8208b5818b035e189822981bb269\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit b04d64b90f2822222f96ddd726d379a2f19b539e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44c367d98520cbc0cef60ce914c304918dce660b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44c367d98520cbc0cef60ce914c304918dce660b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
