)]}'
{"id":"openstack%2Fneutron~780671","triplet_id":"openstack%2Fneutron~master~If56c2c2da1a84aa686fe13360a56592e30fc183b","project":"openstack/neutron","branch":"master","topic":"test-centos-jobs","hashtags":[],"change_id":"If56c2c2da1a84aa686fe13360a56592e30fc183b","subject":"DNM It\u0027s just a experimental centos jobs test","status":"ABANDONED","created":"2021-03-15 20:51:02.000000000","updated":"2021-03-16 06:51:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ce54b623e4cf5406757d6d916e34978024e9512","_number":780671,"virtual_id_number":780671,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-15 22:55:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 20:51:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 21:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 22:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bfb9082ac56c42e306ba00074fd7b875fa80932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 20:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45c95b0344d34fd342aeb4ab305c7e467470cbdc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 20:51:30.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"28a46fe757169aa430a0b6d071637943cb93f078","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-15 20:51:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"e6c19a6674bd380ea10623864c5ed763a34b6dee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 21:39:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/439104d32f684cafa51afc92bd46d77c : SUCCESS in 38m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/059c726a8bd9418aa190d7fe85056985 : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc424eec668126cbc855ac6ee14d3cf67184f296","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:27:00.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/60ddad1982f243869827c5ec670df2b9 : FAILURE in 1h 27m 42s (non-voting)\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ca6631bf25646e7b528775de1f6385f : SUCCESS in 28m 51s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/42d88df5c8c6414e877cc84736928830 : FAILURE in 26m 42s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/9fb81d318ce141d49a80610c2effddfa : FAILURE in 1h 02m 31s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dc7ac59c383044be9c9cfdc18e86bdd5 : SUCCESS in 1h 32m 28s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/908a456ead764ad8a9b0ed58c0da4ed1 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6a2c55eaf78e8593448e5fbf762cbc55f0980ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:55:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6b9a671554574866a2970931fe85228e : SUCCESS in 3m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2eab61ca6d96418aac7f29c94ea71e95 : SUCCESS in 48m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0dd4e98647d941438cd3ec371477bb1e : SUCCESS in 1h 47m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bc65de99a5a2409eb862f4ab0f8b248d : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a35dcb97f863480bb8ddea2d297dfa5f : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/70cd264dd272438ab05b5a5b861fb87c : SUCCESS in 2h 02m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/535507bc439b42029e0aaea53cfd1c99 : SUCCESS in 44m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fa554d8a3a04b518c0c719f785175e9 : SUCCESS in 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e0951ed26f46b3bb87461b78f1e9f6 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab0306d02b3c4fa7915700d9cce09ba7 : SUCCESS in 29m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1b21f51649e47859634f0dae4263513 : SUCCESS in 35m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f957661a57e45f09cdbec83c5f2a8b9 : SUCCESS in 37m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3979ceeed116406cba6e8ddae69e1f28 : SUCCESS in 11m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7db53d0c702a41c295754f975f6b85c6 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afd1e3a479604d70a2013f48ce57d7c6 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0214831c0004998b29862feedf91f81 : SUCCESS in 48m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82a05114dbd441b3a52fca1abf565755 : SUCCESS in 1h 23m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/42d982c6f2f14303950469e5399aaf91 : SUCCESS in 1h 29m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04e0c952da6f49d5adc711c65af0659c : SUCCESS in 1h 33m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7f32d9d60da84c618232dae80deb5abf : SUCCESS in 1h 03m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f17388bb27ac4c69910d192a23950b94 : SUCCESS in 1h 18m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bdc6ad73a1fb4454a075b2a4892b1a37 : FAILURE in 1h 15m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebf51b17ea8146c39f3b8f943e8ed0f4 : SUCCESS in 1h 40m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7263c54b89b4255950752459cfb5e75 : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a71a76a6faf4091931ea1c868dfc8b3 : SUCCESS in 1h 21m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/72043832c81d420987a9140824dd5fb5 : SUCCESS in 33m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7188ce0ba7564a7aba956f52ca9ffa13 : SUCCESS in 1h 38m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/659b1200e40544d28deb4465e37cc5a9 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1ce54b623e4cf5406757d6d916e34978024e9512","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 06:51:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1bd248e15b477cb474af3fcce1abd393a50c758","revisions":{"d1bd248e15b477cb474af3fcce1abd393a50c758":{"kind":"REWORK","_number":1,"created":"2021-03-15 20:51:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/71/780671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/780671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/780671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/780671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/780671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/780671/1"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-15 20:49:57.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-15 20:49:57.000000000","tz":60},"subject":"DNM It\u0027s just a experimental centos jobs test","message":"DNM It\u0027s just a experimental centos jobs test\n\nChange-Id: If56c2c2da1a84aa686fe13360a56592e30fc183b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bd248e15b477cb474af3fcce1abd393a50c758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bd248e15b477cb474af3fcce1abd393a50c758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
