)]}'
{"id":"openstack%2Fneutron~780802","triplet_id":"openstack%2Fneutron~master~I191e020ddb97dcf8fb41139d35bfd699e125379b","project":"openstack/neutron","branch":"master","topic":"bug/1919280","hashtags":[],"change_id":"I191e020ddb97dcf8fb41139d35bfd699e125379b","subject":"Fix VLAN driver tests","status":"MERGED","created":"2021-03-16 12:03:54.000000000","updated":"2021-03-18 17:25:57.000000000","submitted":"2021-03-18 17:24:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"780802-bug/1919280","meta_rev_id":"617c8b8f5ed7ac2f3fcce9865057769e7cd1bf7c","_number":780802,"virtual_id_number":780802,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-03-16 17:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 17:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-17 00:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-17 08:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-03-17 13:37:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2021-03-17 09:56:37.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-03-17 12:48:29.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-17 08:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 12:19:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-03-16 12:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 13:35:36.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 20:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 20:56:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 00:38:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-17 08:44:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-17 09:56:37.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-17 13:37:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-17 18:18:00.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"CC"}],"messages":[{"id":"b9f8c08a3bb7fb1bb130c94deb80e15eef6691dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 12:03:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97f025b1abebbfe2700400b8d73c595c8c7b4f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 12:12:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e928f95a948a16fcd66f6134b40c6905a7496f8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 12:19:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2506c1c518e27b8e328b44fe94b5fcac00fb0554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:21:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dac2e2ec9225488891518e9e446fa821 : FAILURE in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59f055e1b983476ea61fff51d810d591 : FAILURE in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"474cdc73a5c77f62422c0eaea90cd7dd553ddb0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-16 13:35:36.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/02/780802/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"cb2a0b88ddfc32c41f1fbe150c3ea6b6b7a42560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 13:44:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"910472397d09dbf0c4402dfb290edf2a9b1bde95","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 13:45:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fa7df95c8f7d3a576509a015406d64a3dd125795","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-16 13:45:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/02/780802/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9362ca7c8c7fefd4235497dfc185f68e144b236d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:40:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e9ea1e5d52c48b189a240375c80fbcb : FAILURE in 42m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"753c995d7fef69475e3f9b5976ec308aab9bbe62","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-16 15:08:17.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/02/780802/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b2b6e33dc0dd679996d68483bd9456908999aff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 15:43:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85659c225ce418d476ec283e5a1aa9325df99208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:04:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6094501849e4459eb9fcdcb409ec28ca : FAILURE in 6m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28bec762613a473fb7a71337f9f176bf : FAILURE in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14df085213486380562e882c1ecb8bfa735839b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-16 17:15:49.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/02/780802/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a8fc167b38a7e7ba0ad313169e5e9400cdd0b1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 20:13:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7023c66c3d8242bd94626da5b774ebe5 : SUCCESS in 52m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc38bd6683794f9f90a7fbf2ec57527c : SUCCESS in 1h 24m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/32f00943b7f44161b02bc7c36fca9483 : SUCCESS in 2h 18m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5db89097c6014991bf6b87fe68eaa9e2 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c4d242bb9fe94520926332c0ea2791f0 : SUCCESS in 2h 06m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8723511fa2bc441bae0e4b8b877a0356 : FAILURE in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6753ae66bd1e4789ae38e2a3f5931699 : SUCCESS in 29m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a07b5f151c94bee99d5ea32df430779 : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61bb8093ea39453787c6123428dd1acd : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b434c110fa4044a594d30edd3686b479 : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b02a77a00b745619737b694beae9f0b : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5334f0daa584238bb86ea25d60163fb : SUCCESS in 10m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1734a878be63415d9f6b68f83e0575b3 : SUCCESS in 4m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f9b85f0264448d08e0bdc86ce6f7790 : FAILURE in 19m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8efb351c0ef54958abbeca3edb7e9f39 : FAILURE in 14m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5faa7004e54a41c48aaa38310894f6f0 : SUCCESS in 1h 51m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0c95631a2a64b47979920518600a4c8 : SUCCESS in 1h 19m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32c7fdf10b5d44c5a0f5183670393558 : SUCCESS in 2h 48m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/571f85cf76864e5ba9f9f711ff9d1171 : SUCCESS in 1h 22m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/732067ec7c264ae0985a71871d1deda0 : SUCCESS in 1h 38m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36f0c28bd9e84adba12d6995de33bd7e : SUCCESS in 1h 39m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d08e0015b6ec43569fffbce6a7f2c6de : SUCCESS in 1h 20m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8371aed18c3647c39336af6cdaa01cc7 : SUCCESS in 1h 09m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bb7348df74a44311b1e3392b39c24eab : SUCCESS in 39m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c9ae36407de4149a143cbbe857d3a87 : SUCCESS in 1h 36m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1966c5e6f7074bce9892796e34c64ed8 : SUCCESS in 1h 31m 59s","accounts_in_message":[],"_revision_number":4},{"id":"ae8026cba7778401d64d58d05456e2532cf3604a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-16 20:56:35.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"21f21339eb2ca0794844a3c7c34c06efa247990b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-17 00:38:37.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck\n\nthis needs to merge so reqs can re-enable cross tests","accounts_in_message":[],"_revision_number":4},{"id":"adbcaafbbe0620d39c32f8f15533eb723d64e86e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 01:42:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fc11595693d4d5e9a7234527b47379a : TIMED_OUT in 1h 00m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32942b587eee4ae28c03f83ce38acbaf : SUCCESS in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d71fc363e6fcfb6aa0de0a94a26ff3433cb1b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 03:35:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2caa8042abed4a31a46a155bfad3ec89 : SUCCESS in 1h 03m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0caee1fa3ba747328b8608a5c37b226a : FAILURE in 2h 49m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dd2d298e8ea94f83a17694f79792ffdc : SUCCESS in 1h 39m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/43b5fe993aab4736af8b7affbb2b533c : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1614d6ceec554e5b98cfa704f41954af : SUCCESS in 2h 19m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9a68b9e2cd24ee4a7f4edcf56f25557 : SUCCESS in 32m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a1689b803d34ce39205d17daf3ea982 : SUCCESS in 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61403bd4570847e38f5e3aae54db8e20 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ce052c08ce1494393473342eea63668 : SUCCESS in 42m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf0fdd07d0c4690af004596046b430e : SUCCESS in 37m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f4b2d50fe1c4defb021c5895ae3f371 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d743314485418f8ea99c902e64ee55 : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ecda60f4c8494054adc5022f254cad32 : SUCCESS in 4m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82c5284ad8a246fdad5da85c14d11275 : SUCCESS in 50m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb382b1fd9f74d6a9f789e473b8bcf15 : SUCCESS in 1h 21m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3338426720554eec88e3e2e725132784 : SUCCESS in 1h 23m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/395dcd133abf4148960208c87a0632d2 : SUCCESS in 46m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a91774df084a48ea9730eb5ff1b40406 : SUCCESS in 1h 05m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/21a36d7882354b2f9b396608e27d07e0 : SUCCESS in 1h 14m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ac7b627d997f42f198f36588f0f13a38 : SUCCESS in 1h 18m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60dedf0228c943dd8d7d4755a0d22c3f : SUCCESS in 1h 30m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/250cff0362354808829c72c91e206905 : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b6169a41d60d4212ba83fcaf8f71cd07 : SUCCESS in 1h 07m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f911cc4f97824a11b516be9878182ecc : SUCCESS in 35m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e034601777de4f6ca351a1e8b92c0554 : SUCCESS in 1h 19m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0dc62f4981b442308cc3f02eaa1dd50d : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":4},{"id":"cb2cdb202ed5e801f5deac2f5c10fb6245ae6de6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-17 08:08:37.000000000","message":"Patch Set 4:\n\nrecheck\nneutron-tempest-plugin-scenario-linuxbridge ssh timeout","accounts_in_message":[],"_revision_number":4},{"id":"4e066bc9457cf01100954010ce738b47966298a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-17 08:44:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2691c3f98dcdaaae4be6a331fda3d94b20b06185","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-17 08:48:50.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"9ed26d642ee7a7443ada4c3fcf06466dd8de09be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:06:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8853c50861f14b76b904b2c271746081 : SUCCESS in 45m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b326c35d66f4c4ab2b354145986225f : SUCCESS in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"107bf5db3a32b64be85d15ac44ce9220a14425fb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-17 09:56:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9c5c1f95985c31eeca5409ff9291b9ccc6c95ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 10:11:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/14015d0e403549a4b592e40f9ccf800a : SUCCESS in 48m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7580abfb186144b19d4ca6f07238eba4 : SUCCESS in 1h 48m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9004b733d1d43f8b04e24b1ccadbfe5 : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c10d4fe056fa4e108ad8f7ea50e7f41c : SUCCESS in 1h 18m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ffe738c1a7834fbcae38d9618eba4b50 : SUCCESS in 1h 49m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97b10371370d473bba1183ef7c930c07 : SUCCESS in 32m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f71d032e8d54768b497d4c31d2df1d6 : SUCCESS in 37m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e768be729c4a889db46ebee8e5c2f9 : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f40cd0c9b5492b8f967807a57f4aba : SUCCESS in 29m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5be8208044774358a7ce17dbbd39f8b9 : SUCCESS in 35m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/865052a2c38144cfaba5fcee047b6726 : SUCCESS in 30m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2aafc39dd643e1856daedf8372afd7 : SUCCESS in 9m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca36d82747af4349b2f2fd0643bf339b : SUCCESS in 3m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd5a3ac732214818b7790b1841e1f9bb : FAILURE in 47m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4ea3d1d0a5542af823fe932dd26d507 : SUCCESS in 1h 14m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/559ba5ec0f4d49799ad411d23e570d3c : SUCCESS in 1h 27m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a148b007e0264ccc8808be02abc337ea : SUCCESS in 57m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/105e184a64844d869c58fdc998d46ce3 : SUCCESS in 1h 01m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bee563291d53470a9a68451cf1bda74a : SUCCESS in 1h 09m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29053201c3ea4bb9b518dd83661806b0 : SUCCESS in 1h 57m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b53e4464e584c18a763da5cdf9b8d68 : SUCCESS in 1h 46m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb4642de824b4ec687a9d14aa39b5d1a : SUCCESS in 58m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dde05f5e02cd42ffa900349465e63aba : SUCCESS in 1h 15m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/598ec498bd5d4a32a55174be70f05365 : SUCCESS in 39m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f686f0599057424c91398773676cbee5 : SUCCESS in 1h 23m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/500950633a1a48589ed9c582028fb749 : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":4},{"id":"915088892cc96b52e9ecbc8fdcc3fc80dec14648","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-17 10:51:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3fc5e5dc9dabdaa88343bc43a4b8557ef9803cc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 11:48:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ffe0e2322394c7081968d04fe344d8a : SUCCESS in 44m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf5d39a658c14ae2aa21d7e7e43cfbfe : SUCCESS in 51m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"54b8d7c55499952df9630749d1081af5ba2b80f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:42:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af594ff07cbc4797bf662c4c6cea0c04 : SUCCESS in 48m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e83c70c92c544038003ed1c899095c6 : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8df8bb449064de3ae7ef05c5861b468 : SUCCESS in 1h 49m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5b62fbce0e54e34a37bf29e0a424ab9 : SUCCESS in 1h 10m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6816f8fd98de45d0863cc6fafb4c10ba : SUCCESS in 1h 33m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/577da29405c74140ba3e6082afbc5590 : SUCCESS in 57m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0785d290cc374922b90770fd24589c78 : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1beec02261844113ae711281837424b9 : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d3c6528e7c45da9d948bff647ed56a : SUCCESS in 51m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4ac5adb54444ce4a91d1dd792067aae : SUCCESS in 31m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bed7bc11d8124112b31ffbb0ce79030d : SUCCESS in 36m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3365f62385412cb642e38cc81be30d : SUCCESS in 11m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8932dd8277744b17b91385fd5fdb2882 : SUCCESS in 3m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aeba1afac31c4b31a668538f4c923e24 : SUCCESS in 55m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0961d102360544e7a3cf420b72b5efb2 : SUCCESS in 1h 25m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/13bb00d0775549708cf1edd7a991f0cc : SUCCESS in 1h 41m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6196136c56241daa270e7d8ded596c4 : SUCCESS in 1h 04m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f259f5dd710d465ea39ca4d85c449ca5 : SUCCESS in 1h 05m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/683cef08b232410f8bb2ce2d141a0c0f : SUCCESS in 1h 18m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7e0fa7fdfa0429e80d297ed65b553e6 : FAILURE in 1h 12m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/914ea2a8f9a64a9e96e0349786a23582 : SUCCESS in 1h 34m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63881f60c8ba4e4cbe280adb4041d2ae : SUCCESS in 59m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0af5c7908b804800837801ed2357db3c : SUCCESS in 1h 06m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f6cfdc8b898e4cbba2f3474c9f5c247f : SUCCESS in 32m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90b2e3302bb04a36924c00bdfd15f239 : SUCCESS in 1h 26m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0c5734508f0d42ac95c67761eb24e850 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":4},{"id":"854dcf494e6ca28a55d06f453782b5acf5bc4b35","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-17 12:48:29.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"489e10a885dc8edf3c2d3ade7eb2eb0d81563f97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:48:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cef9a98f7d02154ae7d3ff05aebacc94546a270d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 13:37:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"afaade4cbd5d61b7ea34d736a9742a0b01858f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 15:10:33.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af09be255e024c399ba75e6f228ea693 : SUCCESS in 49m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7a82e84e54bd4cffa024362967b4b2d8 : SUCCESS in 1h 36m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/18346146df9746a995705d72637685f9 : SUCCESS in 1h 36m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3855addd72cc429288ea201df4e0d168 : FAILURE in 1h 45m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3f9f0b5e2364b0bb4c28d6ef6fc3d07 : SUCCESS in 1h 48m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a7e30e07a9c40e0965a59cc42734215 : SUCCESS in 40m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb4dd9403f84cd6942d7c16c31d5913 : SUCCESS in 15m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee89b7181c484429b86b768c741610f6 : SUCCESS in 37m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ddad91c02d41248470e80215ecfa31 : SUCCESS in 34m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50635f220d2142199157e1250fffd26a : SUCCESS in 11m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8fd7b0f18546442f8fe0a6dd0aa4ff7c : SUCCESS in 5m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3361ee1feea64c419fa789f9f0998401 : SUCCESS in 54m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/981d6e3464a54f3b927e13419686d297 : SUCCESS in 1h 22m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c1e2f1d8ff04e25a205fc91161e787b : SUCCESS in 1h 19m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f985e70aecc41f8bb7fc971dc8de579 : SUCCESS in 1h 18m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ec4e4e0a93c451cb95f2814b062311b : SUCCESS in 1h 06m 15s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f538bbfb5cff480185f11bd0e9344a25 : SUCCESS in 1h 34m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/851bf144ea914b64b6285a8880fc6432 : SUCCESS in 56m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/86df24f036f04ab8b10994494195f59f : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9011d47fb98b4370a79ccc88e2a786e5 : SUCCESS in 36m 22s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d9709702f82c44fab0ee9b3518fa0afb : SUCCESS in 1h 35m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a013ac714fee41602216a9edad2479082dec93fb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-17 15:37:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6f50188b0b4e4bd5eaec77ac8520f188bd5f4915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:38:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fcb6da56f7e4e87b02217a26e805850 : SUCCESS in 50m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea0a7f896d054d41a0d495f8dc2736f1 : SUCCESS in 48m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"01bd07cc75a0d8bf2e2b08100e613ca6653a50d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:59:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fed208c9afad4a8aaff74e3a14466c49 : SUCCESS in 54m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c053ddcac8ad4075a14d0cf9f8eb4814 : SUCCESS in 1h 58m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2967dfa264284c80baab97d02398085b : SUCCESS in 1h 50m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/013c945d52a04f4285f6535593889333 : SUCCESS in 1h 30m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fa15cbd179704579a39480f4ac6070b0 : SUCCESS in 2h 15m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e96c58a1202c4f80afc4a90fc5b44017 : SUCCESS in 40m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d0011dd1b004d8689539644e44d0ea2 : SUCCESS in 36m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1131adb7ee4f0eb2e407408c317b77 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/172c5bd56d6343ac82923cf0dd1b6152 : SUCCESS in 39m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0468bf8913524a53baae7f4c5beadab4 : SUCCESS in 31m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/686ba71a3b184d9ba49f986e92293053 : SUCCESS in 38m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3191cee11f44249cf7827c052af3d2 : SUCCESS in 10m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afb1f479b3114d49857672412944aa9a : SUCCESS in 5m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9da9f362371a44098ef3ee00fc0b4df1 : FAILURE in 49m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a67c4633c3d4836b1e63c7ffe9b2cac : SUCCESS in 1h 34m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/18aba081b6f7451b9b3259fe17e77ab6 : SUCCESS in 1h 48m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be461b4b59d1452da71c6014d6293c4d : SUCCESS in 1h 12m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1680148f0f5437f82772c90460b3d22 : SUCCESS in 1h 04m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3551a183b3a94bb193ccb21c19126334 : SUCCESS in 1h 24m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2337ba1578894f5d96853389dd67e002 : SUCCESS in 1h 58m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/726bbf36e9ed423aa9cdc18aa48f8ced : SUCCESS in 1h 35m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe96dbfb018d4556b186ae0287c18f30 : SUCCESS in 51m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce197f7780ec4e0aa6389d5734293d3a : SUCCESS in 1h 27m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c0db8b2074e34ef1b06202b7d8bafb49 : SUCCESS in 34m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4695b5a2cee747279b9f25d275093f50 : SUCCESS in 1h 25m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1be4d633d27341ba8c9ffbacd0e5af43 : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":4},{"id":"cb51fc790db745bae0924bcb9b538aa09fe826d7","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2021-03-17 18:18:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d97ea94bf1467017e2db9ab168f7c41cde029d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:36:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2eedf14b0dcd710b24bc02a86a5ed8864fbd47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 20:19:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6dd42d48e5464b2b9b4ea89fb95a7086 : SUCCESS in 54m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d49dbf0598a4367a199ec0e5df308eb : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ed6b3cee80a42e085f29a67a19947a1 : SUCCESS in 1h 55m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/39d46246c8d94a098cfc8b0742e5784f : SUCCESS in 1h 20m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fe16b163ad2a4f7c9f96dff77c8e4bc4 : SUCCESS in 1h 50m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19c94ba5529d47898a14d2943ff0cfed : SUCCESS in 42m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d45ca8c9f55485da39bdd04e28de3a9 : SUCCESS in 32m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cdf3bf52a30481a841bae2f906c1198 : SUCCESS in 17m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35677e0688fe4127a5b83c76d4106d5e : SUCCESS in 41m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b914cc9d2dd4c72b82a910f669ef065 : SUCCESS in 41m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba7b98625cbf410bb4608f1334dcb099 : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2be5aacb9a4c89a2587e00ae3a8481 : SUCCESS in 9m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da78a32fad114cf99e046a24ea67ed41 : SUCCESS in 4m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/395e3e0adb864bd8aa5f0a4b08d79a30 : SUCCESS in 41m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/092ea624410d41a4a54294bb17535a84 : SUCCESS in 1h 30m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8c257febe9d34dca935fe869220534b0 : SUCCESS in 1h 35m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3b250ccc50143f0854f70b34c693c1f : SUCCESS in 1h 11m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b547542e6df42f784cf7a3097d4ab8f : SUCCESS in 1h 03m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd7e6ea0808540148a5bf539cae537f3 : SUCCESS in 1h 17m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c679ded5b1f43e0acbd4d7ab9d5ff13 : FAILURE in 1h 13m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1aa8d1a266b44c22a704534958971838 : SUCCESS in 1h 39m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc4730b54a94c7396397620de51d8e3 : SUCCESS in 1h 05m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/19b9afb1725e409794f05fbce10eabc1 : SUCCESS in 1h 10m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/86825358abc349b598ae02fd281c76f4 : SUCCESS in 32m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/345973c2b6e54c0fb93fdfb443f36a96 : SUCCESS in 1h 31m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/88cdfd5253824b0db0c43423b259467d : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":4},{"id":"8e84b5bbcaa2013e4f5f0888c2b7e3af9eefbbef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 20:19:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d208e975bbc699378a6aa1d1736b2df3f5bb0d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 21:51:55.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d5756011da1eb724f4f448d69ca247a831e93301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 22:06:02.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f075fdee22d59967f4080d47bd2ffb1c3d200b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:36:25.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6209ec7e6d7e76fa5c4c72a5ec979e1f6389c1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 02:51:33.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2eb9bd5458b747c1ad1eb462946ff16d : SUCCESS in 50m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7583c138a1aa4f43acd8924a6fce65b1 : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7565bd760c5045578b35e4cb04e11801 : SUCCESS in 2h 02m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/de23fe045f3a4952ab22eaff28ef923d : SUCCESS in 1h 36m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0fbd434d39340f0a8fd2d214c614c84 : SUCCESS in 1h 31m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/686f0863ca38415399e80e3315dc98ae : SUCCESS in 29m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e1b430b43548b69eac98088cc3af30 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e442a8cd5649ea983008ed7c29e719 : SUCCESS in 37m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1f55ecae3c44e0b7cf54a58ed457eb : TIMED_OUT in 41m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482c70af0cf442a48ab555c6abf7a39c : SUCCESS in 10m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d765d480fc304f0f9e53aeeb1df0e221 : SUCCESS in 5m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7c2c41320c0425d9e6bcf7aec24e903 : SUCCESS in 1h 18m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c1eb7e5b4714b00b65c245dc2dfa126 : SUCCESS in 1h 36m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a7adc3f8a9e04347a8df6a8f15923619 : SUCCESS in 1h 21m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddd28d7f37004932a2a072efdd1f926a : SUCCESS in 56m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4aa06b28ad444aa5bc2c18ade54c4d82 : SUCCESS in 1h 10m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f74e90a0a094988bab1917227d0aeb3 : SUCCESS in 1h 40m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8449c50666a54a9c845336dba8bb6cf5 : SUCCESS in 1h 05m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40d619f8b21f473b97379e007c046890 : SUCCESS in 1h 07m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b6bba100dca84759a861c1665db91741 : SUCCESS in 34m 09s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6dd96e63802d45b38a4f494ffed4d70c : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":4},{"id":"b662fe3c444d0574cacdefae2f37fc9e050c6e12","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2021-03-18 03:18:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9656ecb9829f38389a1e1b1ba759cc0903956b20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 04:19:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3745a178a3444e4afaa7de6b93e735a : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dfee6e3d72f43b5aa438ea772689cdf : SUCCESS in 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c015a591bd676efa90bec6a9b104ca155016f448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 05:30:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/302cfdbeb09347fb9cdc991486c45f89 : SUCCESS in 50m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/44a45638acac41b8b5f32747993a8fec : SUCCESS in 1h 48m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5b4354bd5aa34e999f486eba7167494a : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/615c56b8e59e41029d56d247de1e32ea : SUCCESS in 1h 37m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fd3f44d34fdd420688f4dfe1eb8d8c1a : SUCCESS in 2h 06m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91ca5bb989894f229cc22206833798bd : SUCCESS in 51m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbaab47d8dfa4aa988e39cd9f9cac241 : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9837c0b533a148c7a887ec4e5ddce230 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7111c1a8bd3f4dffbb1f5c39f2f48c7c : SUCCESS in 35m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85960b2c3b44c18b34cac21e802da47 : SUCCESS in 44m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0c60b2b539548a5be91268f78a087d3 : SUCCESS in 40m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77731cdcc5f142ef869366b181010984 : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b397017864747358e2b8970ddc04c44 : SUCCESS in 4m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fe78e289ac94b8c84175d384314d589 : SUCCESS in 48m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4b08ce2c9ff4e80aacfbf657a2197c1 : SUCCESS in 1h 27m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/769c494e6f104420bd8ee904f8a01858 : SUCCESS in 1h 37m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76b6322c761c4a5ebeeea959d94ff41c : SUCCESS in 1h 03m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72a31b97c687429eab97df3a0a7605d1 : FAILURE in 1h 11m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ff76a0780a84cce97b5562254a00a75 : SUCCESS in 1h 22m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/821a92ba1e734aef9f0812b8b5a84f55 : SUCCESS in 1h 28m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47eb02a27d4f4b5b88fdbe174ac18be2 : SUCCESS in 1h 32m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f373c6049fe7483bb4c27f175f09990b : SUCCESS in 49m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34890b4571884d8da65f587f05337412 : SUCCESS in 1h 08m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9e9dcc8c60f94d1e985421eae3ffdba8 : SUCCESS in 32m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/277287c4d8e94874b6ac7ef2f0348766 : SUCCESS in 1h 30m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/27469a4ebdb6438a8217374ae45e1a0e : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":4},{"id":"412e625264860c6aa8a8c5f228f2c5b3969c7473","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-18 05:35:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c16b8265e9a9524e9c6217325a351c1c9f919f83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 05:36:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4e04980adf9522de44547c39cf7e93916f4a0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 07:58:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e073e56b2be446bbcf4138e517703df : SUCCESS in 55m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/10d0dea0e9be479aa895895b9db6ea6b : SUCCESS in 1h 42m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/112a623ea19349dbae97d1a17d52947a : SUCCESS in 1h 29m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6465dd8b4204082a790ab7f6039d1e4 : SUCCESS in 1h 37m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/916e30d6f2d04c2e83a1549623523140 : SUCCESS in 2h 22m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8355999f62a84e178ca0c49f028aaee0 : SUCCESS in 54m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b8de8d7a50f44a182dd7d577303122a : SUCCESS in 37m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d4eed2b9d243138af8fc5faa0a4dab : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ca09a73d364451a2bbbb79622f6bb4 : SUCCESS in 51m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb08863064fc4822b4f926a3ef676571 : SUCCESS in 39m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/440407cb17fa40a38a73d2c951134299 : TIMED_OUT in 41m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b47927b09614c9891a4f5743b46026d : SUCCESS in 8m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b3c2321e4f34f91ae5986d47277ff92 : SUCCESS in 3m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55be1d1ddb21447b8efd4e0c9dc31fdc : SUCCESS in 52m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5e231b64c89472e8816e9417d613dfc : SUCCESS in 1h 24m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cef663b89f0f4c0d9ec1724241857007 : SUCCESS in 1h 36m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/451f82763b2648b3b1dba0b0ea02b464 : SUCCESS in 1h 05m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e5300b2ad883415b880c950d4c0c778e : SUCCESS in 1h 12m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a21a8e6fd8d40cfbfac4be0017715f8 : SUCCESS in 1h 25m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4dc1dbb9023449b8a96d38381435035f : FAILURE in 1h 12m 43s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3d18a3de19a4d21aaea065e4a589a05 : SUCCESS in 1h 41m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24d5bbed6cbc44b3b256feaf8b4ca1e8 : SUCCESS in 59m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e6b127d121f648fca06e75eb27649fc2 : SUCCESS in 1h 07m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12317f59a37448b9813425ca22a5ae95 : SUCCESS in 36m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f3421a98ff14fbd8494600bda0e7de9 : SUCCESS in 1h 43m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/71cfbf30f3d54415a6b6b6d69a01b1c5 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":4},{"id":"6c5e4834b20ef091c2ee59bf50e7b10cab083636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 07:58:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42f03dbc04a1081fbecea808090fe73b769d03f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:55:10.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4536d2651d994ca8a8d1a624e5b10c4f : SUCCESS in 52m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/711c74a5114a432db36340bb3cdd1729 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11e31cb87dc34133983cf99e42db4824 : SUCCESS in 1h 41m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3045842ad7de4565bc48a1ac9f0e8735 : SUCCESS in 1h 42m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef62e6f3662647a9ac6b2fadf1d8b297 : SUCCESS in 1h 33m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4c71a961f264338afe64b38a5b32ab3 : SUCCESS in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a160c4714ca47c8970d530602ab3d9a : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c34261ee38444f50b777aa47464cb4e3 : SUCCESS in 38m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9159a2f9ede34c569f01a3df2b41bcaf : SUCCESS in 36m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56252ba04c4148f8988ce9a917d97eb2 : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a07661d9216f4a03aa641e2da3031124 : SUCCESS in 3m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/710c5b5a832748e7816e494ec8214123 : SUCCESS in 54m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed4b879caced47eea7aaeb44ee18efa9 : SUCCESS in 1h 36m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/35b343bc256445dfaa929b9116318d9c : SUCCESS in 1h 23m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c5c95039a32477188e52ad67ec75c3d : POST_FAILURE in 1h 02m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3ebe398d69ee4767b69a30cc5b01dca3 : SUCCESS in 1h 05m 59s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa02f9dd2bfc47de916ab7c0f76b9f19 : SUCCESS in 1h 48m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c76604c073042fc9669e7898c599ec7 : SUCCESS in 1h 09m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9de481a6ec274ed59bc8e6dd80f8344c : SUCCESS in 1h 18m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/32ea0814dbd942dea1ddb65c91fef8fb : SUCCESS in 37m 12s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/580fb26bc968436c9a2fb0dd03db5568 : SUCCESS in 1h 36m 10s","accounts_in_message":[],"_revision_number":4},{"id":"26bb89e65237bd0b556914e14a536de12e6f222c","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2021-03-18 09:55:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bef518b8142c74e5f44becf22926bf560ca839a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:00:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20c8fd4941e841cb8be66e47b545b5ee : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f4eaf749a994f29be25d67fdbdd0bf1 : SUCCESS in 49m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"87b994586c0d0425ff0253ef0f9b29a09fee5c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:02:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/88de55a67e0e481d83281570d476c896 : SUCCESS in 48m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b97cfbdd6ed8486a914944d90660db30 : SUCCESS in 1h 29m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c5fe583b6e214365aeedc52614ddd47d : SUCCESS in 1h 21m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7216858576c490d81411e218f6dc766 : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5562da06601141bab140ea5f3ac97112 : SUCCESS in 2h 01m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae603defeb4f45a88e6b235a92f5866d : SUCCESS in 33m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c348891a6e27457e9aa0fc049e66e2e1 : SUCCESS in 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb8072e6c384209a84e381b983a534f : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0097eb22fc34fc39858f8e481455694 : SUCCESS in 37m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df835889f7542da9cc7eed54a2645db : SUCCESS in 33m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2c2700cde5942d3a7b563c2562fed91 : SUCCESS in 34m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0dc167fe7f496688757d5bf5d42743 : SUCCESS in 10m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2c3d308fbeb24e2ca862bec321c69565 : SUCCESS in 3m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b83a725d439421ab00306629985b821 : SUCCESS in 50m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cf82083e2334589a6ea92fed0501f0c : SUCCESS in 1h 18m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7f849af9139048f69d38e7805cf76643 : SUCCESS in 1h 30m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3a7a95aa7354a3090e5fc4845798f77 : SUCCESS in 55m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/722ca9cb592d4a64847b5a38ecedf6b7 : SUCCESS in 1h 42m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/84fd5e8fc27c4652a9634b5291bced02 : SUCCESS in 1h 21m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b22070d35a2434fb0d00166a3ac2c26 : SUCCESS in 1h 26m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a6496af672f4931bcb3d3e34df80310 : SUCCESS in 1h 35m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de95ce0725fa4285aa3ed9326ab9859e : SUCCESS in 1h 09m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/45a9f238e31b49e48e234a2a972adddd : SUCCESS in 1h 50m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6ebb7d0f4f4c47b7bf6daab4b3e147b7 : SUCCESS in 33m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1a759a1f2af4d7781cc573d57833b91 : SUCCESS in 1h 07m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5fa9462d40e643678db456b5966ca273 : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":4},{"id":"fc06c3d43f927405e7c94cffb634cce06f8ff276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:06:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43abb0c2bf5eda4aa8d27420cea386b281ef4f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:23:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/289c5de450fd4554bf8bac1d60cf6ab4 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4e3dd2432ee4466a878e7258599c04e4 : SUCCESS in 2h 03m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f72f7ff3499447a09a306a0aa16c6143 : SUCCESS in 1h 56m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71afe357568c4055a5a81e8fc6196615 : SUCCESS in 1h 29m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a136b43dbc0d4ccea1d1ff8b109224fe : SUCCESS in 1h 18m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/355d05c1ea9f4e3ba6a2ba243d9cd7c1 : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f43345f4a6f47a391c888e82f2ed7c3 : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7786dc5b3e4718bf11cb773f911145 : SUCCESS in 32m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f1c2a41710542bc9c70bfa818059bb9 : SUCCESS in 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e8dbd57d2b4d8382c567f126bf44cd : SUCCESS in 10m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/53d1fb4207384ff48ffffd37639429d0 : SUCCESS in 4m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53d3884cf93248adbfe8335e174f33fb : SUCCESS in 55m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/248528944cb9430fbb240ad0cf03db69 : SUCCESS in 1h 24m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7aaeedaebd8a4338a0bf172adc6324cd : SUCCESS in 1h 38m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb44f28bc7fc46ac8d95127fe18afef6 : SUCCESS in 54m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f1dfd713d72432ca40d4c1011408ada : SUCCESS in 1h 07m 24s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae0926ae20884bf499127ad6f3f65dff : SUCCESS in 1h 36m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43c9eba82aa74d99a9b1d9954c45ae26 : SUCCESS in 1h 01m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/862636959ad34711b67d2849425a9365 : SUCCESS in 1h 07m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e92a7dd5baa450b943bf6bb3e8a0ef7 : SUCCESS in 36m 04s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bef91ed2f7c4416b6e42ce144ca4ec7 : SUCCESS in 1h 33m 03s","accounts_in_message":[],"_revision_number":4},{"id":"e3fee96b3275b704cf62227b49dc0ec665775457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:24:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"617c8b8f5ed7ac2f3fcce9865057769e7cd1bf7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:25:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e14eaf172a4e99b0935a770a48b617 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f62890527974e1fea286373ba47b463671360947","revisions":{"77c5bded155d4ee482d595d4d1920af97de21603":{"kind":"REWORK","_number":1,"created":"2021-03-16 12:03:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/780802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/780802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/780802/1"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 10:09:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 12:03:28.000000000","tz":0},"subject":"Fix VLAN driver tests","message":"Fix VLAN driver tests\n\nSince [1], the physical network VLAN ranges parser populates the\nranges for those entries without a defined range, allowing all\nvalid VLAN ranges ([1, 4094]).\n\nSome VLAN driver tests, relying on the previous implementation,\nconsidered that the physical network without a defined VLAN\nrange does not have segments to allocated (those segments\nare created on the fly by\n\"SegmentTypeDriverallocate_fully_specified_segment).\n\nSince [1], all physical network segments are stored in the\n\"ml2_vlan_allocations\" table and set as non allocated.\n\nThis patch also reverts [2]. When the physical networks are defined\nin \"network_vlan_ranges\", there is no distinction between tenant\nand provided networks; the physical network segments are assigned\nby the user. It is possible to create a provider network without\ndefining the segmentation ID, it will be provided by the Neutron\nVLAN driver, if there are free segments for the required physical\nnetwork.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/779515\n[2]1376df7873c2ac77c256ab2fed928de41a2c1d58\n\nCloses-Bug: #1919280\nRelated-Bug: #1918274\nRelated-Bug: #1649750\n\nChange-Id: I191e020ddb97dcf8fb41139d35bfd699e125379b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77c5bded155d4ee482d595d4d1920af97de21603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77c5bded155d4ee482d595d4d1920af97de21603"}]},"branch":"refs/heads/master"},"ffc5e0894cf87c249efd456df91044b327516197":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-16 12:12:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/780802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/780802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/780802/2"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 10:09:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 12:12:06.000000000","tz":0},"subject":"Fix VLAN driver tests","message":"Fix VLAN driver tests\n\nSince [1], the physical network VLAN ranges parser populates the\nranges for those entries without a defined range, allowing all\nvalid VLAN ranges ([1, 4094]).\n\nSome VLAN driver tests, relying on the previous implementation,\nconsidered that the physical network without a defined VLAN\nrange does not have segments to allocated (those segments\nare created on the fly by\n\"SegmentTypeDriverallocate_fully_specified_segment).\n\nSince [1], all physical network segments are stored in the\n\"ml2_vlan_allocations\" table and set as non allocated.\n\nThis patch also reverts [2]. When the physical networks are defined\nin \"network_vlan_ranges\", there is no distinction between tenant\nand provided networks; the physical network segments are assigned\nby the user. It is possible to create a provider network without\ndefining the segmentation ID, it will be provided by the Neutron\nVLAN driver, if there are free segments for the required physical\nnetwork.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/779515\n[2]1376df7873c2ac77c256ab2fed928de41a2c1d58\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/779515\n\nCloses-Bug: #1919280\nRelated-Bug: #1918274\nRelated-Bug: #1649750\n\nChange-Id: I191e020ddb97dcf8fb41139d35bfd699e125379b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffc5e0894cf87c249efd456df91044b327516197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffc5e0894cf87c249efd456df91044b327516197"}]},"branch":"refs/heads/master"},"b546f7c8facb18ee10bddad9147142878c4c707a":{"kind":"REWORK","_number":3,"created":"2021-03-16 13:44:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/780802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/780802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/780802/3"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 10:09:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 13:44:03.000000000","tz":0},"subject":"[WIP] Fix VLAN driver tests","message":"[WIP] Fix VLAN driver tests\n\nTODO: bump neutron-lib\u003d\u003d2.9.2\n\nSince [1], the physical network VLAN ranges parser populates the\nranges for those entries without a defined range, allowing all\nvalid VLAN ranges ([1, 4094]).\n\nSome VLAN driver tests, relying on the previous implementation,\nconsidered that the physical network without a defined VLAN\nrange does not have segments to allocated (those segments\nare created on the fly by\n\"SegmentTypeDriverallocate_fully_specified_segment).\n\nSince [1], all physical network segments are stored in the\n\"ml2_vlan_allocations\" table and set as non allocated.\n\nThis patch also reverts [2]. When the physical networks are defined\nin \"network_vlan_ranges\", there is no distinction between tenant\nand provided networks; the physical network segments are assigned\nby the user. It is possible to create a provider network without\ndefining the segmentation ID, it will be provided by the Neutron\nVLAN driver, if there are free segments for the required physical\nnetwork.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/779515\n[2]1376df7873c2ac77c256ab2fed928de41a2c1d58\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/779515\n\nCloses-Bug: #1919280\nRelated-Bug: #1918274\nRelated-Bug: #1649750\n\nChange-Id: I191e020ddb97dcf8fb41139d35bfd699e125379b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b546f7c8facb18ee10bddad9147142878c4c707a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b546f7c8facb18ee10bddad9147142878c4c707a"}]},"branch":"refs/heads/master"},"f62890527974e1fea286373ba47b463671360947":{"kind":"REWORK","_number":4,"created":"2021-03-16 15:43:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/780802/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/780802/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/780802/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/780802/4"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 10:09:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 15:43:33.000000000","tz":0},"subject":"Fix VLAN driver tests","message":"Fix VLAN driver tests\n\nSince [1], the physical network VLAN ranges parser populates the\nranges for those entries without a defined range, allowing all\nvalid VLAN ranges ([1, 4094]).\n\nSome VLAN driver tests, relying on the previous implementation,\nconsidered that the physical network without a defined VLAN\nrange does not have segments to allocated (those segments\nare created on the fly by\n\"SegmentTypeDriverallocate_fully_specified_segment).\n\nSince [1], all physical network segments are stored in the\n\"ml2_vlan_allocations\" table and set as non allocated.\n\nThis patch also reverts [2]. When the physical networks are defined\nin \"network_vlan_ranges\", there is no distinction between tenant\nand provided networks; the physical network segments are assigned\nby the user. It is possible to create a provider network without\ndefining the segmentation ID, it will be provided by the Neutron\nVLAN driver, if there are free segments for the required physical\nnetwork.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/779515\n[2]1376df7873c2ac77c256ab2fed928de41a2c1d58\n\nCloses-Bug: #1919280\nRelated-Bug: #1918274\nRelated-Bug: #1649750\n\nChange-Id: I191e020ddb97dcf8fb41139d35bfd699e125379b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f62890527974e1fea286373ba47b463671360947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f62890527974e1fea286373ba47b463671360947"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
