)]}'
{"id":"openstack%2Fneutron~780906","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ic9e64aa4ecdc3d56f00c26204ad931b810db7599","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1917811-stable/queens","hashtags":[],"change_id":"Ic9e64aa4ecdc3d56f00c26204ad931b810db7599","subject":"Schedule networks to new segments if needed","status":"MERGED","created":"2021-03-16 15:46:44.000000000","updated":"2021-03-27 17:53:59.000000000","submitted":"2021-03-27 17:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780906-bug/1917811-stable/queens","meta_rev_id":"6ec7665e41903514e37ff6613da15b1b59dd9986","_number":780906,"virtual_id_number":780906,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-27 17:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-27 07:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-19 09:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-27 07:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 19:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 20:58:50.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 09:48:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-27 07:07:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"a333b4fbfe4dbc75c980c89b57cd33b060817c7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 15:46:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0674f96373b339e1b6a7c61fc27d35f73cbc264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 19:12:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/fcda1a069fa6408990570d1f079f808e : SUCCESS in 43m 09s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a256bfc06df445ca97214a4d962d466e : SUCCESS in 59m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/e0de75a1b2ed438e8b106edc3c08d907 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50113ef320f64a3f9b8a7992988feac5 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e7c84e42f7b4fe79e5cb135f041a547 : SUCCESS in 21m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/323788e952534865a0cdd58d916f153a : SUCCESS in 20m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a44eadab72d74783ab1705d27ca6e389 : SUCCESS in 12m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/29f11eb7ed07491b99814d7b3fd048a8 : SUCCESS in 1h 22m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a6562a040eb450cb5730639bb11b491 : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7595bb7a0a574fc8ac26c97fe8e439f8 : SUCCESS in 1h 33m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b4763b17edd4437bb0d6a6677551421a : SUCCESS in 47m 09s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/d0f0ca886e0d4e1298dc145fa06ace29 : SUCCESS in 1h 42m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e239e558dff046c5a7febac5404f7d9e : SUCCESS in 1h 15m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c815b7fa25a14020bee4f2fb3c7c76b0 : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd5f5c99864842b88788651a51e5a5e8 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fa5d55b65ce0518e48338fd2ef26287f438005c4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-16 20:58:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"8f27928f595e5bd02c5f106fb89f934527caae3b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-19 09:48:34.000000000","message":"Patch Set 1: Code-Review+2\n\nConflict change is just in imports list","accounts_in_message":[],"_revision_number":1},{"id":"99a56f426adbb2b4c7df06f495f8ff4f75c3dcdf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-27 07:07:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e15f5184728a08bd22d176695f1c726d6f0825fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 07:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b8b054397fee0c611090de55e1125069d25ebc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 14:13:27.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3d64a767d47e790ebc1c1e09453acadfa47a99a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/1f918bef4b5d4705ac692ef132ce5851 : SUCCESS in 43m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c082fc5780b34cb68e4bc28abca562a8 : SUCCESS in 11m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03565cee7b2e4fa8bdd627391651e0be : SUCCESS in 23m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/daa6e81ff93c45b5973280dfdf6a6295 : SUCCESS in 28m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/850c700978d04936920c8039b957da98 : SUCCESS in 6m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9b74026230e24afbbc0284194b30db49 : SUCCESS in 1h 13m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a719f6f41694d1684a7b16e13da8e70 : SUCCESS in 48m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f054f16747f34ffba8e6ac9155504742 : SUCCESS in 1h 13m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3b31618cb6944e1e86dcd2af0790dab8 : SUCCESS in 38m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/949c0754323d4254bfe10d89d503408c : SUCCESS in 1h 14m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/057089e67b884e25932c14d2d0abe132 : SUCCESS in 1h 12m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c8837dbc49942aa80e0cbdc9a085b99 : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6ec7665e41903514e37ff6613da15b1b59dd9986","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:53:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac78b715c311970a6c055b4bd946d394f5c2f6ad","revisions":{"ac78b715c311970a6c055b4bd946d394f5c2f6ad":{"kind":"REWORK","_number":1,"created":"2021-03-16 15:46:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/06/780906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/780906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/780906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/780906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/780906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/780906/1"}}},"commit":{"parents":[{"commit":"81db4eb473b8a1e166ea674b4ad10771c5a6095e","subject":"Merge \"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81db4eb473b8a1e166ea674b4ad10771c5a6095e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 22:18:16.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-16 15:45:46.000000000","tz":60},"subject":"Schedule networks to new segments if needed","message":"Schedule networks to new segments if needed\n\nIn case when subnet in segment X was created before hosts from\nthat segment was mapped in the Neutron DB, network wasn\u0027t scheduled\nto the DHCP agents in that new segment.\nNow DHCP scheduler is triggered in such case.\n\nConflicts:\n    neutron/db/agentschedulers_db.py\n\nCloses-bug: #1917811\nChange-Id: Ic9e64aa4ecdc3d56f00c26204ad931b810db7599\n(cherry picked from commit 5c931f29132c3c5c01127f089f7d5909d2a19e4d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac78b715c311970a6c055b4bd946d394f5c2f6ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac78b715c311970a6c055b4bd946d394f5c2f6ad"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
