)]}'
{"id":"openstack%2Fneutron~781639","triplet_id":"openstack%2Fneutron~master~Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831","subject":"Change new policy rules for availability zones API","status":"MERGED","created":"2021-03-19 12:07:07.000000000","updated":"2021-03-25 13:56:13.000000000","submitted":"2021-03-25 13:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781639-secure-rbac","meta_rev_id":"ddea76c011527ef917801eda55dc744b475a522f","_number":781639,"virtual_id_number":781639,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-25 09:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 13:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-25 13:56:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-25 13:56:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-24 23:59:56.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-03-24 23:59:56.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-25 13:56:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 12:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 14:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 16:38:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 14:15:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-22 15:48:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-24 21:26:13.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 23:59:56.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"}],"messages":[{"id":"4ae43edae95ea8b9040646da8153c61648305ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-19 12:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a7603a3c77d7b3332169a3735454dd7656f0db7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 12:50:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cc9a2dd27124b61b619dd0b48455344 : SUCCESS in 38m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a735387094c455094bce1efddc58066 : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebe3947f8864ca447802fcfdbf6cc08bde21012c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a59909bd5d0f47f6ba474d4495b2c380 : SUCCESS in 57m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/141c80cc853e49aba5350b0907a7d9ed : SUCCESS in 1h 52m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/32377492c2af44b88434768738234d28 : SUCCESS in 1h 50m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b78e944182c47a4bda6e2419ed64faf : SUCCESS in 2h 01m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e8d812638bae4335a2aaaa578f6eb847 : SUCCESS in 2h 07m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dad0178e0b04a7487da355dddfeb956 : SUCCESS in 41m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf8d72187ba4ccb949bbfb334ac9c69 : SUCCESS in 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62ce4ad7739468c965fd6cfed5c7790 : SUCCESS in 10m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3185e1eabd0d449da2cb060ac8a8ada9 : SUCCESS in 20m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7855d09fd99846a799135a4c7ef0a449 : SUCCESS in 31m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4992a969cdb1425091a1c79adf504559 : SUCCESS in 21m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a43737a87b40908b09bdf3dc833231 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bf39ee5ab33440c8cc5a131f5bdeb86 : SUCCESS in 46m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ef6b1339934404ca61d54a0b94bd3e0 : SUCCESS in 1h 28m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5028418ff1994e2fab4646dfa1f0b977 : SUCCESS in 1h 49m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/645256e530de406aa7b40dcbafd9855f : SUCCESS in 59m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7f1a3c13a515491cbc5a4162641412f2 : SUCCESS in 1h 02m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/31128bd5bda3441e96a1f8ffadd0f540 : SUCCESS in 1h 42m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08a8443d1f754b8e9c8bd913e8e61001 : FAILURE in 1h 24m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81fb3f81358c44d69a393612d39c3a04 : SUCCESS in 1h 34m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b2f9a3f05bd4405b423ad1535a5ffb3 : SUCCESS in 1h 02m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ed4ec4de33bd4c2aacdeec9e3d7d9be5 : SUCCESS in 1h 05m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9e1c8e7d89674da19b0ca322603e3e0d : SUCCESS in 34m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/36f96320ac614721938735e635a44c47 : SUCCESS in 1h 24m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f2b6a78bbea44e209a81cc56991f27b4 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"16ac08c1acde5720e1f2292e31fc97e5a6d2f37a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-19 16:38:06.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/39/781639/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"41492438563d88008096566f00fed648c00019d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-22 10:29:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"813c760a0966b52e96aa95f9fa225da658e299fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-22 10:30:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"134a9ff3345f1ff8ddc086474a2f87c3683d7807","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:09:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef0c6520e461e9d15bcb4e373799a8e64b97e41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:43:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9005b90ae28e4e388097aadee35bdd69 : SUCCESS in 47m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a4e2de958e024e98bc4382498cb56219 : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/05b8ffdfcdc44f2399da715f71f58227 : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4480f8f247c4a0fa6c996dee1793a23 : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fce7807dd8824b06b0da935f9c21f260 : SUCCESS in 1h 36m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f89821b18724e3ca25910544967dd3d : SUCCESS in 39m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fc52e5431594ee79a3b63b9cffa1669 : SUCCESS in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d1236b9d374ce1a302e94942b4aaaa : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d716572f4c454ac0b12b9df3945515fb : SUCCESS in 25m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1708e45a7fc494ebad490a198722df0 : SUCCESS in 24m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd898c95308a4f1bb1aec0787a353715 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa4a9d18e224bf28f3e668e320a0a38 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4fa6e13997d4ff59cd587499a1bf8f5 : SUCCESS in 49m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0d4679a916b4a188375341ce538dd52 : SUCCESS in 1h 31m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b24949b7cf0146bc8c99cd14bceea9b3 : SUCCESS in 1h 47m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38056ba61e0f4ed4953989a2ff49cf81 : SUCCESS in 1h 12m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e122cf590bee43de931c97bbd76eb757 : SUCCESS in 1h 06m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9192a4b12db14240a77aa64b3d9a078e : SUCCESS in 1h 33m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a29245e51ef4a2a84f3a26f53316876 : FAILURE in 1h 18m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4fe0876271840389376fbaf95e2bf74 : SUCCESS in 1h 40m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/742f50bb87a54d27a2e8e79b5c8cfad5 : SUCCESS in 59m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/61714e2ff93d464e90d325958362da4b : SUCCESS in 1h 03m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/43d0e45f5e274aa380834193e7a093b5 : SUCCESS in 28m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efc5302664294adeacb17f2172616974 : SUCCESS in 1h 43m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3435a079ffbf48e4bde13ea8f5e48d26 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b162ee1bab2b85216d1a9ee3d205a0a9b9eb6bbb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-22 12:58:34.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"cc478cadbbdd970bbee508f44ddd723edc3de870","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 13:15:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/39/781639/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e375e74b425b770285925df1f385cf2e4c8607c1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-22 14:15:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"342cd2f4a3389767bc8440a7a12da55fe455969d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-22 15:48:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d6dfa92f9b9dba8200b4d2352e7ac072ab78a9eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 16:26:26.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/39/781639/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2a6d93d170addcfbb43ecb28254f2e9e9de121db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-24 21:03:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"92e0a6c7b82e23a168a6820a1c5a3bcd0d43aef2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 21:26:13.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"e8a7c819f60d7d84ac4df40b10ba89456fd249a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 21:32:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9f65962526dda2fbe81134b55bd3aa90e90afa7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 23:24:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d1d9d60c6d244ff4aba43845019b9543 : SUCCESS in 50m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/96ab13e994cf4a15b5963e8f1454b5f3 : SUCCESS in 1h 39m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f47f6a11dc08465caffd7d73a00eb39d : SUCCESS in 1h 32m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf62170177d44919bfd27e6642faaa90 : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1ab51253ce3841feaa0ccf438880845d : SUCCESS in 1h 59m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/168af66c57f248699616b38129dceae9 : SUCCESS in 37m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5faf78782cd2485a8e79338a5f2c9670 : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e131c997d648d4be546a386f2dcc31 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2956fd2ba81c4cbd8a503d51e4a17765 : SUCCESS in 34m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bb00b84b63e483ea23674699dcdc7e1 : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a617b5a19d804285a134261759b94ec5 : SUCCESS in 32m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99be197ff3e24ad991b22a1775452e1d : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6934059963644dbc8bcae32b6fad8b8a : TIMED_OUT in 2h 15m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0419c5c0b0c1454a834849eb5b30638a : SUCCESS in 1h 21m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a2d8d99116fa465f85c6bd570db7a20f : SUCCESS in 1h 31m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19e57faea59e4303bb73d86ad3f34a6e : SUCCESS in 1h 01m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3c5265fef77d491d89846f5f46305255 : SUCCESS in 1h 12m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/830e6772bdd54f5889a3f83413ab5857 : SUCCESS in 1h 19m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0edd29e17fbe4e99b076eb5cd163a2f1 : FAILURE in 1h 17m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f0c9f18bb884e6da0c4bd7048df8772 : SUCCESS in 1h 35m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afadf58d89da4fc18efa792e7900abb2 : SUCCESS in 1h 12m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3f21d5ebfcd242b588907cf654f2bd6a : SUCCESS in 1h 18m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/21fcce59d5304e2484162313903da78b : SUCCESS in 38m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc0553b78df242f0bb9b3782b8999af6 : SUCCESS in 1h 31m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8c2a594bc283425f813c02e4ea37f9c1 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":4},{"id":"cd3a65c8686e3da28fd1bfcfefc1be8bf0b9df17","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2021-03-24 23:59:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4348e2fd5a86293984e4979ecf09bb4bbccb4e28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:37:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/580b169cfd0b43988e6256b47d1069d8 : SUCCESS in 35m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12deae6b18aa41bf92d40e6146a75e96 : SUCCESS in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7556e4a7c85e94986b0686d63857a00af63e362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:44:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b182df22f38496980511f356a9f5eec : SUCCESS in 38m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9fb33a0d03df446ebc9b757e02727f7d : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a1eabc88982465fa3407390e51c0101 : SUCCESS in 1h 08m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f33cc0fe4f1d4cbca6d41954123eea04 : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9eb6568d7ff343e4ab6eb46cea391526 : SUCCESS in 1h 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b3642e62c54410a2b6dd4217b8da95 : SUCCESS in 46m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4be841727de43238877eca4c0cf0a46 : SUCCESS in 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876ff86927b84094adf27bd1132c8096 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca4edb1036cf4242a9a0bab30758ca41 : SUCCESS in 29m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e641c1887854d3c84edec34108c0309 : SUCCESS in 23m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7c1645ea2fe412babdb27134909e5dd : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987b8d355562456f977a5d28b89d99fd : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa9ef9ac1ef64e0ead0392083dec59f8 : SUCCESS in 49m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9910d23d07084c58828b560d768c48e8 : SUCCESS in 1h 10m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/35b3e14fa41b46ac8778793da3afec54 : POST_FAILURE in 1h 27m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/156c4288fb304ac4b7436207c127189e : SUCCESS in 1h 03m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0047b4c72bfa4109a7ed7c6d5d857690 : SUCCESS in 48m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/089ebb325be9482cb40111a795a197af : SUCCESS in 1h 21m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca84f1e17d9f4e4db5c2bcefd8fc712c : FAILURE in 1h 18m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03f733e6d8a24d94b28d185b7c8c3942 : SUCCESS in 1h 34m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36fd2fb7175b4910bc23fe144e3ec96e : SUCCESS in 54m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/60af9dc5ebe749a18fcf9ac75b966df4 : SUCCESS in 1h 09m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bd00bb5cdedc4b3280b17fb2e74817ec : SUCCESS in 41m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aafdfc04919041e1b761aec7e24c9d48 : SUCCESS in 1h 08m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1c2edc6bbcaf41e8a73b1ddab3b208bf : SUCCESS in 1h 38m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a36a25a21244e7493adaac7c6f751d849f5eff90","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-25 07:25:41.000000000","message":"Patch Set 4:\n\nrecheck\nneutron-rally-task POST-FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8d5416b7dbb2d61c969de69367c8a2fe36700a64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:20:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3a356bf705941afb19da567a365f57f : SUCCESS in 47m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76dd44f5eca5497a8890bff0addbfca8 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07cabf0b381fce2c24d21b2a910ac7ebce779abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 09:17:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fdc972e58c7047a1ba97fac9ae1dd2e7 : SUCCESS in 55m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/43edb453b75043dbad5378cff4503cfe : SUCCESS in 1h 15m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5e24539e91e0447aa7f9864ae9f8794c : SUCCESS in 1h 11m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09ebce561f904035b1e452eaf24d9f33 : SUCCESS in 1h 14m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/15ab6f16fa024a1381e1b85bc8aa9b7d : SUCCESS in 1h 47m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d635ecee8e4ae8b6e21d6910a4f355 : SUCCESS in 48m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bb26d2838274b5eab113889f85664fe : SUCCESS in 34m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f04acc634a4b1f932325042af86344 : SUCCESS in 19m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e49bb3e7f474d4b8fbcdfd79ce6da05 : SUCCESS in 47m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf82d38bd4c94eb4a2a853f4c463c8e1 : SUCCESS in 33m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d48a9d13a7c4e6e96ebb6fa86855971 : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5191fec6fb4e7a98001ee4d9907fc6 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94971538e1dc476580590cdf02a213d5 : SUCCESS in 59m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c93d1b0db9a439a834a8930a8cef269 : SUCCESS in 1h 29m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c7dac93a1de642ba809ea8a4ef90dd5b : SUCCESS in 1h 41m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8aca79b00b6f4756b5241b733b66e14a : SUCCESS in 1h 02m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5523c63547ee43659a5612ecd21a6bcd : SUCCESS in 59m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/13523376673b44a4b24726f9ff1d981c : SUCCESS in 1h 43m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f406c44ad0241f1b97b410ad841d232 : FAILURE in 1h 23m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fdc41863d5048d58d04e91333238e6a : SUCCESS in 1h 45m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcad534ba2d046679e8d6c78c9693c92 : SUCCESS in 1h 16m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a80f4253e8bc4ad8b58c3bd1ade45b61 : SUCCESS in 1h 02m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9de7f211a6fe431394acfa3fe61fa834 : SUCCESS in 36m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/131834fd3c244320a7c6cd194f1051d3 : SUCCESS in 1h 21m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/334c34ee69ac41d6a91ccfeb17a727cd : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2b2b610fc5678572d0b697c597356d8dd96a5fb2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-25 09:34:59.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/39/781639/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"05d4e331fcdf0b09814167193de57a850d20522c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:26:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6bd7a4a8106030ea7b3ff50da918a82f3353dfde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:43:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65e896b269564d1a9f9423b505df6c49 : SUCCESS in 47m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d7f49fd1da774bd4abe38e19a428e023 : SUCCESS in 1h 48m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9de40bb41d9744eeb3e9bd845d698e52 : SUCCESS in 1h 43m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/318194a22f3448e8b93962d861297cd4 : SUCCESS in 1h 53m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/064afd9b1b0d4766a9f9d503136c01a4 : SUCCESS in 1h 21m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4e9e8002f7348dcb86a9d0f3f85e426 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95bc279e772649439da2b08c5c9e9ba0 : SUCCESS in 15m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/406cdf9c657e485da03bcae67b80fd87 : SUCCESS in 31m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7f4efbad684a828020dcd32a69f994 : SUCCESS in 28m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ff25ee1e854f31a72da6955e5383f0 : SUCCESS in 12m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f869e8b7c4464dc99e28b82f90930028 : SUCCESS in 47m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b67cb49709b64d3fb943e1840b943b08 : SUCCESS in 1h 19m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/41ed91084e5649b48cd69e0f94ffe1e1 : SUCCESS in 1h 24m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24da7e95ee1b4b8292dbd7fcc6ea6371 : SUCCESS in 1h 06m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82aa16ace4d44d0e8832043fc3dce678 : SUCCESS in 1h 36m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb09015c128a477fbc9d152c28ac9e61 : SUCCESS in 1h 03m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/14ef302178a040d4aa1ee6d581fc2220 : SUCCESS in 1h 08m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4889466f9b7d42bfa32ee29a7013ade5 : SUCCESS in 38m 00s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8584de13be60460789db3bd1c9ce0498 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5b5f2d78711a22d522fd74b04adb125d1a7e9ee3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:43:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9f1d36241605bad5a94e477b8d1188771449a88f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:56:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbb0cc899394596be74414da4e36478 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e45de322a1a9533e6c1d7de196a28beb74089832","revisions":{"201c915db1959e6c77e5439b3a29d1d89257ca69":{"kind":"REWORK","_number":1,"created":"2021-03-19 12:07:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/781639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/781639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/781639/1"}}},"commit":{"parents":[{"commit":"3eec75e9db90bdf9a1560912c272cf97f4cadcba","subject":"Base unit test framework to test new API policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3eec75e9db90bdf9a1560912c272cf97f4cadcba"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-19 12:05:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-19 12:05:04.000000000","tz":60},"subject":"Change new policy rules for availability zones API","message":"Change new policy rules for availability zones API\n\nIt was previously set that both SYSTEM and PROJECT READER could get\naccess to it. But this resource don\u0027t have project_id so PROJECT_READER\nshould be removed from that rule and only SYSTEM scope users should be\nable to get it.\n\nThis patch also adds UT for policies for that API.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/201c915db1959e6c77e5439b3a29d1d89257ca69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/201c915db1959e6c77e5439b3a29d1d89257ca69"}]},"branch":"refs/heads/master"},"549ccc63d60d3a70d5593809a2dce379f723af10":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-22 10:29:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/781639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/781639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/781639/2"}}},"commit":{"parents":[{"commit":"e34777898fdd609dcf3beb68ca93d3f280f05caa","subject":"Base unit test framework to test new API policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e34777898fdd609dcf3beb68ca93d3f280f05caa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-19 12:05:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-22 10:29:14.000000000","tz":0},"subject":"Change new policy rules for availability zones API","message":"Change new policy rules for availability zones API\n\nIt was previously set that both SYSTEM and PROJECT READER could get\naccess to it. But this resource don\u0027t have project_id so PROJECT_READER\nshould be removed from that rule and only SYSTEM scope users should be\nable to get it.\n\nThis patch also adds UT for policies for that API.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/549ccc63d60d3a70d5593809a2dce379f723af10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/549ccc63d60d3a70d5593809a2dce379f723af10"}]},"branch":"refs/heads/master","description":"Rebase"},"d0fb85d157fc95b545ee721d0a0b04784b24ae58":{"kind":"REWORK","_number":3,"created":"2021-03-22 10:30:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/781639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/781639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/781639/3"}}},"commit":{"parents":[{"commit":"e34777898fdd609dcf3beb68ca93d3f280f05caa","subject":"Base unit test framework to test new API policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e34777898fdd609dcf3beb68ca93d3f280f05caa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-19 12:05:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-22 10:30:17.000000000","tz":0},"subject":"Change new policy rules for availability zones API","message":"Change new policy rules for availability zones API\n\nIt was previously set that both SYSTEM and PROJECT READER could get\naccess to it. But this resource don\u0027t have project_id so PROJECT_READER\nshould be removed from that rule and only SYSTEM scope users should be\nable to get it.\n\nThis patch also adds UT for policies for that API.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0fb85d157fc95b545ee721d0a0b04784b24ae58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0fb85d157fc95b545ee721d0a0b04784b24ae58"}]},"branch":"refs/heads/master"},"e45de322a1a9533e6c1d7de196a28beb74089832":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-24 21:03:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/781639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/781639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/781639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/781639/4"}}},"commit":{"parents":[{"commit":"26072dc65d177e10214ab4bb1fe38ceb007dfeeb","subject":"Base unit test framework to test new API policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26072dc65d177e10214ab4bb1fe38ceb007dfeeb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-19 12:05:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-24 21:03:21.000000000","tz":0},"subject":"Change new policy rules for availability zones API","message":"Change new policy rules for availability zones API\n\nIt was previously set that both SYSTEM and PROJECT READER could get\naccess to it. But this resource don\u0027t have project_id so PROJECT_READER\nshould be removed from that rule and only SYSTEM scope users should be\nable to get it.\n\nThis patch also adds UT for policies for that API.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: Idfaf36c3dd003c9c5c1ce3f9ee889ba54643e831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e45de322a1a9533e6c1d7de196a28beb74089832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e45de322a1a9533e6c1d7de196a28beb74089832"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
