)]}'
{"id":"openstack%2Fneutron~782008","triplet_id":"openstack%2Fneutron~master~I240a78879db757592df138a53b2c22d7f5a9ae13","project":"openstack/neutron","branch":"master","topic":"bug/1920700","hashtags":[],"change_id":"I240a78879db757592df138a53b2c22d7f5a9ae13","subject":"Call install_ingress_direct_goto_flows() when ovs restarts","status":"MERGED","created":"2021-03-22 02:20:25.000000000","updated":"2021-03-25 21:56:34.000000000","submitted":"2021-03-25 21:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782008-bug/1920700","meta_rev_id":"9136c2c102efd8dc0ddf6ef03c3700c83a9397ab","_number":782008,"virtual_id_number":782008,"owner":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-22 10:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 21:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-23 13:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-24 08:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-24 11:41:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-23 13:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 02:24:57.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 03:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 03:42:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 04:12:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 05:50:31.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-23 13:58:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-24 01:38:56.000000000","updated_by":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-03-24 08:16:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-24 08:30:22.000000000","updated_by":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"reviewer":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"state":"CC"}],"messages":[{"id":"4a9251826c8b1d9aaf5f453f9ef5eefb6a9266cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-22 02:20:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38d50845f1c8dddd09a7c75a503f771c62366fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-22 02:23:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fedebdfd5a64c8b152dcef8370aada186dd29c48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 02:24:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/08/782008/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/08/782008/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"3527b7e893281f2633a594ae9c9373743e157aa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 03:10:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f02647552604fca99b70a8116c07f4a : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e27c20d16cf5458dae96f097e9d72f80 : SUCCESS in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0556632229d3395e7dfc372309c84d571b8a7afe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 03:28:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/08/782008/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/08/782008/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"47c638043f7c8566887491a927d4866c218a04d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 03:42:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/08/782008/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"25641f9183617c7606fe7e565fd7e951115ba6b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 03:53:33.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/08/782008/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"91c2e6c33505e4a5082b3b71e296a0977ca00949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 04:12:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf6b9b9358c64d338bf017be1849fe01 : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f76d8f9b86a473ca2933cffa31dda72 : SUCCESS in 1h 05m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c39f15f782da4015a1483ba707238c58 : SUCCESS in 59m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c27bf6489744d49feb0d5ac8b784a6 : SUCCESS in 26m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b89ba2b4629f411eb9748a63373f858f : SUCCESS in 23m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e1a59d7f7f45b398e462d2a2ec2870 : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f50ead340ccc4157a266c4075c83d400 : SUCCESS in 23m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/955a8a06005c4c44a6574d733a11df35 : SUCCESS in 33m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65f39e3c01be46d19c9c10035460641c : SUCCESS in 35m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473664a0a7154ea89d4ad3d767db4091 : SUCCESS in 8m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7b1d02c2a414e01a5940a603ae1ba96 : SUCCESS in 52m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96640251c4d443688a4c8dadec5c9a64 : SUCCESS in 1h 25m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2978c4d0fa184c5db8685d3b993df340 : SUCCESS in 1h 36m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93f38397b2024bf989f39e1bebab6e4c : SUCCESS in 1h 17m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/355e57023dea4a65944555db53e88ef3 : SUCCESS in 46m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/038a5e14f09d490881fa389725478ab4 : SUCCESS in 1h 06m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ee153f48b0e4b9e99e34247e39b3f5a : FAILURE in 1h 28m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d29ac27c75ca4171b6e5772054ce2c18 : SUCCESS in 1h 46m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ba3edaca35745828407fc9a8d17e4b9 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3f802279e16f16b3a8bc921ecba304640eccdf44","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-22 05:50:31.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease update test_ovs_restart (neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/test_ovs_neutron_agent.py) unit test as well","accounts_in_message":[],"_revision_number":2},{"id":"ad1ebf827a1e803401a363c674d481efdfc9b50c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-22 09:26:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ac775e48ef4c7f003b8fbf40e1bbf246e5473565","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:27:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f877ec80069411481c4bab53333ac5a : SUCCESS in 47m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a85960a0033e4c01826b32c2516ff11c : SUCCESS in 50m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"28f60290784744b0ef61fd5f9299d078352b8a1f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 10:55:55.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/08/782008/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8025cb0ac3eb145f5a2ae743cfad7f4195b01c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 11:15:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64f42c250ece49f299ebba6d26b1adfc : SUCCESS in 46m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/973ad4c390f84cc4bd22c6c586db43b6 : SUCCESS in 1h 25m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce361fa9f43742aa881e69c7c460906b : SUCCESS in 1h 18m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ff85daadef4e2492ad35fdc6adb867 : SUCCESS in 31m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94cfef7f1617420facb8a7b9fe2a5bee : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf20487e7910481e8a382e6c1ea122cc : SUCCESS in 12m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10b131f7f9574af09e8590331e5fc05b : SUCCESS in 34m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f1763d56feb4d0a8b3a6ac67417831e : SUCCESS in 34m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d2c936d5ad04473b3265d1444785f13 : SUCCESS in 33m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7def95045b7d40a38c5bf1dcfde79c91 : SUCCESS in 12m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd20c99bea414ff5bde28a41aceff79e : SUCCESS in 55m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0943ed4a5ca74f2595ec17c7dcbdd630 : SUCCESS in 1h 23m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bad89145d78e4f988335571ebe0d34c2 : SUCCESS in 1h 27m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ec128d1be474cba8daafd2fa35c444a : SUCCESS in 52m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/506f0efe352348dea6c1b9471dc2c8e8 : FAILURE in 1h 01m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/28984be5d3e64af1ba4750c21998c60b : SUCCESS in 1h 10m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aaae8241ff0c47cf8a909d5034f19f51 : FAILURE in 1h 02m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/114fa5c67333409781c1ad0f0eeba338 : SUCCESS in 1h 39m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f042f3c7e83488d8aec0cb6565b6905 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b464036d69bb3ab63d8554bd079c62c07625a43d","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-23 06:25:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b015438879f1792cf43d08a0ae0c898564590e07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 07:21:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a83dd6d85fb4a0091b1ca07c2e5fcd3 : SUCCESS in 42m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18c369171b144739ae81fc584e1b876b : SUCCESS in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5a818a619dcde28e3408676c09ff771baab4844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 08:19:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b119e7a22a984b4d915173d2bd2cfac3 : SUCCESS in 44m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/05ce69024a914dd092a88d91650983ac : SUCCESS in 1h 49m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec58a8843e7b41aaaf07a64019662963 : SUCCESS in 1h 18m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c42a9fec7a9404393f5710389fc91c8 : SUCCESS in 50m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fa6183ecc1a41748665d561c62d7f7f : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91d1cbf58d24daab68cf6c30dc14560 : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbfba73371054af9bb0e5ff0b15fe32c : SUCCESS in 20m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b730d0a5a30f4100b7930a86f91e5843 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36b64e2015a945e9ab0e9f2297503b77 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6034164160400c9a37a3e3f5314188 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1480d61dd5894d8e92c40744f5109ab6 : SUCCESS in 49m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6be4a8327a4f447a84fda55e326e9b09 : SUCCESS in 1h 26m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/56b80a46f4f9495f9fb40d6a40550beb : SUCCESS in 1h 23m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19df8a8a10464e45b850f65a176720d2 : SUCCESS in 1h 05m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ece6880019674924be67d81819bb2e52 : FAILURE in 56m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/532dbe0d2219436fb8f97c8515a61b5d : SUCCESS in 1h 06m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c702820496d943c8a5bf76cd94980405 : FAILURE in 1h 36m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef4cfb5c431d464687e5cd2210b0eb75 : SUCCESS in 1h 42m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ae0487fb7694c5cbdd72a83d6f47be0 : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":3},{"id":"195ce06d61a47bceb1ffdd1ad7e70b83b3010294","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-23 13:58:52.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1499dab3fbd194a928d02a6e01fca8621f4e9a67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 14:00:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf38e8c08580ead891010dc3c8654c9b8742376d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 16:20:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1bf59ef0b7fa46e1b27f71e08c54e81a : SUCCESS in 47m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a8f928699ee465fa1fc37ac27abc749 : SUCCESS in 1h 07m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8bd6806083164e14b2aff34796ef4e00 : SUCCESS in 2h 18m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d2de8d92844e4bb864f476f01a4a71 : SUCCESS in 42m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7858ff4b8514e2dbd9493cd7e53ff77 : SUCCESS in 34m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a86f45b3d5b464f8805602156e90143 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac74cf159ab47c8964fd31a88998bf2 : SUCCESS in 32m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54f6c6af6f7454f85fc8136c56d5e7a : SUCCESS in 30m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f606d6a00684a05b9b309a0cac88004 : SUCCESS in 34m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbae8bf27334c5baff37d8459733725 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20b0d722b78f421bad4ef1bd868a8472 : TIMED_OUT in 2h 19m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6dd493df1d047b8bd7b50e9d9662eca : SUCCESS in 1h 22m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8336562986ae48a8b124e91151b5f754 : SUCCESS in 1h 01m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/15d017d8796f4e4a9713594cbd16db4d : SUCCESS in 1h 04m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/436e43d9d5424640b674fcd1a4257f30 : SUCCESS in 1h 03m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96a73e32cf81467abe6963999a70d2f2 : SUCCESS in 1h 17m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e0c72fea7374c1a906496b4d44c95d0 : SUCCESS in 1h 45m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32563a9e6ebb448d92a19c8609248664 : SUCCESS in 1h 14m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7776e5c3a71d4dc4969f305dae434361 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3c7dc9c6e911c046e4b8d03a09c439e1bf421f67","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-24 06:32:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4f646e4da96591a04cef7da456420a21a13f5dee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 07:13:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60231a20b8954df383cd7a1186167260 : SUCCESS in 37m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a94d0bcf1394b199b81ee02ed95471e : SUCCESS in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3a894f26f1ecd677d6be07066f9912077d3febf3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-24 08:16:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7b86d3c7a0bf10d719b98d3ce017c7adb70c2c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 08:28:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/03e336f8453b490f81ef03cacf5cbc59 : SUCCESS in 52m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c91008db0a34bae9db1ddaac6fe247b : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0e8772179e34bf19d5b83b245a0ca65 : SUCCESS in 1h 34m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64397041b1784d22accf546777bafe41 : TIMED_OUT in 1h 21m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc108f8906d4b4190d05d057b36c940 : SUCCESS in 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f47276a936b449bbefb3e30327e9557 : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54f8abdc60f948c7984dcec09849d234 : SUCCESS in 33m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59302b5b23124da6b978c45b1fbd5cfc : SUCCESS in 29m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f81ee8cf2b044c40bac7359f1edc0d0a : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a65c6094c874fe0b82ead0773284892 : SUCCESS in 13m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/517e3f4388164d64b40537b0cedd3d46 : SUCCESS in 48m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e19d90ea4a86433eb8dd7fd69234a270 : SUCCESS in 1h 20m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8817a89b79b040ee8f2ba04641ee825f : SUCCESS in 1h 17m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/252468fb647b4a3f88420529240e51fd : SUCCESS in 1h 00m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21a40393cf924682bdf0dd661d00354f : SUCCESS in 1h 14m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4c2be8d592e4389a8bd922bba84744a : SUCCESS in 1h 26m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98016fd6f4934b5780665bf908ec77e8 : FAILURE in 1h 14m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0def2899b34742729e7b0d595bfa59cc : SUCCESS in 1h 38m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d3a3b09434340b4a3f038c79a3ae965 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4e5786c1dbe7e4e39ee60b45383df4a10d048a96","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2021-03-24 08:30:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"07f563cba64e655e4dd424599b599d8e1167f77a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 09:24:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f80acbfe215447fc8263329f8a08f4be : SUCCESS in 34m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dc5d06f16314fd5a4325f4b7a7928ea : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a311bdaae512abda88fdb46e9141f453e154a900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 10:21:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f9a4a556bb9f4414ba15740146411ff7 : SUCCESS in 35m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e1fe7e1e77c54e72b3e947893dab9a3c : SUCCESS in 1h 39m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3311367ab42341ebb65c81be1b8d386d : SUCCESS in 1h 37m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ca3a96fde894ef2ae10478eb90e87cd : SUCCESS in 46m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00d684474bc34bab951653b32b06571d : SUCCESS in 41m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc9a464248e42c9823f452c63c0109a : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19bbd2ac9a0243648ee876dda9800652 : SUCCESS in 37m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7867eb6c6c84529b2b35dd446c7bf4d : SUCCESS in 28m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caa1fe9f4068444592e1f3963080772c : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc530b33ff6348bf8e02d7c007e67b76 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7501528ed463481a99a42c4e8b5c0bd2 : SUCCESS in 53m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/945d833d62b1488ab1bf3eabc525f7fa : SUCCESS in 1h 32m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6fecccb3b48348bd80d7ea908d892fdd : SUCCESS in 1h 30m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f35097d9ba74782a94b09218d0e3a80 : SUCCESS in 1h 02m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/08f6fe9059bf43ba8ec73d1c85d48891 : FAILURE in 1h 02m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bfb70bd5bec440f69c7071ad40131130 : SUCCESS in 1h 28m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ac6707b870a345e6b731ab4b50d58ea2 : FAILURE in 1h 28m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5129f3b7d8548a880c6e0fd16a8dcb7 : SUCCESS in 1h 45m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37297ca65dad451292fe0271f8aea510 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f825ea10001ab466f8611b5f3976b2f612c99d13","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-24 11:41:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9128f2d096c5a237b5e0ab634436d78eacb47b9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-24 11:41:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"efe90fb49e61467032cfcef01b71717a83b4b361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:25:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a13880e4a372413ab7915639d8ab5afb : SUCCESS in 36m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/309c3cac49484a7fa1770d46abe29e5e : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40a97d4c8dc056bca88fefb79c8a780f2d2dcad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 13:27:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c84fb31622f40f08ac156b48116ebb1 : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9453c5c4c4644de94cacdc43c57e79b : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2cf95f2f18da4779940b27b5a17e6a57 : SUCCESS in 1h 42m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/168bf081edf1423e8f969a604319dee5 : SUCCESS in 40m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d216204602546118bacde0d2ca65505 : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9cdcdee03f5475590fd7b55d6cc8795 : SUCCESS in 16m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0be4c988748a49048caaf6992aa38e21 : SUCCESS in 32m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e65e509bd24a97a2c0fe290f4da5aa : SUCCESS in 33m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34c0b2f7b45d4f75b345018543ad2c22 : SUCCESS in 33m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd22ee98503446b4bc3714b18e8f31af : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d7e1b7370134e00ab62077e0f35a870 : SUCCESS in 52m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d497a6d82dc9460aaaae76cac87235db : SUCCESS in 1h 19m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/02c1b0e0a230437f85f8062ea36f7402 : SUCCESS in 1h 27m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/316c8589bf4847179640c21cd36012ac : SUCCESS in 53m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9972b9f0a6ec44e2888246e84adfb867 : FAILURE in 50m 08s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d9e74ea84f584658a696ac403e479304 : SUCCESS in 1h 09m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ccb19b31041f41d4b06dbad30092d220 : SUCCESS in 1h 33m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d938e7791489493fbcc35dc5704e199a : SUCCESS in 1h 23m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14e2c3d68f524d0db56b17b726f39ad9 : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c38a9caf4a03969b8f09510d9939835ca3f8ca9e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-24 16:27:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"26fbf4605200e11e58d820f3bd0bb44fafafe4bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:22:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d83594b64854619bb93139cfa6a026c : SUCCESS in 46m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/303412160e7040c8aec3dcee7ed838f7 : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"caf6607c1052580b6104eac1b8d40972f165d42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:22:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a934196170cd4c3b8f8cd771e7840c4d : SUCCESS in 48m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/323aa6a488434ae693e5ad8c0b59dfdd : SUCCESS in 2h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/32f6ca649fda4ea8809d518ac6f95b0e : SUCCESS in 1h 35m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/229fb2e50aaa47fcb24f7e534bef1a44 : SUCCESS in 55m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5820342f084845aca219a4f4853066fa : SUCCESS in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8499ba25f8db4187bda88b2702885ad8 : SUCCESS in 14m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/072c53a5a5c74de38484bcaf37a76181 : SUCCESS in 31m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffbff994b9354041b7f64555edd43183 : SUCCESS in 33m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae8aec1e8b74ad2943325429d7f8b29 : SUCCESS in 34m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d79860d61a45c9b749dcce9a8da0cc : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/650dab41a7b6497c9c638f6e2e506460 : SUCCESS in 52m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bed7a5980b7d47d28038f6d013f36460 : SUCCESS in 1h 21m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/10e108855a2a4412bfbc7d731cde96c3 : SUCCESS in 1h 35m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdeb94ddfb314b69bf3b1dedcb24915b : SUCCESS in 58m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ea730ff80b34d3f8e97b080d54b2563 : SUCCESS in 1h 04m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ea921d86df674b8da520de9a07f44c89 : SUCCESS in 1h 18m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/79b72a2388aa4fa284325e86551b87c3 : SUCCESS in 1h 25m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/245abad972dd40f9878aa1aa05e66634 : SUCCESS in 1h 33m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbd46640984642e2b752fe7b4a0305bf : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d74d6bc7cd1eaf88802173c1384d1841577ade9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:24:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e593a36ac3cde22753a63cafb1f6701b6fbe5d8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:08:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d7a5eb8241d44dd2b8a35687ab37216b : SUCCESS in 36m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9606185f595e4b0192e9184339d812df : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24e71bac1a5a4e1ab8b764ae17efe2e8 : SUCCESS in 1h 20m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05ef88e52b5a4f61917eef6862f5b292 : SUCCESS in 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2528551897604a389792e09ab719a871 : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d833c5708fab4f70b07f349302e2c7f6 : SUCCESS in 20m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdbe402849af4d0fb996fceff4db771a : SUCCESS in 20m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c0721e069545d2821448d8dcc3eca5 : SUCCESS in 11m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c44ffd5e12f414faa0dfe8c345a56bb : TIMED_OUT in 2h 18m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbefd0e1b89a432487aaaeff80330e9f : SUCCESS in 1h 24m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d30ce694d8594f7fb69e6ed9f66b981d : SUCCESS in 1h 25m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de942614cdab487299e6444ea44c9ba8 : SUCCESS in 1h 26m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c0e3ae2c3d47428b86d3bc82f5b0436e : FAILURE in 1h 15m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7e2c19ad0d04b77bbd95be78fb450e7 : SUCCESS in 1h 33m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/624826a8509f4e21895fbddc0592ed0f : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":3},{"id":"56f388ef0112d60d9fdef2d5bc7afd6f1191dd68","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-25 00:27:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ea3214fd5fb720573a023efab8a7cec8eef3dfa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:09:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/610b4317877945bba37bfe0578d5038f : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11949ad645564107948921af4e1081a9 : SUCCESS in 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"477feb0d6ab040d35c4bfde9122dd8b812a4e584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 02:33:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd757016e4864fcca7a22dfa1a877697 : SUCCESS in 51m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b1d7ff8eb78248a485fb7194c2b0d1b3 : SUCCESS in 2h 04m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6f6c9b41352437c97e0f69318702dd1 : SUCCESS in 1h 19m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4ea3b553f5b4eec82817085621e4148 : SUCCESS in 30m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd48972084a489ca37eb84581b446de : SUCCESS in 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e1d9e5a24246e88f2d64ffe30adf9f : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b6d00a29abf4c0e8940927c3b48e654 : SUCCESS in 23m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baa5cdf9274141a4883426b6bae0da26 : SUCCESS in 22m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/915864091a2642d1950ffcb7da45569d : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea975400d364019bf07defbe909362f : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b310428efe094395be732d94f94ff9c1 : FAILURE in 45m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38cb2ef5ae844992b4ae4ae8a7b3eae6 : SUCCESS in 1h 25m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bf1f952f70cb48c88b740e19e15daa98 : SUCCESS in 1h 19m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b52b40d09334f7d9e77d7f647be4dc7 : SUCCESS in 59m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cb8f29ef0b184152bb1a3ea9dc96e062 : SUCCESS in 1h 04m 35s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f037af257be46ee878b50222fba02fc : SUCCESS in 1h 17m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb5fa169fc254383bf15ce46e1328184 : FAILURE in 1h 11m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9740c42f0b8646eab0453a901742181d : SUCCESS in 1h 35m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62af197bed2f4893942b6c704603c67e : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7a1917e5fa04ef0f85a837cea400946a87b3a928","author":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"date":"2021-03-25 04:40:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cadb3924b38ffef6177a98593be7b7d41e0b6f82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 05:34:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01543bb1eef743e6989d6822d5a2da5d : SUCCESS in 40m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c25a33924124e7ebde2c1cbfbf42ca6 : SUCCESS in 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7c2a64e8482195c4087cc8bf7d4bb1ffd72c6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 06:43:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a2ff9cd0f104914b42653a333bf10ff : SUCCESS in 53m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6025d9f4ee8444db8db8bf7f477a06ab : SUCCESS in 1h 57m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/554b5b60f173490294bd388df56c4b75 : SUCCESS in 1h 42m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e429051abf748108de3fe2371de2d42 : SUCCESS in 30m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db0648c077f34a4580d2583364ef07e2 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb72b40c5f1843fb9629720fb16743ec : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae59561edd3f4bbba34a664bc9978ff8 : SUCCESS in 24m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cbcdbdf1ed64723bb7a6cceda9a523f : SUCCESS in 17m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4af34835b4a45548f3879edc8bf94a0 : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e89f25292b488e9922cb199dd1f402 : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/56843fbfa6174bc28c429560884a9d3a : SUCCESS in 46m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b62f98688d9345ecb9f66817dc46b8e4 : SUCCESS in 1h 18m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9d47c417d3bb4f468b38b31616edb35a : SUCCESS in 1h 26m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7302448161944031b3b8e6e0ad5c4cbc : SUCCESS in 45m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2ebcc182a91042bca7eb52de855745dd : FAILURE in 42m 16s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b84b5e9abc8b4737b27370be1e4d9bdb : SUCCESS in 1h 16m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/554ce7ad20d14be9a44f27fcc8fe9964 : SUCCESS in 1h 07m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6442e6eedad94c0caa6a8e13cf3d79d0 : SUCCESS in 1h 30m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93758e14ed22422ab464157248e3a509 : FAILURE in 27m 42s","accounts_in_message":[],"_revision_number":3},{"id":"febb03455570f0ecc9a3a26e9e6d2170ef8c7f49","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-25 14:40:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3e21113cf760f8c4fd0ec4b6b3b6c0c5e7aca8c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:39:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ade171ef8b494cbba4a4a21c51192878 : SUCCESS in 48m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c526258cf416469eb9c5869dfb5a9097 : SUCCESS in 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"593eb9e5c4b2fc7721010139241cfac608694e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:57:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74860bea7e45454bb056af8024727903 : SUCCESS in 57m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/795e13a1caf84a0abae67731f82f107f : SUCCESS in 1h 43m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/820b520d78da4eda9b3b5b07c8996633 : SUCCESS in 1h 44m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e925d47359e450c8e3ce463e362ee41 : SUCCESS in 32m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f0a17bf8b524b8abc915830b0d18969 : SUCCESS in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef5adf9429f4bec92fe88d72dfedcb9 : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d04e7a501b44d608641de6e1f77eb84 : SUCCESS in 28m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc6152a3bc0476e8241247f77b821ab : SUCCESS in 31m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/295b7b7b9061494b95d9566d1c7c8b45 : TIMED_OUT in 41m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633bac2aec5f4142a8c490472bc58dc9 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ac6fea75bbf438f8c6d7781a23444fd : SUCCESS in 59m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/020eb7c60f764196928fe35332a8ae30 : SUCCESS in 1h 23m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/baace7130a604c6da9248b2567579c1d : SUCCESS in 1h 24m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bddbb3cb59c4ee780f132de020b64a3 : SUCCESS in 1h 25m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73edcd0ef86f46c68a6ae6ef6e12d7cc : SUCCESS in 1h 03m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8d9090c353ef4f359e028cba49ab5560 : SUCCESS in 1h 20m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e179e47e654e4d8480df6fc7557a26a1 : SUCCESS in 1h 32m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/713d72e354bd4e65bf092ac258d80b1d : SUCCESS in 1h 39m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b80ab1b9ab481dbb93878c8c0260da : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":3},{"id":"82cde03a8c4ddb817217b0bb38d612822c3928fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:59:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3adc33c3d12c8844b3835f6178400448d29df12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:49:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/16f8bdb679b14638985c46c1cf857dcf : SUCCESS in 57m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b3dc8a5ead5b43ae843faf72accd9d2e : SUCCESS in 2h 25m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fad6aaf5caf348bf8d37bb33889e27c6 : SUCCESS in 1h 32m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e9cb1dbd2c445788f4deb5fc8aefbb : SUCCESS in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b65113279e4f25a2a90ec6348411c1 : SUCCESS in 16m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeeedf1ce4de4b299e97c9e29ddf5c16 : SUCCESS in 31m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/019f64224c54487c8029a5d56db8412e : SUCCESS in 27m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ece15e113ab48608c3e9a4237aa53d5 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d1c34cd79b144b294f4a47bddb3f1ca : SUCCESS in 54m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffea78fbd69c4f0fabbc225e7bc67637 : SUCCESS in 1h 22m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f3a913122b345d4a97d7e93e8de4760 : SUCCESS in 1h 18m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfeb13b9977542b4959588ec06b8fa12 : SUCCESS in 1h 17m 38s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be6ab9d9882a4af3986c6611ba72552b : SUCCESS in 1h 34m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/237b0e72047246868cf51cefd9d5bd9a : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":3},{"id":"701b4ed06915d749d88610ea0d7c30124791099e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:49:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9136c2c102efd8dc0ddf6ef03c3700c83a9397ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:56:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd381cc917f4be9b438ce29305125c0 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2bb52d49bec299f070a513fee6f9f7ae5d53118b","revisions":{"0ff4acd6ae9df1d8746fb17e1a8c276d100f3fd4":{"kind":"REWORK","_number":1,"created":"2021-03-22 02:20:25.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/08/782008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/782008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/782008/1"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 02:01:22.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 02:01:22.000000000","tz":480},"subject":"Call install_ingress_direct_goto_flows() when ovs restarts","message":"Call install_ingress_direct_goto_flows() when ovs restarts\n\nWhen ovs restarts and needs to regenerate flows, some flows will be missing in\nbr-int, as follows:\n * table\u003d60,priority\u003d4,in_port\u003d\"int-br-floating\" actions\u003dresubmit(,61)\n * table\u003d60,priority\u003d4,in_port\u003d\"patch-tun\" actions\u003dresubmit(,61)\n * table\u003d61,priority\u003d0 actions\u003dresubmit(,62)\nCall install_ingress_direct_goto_flows() again in the _handle_ovs_restart() to\nensure generate these flows.\n\nChange-Id: I240a78879db757592df138a53b2c22d7f5a9ae13\nCloses-Bug: #1920700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ff4acd6ae9df1d8746fb17e1a8c276d100f3fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ff4acd6ae9df1d8746fb17e1a8c276d100f3fd4"}]},"branch":"refs/heads/master"},"d3c5c1a6903523d7103a9da05c599cf8666b7304":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-22 02:23:47.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/08/782008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/782008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/782008/2"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 02:01:22.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 02:21:54.000000000","tz":480},"subject":"Call install_ingress_direct_goto_flows() when ovs restarts","message":"Call install_ingress_direct_goto_flows() when ovs restarts\n\nWhen ovs restarts and needs to regenerate flows, some flows will be\nmissing in br-int, as follows:\n * table\u003d60,priority\u003d4,in_port\u003d\"int-br-floating\" actions\u003dresubmit(,61)\n * table\u003d60,priority\u003d4,in_port\u003d\"patch-tun\" actions\u003dresubmit(,61)\n * table\u003d61,priority\u003d0 actions\u003dresubmit(,62)\nCall install_ingress_direct_goto_flows() again in the _handle_ovs_restart()\nto ensure generate these flows.\n\nChange-Id: I240a78879db757592df138a53b2c22d7f5a9ae13\nCloses-Bug: #1920700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3c5c1a6903523d7103a9da05c599cf8666b7304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3c5c1a6903523d7103a9da05c599cf8666b7304"}]},"branch":"refs/heads/master"},"2bb52d49bec299f070a513fee6f9f7ae5d53118b":{"kind":"REWORK","_number":3,"created":"2021-03-22 09:26:10.000000000","uploader":{"_account_id":31769,"name":"Jacy Lee","email":"yajiemango@163.com","username":"yjmango"},"ref":"refs/changes/08/782008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/782008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/782008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/782008/3"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 02:01:22.000000000","tz":480},"committer":{"name":"Li YaJie","email":"yajiemango@163.com","date":"2021-03-22 09:24:11.000000000","tz":480},"subject":"Call install_ingress_direct_goto_flows() when ovs restarts","message":"Call install_ingress_direct_goto_flows() when ovs restarts\n\nWhen ovs restarts and needs to regenerate flows, some flows will be\nmissing in br-int, as follows:\n * table\u003d60,priority\u003d4,in_port\u003d\"int-br-floating\" actions\u003dresubmit(,61)\n * table\u003d60,priority\u003d4,in_port\u003d\"patch-tun\" actions\u003dresubmit(,61)\n * table\u003d61,priority\u003d0 actions\u003dresubmit(,62)\nCall install_ingress_direct_goto_flows() again in the\n_handle_ovs_restart() to ensure generate these flows.\n\nChange-Id: I240a78879db757592df138a53b2c22d7f5a9ae13\nCloses-Bug: #1920700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bb52d49bec299f070a513fee6f9f7ae5d53118b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bb52d49bec299f070a513fee6f9f7ae5d53118b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
