)]}'
{"id":"openstack%2Fneutron~782298","triplet_id":"openstack%2Fneutron~master~If01568110ebfa80c46e595fa4818572300baf05b","project":"openstack/neutron","branch":"master","topic":"router_port","hashtags":[],"change_id":"If01568110ebfa80c46e595fa4818572300baf05b","subject":"[WIP] Remove DB Port backref to DB RouterPort","status":"ABANDONED","created":"2021-03-22 17:04:54.000000000","updated":"2021-06-11 09:40:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1922a08b8ca0e1489c47d3f5a9042323680365d2","_number":782298,"virtual_id_number":782298,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-22 20:33:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 17:26:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 20:33:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 23:14:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 08:06:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"792c41ab749ea1186571978ea624ebccc16e8212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-22 17:04:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e679db7de8bd77acabe84b34f58ee984ccd5827","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-22 17:26:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c11a48d7bf4922e589b4a13cc0f829070445c60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 17:26:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/98/782298/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/98/782298/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"ec302c5264bc02b697f87049ac7b25306f6ddee9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 19:22:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/98/782298/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/98/782298/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2ba4939aecfa4eafce08a91fbdd00c7e6e88cd14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:33:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/318b6611aab74e96870bcf67be7daa0b : FAILURE in 31m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c2949801564c4891bfaf47974dabf15b : FAILURE in 30m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/75d785c4acff402e896492070c1ba8ff : FAILURE in 48m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27e04225bc824339b10f88d1479c76a0 : FAILURE in 33m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d3d85320571349098389220b381b02bd : FAILURE in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458007937db8461e8786bf06a6dab0d5 : FAILURE in 37m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f71040f90fc431eb9d8d44018d434ad : FAILURE in 36m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a1408cc925411787aeeff9d32fd295 : SUCCESS in 15m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dc88620f17d4b45a897313800e35e26 : FAILURE in 38m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f73da08c6ea04c60a996cbc32c585237 : FAILURE in 36m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8989c58d994c90947b9d12b219680e : FAILURE in 36m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535fb32999ec49e398719830163de218 : SUCCESS in 12m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ceee16158c44b8e98272bd73b2987c5 : FAILURE in 52m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c78acedc4d3c48149fa0fee4ebd8110a : FAILURE in 1h 07m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b4f406174a464a4599a5e96b345fbaed : FAILURE in 24m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1d413f52c554529aacfe169377070b5 : FAILURE in 1h 12m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2a5ad069a52240ca8ea5d70afc82f888 : FAILURE in 1h 03m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a48619bdb174224b9e832596f96933c : FAILURE in 29m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7de6c2a0c9864f3f8fab498a72edc815 : FAILURE in 31m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a748543d37724924bdd3786fa8badb80 : FAILURE in 33m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f164e5777f4df9b88ac60f6ca078b1 : FAILURE in 26m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6ad1cb9e26d84fa7a74f7811b04a55a7 : FAILURE in 31m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/09b09022676847c5af5cbf1ad6a45ae3 : FAILURE in 31m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5626bd089c3b45f1b8ed0d6ebb2087b0 : FAILURE in 28m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e87e0160c0d466781e885dcd0686651 : FAILURE in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"38e36bd28bbad6f9ed3cc29e18aca1202b0b1af3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 23:14:42.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/98/782298/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"06523e2d86e08c873a14e9146881ec54323b1a32","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 23:20:25.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/98/782298/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1922a08b8ca0e1489c47d3f5a9042323680365d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 09:40:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"60ec097552e3a2b936b5fc3d9763caeaf5c1f080","revisions":{"529464052222b2abc00a0a02df8a23fef26382e6":{"kind":"REWORK","_number":1,"created":"2021-03-22 17:04:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/782298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/782298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/782298/1"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 16:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 17:04:35.000000000","tz":0},"subject":"[WIP] Remove DB Port backref to DB RouterPort","message":"[WIP] Remove DB Port backref to DB RouterPort\n\nRemoved back reference from \"Port\" to \"RouterPort\". \"Port\" DB\nobject does not need to retrieve \"RouterPort\" (and \"Router\" and\n\"Agent\", from \"RouterPort\") when a port is fetched.\n\nChange-Id: If01568110ebfa80c46e595fa4818572300baf05b\nCloses-Bug: # TBD\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/529464052222b2abc00a0a02df8a23fef26382e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/529464052222b2abc00a0a02df8a23fef26382e6"}]},"branch":"refs/heads/master"},"60ec097552e3a2b936b5fc3d9763caeaf5c1f080":{"kind":"REWORK","_number":2,"created":"2021-03-22 17:26:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/782298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/782298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/782298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/782298/2"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 16:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 17:26:01.000000000","tz":0},"subject":"[WIP] Remove DB Port backref to DB RouterPort","message":"[WIP] Remove DB Port backref to DB RouterPort\n\n[Change description, more DB models changed]\n\nRemoved back reference from \"Port\" to \"RouterPort\". \"Port\" DB\nobject does not need to retrieve \"RouterPort\" (and \"Router\" and\n\"Agent\", from \"RouterPort\") when a port is fetched.\n\nChange-Id: If01568110ebfa80c46e595fa4818572300baf05b\nCloses-Bug: # TBD\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60ec097552e3a2b936b5fc3d9763caeaf5c1f080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60ec097552e3a2b936b5fc3d9763caeaf5c1f080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
