)]}'
{"id":"openstack%2Fneutron~782587","triplet_id":"openstack%2Fneutron~master~Id3f71611a00e69c4f22340ca4d05d95e4373cf69","project":"openstack/neutron","branch":"master","topic":"bug/1920923","hashtags":[],"change_id":"Id3f71611a00e69c4f22340ca4d05d95e4373cf69","subject":"Randomize segmentation ID assignation","status":"MERGED","created":"2021-03-23 19:05:16.000000000","updated":"2021-03-29 10:30:49.000000000","submitted":"2021-03-29 10:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782587-bug/1920923","meta_rev_id":"3973e9a765d1085fc9b05898a04269c9b13b2387","_number":782587,"virtual_id_number":782587,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-24 21:10:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 10:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-25 12:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-29 07:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-28 08:27:28.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-03-29 07:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 20:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 20:42:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 23:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 01:30:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 05:14:03.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-24 07:10:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-24 07:52:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-24 21:10:41.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 18:16:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-25 18:16:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9413ff7f7afb1af4afd30163c7d6fdef4cbcf7c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-23 19:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55371b924a1428099b154e022d9a8c6650dcd244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-23 19:13:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6da05009111504e6945e25449e4730975f5fbaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 20:20:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4e5160f525d45c18351225506d202b9 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/497e358d4a174f6d963cfee15e740ffb : TIMED_OUT in 1h 00m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5cbdfbaafb9576686dfb7428772088431338c73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 20:42:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/782587/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/782587/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e99e5f63167d4f077feea7bc8f6e9c98620dccb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-23 21:35:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad64f0abb62516993d6af070daa2d4c26d73c9db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 22:26:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c9f4aa4b3d647969964c62560b61f76 : FAILURE in 35m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff05fe0176f7462cb51c220d91d1556d : FAILURE in 38m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"563556cdbd79d2e54bfec91aae23b4d6637369c9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 22:39:57.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"27160911ed9e0e0c8979f2f2b3b23974badbb445","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 22:41:09.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/782587/3/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/782587/3/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"037f55829698d07ebe56e29fecab36052a2d171c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 23:23:49.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e34a6536dbf241008dfba0c238291e46 : SUCCESS in 44m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aa702a74a9f84e368011bd92e256e30e : SUCCESS in 1h 19m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bdee65ab91d478faa56073c5cd1443f : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6bcea9de2c12400180cf4c365d37efbe : SUCCESS in 1h 34m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cb8f6e0d32a047c796cb37fe36f12ab8 : SUCCESS in 1h 38m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8964d592f9540d7878549dc50fe748b : FAILURE in 47m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/787eef08613b4e2db3ba5be95aef7b16 : FAILURE in 42m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b217e9ae5078437c9a6617d991c6bc88 : FAILURE in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42dcc6141c4740bdad88e318f76be02e : FAILURE in 26m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b118144e99644f5a824f7c429bce19f4 : FAILURE in 37m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32376934eee64f2093a9cf76ca39e764 : FAILURE in 25m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7cf763a0eb4901b8a1313b1ed1ef20 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62584ea659674b7faf9b1c887a36b5ad : SUCCESS in 49m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c96607805f614d50a0a13d6211239254 : SUCCESS in 1h 33m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/511fe8be1b054e728ec9882ccfa6bede : SUCCESS in 1h 25m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16de4fe8e27d450ea4f2427ac2342d84 : SUCCESS in 1h 03m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3ee89afc5d86474f81ab318e45fef852 : SUCCESS in 1h 06m 26s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0ba152a073e4d6e86c0f1f663655777 : SUCCESS in 1h 33m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7439ea2bcf6d411dbb0cd031ccdbb933 : SUCCESS in 1h 35m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92087d849fb04037bbe835b2d06792b4 : SUCCESS in 1h 33m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33e5c154b9a34841a1dd52800f96e5e7 : SUCCESS in 1h 03m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d23f4881e025496caf4ab758466866bc : SUCCESS in 1h 03m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e7870ab1fb984cb192fdec38f2d6fe63 : SUCCESS in 35m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7fd97f39b6f4b848002df4116f133f1 : SUCCESS in 1h 25m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ef6de5bc489341c5adee9e1d5ba44242 : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":3},{"id":"fd825f118080294169d8ab651d28634cb5353236","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-24 01:30:30.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/87/782587/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e20112ae9d3bc52dd44024802f279dfa4fb2e15e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-24 01:36:45.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c5f7d6499e6eb455a1cf2101b4b1a0088d3134f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-24 02:59:18.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/87/782587/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"df72aa8c643a5f1d5fdf81ee27edc98aa70b4713","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 05:38:28.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/782587/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/782587/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"45425c31ce7d6e066cf572877d8c6ae8bc8fd883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-24 12:24:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ccee99935afc796983478437bdd7d6601a58c9e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-24 13:08:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nAnd please fix pep8 issues :)","accounts_in_message":[],"_revision_number":4},{"id":"d892738b9e6042257ec347e8c65c9d46c8aeb600","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 13:33:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f649966f03164f51a31554fd90bc0d38 : TIMED_OUT in 1h 01m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4eb0c3ac77b54a18879436c16e3f3f79 : SUCCESS in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b887fabf72368e557afca4272e8b8ac906cab7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-24 13:55:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"94f6456c11407bd973d9136e9949bc597f97e75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-24 13:56:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e3f6520f9be49f4e6d9ea842673580234e1485ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:57:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b7b128041114287b8778f35bcec55f1 : SUCCESS in 41m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aeb03b97f2fc46fda737d87105404be2 : SUCCESS in 49m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c3f090156e2ce975cbdee0f2ada7f404107e84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:07:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/09bc2ece1b4e4071ab177ca8614f4c32 : SUCCESS in 54m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/32c45cef9b7d44f3ad22a27f2bdff33a : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9cde7182105b46a5a0d2eed51f128987 : SUCCESS in 1h 46m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec6af629c3e943b4a330321e0c9b6534 : SUCCESS in 1h 41m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/143536106ad54c2886bb5bece3303169 : SUCCESS in 1h 59m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da12271c396c406089d1189a01d34dfd : SUCCESS in 45m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b5c0624d40e4611a1b88f9ef00b73f7 : SUCCESS in 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd835114d1648869eda50ee758248e8 : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43928e1a992a41968efdf1dd76b6a843 : SUCCESS in 29m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3120560ec3964494b6d8db59d4a2c13c : SUCCESS in 31m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d084184c5804d058f877a2737bad0a2 : SUCCESS in 33m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9b43c537324af0950b024ec2ab1e6a : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a6a6d390854466db6516174ae2f5d1a : FAILURE in 55m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d351c349859147d5b4032a48624af4a6 : SUCCESS in 1h 27m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5c203ec606e2499698ffd0399a128916 : SUCCESS in 1h 35m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90e50a7057ec414284617e2a19af897f : SUCCESS in 1h 22m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f1178222099d456d96597dbafa2ae846 : SUCCESS in 1h 14m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0988b359744f4ef08982803091e7ae5b : SUCCESS in 1h 28m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/edfa477b0df141bf87df115602cce550 : SUCCESS in 1h 32m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60204bdcc7254b399998fd43fa26ae7c : SUCCESS in 1h 40m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62258542a9c04efeb8c945fad6da580b : SUCCESS in 57m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/61f43b09d5184dafb823b51f42f46d32 : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0343629e92454cec99cc2a124d8e8657 : SUCCESS in 37m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e957d3998eaa4734bd50f72fd1891b9a : SUCCESS in 1h 25m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/354a4cbb00ad4419a9261bbae275cc30 : SUCCESS in 1h 36m 06s","accounts_in_message":[],"_revision_number":5},{"id":"6cc3bf0000825b3ec2246f251bc882442d8df62c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-24 16:30:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1387cd9adf4d077d7bf6570d3a49da085282dd3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:30:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbe97dad67d545e59a9f9e4101cec44a : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73bb95d0136e4c25b59e4db13149a956 : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81580fe6f806f1df0808496b6b77a1c0c5e2ce1b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-24 18:10:52.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/87/782587/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"102db98000411e3b3ea55cabe502e6871ee866f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:35:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3779d59c728741cfbd74f2abf351da6c : SUCCESS in 47m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/30dae9d685794e6894bd028100e09480 : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d6b741314664651be9389fbb80b9687 : SUCCESS in 2h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b138fdde29b4b94a1192de6ee437379 : FAILURE in 23m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eb82c9e87b054e65bc7ca7ced913b8bd : SUCCESS in 1h 46m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70c795aca7944f869b04834d838d0dc7 : SUCCESS in 31m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcc8fe85bb854c449031f1f5e01a8833 : SUCCESS in 26m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a48b9fd0a9841398729f269956cb7fe : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40ce4d34d91a4fda9717c74f65c93757 : SUCCESS in 33m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/004a6fc704db44eea65b365b7934a385 : SUCCESS in 33m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21d15c7386f4b3aba9ce949abbaf863 : SUCCESS in 25m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa12ac4080674a8fb3259465db42ecf0 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb15358d280945b79e9ea490f0b01444 : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/36e1b76327454a2e98900342988978bd : SUCCESS in 1h 24m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/517943660b674a3fa6c3c35d84c76e3d : SUCCESS in 1h 23m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5da88972adf42b686e71a32537d45a0 : FAILURE in 36m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/591426a8af75494ba00fb600f82fdfc7 : SUCCESS in 58m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5bbf1324828e4271b96f121128203cb2 : SUCCESS in 1h 10m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/628a9e6dc4bf48859eb7d0a012e7afe9 : SUCCESS in 1h 07m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a1ebda22772467c9bd8e32c6d184ad6 : SUCCESS in 1h 46m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f79ffbe0b525485eaab95db38462cb14 : SUCCESS in 1h 17m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b163c997be104cc6ae04bb76619c1b3c : SUCCESS in 59m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc5d741d751647ae9ca72f0293479358 : SUCCESS in 33m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/be4f14172a82494881b435e961309b95 : SUCCESS in 1h 29m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e33c3cc88daa442cb676e8ba3b169814 : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":5},{"id":"327137af782cf8f27f55412332ed6f055d77a960","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-24 21:10:41.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/87/782587/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a949ed90b59011825d64e0c0c0d632f7b669c624","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-25 07:41:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"281ac690f1d16b051a333d97fd7826032faa20aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:29:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5625c2150a57428aa6600768ac1c8ef3 : SUCCESS in 38m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c64e3f4f8d44eaca73987e56edafec9 : SUCCESS in 41m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"10a66bc06b05da9f9cd7bca21c2c7914e4acaf44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 10:03:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80a0a57763424f009fa3582dd09fb0f5 : SUCCESS in 1h 03m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df31f4eeecac41b4bd5be8845942272c : SUCCESS in 1h 52m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5e6c46c0a4b6422682e2eb113a3c8a82 : SUCCESS in 1h 53m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3bba6dc7fe347b39dc6697bf5cdf6a9 : SUCCESS in 1h 55m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cf6e8b7816434864bd967d928125def1 : SUCCESS in 2h 07m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2945dd6206f8484198934e5173c13018 : SUCCESS in 44m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f30cf469ef4fe19a17d057f71e88c4 : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719e5022e0484bb09c01811c7e9611af : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aaab719a40a47588613b916b82e6c0c : SUCCESS in 33m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af6631392859400ca0d929b356eeebab : SUCCESS in 29m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b399015b51f4c6296f2ef1a12af3f28 : FAILURE in 28m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53851e3007eb4d6cb2b5416d3d0fc051 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73201092e0004bcabef8492333a470be : TIMED_OUT in 2h 15m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/62873f8ec58340ff9804a2bd80b35a7f : SUCCESS in 1h 36m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/88e2f45d61464add8cc5547252a08ae7 : SUCCESS in 1h 36m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/45b02b69b2be413296c73a824c79273c : SUCCESS in 1h 08m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/87b92abb188e4221baa6e838f05aee14 : SUCCESS in 1h 20m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a52fe89a54b4ce4914bc47b9b7efc3d : SUCCESS in 1h 32m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d55d0116668487195c351e281f3dcf1 : SUCCESS in 1h 42m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d9547cfcb38a40da8c17fe8e06e18c2e : SUCCESS in 1h 47m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17949f62bd374654889a29c5aae1a628 : SUCCESS in 1h 14m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0329cdc884504811b0b1c1ca417b0b8a : SUCCESS in 1h 16m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1af6f9b98cdf4a34a6c6dab3bdb8c0e1 : SUCCESS in 38m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ace5d79d318d4602b123b5c2916a4a6b : SUCCESS in 1h 27m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b070c900b2744586812c55895ea3058b : SUCCESS in 1h 32m 34s","accounts_in_message":[],"_revision_number":5},{"id":"6708cfae0db7f804261cd4cb3f9f6433e27bcd39","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-25 12:59:39.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"86a86f84a8dda728531ebaea99a2504a3413a15a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:01:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81999f4139f0498eb84fe4cc8c7c4757 : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28ee498cb7ed4e1e8afe9dfdec5ee94e : SUCCESS in 50m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62004fa8ec09cbbd5e8127e6e2e5c7327b1ab843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:31:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b0feb1e6257499b9e73ffa35907d53c : SUCCESS in 51m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c434a29d473e4955884f3064a40ded35 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f4c8e0fcb984f43a46e01dc70d6d804 : SUCCESS in 1h 59m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ba53ed6993f4959a473cff4e20674aa : SUCCESS in 1h 42m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/06d3de88de134a4797f17cad180afcba : SUCCESS in 1h 51m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcbec9469ede49229e3a2c98620f496d : SUCCESS in 32m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c210b8e76be4e8791a6d6243f11865d : SUCCESS in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db9c3f806de644eb86e0dc0db22a7bb0 : SUCCESS in 19m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f40702940514635aaa81edd743149e9 : SUCCESS in 28m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d814d78b96a4ac7986ed9c592485081 : SUCCESS in 23m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c5b8b8f22d44e1980ee4fab5de04a31 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b66d75d5dc84e04a28b41dc9a8c3057 : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aadaba88b7364c138fd673436480f37c : TIMED_OUT in 2h 17m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/612976ec34a04cf4a09aedfc20a2ecdd : SUCCESS in 1h 36m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/80bcb843f7d0466ea8178bc6e4f07647 : SUCCESS in 1h 26m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d88752d2fcd44235aaff9798943a444b : SUCCESS in 59m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a66bdb214e5041ef9344260347664b8e : SUCCESS in 1h 23m 41s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4548e8c646bb47b69fabbbbbdb7fc72a : SUCCESS in 1h 10m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8344e1ad88c44c38a57b628f35f10356 : SUCCESS in 1h 21m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/329c32f776954ab7ab211e88a54b483a : SUCCESS in 1h 37m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11695635ef38425cacc0ff0e9da5c6f7 : SUCCESS in 53m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/498c0b5205c9496691b7125e2d02cbd7 : SUCCESS in 1h 16m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f1facb08e3774f3fac4ad2b34fae2627 : SUCCESS in 32m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e868d468bfd84acda5cc8cc743d5c2b9 : SUCCESS in 1h 26m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/47b440e409fa4e3baa2b5e5bf44bc0c5 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":5},{"id":"348ab1b29e3f671c207d6d0e9a978b10fe471b15","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-25 17:52:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2d9f5df5312bb3d4013b88747fb2c7a6d6b95dcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:10:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"af74c985a43dd23de64c770f1cdd3fb4924667df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:15:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f003a82384f449af8c4e605b293632fe : SUCCESS in 49m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/921b8c425ffa4c7a94354591f84b33cd : SUCCESS in 1h 51m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d8d2fd735924b4ca194f9ea2c84e49e : SUCCESS in 1h 38m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/75f02ae3534249ed8cba0a2459113337 : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/04ba89561ff74e39a671de3a484a94a9 : SUCCESS in 2h 13m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95555098e7334906b59b552a8c09c625 : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a8100992047455b9bf57b2e154c8a71 : SUCCESS in 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf0797012294b6c969741cc2e3a656f : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eda5876b9b724e9daf4baf1e9a2eb462 : SUCCESS in 30m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce0a33089d364224913397fce9d2a8e8 : SUCCESS in 31m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63d05f90a47a4aad92a32615169907a2 : SUCCESS in 28m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9bb022b4ad4256bfe0ca42666e357a : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d043ef714fa14a56b558b7edf4753148 : TIMED_OUT in 2h 16m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e109eb192204d838099b0e430138853 : SUCCESS in 1h 24m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/72e4a8526df0493fb47db72f4b13bd53 : SUCCESS in 1h 29m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34efdac765284b5ba10036f1bd32c554 : SUCCESS in 1h 09m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/30f4b53545024bcfb8bc1b5925610a0a : SUCCESS in 1h 13m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7cd8e2a706e04a1ab0c85163ff119a7d : SUCCESS in 1h 17m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f81a2569b3254adea25b17a3128f1504 : SUCCESS in 1h 29m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d475b7f10994c72913092a270531e3a : SUCCESS in 1h 32m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bc1dce0fb404e27b54be0f8368ac198 : SUCCESS in 1h 01m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/68c330d505cb4f0194542ff5e5f5f01c : SUCCESS in 1h 10m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6242907b5b5f4da3997ca59a41636466 : SUCCESS in 37m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/60605cbaa7654124aa551f45df63e3cb : SUCCESS in 1h 30m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b54a1e608b3e4e96abdd6ff542389f70 : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":5},{"id":"918562902e5ce61c75e5afe538777b123ace7fe8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-25 20:17:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"524984deb32659c4362a3f6a19afb0ef1fd2c89b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:28:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fb78716668c4cc3b9be3472f6810aaf : SUCCESS in 54m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/105f19b8e58d44b4801df7d8bfe34acc : SUCCESS in 57m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6f38ad637d9fa3b5d5b5d379f9e8224d68d321a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:40:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2bb12bc917c546f8bd7e38613e8798b3 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dbeb719b0edc4eb0b728424675bf70ca : SUCCESS in 2h 06m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/208fa3ebe637487d9f17dbc2c4ea860e : SUCCESS in 2h 15m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3391afbe2dd048bca0d29c36f97e3b5a : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/49b6d6cb21b04b3396493cee804a88d7 : SUCCESS in 2h 15m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6216de9a39b4f129ee4a2c8c33ae870 : SUCCESS in 38m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52e44a3885c442b1ae42d553dfc11f84 : SUCCESS in 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c594a4897d9449c894980a17ba773424 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d803b4663b24981aa00f1001eb7fb5c : SUCCESS in 31m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a2d0b79b6ed45f0b8a2e61480dfb8cf : SUCCESS in 27m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc22b64c4275431db7399be234fc546c : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc082c4f047443aa1666ffe0e854763 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f0caf70e8f84325b435fe346fdf2d17 : SUCCESS in 49m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/235da5a5b1264da19736ef533d3b9081 : SUCCESS in 1h 19m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2d90df15fa9c4e339df41655e4a14148 : SUCCESS in 1h 27m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74eeba7e5c624188a9ea02f6f4c6fb7d : SUCCESS in 59m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8b752b84f4742c496afdfd4565e126e : SUCCESS in 1h 03m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/63956478ece64c07ac8707f23dda7efa : SUCCESS in 1h 22m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b079921e363443abf292c5a6d8aa095 : SUCCESS in 1h 30m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c94a04199e344b3f9b298c7198dff423 : SUCCESS in 1h 31m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4507e5bcc81541dd9792d708df89474a : SUCCESS in 1h 05m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e1a7ff6b91c64428bcfeabac3e2ec901 : SUCCESS in 1h 07m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c4f754ecf41949b7bc47a032043992a9 : SUCCESS in 31m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/42012c960bd44253a527d40f3bd9bf0a : SUCCESS in 1h 31m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/debee579ff1a40b695d4bad2e38037bf : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":5},{"id":"b1a87d169fca8527b59bb439b5b17f95fd2763d1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-28 08:27:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c7288c33381a858abadb06b67fa76fd01d1b4fc0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-29 07:53:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e3b28a474f576f4f31f8fa34922899d7cb99292f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 07:53:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"05bc2bb70778cf55ed0370b1dc48fc7abdb7f2ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:27:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85b8f4ce3e9e416b89ecaa41557d680f : SUCCESS in 46m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1a9352595887484085e19ce75c2278bb : SUCCESS in 1h 58m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/16902320b902449fa09b8a40acc8f2e5 : SUCCESS in 1h 43m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c7ccca14404941f5be040ab7f7c75583 : SUCCESS in 2h 29m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9db17e7f9a764c7cb163523b6a4057ad : SUCCESS in 1h 26m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff7c2f95fd6e4d01a90f01b88545b826 : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14919be0cc14503a3edbf21f5ac8ceb : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b813249f3bfe40caa505560290bc946a : SUCCESS in 32m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f2294eceb034a5bb952578e371b8fec : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0837a4aeae2b42b49e17a1d640e82b1a : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55b885bb38734b698d6ff25ebba63694 : SUCCESS in 56m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8799d7af8aa44b1fb647a3f707ecc344 : SUCCESS in 1h 30m 25s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7064b4a6999e40478040a84e5a9fcd07 : SUCCESS in 1h 21m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbfcb9490c4b412e856eed53b4312533 : SUCCESS in 1h 08m 06s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/25d42f48a52847619b4c5e445bbc34ff : SUCCESS in 2h 05m 11s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15a79124ab064f7ea4a335293e74b40b : SUCCESS in 1h 13m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d989eac924b41abb7722ca6249f9ea6 : SUCCESS in 1h 14m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b618072a177b4167be459d4e156202c7 : SUCCESS in 41m 49s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/52ca990cae354b8a89646e27db859e93 : SUCCESS in 1h 28m 09s","accounts_in_message":[],"_revision_number":5},{"id":"a7d30a37dfdde59489ec221f468b77b9a0f280bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:27:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3973e9a765d1085fc9b05898a04269c9b13b2387","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:30:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e085ef43254929a0eabeae408fe9da : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6eaa6d83d7c7f07fd4bf04879c91582de504eff4","revisions":{"4de347cee672a09be2d48efcd05ab8fe18b8f677":{"kind":"REWORK","_number":1,"created":"2021-03-23 19:05:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/782587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/782587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/782587/1"}}},"commit":{"parents":[{"commit":"5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62","subject":"Merge \"Log python warnings only once to limit size of CI jobs\u0027 output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 19:04:19.000000000","tz":0},"subject":"[WIP] Randomize segmentation ID assignation","message":"[WIP] Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nTODO:\n- Use functional test (mysql, postgreesql backends) to test this method.\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\nCloses-Bug: #1920923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4de347cee672a09be2d48efcd05ab8fe18b8f677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4de347cee672a09be2d48efcd05ab8fe18b8f677"}]},"branch":"refs/heads/master"},"5973f91253215cb9c7b9b8cc7f28101655516783":{"kind":"REWORK","_number":2,"created":"2021-03-23 19:13:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/782587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/782587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/782587/2"}}},"commit":{"parents":[{"commit":"5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62","subject":"Merge \"Log python warnings only once to limit size of CI jobs\u0027 output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 19:12:57.000000000","tz":0},"subject":"[WIP] Randomize segmentation ID assignation","message":"[WIP] Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nTODO:\n- Use functional test (mysql, postgreesql backends) to test this method.\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\nCloses-Bug: #1920923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5973f91253215cb9c7b9b8cc7f28101655516783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5973f91253215cb9c7b9b8cc7f28101655516783"}]},"branch":"refs/heads/master"},"94d242c19ec45e89f191d2157c6b02e50cbcaa5e":{"kind":"REWORK","_number":3,"created":"2021-03-23 21:35:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/782587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/782587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/782587/3"}}},"commit":{"parents":[{"commit":"5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62","subject":"Merge \"Log python warnings only once to limit size of CI jobs\u0027 output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 19:23:56.000000000","tz":0},"subject":"[WIP] Randomize segmentation ID assignation","message":"[WIP] Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nTODO:\n- Use functional test (mysql, postgreesql backends) to test this method.\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\nCloses-Bug: #1920923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d242c19ec45e89f191d2157c6b02e50cbcaa5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d242c19ec45e89f191d2157c6b02e50cbcaa5e"}]},"branch":"refs/heads/master"},"14a29f3ae74fde08aa16d2ae88066944cfd2e67a":{"kind":"REWORK","_number":4,"created":"2021-03-24 12:24:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/782587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/782587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/782587/4"}}},"commit":{"parents":[{"commit":"5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62","subject":"Merge \"Log python warnings only once to limit size of CI jobs\u0027 output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5190b8c9c1ec7f4a89fe103d65f8bff3e6c8db62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-24 12:23:41.000000000","tz":0},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14a29f3ae74fde08aa16d2ae88066944cfd2e67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14a29f3ae74fde08aa16d2ae88066944cfd2e67a"}]},"branch":"refs/heads/master"},"6eaa6d83d7c7f07fd4bf04879c91582de504eff4":{"kind":"REWORK","_number":5,"created":"2021-03-24 13:56:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/782587/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/782587/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/782587/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/782587/5"}}},"commit":{"parents":[{"commit":"9093a6f06535871d86254e344012501f2a659901","subject":"Merge \"[OVN] Set OVN_BUILD_FROM_SOURCE\u003dTrue for ovs-master jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9093a6f06535871d86254e344012501f2a659901"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-24 13:56:09.000000000","tz":0},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6eaa6d83d7c7f07fd4bf04879c91582de504eff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6eaa6d83d7c7f07fd4bf04879c91582de504eff4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
