)]}'
{"id":"openstack%2Fneutron~783339","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I6c68d15d0af644ab00467bea8e0cd3ca816f5cc5","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1919352","hashtags":[],"change_id":"I6c68d15d0af644ab00467bea8e0cd3ca816f5cc5","subject":"[OVN][FT] Mech driver stop accessing DB at exit","status":"MERGED","created":"2021-03-26 15:41:42.000000000","updated":"2021-04-01 10:00:23.000000000","submitted":"2021-04-01 09:57:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783339-bug/1919352","meta_rev_id":"d2ada6aa4d8d8343fb60ac00230eea4eeda09d43","_number":783339,"virtual_id_number":783339,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 09:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-04-01 00:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-03-30 14:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-29 09:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-29 13:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-04-01 09:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 16:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 09:51:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-29 13:53:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-29 14:57:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-03-30 07:22:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-04-01 00:59:43.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"489339d4f35e38c99da6ea9a92a82300822caaa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-26 15:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e180c63c48e736b10719935bbd35c2c745283ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:42:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cc1f634d44f437aa5696a3bb14809d8 : SUCCESS in 41m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/852a2bf7f40f40398e8b9353a7b48f4d : SUCCESS in 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3244f4ebb3ac4c08b159262e71e580ce : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d87f638ca182497f85ca801ce1d97954 : SUCCESS in 36m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/879600c8db4042a1aa7256fc3ba52dd4 : SUCCESS in 37m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/197e4ca6a6854cec9d519fa2ffa8d8c2 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6709daf184134e769884339b538b1338 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"294c8844b473c48a82b9ec09b07511c9cebb52bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:17:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3157d1c60ebad752931e8c2ef1c6ea7f423048ca","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-29 09:51:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee7b965d8171af059373b3b10b4e1a529cc3a839","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 13:53:03.000000000","message":"Patch Set 1: Code-Review+2\n\nI will +2 since this looks like RC2 material?  But note there are really no jobs being run here yet, guess there needs to be a template update?","accounts_in_message":[],"_revision_number":1},{"id":"cbe14fca0594ae94e48c22de000a0dc99ee1d690","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-29 14:57:51.000000000","message":"Patch Set 1:\n\nrecheck the templates update was not merged yet, let\u0027s see if it gets more CI","accounts_in_message":[],"_revision_number":1},{"id":"e42088c11a9a92f2308dec1bf44fe86d3323e764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 15:54:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa91be4a22b43db948e290e249f6fec : SUCCESS in 40m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/832aa5bd4428440ca1385b30f989c6f8 : SUCCESS in 38m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e6915b01b44c6bb788f14ca99c8b06 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a52d3ac852ad46c68c03417a09ceba02 : SUCCESS in 27m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765b30c984e1438d88a5e31d1863a3f9 : SUCCESS in 30m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/389790bf8cb7488a85451fb986db828a : SUCCESS in 34m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7adbca0a949b41f5b965ed3da0bc922f : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9a10343cbdca8221ca9f98823789d0cbaaf8828d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-30 14:27:37.000000000","message":"Patch Set 1: Code-Review+2\n\nSounds nice to have and safe to add to RC2","accounts_in_message":[],"_revision_number":1},{"id":"10924a3df4c1b6fded4686bfb292d76388880eb9","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-04-01 00:59:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a0dd8727ad99c65412069ecb8d245e3d26c486eb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-01 09:14:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62a509b40070cff81b32565648629cdf1df3e32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85295cd6682648231b3d15ddfc0939ee958e74e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:57:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfab69322daa4013827a37eaf0ea8cdb : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f847b8fcd814f6d866ce290ddddabc8 : SUCCESS in 14m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a31e210925f499fae4de65e495ee28e : SUCCESS in 26m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/166e9d6665404ff3b00253f1da7c597b : SUCCESS in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f48497cac0b4ccbb56b7b789307bc76 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a5e5fd88da684447f900efc109ef0849e89a2f69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:57:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2ada6aa4d8d8343fb60ac00230eea4eeda09d43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:00:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805c962a12c744ac97c084e309a6d0d9 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fb614046f7bede7f524f4ee85d26e3b0f087df2","revisions":{"0fb614046f7bede7f524f4ee85d26e3b0f087df2":{"kind":"REWORK","_number":1,"created":"2021-03-26 15:41:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/783339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/783339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/783339/1"}}},"commit":{"parents":[{"commit":"9dfa29fe0e52e6b73c3067613bdbc79fa2f7eb92","subject":"Merge \"Revert \"Implement \"kill\" method using os.kill()\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dfa29fe0e52e6b73c3067613bdbc79fa2f7eb92"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-16 15:24:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-26 15:38:47.000000000","tz":0},"subject":"[OVN][FT] Mech driver stop accessing DB at exit","message":"[OVN][FT] Mech driver stop accessing DB at exit\n\n\"OVNMechanismDriver._clean_hash_ring\" is called at exit. Because the\nSQL connection is already closed, an exception is logged. This patch\nmocks this method and avoids the OVN FTs to log this message.\n\nCloses-Bug: #1919352\n\nChange-Id: I6c68d15d0af644ab00467bea8e0cd3ca816f5cc5\n(cherry picked from commit 92a4db01fefb3fe224be920abf372436b8e02c8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb614046f7bede7f524f4ee85d26e3b0f087df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb614046f7bede7f524f4ee85d26e3b0f087df2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
