)]}'
{"id":"openstack%2Fneutron~783639","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Id3f71611a00e69c4f22340ca4d05d95e4373cf69","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1920923","hashtags":[],"change_id":"Id3f71611a00e69c4f22340ca4d05d95e4373cf69","subject":"Randomize segmentation ID assignation","status":"MERGED","created":"2021-03-29 11:24:21.000000000","updated":"2021-03-30 13:07:42.000000000","submitted":"2021-03-30 13:06:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783639-bug/1920923","meta_rev_id":"dfc9c69905b102706d66f44896c8fd035f5dec07","_number":783639,"virtual_id_number":783639,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-30 13:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-03-30 12:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-29 13:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-03-30 12:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 12:12:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 13:54:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-30 07:22:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-30 07:22:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e12ee62ae981edce6c2cb1dbf20419fdb393da1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-29 11:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d456cd66659d8079e5f691ebfd6d2c1bac501797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:12:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aecc5b12cc94407b09d20b12250c297 : SUCCESS in 36m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64955f0d67a84c288e6bb56e21314ddb : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf7709412f34e33bfd2d5b1a32e7701 : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0941bbeefe744db59473cd355c7069c4 : SUCCESS in 28m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e1f9cdde8fd418ca64aca9471c69079 : SUCCESS in 24m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287b8802c68443b49dde39c013d6b257 : SUCCESS in 29m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295f1d8f04d64f0fa3e298164c16bcec : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2f6ce086f6206150b20b82f7adf4bfbd9647cfdd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 13:54:26.000000000","message":"Patch Set 1: Code-Review+2\n\nSame comment here as other Wallaby change - assuming RC2 and note there\u0027s not really many tests being run.","accounts_in_message":[],"_revision_number":1},{"id":"cc2e7876c93bfdbebf51f08024dcbb60e3054899","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-30 12:24:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7399f1ae229d9dcfd5732a4991f10493bfb9af02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 12:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a408670475fc1056c0ca44988d9767e7ddaf475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:06:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da701f6873654524a2486928018e0e5d : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debb79f071bb4b27b8b4c8800a97f856 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f392271473b441f4a34126eea750fc22 : SUCCESS in 32m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a1ab8d12549404ca46150c812422fef : SUCCESS in 21m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd8c3035dff4c50aa1702dc9d20fd05 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9d040e22f687ada19fb1e038bbd15701fab04474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:06:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfc9c69905b102706d66f44896c8fd035f5dec07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:07:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1f86365128461c8984b4b0a60b9376 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d06fd915b33034206a995625ec057cac937fbba","revisions":{"1d06fd915b33034206a995625ec057cac937fbba":{"kind":"REWORK","_number":1,"created":"2021-03-29 11:24:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/783639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/783639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/783639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/783639/1"}}},"commit":{"parents":[{"commit":"b0650575096aa8ad12c183d04c7a41d2df6d1d9f","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0650575096aa8ad12c183d04c7a41d2df6d1d9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-29 11:23:56.000000000","tz":0},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n(cherry picked from commit 6eaa6d83d7c7f07fd4bf04879c91582de504eff4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d06fd915b33034206a995625ec057cac937fbba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d06fd915b33034206a995625ec057cac937fbba"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
