)]}'
{"id":"openstack%2Fneutron~784013","triplet_id":"openstack%2Fneutron~master~Ifc7ec227d4084331948c42cd99d96ad10c09a6f1","project":"openstack/neutron","branch":"master","topic":"task/42184","hashtags":[],"change_id":"Ifc7ec227d4084331948c42cd99d96ad10c09a6f1","subject":"Migrate \"ovs-ofctl\" to oslo.privsep","status":"MERGED","created":"2021-03-31 07:12:52.000000000","updated":"2021-04-06 16:32:48.000000000","submitted":"2021-04-06 16:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784013-task/42184","meta_rev_id":"ea08f598334cc643f895d500a75469211930687f","_number":784013,"virtual_id_number":784013,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-04-01 09:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 16:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 10:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-04-02 07:30:01.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 10:22:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 08:12:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-31 08:17:16.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-31 08:21:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 08:30:55.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 08:37:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 11:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f77d7445c9066c9772b625ca9c3767e749f8a54a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-31 07:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbfffbb1687219df621acf9d33bdbe8983f1286b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-31 08:12:48.000000000","message":"Patch Set 1: Code-Review+2\n\nI hope zuul will be happy with it too","accounts_in_message":[],"_revision_number":1},{"id":"8e88dd2be3436a0ed9ecdf16a041c3737501f2ad","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-31 08:17:16.000000000","message":"Patch Set 1: Code-Review+2\n\nwaits for CI","accounts_in_message":[],"_revision_number":1},{"id":"d1fb11a4f18ee743aabcc46565c7f9fe8f1b7264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 08:21:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ef2f811f3c14c21ba49e90641d38212 : FAILURE in 43m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2cca058adbc4aafb0a445bb36ffcd28 : FAILURE in 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"499ad5b2bd3da1ef80828381405182e56c5d413b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-31 08:23:30.000000000","message":"Patch Set 1: -Code-Review\n\nUT to be fixed","accounts_in_message":[],"_revision_number":1},{"id":"496021fe8ad67a4661deaf2854fd83a1e9123ab0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 08:30:55.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/13/784013/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/13/784013/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9fa0a49ed980aa1677da9eedd6b2b1c882b4bce8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-31 08:37:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/784013/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"901287d07b32a5bb8dd15263af331f67d36cb262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:34:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6312a0765e84ddfa20b8cd63ce29ae1 : SUCCESS in 2h 31m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb26a0ad7d38402cb0d5f04092f48193 : SUCCESS in 2h 07m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/137462544e72424da69ad2f8c9c71c44 : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f4ba841f2eaa478088decb6e6b9a1ac7 : SUCCESS in 2h 00m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb98151dc9014b54800449e3589a7ff6 : FAILURE in 31m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/941f8ff5f346411d9f296dfa550bc868 : FAILURE in 35m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df636dd99704b7f92fe9f3c2113ea78 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4085abb176c14d84822324c8ab90755a : FAILURE in 29m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb54985099414a3b86f4254f45a48c46 : FAILURE in 24m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88116b8200f243f6bc7cffba26024de8 : FAILURE in 27m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc235104b484ee287d7d1bd06ab9eab : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/770d6012b1074bd982379f6dc3536caa : SUCCESS in 56m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d90cf0c2f7446d0b84731753e3e533e : SUCCESS in 1h 23m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1d4becd2d56548f7ac37edf8a75f1b84 : SUCCESS in 1h 48m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1dbc0febc414dd492b61bc7a9558eb2 : SUCCESS in 1h 00m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c8221345ca24a189369715cab1b6996 : SUCCESS in 1h 11m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f223a290696641c8b627b3b1eddf2b99 : SUCCESS in 1h 23m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b503109aa7c46c697b07447b927dcd0 : SUCCESS in 1h 13m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b020ca009ad4bfc8873a0e1c1e2043c : SUCCESS in 1h 47m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56201d0ef9a5466cbc5a707ef9be6b4e : SUCCESS in 54m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/67872771d38f4398ab9c77461d0d7dda : SUCCESS in 1h 13m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c9ca27b3b3a64c0781405c6cc65c0407 : SUCCESS in 37m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e06c8a084f0a43718d8fbc3a433529ac : SUCCESS in 1h 54m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6ffba9977e924dad82253cefb05b7957 : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2622a4265bdeae62c55fd2abadd14a0f24ee869f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-31 14:07:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40c3d4b18cd1d52457412d8e20e98ef80c7fd792","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 14:58:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50e486ceb7d24d729bdbf64ff957ccc8 : SUCCESS in 44m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74d54556d4e9446d935743b2f7702d45 : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b9752f371f37b8c988603538b7a0575d1365b6d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-31 15:04:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a74ed08ec1031d5822b6d38a47ea409895d9c1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:16:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3fd732cbab83479583768e3d142cb2bf : SUCCESS in 1h 42m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8854e45fa37a4f88a761424f9a6df0b9 : SUCCESS in 1h 55m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/206082fc36964f3ea02f3fba87d7fcd6 : SUCCESS in 1h 21m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7e04231f5556466eb3eb216fe14050e8 : SUCCESS in 1h 58m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe69e0ca0eb148c09447cb377bd59219 : SUCCESS in 40m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/219073032c644b03a05d0ff3ff1087bf : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce72c6d44f24c5686e8716f5ef14cd8 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e2a589731c647c9a2a39f25eaf9268c : SUCCESS in 28m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcd42e36107f4c04afa88a750b08ccaa : SUCCESS in 17m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfe1e26ec7674284bf78e93498678e5b : SUCCESS in 35m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cebe6dfd55a44960b55c0a4c7485f07e : SUCCESS in 12m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e34012e02ed64859b037b0cb0c3e7a25 : SUCCESS in 49m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4baf04c3346e4cdeb7af110b941ac5bb : SUCCESS in 1h 40m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b14b6677f2d04b1ab0ccb2417a25f768 : SUCCESS in 1h 30m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/695d49926ec441e0933ebed51abc106d : SUCCESS in 1h 24m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ea19f6c2e0e47caa320ca8104ea980b : SUCCESS in 58m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f159123b00ec4b7fbca571c1ce390de8 : SUCCESS in 1h 20m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6163ce06c1464e07bc4635d0d522468e : SUCCESS in 1h 47m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98b19867d6a54d59a3ca378b250a9392 : SUCCESS in 1h 22m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c91929e243448e89d28ebd9176c1f0e : SUCCESS in 1h 07m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9d86c29ad56e460f808be89846c68b14 : SUCCESS in 1h 14m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a77a801c364bb08b415f86f1da8a33 : SUCCESS in 36m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7f680a2d50046abaa14ae7d4f14b5c6 : SUCCESS in 1h 32m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e11905017be9432f8870ff6ec82fdea6 : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d9ed425a34136e02afbd3ff1a6c3099484fb975f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-31 16:44:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8571d32979abf1f5e721a363d748a2be9017762e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 16:45:34.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"cf14b75e8e8d72f22e811921f00feba0314527a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:35:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bee42b4af9474596aceb7216dea8ef28 : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66aaa7bae60a4f259680b79d293b7537 : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e6e0c00ccd2ff20b64449f8198a2911bf8f8fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 19:46:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/097ddad831444ee793a050dab0728af2 : SUCCESS in 52m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df2fa391f1d74590b8c490e8b342a535 : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/118bda6693074ab8bd8d9e4458daaff8 : TIMED_OUT in 2h 50m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25e445bd6d9d48f49c6ff74db6a0fc17 : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7cdfd6a3a9204fbb85bd235d6e580bcf : SUCCESS in 1h 51m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39689d4588104aa287107821e907983e : SUCCESS in 24m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/510e45c51df54516907e28fcfc4a5e93 : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926a68fecfdb402f84c3ecf76ca97d66 : SUCCESS in 14m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/903eab752f964b92961fcf522a9c75d4 : SUCCESS in 32m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e511bddd794b4db7935ca36ca866bc0e : SUCCESS in 32m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38fa884cb6d2415a9aca99dfba48479a : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff874b9d6b094305b9346457b3264d4c : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d53305f8bf6446409eaa7a62895d8d93 : SUCCESS in 1h 05m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab35f5a2238c4a428fbe366b13e51a5d : SUCCESS in 1h 23m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/211f064351e84f76ae1c3026e13617c8 : SUCCESS in 1h 46m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/872f4fa3ca2a46289cefde093aee1698 : SUCCESS in 1h 06m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aeefb79731fc4cb2a373d2a95256c817 : SUCCESS in 1h 28m 04s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1277c7fe8dea4674a64d91b6eac50a39 : SUCCESS in 1h 29m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e86374161494c2da7cdf9ef54159ec6 : SUCCESS in 1h 32m 46s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7b35bc543084d3186d533f4b514d574 : SUCCESS in 1h 36m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c2b70ea26a84c94bb6c87b8fdb4586c : SUCCESS in 59m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/729b29819d844e1d9702466fc70e1c66 : SUCCESS in 1h 08m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/47492af725024f6e8d4cd573a8ff5fab : SUCCESS in 32m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50009f2b257244f0a8ce5a51afe8e11b : SUCCESS in 1h 28m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7416f4d7349e48cab91ecdb342769cb7 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d5cd35b7b809810331828fe2fae88771092962d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-31 20:09:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"62cc76b074ba56b7c3e1c0c763f528700fef03e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 21:37:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/813dbdd08e694835be71083410f7ca3c : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d34a431930a4f93b495e741c53e860f : SUCCESS in 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fdb8cf5692824b11730196f8dfdaf539ac0b0536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 23:29:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/72aa39b7767645ffbd1f189a8e271d20 : SUCCESS in 48m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/faa1f19d2ae248788d8bb000de03fdc6 : RETRY_LIMIT in 1h 25m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/500b29fa935147229aeb87e842bc5513 : RETRY_LIMIT in 1h 14m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/43cc01a5431f4126b29831e166a48226 : SUCCESS in 1h 14m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2f1e11a4e7304fdc95d3d8fec695536e : RETRY_LIMIT in 1h 27m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4d1d5b03244d96aafd2bd9c715ba32 : SUCCESS in 26m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd5a94e610854cbfbd8b4ec6f08d8355 : SUCCESS in 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93bbe4413d444ab1a04482bbd5c5a8e1 : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/469bbbc7e9b646948300f296731c969a : SUCCESS in 26m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bebbbad4b7fb4fa7bd611e4356bd4b03 : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f05b1f4f9d4044b080658023f20c5ddd : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5ddb64ca5848c6ad19f46af8855ca3 : SUCCESS in 13m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/702e71370e05480c95d5491b93af8b4d : SUCCESS in 1h 01m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c40e084be2ac4b83a3ec28d8387dc5d9 : RETRY_LIMIT in 1h 22m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/be22c95fbc6d4223bd9bdb21000b9c7b : RETRY_LIMIT in 1h 30m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c0aeb8a11c944b489c4b1c59cfd8606 : RETRY_LIMIT in 52m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd23b2f55f40491b8968f29d8de8c5d0 : RETRY_LIMIT in 1h 10m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86451bb13f05458b8ec7f528586d46f1 : RETRY_LIMIT in 1h 26m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1ea4f03623fe4ed9ab054b7fb2b9dd17 : FAILURE in 30m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07a5deaadc1a4ae1878a6800310c1ec8 : RETRY_LIMIT in 1h 05m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/708e73c707704ea68b22a4508cfc252f : SUCCESS in 56m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/828213cbba904448b5159f9b6dcef683 : POST_FAILURE in 1h 47m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b2afd7e71c8a44f2b77f08911563ef41 : SUCCESS in 44m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9ab342ff610f49bf966972daabe1f1ce : RETRY_LIMIT in 1h 22m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/19a909c9b1924407802d4f37bf3bc73b : RETRY_LIMIT in 1h 22m 14s","accounts_in_message":[],"_revision_number":3},{"id":"59fdb18b02d7e7268a7178482349256dfb1f53d6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-01 05:22:37.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/784013/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b1d11a2cbb65dfb0f3c7e0eacd0a85f0359bc50f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-01 09:42:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/13/784013/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a8cdc11fc07d993ef841b5c90e06a2524ff63002","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-01 18:08:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8d79286ec5cd90e00d3d61a211641a8693b69a7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:14:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acb1f549af114dbbad1a7e925dfc7ec1 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b403ea70cbb240e6a56825ef27245c2b : SUCCESS in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa356fca038506980f8600d13d2b3162c32b08f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 20:51:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0113d00364144a07b48973c7eeba950a : SUCCESS in 1h 12m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ff661290db024426ba077e8f207524d3 : SUCCESS in 2h 27m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5f5c538a77fc4557be0ce78b477abf55 : SUCCESS in 2h 16m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0098e713212643288d6f568dadfad35d : SUCCESS in 1h 48m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3752d2cc6d5146629c8471608ae7dd10 : SUCCESS in 2h 35m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb6bc3b547264057b1284c33f423e047 : SUCCESS in 43m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fde83070ddde4b18913253cdd8a1058e : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19729b29fa9465a8419ea206bd8b936 : SUCCESS in 14m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4547848adb4643fa81f132c0174d5b05 : SUCCESS in 31m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b74430483a84d2eb3fd88a32daefd01 : SUCCESS in 18m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a1fee2aff244da3b4103a2002165a99 : SUCCESS in 31m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ff279e55184310953843362c13b2ee : SUCCESS in 12m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2135d039e41446768755206912f712ab : SUCCESS in 1h 05m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6530ff14330949089b3f76aea6263d6e : SUCCESS in 1h 42m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/97ed5cd2b52148138c28c8873f9abce3 : SUCCESS in 1h 58m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce746cf029104adaa81e3656a181e23e : SUCCESS in 1h 15m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56edb9b645c84fb2937ed6dbab72f8c5 : SUCCESS in 1h 22m 28s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/27a5632f17ce49c0933e28329eabd673 : SUCCESS in 1h 18m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f4192d7ea6024eff9ec2128a3bc46888 : SUCCESS in 1h 18m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fffb6551b54c4013b5099750e429be30 : SUCCESS in 1h 42m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0afe735a3e664b4aabfe3816d8d86a08 : SUCCESS in 48m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2082b3401d845189dda02575e223636 : SUCCESS in 57m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ec7a6cb215db4a34bbcd196b91ab23b1 : SUCCESS in 49m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/afc88f0ee53d4f0ca6876e0d8ee2ff44 : SUCCESS in 1h 22m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48d6b83325924ec78ef7f26d0ff75443 : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":3},{"id":"890c39662033ba4633fc0246ffe470fe161f71bc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-02 07:30:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1e13b74a1730162c4bb3e0ea54eacd6f71f6b577","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-06 10:22:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a5250f28dc97752f7819eed858d5e8980679355e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:23:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ddfd5d01c3cf1f48b8efcb878d2353a8e5d0d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:29:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/92daab3bc2234255a75f4a580697617a : SUCCESS in 54m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d310fd0ee2644663b83fd32f2acdf6b6 : SUCCESS in 1h 50m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9e4d7c7f9582481a8f9101e84e263ffc : SUCCESS in 1h 30m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8e638c04c6554435baa646e7b7a3605f : SUCCESS in 1h 43m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd3f7c59b25b4f9d9e9a5b6f01f8a5eb : SUCCESS in 1h 25m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/860b645a01484651bca22108b5e3dddb : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4636fd9e2c4f45749f9352a22ec40e89 : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac32fb7e0fd44fa8b1855362ea483fe : SUCCESS in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f7abf9cd4a4becb4c9e685212b5817 : SUCCESS in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8f2ada661b402fb0f1a966cff8098a : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08a8965c388249758e1f256e973a1da4 : SUCCESS in 44m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c123008dca614e4388281be381cdf83b : SUCCESS in 1h 17m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4325d13c64a245638384b08906fb538b : SUCCESS in 1h 18m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9da1d50debaf42eeb0b4991f6ca3add5 : SUCCESS in 1h 19m 46s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9a1912b964845deb9eadbdf96aa3003 : SUCCESS in 1h 33m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edf9d231ac014753a17848373a76e5b9 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/29d105740aba48d9b64c0ca4e1344505 : SUCCESS in 1h 10m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62123da4eae0481eb67484596bf76a07 : SUCCESS in 41m 09s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f71c73a908774378a4e5e55984556409 : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":3},{"id":"47c3d0e59317e64ee23a93872ecb3a1559d490fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:29:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ea08f598334cc643f895d500a75469211930687f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:32:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fd3afddb9f4c8dbf454c720da5b8f5 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"becb42b92e0b838d085a05e68ace71b864c71b24","revisions":{"84a625513aed9286fb24294fac95169cdbad4e5c":{"kind":"REWORK","_number":1,"created":"2021-03-31 07:12:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/784013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/784013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/784013/1"}}},"commit":{"parents":[{"commit":"493286511d227c7ee8189838199c673110af1ee8","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493286511d227c7ee8189838199c673110af1ee8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 07:11:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 07:12:30.000000000","tz":0},"subject":"Migrate \"ovs-ofctl\" to oslo.privsep","message":"Migrate \"ovs-ofctl\" to oslo.privsep\n\nStory: #2007686\nTask: #42184\n\nChange-Id: Ifc7ec227d4084331948c42cd99d96ad10c09a6f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84a625513aed9286fb24294fac95169cdbad4e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84a625513aed9286fb24294fac95169cdbad4e5c"}]},"branch":"refs/heads/master"},"99375f82c11a77528067020c91cf3ae15ae3c394":{"kind":"REWORK","_number":2,"created":"2021-03-31 14:07:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/784013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/784013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/784013/2"}}},"commit":{"parents":[{"commit":"493286511d227c7ee8189838199c673110af1ee8","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493286511d227c7ee8189838199c673110af1ee8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 07:11:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 14:07:10.000000000","tz":0},"subject":"Migrate \"ovs-ofctl\" to oslo.privsep","message":"Migrate \"ovs-ofctl\" to oslo.privsep\n\nStory: #2007686\nTask: #42184\n\nChange-Id: Ifc7ec227d4084331948c42cd99d96ad10c09a6f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99375f82c11a77528067020c91cf3ae15ae3c394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99375f82c11a77528067020c91cf3ae15ae3c394"}]},"branch":"refs/heads/master"},"becb42b92e0b838d085a05e68ace71b864c71b24":{"kind":"REWORK","_number":3,"created":"2021-03-31 16:44:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/784013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/784013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/784013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/784013/3"}}},"commit":{"parents":[{"commit":"493286511d227c7ee8189838199c673110af1ee8","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493286511d227c7ee8189838199c673110af1ee8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 07:11:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-31 16:44:30.000000000","tz":0},"subject":"Migrate \"ovs-ofctl\" to oslo.privsep","message":"Migrate \"ovs-ofctl\" to oslo.privsep\n\nStory: #2007686\nTask: #42184\n\nChange-Id: Ifc7ec227d4084331948c42cd99d96ad10c09a6f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/becb42b92e0b838d085a05e68ace71b864c71b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/becb42b92e0b838d085a05e68ace71b864c71b24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
