)]}'
{"id":"openstack%2Fneutron~784356","triplet_id":"openstack%2Fneutron~master~I532d0dc67ca79814fa1a99a4a7b0d2dd59f008b6","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I532d0dc67ca79814fa1a99a4a7b0d2dd59f008b6","subject":"Add tests for metering API\u0027s new policy rules","status":"MERGED","created":"2021-04-01 09:47:44.000000000","updated":"2021-05-27 17:32:00.000000000","submitted":"2021-05-27 17:28:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"784356-secure-rbac","meta_rev_id":"fae7d103d9fb58e2e01f8d73e9a04dd447e8e23d","_number":784356,"virtual_id_number":784356,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-04-13 01:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 17:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-04-23 14:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 10:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-05-27 15:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-05-27 15:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 10:48:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 10:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 21:00:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 16:29:25.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-09 13:20:29.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-04-23 14:02:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-25 10:18:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-05-27 15:12:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c41c6810ab7c487aa9e265ae16970fbbaa4d936b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-01 09:47:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c6f4e4cad77d2dd95d9dc1b2975b32e548bc39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ee0e191a1484ce0b89f10f21b931043 : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6aaa6c28ec04473e94b3940855e8b00e : SUCCESS in 48m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef9918fc291b47b89c26b0de47b1c4593f32202e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:54:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56eaa1ff838c4977a973c2c50755422f : SUCCESS in 36m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/689926c36f8e4a9eb125195d7ce2aeb3 : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e1b73a8fdc4a71955375846f79a30a : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/587fc24351d84384865bdf332eb52405 : SUCCESS in 26m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5b33563f245418e8d3a52cac484310a : TIMED_OUT in 1h 00m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c958296baaf44fdb2d064610cb05f0b : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a50e0ab5bd40b796c793fee20f7946 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a189102b317b099c227f2b6da028f64f0faf45cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-01 21:00:53.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/56/784356/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"50fa83196caa9be9066c3a42f43e3bae0d41ac67","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-08 13:50:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa5a51d5010e869d35d7517c3dea3bacf652f9b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 14:50:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63bb630c914f486a8fe314d99f84bcb4 : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e271a569a0aa41618db28cb9850357ff : SUCCESS in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd39460f70de61ffb709de27c437d9968664f2bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 15:29:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff67ce85d76434488f8354f9a7d8824 : SUCCESS in 47m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/429f7960bf8a4c31b19be3b06f4d5875 : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502352c12cde44ee8d3bdb837e9d8edf : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95832679b6434dda9fbdaa3ee45a6eb5 : SUCCESS in 30m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b30bd72041864543a9f472d7f8af02b1 : SUCCESS in 32m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8fb8c6344404dfb94bd8c9b9e4a5e31 : SUCCESS in 24m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ec402c013b4e739d769473bbc4a68d : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bdf0b3c85da57b417ed30f7b1e5df6f78d0ee48c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-08 16:29:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/784356/1 : SUCCESS in 1h 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b222837f1ab549fa52629b8f07b3ea468ee1ba0c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-09 13:20:29.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nJust a minor question on project ID attributes and if they overlap with metering labels.\n\nThanks, Slawek!","accounts_in_message":[],"_revision_number":1},{"id":"4698422066cc56cfc1dd661d8d76edc8cc8f696a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-12 13:34:51.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"047ac586fd61107106e14c63e270779b2f8106b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-12 13:34:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"947919de00023da82d03693458ac5703d038f37c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:29:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded767c652024a0eb4fba029b00a3675 : SUCCESS in 44m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15fa69f21c1341b69a707418b58d93b7 : SUCCESS in 35m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4901a58772884ad7965462f49aae69a1 : SUCCESS in 15m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba820537bbd74b70a5a32b52fcd77152 : SUCCESS in 24m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2175b73f1aa747adb2b5cd135962c849 : SUCCESS in 32m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c1ba4a3ba324c63b6ce89a606447d61 : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51849c8dfca44a3889672ecec758bf01 : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"618348bce5cd827057e0381fc09874a56004c69c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:46:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16f983616265acfd05f4e7527a23919dbd80c935","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-12 18:30:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a88d5b7c417d5c219a42779bc56b0d478122eee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-13 01:39:22.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/56/784356/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"4b6c26574dab7ae2e5a8d0a72e4f77ae83fdd678","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-23 14:02:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f507421a408e981885beb71c492070e91405f73","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-25 10:18:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e57370474e72d28da6d64c5e0fb081001987cb9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-27 15:12:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d983bc926402be761461ecf0e06c6ba723ff47e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:14:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32d94e996e0a43ee284cfb1659b38059b66d6033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:28:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/330d16bd6cf4485e8ace8b81d408f6c2 : SUCCESS in 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b952886464b4bac8e236df2520ad613 : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0ed6c5ddf2a4e6fa6b6a42a072c1cc8 : SUCCESS in 24m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a06ae910cf412ea4f4856c031e7798 : SUCCESS in 33m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12a7d13c3854beca7547e73f1025c0b : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"da59a7bb06984c365bbe57688d3279364fb7b3b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:28:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fae7d103d9fb58e2e01f8d73e9a04dd447e8e23d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:32:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714416626b1e4bdfab55ee8527829f1e : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b331b31f21ed808600cd3ee7e2e85f0c2896672","revisions":{"0da6a52294071d8c756cafb554482ecd3a5c620e":{"kind":"REWORK","_number":1,"created":"2021-04-01 09:47:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/784356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/784356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/784356/1"}}},"commit":{"parents":[{"commit":"07c0cc4a6691c35e517d0efa4e922d1804515a13","subject":"Merge \"Unit tests for auto_allocated_topology API policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07c0cc4a6691c35e517d0efa4e922d1804515a13"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 09:46:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 09:46:41.000000000","tz":120},"subject":"Add tests for metering API\u0027s new policy rules","message":"Add tests for metering API\u0027s new policy rules\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: I532d0dc67ca79814fa1a99a4a7b0d2dd59f008b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0da6a52294071d8c756cafb554482ecd3a5c620e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0da6a52294071d8c756cafb554482ecd3a5c620e"}]},"branch":"refs/heads/master"},"3b331b31f21ed808600cd3ee7e2e85f0c2896672":{"kind":"REWORK","_number":2,"created":"2021-04-12 13:34:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/784356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/784356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/784356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/784356/2"}}},"commit":{"parents":[{"commit":"07c0cc4a6691c35e517d0efa4e922d1804515a13","subject":"Merge \"Unit tests for auto_allocated_topology API policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07c0cc4a6691c35e517d0efa4e922d1804515a13"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 09:46:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-12 13:34:42.000000000","tz":120},"subject":"Add tests for metering API\u0027s new policy rules","message":"Add tests for metering API\u0027s new policy rules\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: I532d0dc67ca79814fa1a99a4a7b0d2dd59f008b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b331b31f21ed808600cd3ee7e2e85f0c2896672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b331b31f21ed808600cd3ee7e2e85f0c2896672"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
