)]}'
{"id":"openstack%2Fneutron~784412","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I65130b299541dd4559e2d758fb4ab9d68c6f2cfa","project":"openstack/neutron","branch":"stable/wallaby","topic":"secure-rbac-stable/wallaby","hashtags":[],"change_id":"I65130b299541dd4559e2d758fb4ab9d68c6f2cfa","subject":"Fix create_port new API policy roles","status":"MERGED","created":"2021-04-02 07:33:25.000000000","updated":"2021-04-03 02:38:12.000000000","submitted":"2021-04-03 02:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784412-secure-rbac-stable/wallaby","cherry_pick_of_change":784422,"cherry_pick_of_patch_set":1,"meta_rev_id":"66bd79693692dbc88ea652daf55331d9cdf8222f","_number":784412,"virtual_id_number":784412,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-03 02:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-04-02 07:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-04-03 00:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-04-03 00:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 07:55:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-04-02 08:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 08:49:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 10:05:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-03 00:08:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"407d490bcb8292875c01c5aac311d9dc5658a001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-02 07:33:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f6f426717ec8708701bbb21f2dfaeb3978020895","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-02 07:55:28.000000000","message":"Patch Set 1: Code-Review+2\n\nIn the list of secure-rbac rc2 fixes","accounts_in_message":[],"_revision_number":1},{"id":"9f216e5fa2001b3af9f3bf1cc9b73aa5cffa25f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:14:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6da7b8b7085f4d079171f3354db1bace : SUCCESS in 35m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb0e1faa50154285bc8e62628739a2f8 : SUCCESS in 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ecae89a7baaac684ceac503997533ae3513f10b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 08:49:15.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/784412/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/784412/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"396b7342e5799b83d7dd2d26db3f0727785e66f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 10:05:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c1c9701dde394dcc9d2d2d785cd49dbc : SUCCESS in 43m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b705a293fa304827b617ece71992d3dd : SUCCESS in 1h 58m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/22372f17551745719a8f495bd1653921 : SUCCESS in 2h 26m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee0b25844a9a480492aa2e4c11fce18e : SUCCESS in 1h 50m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6edceea5d10246f1943078b761e57460 : SUCCESS in 2h 14m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f02521e57842e292797166bffb75bb : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdf10e5cfc3245fa873b940c3a2ec09e : SUCCESS in 53m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bcfd1f10b7462b9280e76d9c24d58d : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1991828a1f644bcbbfad9bfca6dd8c16 : SUCCESS in 47m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4413f1d8c8834270bc6ffc2a133a6e07 : SUCCESS in 27m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b25749964fa347a28710eaa2e1cf536a : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24170f4a2e2e4bb6b8d841f8e2952c99 : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1cf874c69bb4a0f86f1aeedd93f455b : SUCCESS in 53m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e16d7b15e9a545c69bed62c74f90d7bf : SUCCESS in 1h 24m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e871b78cf60436ebf8dba934998a14c : SUCCESS in 1h 37m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89a69d66deec446b85841378cc6df646 : SUCCESS in 1h 05m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b247ac1c3f64dc1b6173b348346edb5 : SUCCESS in 1h 20m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2ca39b7bbd34e9897630865e25a6b2e : SUCCESS in 1h 26m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8c109048fd543949a0b12e93b122533 : SUCCESS in 1h 41m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2f84e23ecc741f79e51a167160395cd : SUCCESS in 1h 37m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e64e96615e9843fb84f2b96e3fc6135d : SUCCESS in 1h 05m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa3b696f279341b9805bee7e6fbbd3df : SUCCESS in 1h 23m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fd7aaefa862b484fb2bfda7176d292e4 : SUCCESS in 37m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbd9e2f9122b400b86c33ded755e31ca : TIMED_OUT in 2h 13m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/edb71b70a79d4196a29b294f3d19f726 : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b53f2a14c4b728c6c838a9e3ef3fc269c69e9a0a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-04-03 00:08:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d2c9cf882ee9a6702cc614f6964515b083ae0bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 00:08:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64fd860a0a73284d7bcdcf1d69c57b01c153dadf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:35:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ec7c62ff15b643ff8154e800dba65098 : SUCCESS in 45m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b5634c41c32d47bebcdb5cc0d3aa3b76 : SUCCESS in 2h 12m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6c3940bab33a4c3ca851bbd3b4e10519 : SUCCESS in 1h 42m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/00216d9921e541a2a826be8b9f28cbfd : SUCCESS in 2h 05m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c44f2b30071c40a78221621841642d32 : SUCCESS in 1h 02m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/918a748db215404fa65f7712d5a10225 : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6509bebbaad4036946d814068229edf : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af30bc1a5a0c4921aba142e771da3d39 : SUCCESS in 23m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5467a355bc047aba6acceae605b23a8 : SUCCESS in 32m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da6063078124fa8942146abb6711d47 : SUCCESS in 9m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18b1ee507db94edd9559c665c7f0f6b4 : SUCCESS in 47m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/146e5786c8ee4f68a186cc459c0bb788 : SUCCESS in 1h 35m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/19968dd55e3c444c864f1f5b5bed566e : SUCCESS in 1h 16m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/034a17a8eb03475b80f52aaa8ff948e4 : SUCCESS in 58m 26s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e21106242aa4b838749d80662a8da04 : SUCCESS in 1h 27m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62434b232917450f91f22e82d9084385 : SUCCESS in 55m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8338842464a04b74b5d431b1a683b462 : SUCCESS in 59m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/123679acb95140acb685566b50ed0f56 : SUCCESS in 31m 59s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4cabaa52c3f749fba94c78d13ae8355a : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5e314b817d2e278e061da18c26e139216727d058","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:35:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"66bd79693692dbc88ea652daf55331d9cdf8222f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4467b75421194af0aad40ac39d08f5bc : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a25b9f74d604b042d5fdbd07be401f375b0b71f3","revisions":{"a25b9f74d604b042d5fdbd07be401f375b0b71f3":{"kind":"REWORK","_number":1,"created":"2021-04-02 07:33:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/784412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/784412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/784412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/784412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/784412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/784412/1"}}},"commit":{"parents":[{"commit":"80e1d8269b4963a905ecadaa282f5f5f9e97cf4e","subject":"Merge \"Modify create address scope policy rule and add UT for that API\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e1d8269b4963a905ecadaa282f5f5f9e97cf4e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 14:10:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-02 07:33:25.000000000","tz":0},"subject":"Fix create_port new API policy roles","message":"Fix create_port new API policy roles\n\nSome rules for create port, using new personas policies allowed to\ncreate port PROJECT_MEMBER persona but not SYSTEM_ADMIN. So this patch\nfixes it by adding SYSTEM_ADMIN to that rules too.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: I65130b299541dd4559e2d758fb4ab9d68c6f2cfa\n(cherry picked from commit 2bc1572740d9eb6cff8c8d893c5a24ab10ed37b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a25b9f74d604b042d5fdbd07be401f375b0b71f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a25b9f74d604b042d5fdbd07be401f375b0b71f3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
