)]}'
{"id":"openstack%2Fneutron~784552","triplet_id":"openstack%2Fneutron~master~Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab","project":"openstack/neutron","branch":"master","topic":"test_py38_timeout","hashtags":[],"change_id":"Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab","subject":"[DNM] Testing openstack-tox-py38 timeouts","status":"ABANDONED","created":"2021-04-02 09:57:31.000000000","updated":"2021-04-10 09:33:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44980c6df6d29e75e1bedc2c6b5e1cf6c796a76b","_number":784552,"virtual_id_number":784552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-04-05 15:20:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-05 12:38:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2021-04-05 13:44:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 10:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-02 13:27:19.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-02 17:47:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-05 12:20:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 10:12:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"2f42638038c3bca64cae6e1106fbc9387d0770a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-02 09:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9886b3ca82fa47a69bb0f6aba9e789919c0c74a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 10:41:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/12355d1d7f36403a90aa3b2c81e4cf8c : TIMED_OUT in 41m 14s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/87cf0055032c46d7b2d2fcc0f8d3dc8f : SUCCESS in 30m 35s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/ee497a2211a0423cb9d8d0f8ddb0ac9e : SUCCESS in 38m 28s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/f2369566c1f64870a8701c2d2def02ad : SUCCESS in 25m 25s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/22d89c6a027847acb6ce994a2c44612e : TIMED_OUT in 40m 52s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/6f6031be36084c129ad7585d1fc2fdb4 : TIMED_OUT in 40m 54s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/8db97f4afbda4c78b9835e648bb22ff1 : SUCCESS in 25m 28s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/348cd05c657343f2bedb4432fa8b4b96 : SUCCESS in 28m 43s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/1cd8271b6c904a2f9cd02d430a987e17 : SUCCESS in 36m 04s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/6e88883816a0433aa3b5318bcfad01cf : SUCCESS in 35m 19s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/4b3a890500a04bad813d975f345a916d : SUCCESS in 32m 45s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/5689008784ae4d04a792171354f805c8 : SUCCESS in 30m 47s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/2c92cd14bf57495db8cc9abb45af1609 : SUCCESS in 22m 24s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/16077b609ecf4e9b8f21ed8a81faa2a9 : SUCCESS in 30m 20s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/f4a55a79c08e411ba4ca5ce0e7480f8f : SUCCESS in 25m 38s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/06dd9c3fb6824b6eb7e1e55344a96b33 : SUCCESS in 22m 24s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/76508660e70243ed9696fab9cb6b903b : SUCCESS in 29m 22s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/1005764972e441d89185f51544f43bf9 : SUCCESS in 32m 10s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/a976a0fa1e7247ee9a101eaba56ed2f0 : SUCCESS in 24m 29s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/3a8940002c40431196c1d6e467be6361 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"455792c257dab62eb43f9793b6994c67f81f3b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-02 11:28:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"def0af0efbdce7cd3f49a667862ab64b5e41050b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 12:20:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/69776c81037341d6ac456ff1f4ce7bdd : SUCCESS in 23m 56s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/142a2ed694524639a30c0817df3ed315 : SUCCESS in 22m 49s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/c70941ede539445496b111fb3984582f : SUCCESS in 24m 01s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/ce5267e102144cdb8bf8932855030cf4 : SUCCESS in 22m 17s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/adf4da50526845009ca511cd6e4e1bd9 : SUCCESS in 22m 31s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/fc5ae1f928f6489daf38409ac9da3fcb : SUCCESS in 22m 47s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/fb84cb0e57164207ac4fd4c176da3ade : SUCCESS in 23m 07s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/2bc0310f3f3f43a3a5bfe4cb2bc61f07 : SUCCESS in 22m 52s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/d9c79507e11b43cda9127a3868cf7a15 : TIMED_OUT in 41m 04s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/0dd52563272a4504b664a681e2e2dcd5 : FAILURE in 33m 02s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/96463d844bcc4335b01a15fc7e679835 : SUCCESS in 28m 11s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/6e1c0d9ca493486088f1e40ed8dc3898 : SUCCESS in 30m 25s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/b22589e0ea764a95b3b6f6f2f461d23c : SUCCESS in 29m 30s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/93729cc98b9c4cfca2a57013ff07c37e : SUCCESS in 27m 38s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/e7f5960f292c418a816e9cc33cb1eb98 : TIMED_OUT in 40m 54s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/46dc4b5fe0ae48548201bd03037e3449 : SUCCESS in 33m 30s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/7822260ebcd0423abbaf502e249f53c7 : SUCCESS in 17m 10s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/5fa26a39d0c84a118e5a2c19d0ed7270 : TIMED_OUT in 40m 53s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/472383184439424a9b701309f9634f60 : SUCCESS in 30m 22s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/bbf5de42283e41e49e2b033f5770faeb : TIMED_OUT in 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c3aa2e8394bb67d0cd0f2e0059c7d0f78bd7cfe8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-02 13:27:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/784552/2 : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d01013e8ee171f3037b177c677fb89b4b9adeb17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-02 17:46:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9e2706951efcdbcbd847cc4d81af2241e02867d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 17:46:25.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"93db2fbd33e1cf64474b9849740328e44c31882b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 17:46:30.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"0b693a0eb532ac5e2710cae6fcde196d87271f18","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 17:47:38.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"a22ce198e1439fbcd65f126df40229919cc9c92d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-04 14:58:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"df1bcb5bee5882b41cee6123dc754d61a185513c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 14:58:16.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"509c9b3beac2f24994ba8145331b96345cc6a14e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 14:58:19.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"5a5664f5931f674d86188690a1e6e48dd9396b30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 07:12:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad3e8fe34cf728b0686647b3ba9c9c615bc04084","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-05 07:14:08.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"f1ef33d8ac4fd3fa173d8b0f00056f6abaf4b46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:14:33.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"33bc00182846fa60501d0aaf65050aae5cc9b060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:14:38.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"0122819b5f8e3159aa08b98293a75542d7a7c982","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 07:22:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d9b72484301805d6f32f5751d0a55d9590b49b1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:22:42.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"b407ac929208cb82d8bf70cea3353429181f252d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:22:45.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"bb16f19a9c43f8b3969932970fd40bab556f83d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 07:27:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3ad7881c3f83d9280bd5e264471460c0fc2df486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:37:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/2b2d8abfee804f6fb9bc5d8ea3238419 : FAILURE in 4m 06s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/b9a1f1adeb6f4330af2928f3f66eb658 : FAILURE in 3m 50s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/76b392cebcc94c7ca850c5a236287a37 : FAILURE in 3m 53s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/0576d128b7e14d358e5ab381c6bbc91f : FAILURE in 3m 45s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/78a13f3aa4a84925b7b00d758172253e : FAILURE in 4m 05s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/f8910d5a93b7488eae7be9404ebfdb36 : FAILURE in 3m 43s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/d1ac666924554cf38aa85334ccd50a7c : FAILURE in 3m 59s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/36920256dea34e91a8b9ce739351214a : FAILURE in 3m 53s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/7e71927d4e4f4a2484242204371df91f : FAILURE in 4m 03s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/578f67f105a849c6a39ab27a24d8a7fe : FAILURE in 3m 59s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/ba72e7d8e4d545b99a3203331570dd6e : FAILURE in 3m 52s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/9452670e60444ae992fe1a1966917a0f : FAILURE in 3m 52s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/854df734b331437db98d1d630f544012 : FAILURE in 3m 43s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/b94dfeb485ef4e129eecf5e28d0eb8eb : FAILURE in 3m 56s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/866a88079f224d0d8deda89fdef0867b : FAILURE in 3m 46s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/f9e09287b4e9432fb3455182c54bb2fe : FAILURE in 3m 43s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/716e8d6497044df5a3fce872fe50ab54 : FAILURE in 4m 04s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/46631154847a473a857cf932949e3fd0 : FAILURE in 3m 44s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/62e93f2919d640fb828683b8023b94a5 : FAILURE in 3m 51s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/d1d7b3de095a4104b9a395e9556c2854 : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"953355e60b98bf20097a507a27b61566552f78d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 07:41:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2cf7d5331fae950bce8b073e126bdfae216748ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 07:52:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/ab0cab9d534840688b4511cd9c41c599 : FAILURE in 3m 48s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/6447e797295244bf9f6602075d4b9e6d : FAILURE in 3m 52s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/0314e5b83bc44c419c4a177a942a7fd8 : FAILURE in 4m 11s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/33fcb121adc3475698e89786eb3a4058 : FAILURE in 3m 55s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/2922023ac940406785eeac816daaa023 : FAILURE in 3m 58s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/bd1b5ce716184232a996a0a426dc30cc : FAILURE in 4m 22s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/8a2ec79888b648429a2fc243748562c5 : FAILURE in 3m 59s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/2ac9c761adca46b4a710d6219df24338 : FAILURE in 3m 56s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/539ea2cb3fb2491898276eb1837157cd : FAILURE in 4m 02s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/14b2513493c54713abf2e29a62dea600 : FAILURE in 4m 03s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/d7d39915772449048d3e93dd040cf376 : FAILURE in 3m 50s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/c708c7829b554cc98f1df61edb4ad32d : FAILURE in 4m 22s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/9fc1e6885c984bb3bbef7a9dbcdc93f7 : FAILURE in 4m 31s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/e725d2458a324fd18cc5dfa517cb1179 : FAILURE in 3m 50s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/c77d75157a3b4fb9977f2a325406172a : FAILURE in 4m 11s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/4bbfd69657434bab923fe4ae9ae490b1 : FAILURE in 3m 47s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/bb7b6fb59f8847289866468010dcc1ca : FAILURE in 4m 03s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/91b736152c1d4565a7d20deb442faba2 : FAILURE in 3m 58s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/eb7887c24a01455f8ec262c80c44f150 : FAILURE in 3m 50s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/e34b2d683f884d90ad063aceff250068 : FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2ffe34ccf10cf653ef356e17c94fda745b908a8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 08:26:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c89130337d9ef891f058d51f05b5d074e66e1e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 09:11:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/550e359018a746b8b9bcaa44f2d8f220 : SUCCESS in 24m 20s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/56d39c7505544a4b98a324e6d9b9cb4a : SUCCESS in 31m 24s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/30a194a3e87b4ba5b37c932bb1755eb0 : SUCCESS in 30m 29s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/91cfaa22885a46578498c71388b89d58 : SUCCESS in 28m 23s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/c9ab7d9a2d3542a89224dd4a415ad6e4 : TIMED_OUT in 41m 09s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/501a94f10f6a4869b6aa0eeb2653e219 : TIMED_OUT in 41m 04s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/27b7f49ab4ff40f8b74a1f591fd68fe4 : TIMED_OUT in 41m 07s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/aca01292f9bc4ca5819373753dea0910 : TIMED_OUT in 41m 07s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/1e766478ee194ebd9394352094cd6efb : SUCCESS in 31m 04s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/aacc50ca3e1e4fa3b657bda8701e450b : SUCCESS in 34m 33s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/e0753e713e6348778a09b051318b2e40 : SUCCESS in 30m 42s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/601a3001510d4031948119d5e1b3d20b : SUCCESS in 27m 33s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/1b0304895f2d42ab909d55587e7e06c3 : TIMED_OUT in 40m 50s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/c1c104f006054d08acc8a256d7ef9e20 : TIMED_OUT in 41m 02s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/32d92c87dfc747ed92f256a1466748a4 : TIMED_OUT in 40m 54s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/b6dc8a4042494cf2a7b59186f3622a59 : TIMED_OUT in 41m 07s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/ae8e171c56ba44f1bdc591ca01a39290 : SUCCESS in 29m 58s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/efa292535520489d9410e8142473683a : SUCCESS in 41m 09s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/90ffd2bb9f18474c9c76220cac33eef5 : SUCCESS in 24m 26s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/5c00d8b4990c482da1bb21fb84ec5a3f : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":7},{"id":"7271bd6c021e7c74805caf70876abd15aeb4374f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 09:47:16.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"a814b1362d87196f455953bcd12684c3248de83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 10:29:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/aeaa48fbde8c43b7809229ab6b766c99 : SUCCESS in 28m 26s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/4955bf4d96a94a53a653629e5a8a7f03 : SUCCESS in 26m 37s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/da50ab83919a443a910b9c3beceb50d3 : TIMED_OUT in 40m 47s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/2ba5e8bcf40d4f5cbdf401a01cef2d8a : SUCCESS in 23m 51s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/ae03f2087fc84e81a648d44a0ff91e6c : SUCCESS in 24m 06s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/6355d79623e1411a9146b135dbed25b8 : TIMED_OUT in 40m 59s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/8d780c5d9fd142c4872cda7fc4a1fdb6 : SUCCESS in 23m 28s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/a0e898c4461b4c348d194f48a178de3b : SUCCESS in 23m 59s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/52b9642bb69e482a9334910d88b8578e : SUCCESS in 23m 59s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/3fd289b6d55c4d2dba461146e3c45db5 : SUCCESS in 23m 58s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/e46419c2c7ef4c29ba86e9d3611549d9 : SUCCESS in 23m 25s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/735c0f5cc81a44d3b1758e116b377f75 : SUCCESS in 23m 37s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/4b01c98cc3d04a2a9ab36f1e26dababa : SUCCESS in 22m 17s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/77b52327dc95430683fbf855ad20f974 : SUCCESS in 22m 38s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/b80df8c8ffd34ce8b27b75fbc6573bf4 : SUCCESS in 22m 31s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/babe29fd62d5484a9f3f5f5d86a84af7 : SUCCESS in 22m 21s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/b102461797fd4457ab5fcff93a95f603 : SUCCESS in 22m 21s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/848a1a7808e54716aaf36d7fa63bd0e0 : SUCCESS in 30m 41s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/4ce5d2c4130648e78b85912b35f60593 : SUCCESS in 22m 23s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/e1feb03549fc49fb8ce355e771bafbe2 : SUCCESS in 27m 15s","accounts_in_message":[],"_revision_number":8},{"id":"e3d84ef5bcd219bfb9b46c68234c558a37a96b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 11:09:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2921d5e0e79515cd7f1869f739443e8a4f8a667c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 11:53:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/a2c4f9a34d154636abe5a24d181ab895 : SUCCESS in 37m 28s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/419d24864ed643d9a111d27e680caf8d : SUCCESS in 32m 20s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/8378a8f7836641b28c91d65a4eff0f19 : SUCCESS in 37m 41s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/2dcb78c0b9ae49e6a79e9d8582532f32 : SUCCESS in 29m 15s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/5151ff52d0814b4288ce66b5b8018962 : TIMED_OUT in 40m 58s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/8dc77a4e78ea4d07ab231ab9fdcfebb2 : SUCCESS in 39m 51s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/a63dbdcecb2742a7a6b3c7e4965a177c : SUCCESS in 35m 55s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/ec7e456b2d564902842b7685afae457d : SUCCESS in 34m 51s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/967ac2b9046e497b909069fc53c92146 : SUCCESS in 39m 40s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/363e78fa2a0d4bb783e2cdca80d5a999 : SUCCESS in 35m 52s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/d2fa3ecd4be847ada4102ff615cdd618 : SUCCESS in 38m 59s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/bfccd068cb9b44dcba724c93508b397e : SUCCESS in 35m 48s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/74a6fbf6488244c88a689b9cfb3659dd : TIMED_OUT in 40m 50s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/b0c2ef753d424e38840ac8e378c1faef : SUCCESS in 24m 21s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/ed0311a34aff4e609996f601b849600a : SUCCESS in 24m 46s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/9f5a09fbb63a4162b59cd31818245216 : SUCCESS in 24m 02s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/32ae3cc29d7a45859c9dac615e95fe1e : SUCCESS in 24m 04s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/75657f6426ac4b31a5897acbc2d2f957 : SUCCESS in 23m 39s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/6a498523ce5c40a4b00f855b6a6da954 : SUCCESS in 24m 04s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/bddd7cd49ad340d39175a63483131052 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":9},{"id":"f58b9eba60ddc629e1496cda33b198a523934f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 11:56:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9a3291bfea734d6d7681ba9930e859431dcff085","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-05 12:20:53.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/52/784552/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"387b85a9b37df2cf80511847529209a9f7dbcdeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 12:38:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/e07531cdc079429b9ef0ec6097b387ee : SUCCESS in 17m 42s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/51023e94353742f2ab6812684f0cdc7c : SUCCESS in 27m 18s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/95a6a8b0ccd848d6a575fff5342f8d73 : SUCCESS in 37m 46s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/987fa461930046fe9cd5bd6ee74b1d45 : SUCCESS in 28m 43s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/9ac80a8418db46388f429ac647d6ced0 : SUCCESS in 29m 43s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/d4c94597f62843ada7e07877b45cf211 : SUCCESS in 33m 01s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/a7c0342314304ffe958c65857763e4a5 : SUCCESS in 26m 52s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/74d8dda1daad4fc2a28b7be068664d8c : SUCCESS in 37m 23s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/06a592ba5512493b811270e76253c831 : SUCCESS in 37m 35s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/9875247864c74569aa3acac019fa551c : SUCCESS in 29m 45s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/180130f28a8e423cb7675d5ed0908a8f : SUCCESS in 30m 55s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/4cf0dd68c7af498ea65c61e03b9a9ee1 : SUCCESS in 28m 35s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/0361f6b85baf4183854b3fde657f8860 : SUCCESS in 28m 30s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/7b73b44085a04617b09815ffbf570025 : SUCCESS in 26m 20s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/8eac994baa56437b970f71531fc353f6 : SUCCESS in 29m 00s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/0985304a1dc945898fad9123d2457b03 : SUCCESS in 33m 07s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/dc1d790f847b45d484f5bfd638108fa7 : SUCCESS in 27m 08s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/2094725481de4499b08ef6aeb3bc93c3 : SUCCESS in 26m 49s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/c3b9803d4a824c53a00a15d3d3f131e9 : SUCCESS in 26m 43s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/b6732d9babcc42db8c57d18cacc47c4f : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":10},{"id":"0c63d810451e57b0fdbfb773267eb66f2d80dbe6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 12:44:49.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8a0709224c00059ad9a12732418ffcc66ee682c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 13:31:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38-01 https://zuul.opendev.org/t/openstack/build/aa1d57ecc5de483e979cc2b6c760ecba : SUCCESS in 27m 24s\n- openstack-tox-py38-02 https://zuul.opendev.org/t/openstack/build/3bf5cedd4a4248e5b311e9ffb6cc0fd6 : SUCCESS in 33m 06s\n- openstack-tox-py38-03 https://zuul.opendev.org/t/openstack/build/0f173707fc8b417a95cc14b81b61e7ee : SUCCESS in 35m 22s\n- openstack-tox-py38-04 https://zuul.opendev.org/t/openstack/build/c14608b7e389482dae67cf251e302278 : SUCCESS in 28m 29s\n- openstack-tox-py38-05 https://zuul.opendev.org/t/openstack/build/10c130f16c064b06ba8390a2c3bf4db2 : SUCCESS in 30m 05s\n- openstack-tox-py38-06 https://zuul.opendev.org/t/openstack/build/6565fa7ba9d7432dad89d35f7093246b : SUCCESS in 33m 42s\n- openstack-tox-py38-07 https://zuul.opendev.org/t/openstack/build/b73fd3a9e487434f90e0012f96a4c624 : SUCCESS in 39m 40s\n- openstack-tox-py38-08 https://zuul.opendev.org/t/openstack/build/f060dc510dcd4308a9d47b2730408256 : SUCCESS in 28m 08s\n- openstack-tox-py38-09 https://zuul.opendev.org/t/openstack/build/da2abf7aefc0443cbbb2b46e0498d428 : SUCCESS in 39m 43s\n- openstack-tox-py38-10 https://zuul.opendev.org/t/openstack/build/c25abfac7cda49ea9e17714ec3be7086 : SUCCESS in 26m 16s\n- openstack-tox-py38-11 https://zuul.opendev.org/t/openstack/build/1a9a03c0e4f1459284f5a5ff9c6baa2e : SUCCESS in 26m 17s\n- openstack-tox-py38-12 https://zuul.opendev.org/t/openstack/build/73378aaaffdd4e94954d73d65e82f09b : SUCCESS in 26m 17s\n- openstack-tox-py38-13 https://zuul.opendev.org/t/openstack/build/905a78cba71d45009920135a7d5dc496 : SUCCESS in 27m 05s\n- openstack-tox-py38-14 https://zuul.opendev.org/t/openstack/build/f7f0fc4fb543458985e57cb24fcf7bf3 : SUCCESS in 26m 11s\n- openstack-tox-py38-15 https://zuul.opendev.org/t/openstack/build/d63f5590ab40431ca7bbd8e0bad2148f : SUCCESS in 36m 02s\n- openstack-tox-py38-16 https://zuul.opendev.org/t/openstack/build/18c0f48fdb5245c0be4c190c916e0369 : SUCCESS in 27m 04s\n- openstack-tox-py38-17 https://zuul.opendev.org/t/openstack/build/abbf02e74e354ce1b5e10de26342775b : SUCCESS in 29m 59s\n- openstack-tox-py38-18 https://zuul.opendev.org/t/openstack/build/153a80db38cd4fd9824cc1bf12e095a9 : SUCCESS in 26m 16s\n- openstack-tox-py38-19 https://zuul.opendev.org/t/openstack/build/cbd858c73325406db1536bd9830294c5 : SUCCESS in 40m 09s\n- openstack-tox-py38-20 https://zuul.opendev.org/t/openstack/build/d6aea0cb3c2b422dbfcfe78dd03e0d6b : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":10},{"id":"d977b8bca3ce6482a8e6cb3f47865c5c59eb45f1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-05 13:44:11.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"e5b8fabeff2709a29f8d6a7cbd1c6daf1392b18c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-05 13:44:26.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/784552/10 : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":10},{"id":"0755e68661aadd8c9623c77ea231b3d3588f4490","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-05 15:20:51.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/52/784552/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"44980c6df6d29e75e1bedc2c6b5e1cf6c796a76b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-10 09:33:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e2f22b9730c33ef4cd7f371809b735d5ba280396","revisions":{"515c9c1bb844946b050977d55eaaff2aee47a69f":{"kind":"REWORK","_number":1,"created":"2021-04-02 09:57:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/1"}}},"commit":{"parents":[{"commit":"b33539ee26222230157f9b1f406ace94f9b4a258","subject":"Merge \"Fix create_port new API policy roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b33539ee26222230157f9b1f406ace94f9b4a258"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-25 20:19:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 09:57:07.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/515c9c1bb844946b050977d55eaaff2aee47a69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/515c9c1bb844946b050977d55eaaff2aee47a69f"}]},"branch":"refs/heads/master"},"818aaf71db05c003b6eefdae207cb50ba5655375":{"kind":"REWORK","_number":2,"created":"2021-04-02 11:28:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/2"}}},"commit":{"parents":[{"commit":"b33539ee26222230157f9b1f406ace94f9b4a258","subject":"Merge \"Fix create_port new API policy roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b33539ee26222230157f9b1f406ace94f9b4a258"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:28:12.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/818aaf71db05c003b6eefdae207cb50ba5655375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/818aaf71db05c003b6eefdae207cb50ba5655375"}]},"branch":"refs/heads/master"},"ce36ad46f4f624931810a316e482d49491849395":{"kind":"REWORK","_number":3,"created":"2021-04-02 17:46:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/3"}}},"commit":{"parents":[{"commit":"bdf1416958f79e8076397fcbf2fbbc920f964b37","subject":"Remove redundant _ensure_default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdf1416958f79e8076397fcbf2fbbc920f964b37"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 17:44:12.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce36ad46f4f624931810a316e482d49491849395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce36ad46f4f624931810a316e482d49491849395"}]},"branch":"refs/heads/master"},"e2e63a00f89f95e619faf3fa201f4cd19bf47b17":{"kind":"REWORK","_number":4,"created":"2021-04-05 07:12:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/4"}}},"commit":{"parents":[{"commit":"bdf1416958f79e8076397fcbf2fbbc920f964b37","subject":"Remove redundant _ensure_default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdf1416958f79e8076397fcbf2fbbc920f964b37"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-03 01:15:56.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2e63a00f89f95e619faf3fa201f4cd19bf47b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2e63a00f89f95e619faf3fa201f4cd19bf47b17"}]},"branch":"refs/heads/master"},"158ea3eb7bd44a423a8d83f4bd5e633a54497e7d":{"kind":"REWORK","_number":5,"created":"2021-04-05 07:27:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/5"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-05 07:27:28.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/158ea3eb7bd44a423a8d83f4bd5e633a54497e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/158ea3eb7bd44a423a8d83f4bd5e633a54497e7d"}]},"branch":"refs/heads/master"},"942e28b83433a98a901ddc1c7289aa2126b3ea99":{"kind":"REWORK","_number":6,"created":"2021-04-05 07:41:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/6"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-05 07:41:17.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/942e28b83433a98a901ddc1c7289aa2126b3ea99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/942e28b83433a98a901ddc1c7289aa2126b3ea99"}]},"branch":"refs/heads/master"},"4bbbcbd424bcfa866c9cee070e7eaced4829def3":{"kind":"REWORK","_number":7,"created":"2021-04-05 08:26:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/7"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-05 08:24:32.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbbcbd424bcfa866c9cee070e7eaced4829def3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbbcbd424bcfa866c9cee070e7eaced4829def3"}]},"branch":"refs/heads/master"},"5cdf50ec4df5a4dc14130ffe2650784a5343ff60":{"kind":"REWORK","_number":8,"created":"2021-04-05 09:47:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/8"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-04-05 09:47:06.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cdf50ec4df5a4dc14130ffe2650784a5343ff60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cdf50ec4df5a4dc14130ffe2650784a5343ff60"}]},"branch":"refs/heads/master"},"8804a957d331815009661fb06bdbe44b1dceef68":{"kind":"REWORK","_number":9,"created":"2021-04-05 11:09:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/9"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-05 11:09:38.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8804a957d331815009661fb06bdbe44b1dceef68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8804a957d331815009661fb06bdbe44b1dceef68"}]},"branch":"refs/heads/master"},"e2f22b9730c33ef4cd7f371809b735d5ba280396":{"kind":"REWORK","_number":10,"created":"2021-04-05 11:56:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/784552/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/784552/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/784552/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/784552/10"}}},"commit":{"parents":[{"commit":"141469d328d8e2f57d0198b1cb68f32cc160e68e","subject":"Merge \"Open Xena DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141469d328d8e2f57d0198b1cb68f32cc160e68e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-02 11:27:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-05 11:55:47.000000000","tz":0},"subject":"[DNM] Testing openstack-tox-py38 timeouts","message":"[DNM] Testing openstack-tox-py38 timeouts\n\nChange-Id: Ib9b46dec5eaa42e45075bc32e120b5216f2a08ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f22b9730c33ef4cd7f371809b735d5ba280396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f22b9730c33ef4cd7f371809b735d5ba280396"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
