)]}'
{"id":"openstack%2Fneutron~784576","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I9f8a9f0ae24683458aa3b5f8e49c6f111ccf815a","project":"openstack/neutron","branch":"stable/wallaby","topic":"secure-rbac-stable/wallaby","hashtags":[],"change_id":"I9f8a9f0ae24683458aa3b5f8e49c6f111ccf815a","subject":"Fix API policy rules for new personas","status":"MERGED","created":"2021-04-02 12:52:38.000000000","updated":"2021-04-03 02:37:06.000000000","submitted":"2021-04-03 02:35:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784576-secure-rbac-stable/wallaby","cherry_pick_of_change":784434,"cherry_pick_of_patch_set":1,"meta_rev_id":"9af70f62708ffe60da777fd8956cd32a4a12d26f","_number":784576,"virtual_id_number":784576,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-03 02:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2021-04-03 00:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-04-02 15:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2021-04-03 00:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-04-02 14:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 12:52:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-04-02 13:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 13:57:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 16:20:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-02 18:06:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-03 00:07:43.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"702ac3bddb4c60c3d9fa12cfcc25ce1c113faae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-02 12:52:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2972ae5bba782832463c9032791d855c7f6a61e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 13:31:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d629b14caf4d473f8ac5e518d3319768 : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/577f68bf311745e897e180c067d4fbd6 : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b2ff34aa3f2c7f28f73083d3fddffe1c6be3e7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 13:57:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/784576/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/784576/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"99e8ea33bbc0043b0cb5cbb1677b8723a44f871f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-02 14:34:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74f984e9fab800d7abf7d207771f75469de0cef5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-02 14:35:08.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"c0585cfd8b7f704b6266ce37e0b326677a445b0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 14:43:21.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"df8532fe6b7b122f3ab1002206827c3335414815","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-02 15:17:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"488e75d29077220596cf40e207478181712e372f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 15:26:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b03919065b0844e0bbe36c0895a1b2c2 : FAILURE in 46m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d24b4d2a97264c1894e302015b43e7ea : SUCCESS in 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9776fd88e1b448b2fee77e96fbd2a315b714930c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-02 16:20:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/784576/2 : SUCCESS in 1h 44m 55s","accounts_in_message":[],"_revision_number":2},{"id":"32c70a6f9303623279fdacd974123054ba3c3e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 18:06:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0cf3d63f0c7f48eb90a9d396c63d535e : SUCCESS in 1h 00m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a33a695230744b6917bbaf263de7cdb : SUCCESS in 2h 02m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7db5489a87c341f2926c8183b279c4cc : SUCCESS in 1h 31m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6389f9eaa39480d98579a6f90c4b374 : SUCCESS in 1h 41m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b02f689ba2cc48d7b9e5643d7b6229d5 : SUCCESS in 1h 45m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c361ab750a04c6b954fbe4946b24932 : SUCCESS in 38m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a32bb18aebcd436ebe6b02e45c764f70 : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6ef5dcc09648a58f5a90d78b3f884b : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8862c8b500f4463a8f3552104a9e961b : SUCCESS in 30m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f71a3b1f41c04fedaea8128900baa1c9 : TIMED_OUT in 1h 00m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fea76612a3d445baac36c22802350f4c : SUCCESS in 18m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb917bb5d0744a28e2a30f68336408d : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8865628d68534b958278bd560ee29f5d : SUCCESS in 46m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a37935c853844ac593d7d410b929f589 : SUCCESS in 1h 20m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/be85fc0beab94cdf85f056c8630592d8 : SUCCESS in 1h 25m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84dd7f4d4959495b80d012dad21ace59 : SUCCESS in 1h 21m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca057c6e4c034053954d10b4e267f0c8 : SUCCESS in 1h 17m 44s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/206b1d47d708440da722c1e9a9d3b62c : SUCCESS in 1h 16m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c04f42c267b494ba71fb2d4bc961854 : SUCCESS in 1h 23m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0762f8bc0924dd2bd5426640d65b488 : SUCCESS in 1h 19m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b83a799dff94aec90ba859231a87e6d : SUCCESS in 55m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f14e3089afad4000a95bab780ca2c85e : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab6fbd4008784ade936a94f262a6d217 : SUCCESS in 44m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d422d8cce45342fdb73364a9fd571cd5 : SUCCESS in 1h 23m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d2db63d743a5430697a83615dd288cd2 : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":2},{"id":"48c29c7621c5b04e92ae4ef7072332ecb962b0b0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-02 20:32:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1e76668ee97199cdca4a1481682785de98b199d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 21:19:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbfbba11434d46ceb19610056514eabe : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dd168b0d5314fcebc088600cff2aa48 : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa826e00cefbb4f771acab34f61c5437c3f20d4a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 21:46:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b38ff63fa11d3d5093fff5a517add2298e1aaae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 22:13:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2307c077a86e4ffb88e66c900b79a072 : SUCCESS in 50m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cdc142a53689421f9023662ce3f5eda8 : SUCCESS in 1h 35m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fee09d64ee1c48f3901b78a879eed77a : SUCCESS in 1h 34m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ddc202acce83402284a2bc8de1f6cf5e : SUCCESS in 57m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e3dbcf6cc9ed49c891dd89702c04cb44 : SUCCESS in 1h 24m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2249564a775b484e8b26259ff9b13ff4 : SUCCESS in 28m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69916e7b491c4ec68ab73d895997d1f7 : SUCCESS in 41m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a58ca9218ed4371a9b17e169aeb323c : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69426ad5aa7a4e2d9722533e1ffabb4d : SUCCESS in 22m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a47b3d8260641a3afed3090bd19b294 : SUCCESS in 19m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52b3837b511a43269c543c6310c92869 : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c2f530ceba4fe496f6d7b46b2eb0b8 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c950e929981f4146838f0d97c5ac25db : SUCCESS in 53m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61c400d969104cb68efd93b692fefa41 : SUCCESS in 1h 07m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b8cc8ee49b084976aec5484532752bd1 : SUCCESS in 1h 25m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34673c147ca6477184a897b6c618c008 : SUCCESS in 1h 36m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ecd37e55b214c79a9c3a7104006e544 : SUCCESS in 1h 20m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc7b546a289d4af7a2062ce62fcb7148 : SUCCESS in 1h 18m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a506fc8b0ee44ca6abb1d7c08f03c8b9 : SUCCESS in 1h 26m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9cfea5a77694a86ab413d6c57f3916f : SUCCESS in 1h 30m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0adb63ce79d04d6ba2402db0ac9c7d52 : SUCCESS in 1h 03m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2e46d8c8673d43d2b0f277292f95620c : SUCCESS in 1h 08m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3cdc00e964ed482d911efbc7407fee3a : SUCCESS in 36m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f88ecde55d654259952252219e333d02 : SUCCESS in 1h 35m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/52b9d1dcfab94eafb34848a4f6a35f78 : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":2},{"id":"855a949604eb2214f5cb16544d5c288c96e9055d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-02 22:27:31.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/784576/2 : FAILURE in 1h 54m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d49a57d198a63892f2e2a5535de2724947b79b99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 23:32:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/784576/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"1883fc56f7b80499bec7a7d880bbdbd488d284f7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-04-03 00:07:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42f1c114af68540f54895351424798709810f5b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 00:08:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9287b3ee67d46a331fe72a2a2ac46c43dfc9e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:35:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c689bb960174c2d9f705d57fa4e0c52 : SUCCESS in 48m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/731640ad41af4840a643271eb81d2ad3 : SUCCESS in 2h 27m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3463cace154448438ce0f12e7a62c890 : SUCCESS in 1h 53m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3bfaaf1574a84afa972b170b53aece40 : SUCCESS in 1h 45m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d424c9750d4f48d2b59d9f1c3265d43f : SUCCESS in 1h 15m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca352581fec5452ab7a32ac0279d1e78 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0cc3f7301a45f6887a143e535d894d : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9039424d5bae46ec8a6c06e3cafb8895 : SUCCESS in 24m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc0f484ec96c486aa984565a0eead1fe : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31602a64f04b45bfa511fda9ca968382 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ef0d6aabd444f109874ca26419696f5 : SUCCESS in 44m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/efe53bcf0a0048c7bcb9a3622046070c : SUCCESS in 1h 25m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2293e59e3b7c4b5386ac22d6ce6c4d84 : SUCCESS in 1h 23m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9721e1b727449869d86d965b4774ac6 : SUCCESS in 1h 05m 16s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab5c4c148a8c414798a8cbd994ec548f : SUCCESS in 1h 37m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24071912995640ccb5036bd752ed5580 : SUCCESS in 59m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a27f5c67c2f49448f22d183b727df7f : SUCCESS in 1h 10m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/15e53c69621d44c0a40d83d68c9491e2 : SUCCESS in 32m 05s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/559b8dc3a1f7413c912d7a80728d94da : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2aaa6a5719e92f2e20ea2a83b2051232a53a1e1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:35:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9af70f62708ffe60da777fd8956cd32a4a12d26f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 02:37:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed335eff677b4e9b82219b2d2e8b6edd : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15fee2f573c330e4e533ff3841cb008afaba299f","revisions":{"bdee94bce9ad875f12e81664be99a1adcbb943f6":{"kind":"REWORK","_number":1,"created":"2021-04-02 12:52:38.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/76/784576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/784576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/784576/1"}}},"commit":{"parents":[{"commit":"80e1d8269b4963a905ecadaa282f5f5f9e97cf4e","subject":"Merge \"Modify create address scope policy rule and add UT for that API\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e1d8269b4963a905ecadaa282f5f5f9e97cf4e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 15:00:59.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-04-02 12:52:38.000000000","tz":0},"subject":"Fix API policy rules for new personas","message":"Fix API policy rules for new personas\n\nDuring migration to the new personas in the API policies, we made few\nmistakes and allowed creation of some resources just for PROJECT_MEMBER\nrole. It should also be possible for the SYSTEM_ADMIN persona and that\npatch fixes it by adding SYSTEM_ADMIN persona to the default rules where\nit was missing.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: I9f8a9f0ae24683458aa3b5f8e49c6f111ccf815a\n(cherry picked from commit cabf19a984a575d14cc4cde0312807e859765aa8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdee94bce9ad875f12e81664be99a1adcbb943f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdee94bce9ad875f12e81664be99a1adcbb943f6"}]},"branch":"refs/heads/stable/wallaby"},"15fee2f573c330e4e533ff3841cb008afaba299f":{"kind":"NO_CHANGE","_number":2,"created":"2021-04-02 14:34:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/784576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/784576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/784576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/784576/2"}}},"commit":{"parents":[{"commit":"80e1d8269b4963a905ecadaa282f5f5f9e97cf4e","subject":"Merge \"Modify create address scope policy rule and add UT for that API\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e1d8269b4963a905ecadaa282f5f5f9e97cf4e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-01 15:00:59.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-02 14:34:50.000000000","tz":0},"subject":"Fix API policy rules for new personas","message":"Fix API policy rules for new personas\n\nDuring migration to the new personas in the API policies, we made few\nmistakes and allowed creation of some resources just for PROJECT_MEMBER\nrole. It should also be possible for the SYSTEM_ADMIN persona and that\npatch fixes it by adding SYSTEM_ADMIN persona to the default rules where\nit was missing.\n\nRelated-blueprint: bp/secure-rbac-roles\nChange-Id: I9f8a9f0ae24683458aa3b5f8e49c6f111ccf815a\n(cherry picked from commit cabf19a984a575d14cc4cde0312807e859765aa8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fee2f573c330e4e533ff3841cb008afaba299f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fee2f573c330e4e533ff3841cb008afaba299f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
