)]}'
{"id":"openstack%2Fneutron~784889","triplet_id":"openstack%2Fneutron~master~I4846ecdd8312fa2ff915c4461da07a607c7a8f8c","project":"openstack/neutron","branch":"master","topic":"bug/1687027","hashtags":[],"change_id":"I4846ecdd8312fa2ff915c4461da07a607c7a8f8c","subject":"Remove FT \"test_has_offline_migrations_*\" tests","status":"MERGED","created":"2021-04-06 08:25:22.000000000","updated":"2021-04-06 21:14:02.000000000","submitted":"2021-04-06 21:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784889-bug/1687027","meta_rev_id":"6ad4ea89921726eaf4dd40e709715a0027d5242d","_number":784889,"virtual_id_number":784889,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-04-06 12:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 21:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 09:08:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-04-06 09:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-04-06 09:33:23.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-04-06 09:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 08:42:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-06 09:08:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-04-06 09:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-06 09:33:23.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-04-06 09:41:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-04-06 10:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 11:34:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 12:07:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"06f82f8626c9256282264bf9ea35644d282745e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-06 08:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd172162f2f22c51885b436265c026723996d9cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-06 08:27:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffd1dc7b37eb8cd9ca1aecc4ee94ed65174f98c2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-06 08:42:12.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/784889/2 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a70725fa8f0e55f00165c0842f3bbb28b93f9440","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-06 09:08:11.000000000","message":"Patch Set 2: Code-Review+1\n\nThat sounds like a really nice cleanup for functional tests!","accounts_in_message":[],"_revision_number":2},{"id":"e04e316581b84cafd574b73d1c0e4a504fba053d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:09:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1258c8ef032e483e97ca116aeb702f27 : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e5a9fe7a44946278bbe52a8c22666fd : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9cccaa1ef9f5ad716e9aeb5df73e2bb40e303ef0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-06 09:33:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be834052a9c761f833cb39738c7e0ada23001baf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-06 09:41:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6db7d7bd6d5cc1a43ca07836469eaeb2a69673bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:38:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/824dae5c7ccd4a52820ac0976e2e0af7 : SUCCESS in 48m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e4b1fc2f2904222aef37f4198344951 : FAILURE in 2h 00m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9bfc6b8e307345acbc63c2e449fc19df : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b1276703442409ba8c0803eb407263a : SUCCESS in 1h 34m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c88f1348c46149918589b005d6033e1f : SUCCESS in 2h 07m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b04ca176d004eb88cf38dadc1619548 : SUCCESS in 40m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81d3af32fd3b480ea20f90434cc9befe : SUCCESS in 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173aac7ba37148febbdca9004283d3b2 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e70c85966146b39eaa12cee9fccf20 : SUCCESS in 30m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78874527043c4052ba1804e7f43cd681 : SUCCESS in 22m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c387deea74db47598b4f34da60fb299c : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b1c80c68ca41debe1e5745e560214d : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c91068fa6c8a40ecb9bdee0d33117d4b : SUCCESS in 49m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/95de84ddf53741e79d62d925dceb3854 : SUCCESS in 1h 21m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/55432694cc424ceebe1aeac562c31918 : SUCCESS in 1h 32m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6d1916b7a6f4ef29c0d35045b26367d : SUCCESS in 1h 03m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/30eae050bec94da2b295c6f4f4d298c2 : SUCCESS in 1h 05m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6e87dd148a6a4bb78166235090a95356 : SUCCESS in 1h 16m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1529be338e5046449af09ce5436132e2 : SUCCESS in 1h 32m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/813cc7601d5349bc94827567858dbe9f : SUCCESS in 1h 30m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c44a3e7d2c41b19048f4b81a749718 : SUCCESS in 1h 03m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b35c5781856b4d318f77e41228bb5995 : SUCCESS in 1h 09m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/31f760f7694342d6b8b7cae30d02ccf4 : SUCCESS in 29m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec41504d253741fdbeb87c30c074229d : SUCCESS in 1h 35m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a256aff57df142ba95fda82101f2071a : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fd230a16e722225bfee8c6589e400cfefac476e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-06 10:50:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba6667143933275b46e97f482699f7dd9ef6c571","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-06 11:04:29.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/784889/2 : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5cf944481675e3137b248dc890dc864a4a9316d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-06 11:34:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/89/784889/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9559d48fd5895cd3a9d253d465133a620c6dd368","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-06 12:07:47.000000000","message":"Patch Set 2:\n\nrecheck\nneutron-tempest-plugin-scenario-linuxbridge","accounts_in_message":[],"_revision_number":2},{"id":"342279384a0a9487ebede103ccd61eb245b5e4e3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-06 12:22:19.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/784889/2 : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4ea14742e2830af3c82865fbaa54d988fef25f6e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-06 12:54:21.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/89/784889/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cc230982a2aeaf7b456d8cf113433a143413df0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:45:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8636ce74f30407ea5d1bfe2854973bd : SUCCESS in 43m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3c3e5c2687d24480a36e7d4c2eab2280 : SUCCESS in 2h 21m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6f258eea760e487cb87e67fc02f3febd : SUCCESS in 2h 30m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2241435f59d45adbf44d9cffbd7063c : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7b6ce1e32b1c4876a3a5bf76a09b3b44 : SUCCESS in 2h 11m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/408144d1423643359e71105281497769 : SUCCESS in 45m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e3c56d39fc4701bdf3328fd3a024f0 : SUCCESS in 37m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6475ac77df494bb5494dbc321e9353 : SUCCESS in 16m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa8e899a563444cabcdaef2a817c5212 : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6bf7172e1ab4ab1be5234160b0da364 : SUCCESS in 30m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b109c6c58fd4be3ad7653e4f6853c5d : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d2ab22ac784b41acfa85ef02799f3c : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cf2de9c6ec1401bbe2979ef716f29b0 : SUCCESS in 54m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4bf3704a8da48e092aff8579571f087 : SUCCESS in 1h 43m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/75b1b69ff0c44ea4b7171766780113d6 : SUCCESS in 1h 56m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0638560c8b524e78b68f4cd1bbaccb1f : SUCCESS in 1h 05m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce33b181556a42df930ade618bf657ad : SUCCESS in 1h 32m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a0d80acf5f0d4af0b85b45e53f692ea9 : SUCCESS in 1h 14m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53dbc3d2f73a49e5b40ca368ccdbd95e : SUCCESS in 1h 28m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13f0ffbf59e9475cbf767e20fbe081a4 : SUCCESS in 1h 37m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31a9fcc9981c42a287072a1a793d571e : SUCCESS in 1h 00m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4b986ede65ce407d983de7ac2112bf69 : SUCCESS in 1h 14m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f2df5e32fcbf493d93b61945c531b0e9 : SUCCESS in 36m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a6563632fea4b3cb8e99f95d107fa0c : SUCCESS in 1h 34m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ffc0a1f512fe49bf923a6604aa47b691 : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4fe2deb31138d61d9fe698abf4be65966f40a4ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:46:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c35bef93e3d2a44eb66baeec70e2a5e00815e91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:12:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cb7e5b3dfda4d3d83b57473c39b6029 : SUCCESS in 53m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82912000c203494c9e9e7e3675eec09a : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c78c8df20508c9d18245766593144c955b0b3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 21:08:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c1731f70ea00445ba11c4a164e7ca8a3 : SUCCESS in 41m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/305a374a9c6540de9ae18d3d906caac0 : SUCCESS in 2h 20m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4e121fba908849ff9e1a3476b7217c30 : SUCCESS in 1h 48m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd20dca328dd447b8d198138cada2567 : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bdbb1f1d51964990804a259368302c34 : SUCCESS in 1h 10m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1624dc941d8044edbee0a6621b9a5ea2 : SUCCESS in 35m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97efae4b87fb45bf95875cf2136cf05d : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fffd159a15424affb244bcad3befab02 : SUCCESS in 33m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab3872a1b2b490083fae6975f391f50 : SUCCESS in 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cbdc374f70410e9d7f525113fc7dc7 : SUCCESS in 11m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d35c331f7c714be7b96ebe8b7a7b241a : SUCCESS in 53m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cacec6b3613a4cfdbd8708fa2259fa97 : SUCCESS in 1h 25m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ec1bced0b2bd49988d31ab1bcee866eb : SUCCESS in 1h 19m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e54073bc8b544ac48a88448022148583 : SUCCESS in 1h 12m 28s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5f76aeab131449f93860eec6f0c3d87 : SUCCESS in 1h 46m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fd2b4cdc00b40be8a78430c37d28805 : SUCCESS in 59m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/495f08df2c4e416197caa45d549e8f47 : SUCCESS in 53m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a13a6c88ff304da886707ef62bb238f8 : SUCCESS in 33m 52s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c429f8be3bba4ddb8e6301b27119f347 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d5c5f8efd190379063babb7cb07112c81099d640","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 21:08:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6ad4ea89921726eaf4dd40e709715a0027d5242d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 21:14:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8df972a24b94fc597c57185eb013a86 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8522628f2f6f59344c04568b31c444f7a290562d","revisions":{"cd3e17b5c998e34bc71787b3b494a70a8bcf2ac8":{"kind":"REWORK","_number":1,"created":"2021-04-06 08:25:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/784889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/784889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/784889/1"}}},"commit":{"parents":[{"commit":"58c9912be0ce5d9bf9eb9e1c44b87cdf90aab452","subject":"Merge \"Implement namespace creation method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58c9912be0ce5d9bf9eb9e1c44b87cdf90aab452"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-06 08:14:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-06 08:24:52.000000000","tz":0},"subject":"Remove FT \"test_has_offline_migrations_*\" tests","message":"Remove FT \"test_has_offline_migrations_*\" tests\n\n\"test_has_offline_migrations_pending_contract_scripts\" is testing a\nnon supported branch, \"Kilo\". The usefulness expired when this branch\nwas declared EOL.\n\n\"test_has_offline_migrations_all_heads_upgraded\" is tested in\n\"test_forbid_offline_migrations_starting_newton\". Since \"Newton\",\nno more DB contract migrations are allowed. Both tests are checking\nthe same behavior.\n\nDue to the time extension and the lack of value of those tests, this\npatch removes both.\n\n  test_forbid_offline_migrations_starting_newton [683.133807s]\n  test_has_offline_migrations_pending_contract_scripts [123.091853s]\n\nChange-Id: I4846ecdd8312fa2ff915c4461da07a607c7a8f8c\nRelated-Bug: #1687027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd3e17b5c998e34bc71787b3b494a70a8bcf2ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd3e17b5c998e34bc71787b3b494a70a8bcf2ac8"}]},"branch":"refs/heads/master"},"8522628f2f6f59344c04568b31c444f7a290562d":{"kind":"REWORK","_number":2,"created":"2021-04-06 08:27:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/784889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/784889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/784889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/784889/2"}}},"commit":{"parents":[{"commit":"58c9912be0ce5d9bf9eb9e1c44b87cdf90aab452","subject":"Merge \"Implement namespace creation method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58c9912be0ce5d9bf9eb9e1c44b87cdf90aab452"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-06 08:14:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-06 08:27:16.000000000","tz":0},"subject":"Remove FT \"test_has_offline_migrations_*\" tests","message":"Remove FT \"test_has_offline_migrations_*\" tests\n\n\"test_has_offline_migrations_pending_contract_scripts\" is testing a\nnon supported branch, \"Kilo\". The usefulness expired when this branch\nwas declared EOL.\n\n\"test_has_offline_migrations_all_heads_upgraded\" is tested in\n\"test_forbid_offline_migrations_starting_newton\". Since \"Newton\",\nno more DB contract migrations are allowed. Both tests are checking\nthe same behavior.\n\nDue to the time extension and the lack of value of those tests, this\npatch removes both.\n\n  test_forbid_offline_migrations_starting_newton [683.133807s]\n  test_has_offline_migrations_pending_contract_scripts [123.091853s]\n\nChange-Id: I4846ecdd8312fa2ff915c4461da07a607c7a8f8c\nRelated-Bug: #1687027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8522628f2f6f59344c04568b31c444f7a290562d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8522628f2f6f59344c04568b31c444f7a290562d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
