)]}'
{"id":"openstack%2Fneutron~785591","triplet_id":"openstack%2Fneutron~master~I1402916b69db51ee4c451118be2c89d6be7ac155","project":"openstack/neutron","branch":"master","topic":"rehome-ovs-constants","hashtags":[],"change_id":"I1402916b69db51ee4c451118be2c89d6be7ac155","subject":"Remove BAGPIPE_FLOOD_TO_TUN_BROADCAST constant","status":"ABANDONED","created":"2021-04-09 10:27:32.000000000","updated":"2021-04-09 12:46:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4051d8107298f97b602dd9eb4e41daec194676a3","_number":785591,"virtual_id_number":785591,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"CC":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 10:52:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-04-09 11:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 11:27:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-04-09 11:27:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2021-04-09 12:46:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"11a4b9c53c66044a042c9903fea0cb3c3d74d732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 10:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e339181e13eb7c5bde41a1021fbc570dc1d3f60","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-09 10:52:27.000000000","message":"Patch Set 1:\n\nThe idea came from this review:\nhttps://review.opendev.org/c/openstack/networking-bagpipe/+/775398\nperhaps not the finest place, but all other flow constants are here, and now there is no common place for them to avoid overlapping flow rules/tables whatever.","accounts_in_message":[],"_revision_number":1},{"id":"1abefa406100c44615e3c20dd1fd92ca3ef463c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 11:12:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c94da5e96f84f88b0c7b4bc77ac2b02 : SUCCESS in 37m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd99db823f5443c69db10b65846b64ec : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbf1d465d569754c5da38ef3e887a061aa259c75","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 11:41:13.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e The idea came from this review:\n\u003e https://review.opendev.org/c/openstack/networking-bagpipe/+/775398\n\u003e perhaps not the finest place, but all other flow constants are here, and now there is no common place for them to avoid overlapping flow rules/tables whatever.\n\nThx Lajos","accounts_in_message":[],"_revision_number":1},{"id":"db494354c8c90435aedcc3db8652981b73094052","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 11:41:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"4051d8107298f97b602dd9eb4e41daec194676a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-09 12:46:44.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/91/785591/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/91/785591/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"932d3504bd7eaeb06accc911c6499ec4b0e2c6b9","revisions":{"932d3504bd7eaeb06accc911c6499ec4b0e2c6b9":{"kind":"REWORK","_number":1,"created":"2021-04-09 10:27:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/91/785591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/785591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/785591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/785591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/785591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/785591/1"}}},"commit":{"parents":[{"commit":"bdd661d21898d573ef39448316860aa4c692b834","subject":"Merge \"Fix \"_get_sg_members\" method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdd661d21898d573ef39448316860aa4c692b834"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:25:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:25:36.000000000","tz":120},"subject":"Remove BAGPIPE_FLOOD_TO_TUN_BROADCAST constant","message":"Remove BAGPIPE_FLOOD_TO_TUN_BROADCAST constant\n\nIt was defined in the openvswitch agent\u0027s constants but it isn\u0027t used\nanywhere in Neutron or any other OpenStack project. So I think we can\nremove it.\n\nChange-Id: I1402916b69db51ee4c451118be2c89d6be7ac155\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/932d3504bd7eaeb06accc911c6499ec4b0e2c6b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/932d3504bd7eaeb06accc911c6499ec4b0e2c6b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
