)]}'
{"id":"openstack%2Fneutron~786027","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~If04e360c01cdb16e956d16357e2181db55445137","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1687027","hashtags":[],"change_id":"If04e360c01cdb16e956d16357e2181db55445137","subject":"Group execution of SQL functional tests","status":"MERGED","created":"2021-04-13 09:41:27.000000000","updated":"2021-04-14 21:10:22.000000000","submitted":"2021-04-14 21:07:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786027-bug/1687027","meta_rev_id":"f67a65b2ed8db950e92496236fbf2ab9aaadcaec","_number":786027,"virtual_id_number":786027,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 21:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-04-14 13:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-04-14 12:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-04-14 13:59:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 11:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 11:39:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-14 12:18:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-04-14 13:59:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"1928f1ba3321e2a7d11230f70445a3cc75bbdcf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-13 09:41:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22b233417e83c93715e91be5c94b70ccc8e06752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:06:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b63141cc9d402993d262a22e8bd4bd : SUCCESS in 46m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581101060d1448829830644b3ccfaa8f : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ee357022da4ab1960e6ad8c3217242 : SUCCESS in 27m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78f64d1339b24562b56e800a4ef6e6dc : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ba0164b6c245f29969e15cf67a89f4 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a289b4010fd34f2085321cb38c08bfb7 : SUCCESS in 13m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f55ae36de4a14100a1d6b33a54e9de51 : SUCCESS in 1h 13m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b99142a266ea46b8a89921b342fdb38a : SUCCESS in 1h 22m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f59496ced7584cd68172a179cdb134ea : SUCCESS in 1h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e51e92d0937fe515e1a65b3a886e25915eeac523","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-13 11:39:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786027/1 : SUCCESS in 1h 54m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1205fa48d6c55f53e7c36bc528c0c3984662d30c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 12:18:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38bcc4c41d5e9857d279ff28de690b9fc3b0b59a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-14 13:59:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d0fb0f396d45e5952fa3237d6ba3e654a280221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96984dbbfd8fb7c6b7ea92f37d96e57d86896bec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:07:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ee17eaab5e41e5875c4af43dc8c737 : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a5dfc366484ce8b251b4ddfc353d7a : SUCCESS in 28m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5d647597fc94bc7ada95f513e432a2c : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9643576f544f5e8c4c0a905db5a275 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f949f3540a344f0b69eb8a61499630b : SUCCESS in 11m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffe5cbd50fea4d03a34bd21edc8c3515 : SUCCESS in 1h 18m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c789e441903f4644b5b5d1119edc757f : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f345739d34a9bff845e0288476f6c979a21d8f11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:07:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f67a65b2ed8db950e92496236fbf2ab9aaadcaec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:10:22.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd0fd0eac2b4efcba28a545981338b5 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d9330e762a4d58b2602eb416a27084 : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfeafa5ae7f473438a4d59408733203916d4cff8","revisions":{"cfeafa5ae7f473438a4d59408733203916d4cff8":{"kind":"REWORK","_number":1,"created":"2021-04-13 09:41:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/786027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/786027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/786027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/786027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/786027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/786027/1"}}},"commit":{"parents":[{"commit":"49e914849ff876d3017be1a68ff7b9f48a7e1ac9","subject":"Merge \"[ovn]: Remove unwanted IP addresses from OVN ports\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49e914849ff876d3017be1a68ff7b9f48a7e1ac9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-09 07:29:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-13 09:41:06.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process. That\nwill ensure those tests are not executed in parallel, interfering\neach other.\n\nClasses that are grouped:\n- TestModelsMigrationsMysql\n- TestModelsMigrationsPsql\n\n[1]https://stestr.readthedocs.io/en/latest/MANUAL.html#group-regex\n\nRelated-Bug: #1687027\n\nChange-Id: If04e360c01cdb16e956d16357e2181db55445137\n(cherry picked from commit 7326fd622f2c3cb36681136d79e39e4efdda5411)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfeafa5ae7f473438a4d59408733203916d4cff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfeafa5ae7f473438a4d59408733203916d4cff8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
