)]}'
{"id":"openstack%2Fneutron~786399","triplet_id":"openstack%2Fneutron~master~Iaecfda2700c5316cb25a93496d24ece366e40a4a","project":"openstack/neutron","branch":"master","topic":"subnet_create_opt","hashtags":[],"change_id":"Iaecfda2700c5316cb25a93496d24ece366e40a4a","subject":"Improve Subnet create performance","status":"MERGED","created":"2021-04-15 12:53:15.000000000","updated":"2021-06-05 10:04:30.000000000","submitted":"2021-06-05 04:06:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786399-subnet_create_opt","meta_rev_id":"188d8823fc56c63f48048f1aeb5e8440f6f4505b","_number":786399,"virtual_id_number":786399,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-05 04:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-03 16:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-05 10:04:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-04 07:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-03 16:13:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 13:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 13:53:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 14:41:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-15 15:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-16 03:39:47.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-16 09:44:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-16 14:20:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-03 16:13:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-04 07:56:02.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"1103cf3dc563ae9782cb97c4f06a86dd157f14ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-15 12:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c89d087ebe85658e91d0ee0e8fd53c5ef39de4e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 13:40:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2f0d4d6837c44ff8ed821edb8cf821c : SUCCESS in 41m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a676369fb5944fcb0b5d35bb5a5883a : SUCCESS in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d3303c562183b93893f1590c9675fe53c18b7f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-15 13:53:08.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/99/786399/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c7fda82420bdad024808656098594b4bc26c5595","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-15 14:41:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/1 : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e2791bc9fab67240f91933715224396a8edaf390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 15:14:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/32bd7c6074c2448f9a95c982c837da87 : FAILURE in 26m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c1d7fabe122544828ef663f9d25d1fd5 : FAILURE in 29m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d0a390d09ad141fa8dfacd90bf21d80a : FAILURE in 24m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f320179c3b5742edb707b4ebc5a7191b : FAILURE in 23m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d21b85dcd3eb441f943421d74e9e1bd8 : SUCCESS in 2h 02m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52ae314a88c140edb862a89723121aac : SUCCESS in 38m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd9c2c9a112e42e4a0a4bbdc98d87bab : SUCCESS in 36m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab88d133b7a4e69b91f32960eefad31 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d6fc3d72cd45479804ced839695852 : SUCCESS in 31m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b8835e69c664fba936dff90682d5718 : SUCCESS in 28m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/212bc99c799644a6b74c9cfe1e2b808f : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd550e43683470c954f5e2e5536a400 : SUCCESS in 13m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bca7424815e34666903cb0079186113a : FAILURE in 47m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a2504a9d2104756924ed117936d7748 : TIMED_OUT in 2h 15m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c887999667ab430982a35dfc84ba427b : FAILURE in 27m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79ca422949924c859b02439b187f03e2 : SUCCESS in 1h 07m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ffa24b4ef7c641548cd90c437c1568f1 : SUCCESS in 1h 04m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/06b516c092ec436a8b8df9c560638a4f : FAILURE in 20m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9d410f802f464114993a2f903e96c55c : FAILURE in 24m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9e37a6cf9f847859293191f1e4951fe : FAILURE in 19m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80bf10de43814faa88b5b7f7e091071c : FAILURE in 16m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/efb7b7e0674a46fcb45d8b2ede42677c : SUCCESS in 1h 05m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/46c79c8956274e25adab09f818d99efe : SUCCESS in 34m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/928295615a1f40d982e9ebebfba27333 : SUCCESS in 1h 18m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7a664c4865fd4281a1f70de91f445fb7 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4094a2b198d1ad8589bc1e65e1d3a61c6ad3af58","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-15 15:22:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fbeb74ab3790a116e8fea1e8502563934778fec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:23:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78fb0440c72b4f63bbe5249151eb18ab : SUCCESS in 53m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0a2ed205e3844e69291f934d2ee2b10 : SUCCESS in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59fbabbb4d8f5831f33ca8fad6cefb09d7a010fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 17:21:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad81f9e1750e49cb81a87d6f4117a0e7 : FAILURE in 24m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8daff25b143446b49b06b7c603d6ea3f : FAILURE in 19m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/691afda48a064e3cb1c179623f8e642b : FAILURE in 17m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2a85f7a73b154f0a90e0c7a0fde026bf : FAILURE in 26m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/25fa982b75b34a0ca8baa39d44e3ba77 : FAILURE in 36m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4620404a2264bb0828e4a2c318b319f : SUCCESS in 26m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37b5f1c9ba8e41a3aa5687c3508023ff : SUCCESS in 35m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49da41a35864e26af7e2b2c8c98bc72 : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54993d7110cc48729202c9d7130ef1ec : SUCCESS in 23m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196d6a0c8247457ab29cbf3dd1866f04 : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b548eebfb7cc4c37883b4e033430d9c6 : SUCCESS in 30m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3a18958a904f248c3815987951dced : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f45d309cdbd14927bac1703b897b9b09 : FAILURE in 50m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2ddbd4b09b0448ebd568a91a2547ecf : SUCCESS in 1h 26m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e450cc64fde142cbb97d694b8af5a441 : FAILURE in 24m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4474ae42ec68409e85cdea997eb8eed8 : SUCCESS in 1h 09m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c58e970c4e441068cb40bb7fd8a2d41 : FAILURE in 34m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9798931409164dda950d46727ce6810a : FAILURE in 32m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3436f213576471bb80418704bbf0c74 : FAILURE in 25m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d034a1aa97fd4541a6d55647ce0a110e : FAILURE in 22m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69a9fd86cd22472fac34959d5b1a9dc5 : FAILURE in 24m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b2bbadf84a614606b34e01dfec7aea28 : FAILURE in 36m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a2be6c6f4e434731b5cf1c4067083090 : SUCCESS in 30m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b953e76e462b41108681ad3fbbda63f0 : SUCCESS in 1h 43m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3fd0bb24e5d94324bd72beee45cb7d0c : FAILURE in 35m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b912ac200b594b2f06ec7532eeeb8404ceb00c88","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-15 17:57:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/1 : SUCCESS in 1h 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"df06d7b921e18bef7aa72139d0a84aaa959979af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-15 19:36:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16023abb8d07ede2559ac3cccc72a3429a1d8160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 20:42:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/566c59035aa1427cb4a69272d0f1b2e3 : TIMED_OUT in 1h 01m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cafb025f651b444596f2f9b698c1300b : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4af7bcec2a6bdaef1cce937c40e8f1b1b9efb043","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-15 20:44:49.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/99/786399/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"70a86d75b5b20a1c045c3a3d4b336b90118957de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-15 21:17:38.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/786399/2 : FAILURE in 1h 40m 32s","accounts_in_message":[],"_revision_number":2},{"id":"748e4cc693aa882fd271e6f75606a7e934296225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 22:08:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bc4f24bc9b6d4b2db28782175d42cc4a : SUCCESS in 45m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ca75e7acb2c34712b333e4e583c014c7 : SUCCESS in 1h 47m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0bfdba449cc746269fa1c5c02ebd899d : FAILURE in 1h 42m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9963fa3e739b4830a8be72d37e30185a : FAILURE in 1h 42m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9be91e1649d14f06bcd4aad10b017d1c : SUCCESS in 2h 25m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66d734e2f61645229cd01ba49a90fa23 : FAILURE in 41m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b617c619a9442fca37618e2db166b27 : FAILURE in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42961ed0991145968d8b16615d371022 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a374af828b40a1a2b485b809f54b02 : FAILURE in 25m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/098cd74ecd914614a716c8cdf92a6c14 : FAILURE in 23m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97003ea2b6394c3b8f01856478b86da9 : FAILURE in 18m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0d0fd3a8414f94960a16628bada87b : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2567f9f38ba42dabf658f64237c1c70 : SUCCESS in 49m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/057566f00d204af48152332033adc6d0 : SUCCESS in 1h 20m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ba4e1f1ef9044527896f98189182932a : SUCCESS in 1h 17m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b67a02fc8f2744539340a933481f2588 : FAILURE in 1h 02m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c58188bc2f614503beb165d47a63f659 : FAILURE in 1h 10m 25s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4ffb15ffc33f4079b4abbb0acab43211 : FAILURE in 49m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb2a3c4b076b457e893d4a8f1142bbb3 : FAILURE in 1h 05m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/318481a776f840cf945cbd15aecc6a6c : SUCCESS in 1h 44m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c0b963321a4884bad77fe342aaf72f : FAILURE in 1h 00m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/80bfbdace34f43c581ab10dd4f85049a : FAILURE in 41m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/10fc41bf2ed24af9b83a895be5996939 : FAILURE in 28m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cceefac7965240e58df4911db8bdf921 : SUCCESS in 57m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5c04620d4d754370b024b9b81cd45df0 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a62a153a3b150e1cbd7e8303fbe965255a49492e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-16 03:39:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"9f0530f3a0225443134dbd001426c5c7639dc70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-16 05:44:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4b6f3483a127fb53ca92979e6c5e9494b6b5f7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 06:36:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/498dc2c9031f440bb1b45e40e23a0a3f : FAILURE in 49m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47271e0b398048fca811b05592973d33 : FAILURE in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cada0a6bc05e5e7d26cb5e93ab012ec9074c3836","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-16 07:03:08.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/99/786399/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"842b4fcd02aa35c0e7e5c78b64ea6cc1af28abb4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-16 07:27:54.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/786399/3 : FAILURE in 1h 42m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c1bd29bffb07d9dc00ecf504cbcbb62712a1d90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 08:01:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b3bf578079a4477810f2f24b89271ac : SUCCESS in 43m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8c2c2089d80d4e39b39c708bf5ebf91f : FAILURE in 1h 43m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5cd10860cc2b4584aca9ddd6107f1223 : SUCCESS in 1h 28m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b767e72cd7224fe2bdfbeb81af248509 : SUCCESS in 1h 12m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d5bc8f344f5244fe9b94e9d9929b3ee7 : SUCCESS in 1h 37m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92ec61611af541a89f79c4d8ce8a0ca4 : FAILURE in 23m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9397a11b831d40a6bd883943c0d8f0d3 : FAILURE in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc0c65b9df24545b00e122a2a4a5867 : SUCCESS in 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bc9f9c82e54452491899042d80e4121 : FAILURE in 19m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc6b475c00a64a4bb0691352ec1fc8c9 : FAILURE in 20m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aee7ea625cbf4c3d8eb294cdd03b0835 : FAILURE in 17m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde05ec3d3084431acb171756f77941b : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/367cae4768a7452cac9e13bde9b745a6 : SUCCESS in 58m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35c6e1ff8b38453b8c8760344918bd04 : TIMED_OUT in 2h 14m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c3d84530b7ca41b79e647b9500ef3e2d : SUCCESS in 1h 06m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/baf47114dc0e484cb54d2c7bb5855490 : FAILURE in 57m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06d7035ca9a94058b43cd7be01405c4f : FAILURE in 56m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/317830ae0e3043c7be74ec09365677e1 : FAILURE in 1h 03m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a2d133ec7c34c129a6d188b322cc670 : FAILURE in 57m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e34815b3cbf43b48d84e346e3ca7aa3 : SUCCESS in 1h 14m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f37be12fd1fd4f3ba03f10f92b08d3dd : FAILURE in 39m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07f5708aacf94b23bbdb39faec300e5d : FAILURE in 58m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d9bd2672a63745428d50d07d805510b1 : FAILURE in 33m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3eb86c6f66d34e45b78f2394d76a41f3 : SUCCESS in 1h 04m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/756ffa3edef94a31832331a57d9c455d : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"be264cb7db069dfdaf65148efbdfab1199730af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-16 08:30:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d2f28c4a1572abe6526e6f47802ef95acaae157f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 09:36:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c19564803644c39b3f7efca8f6075b1 : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f11ed086e3bd4d9b8dc1868303c86f89 : SUCCESS in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cd972674109c8f3ade91d130a7d4fe3036162ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-16 09:44:43.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/99/786399/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"143a3d79622df62925baa81a03922ad15ecae293","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-16 10:07:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/4 : SUCCESS in 1h 36m 07s","accounts_in_message":[],"_revision_number":4},{"id":"bc960a02ea24d0ca77158a83bb89173de2de4313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:10:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/410859a3e9e643358bc9ad7bdf37bdf8 : SUCCESS in 47m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/601090a53a754549b0ed4e9e0cd9bd11 : SUCCESS in 2h 33m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1521359644564a0bb573bcec4e25724a : SUCCESS in 1h 31m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bafdf410360240ff849179ca26d17833 : SUCCESS in 1h 30m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/554ce3a07cb04084bbf2c8fe28c1cff9 : SUCCESS in 2h 05m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0ca75190c1e45b394bdcfa06fca3f57 : SUCCESS in 36m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2dee46d262141aba289b7174c62123a : SUCCESS in 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053e6c75f2e9448f99a05538f8565921 : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/093f6ea420bf4d0095105f9654fda661 : SUCCESS in 29m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/700d546c2c3643779eb07ce7b28986e7 : SUCCESS in 26m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1554ce32a2b6497d90a20d20dbe09978 : SUCCESS in 29m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d046ff7ba2694e2f9982420ef6f18b4c : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e991e5727524767a14a3045dc99d2be : SUCCESS in 56m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab2a7fc37d324cc4b1d7bdf1d966fd41 : SUCCESS in 1h 27m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/621f08b8c97c477e87404dcf7610635f : SUCCESS in 1h 38m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c2ec673a2634eb287a363d1cd6d7794 : SUCCESS in 1h 04m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60e13cf91e034bbfbd183fe9495bcdc5 : SUCCESS in 1h 13m 27s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2464a3c35bb84c86a72f606bc94cfa50 : SUCCESS in 1h 34m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa0c71a72e2f4348ab0f749214549a40 : FAILURE in 1h 11m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/891fa0f6e8ff443499d6283c57374ad7 : SUCCESS in 1h 41m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0356312f5b8843128c6377531047f386 : SUCCESS in 1h 10m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f0d3218a52de4238ba2ccbdf86324f35 : SUCCESS in 1h 13m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c770054a23324badb272ea8d3f57415c : SUCCESS in 36m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b176329988c4eacb0168a050dff805a : SUCCESS in 1h 40m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0cc06278a0494adbae3ff7a02a4bc160 : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4f6f9d3951292fc255925842ea46f42e54518cc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-16 12:03:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"bb2b23e05c88e7545fc4b5cb891732a5ca888a84","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-16 14:20:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9fa0731675d9991a6c48ea8dc75caba45f4d0489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-28 13:59:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a52f07645fd22bee2f7ae5b3b77783d2487cd297","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-28 14:07:07.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/99/786399/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"69ec6d3caa3537c86a38afe61d580511628a49bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:41:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31d258c29d7c41cf9d8a62c3ca79ddb9 : SUCCESS in 35m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fea8677119046b1999d5164f591e143 : SUCCESS in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e2ae8aacbb1b94935fe00221b653b4c6921a0799","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-28 15:18:04.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"68bb652c12f86619198e071885dbba40c7c6622e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:20:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/856e56c69b1f43b296284b85a12e4bb8 : SUCCESS in 51m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b5901a8e7f9a47ff832089a0aec8211a : SUCCESS in 2h 14m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a0d89b33a20c43cfb7aa7d1f30c65f73 : SUCCESS in 1h 53m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/463d653fadfb45cf82b991c8394ff550 : SUCCESS in 1h 43m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a1e58f47fd054e05b7713cc4d9477913 : SUCCESS in 1h 58m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be3e4431720f4e5bb34449f24ac42c26 : SUCCESS in 37m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b7891b42a04139bcdfa51cfb8d5ee0 : SUCCESS in 33m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b898739c785d4c249061e388df23b167 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8515ea46fb2147f7a7c8bb84381ad239 : SUCCESS in 27m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d666d681994c5d9d0625cdb6967d27 : SUCCESS in 25m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e552310cd84634ad9f2b19ea12110a : SUCCESS in 27m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3837c3de7a8943cfab6921072b36dfb5 : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a45a0231da34396b66dcd49a01d3ca2 : SUCCESS in 57m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/183b6bacd9e1455baa945515312cfdb1 : FAILURE in 1h 07m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/620efce8ae864b59bea1eb1889aebaad : SUCCESS in 1h 25m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d4089a8550844f8aebe5ac9595e815e : SUCCESS in 1h 10m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4d8de6fd3b654a8a9ae6f08326d85c5d : SUCCESS in 1h 01m 35s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4590263fb2954edfb921f560afda1920 : SUCCESS in 1h 18m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e34aa186e3cc4f42849a3c2f56da8ce2 : FAILURE in 1h 08m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a3c15c0a9fe418fb335dc06fd40c340 : SUCCESS in 1h 49m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9e98b91fa1f4fce9660a270a4b5ec49 : SUCCESS in 58m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4377f111a193420283969309b3cbb738 : SUCCESS in 1h 01m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6657a40c164d35b33c39477ec812ac : SUCCESS in 34m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f22980bde14b4626bb0c9497fe2e314b : SUCCESS in 1h 21m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0190f748eadd40f0ba6a166f8e5ff074 : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":5},{"id":"c91ebfd8aeb77b0da638c948d6341a07b22fd20b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-28 18:41:12.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"dbf2c05304ea3abb7b60bc057d2777ab62071cb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-28 19:20:48.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"33015d2baa72ca7bd00e4b98d02e6c60b7c60596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:27:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1613b6ff7b5c4556ad7aff4ad83e7023 : SUCCESS in 38m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c689a57cf49a49b8926db3ddc97858e4 : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"86790eb9883f85d2e1bbe7a3d5654c3921564213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 20:54:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70377acaf19440659f4886fba10614f1 : SUCCESS in 42m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f4a1935d22c84406a4774b6e18770e1d : SUCCESS in 2h 06m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/672e63eecf1c49dca8196554e28cadc5 : SUCCESS in 1h 40m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5006c7496b714d76888bb85ee2dbaafa : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/551811602df74408aa6a5ffcf0ba3863 : SUCCESS in 1h 41m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70cde2967c64b2c893c834dba86d21f : SUCCESS in 30m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74f96c52ff3040799c0039a2ee0ff249 : SUCCESS in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24390495d58c4c3585385e6f18cd5f6a : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72005611d79d48a4bb4f48ac31780aa0 : SUCCESS in 26m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19565eb90fd24e1f905151af54342c55 : SUCCESS in 33m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54554da828954ab290c9bbf897bde432 : SUCCESS in 28m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7010fa143364ccebe3b413ecaacdc9c : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b938e88c9664b22b6457d4e3e632a7d : SUCCESS in 51m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/85cccfc44d3e485c9e3b0db16e952d2d : SUCCESS in 1h 29m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/183b32e83c4d435e85707fe67cb1f604 : SUCCESS in 2h 05m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e64f5df4ee3433d868dae7dec1eae6d : SUCCESS in 1h 09m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/684004fefcf34c7cb75ddcb97576456b : SUCCESS in 1h 06m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/917cf185abb14c4aa2deda9219e1fea2 : SUCCESS in 1h 25m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8977bf0b1918437e85b91406c2142f3b : SUCCESS in 1h 22m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc354542214046b79ae5ae6280c0c926 : SUCCESS in 1h 38m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc0494f4193c4dbaaf10b791cef67f57 : SUCCESS in 59m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/480cc1ae09734919b31e282deae36823 : SUCCESS in 1h 03m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d3869b4ccd224ac5a90aaa44526cab74 : SUCCESS in 39m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3aeafa4fbc3d44d99132813dd4d87e39 : SUCCESS in 1h 20m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1d2318cd59324745abf98ee6ab794aee : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":5},{"id":"4043158dc5e710131efb98c5f76514c1d05a6a37","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-29 06:52:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0c7cd61ecc31f2450be4d231ad16b43c615f3933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-21 15:03:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"131fb4ab03a10d4071da3c20cf21689c0e634cd1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-21 15:14:12.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6cbdd3de771edb656cd7edbb0203499bfdf14a5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 16:01:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88be4bcfdfa7423491eb7e14f80c74c1 : SUCCESS in 51m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4fefec3804244878b025214d95cb587 : SUCCESS in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"54ab56ddda18dff55d41c173ef3b646832a2d186","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 16:45:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/6 : SUCCESS in 1h 39m 55s","accounts_in_message":[],"_revision_number":6},{"id":"ea89c29d032c9daeb52af6329cc16cb9f2363634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 17:17:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc7cb587f79345cdab22e29e3ce756af : FAILURE in 20m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/240e307541da4ec182320db44171128d : SUCCESS in 2h 08m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d15b06b846294adea6f8b61b73eba33e : SUCCESS in 2h 00m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a26fdace21704e4c8b1caa6c59704e57 : SUCCESS in 1h 28m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f65fcd3c6d384e63a7bb6ff0018b6226 : SUCCESS in 1h 57m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85507c55ddc47e0ba5d9870101fc5e7 : SUCCESS in 37m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dea84a08c6b4092925d322cfa6c978e : SUCCESS in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303bda23f0ce4ad9893bf9b38cd84959 : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e16a29af7c42829f8b33ad0ae23f61 : SUCCESS in 25m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd68a8f36b94578bd61930128a96cee : SUCCESS in 25m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9534e9488a1a4520acb1b4dfcf68bc5e : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493ef69c74974dd9a75647d20781c0a9 : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cd341035cc943a892af233dc6d1ab16 : SUCCESS in 46m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6613a5fd4894331aa254a3435a450e6 : SUCCESS in 1h 19m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/94abb883a7a84a689783c8e08b17e74b : SUCCESS in 1h 13m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/beccf6a887264c5080ad645000f1ab0f : SUCCESS in 59m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/793f9cfcbf2f450c89a5450d4e239038 : SUCCESS in 1h 07m 06s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9f0d9821bfb7493c9f7078948753c53c : SUCCESS in 1h 16m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3ffe0cdf54e4e3b8aa46cd6f22668f6 : SUCCESS in 1h 21m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52bab589ada248b9a636fe4db39627d2 : SUCCESS in 1h 26m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27ed091e5f2743a28012856c4554caed : SUCCESS in 1h 00m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b56f14d11b8a440886e69d1775a161c4 : SUCCESS in 54m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/94b61b2c4a6341a397b6c6094833c127 : SUCCESS in 30m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/011c6da9192c42a5bf5da90bd20510f6 : SUCCESS in 1h 19m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ffb6cc148974cd998a8f84510b4619e : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":6},{"id":"e9629e391068b7eba1b58912b6a3941da274efde","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-24 08:05:22.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f84963196ff5fd1da268721c9509fe960960f73d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 08:51:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/752c98fbde1f4b9a852acadf5adafc3f : SUCCESS in 43m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2911a8370cb84b70ae0d9aedde7ff6a8 : SUCCESS in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b59bebf31abd1580fc1d4e271b79ec50455f06e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-24 09:13:37.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"b3128bfe045b707e86637f0335e313f9505f9ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:00:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0130ba137f844796b577f79075b5f7e0 : SUCCESS in 41m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11924c74c3a54846bd8901e46bcc1a86 : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/745edc60e4764e2798b50fc1d72585b1 : SUCCESS in 1h 53m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64b0794bb0f443268e04e53b37de107f : SUCCESS in 1h 23m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7146f644506f4b03bc7e8ab7571cf2e1 : SUCCESS in 1h 40m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/464a9f5742e6492c88334cfa8dedfa60 : SUCCESS in 30m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0321016518f406998e1690040747eca : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c3381412d04d1a830afa83837971d2 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/011ef75498ae45689607f0d5e29a4ea9 : SUCCESS in 24m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124520e543f74a68be9ffb19ba7ae138 : SUCCESS in 21m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2945082a3b547b39f89433e76e31cc4 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95685b77178c45d6bb308052b9982079 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49e45361e36740c483d4a07f9cd5cd6a : SUCCESS in 45m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce7bd14ccaa7460ba7d324f1fa942a7b : SUCCESS in 1h 22m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7effebc175e041e889a49b20517c0b89 : SUCCESS in 1h 11m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac58d270f58d4bc6a5760307353d11a4 : SUCCESS in 59m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/62c2858fcb254621aeaa3ab04acdc819 : SUCCESS in 58m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/71fcae5c9636415085ce53cca527018c : SUCCESS in 1h 08m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15adbb2bbb9246049c454923f46ed865 : SUCCESS in 1h 18m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47784c11b48743eea6c7d8f4b6ceb83f : SUCCESS in 1h 25m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c06de9ceebc4caba7a8c655b2dcead1 : SUCCESS in 53m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de98cb89fc814a4cb1d60f47197c3825 : SUCCESS in 1h 01m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3997d551e44bba9a8b8b00fcdd47a4 : SUCCESS in 30m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a356ea3ff2342da90eb57b48c88d548 : SUCCESS in 1h 16m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a59d1027122a4e1b8d8f22a80648ecb7 : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":6},{"id":"c0f2a942ed5378ac042e758c81d381285e29622d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 10:22:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/6 : SUCCESS in 2h 16m 29s","accounts_in_message":[],"_revision_number":6},{"id":"7285e66fff6f282cc1e97d4f61cf78f244ad6680","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 16:13:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"cfc714e45275964ab6c9a3a71e0db8452ff6165a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:14:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9d75d4244a4f97a908f611d6c82030ea7275aadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:37:52.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0133b6c5a0014818a47f4072fa511245 : SUCCESS in 45m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ada99815c665436bb9ac1c4b86ec6b2a : SUCCESS in 1h 58m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4c62c49330e340a989f0b08c8b99872b : SUCCESS in 1h 55m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6b95fdf3bbeb4b0e9631246430e80541 : SUCCESS in 1h 53m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60dca3f7951f43af8e03322037b3bae4 : SUCCESS in 1h 06m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc33e4dff2404e05a36fedae511c2c3f : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09501d5efa8c4203b07b7486490115b7 : FAILURE in 15m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67d9bd49eb794bcf8ef62184e4be13d1 : SUCCESS in 27m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75c61ccb2e184a5ea36d4433942160c5 : SUCCESS in 28m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f8fde4076e4a0d86d80b9c54b038e4 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/282ee34a804342e8acab842ddcf5e0e3 : SUCCESS in 51m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecabfeb3bf0645389a76715c4527967d : FAILURE in 1h 12m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/03ddf8c1aef64b2a92c82071362fe2ff : SUCCESS in 1h 11m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/524716a67fb9400ea80a3e400306f97b : SUCCESS in 1h 00m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0db3e1667fa444294e8554f268c0cee : SUCCESS in 1h 00m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2f9430d261f2438da1a1db04a5b8cbe5 : SUCCESS in 57m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/387f3bbf6de94a0bbf88f1cb53b87792 : SUCCESS in 25m 53s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":6},{"id":"5b5685b3fccb4898255b0611ffb610a902046ddb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-06-04 07:56:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f30254f15cdaf83a8670a4dcb06ca4c1cada99aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 23:02:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ca9a8e8d508e7e59bdbc01edb227083401f47690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:29:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d23115d8ee4a59a17916d4cd9811cd : SUCCESS in 34m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca0a51afd4334d3690c65eb6ca4c33f6 : SUCCESS in 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da686cec46564e36a758bd8cdb7a82fe : SUCCESS in 16m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f533a6baf6554f0a8bc255d0921cf7d9 : SUCCESS in 23m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c495022c3d924ee7b88ec2a500047a2f : SUCCESS in 34m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c9ea0fd2bd48d9869de400fe365311 : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97fb7a219ab419ca6a95e4b20c6bec0 : SUCCESS in 9m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ad061671f8b4a269d7970a1509a1470 : SUCCESS in 1h 25m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f19d662e2a94178805f458d5e228d9f : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b7a12204eb6f8aa223c159ee66348dee04e52c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:29:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8edae99b184c93e931db66ef74aa3ed7e5633967","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:31:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e6afb241fa2406c97192a1258c37176 : SUCCESS in 55m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e507bfbc3b6a40819aa2aa6eea35c58c : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4afe9b3b6e377712b481ac52c49c1cc597d764a3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 01:24:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/786399/6 : SUCCESS in 2h 00m 56s","accounts_in_message":[],"_revision_number":6},{"id":"b10e8cbf38a8e010a37ec78052b10e783fe3ae3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 04:06:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be06960701884b2ab493adbe5b0914a9 : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4cd93fe5a34c99b9ce752a71d4f846 : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77190a2179c8450692b4ea7b8aee426a : SUCCESS in 28m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e07df6168fba41e2a9f4afddd39a9db7 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d489146060964a1c8ea6d1fe1608b3f0 : SUCCESS in 9m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee83f16199924200b28edab68df7cd96 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":6},{"id":"032529ff8e91e05f43ea2bb52ee956df1b646184","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 04:06:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"190bd37aec56c19528e50016fd6b308c363cb1bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 04:07:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2aac4358551471c8fff2d13a024929e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6},{"id":"9b6984840a3416a00ae229a4c00b20c8fd7b739e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-05 10:04:30.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/786399/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f52280287f956f67c7b9c5c53ad945d96dd69103","revisions":{"9b375d8acfe72af378ec20c84debf43b140279ef":{"kind":"REWORK","_number":1,"created":"2021-04-15 12:53:15.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/1"}}},"commit":{"parents":[{"commit":"504b15d8e2a516553cca33e382b38935f8c88f03","subject":"Merge \"Revert \"Cache default security group IDs in memory\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/504b15d8e2a516553cca33e382b38935f8c88f03"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:50:47.000000000","tz":180},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network dict from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b375d8acfe72af378ec20c84debf43b140279ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b375d8acfe72af378ec20c84debf43b140279ef"}]},"branch":"refs/heads/master"},"49ca7f9e1093fef9eedd7308b43f22a2e5842c2c":{"kind":"REWORK","_number":2,"created":"2021-04-15 19:36:26.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/2"}}},"commit":{"parents":[{"commit":"504b15d8e2a516553cca33e382b38935f8c88f03","subject":"Merge \"Revert \"Cache default security group IDs in memory\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/504b15d8e2a516553cca33e382b38935f8c88f03"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 19:35:26.000000000","tz":180},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network obj from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ca7f9e1093fef9eedd7308b43f22a2e5842c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ca7f9e1093fef9eedd7308b43f22a2e5842c2c"}]},"branch":"refs/heads/master"},"384c2917dbbad5cb4d1f059bac0b9b2cff75c20f":{"kind":"REWORK","_number":3,"created":"2021-04-16 05:44:56.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/3"}}},"commit":{"parents":[{"commit":"504b15d8e2a516553cca33e382b38935f8c88f03","subject":"Merge \"Revert \"Cache default security group IDs in memory\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/504b15d8e2a516553cca33e382b38935f8c88f03"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-16 05:44:08.000000000","tz":180},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network obj from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384c2917dbbad5cb4d1f059bac0b9b2cff75c20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384c2917dbbad5cb4d1f059bac0b9b2cff75c20f"}]},"branch":"refs/heads/master"},"6c507b5c2dfec319e690e50c06809520a41b9b38":{"kind":"REWORK","_number":4,"created":"2021-04-16 08:30:57.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/4"}}},"commit":{"parents":[{"commit":"504b15d8e2a516553cca33e382b38935f8c88f03","subject":"Merge \"Revert \"Cache default security group IDs in memory\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/504b15d8e2a516553cca33e382b38935f8c88f03"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-16 08:30:11.000000000","tz":180},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network dict from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c507b5c2dfec319e690e50c06809520a41b9b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c507b5c2dfec319e690e50c06809520a41b9b38"}]},"branch":"refs/heads/master"},"88efe49e6d2ba6e1153ab4a348a38b0d42fd83bb":{"kind":"REWORK","_number":5,"created":"2021-04-28 13:59:26.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/5"}}},"commit":{"parents":[{"commit":"52b14863effea218780b41f1f698aabfcbdaf19a","subject":"Merge \"Fix \"nftables\" ansible role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52b14863effea218780b41f1f698aabfcbdaf19a"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-04-28 13:57:06.000000000","tz":180},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network dict from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n- don\u0027t count subnets (DB request) if subnet has no segment\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88efe49e6d2ba6e1153ab4a348a38b0d42fd83bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88efe49e6d2ba6e1153ab4a348a38b0d42fd83bb"}]},"branch":"refs/heads/master"},"f52280287f956f67c7b9c5c53ad945d96dd69103":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-05-21 15:03:52.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/99/786399/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/786399/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/786399/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/786399/6"}}},"commit":{"parents":[{"commit":"4882fa34bb5cd8ebce52f9333bf564c519b6f26f","subject":"Merge \"[DVR] Send allowed address pairs info to the L3 agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4882fa34bb5cd8ebce52f9333bf564c519b6f26f"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-05-21 15:03:52.000000000","tz":0},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network dict from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n- don\u0027t count subnets (DB request) if subnet has no segment\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f52280287f956f67c7b9c5c53ad945d96dd69103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f52280287f956f67c7b9c5c53ad945d96dd69103"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
