)]}'
{"id":"openstack%2Fneutron~787304","triplet_id":"openstack%2Fneutron~master~Ibf7c08e147ace055114556eb14ca013fcf71fe6e","project":"openstack/neutron","branch":"master","topic":"bug/1922653","hashtags":[],"change_id":"Ibf7c08e147ace055114556eb14ca013fcf71fe6e","subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","status":"ABANDONED","created":"2021-04-21 00:46:37.000000000","updated":"2022-02-08 08:52:33.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"06b4338873d9c8f1d3c756724e1f16ccd133d002","_number":787304,"virtual_id_number":787304,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-28 17:35:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-24 11:07:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 01:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 03:53:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 04:10:30.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-21 05:56:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-21 07:54:16.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-04-21 13:41:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 10:45:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-07 11:52:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"287084bb9d4bf53111b5b07651f2109b3ab5b05d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 00:46:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f447141f4fd0a2ec4be7aa96e0f2ef3cb9fd688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 01:39:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5af8d45fc6f541c3b0bc54f7cb1d1c8a : FAILURE in 26m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d30a62cf43534fa5877032d11c8729e9 : FAILURE in 28m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d049c3d8a111456c8b8f736102f6e79b : FAILURE in 21m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bac5d384ca174393a90598453e5682b3 : FAILURE in 28m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1a24c9e1f20849a4b395dbd4aaf0f094 : FAILURE in 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52f49096f0414552b72afbeabb0fce6b : FAILURE in 44m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f35afda395e446196f3c2b088650282 : FAILURE in 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d65a6d00c04c3da040515ad1967022 : FAILURE in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/783e31d6951b49aea28f78891cef61f3 : FAILURE in 32m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda3d78159f3420a9328cb996fda2fe4 : FAILURE in 22m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa273c97fee49e5b6e62508bd522201 : FAILURE in 22m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848ba4cb968e43c8916c15459a620aa2 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9eeef2198f443a1ac52ab2ace351ce6 : FAILURE in 47m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/452e034a7b1745ba988042bbde2c496a : FAILURE in 15m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/69d2faabccd54b64bbc977b2896b8368 : FAILURE in 19m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/432b8b3af1be4951bb7e792e6d5b05b1 : FAILURE in 50m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bca922de9bbd4da88c8bb555fb978dcd : FAILURE in 49m 10s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f48ffe53ce4146c0ac4fa33e3803ce72 : FAILURE in 22m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e2e6909d6283421381965f96d565fd50 : FAILURE in 22m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5551f324d914c0da89140c300f83de2 : FAILURE in 22m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/669e36ce00d24f81bc83119a50f74caf : FAILURE in 18m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a54f3c93e71d464a85bfc2e95381ce33 : FAILURE in 20m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7425c4520cda4edfba33898e51ea359c : FAILURE in 20m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e5f45033832e451a9744c74e96690cb7 : FAILURE in 21m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e02384597e0a4b39ae08cb53c62284fb : FAILURE in 26m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cd80d40d00b5ca0faaad399c356b87fbd032b72f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 01:53:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f989403914844e8b87e7319ed57e03ad : TIMED_OUT in 1h 01m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/531e639eb2c144c380ca84715e7011eb : TIMED_OUT in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9a93efc1eb1db93ed75a479363fb1b17ec3b3da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-21 03:53:52.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/04/787304/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1440c87d114511e354c96499da5617a28acfc219","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 04:10:30.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/787304/1 : FAILURE in 3h 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e587011e24c5e273594dde6451d8f418f72aee50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 05:40:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc5992c1a32ad0ea6a3c7f456968195bf0bc7a36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 06:27:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3af8a39b189c48298500386dec3be85a : SUCCESS in 45m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5eaa5e13b164c6a9d01c777d7b83224 : SUCCESS in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adcc6bb1eb5c6ac2ade001db30c5815b62d6716f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 07:34:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/2 : SUCCESS in 1h 52m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f65313d9b25322682bc116725579a8d6305c312d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-21 07:54:16.000000000","message":"Patch Set 2: Code-Review-1\n\nI think this API change deserves a release note","accounts_in_message":[],"_revision_number":2},{"id":"d7c521f6b9ff355e1f1bb0c3341d8a2b6378caac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 08:20:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87d9ebf0425444618328a3d514287ed3 : FAILURE in 40m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1db0f2e54ea04ef4a9cc9fb8823c25d4 : SUCCESS in 1h 35m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f33372e7ba5c4177a28146de5d59a10d : SUCCESS in 1h 21m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c065a4bb75a14c6995368b2e18aff5b9 : SUCCESS in 1h 21m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67f4edb2f0004ea58d47bc585821fe6a : SUCCESS in 2h 37m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cea0632cfce42119324bd68cf2e79d4 : SUCCESS in 26m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74b09f98dc0d4dc18ac8c51bd5cd79dd : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3098642caf974e2cb55549ab28552d34 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03ae91a7987342008fe201ca2b222f97 : SUCCESS in 27m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7698e03f26fb4f8eb4aac8a1451dddd9 : SUCCESS in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69e40deabf17445b936e8a9b065cbf4e : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecffb22f91e4abba3658a3c63340e6e : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/747742beb8874218835e61ba385fe89e : FAILURE in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e445805ec2fc44969509112425c10738 : SUCCESS in 1h 23m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cb82d5aa5cb2420699dce752f03b2abc : SUCCESS in 1h 36m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5a74ffe58e748dd91e3ff05f476b3e2 : SUCCESS in 59m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2319c44f744d44c38f0d324c9039c31e : SUCCESS in 1h 07m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4aaeac9b2d514539972ed152d0663438 : SUCCESS in 1h 03m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6519b1d224a42899d64995c46e286d7 : FAILURE in 1h 05m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dde022c420744d1c95a6eaf2958e71a8 : SUCCESS in 1h 28m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d0a0bdde8f54fdca0ec8888b5261f46 : SUCCESS in 43m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e9fb7a1a35c14fcab4bffa3934c344fa : SUCCESS in 1h 04m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6ebaccdba542f9abf02fd784c73d0a : SUCCESS in 25m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9e07833ad2d4920826c97399ae8b664 : SUCCESS in 1h 37m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d2c92226761242ffac3e0bd81b80f3f6 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7c332246eaef8cddc7192324fb832fbd83b82fc3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 08:53:51.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e I think this API change deserves a release note\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"f1be5e84ff5dd0ad3b12c2bbd5cb640458987564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 08:54:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24b6f3bfb152982901017d737bae562e13d9b2fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:45:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2a18bf31d664b8891f4a8768592524b : SUCCESS in 49m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4595d95f350349b188363aa2173c0873 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8643ddb308479f2682d8cf10b44f8594c2bd99a2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 10:56:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/3 : SUCCESS in 2h 01m 44s","accounts_in_message":[],"_revision_number":3},{"id":"cc3159bcd0f714884f8b6d0ec5da2de7906d969b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:06:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ac49ecf11554b42a7b1ef358057e640 : FAILURE in 42m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7dd7ed2141134e3984ecb723093f3c9e : SUCCESS in 1h 49m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8a9bf066a4f046d08618a954504a98a1 : SUCCESS in 2h 05m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c18e0b8837a74544aad53a4b668ce0ec : SUCCESS in 1h 04m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/74b8784049004a2398f1269f26813550 : SUCCESS in 1h 52m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ab8d34f8ebf4ca19239253527306544 : SUCCESS in 36m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba78d2c870f5498d9312f9fb7a343843 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d07079ee2e4864af460d4e9711ac3c : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/001c8adc64e24625ae4340cee002a318 : SUCCESS in 28m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0759ca28e0324378a471464cd6b2d067 : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00bc38bc0d58401c96bb532ca0ce3051 : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86901f82a8e84411837dbac707ff8d03 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c03b147aac484339a01a7fed88d7777a : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/822f59922a0746a292b89cc012bbe1b7 : FAILURE in 47m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3c9904e73034b59b16af3adbad08430 : SUCCESS in 1h 22m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3d31234df51b40bc9f4d33b89f34d0f5 : SUCCESS in 1h 17m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6288bb570d5441888e31500d961a14b : SUCCESS in 57m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3587ae4ffd9e4107badcebfddfb48b85 : SUCCESS in 54m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/49a148e975a34715998b94cff0acb349 : SUCCESS in 1h 16m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6075ed21182945a3b241842bec3600ea : SUCCESS in 1h 36m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6fcda7d2f1b450e9c5d2e110ffb874e : SUCCESS in 1h 24m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7c2350dbcf64937a3963f7624c5b08c : SUCCESS in 1h 02m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7163d628ba10459ab444a9bade97858b : SUCCESS in 58m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/29bb5dfc6faa41229a450d053c4a7dab : SUCCESS in 34m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2367af36d65431ea3f025679f814250 : SUCCESS in 1h 39m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c0895bb6afc471f81f0dcd6e55cf15e : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f609d6dbe5d7e620655d8e0d0016cb919c59cef8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-21 13:41:25.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/04/787304/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e01a443acfb1569e4c4fa2243ba4f34896ce9ce6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-21 19:35:36.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/04/787304/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"bc330ecb8352817de7b0c3c01df2fb15b9cca258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-22 03:39:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f26258eb5c992d9ec31f3f8916602ee5fb98ff9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 04:24:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cad44590b8974d65b708e0849bb50d27 : SUCCESS in 34m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27cf44fbcb894d26b851b39fb872ddc0 : SUCCESS in 42m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3a1eeecc9458cd39b9374327880b5c90889d4b2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-22 05:43:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/4 : SUCCESS in 2h 02m 59s","accounts_in_message":[],"_revision_number":4},{"id":"9f39e2ea2ed43c4e262083bb03826d5b2b48e694","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 05:45:03.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/04/787304/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3ef80711079308551cad4a1430f976f155265f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 05:47:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ce4b07b4493441de80118592c0d0faa5 : FAILURE in 41m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0a17868010c4082a373b4c6e7ff90a1 : SUCCESS in 1h 31m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/60cee3586ce24bc69c1c0d816c3a251a : SUCCESS in 1h 30m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/866c6c376c3c4e52afb6f4c84f631c01 : FAILURE in 1h 18m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/76e6e4668f2342e29391f797828496d9 : SUCCESS in 2h 06m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc3f15e6bc734724ad6a17e2ceb0dcb8 : SUCCESS in 34m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e1abc98943440dca03648df68a9a0ed : SUCCESS in 44m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/732aa4587f904676aefe5f2497add16c : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6283f7badbf849c29f1e658d40f15e83 : SUCCESS in 29m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29ec80a6300f4f9f9ec3f403af3decd3 : SUCCESS in 27m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/494e26d84f604e0c9acdb4e9dced4fbf : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e46147ac784b94ae3f56dedcbc1267 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fddf886bae9440e38b30bd3d0e73c8d7 : SUCCESS in 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54722a118de84df0aca92fc7adcb59b0 : SUCCESS in 49m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10e19bd0596f426c8f3a17b9d5344086 : SUCCESS in 1h 10m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bb2e5e67956e4c39b019d42232070c96 : SUCCESS in 1h 11m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ade5a568a1f4437395c1c971574e0c0b : SUCCESS in 55m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97acf50fc1bf48e396ac5aa4cadc0ad9 : SUCCESS in 1h 15m 09s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/728edd1cd3ec4211b92d8781c9ba95ba : SUCCESS in 1h 09m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ea3ca9202854989af3ee9020417c8ea : SUCCESS in 1h 18m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c05d1c16c3ae49e4b2f535e4c9b7dfac : SUCCESS in 1h 37m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f8038507e7849f4a7f049c4a55cd9ca : SUCCESS in 56m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/85bd2c64f0dd4e77a9168b6b2a4b645a : SUCCESS in 57m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a5b6d14e2b2248b8be9f285c1dbfb2e4 : SUCCESS in 28m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35b9285e4cd8405bacabddc92e2a4f27 : SUCCESS in 1h 12m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/768866619596456ea40526b7da7e6264 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b64e91f5a2e5baabbeadbeec30e3bf90ca841f9a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-22 07:07:23.000000000","message":"Patch Set 4:\n\nNeed to remove the case in neutron-tempest-plugin [1], patch uploaded here [2].\n\n\n[1] https://github.com/openstack/neutron-tempest-plugin/blob/master/neutron_tempest_plugin/api/test_port_forwarding_negative.py#L66-L84\n[2] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/787474","accounts_in_message":[],"_revision_number":4},{"id":"41320ba4b676c4fb9afcc3c6e8298d8373baac17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-22 07:11:45.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f0166423e7725bc8d9cfef99729ca3b4d5b1f850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:16:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10e29c8c91014f25b905428da4f8bf82 : TIMED_OUT in 1h 01m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e08d4b69cacc4c629a8015ffaf447f1d : TIMED_OUT in 1h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d2507b5405d7da6d42e13e27b1e2d6f132c30eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-22 08:51:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/5 : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":5},{"id":"f1d835424f66121d3ca0a21730437acd1f5fb8b8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 09:43:50.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/04/787304/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e8f4851a2ec8373b8736cb7bcb373dc93556b656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:38:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/54074923b5ac465a97af50e58f121689 : SUCCESS in 59m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c273670e65b4c6d9f15630d1f3d969e : SUCCESS in 2h 33m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a521bac365ce4b05be24f5f558717f2a : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/427265e750334728817e3a76349ac89e : SUCCESS in 1h 43m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0af89f542fa340a2a33e8789133658c0 : SUCCESS in 2h 34m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6caf3393a0e455da7349a7e8baa499a : SUCCESS in 46m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/418b273f48b348709b217cb07bbc2af8 : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654f9c9f70434905af04a4edd988b4bf : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3267a7d75c4c4e339d0d86a1e0e0e827 : SUCCESS in 26m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35e05ea996b04ce18be8542433160d1b : SUCCESS in 23m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d56ed9330545418e8152ff3c48fd0b : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49321817c71f44b7a65ad3fe17dc4cec : SUCCESS in 22m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d0742f42a2a4ce1b3c73c1b1ce4c53b : SUCCESS in 11m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/373ec20ec1d84b9eabeb180f4bda96e8 : SUCCESS in 50m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/944a4a8ba55845a7ae25714b1e27deaf : TIMED_OUT in 2h 15m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/00c7f78211fc4659880d13e5bdb85081 : SUCCESS in 1h 42m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d85d85f66c83475eb498e9e45b73a37b : SUCCESS in 1h 04m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28bf3e40c8c441339edadd2a1a9de5b1 : SUCCESS in 1h 01m 53s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9be0493e87b14dc0a99558631cee7e9f : SUCCESS in 1h 13m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c3ae63efd2049bca9fb88a40ec7e6b9 : SUCCESS in 1h 22m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d300cf3595b44c958c588da112fd3b2c : SUCCESS in 1h 45m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1ada74e7ff34175934cee2e74382e69 : SUCCESS in 56m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/900a84f1a4db4b7bb83fb87f7e92756e : SUCCESS in 1h 08m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a3ca48e7644942b5880493d3ef9e1109 : SUCCESS in 30m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26af8051de0341a2a41715b8b134ed02 : SUCCESS in 1h 34m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bc23f8baa58f4763b9eafcb92b3fa14d : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":5},{"id":"25fe0eddf742b151d363fb0000e4c71a5448ff62","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-23 10:45:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nThis is API behaviour change so IMO we should have API extension to let people discover if such association is or isn\u0027t really possible in the cloud.\nAlso I think that this change deserves release note.","accounts_in_message":[],"_revision_number":5},{"id":"7dbc518be4434313c101b71a3c2bf5c726ad2266","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-24 13:01:02.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e This is API behaviour change so IMO we should have API extension to let people discover if such association is or isn\u0027t really possible in the cloud.\nDid you mean a shim extension for this? I\u0027m not sure if it is really needed, we changed the DB uniq constraint once [1], and we did not do that work. And, IMO, it\u0027s not so much necessary, just a bit waste of engineer time. For the users, they have many ways to get this change such as git log, release note, LP bug and gerrit.\n\u003e Also I think that this change deserves release note.\nThis patch has a release note now.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/613549","accounts_in_message":[],"_revision_number":5},{"id":"0f5d4801e440ad0a1c979edf15002d2e44edf021","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-24 13:02:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7cd22310fdabbf32217229ec6c0331f8cb49e1e4","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-25 03:11:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3fa0b65f9e68ce53e8491a6b602fa54385b3b972","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-25 03:13:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b0dcc96add109cc9e8991444705f520167c33559","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 04:14:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea963346304342d193117c41d490739c : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9df0f2ab73b44eb08d9ed30f1548752b : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"318a0d03643b0ec95ace9aecb17627fde257aadd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-25 05:14:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/5 : SUCCESS in 2h 01m 52s","accounts_in_message":[],"_revision_number":5},{"id":"0ef1627305e9175505d0433879634c34218c9ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 05:24:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7a92de99f34e49d189ac3cbc8b6170cb : SUCCESS in 34m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/09925042df1244f584351317e72a5cb5 : SUCCESS in 2h 06m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/881f3dcd76e1424e872596959738cd08 : SUCCESS in 1h 56m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7a8f41d2b044833a8b806548d95a0ac : FAILURE in 1h 52m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ee77fe6913aa4ba5bb03890f864e5dd8 : SUCCESS in 2h 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ec438e675274da884f9706b6d920024 : SUCCESS in 32m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d661bedd2d47c68f8ea905d72d26c2 : SUCCESS in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5ce3b86fcf4c1a8dbd0c8b9e660961 : SUCCESS in 16m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ae8166443ba4f45a3bb669f9d1069a2 : SUCCESS in 33m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb45b52166fa410cae10da215994655d : SUCCESS in 30m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf7c8ee242784c56ba6744856e54cb6d : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73593f043154088ba073e7b0abd4913 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc185949574443c6bd6eae655a830cbc : SUCCESS in 11m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29451be88ffc4eda904add8b71e36954 : SUCCESS in 1h 02m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64368f16c48242bfa3b46df445baf60d : SUCCESS in 1h 20m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b0045e60b12d4c1e9aa503122f012080 : SUCCESS in 1h 51m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ef208dd1b84464ca0f02562b8e61b0b : SUCCESS in 1h 01m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b8356121a25a4f268088b80d3ee8b414 : SUCCESS in 1h 08m 53s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3f095f2d163d4e02a8a12e6414928887 : SUCCESS in 1h 07m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1ee71fc007af4ca5bf7bb662b648dece : SUCCESS in 2h 00m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87f5f78f52de4069990ca5df4b164a70 : SUCCESS in 1h 31m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b18e25747d864a939e551cc99f23696c : SUCCESS in 48m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ea071544d8645c0bc8e3702b5a763e0 : SUCCESS in 1h 06m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/06af324c67f341c0bdaa886b85838f8c : SUCCESS in 44m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd71ddb822744152bdfba012adba06cc : SUCCESS in 1h 21m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2ef073644c7b4793823035fd2d08ce3a : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":5},{"id":"3822ed32940436353a8be9b6296e7419003bad97","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-25 06:09:25.000000000","message":"Patch Set 5:\n\nneutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7a8f41d2b044833a8b806548d95a0ac : FAILURE in 1h 52m 58s\n\nFAILURE is not related to this.","accounts_in_message":[],"_revision_number":5},{"id":"6d1edf6726f51e608eb5d8a8b0ebf192078d3bc7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-25 06:09:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bd716f440716d1b2e354ea80df40d2dee41ccd7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 07:12:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d1df66373014cb4a8db75e3044b1e12 : SUCCESS in 54m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c919d50b40da4211bddb6a2f0a8a6ae6 : SUCCESS in 55m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1161fccd160255e177d98fc2553559c6b0051a1f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-25 08:00:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/5 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":5},{"id":"04da67f1a0e9f9e9636504fb675cc91f968969f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 08:37:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/930af6ea42ed4e788531291849bba4a3 : SUCCESS in 45m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/41391572370a462d89ce7d48a8e913e6 : SUCCESS in 1h 55m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/49d0c220566c434989c8669217a37c3e : SUCCESS in 1h 36m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb9e9a982afb40f789b3c9d08260020d : FAILURE in 1h 56m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7951c98970614a7b905499670555442e : SUCCESS in 1h 51m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a33ad5df78f84381b4e0813bed8078b2 : SUCCESS in 39m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72be349f0ee54a5c90538693386d6488 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9cd6f7292a4205bd78697e3ddb26cd : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d3c9b0334ec4e55a7eb37e90fd4c8b1 : SUCCESS in 29m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a916b682d86243fb9c4c6b6e9db078f1 : SUCCESS in 29m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b2a773e3b1e40f5be633e54db31eca6 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88369db6e3474598b878b324677b6235 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b25d41c00d104c0ea38f22d0d77ff79a : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f82dcc29f95d40a7bc574d99d527fbfc : SUCCESS in 46m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3167dd6dc3c47d492e8820ad0629b71 : SUCCESS in 1h 26m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fcde05dc606642d4b8e75e15409d6388 : SUCCESS in 1h 25m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d14ac61c7bb4ae486af572548265f81 : SUCCESS in 1h 14m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/08ce2c5358cb4bf28c498c4348e0c73d : SUCCESS in 58m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/06f9f869e8064ddd8f2a2df4ee2d64bf : SUCCESS in 1h 31m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2a205a4bead94382a9f33195e6e1a24e : FAILURE in 1h 14m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26d7420f71bc40359b0eb9679d2eb46c : SUCCESS in 1h 24m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7785a6e90474d7cba1e4234c5224478 : SUCCESS in 1h 03m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1b46f8a06c9345aeae31dbe719e9b5df : SUCCESS in 1h 01m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3c46b4271d1c4f11b43f272a98cc8938 : SUCCESS in 30m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e321a7bfde64f20be1c5f341c3a6bd1 : SUCCESS in 1h 15m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43e190ab50ed420895a25973c750a044 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":5},{"id":"8ae9e22b020fb916fea76fd14619098eea02e1c4","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-26 00:44:06.000000000","message":"Patch Set 5:\n\nFile \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/tenacity/__init__.py\", line 426, in __call__\n    result \u003d fn(*args, **kwargs)\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/neutron_tempest_plugin/common/ssh.py\", line 171, in exec_command\n    return super(Client, self).exec_command(cmd\u003dcmd, encoding\u003dencoding)\n  File \"/opt/stack/tempest/tempest/lib/common/ssh.py\", line 162, in exec_command\n    channel.exec_command(cmd)\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 72, in _check\n    return func(self, *args, **kwds)\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 257, in exec_command\n    self._wait_for_event()\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 1219, in _wait_for_event\n    self.event.wait()\n  File \"/usr/lib/python3.8/threading.py\", line 558, in wait\n    signaled \u003d self._cond.wait(timeout)\n  File \"/usr/lib/python3.8/threading.py\", line 302, in wait\n    waiter.acquire()\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/fixtures/_fixtures/timeout.py\", line 52, in signal_handler\n    raise TimeoutException()\nfixtures._fixtures.timeout.TimeoutException\n\nMeets this error, there are some bugs related to this project \"paramiko\":\n\nhttps://bugs.launchpad.net/neutron/+bug/1892861\nhttps://bugs.launchpad.net/neutron/+bug/1571486\nhttps://bugs.launchpad.net/neutron/+bug/1253896","accounts_in_message":[],"_revision_number":5},{"id":"68c164204fc7abf9078424239974cf8475d254b6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-26 01:42:11.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/tenacity/__init__.py\", line 426, in __call__\n\u003e     result \u003d fn(*args, **kwargs)\n\u003e   File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/neutron_tempest_plugin/common/ssh.py\", line 171, in exec_command\n\u003e     return super(Client, self).exec_command(cmd\u003dcmd, encoding\u003dencoding)\n\u003e   File \"/opt/stack/tempest/tempest/lib/common/ssh.py\", line 162, in exec_command\n\u003e     channel.exec_command(cmd)\n\u003e   File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 72, in _check\n\u003e     return func(self, *args, **kwds)\n\u003e   File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 257, in exec_command\n\u003e     self._wait_for_event()\n\u003e   File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/paramiko/channel.py\", line 1219, in _wait_for_event\n\u003e     self.event.wait()\n\u003e   File \"/usr/lib/python3.8/threading.py\", line 558, in wait\n\u003e     signaled \u003d self._cond.wait(timeout)\n\u003e   File \"/usr/lib/python3.8/threading.py\", line 302, in wait\n\u003e     waiter.acquire()\n\u003e   File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/fixtures/_fixtures/timeout.py\", line 52, in signal_handler\n\u003e     raise TimeoutException()\n\u003e fixtures._fixtures.timeout.TimeoutException\n\u003e \n\u003e Meets this error, there are some bugs related to this project \"paramiko\":\n\u003e \n\u003e https://bugs.launchpad.net/neutron/+bug/1892861\n\u003e https://bugs.launchpad.net/neutron/+bug/1571486\n\u003e https://bugs.launchpad.net/neutron/+bug/1253896\n\nAdded a new bug for this:\nhttps://bugs.launchpad.net/neutron/+bug/1926109","accounts_in_message":[],"_revision_number":5},{"id":"488dc0da1da8faeb2a17594bfaa2fa9faf91b155","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-26 01:42:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7573bf9e75b09825dc8e1ed635f3fdc3019d9506","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-26 02:21:03.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/787304/5 : FAILURE in 38m 07s","accounts_in_message":[],"_revision_number":5},{"id":"d673ddae87d945ef06f01e9d073610d9d0435772","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 02:44:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3846c0fcb1842b3b4a45d47bbc4ea09 : SUCCESS in 58m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adbb4118b9a747dca630e2f83be4f445 : SUCCESS in 58m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b39eab24562f7d9f4b2e5f1499a0d180f1542aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 03:40:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0d0c08d107e4ff9a4d0f5c79d2d86fc : SUCCESS in 40m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/37818331f97844ec8a09f1a7863954f6 : SUCCESS in 1h 49m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/05d119e7dff6400cb82ea11940bc7636 : SUCCESS in 1h 39m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06d611202b3a43999889ca6ecd6a9694 : SUCCESS in 1h 25m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/496c7cca73654c49b5c84d4c8eabdaa6 : SUCCESS in 1h 45m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ac4e1c6ac8487486412585c7b563fb : SUCCESS in 30m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/803de21b804148f3a610d5d32d2fe73f : SUCCESS in 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b3c21633ba46aa8dba1bbbb91954ac : SUCCESS in 12m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c6cc574099f4d25b60130a7660d89ad : SUCCESS in 23m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e10d1a19c1143969b5e0f6769d2a1d8 : SUCCESS in 22m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1019f87663c04d40bc5c1e887f997bf1 : SUCCESS in 31m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d883906f9841e4b2d815d90f4a662c : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e69bcb061e414e7295ad4932593d0fe3 : SUCCESS in 8m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb3a811f5644448eb3fcd925d93f1bf8 : SUCCESS in 44m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3552f35cfe64d4cb9b717e666769bb1 : SUCCESS in 1h 24m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3a7763bb4c4f45b8a29986e0434c1741 : SUCCESS in 1h 27m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3a168e5d6624346a9cc284c6f2ad0d1 : SUCCESS in 55m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f4eaa30ea9df4856b2d3f2a9373cd72f : SUCCESS in 58m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa00134b2bb94d2b864c9d409a733f33 : SUCCESS in 1h 45m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/399bf10bbee34ec8a9791c18f8498bf4 : SUCCESS in 1h 17m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3809075c0d634b4abbd03fe631187abe : SUCCESS in 1h 36m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a95c7c42204340b184676d53804617 : SUCCESS in 48m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/46617d80fea2478487f1ddd4df6e8fe3 : SUCCESS in 1h 23m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/13ab0f60291e4f71b4fa45ffc4fb2fa2 : SUCCESS in 35m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f83f290f7ddd421fb75a4c5df684898f : SUCCESS in 1h 15m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f21b79f5b0c5456c8d74fd9d75f8d21e : FAILURE in 1h 53m 04s","accounts_in_message":[],"_revision_number":5},{"id":"88e1b4718c4b9f702b6d842e42e50ee236e4417a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-26 07:28:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"33bf48d941ab84cbdb736eee170ffdae1961361b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 08:40:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d23572690c78442b827a635092371d5a : TIMED_OUT in 1h 01m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ceda418f9c24af4a5f46e1fd0736e92 : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f2c79aff767d13b17531bd33cbe971ded7668a57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:20:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ea463ba5047e48799dd6ec5712614f8e : SUCCESS in 1h 02m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bd61838ffbc24de9948047d605a2d5ac : SUCCESS in 1h 49m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1f45ff425ffd490bbf2cf7f6f0dfbe5f : SUCCESS in 1h 46m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b459692da3ec4e8d9d9667446743b0c5 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f2b1df446d2f49b6ac0bdad43c5822bc : SUCCESS in 1h 48m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/364bef4c09604f9f82d8a239a8237ad4 : SUCCESS in 49m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a69b71e85d5d43fab92139a8ef167264 : SUCCESS in 45m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a3d3c898f34d7f9b775b6484348f72 : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea3efe3fe1aa481e94956d2697a23579 : SUCCESS in 33m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a377550c3b3b4d6084194a3b96a0ab72 : SUCCESS in 18m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04c7377db1c455b98644c54873ac949 : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab6d5df1b11470f8c678117987fc55a : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1697da91ff5f4986a9cc2514c5a6a18a : SUCCESS in 12m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d6382b8d2fa4cc08f982cf265272937 : FAILURE in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7690f990f0834a05a60251d11e767fd0 : SUCCESS in 1h 10m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7ce9e8a364884b6ebdc31a0f023774ec : SUCCESS in 1h 09m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/709b96828fd6472983a98ef9c74b68ca : SUCCESS in 1h 05m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/33a605518bf74b769b783e3d1eaaeb5d : SUCCESS in 54m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/49245a33d8bb4bc598a80ff9a8d11631 : SUCCESS in 1h 02m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ead7cca30e449909e72b50ac470b6a1 : SUCCESS in 1h 10m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbe3824923154c1c9eab2f25181ea1ee : SUCCESS in 1h 16m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f0d63c671054e3fb5c869996a348a1a : SUCCESS in 47m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7b391aa60ef949cda0df15dc65ff9cf3 : SUCCESS in 1h 03m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c95636ee488e40ed904e399245655d61 : SUCCESS in 29m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efe0d03e4b094c6ea9385a2c881eb62f : SUCCESS in 1h 19m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dd768a35ee0c4fdca9a1c882b603b3d7 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":5},{"id":"448565ae59e29710d60a79b9a91efe83aa2dca86","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-07 11:52:20.000000000","message":"Patch Set 5: Code-Review-1\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5: Code-Review-1\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e This is API behaviour change so IMO we should have API extension to let people discover if such association is or isn\u0027t really possible in the cloud.\n\u003e Did you mean a shim extension for this? I\u0027m not sure if it is really needed, we changed the DB uniq constraint once [1], and we did not do that work. And, IMO, it\u0027s not so much necessary, just a bit waste of engineer time. For the users, they have many ways to get this change such as git log, release note, LP bug and gerrit.\n\u003e \u003e Also I think that this change deserves release note.\n\u003e This patch has a release note now.\n\u003e \n\u003e [1] https://review.opendev.org/c/openstack/neutron/+/613549\n\nI add another viewpoint: for this we need new tempest test (https://opendev.org/openstack/neutron-tempest-plugin/src/branch/master/neutron_tempest_plugin/api/test_port_forwardings.py ) or change negative tests and for that \n* we need a new extension if it is really an API change and not just fixing the buggy original API behaviour. We need that for tempest as well to skip the new tests on older branches.\n* if we say this is more a bug fix of the original API behaviour, we need to add this as \"network_available_features\" to the necessary zuul file to skip the new tests on older brnaches (https://opendev.org/openstack/neutron-tempest-plugin/src/branch/master/zuul.d/master_jobs.yaml#L85 )","accounts_in_message":[],"_revision_number":5},{"id":"f4169bb3f06fa94a82984dde9a9c612d52df7710","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-10 02:47:05.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 5: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e (1 comment)\n\u003e \u003e \u003e \n\u003e \u003e \u003e This is API behaviour change so IMO we should have API extension to let people discover if such association is or isn\u0027t really possible in the cloud.\n\u003e \u003e Did you mean a shim extension for this? I\u0027m not sure if it is really needed, we changed the DB uniq constraint once [1], and we did not do that work. And, IMO, it\u0027s not so much necessary, just a bit waste of engineer time. For the users, they have many ways to get this change such as git log, release note, LP bug and gerrit.\n\u003e \u003e \u003e Also I think that this change deserves release note.\n\u003e \u003e This patch has a release note now.\n\u003e \u003e \n\u003e \u003e [1] https://review.opendev.org/c/openstack/neutron/+/613549\n\u003e \n\u003e I add another viewpoint: for this we need new tempest test (https://opendev.org/openstack/neutron-tempest-plugin/src/branch/master/neutron_tempest_plugin/api/test_port_forwardings.py ) or change negative tests and for that \n\u003e * we need a new extension if it is really an API change and not just fixing the buggy original API behaviour. We need that for tempest as well to skip the new tests on older branches.\n\nWe do not change any API code, but this DB upgrade indeed expand the API abilities. Some actions are not allowed, but it can be done after this change. The bad thing is, this is a DB change only. We have no way to add a config for this to decide whether enable it or not. Or is there any way for us to disable the neutron-tempest-plugin cases for stable branches only?\n\n\u003e * if we say this is more a bug fix of the original API behaviour, we need to add this as \"network_available_features\" to the necessary zuul file to skip the new tests on older brnaches (https://opendev.org/openstack/neutron-tempest-plugin/src/branch/master/zuul.d/master_jobs.yaml#L85 )\n\nI\u0027m not familiar with the zuul config, so I have some questions. The Neutron stable branches will not have such DB change, but the neutron-tempest-plugin will be use with master branch? If so, remove the negative case seems no harm to stable branches.\n\n@Lajos Katona, thank you for the comments. Please see my reply inline.","accounts_in_message":[],"_revision_number":5},{"id":"3a60ce5f7848b5555b9680b279c49f59cc250857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-24 08:28:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3798858da0373198e3d5e48c5a7d6b6eb690d1c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 09:18:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5079ec9a19c7412e9726760b1bdc2fd5 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b083a0a0f9ec4af8b67b203eca837027 : SUCCESS in 40m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7d45d7e71b7489c356a68ba8607e5448996c9d17","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 10:36:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787304/6 : SUCCESS in 2h 06m 16s","accounts_in_message":[],"_revision_number":6},{"id":"bf0784da3301b3ead911e96346038005d5eb8f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 11:07:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a9a8665666144f6db0ab2bf6cf651956 : SUCCESS in 36m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dd195b09c273464894490e8e6f520ff4 : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7cf993903f4c4625b7f96cf482e4a51a : SUCCESS in 1h 30m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53ca92d82d5b4eb893849e9a6e39920f : SUCCESS in 1h 44m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/83e703b6808a45a99d4ede3a88006f1c : SUCCESS in 2h 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bb1cc7a7e654f5c94ae0128183c9cc4 : SUCCESS in 32m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01188eb9ac084212ba03daf360b88780 : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d1e62f3574439db99672f49492dfa5 : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16b503661e3f48e2a2cc0dbeedc3bb63 : SUCCESS in 25m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aaa640b008647ab95cd6ed9d0bb2f76 : SUCCESS in 20m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6249cef7bcf84ffc9c94b86823153f92 : SUCCESS in 31m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75caa4f3637544f784f5d29afa1de5c7 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7313ab3471346ceb974a725ffab1fb6 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06c184d970b7473bab89c7fa742c7663 : SUCCESS in 1h 01m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1792c6b7038c46b883dec5e24bd3dab1 : SUCCESS in 1h 25m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/45cd2153573c409bb606ba91a0c3d79a : SUCCESS in 1h 02m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d247154b509f4a2eb864ed23d11716c9 : SUCCESS in 54m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/547828e2a5734b848932e5ce7b7872c6 : FAILURE in 52m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/407ae4e52e184aa4a05d56998d213c34 : SUCCESS in 1h 13m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b70a165e1be24cdcb85140b959760c6f : SUCCESS in 1h 10m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f5b65a2dacb4e94a4870ddf88c70381 : SUCCESS in 1h 31m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fc416e1d49347068bacd648c5ec8051 : SUCCESS in 57m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7d01e128545043228b2d296fd36d8de5 : SUCCESS in 1h 02m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d2fe37881b3248d2863e3afcdd619eae : SUCCESS in 31m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f58ad3f798d411a91d9865753c8c412 : SUCCESS in 1h 15m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6caaf1e143fd42d3b81138616591ea79 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":6},{"id":"df31617d431846f3e1ff58ae2f6d5a9482ca93c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-28 17:35:35.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/04/787304/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"06b4338873d9c8f1d3c756724e1f16ccd133d002","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-08 08:52:33.000000000","message":"Abandoned\n\nRestore if someday we want this.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d61950d58274137fe57095b4de97a8e4e72f3ac","revisions":{"954f8c0c19f50fb3d266ca2d2dfdb7351e5d0ffb":{"kind":"REWORK","_number":1,"created":"2021-04-21 00:46:37.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/1"}}},"commit":{"parents":[{"commit":"77b7d8b452fe084b28d3092c35c639887ea2a4c1","subject":"Merge \"Initialize privsep library for neutron-ovs-cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b7d8b452fe084b28d3092c35c639887ea2a4c1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 10:34:01.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                         \u0027external_port0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n    )\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                         \u0027socket0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n    )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch remove this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/954f8c0c19f50fb3d266ca2d2dfdb7351e5d0ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/954f8c0c19f50fb3d266ca2d2dfdb7351e5d0ffb"}]},"branch":"refs/heads/master"},"09b184011e46d7ebd114f99c5cc3540d603b6aa8":{"kind":"REWORK","_number":2,"created":"2021-04-21 05:40:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/2"}}},"commit":{"parents":[{"commit":"77b7d8b452fe084b28d3092c35c639887ea2a4c1","subject":"Merge \"Initialize privsep library for neutron-ovs-cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b7d8b452fe084b28d3092c35c639887ea2a4c1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-21 05:40:10.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                         \u0027external_port0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n    )\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                         \u0027socket0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n    )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch remove this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09b184011e46d7ebd114f99c5cc3540d603b6aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09b184011e46d7ebd114f99c5cc3540d603b6aa8"}]},"branch":"refs/heads/master"},"99d9dbf48eae6c0fa3041e829989a3956c0834d0":{"kind":"REWORK","_number":3,"created":"2021-04-21 08:54:01.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/3"}}},"commit":{"parents":[{"commit":"77b7d8b452fe084b28d3092c35c639887ea2a4c1","subject":"Merge \"Initialize privsep library for neutron-ovs-cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b7d8b452fe084b28d3092c35c639887ea2a4c1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-21 08:53:19.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                         \u0027external_port0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n    )\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                         \u0027socket0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n    )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch remove this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99d9dbf48eae6c0fa3041e829989a3956c0834d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99d9dbf48eae6c0fa3041e829989a3956c0834d0"}]},"branch":"refs/heads/master"},"e540683916bf7c5d53bb861c939be55fc5b9de1a":{"kind":"REWORK","_number":4,"created":"2021-04-22 03:39:57.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/4"}}},"commit":{"parents":[{"commit":"77b7d8b452fe084b28d3092c35c639887ea2a4c1","subject":"Merge \"Initialize privsep library for neutron-ovs-cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b7d8b452fe084b28d3092c35c639887ea2a4c1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-22 03:39:11.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                         \u0027external_port0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n    )\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                         \u0027socket0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n    )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch remove this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e540683916bf7c5d53bb861c939be55fc5b9de1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e540683916bf7c5d53bb861c939be55fc5b9de1a"}]},"branch":"refs/heads/master"},"abfe7f9e3e379b07d0dc2621658b6df26df71627":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-04-22 07:11:45.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/5"}}},"commit":{"parents":[{"commit":"77b7d8b452fe084b28d3092c35c639887ea2a4c1","subject":"Merge \"Initialize privsep library for neutron-ovs-cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b7d8b452fe084b28d3092c35c639887ea2a4c1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-22 07:09:26.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                         \u0027external_port0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n    )\n    op.create_unique_constraint(\n        constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                         \u0027socket0protocol\u0027),\n        table_name\u003dTABLE_NAME,\n        columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n    )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch remove this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nDepends-On: I95d9fd3224300c4406a1a397e514989d7495b370\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abfe7f9e3e379b07d0dc2621658b6df26df71627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abfe7f9e3e379b07d0dc2621658b6df26df71627"}]},"branch":"refs/heads/master"},"7d61950d58274137fe57095b4de97a8e4e72f3ac":{"kind":"REWORK","_number":6,"created":"2021-05-24 08:28:57.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/04/787304/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/787304/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/787304/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/787304/6"}}},"commit":{"parents":[{"commit":"b24d2fd7690a01d56e4b09588e52ff7f9481e8c5","subject":"Merge \"Deprecate and remove \"get_agents_db\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b24d2fd7690a01d56e4b09588e52ff7f9481e8c5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-04-20 07:22:13.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-24 08:28:19.000000000","tz":480},"subject":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port","message":"[L3][Port forwarding] multiple floating_ip:port to same fixed_ip:port\n\nFloating ip port forwradings table has constraints:\n\n  op.create_unique_constraint(\n      constraint_name\u003d(\u0027uniq_port_forwardings0floatingip_id0\u0027\n                       \u0027external_port0protocol\u0027),\n      table_name\u003dTABLE_NAME,\n      columns\u003d[\u0027floatingip_id\u0027, \u0027external_port\u0027, \u0027protocol\u0027]\n  )\n  op.create_unique_constraint(\n      constraint_name\u003d(\u0027uniq_port_forwardings0internal_neutron_port_id0\u0027\n                       \u0027socket0protocol\u0027),\n      table_name\u003dTABLE_NAME,\n      columns\u003d[\u0027internal_neutron_port_id\u0027, \u0027socket\u0027, \u0027protocol\u0027]\n  )\n\nIt does not support N-to-1 rule (failed on constraint\nuniq_port_forwardings0internal_neutron_port_id0socket0protocol):\n\n172.24.4.64:22 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.64:122 -\u003e tcp -\u003e 192.168.111.45:22\n172.24.4.168:22 -\u003e tcp -\u003e 192.168.111.45:22\n\nThis patch removes this uniq constraint to support multiple\nfloating_ip:port to same fixed_ip:port.\n\nDepends-On: I95d9fd3224300c4406a1a397e514989d7495b370\n\nChange-Id: Ibf7c08e147ace055114556eb14ca013fcf71fe6e\nCloses-Bug: #1922653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d61950d58274137fe57095b4de97a8e4e72f3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d61950d58274137fe57095b4de97a8e4e72f3ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
