)]}'
{"id":"openstack%2Fneutron~787337","triplet_id":"openstack%2Fneutron~master~I708904b982d243359f2eeda809beae0321f1a7db","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I708904b982d243359f2eeda809beae0321f1a7db","subject":"[OVN] Only account for bound ports in metadata agent","status":"MERGED","created":"2021-04-21 09:46:36.000000000","updated":"2021-04-26 12:14:59.000000000","submitted":"2021-04-26 12:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"787337","meta_rev_id":"78de8d0bed92caf6f1636e0644392d909b219d9f","_number":787337,"virtual_id_number":787337,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-04-22 14:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 12:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2021-04-22 15:54:16.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-26 08:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-04-23 16:12:44.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2021-04-22 15:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-26 08:15:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 09:56:43.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-04-21 10:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 10:56:03.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-04-21 11:46:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-04-21 13:05:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-21 13:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 16:24:00.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 20:48:15.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 14:36:52.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-04-26 08:15:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"9d3dfad52dd6742aaaa4d19ddf3647a51b09d281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-21 09:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"833999c72869b41f57fcb995a6d4fd715137064d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-21 09:56:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfd49a6826a94cbd8f9954da0fb3aca630543639","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-21 10:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nthanks Kuba!","accounts_in_message":[],"_revision_number":1},{"id":"61b5cc69946eff783173c1f8c3fa4cc7b1d8ddb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:31:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72502107c25a4df0a6fceebc3305afdd : FAILURE in 41m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ee23a52bac640fdbd0c25174885b63d : FAILURE in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4e366185b5d49291c020dd00c6d29f864ac3dbc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-21 10:56:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b39213a567f862b427f579dd7b6c21534e81e0b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 11:46:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787337/1 : SUCCESS in 1h 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"05763f35550c28f1b31a1728e24c38f68f6d51d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-21 12:29:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69a89d723a02b8b85df5a1ba6ac57df32e06a408","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-21 12:31:37.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ddfb8ebcf93178bcd40a6ae971c6e3e88b17f91b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:42:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06992ff3ddc54d3ea2e71b412ffd3291 : FAILURE in 8m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50542e90ea6141b882aee1c0dde004e6 : FAILURE in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6005f2f3f800dfac221c322b1f7cfa7c23318928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 13:27:25.000000000","message":"Patch Set 2: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e805ce663424a7cb0c6dc15b007d8f5 : SUCCESS in 49m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6abe0f9320c048329c1e1ecaee1a4eb1 : FAILURE in 26m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d96bb2fbc7e14a239251e3e96e5caa3e : FAILURE in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0e934cb7dc44f0cb96bdc390ebe70dd : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5aa7e6ddf1468cbd8111913e413cd9 : FAILURE in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aeb832267da483a8f6888018e797723 : FAILURE in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a1fb9001469457a8ca2fa0190b8b31d : FAILURE in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a5bf23a18714d0e8a122e544469acc9 : FAILURE in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ddfb667c884f469f694d7b1074a1bc : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28e3b3d276f844dab1dd9109a467ba30 : FAILURE in 15m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/44a8d2ec7693438c859a26ff906be4e0 : FAILURE in 24m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cfd2ff3523324227a573dd72a603ad2f : FAILURE in 21m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39c0e8c064814b08adf2dd04269dd6cf : FAILURE in 31m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/35518629c1414a9f8758451a8596c17d : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b290fe45d15e84f7ad93f54ae86ad9a78c293b2e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 14:10:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787337/2 : SUCCESS in 1h 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d6384d501e060200d45a9c575cf922d51c75887f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-21 16:24:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/37/787337/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/37/787337/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b425c16afdc41f2e19a577977d728d35cb746102","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-21 20:48:15.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/37/787337/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9c7627b8f8a2d1cf77b16b28f1636eedbeebd084","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 00:20:17.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/37/787337/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"05a9661a2ad9ca7eb34a144ef16b810c13cabd00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 08:45:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4eb8d137b198aa296f450873df146b3af2d4ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 08:45:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cc744c7e7c19bb2f591f76d4f9508cab0bb71256","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-22 08:58:06.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"77f1c1f188963c360c282759acb4b78d05b8502a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-22 09:04:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nQuestion in-line","accounts_in_message":[],"_revision_number":4},{"id":"3d31a109f232444723d3b057e733d046f003c5ad","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-22 09:05:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2daca075db99af6a03ac7b7fe88bce149ae3f3f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 09:25:57.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"a30fee436a3edb99f5226df2d2805354280749b2","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 09:28:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cdc863c3fac6c145faebc177de93f869b79ac608","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-22 09:38:21.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a42df19c2f4dc31289a2c67d4100d6c74b4abb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:49:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f62e1a1bfb8a4885a3416f510a679e3e : TIMED_OUT in 1h 01m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac33a4632c014184a93439ff3ec14785 : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3278724493f7dbe732a577ba3bf4c199cb8ee8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:28:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ad70ee92e5e4b65bd4741fd8e9db4bb : SUCCESS in 42m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f4b20f092a3c45bd81c75e3c38b0344b : SUCCESS in 1h 44m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d0aa835cb744459b56bec88bd21c703 : SUCCESS in 50m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb72f7bec45f4df590f37b307590d832 : SUCCESS in 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60090bbb8e124028a6ba810d347997dd : SUCCESS in 15m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5056bddd0b2c4ba3b7fea9f841bea322 : SUCCESS in 29m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d059186244443e48a22336272dc53f7 : SUCCESS in 28m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2423a4ca5a784875a117f8ded3fba8d5 : SUCCESS in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d86fc7014204f43a2c3d33158999d2b : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6da6a5940fb94b26843afd6f8ef55142 : SUCCESS in 53m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c207b35a420849798c142f60a9d2bda9 : SUCCESS in 1h 05m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f8b69ba979a94131a9d2c44438e04457 : SUCCESS in 34m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe23ff3d932a497192f81d2a0ba4f2d2 : SUCCESS in 1h 24m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d0bf794f3eae4998809ca4b051fdc5f0 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":4},{"id":"095a3288fcd9fd9248b927eb0f94385c7d672c8f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-22 12:23:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/787337/4 : SUCCESS in 1h 50m 01s","accounts_in_message":[],"_revision_number":4},{"id":"caea65b653b621ac85bcab70c82904becf83c614","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 12:24:57.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/37/787337/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"667736068571365f6fe5a3b6216af9efc082e1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 13:24:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"89f8b9e841d4c188a785b96c4a9aeabd2e41b457","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 13:39:11.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/37/787337/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"196cef0ae87244b75315f5c2d0267c154f529fdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 14:28:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/675ef3eade444c29a96a04dfce55de04 : TIMED_OUT in 1h 01m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcb2c597dcd64e699632ef9108140ff7 : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7dab52f797a7ab8be49e2fa1696ca4944e86a710","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 14:36:38.000000000","message":"Topic dalvarez removed","accounts_in_message":[],"_revision_number":5},{"id":"7739f8e80b7ca22422807a2c6f1928ad280c3e68","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-22 14:46:35.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/37/787337/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2c1b1bd784f35104df6db90ef40d7ae3fa96803a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-22 15:11:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b68d47f517eb90b53d00f3892d4f073674bac79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 15:38:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/834b28ce43684bffb8df11746a6e0073 : SUCCESS in 45m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4392acef096241e2bfaf6b6469234cc0 : SUCCESS in 2h 12m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66e10a2b39b348ff977e71f177bdc46d : SUCCESS in 42m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/857bd06b281642709616a0745e1c0bfa : SUCCESS in 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09e134fc84741df9f3e962edae05946 : FAILURE in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be5935f28fda47198d2b42b65b9374e7 : SUCCESS in 37m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1cd13ee1a234dc8b719c7973a573cc2 : SUCCESS in 30m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84ee1f6b4ed0460a81ff00a8d8c2a111 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2356debdb956489a94d65f6d4d796e54 : SUCCESS in 14m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b844763111b49c183c9cb80fc7ca272 : SUCCESS in 1h 12m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1dba88b562394bf6b5b3d3e180aa7a1e : SUCCESS in 1h 14m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0f711d983c784b34b677f42d70dcfa32 : SUCCESS in 35m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9441d882a574ad0be4590a213763dfb : SUCCESS in 1h 49m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/93f0fe6e438a4f0dbedd7ad9e9c1d442 : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":5},{"id":"0f09c8e3467ad4a2b1aceabb655783d159ae6a7b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-22 15:54:16.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck\n\nPEP8 unrelated failure:\n\n2021-04-22 13:51:40.921112 | ubuntu-focal | WARNING: Retrying (Retry(total\u003d4, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)) after connection broken by \u0027ReadTimeoutError(\"HTTPSConnectionPool(host\u003d\u0027mirror-int.dfw.rax.opendev.org\u0027, port\u003d443): Read timed out. (read timeout\u003d60.0)\")\u0027: /pypi/simple/hacking/","accounts_in_message":[],"_revision_number":5},{"id":"b1500ca9e1d7f09bea77eeed974fba1bfccf4c0e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-22 16:06:35.000000000","message":"Patch Set 5: Code-Review-2\n\n(1 comment)\n\n-2ing just to get an answer before it merges, sorry.\n\nWithout modifying MetadataAgentOvnSbIdl, when adding the Chassis table won\u0027t this watch all Chassis tables if Chassis_Private is supported? Under if chassis, it will only do the conditional monitor on Chassis_Private, right? Don\u0027t we want to limit to only watching our Chassis table as well?","accounts_in_message":[],"_revision_number":5},{"id":"a9f6a91d6a30ee58a3095d7aae292949534d45f1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-22 16:08:33.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/787337/5 : FAILURE in 1h 58m 48s","accounts_in_message":[],"_revision_number":5},{"id":"3414bd4e663a68399f6bc0a9fc3c3c44f3dbef19","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-22 16:51:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8b80df31dbdd9915fa5db7be7416a78efbd4e4e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 16:58:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f01932906734651ae1393cfabb91af7 : TIMED_OUT in 1h 01m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cca456a6b345400da829796a5474f5c9 : TIMED_OUT in 1h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a4862532fa140fbbbdeeace2a04a8304239084b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-22 17:22:41.000000000","message":"Patch Set 5: -Code-Review\n\nUnblocking since this specific patch shouldn\u0027t cause additional issues.","accounts_in_message":[],"_revision_number":5},{"id":"3cbc491d54bb8fd03c7e879d35f3b0de0471e220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 18:18:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e2bffe558994a31a3a8326685cfcaac : SUCCESS in 48m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2b5f4cbe9863438f8d4a826f7e0985f1 : SUCCESS in 2h 13m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/365eac4fb4db4854a1f14dacb8210b78 : SUCCESS in 46m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a249ab8f4b24451391ab5d909e8594c9 : SUCCESS in 42m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b134c5ce0047258e8cdfca8d8185fa : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e783132731b2418da2831bc59dee5fed : SUCCESS in 30m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d3c7f9f5064882b1bef71b2d232064 : SUCCESS in 23m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af525904bd804508b344a9095a353af1 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9b26aef32f467eb3d5c48be5e49d3c : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01a42dc3d33c467c9a368dc976c8700e : SUCCESS in 45m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/259291afcaea4decaa529f88a738d45f : SUCCESS in 55m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/95d5ade3b5744b07ba728c135d4e1430 : SUCCESS in 38m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14738b7feb864584bf90867871560742 : SUCCESS in 1h 01m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d22484fc224b42d88a893030f2ad6d24 : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":5},{"id":"65381d65f457a33069daf6f65627964924f54b90","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-23 08:15:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b885764f4434dc8165f7a3ddd9ccc6028a888b2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-23 16:12:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55a75f83537c7524d1ef978bef2f6e4439fcfbd0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-23 16:23:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"28b0a9dcbf6b85236439b9ac30ce55d996e593e2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-23 16:37:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"62cd82b45ef6432edc7bdf3f0c0d26d937678b40","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-26 08:15:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nIt seems to be good to go now as we have also follow up https://review.opendev.org/c/openstack/neutron/+/787780","accounts_in_message":[],"_revision_number":5},{"id":"613ee7ed86a7ea95d3c28b700f2da7b667336c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 08:16:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"486c604d4d313f5cfe4d399d7b3b43cb60735929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:13:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c1f518d328b24c0ab639dd4c5a3246f5 : SUCCESS in 37m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd1e971845b6432dbe52be5129769863 : SUCCESS in 1h 51m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e26dc43760bc4773906d95b3b2439eaf : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58600258e338451ca0ca67809c3cef3a : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075d782f5bea4830af60e378a712ae4f : SUCCESS in 23m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/011a021709124463842c92f48e2615fd : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f5abffb7894e81bd87911d2db700fe : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a06aedf0df54cedabafc09bf571d0ae : SUCCESS in 40m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ad443fc45be4618ae928304b78549a5 : SUCCESS in 1h 10m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8052a59c55534496ba3e91dacb008bd0 : SUCCESS in 31m 55s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3d37c108f2a14e4a9322a6d403308825 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":5},{"id":"dd0a82c9705a4f960e4ab847dfe9523d243b4afd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:13:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"78de8d0bed92caf6f1636e0644392d909b219d9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:14:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573eaea66b764933affd7583a44c3efa : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"32e938e698f260af6896f6179a5ffc0838677124","revisions":{"17b9a3a5cf62f49c8f8180287dad106257553dbb":{"kind":"REWORK","_number":1,"created":"2021-04-21 09:46:36.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/37/787337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/787337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/787337/1"}}},"commit":{"parents":[{"commit":"300f6c1c4d2721e4c465d7dfc312184381eeaa5b","subject":"Merge \"Initialize privsep library in Neutron commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/300f6c1c4d2721e4c465d7dfc312184381eeaa5b"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"subject":"[ovn] Only account for bound ports in metadata agent","message":"[ovn] Only account for bound ports in metadata agent\n\nDue to bug #1922934 there might be situations with stale ports\nin the OVN database. When instances request metadata, the agent\nwill try to fetch all ports for that particular network with the\nsame IP address and there should be only one.\n\nHowever, when there are stale ports in OVN database but not in\nNeutron, those ports may have the same IP address and the result\nis that metadata won\u0027t work. As the stale ports will never be\nbound to any hypervisor, this patch is accounting only for\nbound ports so that those ports don\u0027t interfere until they\neventually get fixed by the maintenance task.\n\nRelated-Bug: #1922934\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I708904b982d243359f2eeda809beae0321f1a7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17b9a3a5cf62f49c8f8180287dad106257553dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17b9a3a5cf62f49c8f8180287dad106257553dbb"}]},"branch":"refs/heads/master"},"330706194aa8aac1ba0bb38eadd87761dc9bc4c3":{"kind":"REWORK","_number":2,"created":"2021-04-21 12:29:31.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/37/787337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/787337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/787337/2"}}},"commit":{"parents":[{"commit":"300f6c1c4d2721e4c465d7dfc312184381eeaa5b","subject":"Merge \"Initialize privsep library in Neutron commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/300f6c1c4d2721e4c465d7dfc312184381eeaa5b"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 12:28:48.000000000","tz":120},"subject":"[ovn] Only account for bound ports in metadata agent","message":"[ovn] Only account for bound ports in metadata agent\n\nDue to bug #1922934 there might be situations with stale ports\nin the OVN database. When instances request metadata, the agent\nwill try to fetch all ports for that particular network with the\nsame IP address and there should be only one.\n\nHowever, when there are stale ports in OVN database but not in\nNeutron, those ports may have the same IP address and the result\nis that metadata won\u0027t work. As the stale ports will never be\nbound to any hypervisor, this patch is accounting only for\nbound ports so that those ports don\u0027t interfere until they\neventually get fixed by the maintenance task.\n\nRelated-Bug: #1922934\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I708904b982d243359f2eeda809beae0321f1a7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/330706194aa8aac1ba0bb38eadd87761dc9bc4c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/330706194aa8aac1ba0bb38eadd87761dc9bc4c3"}]},"branch":"refs/heads/master"},"a6d97f91924b96d9a397bb86152a67f9bdeeafc1":{"kind":"REWORK","_number":3,"created":"2021-04-22 08:45:03.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/37/787337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/787337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/787337/3"}}},"commit":{"parents":[{"commit":"d5b5f1ac9c19bc93037d35bf75d8805a02d952bc","subject":"Merge \"Configure branch on which tripleo jobs should be run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5b5f1ac9c19bc93037d35bf75d8805a02d952bc"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-22 08:42:04.000000000","tz":120},"subject":"Only account for bound ports in metadata agent","message":"Only account for bound ports in metadata agent\n\nDue to bug #1922934 there might be situations with stale ports\nin the OVN database. When instances request metadata, the agent\nwill try to fetch all ports for that particular network with the\nsame IP address and there should be only one.\n\nHowever, when there are stale ports in OVN database but not in\nNeutron, those ports may have the same IP address and the result\nis that metadata won\u0027t work. As the stale ports will never be\nbound to any hypervisor, this patch is accounting only for\nbound ports so that those ports don\u0027t interfere until they\neventually get fixed by the maintenance task.\n\nRelated-Bug: #1922934\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I708904b982d243359f2eeda809beae0321f1a7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6d97f91924b96d9a397bb86152a67f9bdeeafc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6d97f91924b96d9a397bb86152a67f9bdeeafc1"}]},"branch":"refs/heads/master"},"9d066b0303692d8be9ac055a2d2af4d792de2a5e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-22 08:45:56.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/37/787337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/787337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/787337/4"}}},"commit":{"parents":[{"commit":"d5b5f1ac9c19bc93037d35bf75d8805a02d952bc","subject":"Merge \"Configure branch on which tripleo jobs should be run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5b5f1ac9c19bc93037d35bf75d8805a02d952bc"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2021-04-22 08:45:42.000000000","tz":0},"subject":"[OVN] Only account for bound ports in metadata agent","message":"[OVN] Only account for bound ports in metadata agent\n\nDue to bug #1922934 there might be situations with stale ports\nin the OVN database. When instances request metadata, the agent\nwill try to fetch all ports for that particular network with the\nsame IP address and there should be only one.\n\nHowever, when there are stale ports in OVN database but not in\nNeutron, those ports may have the same IP address and the result\nis that metadata won\u0027t work. As the stale ports will never be\nbound to any hypervisor, this patch is accounting only for\nbound ports so that those ports don\u0027t interfere until they\neventually get fixed by the maintenance task.\n\nRelated-Bug: #1922934\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I708904b982d243359f2eeda809beae0321f1a7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d066b0303692d8be9ac055a2d2af4d792de2a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d066b0303692d8be9ac055a2d2af4d792de2a5e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"32e938e698f260af6896f6179a5ffc0838677124":{"kind":"REWORK","_number":5,"created":"2021-04-22 13:24:04.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/37/787337/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/787337/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/787337/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/787337/5"}}},"commit":{"parents":[{"commit":"d5b5f1ac9c19bc93037d35bf75d8805a02d952bc","subject":"Merge \"Configure branch on which tripleo jobs should be run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5b5f1ac9c19bc93037d35bf75d8805a02d952bc"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-21 09:41:57.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-04-22 13:23:43.000000000","tz":120},"subject":"[OVN] Only account for bound ports in metadata agent","message":"[OVN] Only account for bound ports in metadata agent\n\nDue to bug #1922934 there might be situations with stale ports\nin the OVN database. When instances request metadata, the agent\nwill try to fetch all ports for that particular network with the\nsame IP address and there should be only one.\n\nHowever, when there are stale ports in OVN database but not in\nNeutron, those ports may have the same IP address and the result\nis that metadata won\u0027t work. As the stale ports will never be\nbound to any hypervisor, this patch is accounting only for\nbound ports so that those ports don\u0027t interfere until they\neventually get fixed by the maintenance task.\n\nRelated-Bug: #1922934\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I708904b982d243359f2eeda809beae0321f1a7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32e938e698f260af6896f6179a5ffc0838677124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32e938e698f260af6896f6179a5ffc0838677124"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
