)]}'
{"id":"openstack%2Fneutron~789664","triplet_id":"openstack%2Fneutron~stable%2Frocky~I59b0f99463ab00743f19016a890561916df900ed","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1891309-stable/wallaby-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"I59b0f99463ab00743f19016a890561916df900ed","subject":"designate: allow PTR zone creation to fail","status":"MERGED","created":"2021-05-05 06:44:32.000000000","updated":"2021-06-07 16:24:20.000000000","submitted":"2021-06-07 16:24:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789664-bug/1891309-stable/wallaby-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky","cherry_pick_of_change":789663,"cherry_pick_of_patch_set":1,"meta_rev_id":"515582d0dc0d6041038e31d93716cefbd7f25f0d","_number":789664,"virtual_id_number":789664,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 16:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-07 11:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-02 08:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-07 11:00:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 06:44:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-05-05 07:22:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-05 08:04:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 08:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 14:38:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"549973e718566b6000aba247c1ad7d5af4e73c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-05 06:44:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"af3a8f1dd3ed2da6bbeae3a4f60a178404fc114a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 07:22:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e38b725af75a1b788ddef4913878895f0efac5dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 08:04:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/64/789664/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/64/789664/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fa016040c9dc0272e5432096882bc5a0c3e84501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 08:27:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/82cba84366cf4f8db6b3e093dcba8e63 : POST_FAILURE in 43m 07s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/17a985ebafe24d8b8691ba26d992a158 : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/eae0070171db478db82dd14453c93053 : SUCCESS in 1h 06m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/41412c0be93d4066994a23f1009d0ad9 : POST_FAILURE in 1h 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdd8b99617e4846ac5b8ed0ab2e31e7 : FAILURE in 12m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a09c72cab7f4e259ef7b87340b10ffa : SUCCESS in 25m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08fc1d611ed0411ca1359052ca4c4c88 : SUCCESS in 29m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec87ba76b839420ca173bf595a57c6f6 : SUCCESS in 8m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62104ac6a52545c5930c952fe44808ae : SUCCESS in 1h 20m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/512e47ba1a1a450db27bcab35bae6357 : FAILURE in 12m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e8ed79d72914d1199255621706c2303 : SUCCESS in 1h 01m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/efe28feae8b44220b1de848d8958707f : SUCCESS in 35m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ec9f49251222473d9082984820bdd9e5 : SUCCESS in 1h 22m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0a733d0c14784de3a4f819741b6f243b : SUCCESS in 1h 24m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb13b37329af448e9a56c695871ef6e5 : SUCCESS in 1h 23m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2fe41da52e442668ebf6df264b3a3fa : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb491bbe807e89a90b01b5cfc11543f2d614db9a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-06 14:38:04.000000000","message":"Patch Set 1: Code-Review-1\n\nDifferent times, different pep8 😊 (same as queens backport)","accounts_in_message":[],"_revision_number":1},{"id":"b6fef01b02a7c28a19a23eed3c6f9339dd4c203b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 08:00:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"892fe018e2afc8520aa1b6c3953e81fc157c5a4e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-14 08:05:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"23ec39aff22dba1a32dbe04d74819ea8cc6a0b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:55:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/146fb87302004fcd804976763584a6d0 : SUCCESS in 37m 32s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/b791f2d6bf314ed794e68243139fd900 : SUCCESS in 1h 05m 53s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/20128545059b43c8ad736d57bbc0b1f5 : SUCCESS in 1h 04m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/ff297719443a490fbb2bd3b90e29b385 : POST_FAILURE in 1h 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae769e9f88c6437eac448aae209e083b : SUCCESS in 10m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3953ef5feb044b3a5380aaacc3eb64c : SUCCESS in 26m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ace54fca4d114a77b4c80223923588da : SUCCESS in 27m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fe357a92f7b94449a0d277166a737ab4 : SUCCESS in 7m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cde7c397929e440baf89cdce70e0d5c4 : SUCCESS in 1h 18m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad88ab814b384024b04265bfebdfaba1 : FAILURE in 10m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1dff27ed8694dabbf98b33a595c4acd : SUCCESS in 1h 12m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/955085e77b5e4ca9b53315eaa4954ad1 : SUCCESS in 32m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f470405489b47fe9a85ad5a292a7bd5 : SUCCESS in 1h 14m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/52e615a9d79f4080b49d0164c752a914 : SUCCESS in 1h 21m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aafb4beb6e0541cda0614def707d229f : SUCCESS in 1h 18m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c622cc99f3b8474aa68807119d2f6377 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":2},{"id":"dbd885e9bace2f86052ae553f67d3ec62d0144db","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-17 20:20:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0092495e9c6095ebd8ba045a999d74db62c66e3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-17 20:24:51.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8d2c25dafffedb9da58e3e5b7268a9205548a313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 00:16:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/38bab9657455456ca0f9e04a1b1ca108 : SUCCESS in 49m 03s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/19b0822eada74468b1ec31b7f2b83bbf : SUCCESS in 1h 15m 51s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a11d4014f16342bc9be8de6623a07b0f : FAILURE in 1h 13m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/b76764025c114169b2c69f7fb8b0e82e : POST_FAILURE in 2h 04m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0579b762eef9422e80c2bd61d37b05b6 : SUCCESS in 12m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbdadc88d9bc4f6b89d0a6f561c7ccaf : SUCCESS in 29m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3dd186829fd49b29b75472b10593bed : SUCCESS in 26m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/522fb3a34c2d4deda309ac50d9612428 : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca69fb3d42084c718192d0c67de3e6bd : SUCCESS in 1h 14m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5f46d4f6b37c44bfbcb8560d2d91ed24 : FAILURE in 15m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd606d424f584ed08e3414b54d4c0709 : SUCCESS in 1h 16m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0b4851098ea24067bd64710447f53214 : SUCCESS in 38m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/57c91065e8d242308798eb155f8c7d0a : SUCCESS in 1h 17m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8841bb6b375341bf84530f573d040190 : SUCCESS in 1h 18m 05s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/61e580154cda4f93b6cb5b7379713ee6 : SUCCESS in 1h 19m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3e58aa5c64e4000981a82b4babb4429 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4008d08b474bf2adb6010a050da59c022014558a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-20 12:18:46.000000000","message":"Patch Set 2:\n\nrecheck rocky should be fixed now","accounts_in_message":[],"_revision_number":2},{"id":"bf545375b6e5a77e978556eb4c559432c1741e49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-20 12:23:30.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"67d48348082e9a4914ea397743e044ec500fede8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:39:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e3149d55483e4f739300cda600c6d8bd : SUCCESS in 45m 54s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/15f5ef57d0f64564899ce9f0b50d1b5d : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/7ab71040188a4f3c8f4de8114d53698e : SUCCESS in 56m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/5f3e94a18dd648969b7b48c080fb3e98 : SUCCESS in 1h 00m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1ac36ea0de4ef28928f1813b1b73e4 : SUCCESS in 11m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92a920cfeff843af8f106beec297013b : SUCCESS in 16m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/718418105bd448b1a8ff632b53829729 : SUCCESS in 24m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5227ddd930854863ab2066273f3faf88 : SUCCESS in 7m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35fc92139b804aed85f7fb6b5920bf62 : SUCCESS in 1h 15m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0d09680afeb74adbacc7986d045c3553 : FAILURE in 8m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b566ab38ffc40cdaae2016c1f509699 : SUCCESS in 1h 16m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9a6e4970956a4727b2dce711c34960ff : SUCCESS in 36m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6372271590d9407e972ca3fb4cd1359c : SUCCESS in 1h 10m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a42adc5012e1401993ea3841610c2c9c : SUCCESS in 1h 19m 00s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ca9eb6c72144767abfce717a9f94c34 : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac39dfff9dc41c3935cc3f537ba156f : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"76b80ca20d89ae64894d8eb7ea747e1ca8acfca3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-01 15:00:07.000000000","message":"Patch Set 2:\n\nrecheck grenade","accounts_in_message":[],"_revision_number":2},{"id":"178444af798f128966c788a467fe4acbe022302a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-01 16:31:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/64/789664/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2603916a68ab6d91792abeaba4d125bbca16fbcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:07:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/54a1e129aeea4b5495251770586ab998 : SUCCESS in 50m 13s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/a14c8c64daef4a4383fde713e0a59d89 : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/42d7802dbd0641418f8f3ae40d00cfc3 : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/824eb2ba6ddb4cb3bcb581521a54d1d2 : SUCCESS in 1h 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7ee2f2079346de8c59eedf9022d4ac : SUCCESS in 14m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57a806f8e5ad440bb6f8ed08c6350e27 : SUCCESS in 26m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/09419e5cb1fe4903a8dbd24394f4394b : SUCCESS in 30m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f147f992c3a47ba961ae74855b290a3 : SUCCESS in 8m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2b7d4c5a2db4dcea53dd0c1b66542d3 : SUCCESS in 1h 26m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1678959618b343b5b2849c1537c2d820 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed3f6d93abb34d398a59b874801c2faf : SUCCESS in 1h 21m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/821bb5c44e054ca5af71d9a9b1bbadd4 : SUCCESS in 43m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ed52aa5c73414cf6abc6092663c6359c : SUCCESS in 1h 36m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/328ea59990ab46e08a67b6937d269e97 : SUCCESS in 1h 28m 14s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9d65fa70c454305a16b2e1df2e447d9 : SUCCESS in 1h 39m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e031fa993454bdb9552ed88d8d374db : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5324b17a8b16490043be7e6e0a2eb8a90bc689d6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-02 08:54:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35a22816bd7936d7230c6fd7ca5c8671bbd73edb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-07 11:00:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"314f03ddcaa2d32a527be934d020b62adbcb85a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:01:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8074986d9966703169d5f928aa1e6d41594671b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:24:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/ed2492ab836a41aebfc3945dfad5a822 : SUCCESS in 50m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff9744356cd45b1b88d48d3cb74a87b : SUCCESS in 12m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/152587e146a64ff1bb2819e39bdb7906 : SUCCESS in 24m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47d5b2edbebd455b91bfc3dffbfbd725 : SUCCESS in 28m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b5d422eff9bd4c84b2e06677cc6d78fe : SUCCESS in 7m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc5a3b51f5dc4958b96045efb011b0b7 : SUCCESS in 1h 11m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bbcac28e39e44bdbbc3693e73a484a61 : SUCCESS in 45m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4828f9aa03f4495c84250d8f48a68c69 : SUCCESS in 1h 19m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23369e7f787d4207859d97813b9da13c : SUCCESS in 43m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/417dbadbaaec41e9a8f5c2adfb749e79 : SUCCESS in 1h 13m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fe87e614bd4a467592026d7381aca5da : SUCCESS in 1h 18m 08s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e56e6d7b0517435e8f00b6d72eb4bd60 : SUCCESS in 1h 16m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5820756ba4e4e10b8fb4dfc3fab6add : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":2},{"id":"515582d0dc0d6041038e31d93716cefbd7f25f0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:24:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8977e2e7a49804b0464df9da13c6838362b0bef1","revisions":{"437120752e74648cc4ce812692f9ab9f89ba062f":{"kind":"REWORK","_number":1,"created":"2021-05-05 06:44:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/64/789664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/789664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/789664/1"}}},"commit":{"parents":[{"commit":"308de8290a8fd66c89e81968775c51825a0c4bfa","subject":"Merge \"Fix OVS conjunctive IP flows cleanup\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308de8290a8fd66c89e81968775c51825a0c4bfa"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-08-19 12:39:35.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-05 06:44:32.000000000","tz":0},"subject":"designate: allow PTR zone creation to fail","message":"designate: allow PTR zone creation to fail\n\nWhen multiple PTR records are created at the same time, Neutron may try\nto create the same zone multiple times, resulting in a conflict. Ignore\nthe resulting error instead and try to create the record anyway.\n\nChange-Id: I59b0f99463ab00743f19016a890561916df900ed\nCloses-Bug: 1891309\n(cherry picked from commit ca1c6fd69bda5f392712a8a966bf4f9b85ecb57a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/437120752e74648cc4ce812692f9ab9f89ba062f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/437120752e74648cc4ce812692f9ab9f89ba062f"}]},"branch":"refs/heads/stable/rocky"},"8977e2e7a49804b0464df9da13c6838362b0bef1":{"kind":"REWORK","_number":2,"created":"2021-05-14 08:00:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/64/789664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/789664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/789664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/789664/2"}}},"commit":{"parents":[{"commit":"308de8290a8fd66c89e81968775c51825a0c4bfa","subject":"Merge \"Fix OVS conjunctive IP flows cleanup\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308de8290a8fd66c89e81968775c51825a0c4bfa"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-08-19 12:39:35.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-14 08:00:40.000000000","tz":0},"subject":"designate: allow PTR zone creation to fail","message":"designate: allow PTR zone creation to fail\n\nWhen multiple PTR records are created at the same time, Neutron may try\nto create the same zone multiple times, resulting in a conflict. Ignore\nthe resulting error instead and try to create the record anyway.\n\nChange-Id: I59b0f99463ab00743f19016a890561916df900ed\nCloses-Bug: 1891309\n(cherry picked from commit ca1c6fd69bda5f392712a8a966bf4f9b85ecb57a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8977e2e7a49804b0464df9da13c6838362b0bef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8977e2e7a49804b0464df9da13c6838362b0bef1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
