)]}'
{"id":"openstack%2Fneutron~789831","triplet_id":"openstack%2Fneutron~master~I8572906cc435feda1f82263fd94dda47fc1526e1","project":"openstack/neutron","branch":"master","topic":"bug/1926273","hashtags":[],"change_id":"I8572906cc435feda1f82263fd94dda47fc1526e1","subject":"Sanitize MAC addresses","status":"MERGED","created":"2021-05-05 13:30:35.000000000","updated":"2021-07-28 14:29:11.000000000","submitted":"2021-07-28 14:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"789831-bug/1926273","meta_rev_id":"c1c904542e2194cd72304bce37792d73ef1f60d1","_number":789831,"virtual_id_number":789831,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 14:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 11:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-07-21 01:40:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":2,"date":"2021-07-27 14:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":1,"date":"2021-07-27 14:55:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 13:47:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 14:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 15:26:18.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-05 16:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 18:50:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-06 14:55:46.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 19:17:02.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 06:55:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-20 13:31:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-20 13:31:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-21 01:40:05.000000000","updated_by":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"reviewer":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"state":"REVIEWER"}],"messages":[{"id":"cc203f10aa7d5345927b88dc8d41c95d935d3882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 13:30:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"977b1db043e3ff0d182efcaa393fda1b748b0d68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 13:47:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"139fd3e5a775f47d8fb9dc16feda5aae251edaeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:20:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2296f702a5024003b4eb9e5cb072a87d : SUCCESS in 47m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a4a6344d78a46b496ac40ad0cf45d4d : SUCCESS in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29a04456ef2a9a4b5d10703e8ece19706d741e92","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-05 15:26:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 49m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f4a63e4adf3cddad5addc8d2a65fe6f5088d1b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:28:37.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/474f8b49163e43b388e352ccf66dd783 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0f1c8506c47b46539c254296376be14b : SUCCESS in 58m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/318b44658df34e12bb8c23df2a1603bc : FAILURE in 1h 26m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9d9fea05ec6a4ea191c5fbc911dd2e4a : SUCCESS in 2h 38m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4df32c9b0c7149028b1ef605c5085503 : SUCCESS in 1h 39m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4f558ba933d84e55bbed59fb1031cd30 : TIMED_OUT in 2h 51m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f31b5ef7534b47528e03b0c698239505 : SUCCESS in 39m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f36c73fd39a40e29b787342ef5e1865 : SUCCESS in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c40046f8b16446d967a71872fcf338e : FAILURE in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0595c4631144a60af888c840db2f023 : SUCCESS in 28m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33c4d09089124914950ffe6a1899ca7b : SUCCESS in 28m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d34f98330e9d4f948d2f3f5d0d930a03 : SUCCESS in 32m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180fa2a8933b4426896cf858e23fc852 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/102f5e28aaf646c9a058c44cad2c8381 : SUCCESS in 53m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/276b5e7aaf4b40648cf9e0eb31e4913d : SUCCESS in 1h 33m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/58515c358837420b9dd5c0e0fb954ec6 : SUCCESS in 1h 43m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2ff2b59fe9445c7980966fe46966469 : SUCCESS in 54m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e76b2ff5b2d74e0d9bad90fa4fab46e1 : SUCCESS in 1h 09m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c5dde1d1013f4ba2a21c2c163d05210e : SUCCESS in 1h 32m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/880ec40543d6460a88bc46761c1b0a58 : SUCCESS in 1h 43m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea1a143db0564381b36ae1c79ec557f0 : SUCCESS in 1h 18m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caf6b18f67094e07960cfe7e5cfa0496 : SUCCESS in 53m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3d80afa0198a4d62a82f665ee669f1c8 : SUCCESS in 1h 09m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5f6ed2dc9f984798898e234275c6681d : SUCCESS in 35m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1d587222b03435587673d8f7e4148fd : SUCCESS in 1h 22m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/49cf21b03044417aa1f47d7ff19e68b3 : SUCCESS in 1h 35m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a1ccea7530efd79ef3eddf3004231f7af10f020c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 16:55:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d7f473adfc3e83c65a8cdbfc7006168e6b2a06a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 17:11:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a8af3508bfa04e17df59cff187731e213db28608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:41:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2408607824f048cda9106f2482dc297b : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8682878848c94f4db502cc9b2591626f : SUCCESS in 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a2f431805581f57537b2d7d12370b43c4b547f1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-05 18:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d347f8e24629dd41d0d13433ce0a2918f2df32a8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-05 18:50:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46a85d2412688246a521d14ec31199503c1ac3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc78fe9ce98f479988d5456faa715f83 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/677bb654629b48478c1ea745bb6aa002 : SUCCESS in 35m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/874c112d163d4f37ac9c120f31754ef3 : SUCCESS in 1h 17m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/86a68adebb824469b3dd77fc48dc53e6 : SUCCESS in 2h 07m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42fa3a2f4c1f45c6a0321727719f5188 : SUCCESS in 1h 49m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9de4cf48cf11402187dc0229c46075db : SUCCESS in 2h 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df38407928124489a4e15fbf8c98668d : SUCCESS in 33m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b920942e479415e845cfb00ca02e4ac : SUCCESS in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795c3684c71148bfaa1fd627dfea686e : FAILURE in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e32fbefe087d481bb2b24b2fa43843a8 : SUCCESS in 22m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66d9ae8c6632417ea3b2fed742d407d4 : SUCCESS in 29m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c2cfed6bdc485b81a8f69ae75802e1 : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ed1953c21344718fd066f06e2411b4 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e82c4753dd11484bbd7e9a477339f95d : SUCCESS in 46m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c710e4fb0dcd4ffe98869ea7d362aace : SUCCESS in 1h 18m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1cddff9274c24539a31bf93c7001e89d : SUCCESS in 1h 01m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f0e3d0ddbb14f769e1de03df0a85ed4 : SUCCESS in 1h 11m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8765005cf7d1423e899e7771155838f0 : SUCCESS in 1h 04m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/402ed7bfd98c4d7ba8c196d2486f36f1 : SUCCESS in 1h 15m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/627a7f8442884fa29e7a4ea373056885 : SUCCESS in 1h 45m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38aff01cd1484cc097c2459e3f91d096 : SUCCESS in 1h 43m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eb7cb6e11d54143b2982c57cc0f87bd : SUCCESS in 1h 09m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e9edf9357abe4d7bbb18b1a20cc8110d : SUCCESS in 1h 02m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/201f32c2430e408b9baf3ad7acbb4238 : SUCCESS in 33m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b4cc65a4558489e9b57b12334dcca22 : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/657cdc21cf2c4fad9bef10872c5fe956 : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dc10e74a391ec3a488ebedfcee682faa5fca0163","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 19:25:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1916637afccac768034485b7f6390e4fce12e7b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 19:45:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fef678dfece26734b31a55acf0e727649d55bee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 20:09:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a394f94ded374a78b52f4420c7b04da2 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a9930229ea5473bbee94f6da3e2588c : SUCCESS in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a036d085ac390f907327313a5ed9685f0b23570","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-05 21:14:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 47m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f3e2f0f8a9681e2dd54722537ea87df314ff8a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 21:36:21.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35b5dde91f1c42c5b2777b041ee3ed25 : SUCCESS in 2m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa8cf0762daf4644a4497dc8f6d46650 : SUCCESS in 41m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4292ab9cf7944357b0db7526fde529a4 : SUCCESS in 2h 04m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b617cf18311e429c9b59f5a6bcaec2e6 : SUCCESS in 1h 45m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98f2ec8926e64e798c36cff205d4066a : SUCCESS in 1h 31m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5c4667fe0ff046a1a236ceee80e538b2 : SUCCESS in 2h 07m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48c6197b6db94b04a26d7718daeea583 : SUCCESS in 51m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7df1056edac487daa19658a6cac5f2d : SUCCESS in 39m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b87a2df74ff4574812cbbe849a93078 : FAILURE in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5253422f16a4f99b6163634608f2600 : SUCCESS in 31m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e3ab52bda9241888decf32c3e60676e : SUCCESS in 34m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4135c4a352fc4e85864c004d3b12f902 : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e025922e231f49c690b0982a013b52cf : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97fd7ff0ed974482a34d658673dbb6ec : SUCCESS in 51m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/185403ffdbdf429580c7a8b05c711b52 : SUCCESS in 1h 19m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7e7781d5aec842aeaa193186c0322d1e : SUCCESS in 1h 15m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cf43a15e6e043f1891815fe894c3a7e : SUCCESS in 1h 36m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ebbcaaaae19457aa53e788f76417b13 : SUCCESS in 1h 34m 05s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8b03d5795cf24e468cf5c694cb63384a : SUCCESS in 1h 11m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa831dcb6f8b49738f7b0edb8efcae68 : SUCCESS in 1h 21m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0f54cc1bbe94c95844ad41b29cc94e3 : SUCCESS in 1h 43m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/935d6d2c769e4df48a9d9842e6a25efc : SUCCESS in 1h 03m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c75812cfcc2147698cfd231c3056457e : SUCCESS in 1h 21m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7d087b2930f64ad98ab59ad41bf84307 : SUCCESS in 27m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17c273021aba420fb5349720a82ca6c8 : SUCCESS in 1h 47m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/583648a7a01941b2a3406726b5a519b4 : SUCCESS in 2h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6617184bc7cee4acae4ff4c16932e18f871633a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 21:56:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0171f12574710f033e95331a73b4ec44f7c89633","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 22:43:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3628d21c7a59478994f595576da290cd : SUCCESS in 44m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62c4a948b35b41768d74ef25cd0e2cbc : SUCCESS in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53efed1d59e3467b2e5b0e334d8ba8124ae4734a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 23:10:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5443bde6cc71648eb0ac7110b42ba01082c1dd54","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-05 23:27:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a859fb411c7b868fabf8a9e030d0195f8fb241d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 00:17:05.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f1e512667ef4e5ab72603c9aaae6af7 : SUCCESS in 2m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b7ea4dcf8be4cb98633a293fd6617ea : SUCCESS in 48m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9a5a6b66a3c344599d57c77d6286462c : SUCCESS in 2h 13m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/06b769de1ffb4dceb2e46f074b02d421 : SUCCESS in 2h 15m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ccca8c81ff6e46258164c75d5414d578 : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2820cf97be16462eab80455f8e994654 : SUCCESS in 2h 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c1d0a6be4343b7831a5744ae9b46b8 : SUCCESS in 42m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/076009cce7314457a9fadb4c9a7f1772 : SUCCESS in 33m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75fecd45cac446792006940ec4d143c : FAILURE in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0710cead86d44eeb7f8b01d297cc794 : SUCCESS in 30m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212c9e81c3cd4b8eac11d260a4074ee9 : SUCCESS in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b80d5c80ed084cecbd6b8650724106be : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97216cd92ef4709a1ea8fcfe4029130 : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b29241e0b364e419bb2be3fb067e697 : SUCCESS in 56m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4218eae9df064c2e958a0877b2d8c27c : SUCCESS in 1h 24m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c15de6b3bbe849be8e48f5bb1d503a97 : SUCCESS in 1h 28m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c576977bc2b0425a8a6fd6c8059ed88a : SUCCESS in 1h 10m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4f014aa5fa74276aa7b5204bf98fc15 : SUCCESS in 1h 18m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a12fe6058c5243bd8b622de119d204e1 : SUCCESS in 1h 20m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d03de080df054c8dbf12a2222bc5e535 : SUCCESS in 1h 44m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/64c3fb82375044799c7ad51eba6a9e3f : SUCCESS in 1h 53m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1d20500e9e24d8790d58a6f7dd32464 : SUCCESS in 1h 08m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b0f316ec9d894f6795ceccc434d05a5c : SUCCESS in 1h 10m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f3be6dc0608148a6a213378d37786048 : SUCCESS in 39m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2191345c4b94baa93b941a64f8e5a44 : SUCCESS in 1h 29m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/edf93e7ce8c242f4bec707e496099a31 : FAILURE in 2h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"060498cb204050f25779f856ef3d66a6ace64935","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 00:26:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35df7eb07f81305c6fdbdc7b40029225db87f35e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 00:47:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dc27c268670e4210213bf83b2db1e325789f5c86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 01:09:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf7db0ab5e7f45c5b087c2312529e650 : SUCCESS in 40m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bab63a5d0e44844b5696f0b1bb8f6d1 : SUCCESS in 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1398d96a58869dc60d2fb96969b1ad8d3803e2a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 02:17:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 49m 47s","accounts_in_message":[],"_revision_number":1},{"id":"64befe1c405bb9d1e44efc1b7631d3da6f5e85ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 02:49:59.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c1722a027b1485f882a44d6d03e70c0 : SUCCESS in 2m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0d6705b0b26d42f3b4dbbfa1e0ad4e3a : SUCCESS in 39m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/60f715300d544a3482fc683a1a34062d : FAILURE in 1h 38m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c73edbfaae594206a4ad3fcfdaeddcbc : FAILURE in 1h 54m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20b3d7b6db8d46fbbecb0e4b88653730 : SUCCESS in 1h 35m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c366f334299f4224a010221672e7e71c : SUCCESS in 2h 21m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2c32adf8ce64c17b647ec46ad47f348 : SUCCESS in 30m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/674ba293dccd465abd77a1ca282f26d8 : SUCCESS in 26m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ece5d34995341f48df58709ef3c53ab : FAILURE in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9cd4da93884dadbac35c79bbdc4026 : SUCCESS in 25m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7699bef729d4ef68f5a0c74a96f5f15 : SUCCESS in 32m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14e47fac113147f5a7266e32cf9d0c05 : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9d0c4d4434405598896479b66de18f : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fa33eb8e09c4d559eee51f170400402 : SUCCESS in 46m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a521e48cb584f669abd711cdadb9a79 : SUCCESS in 1h 18m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/96c117767ea5410da77a8d3adeb473b3 : SUCCESS in 1h 10m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0454534a82f4038ada868d172fb1d2d : SUCCESS in 59m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3acaf4a2fa8e405d981918426cc4851a : SUCCESS in 59m 11s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f363d4d1fc3f4329a2fe2942d710a25b : SUCCESS in 1h 09m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7ccfce13efe413a836d59ee6de0e2e7 : SUCCESS in 1h 20m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8524f86177b74574bd767888abb354c9 : SUCCESS in 1h 30m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3140cb0a6fe448a8670a3ad2cd8b954 : SUCCESS in 50m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c539015a9b8346baadd0933e430ff614 : SUCCESS in 1h 00m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/65cabc325e614178b35c16d7703c0f72 : SUCCESS in 30m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3075c2e5819491196dd48ebb519e7c3 : SUCCESS in 1h 14m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87fe4058957443fd83a60983e14afdee : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"42728ac71faa6b0d011670017f2cf9b407b0f1d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 02:57:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75b18eb5c5c817fddcabd93245261c10196632f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 03:17:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"343c5d735cdb4fe9fc44de6670e91342704b7871","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 03:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b4aa1d160143268ba8ed951eb4ae0f : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85997a89d4eb43b7800a71e91afff2f0 : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57a48925702577c455acf7ac6b7a268ab0187695","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 05:03:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 54m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8be1425e71fd183e3d7daa3ee0d17ac1828f56c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 05:13:20.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aecc14e3a7d54ccfaf2ba86e69614b79 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/888aea1960d14af098388aa09c80dd22 : SUCCESS in 35m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d9a1fea82108426dae05d05ec1e94a83 : SUCCESS in 1h 48m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a294f7a6b8ef405fb20b6bb24919e530 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/248020009e9040c0939ea0dc2dd11de6 : SUCCESS in 1h 21m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cd7cefe842614bc08e0c64bd0e22d4d1 : SUCCESS in 2h 13m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4001e6acc9e44fe5a5d352120bec8213 : SUCCESS in 43m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26a7af36587d4468810f380620293619 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2506e93908240c0b964bd9c1507ac8f : FAILURE in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/829457cf817a41929c7c4efd81439471 : SUCCESS in 23m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eb04bb3c3d6485bb4018ac755b1634a : SUCCESS in 22m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb71202c43874bde945055fa1775dd25 : SUCCESS in 29m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd6b353a2d04b50ac14025a3422edbe : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fcc2ce88b2d4e1cade90d93f3f0142b : SUCCESS in 36m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/500ae3c70d344f3fbd74363388d3a3ae : SUCCESS in 1h 27m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/751acfd7bd104e7f8ed003b3a29c34e3 : SUCCESS in 1h 01m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/374363c936b84b8e8d56324f273ba85b : SUCCESS in 48m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/270ba304508b41a1897853b3748e613e : SUCCESS in 53m 10s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cf9e90ab07d34a209a520372e18783e5 : SUCCESS in 1h 09m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d51c770d98574f29a0fb999de262e268 : SUCCESS in 1h 32m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a21f51c731f546e48991f093957f13b0 : SUCCESS in 1h 37m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1ae72ad5e84cdcb6ea3fb0cc30fb22 : SUCCESS in 53m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a41b6d08eb644a8db75d2b24d7fd859e : SUCCESS in 56m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d1a6ea8229534f52a15007f4f44cfc8e : SUCCESS in 38m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c86b0a46edd64786aee30c2fceffdc75 : SUCCESS in 1h 42m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/85d22dd0e12a4022ace6a20a090507e7 : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4d21a8c5fa9c872ac97637f49618f9c1935eeebf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 05:27:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a4b9830a016a18389e255ca3119393f8387b974","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 05:47:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5a1ed8aa5f991031f9d6443e3eb904dcedfad7f2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 07:00:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/1 : SUCCESS in 1h 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"39e90f11f1d355c9bb3b48014095981b957b6899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 07:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3aab0c4bb054f6ebcab5dab28f9916d : SUCCESS in 41m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b54c374245524c6683e7feb2e9d4865f : SUCCESS in 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e13a071e2e0d8a48820381af567933b185ab2d5d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 07:31:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9194df41fb1ff4f308c4795665ee953dce428c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 07:32:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f84bb1da3f7089d5e1ef43328ffbc688440407de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 07:55:40.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"247ff615525bdd05569ce693a83b12f814014d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 08:23:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1295f2edc224d38b0a1e62f6bdd5e72 : SUCCESS in 46m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/410bfa6b5abd4af1a77b8d3f39a33913 : SUCCESS in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d385430df435d3aeeb0048d10c4348932c4398f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 09:16:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/2 : SUCCESS in 1h 43m 04s","accounts_in_message":[],"_revision_number":2},{"id":"dc16bf4d598d21734323e515c810adafbeca997c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:53:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb9906e559664cda9578d2c6e6aa81b9 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8e0d2a53a43a4bb69aff876d565d6d71 : SUCCESS in 49m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6d75379d193b49fe80eb91636b906c11 : FAILURE in 1h 58m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7766762e1b084b3b8db5db5847486aed : SUCCESS in 1h 41m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02f84df62a2e4dde86a430157291d39d : SUCCESS in 1h 28m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fb1b9399765c4654a6bbb95552f105ad : SUCCESS in 2h 01m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ba149b8a4c0457ebe8ee0e04e5af3fa : SUCCESS in 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e8e43988c7e4565a9c8f7463befd942 : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc97beed89e4c409aa4a3199e9c8964 : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5cb476763b54e93b92ec81dfe9e8953 : SUCCESS in 20m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/265dd6e8258849d4a34e2994395af6a0 : SUCCESS in 20m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8153afa9af14b95a90606047636b556 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b772d025514a17bcf12068b55a4c57 : SUCCESS in 12m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1145a3fa6a12415e9851be436e5beae8 : SUCCESS in 50m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/11b344bffd634fe9b1cc6c1d9cfe4953 : TIMED_OUT in 2h 17m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b9406ff268644b859a76eb9a51e78347 : SUCCESS in 1h 12m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cae80042c4e84181929f906ce9bf397e : SUCCESS in 1h 06m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eaa75c5fcd3e4f6185204541ad612f0a : SUCCESS in 1h 09m 30s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b91cf0309cab4b0da9f70ac4771ec36f : SUCCESS in 1h 13m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/559e1f89557b4a7981c8e2267ed5795f : SUCCESS in 1h 13m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d00d465ad0744d88ed5e32d6c40fe7e : SUCCESS in 1h 39m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daf90f432c254c66a70979a31345dbe0 : SUCCESS in 58m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c810311d13824b08a527369bbd64786a : SUCCESS in 1h 03m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d163ddbccde64db88fa589b3d4abae64 : SUCCESS in 35m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c71f7ee0ccae47d581694d335a6f0ff7 : SUCCESS in 1h 22m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/92632b401dc94b75bb9c25423390a789 : SUCCESS in 1h 17m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5edcf470800d659fe8a9094708a39c9369d57405","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 09:58:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa70db5c9ce5043229a39fffde343112c9562333","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 10:19:16.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e6984a045ed1e6171dfbf2ca7d4483c25c36b129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:56:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec00bd664c904b2a9667975990004ee2 : SUCCESS in 45m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55b098ed5c5a4accaee8d4373885b9ca : SUCCESS in 46m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"388b836d64a55df78ec8b7f5fdceeba4ad59dc63","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 11:33:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/2 : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":2},{"id":"dc25099bc1423f652790fa2a4217c0e959b3ed1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:21:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da7715e36f594e7eb675304d9ee15d6a : SUCCESS in 3m 18s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bba391a5c5c5424291ef24da6112051b : SUCCESS in 46m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fa44b99bd919426a8d7d56f011f9bbc5 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9815c8e30d454599a0a296d49ee4ef27 : SUCCESS in 1h 44m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/462ad53af3494ed098afe50746c57d0f : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/755b76e946c540f69affd39a92c74299 : SUCCESS in 2h 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbaaa7c24ce648dd8ece72b118cd2bb1 : SUCCESS in 38m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48c3c38346c14a61b1aefaa99f4d1b53 : SUCCESS in 41m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013427c5af634f068389335b4b719412 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed43c755bed4007b9a3ecc6814c08fc : SUCCESS in 29m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6289476dca8548ad826e6afa113b9bcb : SUCCESS in 36m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28dfaa1e66c34bc59c35e5e479bd3b5a : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46dc7a0ea89d470084dec2bb6491c136 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/74d0ed3d6f024ec8914266e25fbf58cd : SUCCESS in 58m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e7f644c9a57412bb39e16ceb4bda7f9 : SUCCESS in 1h 19m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5912e87251e84d5cb40663b88dbb5aaa : SUCCESS in 1h 23m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24918070b21140ccbba199b8523c7721 : SUCCESS in 1h 08m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cd20e4e3589f42b9ba7a76e62002aae2 : SUCCESS in 1h 06m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b4279e4ae13244dd8cb87f3bc8c83eae : SUCCESS in 2h 12m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99cfdf762afb4412b6c7be7b553abb9a : SUCCESS in 1h 22m 26s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d2ca3a0c3674028aeb1873117bd3072 : SUCCESS in 1h 33m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67eb8efa94ff40f7ae66c4b659f025d1 : SUCCESS in 58m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/04088f15016e45c183b922cad0c6078f : SUCCESS in 1h 04m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b4b19944166c4979bea70760d791632a : SUCCESS in 33m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb98b6b15e9f4a048560e5fcbe991a41 : SUCCESS in 1h 10m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1508b7a0b45a43659348d067ec518396 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ad924764e3ff6177550dbce13e16b999ab579b82","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-06 14:55:46.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ff5a04d74107503dfa002576ae6bf99b45c0202f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-06 19:17:02.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/31/789831/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3eb12bd0b081489aefcbee9d3dd21a8281bfd262","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-07 06:55:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\n-1 just because I have some questions. Code looks good for me","accounts_in_message":[],"_revision_number":2},{"id":"44d1fc4fdf805e5bf3c63ea24809b7be0414a9ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 07:20:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"71836ac87a19ab07ed2d0cfa7404cedc92919376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 07:29:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8eda085fa46529c6236107713dee3b397134d26b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 07:40:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a05e2e09a2940f2a53bd538f80ec120 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/253db1b477ca4e43ab7bf2fb226e5efd : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a933424876595c6194d01a581eae36750b90313","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 07:44:03.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/789831/3 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":3},{"id":"731f4170edfb1a60d83c3f00a8d7e0745396064b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 07:45:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"4dc3b068d15b0372082464f69aa21f853b686ab4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 08:52:25.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/31/789831/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"034d9a80dd4bb7f097eb40895cdff1d5468c939a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 09:05:59.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"8d4edef30eb6b6e5ffc55310740403ea300cbeea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 09:25:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce8bdb95a2a2443fbf2fbda915c9eece : SUCCESS in 2m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0552888bdd344b1182089ae65ffc2e1f : SUCCESS in 50m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e3a8bc73e2ea47c8974945711cd6ab65 : SUCCESS in 1h 54m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00b9581c7cd94f9986b58fb94b6db820 : SUCCESS in 1h 29m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f0d691980be494ea54932f0448c62f6 : SUCCESS in 1h 27m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/407d50dcc1a84f3f90aa38ca98cb3003 : SUCCESS in 1h 49m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97ffa85ce9ce4b8e808c263b13607ba6 : FAILURE in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef2a01d6c93d43989252d1eeeb5d7bd3 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd9bf8260ebd410babd852628a4202cb : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f689b3319d496b99b7d8a3120b7090 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e53b813be634eca9ad58e3a4d546a31 : FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/292413288fa044b3b035a65a5548ea39 : FAILURE in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2f8e37a4b449018255616bfeb38699 : FAILURE in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9954c1f627ed4182970d9d88edabf936 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c740d170a14d493cb7418ef9a2290283 : FAILURE in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da82c14e9a7e4bebbae7183dbc8e7b01 : FAILURE in 15m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fefc83dc43448cea844e1f0a438208d : FAILURE in 14m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3f59fed457d048a3ab3cae15d93307d5 : SUCCESS in 1h 21m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/617660f74b81472dbe246b9083764cfa : SUCCESS in 1h 01m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7840be7a39a941aa89ec8e29f6c99a05 : SUCCESS in 1h 03m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c402b086343b4e52a5682f4889b65b5a : SUCCESS in 59m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/75dde806d68849abb4eb2d8297e3748b : FAILURE in 1h 08m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bdfc42b2c3945ae9594ece982e1d6cc : SUCCESS in 1h 29m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab93978619b54ad895715b26c42a1ae1 : SUCCESS in 50m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/43170ffcefc74ea0b617bf03d1d04b57 : SUCCESS in 1h 01m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec60ee53c5b45ffa745489432411d83 : SUCCESS in 24m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19bb9bec278f45559cfb9db7d3b4a826 : SUCCESS in 1h 03m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8356086e31c54eb888bf702288cb7e75 : FAILURE in 1h 13m 09s","accounts_in_message":[],"_revision_number":3},{"id":"14aad2081513e5ec3c31e74213a15ccb270262ae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 09:33:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dbdb22e98e0c2ff949039ac932af87ecae056382","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 09:48:14.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/789831/3 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":3},{"id":"c62b9b500d8e9fdc570cf85a8d57a96de134c6f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 09:48:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c16735b61de94f59b56fd3a993a5ae33 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94c702d518024ad7b824a903fc7a1850 : FAILURE in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70ced648c1b1e6b97fc436fc972fb94e6909d926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 11:51:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b75d79cde486483f8d4dba60ccd19a48 : SUCCESS in 3m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe18d815975044bd96e900e5764ca136 : SUCCESS in 41m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0757ecf4cdc4b7689ac90074d4820b8 : SUCCESS in 1h 41m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c07ada01559e4b34874b39c6ee8cdb54 : SUCCESS in 2h 12m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aa8d88b0eac24016b01ca1db5063bca1 : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/84d7971ebfee49b98c61d73956f99c1f : SUCCESS in 2h 12m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b788ae632b8348d49962ee062b1db483 : FAILURE in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e70b7d55954249469750d4f8b01ffd02 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72743968e33241d1b7d85486340d7d16 : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f1f858bbe044116b6b610edfa88ca78 : FAILURE in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64fb6e13d7a949a49e1bd3cdd32d4990 : FAILURE in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d499886673674d85a8942193411ef674 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f98486a070400385f932a6d0415833 : FAILURE in 5m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/47b5795ae6af49f89846e5112ec6ddd9 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e864549136304a08af6cdcad5d2784b1 : FAILURE in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29f2f4965b114235a375f2bccfb23ef4 : FAILURE in 15m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca1084000a32406d94d66cfaac98778f : FAILURE in 15m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/871aacb02ec642dabe01f00420adaebd : SUCCESS in 53m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f5a0b7f9ac845eebcf57b31ca415bdb : SUCCESS in 47m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e28db85d168f4f31bf93e658e0bde9a2 : SUCCESS in 1h 02m 51s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3ea22f55ee144841a9d00ddd403ac91e : SUCCESS in 1h 16m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/377efd74a1be4d4898aaa14d328fd23e : FAILURE in 1h 29m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e67cbceef7b43bfb46273644c086a61 : SUCCESS in 1h 26m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8518155f4182460d85dc47c734593109 : SUCCESS in 56m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/559eac157dd545c295f7f1b013db38e3 : SUCCESS in 1h 00m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/739b8f70422b4b86913330c97b813bc7 : SUCCESS in 29m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02001efdb2c14f70ad573acc943a1868 : SUCCESS in 1h 16m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/118ca8f31ce94bf9b2244b6aa3a67f1c : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":3},{"id":"15a934640e64cbac667e378cdeddb57ff586e1d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 12:04:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"87f2bdf8d1c392d860a948d639de3cf8fa5494a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 12:05:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"bb253f52372e6d1ac5cf21436f13defdae1af362","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 12:13:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/647a389e88704abcb87694f48c8e16b5 : FAILURE in 7m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e244dd613b94f0799eb73b1b4c1cd3c : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"807f1daa156e81462661fde5f2b49a0906fa1564","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 12:24:46.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/789831/3 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d4d28ea5b0a540c605ad92908f8a1e695de38e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 14:15:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c3cbf243970444ba31e98276aa8b456 : SUCCESS in 2m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3a19c06c788c4f059524a9d55118847d : SUCCESS in 44m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8ac330a21e0d4b248c36afd2cda3fecf : SUCCESS in 2h 05m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a8f575adb62d476eb0a9c7c6141639d0 : SUCCESS in 2h 03m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/39b04fbcd3a44980aeda9076112c8f12 : SUCCESS in 1h 34m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c0dc1cbf1aff42ea8e04a740b7becaca : SUCCESS in 1h 59m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/741e4be1a9144b8a997e45f814befa0f : FAILURE in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7631857b33dc4fe3805754df2382658f : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa3e7648aeb4934bb7989710b624394 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3ccc0c5a1ff4e779694a257d13303d5 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5db6a003b4487b9fb28bb3c9344ee1 : FAILURE in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2209f798b454eb6964f599f9c2d5c4a : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1342734f084fdc927e7c640a2157f4 : FAILURE in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/406c173a3644454a9bc945cbfeb855e6 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52e44e2248724926b0bfc3a862828652 : FAILURE in 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/467535e3a8234a7491717e178d7049de : FAILURE in 13m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e9d83f4a63548b0b1eb9be7708d4f7f : FAILURE in 18m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/77b9a40d1f234e62a4b6a158ab210747 : SUCCESS in 1h 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e77380a1b15a405fa21010da5c05fa03 : SUCCESS in 50m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a909edc87634030bfdc833cb06fcb38 : SUCCESS in 54m 19s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4e4b70d5646946278c61c72fd1deea4c : SUCCESS in 1h 24m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/56eea4aee96d49b1b04a69d4c300eafe : SUCCESS in 1h 36m 46s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4bf9ef90210d4b54817868c30dbcc3e3 : SUCCESS in 1h 40m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e806708d96f745f5ad49cc22608e5f7d : SUCCESS in 58m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1681f7a3e8f04a8bbab18606a2eb7ba5 : SUCCESS in 1h 13m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/750d5e2667bf48668a6aac2182fc7e3c : SUCCESS in 29m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f11ef72e3e91472ab30d80173dd47918 : SUCCESS in 1h 20m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fd6fb8babb8a43e2b713fc77fcb362fb : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0424fb089524bd3a904160ed5f8561685b422168","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 16:03:49.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/3/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a0bd921df79927f1198ba583d24156182d335b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 09:21:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7f20afa0f463acb2a4df98c830547aa7c8b75653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 09:32:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80012dd47306416da2a2daecd1bfe553 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bce1ea849b04e778a800266bf408f1e : FAILURE in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b435e6b6124a4179387dfe3e88f9d61a267e7d5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-11 11:46:34.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"2332d4af99dd758c6738634475cc4d773590b139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 11:47:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0ddeec3cc8a4419b52b707a3a427b7f : SUCCESS in 4m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e52d111af6546738ef332a649e6efed : SUCCESS in 45m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eca8a9c5eeae4137a2178095d37da4a4 : SUCCESS in 2h 00m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e6d3c3f8c2e9418a8603ff3f364fa016 : FAILURE in 1h 40m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98f8757b35674988b9751453b8d5d0c2 : SUCCESS in 1h 46m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be99ef2312b44eaa839cbe3e8657d505 : SUCCESS in 2h 01m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1559cfe38f124327a3e63d3c84db2ccb : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15058d82bee94aaebfa0fe522a2de75d : FAILURE in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d48e4f1bb5e4d5386e12cd962539e21 : FAILURE in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f009c78022fd448c84dc9a1cc32658e6 : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9062dc927bf7490587fcb2f1e88755ca : FAILURE in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba1d060375144645a38e733525095172 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6085803ebd4ee4bbeb6451fd6a38db : SUCCESS in 10m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2b6b43038454513bbf1cbf707d3008c : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de509fdc94fd43479140e938c03b098d : FAILURE in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d68abd82955499d9381bfa625083482 : FAILURE in 16m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfc9ff13c7254904bb56999c30a1c6ed : FAILURE in 19m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a584b1efd38d4a86bef65321b7529404 : SUCCESS in 1h 28m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ec34b8d8d834f0095eb0337a527cc5f : SUCCESS in 1h 08m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19b1ac9264c046329d1452e45f66da9d : SUCCESS in 1h 04m 51s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3bf6185ba90140e5bfa396a4815c9139 : SUCCESS in 1h 14m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d672e2ccb854532940c4cd332537cbb : FAILURE in 1h 02m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93f26ac1ed944ee6bc6b07a52428c202 : SUCCESS in 1h 49m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9c6c86fcdd44773bee232a70f908d82 : SUCCESS in 1h 02m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8fa88c9f8a164de99f5ac68d20e26045 : SUCCESS in 52m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e46ef442a9584333b8e86baadc9f7528 : SUCCESS in 45m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1163b363adb742a2a5acd333013ffe8a : SUCCESS in 1h 29m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d6c083c3bb19429db84356ff71945b2b : FAILURE in 1h 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72114920d919524e1750c179c7186cc574ebc803","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-11 11:54:14.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/789831/4 : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":4},{"id":"08f793872f18ae04d0e4898a2ef86ecb4c7b4507","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-11 16:18:48.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7d91b2f1b88ac7023508ab8a3c385cd09534f6e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-18 06:36:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a2a5a19cf07eb8abeb7742ef26f52dacaf86a70c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-18 06:41:48.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"13541996c8856e276f9f43a34884397ffc39259c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 07:24:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/331c713718c348d2b606fe6c3986d430 : SUCCESS in 41m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05b149d3d39346f482eb9ab1c2fa44b2 : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6df286d8fcad75ffe3d4b30639103e15f6a61b75","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-18 07:58:51.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/31/789831/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"da7f4a76bf6f566b1f0071fa50b6bd93059ea691","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-18 08:09:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/5 : SUCCESS in 1h 32m 24s","accounts_in_message":[],"_revision_number":5},{"id":"f0b71c525e0b4911e47a4b8b53f0bc89a662c0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:56:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e021a4efe2df406e924b32b56db795e6 : SUCCESS in 3m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5bb5afcd1bb465194af846c72f945e1 : SUCCESS in 39m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/436de594f0fc436aa1e46ded2165d7a3 : SUCCESS in 1h 53m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/89c4663de71d455a9b803de2b4b9c83c : SUCCESS in 1h 45m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a58faba8995745b89d4b4bd81103d24e : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71d31b67c80f4231a6b372dc96861df3 : SUCCESS in 1h 57m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd8cde747e54ec0b0cdc30cd7f8699d : SUCCESS in 30m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca0bedbe53b5455cbd2d6afad91f0e54 : SUCCESS in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b4c0468a1f4dc1bb516f5036fe8386 : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/454a09e742d64612a4263fb28035f7b9 : SUCCESS in 24m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f19fd11bb1b0486db456fb7af6e150f5 : SUCCESS in 24m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb106eb1ecea40279eb0b8a7f996fe7f : SUCCESS in 25m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe5e1045db14d71834f07eea0273fc5 : SUCCESS in 10m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8dd2db3cec2e4e54b77c8b0b7d2b6a91 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f774556c25a483399a9ad34d37e5700 : FAILURE in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/edd9018fb5d54771beed614126003d38 : FAILURE in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74dd64ffec34fdaa1d05fb51d99fc6b : SUCCESS in 1h 33m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d20e50615dbd44d9b0d5275b69a4f40b : SUCCESS in 1h 17m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/354931799a7f409ab1e6c683b78defdf : SUCCESS in 1h 07m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/15b33fd17282488baa4813e59623b764 : SUCCESS in 1h 23m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e8beb8b5e6c94331b421957941cfb359 : SUCCESS in 1h 13m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c27855d63cf54e3dbbc00f51cac6b70c : FAILURE in 1h 06m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b4961d0485845a98d8fd175a302d4f3 : SUCCESS in 1h 43m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d753fa911a6e4ca5b64c1d4ea83fb3d3 : SUCCESS in 55m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ce18a0a528a4cec880786521e47a66f : SUCCESS in 1h 09m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4dd4833a782c4926bf157aec555d6537 : SUCCESS in 31m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/299ebdeaa42d4ed6a9622a60c47c4719 : SUCCESS in 1h 31m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/89e11d632eb34e2ab587c2313b9f55de : SUCCESS in 1h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2031f3692100a5155781f156a2a706643303837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-18 09:21:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"338fdef61f051282ca279fde73d26cbb89d8e45f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-18 09:27:41.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"2b3caee6659fb56e9e51e2c87d7cb73d362eb1d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:14:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82c2d1eb7a974203ae96a09ceddf9af5 : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54f1435953e74979aa95cbb9ada9c62f : SUCCESS in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"50bf3c90020c4a32634438fab1420f161b7bedd0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-18 11:16:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/6 : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":6},{"id":"ef648519945f50e14606c55fdd7c111fb2eae0fa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-18 12:05:52.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/31/789831/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"00272183ba34e2c9d4478680a673ebbd05c36fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:02:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e867e918eeb4fe0b2c13d247d5226f7 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7400622487d141939c5797b8fbdc2266 : SUCCESS in 42m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5257c120c03248af82e30835adb7ab2b : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/16ef2efa821e4b3ca6c5e9ba906cc3e5 : SUCCESS in 1h 57m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae7ea0eb0db7424eb4b75166bca4112e : SUCCESS in 1h 24m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a6b12a1fdad2434d956faf38decebf6c : SUCCESS in 1h 59m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2750ec6a7da440798e02fcf811dbaa4 : SUCCESS in 30m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a705983059af45a2993efcdbc9ea0f86 : SUCCESS in 39m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c7e449ede84ae2ae87bc991a4bb6ad : SUCCESS in 16m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeb367f7546242f0a79b5602f2feb8ed : SUCCESS in 22m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec7a6f472fc41b89df5942deb453db9 : SUCCESS in 21m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dcc5341383245428d21ade4effa6327 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e1ca71fb374117b6b3ec631b456e9e : SUCCESS in 10m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4561da78d8d64f31b91cab5adc25fee1 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1202e5cda874813827fb8ab3c64af44 : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed1a87488c884178817cbf5d2421c345 : FAILURE in 48m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/960f6a5005234d88ba48936445ec91ea : SUCCESS in 1h 25m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b0a102e8865a4fc5833c2476bc2c234c : SUCCESS in 1h 16m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3fa2e54bbc24b6c9c068b4062e997c9 : SUCCESS in 1h 06m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4257ce33ff454e548bbc45670f9a1faf : SUCCESS in 1h 01m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/129004bc6a544d4aa830aeda82ddde1d : SUCCESS in 1h 15m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e520bf7a477f4d45a0b9c043ae4554c1 : SUCCESS in 1h 19m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a363e9854a3458ea8ab97568b0f4609 : SUCCESS in 1h 32m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b19fbe51a6443a5b8b0ec4bc7e9bbe8 : SUCCESS in 56m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/be05766e38a648adbeb101a000fa733b : SUCCESS in 54m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9347879d791e4875967ad0373182316e : LOST in 38m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6665dd7b91847dfabd13b8ec95dcd0d : LOST in 38m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8293e1a2490a4796acd561fa3967c6a9 : SUCCESS in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"463156e826d98568302c170d85198c881898f6e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 08:41:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6f6f557b03f6eba7d9e98c5a260ec8efc519f9bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:25:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bb0bff010384917945fa60e8816b31b : SUCCESS in 34m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d47225f440448c3aed9983b949ac108 : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e745efb13302ca06b2f11effe5f3849a0e0262ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 10:54:13.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5ed5ccd0bcd2c98cd8ab8f003804709486ffc350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 12:44:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f506c462726c4d7382f10cb4ddeb7c57 : SUCCESS in 2m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5850753124634b4ea3a01b59daf371db : SUCCESS in 39m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e67496cc03864cfbbed4e5a0ae6454c9 : SUCCESS in 2h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e9b343d32dd14bb6b35b8ef0a2b40ecc : SUCCESS in 2h 29m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d3179394cd124c5f92dc87c417b79966 : FAILURE in 1h 26m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/95773f8ef304457d83be13d8bc7c79bb : SUCCESS in 1h 47m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/673f432b7b5d4c599177da80e56ad2c7 : SUCCESS in 29m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d153f337c7354927ba3669ccd721e009 : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763ad55132b742f5ba75a3e4424483d9 : SUCCESS in 14m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edf2db2f2cb84140be1f51b9c4277934 : SUCCESS in 28m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cefe99d3e5144bd0bb29bbfe48b5a9b5 : SUCCESS in 28m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dc1b5cc3ffe4cb086e587569d6095a1 : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942b6f4c9fcc40da90cb272e63bb202b : SUCCESS in 10m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff0c64ee9bdd4ecdba5ee62ca1c651d8 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80f5008d4604e1f8e58acaeed69255e : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41ffac72d8704791ac1b4f7b1c7c6c41 : SUCCESS in 54m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbda265988614f6c801dadedc61030ef : SUCCESS in 1h 24m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c0463968d1524248a441e0cd02d48d4b : SUCCESS in 1h 06m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da1fb72264cf4fc1857dbfcc63be3f8e : SUCCESS in 1h 06m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66212f4d231a40ca81ce57fd90580519 : SUCCESS in 1h 15m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7db4b0f43beb4721b00aba8fa7b1c3de : LOST in 36m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57c72f23ff2248ef90118da0b33eb1a7 : SUCCESS in 1h 24m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aaeac613e80440789cb0338673c491ad : SUCCESS in 1h 36m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a8f05d6264c4e9cbb1a255d1cb8b882 : SUCCESS in 52m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7fd99384f28a4f9ea8bdb692e83c118c : SUCCESS in 53m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9054b992f06c4c0ba1906300b9910aeb : SUCCESS in 29m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe58b67cdfa849e7a661045ce2c22701 : SUCCESS in 1h 11m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2e13892522f6413cb18751cf52a075a8 : SUCCESS in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3b6477d38b3d85057649498aa97951ef56b6fd3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 13:40:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/6 : SUCCESS in 2h 14m 34s","accounts_in_message":[],"_revision_number":6},{"id":"4fc9e2a9f95302d45329a7b43ba925dd5876ed89","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 13:43:34.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4f9ae1e41d4905d95bc201f64e952faaef048c46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 13:54:56.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"107c04917451aa8ba69168aff1c6c89165e5df33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 14:24:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76c452760b7d466a87ae1ac009f84ff1 : SUCCESS in 30m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76e4669d954944f88fbe95ff07f0d9e3 : SUCCESS in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"20f10abc430891ae9f5fe5bd95dc3dd0cbcc3b5c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 14:28:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3a549b776ce5df2abc448a23638fe77fb9839706","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 14:39:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"fd60bd8635b1db58541e7c5962fb83ce841572c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 15:10:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79ce0e43cc08446f8b89e2c0eeb1773d : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c195a8954e384b18acfc1878b823d220 : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e9f2c681b43dd45719909f98d06ab31ece5eb32f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 16:27:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/6 : SUCCESS in 1h 47m 27s","accounts_in_message":[],"_revision_number":6},{"id":"c3521471a42a9d924f7bb004147c75451aa38897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 16:34:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40506e0f37b9414fabb64f34f0d2bdf8 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0efce6c4cde941c0908c1d7bdcd0a82a : SUCCESS in 35m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8e3648d9e27a45d1b62a8fdd9786b485 : SUCCESS in 1h 43m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/593b259ab05243f9939c07ef23b0a1df : SUCCESS in 2h 17m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f45ba138d095427c87a8dcfc95823710 : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f47d64a04a0740b888173812a6c2468e : SUCCESS in 2h 05m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97e45b10f1834e57b86706ddc6afbcb7 : SUCCESS in 34m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d16be433872b4b5f99735544d13ea1e3 : SUCCESS in 38m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60e62b8349e446a8a385011aea1b242 : SUCCESS in 17m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c668b24cf8be4aa6a3a2984e4ecc7901 : SUCCESS in 24m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69e8d23135fd4a8b97c2f65c13212ed7 : SUCCESS in 22m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91a49099bc13492aa86403407b5dccd8 : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc72227eadb54e3fbb2ffbc868a5ea80 : SUCCESS in 9m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/17a9362d491a4b1995ad5e352c2bc7f3 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfbc4f5a995b43e194833f24d1754162 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e02010b9e8bb43d59bd3128c35ee5bb0 : SUCCESS in 44m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfe76e8f43ec46598ff19b85cbb1c2d6 : TIMED_OUT in 2h 17m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d44f45cb592b433393c8e786c42aad54 : SUCCESS in 1h 07m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6461684e79524919a565fffd8e0ae616 : SUCCESS in 1h 05m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20fed818dbfa423d8fca2d773180e691 : SUCCESS in 59m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b97eee75d99c41c69a01a837eb3e2600 : SUCCESS in 1h 09m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7cde72ff283d4bbe98b5caa2ef45c1e4 : FAILURE in 2h 18m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/647f2fefabab46a88fa0c8ac12e4462c : SUCCESS in 1h 47m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/223f22836e0643aebb8035eef226383e : SUCCESS in 54m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e2860dbeea0f499f89fdf0174735aa5b : SUCCESS in 51m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c348ff8a20243af91a031486216d546 : SUCCESS in 47m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37f2f647e96643bdafb4863cb4f5dc86 : TIMED_OUT in 2h 14m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/97e77a0c2b3d4a5b8ccce7176931efda : SUCCESS in 1h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b39cfbf2320bd0969a1257ac9e76bb32ec46bdad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 16:44:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e6806d79605daf4641afb4177b25b20e90079af8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 16:54:24.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"7650920bc63d7f3078c64bc1679e7f4832231332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 17:19:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b679aa65e264741943dc3a4703d1911 : SUCCESS in 30m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2628f71521794be38559b3544f6e66bf : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3e719461af53c82c63ef45dadef77bc7a246d1c0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 18:43:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/6 : SUCCESS in 1h 58m 49s","accounts_in_message":[],"_revision_number":6},{"id":"ce22a2f6cb795ff77b6251f9ba3a16d959889cbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 18:52:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e18bf6acf5b49938a06cc48a1e208a0 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ab0362f5a8f54b51a9d634bb79355ae0 : SUCCESS in 39m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/20754456fe2b44bbbe3d734ac867f92e : SUCCESS in 1h 57m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1780db2448df4c00a11ce0c12d7b9073 : SUCCESS in 1h 40m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3da5dc22542544c9bd0e22155a1e1d75 : SUCCESS in 1h 34m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0fb55b74e7514a169bed4539fb0db39e : SUCCESS in 2h 02m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd16fa9728514581a44b3412b8769587 : SUCCESS in 36m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58da2fdee254f95afbc40dedbb58a3b : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76572b9a3d7446e0bb64291a2aa3135c : SUCCESS in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfb95c6deaf64c11875cf2a2c0c56e8e : SUCCESS in 24m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5346beb3a5439cb73ff4f2c28380f7 : SUCCESS in 31m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a515886c64b4ecebe0e20fa10a30eba : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c999b2db324270a502332b059c571f : SUCCESS in 10m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/814154041464467da2f99c2a4e8a31b3 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf7910b37dd44049bde6b3f5d614ce2d : SUCCESS in 11m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f5479e86d8540e0a2e3e0d94ff07604 : SUCCESS in 48m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b4bdc2d582d406fab4cc8661a8cbc72 : SUCCESS in 1h 23m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ad70961a1db44d3c8d184985ad6f2097 : SUCCESS in 1h 07m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3fd38bcf21e4844a439e2dbc4a91f2c : SUCCESS in 1h 04m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be409ffbe6af42079230d6f0989d2257 : SUCCESS in 1h 05m 27s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/05952dbe460347359b1fb17224cd59d9 : SUCCESS in 1h 09m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/512a8f024e1b472b8002610fd9010b52 : SUCCESS in 1h 21m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/946c3c068dce4aec87da361bc7da4ce0 : SUCCESS in 1h 22m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/997f24ea7e28494298225a588b31ee0d : SUCCESS in 58m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e8f61d6c48a745ca9203fd1e439200c5 : SUCCESS in 54m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7ef98cc61c144b11b073d38c6a4e30c1 : SUCCESS in 34m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49505e13b0d4485491068b595f5870fe : SUCCESS in 1h 07m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/54a4d0e74a7b40b9a5527f061259d8f4 : SUCCESS in 1h 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ad2726c63fb831e7941f5ff106f37864e9edca7d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 20:45:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"edda2f677b8e759a82a2ea8f703d155242f25ec7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 07:18:13.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"726675ee2b4b57c43d5a926fb0a7fdcfc62582b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 07:38:09.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9c1323d428712086ed6e0ae7c62bb1435a859eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 07:41:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b283c7954e031c1dc728071c02943bb856c99727","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-23 07:44:50.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/6/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"f1a5139875287211e2041f7ca5788c72f4cc42e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 08:21:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f13c221afa3447ca94ee6ce4f50a531 : SUCCESS in 30m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f81fe1cda9d48fbb6fe783e620c82d3 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a7e6d0fd55dc7fde0cf3bffe228d205aedf72fe6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-23 08:48:12.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/7/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/7/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"a015b30338a2d6f0faaaaecb9cbcb0210c081bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 09:30:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35789d6de90c4a4283b696f8830d546a : SUCCESS in 2m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8b42112a790048249831c54226ee9816 : SUCCESS in 39m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f854b2297d547f2b584955a616d5a96 : SUCCESS in 1h 47m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/722d535f46cb49dc98abc10ba5e2a0d8 : SUCCESS in 1h 42m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98ac99fcc99d4665906ad83e6f5fa276 : SUCCESS in 1h 20m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/159ff06fc8d944e4bf38c33ac5b3dd5d : SUCCESS in 1h 42m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deff9dd6cdc64bf0a5cce13011589f9c : SUCCESS in 44m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be194b65e9924df1bdc1a5776789fbf5 : SUCCESS in 30m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ac46edf5984728830d6c465710c9d7 : SUCCESS in 17m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5647cec16ba64674846847e1fa102bd0 : SUCCESS in 21m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d88e1c2e73ca47d3ae0b1f27c540cce1 : SUCCESS in 21m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb753788c71f48ff96db98ccd4f7d43d : SUCCESS in 21m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c75f50cc2746fbb595a7f4de8a0b42 : SUCCESS in 10m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f09174fdad254c319cae531f24771ed7 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f1e3a0c82814d84b8457724f4043504 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e349cb5001c04fa2af48c76b3f2de9f5 : SUCCESS in 51m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/489142fc486e4903bc194b885f181bf6 : SUCCESS in 1h 21m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9e33b553e169437b9b2c06ed6d20b55a : SUCCESS in 1h 03m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c713f9187afe48fda593e990ca152960 : SUCCESS in 58m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/372e963648a14205a22f924d776a99db : SUCCESS in 1h 02m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d4983613e7ec4d5085164bf7da5bbe5b : SUCCESS in 1h 11m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9c7d261ea06496a98fa809a6b86238b : SUCCESS in 1h 18m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/992b208ac7234b4aaf977d2e12044d47 : SUCCESS in 1h 21m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e61c491ff92547e18ed92e5abc52bb30 : SUCCESS in 47m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/58642fa9b44445cfbe0d10cd7650be24 : SUCCESS in 1h 01m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/92899aab22074eb0aa9f48d9d82e5086 : SUCCESS in 27m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/595c77ef449f4c99b8fcd5614c240efe : SUCCESS in 1h 17m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/087fae30483a4e4390cf5cc12a45b5d0 : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"91bf131ba570e86838e4bbaf2cc88f7829486562","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-23 09:46:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/7 : SUCCESS in 2h 04m 10s","accounts_in_message":[],"_revision_number":7},{"id":"ebf83a64de958612b2f8fc1450f59fbf5001c9a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-29 11:27:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b47bee147830a1973aa53db9f646d63ed869e583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 12:10:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2972790bad6e498c9c51e9529b42a0a3 : SUCCESS in 40m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/467e56bee5584406bc4b3605c814c33f : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"75c6a25437458efafc6eb1d5da0f1bd80e517cc5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-29 12:57:39.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/31/789831/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"9bd410eceec048bd777934d0bb8eb530de68cfcb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-29 13:19:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/8 : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":8},{"id":"0d766aa666eefe5767684c6206c2c5adaef5caa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 13:39:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/38287f69c2214062b3f78a2515f4be2b : SUCCESS in 4m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5f72d5c17a544aba2737d987430bec6 : SUCCESS in 47m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/74a129d10b1542f9b67353bcb1141f2c : SUCCESS in 2h 10m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f63d6047505b44fda9320940c2db07b8 : SUCCESS in 2h 04m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a88232b77944a1d8b97233d7445493b : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d6df5ff8ee794e28b45db727a977b6ee : SUCCESS in 2h 02m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/95c0af733e1d4a5ba029b8f928968a02 : SUCCESS in 35m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3fcc718aa304a9ea2973f5873b92fbe : SUCCESS in 31m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/834ca2c223334aeb8a899f3b155c9a46 : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53c4864d2434267b1ff8518a5a082b7 : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8805e0db306b4a639f11585f004f2364 : SUCCESS in 28m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d726dd0acc474cc7b17c0c092c791881 : SUCCESS in 26m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21c85717fbb8452294d514ff581944b9 : SUCCESS in 28m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f3ffbd40974953b0e7c3dad790981e : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f3c0e8bafb48aa85594514405be71d : SUCCESS in 12m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff6ea9389f694003a7af9b47faad4f97 : SUCCESS in 50m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e73a14b7f62400a88b231565266eadc : SUCCESS in 1h 28m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ab04e447741e461b8fa9cdf048d1c6b6 : SUCCESS in 1h 10m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cfd3ebebcee48098d2e655e97b0e4c6 : SUCCESS in 1h 03m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c523bacd2f9e4d11bd09f6e8394d3629 : SUCCESS in 1h 08m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dac90f46cc2f49f890924e91b223b0c9 : SUCCESS in 1h 11m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/89b5b7717ea74ae89ef60a04ec43ef51 : SUCCESS in 1h 32m 25s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3fcc915b21bc41c49fcf4a2f5652699c : SUCCESS in 1h 33m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93562b37d7694fb7b7248bdc8a9e5898 : SUCCESS in 1h 02m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/04787018bb1a4ab29f306dbd27b016b1 : SUCCESS in 1h 02m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e626850faea247a6ab2eb0d77fad933d : SUCCESS in 31m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/944f0d61fd33494ca17b03d7423af272 : SUCCESS in 1h 21m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/93ae01f72c354234b1acb12b70bcc0c8 : SUCCESS in 1h 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b82f63aebe52dcf901a32ec7ccf430f5d77d5a69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-29 13:45:15.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/8/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/8/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"8a645506b74aba6eb0bc3890ab309cee56d71553","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-01 06:59:37.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nWdyt about adding upgrade check script which would check if there are not sanitized mac addresses in the DB and then display warning to user that sanitize script should be used?","accounts_in_message":[],"_revision_number":8},{"id":"9db2c50df8a228455c91ebb145dea5627c26209d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-01 09:22:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"834f42c71654b8f2ea21bdf87dd2f99214f1e3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-01 09:35:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d07f555609943757381df26b132d4506ede95594","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-01 09:54:11.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 8: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Wdyt about adding upgrade check script which would check if there are not sanitized mac addresses in the DB and then display warning to user that sanitize script should be used?\n\nSorry, I don\u0027t understand what are you proposing. We have the sanitize script for the DB. How can I know if user has executed it or not?","accounts_in_message":[],"_revision_number":8},{"id":"7622d891988045778af390d67219b181f2aff85a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 10:51:55.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/31/789831/9/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/31/789831/9/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"32e0792ed28d6bbd9e90665cb1e648660db38cd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 10:52:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4ad76b9826c4518b956425ef6ab090f : SUCCESS in 39m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c480994d6f14456a435f60b07c39542 : SUCCESS in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"68ddbbbfcaf780862859b8678fc2748711b27e84","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-01 11:15:50.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/31/789831/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"fcfee85aa449bb0d90cb1244219024a596ee5a6d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-01 11:26:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/9 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":9},{"id":"e7b700aeeb86453168745abd3748c326282b2711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:59:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/80c65093a7ac4bccb3b1ba0752e8a7d3 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/21544de7c36442d9a25b9554062db4a5 : SUCCESS in 34m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4168367646904e9f84c07e3ff241f9a7 : SUCCESS in 1h 32m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6999840f22b34316807cca83d7ca3da2 : SUCCESS in 1h 56m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b61559bf569a43348122d10ae5f0f828 : SUCCESS in 1h 19m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d78bc5cbbf864d008ac252ebff72b46d : SUCCESS in 1h 47m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02d452199fd643e1a7008bea8fefc61f : SUCCESS in 36m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a5290d95da4d688538432cc5c74da5 : SUCCESS in 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b2f86897dff45e39142e52ed1ac2151 : SUCCESS in 39m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b8ad8a78d2468aa0ba1d39d21cd65b : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e10aaf125674a4cb25222731e230a3f : SUCCESS in 29m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4427f93cf3c44ce18e6bcd934b5bdfe6 : SUCCESS in 28m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c7174f98b6345dc8087133f6476f9e5 : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c80b80d6874065a5ba7a048b209cca : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f2a2bc474aa4761ab113d5b0bb5b764 : SUCCESS in 11m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ef157c19f864d3eaa4d7e3823c519fa : SUCCESS in 51m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ea334d41efa471bac3733b4f0872b17 : SUCCESS in 1h 21m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c5924092969d432e8f5cdbd647c0cec9 : SUCCESS in 1h 05m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/667ed7c9c7e1417ab893705b5e9ef8f1 : SUCCESS in 59m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8ef28c2974b4eaa957a0b11518dfef8 : SUCCESS in 1h 03m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1846619062b84fb1953f628d44225bfd : SUCCESS in 1h 12m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1cbdcdf8d614ea183fac00294cef345 : SUCCESS in 1h 14m 48s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2400ddf4f4bf4ce2ac89b03f1d4e8ff8 : SUCCESS in 1h 36m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b30c95757b4d96af0aca4385fcc01f : SUCCESS in 52m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6f768c705824e3ab81ad36be18133e5 : SUCCESS in 48m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a499637ebbe5402d8cb2abfd570ce3f2 : SUCCESS in 32m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/949ebc450eab4995a50c112167fee7df : SUCCESS in 1h 08m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7e62607270a449ceabbac87b7d46945c : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"666ba975f90fb15474705e8daa6b6788af39ee82","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-06 06:22:09.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 8:\n\u003e \n\u003e \u003e Patch Set 8: Code-Review-1\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e Wdyt about adding upgrade check script which would check if there are not sanitized mac addresses in the DB and then display warning to user that sanitize script should be used?\n\u003e \n\u003e Sorry, I don\u0027t understand what are you proposing. We have the sanitize script for the DB. How can I know if user has executed it or not?\n\nWe have this neutron-status command https://docs.openstack.org/neutron/latest/cli/neutron-status.html\nYou can add script (check) like are in https://github.com/openstack/neutron/blob/master/neutron/cmd/upgrade_checks/checks.py.\nNow command \"neutron-status upgrade check\" is intended to be run by admin when there is already new neutron code in place but services were not restarted yet. So such new upgrade check could look into db and check if there are not sanitized addresses there. If yes, it can display warning \"please run sanitize mac address script...\". Does it makes sense?","accounts_in_message":[],"_revision_number":9},{"id":"8ad57fcecf84c710e3cf99eab5ea7046343599ae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-08 16:09:27.000000000","message":"Patch Set 9: Workflow-1\n\n\u003e Patch Set 9:\n\u003e \n\u003e \u003e Patch Set 8:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 8: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e (1 comment)\n\u003e \u003e \u003e \n\u003e \u003e \u003e Wdyt about adding upgrade check script which would check if there are not sanitized mac addresses in the DB and then display warning to user that sanitize script should be used?\n\u003e \u003e \n\u003e \u003e Sorry, I don\u0027t understand what are you proposing. We have the sanitize script for the DB. How can I know if user has executed it or not?\n\u003e \n\u003e We have this neutron-status command https://docs.openstack.org/neutron/latest/cli/neutron-status.html\n\u003e You can add script (check) like are in https://github.com/openstack/neutron/blob/master/neutron/cmd/upgrade_checks/checks.py.\n\u003e Now command \"neutron-status upgrade check\" is intended to be run by admin when there is already new neutron code in place but services were not restarted yet. So such new upgrade check could look into db and check if there are not sanitized addresses there. If yes, it can display warning \"please run sanitize mac address script...\". Does it makes sense?\n\nRight, this script should be executed first, before the \"sanitize_port_mac_address\" one. In case of warning, the sanitize script should be executed.","accounts_in_message":[],"_revision_number":9},{"id":"4695729267d347651bd8946c94d19a9ca4995677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-08 16:48:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"af07277f9c0082c17ef1ae967e6d4cce73867acc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 17:35:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a217a8405a04b96a08822f95ef9f2d8 : SUCCESS in 40m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43e987952b294d3293e2e5889d175478 : SUCCESS in 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c9f0c1c3f867da2fbfa9d33fa43067bd86748164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 19:27:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7cc7f2b69010433982957468543fde05 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cfd50e8cb6e343b2bbfdeacb8d1a3901 : SUCCESS in 41m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/979909f2f10f4c51b44484c99156577a : SUCCESS in 2h 05m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/46123b50741b4391860910e15dafcc87 : SUCCESS in 1h 48m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26d568f4caa84bd39a63938665a0869a : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ea32cbea6712475891705382e2a3d490 : SUCCESS in 2h 28m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e94422f98004d85a7ef874bd4d04229 : SUCCESS in 33m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47a3ed7f981a44518f981a1222ff5437 : SUCCESS in 52m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f415a5cbac945e6ab078e4c8b8047b2 : SUCCESS in 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01556c0387b84fe4aac1dc44f712f82a : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09617839d667418896c767d7c94d2c33 : SUCCESS in 27m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80cbba7e0aa64c6990be9fa06b0380f6 : SUCCESS in 25m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eecf000f269c40678acb3ae81abfffbf : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc7d0c5b5244d4ea312ce3bce376868 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995ac16624534b38861810b60393b8d7 : SUCCESS in 11m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91a4e9ceda374cb0bf1882e454d714b2 : SUCCESS in 45m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2da18ab9ddf9435a94119c1a46c5c5b2 : SUCCESS in 1h 21m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/14140253a6944aa9842210da31071d08 : SUCCESS in 1h 07m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4907e83329145dfaacb61e85a2ad00b : SUCCESS in 1h 14m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32194492e52f4896986262329468ee7b : SUCCESS in 1h 04m 22s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f7bfa112a38f46deb9a6fc2c3a257ce0 : SUCCESS in 1h 38m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eaa58df0253340d4bdc48eaac148f27f : FAILURE in 1h 06m 15s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6b752be15d45410da2c8626fb9ea1e06 : SUCCESS in 1h 26m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/343df4ad41064671bb7a2883a4248cb9 : SUCCESS in 52m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/50d774ffd691414193ebe861bfbadd52 : SUCCESS in 48m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab541a8536a64d7a8a23b8b0edc7c572 : SUCCESS in 27m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/464e452010ae479ebef57b795679dd99 : SUCCESS in 1h 02m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5a3bcc95d97743d9b59f53408de61d61 : SUCCESS in 1h 06m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b6719438055905b5725b2b0124d50c3e836cfdd9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-08 20:35:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/10 : SUCCESS in 1h 36m 39s","accounts_in_message":[],"_revision_number":10},{"id":"8e59bb15dd140471e326c1ff40f6c140ba655add","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-16 04:10:29.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/31/789831/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"8ce79a29e8a5f2da43706e3900b98b42a614855a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 11:13:05.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5fe45927a70f74f6b27928d765a586d44cd6529b","author":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"date":"2021-07-21 01:40:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"dec5e8af5938dc8dbe3c37107191e47b580e56b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-27 14:55:25.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ce43fcb7bfd07536b56d477424e9b5a0d2dc2cfe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-27 14:55:52.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5542e1bfa60ffe06cf41006f3582a99bd4d870d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 14:58:05.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"9295464937c377c3d7f628c2d5f422ed3afde076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 17:24:07.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac51a82c037940e99c9ad4179f12f7e0 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/72583a988995447bba504cad9e9f2c5f : SUCCESS in 40m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e458c70076324840b2800d313c6314a6 : SUCCESS in 2h 25m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e6e6b28cdf44c91a8a9cf5416206707 : SUCCESS in 1h 39m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/701bb4150a774e06a0c07986aafdcbb0 : SUCCESS in 2h 24m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e94eb3472ba449098a8b541be8be6aa : SUCCESS in 1h 25m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44bd2d2fe64748638d9c4c58797d52ad : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1425f2976b54ccfb86b32cd48127a95 : SUCCESS in 14m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5b681af4e141f1bece4783a2b6d883 : SUCCESS in 24m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde357757ebc47f681e174b3ba6f47d0 : SUCCESS in 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595db395e7704d8b892e0ce23c59c873 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85601f2645de4e9f9d19c2feedc5980a : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50a529b767f549ad8b7a06818563e60b : SUCCESS in 46m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0761c8f3ed9344f8ac36da71cbe9d290 : SUCCESS in 59m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6dcc6d1684324c45875ff36e2d4791bf : SUCCESS in 1h 24m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7129e9680f8e4aa689fd5323de3929e2 : SUCCESS in 1h 02m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a32a0654483e42a48014c219780ab81c : SUCCESS in 1h 23m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab6a52e957054b9da2a830efa87460bf : SUCCESS in 53m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aadda10936b64141905e7ebb4d6030c4 : SUCCESS in 53m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/59c34944db6942f4b964a47747e7819f : SUCCESS in 28m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04dabaa28d214cc4a7aecc5996e7414f : FAILURE in 1h 15m 31s","accounts_in_message":[],"_revision_number":10},{"id":"7fd545f2b18b1c439c50ac881a8ccf7aea79d13d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-28 07:38:43.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"f4ff97010f99afca1f8c8fbe3c6fb4c3f8ea6ca7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:11:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/325ac85c275a42079ed002a557fa237a : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/041d3ae451e54c109bf0ecd22112c609 : SUCCESS in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f6f4510afc3858eee16aa41af2a7ce47390dce3b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-28 09:24:22.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789831/10 : SUCCESS in 1h 45m 07s","accounts_in_message":[],"_revision_number":10},{"id":"ab564dfc1dfd3abf2d519220bfb30b897c106ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:43:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/505c3fe9dcde4c04a9777974d02aaf2a : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfb4173e7b9c4627aca394d31b6b9a3e : SUCCESS in 42m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d9639cb844c4cf1976d5655de05fa1b : SUCCESS in 1h 58m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/46627c71d9e34df186072a50fe599ad6 : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5283f569fcce40b8a7707e58b006cc50 : SUCCESS in 1h 10m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb179de288f6449b90a6e95d081d2f9b : SUCCESS in 2h 00m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b674334beb947bf8ba0d866a6e7402f : SUCCESS in 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bff3ecc48564c929a5468065d99d1d1 : SUCCESS in 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3df6080571c645399a1fca85f5603c16 : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cc341226784e3bbaabffbe583cd7f1 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9be60b2565db4afcb8b8af58d646b585 : SUCCESS in 28m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b35b2a7dbc146629f80167fa70211ad : SUCCESS in 26m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7be07a14f13046c499637a5ccf87bd1b : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc701f832b614bd0a24ef9cf973bdc65 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/393310a2dd8b447cb495011de58fb267 : SUCCESS in 13m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b8a58bcf4614cf0996748d61295ce38 : SUCCESS in 42m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f48fb6d097564c19b75ece9e59fcec1d : SUCCESS in 1h 27m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef0cbaba8cc6462a8441f31d2e3e75e4 : SUCCESS in 1h 04m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3985919fc37148c890298c214f5381ab : SUCCESS in 55m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/91f805c759b54236902069f98e512040 : FAILURE in 1h 28m 38s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa8915150b3f43408cc090b418d8971a : SUCCESS in 1h 02m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/012abe0387e84599b582e77d0bc8f4a7 : FAILURE in 1h 04m 12s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/b922913c242747b78ff4fb69231d9630 : SUCCESS in 1h 22m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50f5ae318c724e3f9bea188d3aa23e4a : SUCCESS in 52m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2eba878c3a9443cbbfd0ebc67f06fa17 : SUCCESS in 48m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dab5c8ec7ff441e1a106effb40d64ef2 : SUCCESS in 42m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19a967b75f5148e0a9e1b6f0d48ea5f1 : SUCCESS in 1h 06m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4758d1180af242c28b1f5900eca1c0a6 : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":10},{"id":"25894d71d4ec111a0e39f271799add37d4da3d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:43:40.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fb8cbda0a2b41c75abfe5e088103c2e7fbce7a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:27:03.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d507b4c51d9b4a388556fa9309241a89 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94be0001e98d462fa56b23336eefd525 : SUCCESS in 41m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9965e179fe1145a3a80032b0de349bcf : SUCCESS in 1h 48m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f36e26cfc288471bb80f512fdafa0fc8 : SUCCESS in 1h 48m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ad670279a3945f6b32122cf68d008e0 : SUCCESS in 2h 11m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62ddc89a325449ffb54fd0ba94e25c69 : SUCCESS in 1h 22m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8917da9aa04b43e6a0597503955dcf68 : SUCCESS in 38m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f32549d59b4fbdae103bb6ddf63ef6 : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb1f88d201e455796d11d003bdff185 : SUCCESS in 26m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee5403366d740b481f0b8971df43193 : SUCCESS in 29m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed421602d3c4c1588806b93b7e24756 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b92367d6e7324fa1a8d3d61a532d98cb : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7e7264d9bf94fdf8cb46232b23f5dde : SUCCESS in 49m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/37eff68470b44b09af08d185288bb44e : SUCCESS in 1h 24m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/47aeef632fb543ca997dce20aca4da3d : SUCCESS in 1h 12m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/405424e8f3ee4580ab0dee7b4b2152d3 : SUCCESS in 1h 08m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/1b8ae311d91b40ae96fa1af26364991d : SUCCESS in 1h 33m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f264d03820774b409292897cf4c85f39 : SUCCESS in 54m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b776e65358da49fbaa57af80bd40fabd : SUCCESS in 1h 05m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9d970d66cc4d465cbc2e91fbb3be56fd : SUCCESS in 35m 40s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/283b77a7341d49f79f065b3b47211672 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":10},{"id":"9dd6d1ff6ce4cb5b031c6f05f667c2be43a99e31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:27:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"c1c904542e2194cd72304bce37792d73ef1f60d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:29:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ee7363f2fb44479982777d2f12932a : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a83731245d34d4682aad0a77ef2e302 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"827cca2ed75b159c3a1a2e5f193150fe32491b1b","revisions":{"301270c2d2226949fbdec6498dcdac014faad7f9":{"kind":"REWORK","_number":1,"created":"2021-05-05 13:30:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/1"}}},"commit":{"parents":[{"commit":"ce22171b05d554b2072211a6ed6df9965a8f1d63","subject":"Merge \"designate: allow PTR zone creation to fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce22171b05d554b2072211a6ed6df9965a8f1d63"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 13:24:20.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/301270c2d2226949fbdec6498dcdac014faad7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/301270c2d2226949fbdec6498dcdac014faad7f9"}]},"branch":"refs/heads/master"},"19773d1ea29b18e0f6ce2c44187800d3ad9858e8":{"kind":"REWORK","_number":2,"created":"2021-05-06 07:32:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/2"}}},"commit":{"parents":[{"commit":"ce22171b05d554b2072211a6ed6df9965a8f1d63","subject":"Merge \"designate: allow PTR zone creation to fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce22171b05d554b2072211a6ed6df9965a8f1d63"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:49:37.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19773d1ea29b18e0f6ce2c44187800d3ad9858e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19773d1ea29b18e0f6ce2c44187800d3ad9858e8"}]},"branch":"refs/heads/master"},"3878583494505bc10ffe83b5f9ad11c419137a76":{"kind":"REWORK","_number":3,"created":"2021-05-07 07:29:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/3"}}},"commit":{"parents":[{"commit":"ce22171b05d554b2072211a6ed6df9965a8f1d63","subject":"Merge \"designate: allow PTR zone creation to fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce22171b05d554b2072211a6ed6df9965a8f1d63"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 07:28:40.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3878583494505bc10ffe83b5f9ad11c419137a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3878583494505bc10ffe83b5f9ad11c419137a76"}]},"branch":"refs/heads/master"},"785e21286d30c0cffb3c0cc135f6392a38568db5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-11 09:21:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/4"}}},"commit":{"parents":[{"commit":"389584c389f7cb7a13e75587dae558658e1ccb1a","subject":"Merge \"[Doc] Remove one of the known DVR HA limitations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389584c389f7cb7a13e75587dae558658e1ccb1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-06-11 09:21:19.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/785e21286d30c0cffb3c0cc135f6392a38568db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/785e21286d30c0cffb3c0cc135f6392a38568db5"}]},"branch":"refs/heads/master","description":"Rebase"},"3a6b2f8f8f9275f69a7afc026ca069aa30f6f3db":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-18 06:36:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/5"}}},"commit":{"parents":[{"commit":"b08310df9f9061cfdcbc1d4428739db6de3e0e7f","subject":"Merge \"Add localport const and refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08310df9f9061cfdcbc1d4428739db6de3e0e7f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-06-18 06:36:01.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6b2f8f8f9275f69a7afc026ca069aa30f6f3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6b2f8f8f9275f69a7afc026ca069aa30f6f3db"}]},"branch":"refs/heads/master","description":"Rebase"},"bdc59ae6f1040a3246ce5ba3de6b54b98d62bb29":{"kind":"REWORK","_number":6,"created":"2021-06-18 09:21:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/6"}}},"commit":{"parents":[{"commit":"b08310df9f9061cfdcbc1d4428739db6de3e0e7f","subject":"Merge \"Add localport const and refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08310df9f9061cfdcbc1d4428739db6de3e0e7f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-18 09:21:11.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdc59ae6f1040a3246ce5ba3de6b54b98d62bb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdc59ae6f1040a3246ce5ba3de6b54b98d62bb29"}]},"branch":"refs/heads/master"},"83b73c80a94dcf28a526335ed912fac356137f0c":{"kind":"REWORK","_number":7,"created":"2021-06-23 07:41:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/7"}}},"commit":{"parents":[{"commit":"ba97a73783a33ca38702687795009bed96fa1990","subject":"Merge \"Fix devstack path in plugin.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba97a73783a33ca38702687795009bed96fa1990"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-23 07:39:52.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83b73c80a94dcf28a526335ed912fac356137f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83b73c80a94dcf28a526335ed912fac356137f0c"}]},"branch":"refs/heads/master"},"45c99600c2a32bb8211e50263cfd82ea62af1bc9":{"kind":"REWORK","_number":8,"created":"2021-06-29 11:27:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/8"}}},"commit":{"parents":[{"commit":"6196c0873b4af9abd9055bf92478ff05ae090104","subject":"Merge \"Refactor DHCP common config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6196c0873b4af9abd9055bf92478ff05ae090104"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-29 11:27:37.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45c99600c2a32bb8211e50263cfd82ea62af1bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45c99600c2a32bb8211e50263cfd82ea62af1bc9"}]},"branch":"refs/heads/master"},"7dc0ceb1f9801b8b2adc2dabdbc813f268c27f4a":{"kind":"REWORK","_number":9,"created":"2021-07-01 09:35:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/9"}}},"commit":{"parents":[{"commit":"6196c0873b4af9abd9055bf92478ff05ae090104","subject":"Merge \"Refactor DHCP common config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6196c0873b4af9abd9055bf92478ff05ae090104"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-01 09:35:34.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dc0ceb1f9801b8b2adc2dabdbc813f268c27f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dc0ceb1f9801b8b2adc2dabdbc813f268c27f4a"}]},"branch":"refs/heads/master"},"827cca2ed75b159c3a1a2e5f193150fe32491b1b":{"kind":"REWORK","_number":10,"created":"2021-07-08 16:48:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/789831/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/789831/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/789831/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/789831/10"}}},"commit":{"parents":[{"commit":"6196c0873b4af9abd9055bf92478ff05ae090104","subject":"Merge \"Refactor DHCP common config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6196c0873b4af9abd9055bf92478ff05ae090104"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 10:45:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-08 16:46:55.000000000","tz":0},"subject":"Sanitize MAC addresses","message":"Sanitize MAC addresses\n\nThis patch sanitizes the MAC address coming from a user input:\n- The \"base_mac\" address configuration parameter.\n- The \"port.mac_address\" stored in the database, if the script\n  provided is not executed.\n\nThis patch relays on [1], that will sanitize any input coming from\nthe server API.\n\nThis patch adds a new script to sanitize all \"port.mac_address\"\nregisters stored in the dabatabase.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/788300\n\nRelated-Bug: #1926273\n\nChange-Id: I8572906cc435feda1f82263fd94dda47fc1526e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/827cca2ed75b159c3a1a2e5f193150fe32491b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/827cca2ed75b159c3a1a2e5f193150fe32491b1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
