)]}'
{"id":"openstack%2Fneutron~789929","triplet_id":"openstack%2Fneutron~master~I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0","project":"openstack/neutron","branch":"master","topic":"bug/1925368","hashtags":[],"change_id":"I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0","subject":"Check router routes connectivity when GW port is updated","status":"MERGED","created":"2021-05-05 17:27:33.000000000","updated":"2021-07-09 16:01:54.000000000","submitted":"2021-07-09 15:59:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"789929-bug/1925368","meta_rev_id":"6467ef0004ac2d924dfaebceed4705464705b32d","_number":789929,"virtual_id_number":789929,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-09 15:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-05 06:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-07-09 13:49:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-07-09 13:49:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 17:48:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 18:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 19:17:55.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-05 19:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 07:21:20.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-05-06 15:06:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 20:36:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 14:17:11.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2021-06-15 08:08:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-09 13:49:12.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"ebf73b7566508a717c6ea0addd5922e9b458544e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 17:27:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f84e34e1a5234a51e19bb975f564f64a4c5dd2d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-05 17:27:44.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"626019223c8a263bf9754b63175cc51337f9758a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 17:48:35.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/789929/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/789929/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"251b578e5a0beb757182787e33f443ab09559608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 18:18:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbaeaa94545540d8a268442e45801ea5 : SUCCESS in 48m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca8008e4ef894615ad94670f587f461f : SUCCESS in 48m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68d678990b7354be7060ce9a2de182b9456ef647","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-05 19:17:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/1 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8113e248d8fa07de952d895f890d861ae210360e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:36:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f3cf9aa4bf2044919b0a3758e95eb480 : SUCCESS in 38m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8dcb230857f64b78b59995b0977fcf5e : FAILURE in 1h 50m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6b708bac41d47ab863146b78d5771c1 : FAILURE in 1h 48m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a01c02b6e14a401f9ff452467855d729 : FAILURE in 1h 13m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e8f67e96be2e4584bfcdc5b582a60f01 : FAILURE in 2h 07m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0e42b18bd844cdcbed51603e67f0518 : SUCCESS in 40m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4832018e98084de1be997f4c7888874e : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67aa8355a2a8476ba821c23e20c77f81 : SUCCESS in 13m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7af960e2b834950b8bc6686b86fa2b7 : SUCCESS in 27m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4eafd3b40a64e9aac834c851623010a : SUCCESS in 29m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cea54f9689864e858e8e21c93ab68bb4 : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1a86aa2ec44922a641902c0fda8431 : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4718ba7f0514d3396555f377ec0f281 : SUCCESS in 58m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/90e7b76f66694c6b8a5d9a778e9ce46b : SUCCESS in 1h 27m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fb0d8c87984047debdaf7ad24c474853 : SUCCESS in 1h 32m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4db47b27aa04250bffaf1be592c1164 : SUCCESS in 1h 15m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bc88046da36444f3b0484f8a38529fdf : SUCCESS in 1h 10m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e52af5676fa940d4a08451f7300518eb : SUCCESS in 1h 26m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3d195c5d6794730adb48ca465a59905 : SUCCESS in 1h 38m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50c91c8dda2047528ce8fedb3ed24467 : SUCCESS in 1h 40m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5118a87bd11642ff978088bc896ba215 : SUCCESS in 1h 10m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/18e1c38a5de54c8ea9a7b2c73c7466a8 : SUCCESS in 1h 12m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/322993a73a1e4327ad5a29656f543591 : SUCCESS in 37m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8e62b64a0c8749af9ea302e98b0a787c : SUCCESS in 1h 44m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ba4fd3bcbfc8483a89b53f24229f223d : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0f850218b15c5625080d672d3f802e5b0f7a4ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 09:36:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8789772da5a7dc04d988ac91117410bb7ba2d25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:29:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acfcbf5c3bf64b8d90fb137bcb529207 : FAILURE in 48m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3087030f9b24c6cb08603fd2a694ba9 : FAILURE in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80d95b3dab918e13b7c815d9fa207bbae746d7ff","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 11:25:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/2 : SUCCESS in 1h 48m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f73858bff4dabc641ef053c720b45fac96870b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:07:57.000000000","message":"Patch Set 2: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7888465b87a1433eb0554e4b09e88d21 : SUCCESS in 49m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/890a9131d5994358bac4a89e908c6c85 : FAILURE in 2h 24m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7aad10dee0bd421f9f0f0a00ed9a41ab : FAILURE in 1h 47m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f08d43458a94466a0ccb9eb0afdb563 : FAILURE in 1h 46m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6b7da2dab9f84122b776ff266cee633e : FAILURE in 1h 54m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed909ba7460458cafd580be533e18e2 : FAILURE in 37m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87ece4a06fea49e1bf9ad7b914f448bd : FAILURE in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e5a0028b414210921110a3ffad466f : FAILURE in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/923ba59080d9407abe779f331fba5bda : FAILURE in 24m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0f9136d6d974a0789f11b0ba78e7a9b : FAILURE in 27m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9ab599a9f324b24a65c1339855bae7f : FAILURE in 18m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7094177e90488184ff96ef226bc38c : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dacada75c69d474e9a6102dfe877ed65 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13a7eadf83044f4cb2b8523c44ab28ec : SUCCESS in 1h 00m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a37c707fe4246c1bedc45308a9a267e : TIMED_OUT in 2h 14m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a19fad0820ee44b29666e0cf052159f3 : SUCCESS in 1h 36m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ccf8c3eaf7d4369b7833405ed2c0578 : SUCCESS in 1h 06m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9756dd79eb5b445cb3ed923aa85e0af7 : SUCCESS in 1h 07m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4e1cc1dd31e044828e0bd3f6ffb7fe82 : SUCCESS in 1h 18m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff821d3f717e4c45a50d361fe01c71d0 : FAILURE in 1h 11m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cfbd4f9bf1a54eb6b1ab042f16f5c2c7 : SUCCESS in 1h 43m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86e29abdccb746738afe2d974fa940e9 : SUCCESS in 52m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/505ea805e4b6405cb7f08bbb68e9b2b7 : SUCCESS in 56m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9b0ea12f8d76457c8d2a19fa16bac88c : SUCCESS in 36m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/541c389a6e964ecabd5d403545dd882d : SUCCESS in 1h 36m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/69e76d2e8d5d490a8ad9862c88238eb0 : SUCCESS in 1h 22m 34s","accounts_in_message":[],"_revision_number":2},{"id":"18a9aca3fedf85b173c509af188bfeb6b11d2971","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-06 15:06:23.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/29/789929/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0afc5813550548d59acfb021fda692848870fe47","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-06 20:36:57.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/29/789929/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9c48c81b8289eb0b2650c1d53766063fea57d844","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-19 14:17:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8548fd3037fe4264e415a52bae3a53d6db958851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-20 17:25:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ca39c70793ab221c0b47cc855bb0ba1b323bb377","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:06:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba10e1f5db994f6b93995a1dd6a9489e : FAILURE in 34m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0ec58febe864bc6a74774edcc02b2ac : FAILURE in 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f0e5f69e6f8b6a15b34ed6da5e31efb937963827","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 19:20:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/3 : SUCCESS in 1h 44m 10s","accounts_in_message":[],"_revision_number":3},{"id":"7e882f3db19dbf910ecc7ff62776788215b5117d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 19:35:00.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df8dca5ec6314ed0843c2f9590fe2e1c : SUCCESS in 40m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61b802eb0ffd4e6da765c9242ed32d6b : FAILURE in 1h 45m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c0e26a8eb6744b409e64042e8de6dac7 : FAILURE in 1h 37m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1886c75a3eb24244bba99a0dfaa4a966 : FAILURE in 1h 15m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e768f02af6f54614aa42de6c22ecad32 : SUCCESS in 2h 08m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd4926ce30941b49d75f090312b7975 : FAILURE in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93d2344ae59a46db8b00be84f9db01bd : FAILURE in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bc0dd426334fdb87b434d088c94345 : FAILURE in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e819ad9d8e2f4b74949d0f2f0bf528d9 : FAILURE in 25m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5ffca885e384426bbe6432e95c2e7f4 : FAILURE in 23m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08a5a465aeac4f8290ed37b9811c86c3 : FAILURE in 23m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa0f41bb29e41579bad093a2641cf49 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6caae3ad3ef46dabd79a11c90515228 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/775faed374a94bcdba252e010db6e14b : FAILURE in 38m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e09675813b0547828e08d440dc4f4e57 : SUCCESS in 1h 19m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7d92248fe5f24496a777bb78b174b53a : SUCCESS in 58m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e07b0779fcbe41a29d295a332abd3700 : SUCCESS in 58m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d4df6c841bbd472d80f27a6a546d5f09 : SUCCESS in 1h 02m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/74dc5079bc1241c8ab20686c688eddd5 : SUCCESS in 1h 22m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f004f78a797433bbb81f8bd2019ee40 : SUCCESS in 1h 10m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b6220b977c64c689f6587c9f3100848 : SUCCESS in 1h 35m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bc2b168bb034f8e97e5957b40c7cc25 : SUCCESS in 39m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/02f54ffe40ec4aa1acd55aab37bc345f : SUCCESS in 57m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f99dc5c95b443da5b904e33ddcc4ea : SUCCESS in 28m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a521f2d49d98428d95beb67535a9a24b : SUCCESS in 1h 35m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9e0891b001aa4f6d974e7e09c0819afa : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":3},{"id":"410c291b80182d4d4b8b09863fe97a62d619eb36","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 08:35:39.000000000","message":"Abandoned\n\nThe use case tested in the testcase is valid","accounts_in_message":[],"_revision_number":3},{"id":"afc384f325201a91546f6086276adba008c272bc","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 08:44:15.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"87cd7171868422f6909aab6970c88e2cb8b1fe41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 09:22:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f785b0455ba35507d87fab8a31a7db9d3ff0f7a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:12:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bb98527f05f4a69b0969f8afec94523 : SUCCESS in 45m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/536d1aadcb27445ebe34a26ee4293540 : SUCCESS in 48m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a396764f6f0609a12dab9e2848d883ab7318ed74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:04:09.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ed918f2bfd1249cc82f537054e1a0b94 : SUCCESS in 46m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6c9ee559ec864f21a64335bebbadc1d3 : FAILURE in 1h 56m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/901a2d51949a4bb08969eaf1d5261f49 : FAILURE in 2h 23m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9f54d5cda824ddf83c1182b7af8ac3b : FAILURE in 1h 18m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cfb9b2dc99b2464abdd7206f28d11727 : SUCCESS in 2h 39m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb645e85b7064d1283496a82a64452f4 : SUCCESS in 44m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06cba0d2c2364825a38f484c28c90bea : SUCCESS in 44m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75bf015116ae4a0faff19d6d9d01d09f : FAILURE in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a6544fcdce465b98aad0e35c6a458c : SUCCESS in 28m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c89a389744174a2383d92cbc3b53713f : SUCCESS in 27m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f56f0325774920850d1b23e127d5c3 : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77cfdca13be843ce95237287e2511035 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7637c301a974f5a921855784fc7f948 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64f7fd4dea96478c84fa20badde18712 : SUCCESS in 48m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f35e053cfbe64bc1aa63ed70e8c94d91 : SUCCESS in 1h 20m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cab4235f078a4e66988749a23d0d644d : SUCCESS in 1h 22m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e752fdaddb814da99a69b706acf04c6c : SUCCESS in 54m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9d01b9b6d2d49439b588ee4ff608a4a : SUCCESS in 55m 28s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ded8d15883e4863b792e4497f8dfe69 : SUCCESS in 1h 01m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2cc26d34ac0b47a58935c51459183e34 : FAILURE in 21m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ca6482d90d643fb9b9fd373f45e72e5 : FAILURE in 22m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f74160675ddd4258bb9147ee38d53c1f : SUCCESS in 52m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/addbfd6ef1f849efa935b0995fe0c184 : SUCCESS in 1h 00m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2647c0784d6d4b778592b473d2be4230 : SUCCESS in 28m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44aa72b797d84cf4b04dacc915f2277f : SUCCESS in 1h 11m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c8e6371c05ff40ffbc959f12c87534b0 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7494e933088fa8823c24cf013d38c59ee9fcf65b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 12:52:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/4 : SUCCESS in 2h 19m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7369377cfc134e4620c379bd8895c2c9ba99093d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-28 21:09:51.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/29/789929/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"566decd7cd6c43ed78265aab86c2e3f70e7230cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 15:55:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"40a6dd80f74b516415de073b2d08b16809607caa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 17:14:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b72f40e23a2846acad6d15765d0768bc : TIMED_OUT in 1h 01m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8d715dabf304341bcae89c8f19090c5 : TIMED_OUT in 1h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"10b2b4dd42f18f6007e163a99550401e64f1ac45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:45:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/141a45cc65ff48be82dd6c4effec1b54 : SUCCESS in 43m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/de7df54157f04f59b024dfb8ea9c5129 : FAILURE in 1h 48m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6f861ed2637a4a549ca2b8b8017955e9 : FAILURE in 2h 03m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8eaab6137b0f466f9893dcce24ccc0d6 : FAILURE in 1h 29m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/09914df971f64d6a811952a1ecd1cd4f : SUCCESS in 2h 17m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f6a8b6352247b49f21ed9537bf6b90 : SUCCESS in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8110dd5a62ab423d9fce8007b9d804d7 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f599ec779e0e4b548b69b24414f3c548 : FAILURE in 15m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edd7b656c7c2451f9a0515cfddd975fa : SUCCESS in 30m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8817dab855244248b442173f5d1893d1 : SUCCESS in 29m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/694dc50f958f4a6e80906a86820f78f2 : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bffa5d8ec8467d9c942fe228f41ee1 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/753c46b2050347db966006aafbfeec70 : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe7cdc51eb4641d68bb4e4274fbb34dc : SUCCESS in 53m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c8a602f229449829286c2ddde8976cc : SUCCESS in 1h 41m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4b0e1c1a584747bbb1e7dd975a7ded37 : SUCCESS in 1h 28m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/776a5ce514854866b18a19f777db714b : SUCCESS in 1h 01m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58b0ac0b4446459eab789bbe578fdea9 : SUCCESS in 1h 03m 30s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9826e17ee41e46a8bfcd5dd886e88ce3 : SUCCESS in 1h 20m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9a4f66788e345bb985d97cc6d2f5ca4 : FAILURE in 25m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29f90c35ec7140888d398e0a71f89a03 : SUCCESS in 1h 38m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee0c41ae5118439d88c417583d10e88f : SUCCESS in 1h 05m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d13b74f6f98d4b00be7cfa0982c19347 : SUCCESS in 1h 11m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5e5ca3eb550b48e58d1fa0c4c1f4a5bb : SUCCESS in 36m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43a68bb1789f4dcd885a51b11ac3ff2d : SUCCESS in 1h 24m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8eb5110d8f714fea951cf7e4f38f78e1 : FAILURE in 2h 03m 41s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":5},{"id":"bde72f5544c6497567946fdae1e779a903e47ff9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 18:49:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/5 : SUCCESS in 1h 43m 17s","accounts_in_message":[],"_revision_number":5},{"id":"bb9245a29665d8ebd143c54a96305439c5acd131","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 23:10:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2ffd9fd1171ea2dd517fd5f01af84c9a6fa831d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:12:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7495266cd23c4b1bb329ad9391dd8a64 : SUCCESS in 34m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1388f7023ec84118b2995b01fe5539f0 : SUCCESS in 40m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39164e0700fb413bb3631b17902652c8 : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06266c3aaf845e9a5976b9dedcfe329 : SUCCESS in 34m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a1174a4a3e1470dbb30cdbd1ccb949b : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a549a4a941e4f62b700252fde68810f : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26aa0473ccf49259854a92f7da31037 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d46b95df215244d1997a32eeca6f8927 : SUCCESS in 11m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09609223eb4c4f2e86964581e2a1134b : SUCCESS in 55m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a506776d37d04389ab3c6964ff1af2e6 : SUCCESS in 1h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"df3a40adc0a178eee8fdc72f6f1194697c749874","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:15:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e344eaea338545e78349686937bf755b : SUCCESS in 44m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2ddb2a42f9c4edc8a0ed40cb11e22bf : SUCCESS in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"134dca026d99fdd12177356f4fd0f9038d28f75e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 03:07:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/5 : SUCCESS in 1h 38m 35s","accounts_in_message":[],"_revision_number":5},{"id":"372613f7a321c1039c660d7a869f348c31a6e28d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:20:59.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"92c517f8437ecac5aa9b2774238ad4dfe411f70c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-09 17:26:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3754d3c2d86115cea990e5b718c17acf0ed3f586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:09:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da15de23da4a4dfeb5b3daee58a08f23 : SUCCESS in 40m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a7972fc7b234cf99c3aa4924f43d834 : SUCCESS in 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b4e4952a6eaa2ac5ea702079a9cfcd34cdd871c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-09 19:01:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/6 : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":6},{"id":"11cfb462d090695786b3a107ab177aac6182e352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 19:15:26.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74ef13ae1c354313ac869b9d54ee6c0e : SUCCESS in 41m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a2518ec0b0104e3ea5f290d167e8f3ea : FAILURE in 1h 46m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c9be7a0c34ab4493b4ac2b7fdb4861b4 : FAILURE in 1h 43m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fdefae12c8d4bffab63d5d650cbd92b : FAILURE in 1h 16m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/86b43132894d422e82af964e0ac7f8d0 : SUCCESS in 1h 28m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd14fb0f2f034d70a6dae1ee556fdc1b : SUCCESS in 31m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4daedc9e3367474fafe74378273b0d09 : SUCCESS in 25m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79906575f00a45a3b606e832f260cdb5 : FAILURE in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aac8fe142bb40ce8eb905546a4342d1 : SUCCESS in 25m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec8b570cfe424e0d91af2046f995c33a : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80612f6a10434bc09f4970eaa6f9ab7d : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45bd3a0d5ce64594a66bc9812e1e1f36 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/191e8486e9714b6b88b37266873fa3ba : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d463ac212e0472d8061956bfd4546af : SUCCESS in 43m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/849fc3ce78d0497dbda033c73fea449e : SUCCESS in 1h 22m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/117ae6dbc8a34cf39b59b8f057edc49c : SUCCESS in 1h 16m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/622c4f64344942cdb1a6dbf372a9f257 : SUCCESS in 57m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dcdceb3f3b464b0dabc95e35ede54ac3 : SUCCESS in 1h 01m 04s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/047ec1ef896b464897437daa9077edb1 : SUCCESS in 1h 11m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1dddec48943f4584ad0cec5ece2cb35f : SUCCESS in 1h 19m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/946650cfd0f0476ea26381c2b76c32df : FAILURE in 1h 24m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/437a239f4db843969579eacb2de4b573 : SUCCESS in 55m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de2cdc7ac86342a3b27ad778927a48f5 : SUCCESS in 55m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9a718e78d6442c3b7f4fb25d938e203 : SUCCESS in 31m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cb81ed5cd83422a8310f49b74b03e10 : SUCCESS in 1h 20m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/255cd7e03cb74f3c92ff894064513ba2 : SUCCESS in 1h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e5a0ee35a5e153a80696fa162558e22e71bb7b79","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 08:08:36.000000000","message":"Patch Set 6: Code-Review-1\n\nIt seems that neutron-tempest-plugin scenario jobs\u0027 failures are related to this change. Please check it.","accounts_in_message":[],"_revision_number":6},{"id":"3cb1f6281e0b4f9092f283ac89b03f19a866d561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 11:54:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8157a96839cc0c15a8ad986653d57133b32236c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:37:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/593326e19590406ebab5bd8c9de00cba : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/668a2e9b864445b28ee33a2803c8e95c : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"01ba5e880aa9d023a4ca5a86ebc35c4bb2aff69b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 13:23:49.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 6: Code-Review-1\n\u003e \n\u003e It seems that neutron-tempest-plugin scenario jobs\u0027 failures are related to this change. Please check it.\n\nI found the error: https://review.opendev.org/c/openstack/neutron/+/789929/6..7/neutron/db/l3_db.py#376","accounts_in_message":[],"_revision_number":7},{"id":"73c57d2d6f190685751888b4bbb6c92e0b42617b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 13:54:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3bb182fe450949439a348bbeb9aeedac : SUCCESS in 39m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5f80599ca04b4793a2b737c444809185 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e46b1edbdd294b6896b7627a8c92af18 : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ffaee22dce942b69fea5b69c59fe4f2 : SUCCESS in 1h 29m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0f4870651cd14440af2b33afc886819b : SUCCESS in 1h 42m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93f256d05cfa4c0fa609f69d8b172d28 : SUCCESS in 28m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb7a245910a54398840b74c163ecc2d6 : SUCCESS in 32m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f16e2081434f2780ea34500847020e : SUCCESS in 36m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aad280d6c314f73be4a67c70758a23f : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b684f9cd1d479683406e708011db64 : SUCCESS in 27m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66003502242347bdac2590f2d46168ef : SUCCESS in 27m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d937408a9d4944b79a408b4d1407a8b4 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1d0e6819da41a0acc12b2f7cb6ce5d : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c008452949d479ea840c55980b1cfa5 : SUCCESS in 12m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6762a9338df43e5a71a9645ca55e027 : FAILURE in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e3d0b661e4e45a49b86ebf9716ef650 : SUCCESS in 1h 27m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9a7f78be5fb442529211d1ad99ddfe32 : SUCCESS in 1h 11m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f3e48c2e99b46e082cb43ac3b170c5c : SUCCESS in 58m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f31e774534f8498fb7a7daa8b9e8be1f : SUCCESS in 1h 01m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/97761fe0176741aca07d8278fa43af36 : SUCCESS in 1h 30m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7553c1066a1c4c5b84d718dc3d5f7be2 : FAILURE in 1h 09m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3748abcc217543aaa05fdbc648a816ca : SUCCESS in 1h 34m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcf8fb2ddb524562b17c2e1ac0b66260 : SUCCESS in 58m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/354ab8e79323462c82cdeb70a6ce3a8e : SUCCESS in 57m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c815f582eba74c929b29b3065c1ad311 : SUCCESS in 33m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f4812119cae4a63afb6da80d4dd4f28 : SUCCESS in 1h 14m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/76acccc14a754960807d0d7e859bdc2a : SUCCESS in 1h 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"512185cecfb30b4c553d4e6afdf5a2435d1157cb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 13:56:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3d4da9f844d390891ecd93efdf905b6efeaff787","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 14:45:21.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/789929/7 : FAILURE in 1h 32m 00s","accounts_in_message":[],"_revision_number":7},{"id":"9d9fe4c073387148d41731610f9c4d57ae33628b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-25 08:55:16.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7ae5ad6455cd014cd87d964014efcd4460b9cc4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 09:38:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53d1319ae00749ff8e51e0a908527430 : SUCCESS in 37m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/881553182f6c43ac8b3192b0be7332f8 : SUCCESS in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c1bd833c8c0843513315930551bf7a04469fab44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 11:22:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fceb02900a9c4a13a54f6b6d69ad037a : SUCCESS in 40m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2535c352c2b34b698d04e3eff13fa713 : SUCCESS in 2h 02m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3124f4e7570c40f89b6b2d46ca179841 : SUCCESS in 2h 01m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bdae7b05472b44869f79f48ae9069c28 : SUCCESS in 1h 20m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eec030961d68458a85988e8c8cae7f68 : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9305a3126d6140c89005f99c25a2c704 : SUCCESS in 34m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7ab71fb7e0e4f8c817d65ced2e2c915 : SUCCESS in 36m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70703434387e469e9264a034588f6f85 : SUCCESS in 35m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0156bb40d164a258774207de04fd8ae : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fc529bacf924df0a07eb6a3d5e0c4d9 : SUCCESS in 28m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b12614b7006c48da8ddaeab32076feac : SUCCESS in 24m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cffbfb13ebc44eeacdb5af96b203812 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7c4e0f6fc441078ba10461ef13e135 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a34534fcc75419db8aa13dd63d7c618 : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbe28374e26b4f208c226c529ed28529 : SUCCESS in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2aab72a1ada24c88921eddec44355825 : SUCCESS in 1h 23m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f97e5ff84ad74626b464d43d256a4911 : SUCCESS in 1h 05m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0bc92720a724faebb73845c01bf4b0b : SUCCESS in 49m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8e716cd7e9e4283a5d9d34b2a4008fc : SUCCESS in 1h 03m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/17d98df7a6b5456d9d6216834309a4dd : SUCCESS in 1h 08m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1eebcc582b754aa6a1a3696ffc7e16a0 : SUCCESS in 1h 12m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbbadc8ff9224eceab318dd62f6d38a7 : SUCCESS in 1h 44m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5994748f22240618254798bdb9dad2f : SUCCESS in 1h 00m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b9cd0e7981d49aabd01cc9ef664dd91 : SUCCESS in 1h 01m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b01f160693384957b76a90c034e703b6 : SUCCESS in 36m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32bc0719016249d1986717e73a382110 : FAILURE in 14m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e7dafc39f8a3419d9a5d7dc22ab2ec91 : SUCCESS in 1h 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e6686c7755ce9244acb1d2ed4de99d3143d5ea8a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-25 11:28:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/789929/7 : SUCCESS in 2h 23m 23s","accounts_in_message":[],"_revision_number":7},{"id":"fd601d7b3b713b4726b6561052f0e6b2027ea560","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 06:14:52.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8207659c3c28e7f624235021d13d5e40161847b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-09 09:46:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b581738d2012f11ffa5dd742aa8630db57efa2ac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-09 13:19:24.000000000","message":"Patch Set 7:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":7},{"id":"b790054d7428d3c42ca906c0cbda1d001fe10101","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-09 13:49:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b982a6b1b6c2f2a374667a7a4ea13bb20c7ee44f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:49:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f51d5d30198402109f48df338bdb76eb2bf69c0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 15:59:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f9193606fda441cbaf21a66851ec564 : SUCCESS in 45m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3f8b4ce81381432b93cbaa5fe75f013a : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/861dc80af694484e8b2aa259a37fd110 : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f05c5d4aaa374b059e777fb103cc7686 : SUCCESS in 1h 31m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/226ceacf918f454193b38237e498b10b : SUCCESS in 1h 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d8c1e4e46aa428980d4bb14c4a9fab4 : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48731e5bfbb7425f835c4cd2f941437d : SUCCESS in 15m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4afaee1f7a4412481919f8d3ad54cb4 : SUCCESS in 29m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c83b901d8ace46b1a5850059fa47c1cf : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a98dedad42d4683b25de2280d98a862 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10d55726a2264617a35bd7f97bda831a : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e2d5b4e425646079e32a0092f3a60a4 : SUCCESS in 46m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a62f1c8e7084e85a7926f1544b2a7b7 : SUCCESS in 1h 17m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c3276d1b3b9c4e8daa70a7f245bf1a38 : SUCCESS in 1h 11m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc610f5f6d4741cfa35963e7cd748918 : SUCCESS in 53m 37s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d1d18861c4504c8da3693b94280020b3 : SUCCESS in 1h 36m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/343a7bf6770e4e1887800f783dea9f83 : SUCCESS in 1h 01m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de7447954b4f48b2ac34d678327663b9 : SUCCESS in 57m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5f9548c2d6314352aea35b6c8e6ef60c : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":7},{"id":"a71459a64e07c3b1f26773934ce3266eeeb6a914","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 15:59:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"6467ef0004ac2d924dfaebceed4705464705b32d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 16:01:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a96d77a58441b0843ce8f9210af7d4 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fce68dc33de040b288fcddad61c7c405 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5fb5653ffe700a8f9fb0949d16ecbf0978b61fb6","revisions":{"23e2201a5439fde515a724618bd79604f20bdac2":{"kind":"REWORK","_number":1,"created":"2021-05-05 17:27:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/1"}}},"commit":{"parents":[{"commit":"4d26c3a5449b2a9e99091204606023040b7635ed","subject":"Merge \"[OVN] Check for lock in check_for_mcast_flood_reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d26c3a5449b2a9e99091204606023040b7635ed"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"subject":"Check router routes connectivity with external gateway","message":"Check router routes connectivity with external gateway\n\nTODO: testing, documentation, release note\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\nCloses-Bug: #1925368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23e2201a5439fde515a724618bd79604f20bdac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23e2201a5439fde515a724618bd79604f20bdac2"}]},"branch":"refs/heads/master"},"d430fff1bf69f0af4fcad4f4ea40d89bf978bb89":{"kind":"REWORK","_number":2,"created":"2021-05-06 09:36:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/2"}}},"commit":{"parents":[{"commit":"4d26c3a5449b2a9e99091204606023040b7635ed","subject":"Merge \"[OVN] Check for lock in check_for_mcast_flood_reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d26c3a5449b2a9e99091204606023040b7635ed"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 09:35:53.000000000","tz":0},"subject":"Check router routes connectivity with external gateway","message":"Check router routes connectivity with external gateway\n\nReject any router route or gateway update if not all route nexthops\nhave connectivity with any gateway subnets CIDRs; in other words, all\nroute nexthops IP addresses should belong to one gateway subnet CIDR.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d430fff1bf69f0af4fcad4f4ea40d89bf978bb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d430fff1bf69f0af4fcad4f4ea40d89bf978bb89"}]},"branch":"refs/heads/master"},"cf38a191525f58b589d26372ec17bbae63bc870d":{"kind":"REWORK","_number":3,"created":"2021-05-20 17:25:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/3"}}},"commit":{"parents":[{"commit":"4d26c3a5449b2a9e99091204606023040b7635ed","subject":"Merge \"[OVN] Check for lock in check_for_mcast_flood_reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d26c3a5449b2a9e99091204606023040b7635ed"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-20 17:25:21.000000000","tz":0},"subject":"[WIP] Check router routes connectivity with external gateway","message":"[WIP] Check router routes connectivity with external gateway\n\nTODO: change commit message and LP description\n-\u003d-----------------------------------------------------------\n\nReject any router route or gateway update if not all route nexthops\nhave connectivity with any gateway subnets CIDRs; in other words, all\nroute nexthops IP addresses should belong to one gateway subnet CIDR.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf38a191525f58b589d26372ec17bbae63bc870d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf38a191525f58b589d26372ec17bbae63bc870d"}]},"branch":"refs/heads/master"},"62064d38f80cd0e95b14ebb306dc89f954061a81":{"kind":"REWORK","_number":4,"created":"2021-05-24 09:22:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/4"}}},"commit":{"parents":[{"commit":"4d26c3a5449b2a9e99091204606023040b7635ed","subject":"Merge \"[OVN] Check for lock in check_for_mcast_flood_reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d26c3a5449b2a9e99091204606023040b7635ed"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-24 09:20:40.000000000","tz":0},"subject":"Check router routes connectivity when GW port is updated","message":"Check router routes connectivity when GW port is updated\n\nWhen the router GW interface is updated or deleted, the routes should\nbe checked in other to prove all of them have connectivity to any of\nthe router interface subnets.\n\nIn a router, all routes nexthops must have connectivity with one\ninterface subnet. If not, this route cannot send the packets to this\nroute gateway; therefore this route becomes invalid.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62064d38f80cd0e95b14ebb306dc89f954061a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62064d38f80cd0e95b14ebb306dc89f954061a81"}]},"branch":"refs/heads/master"},"e612b3c11890961adfea3dad4a20497a5a870b45":{"kind":"REWORK","_number":5,"created":"2021-06-03 15:55:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/5"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-03 15:55:10.000000000","tz":0},"subject":"Check router routes connectivity when GW port is updated","message":"Check router routes connectivity when GW port is updated\n\nWhen the router GW interface is updated or deleted, the routes should\nbe checked in other to prove all of them have connectivity to any of\nthe router interface subnets.\n\nIn a router, all routes nexthops must have connectivity with one\ninterface subnet. If not, this route cannot send the packets to this\nroute gateway; therefore this route becomes invalid.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e612b3c11890961adfea3dad4a20497a5a870b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e612b3c11890961adfea3dad4a20497a5a870b45"}]},"branch":"refs/heads/master"},"a475fd555906e19c5a751ec727cb3fe7fbd7aae8":{"kind":"REWORK","_number":6,"created":"2021-06-09 17:26:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/6"}}},"commit":{"parents":[{"commit":"9f1fbbb83462ef24bb623155eac62640a7fb49b2","subject":"Merge \"Switch neutron multinode jobs to ML2/OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f1fbbb83462ef24bb623155eac62640a7fb49b2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-09 17:25:51.000000000","tz":0},"subject":"Check router routes connectivity when GW port is updated","message":"Check router routes connectivity when GW port is updated\n\nWhen the router GW interface is updated or deleted, the routes should\nbe checked in other to prove all of them have connectivity to any of\nthe router interface subnets.\n\nIn a router, all routes nexthops must have connectivity with one\ninterface subnet. If not, this route cannot send the packets to this\nroute gateway; therefore this route becomes invalid.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a475fd555906e19c5a751ec727cb3fe7fbd7aae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a475fd555906e19c5a751ec727cb3fe7fbd7aae8"}]},"branch":"refs/heads/master"},"5fb5653ffe700a8f9fb0949d16ecbf0978b61fb6":{"kind":"REWORK","_number":7,"created":"2021-06-24 11:54:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/789929/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/789929/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/789929/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/789929/7"}}},"commit":{"parents":[{"commit":"e12ba06e8bc959b35ccbd338e0e359a4224d4c9d","subject":"Merge \"use payloads for PORT BEFORE_UPDATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12ba06e8bc959b35ccbd338e0e359a4224d4c9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 17:26:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-24 11:54:17.000000000","tz":0},"subject":"Check router routes connectivity when GW port is updated","message":"Check router routes connectivity when GW port is updated\n\nWhen the router GW interface is updated or deleted, the routes should\nbe checked in other to prove all of them have connectivity to any of\nthe router interface subnets.\n\nIn a router, all routes nexthops must have connectivity with one\ninterface subnet. If not, this route cannot send the packets to this\nroute gateway; therefore this route becomes invalid.\n\nCloses-Bug: #1925368\n\nChange-Id: I7ce93b863b0dc0d4a2376fcfd602d2facb6fb2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fb5653ffe700a8f9fb0949d16ecbf0978b61fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fb5653ffe700a8f9fb0949d16ecbf0978b61fb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
