)]}'
{"id":"openstack%2Fneutron~790060","triplet_id":"openstack%2Fneutron~master~I4f98c6fcd781459fd7150aff426d19c7fdfa98c1","project":"openstack/neutron","branch":"master","topic":"bug/1926787","hashtags":[],"change_id":"I4f98c6fcd781459fd7150aff426d19c7fdfa98c1","subject":"New Quota driver ``DbQuotaNoLockDriver``","status":"MERGED","created":"2021-05-06 12:57:44.000000000","updated":"2021-06-04 23:22:08.000000000","submitted":"2021-06-04 23:19:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"790060-bug/1926787","meta_rev_id":"e095f3ba64bf8a8751bf5de9aef9103619aa6ddf","_number":790060,"virtual_id_number":790060,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 23:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-05-21 06:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-24 19:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-20 10:57:37.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-24 19:53:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 13:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 14:50:15.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-06 14:51:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 15:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 21:47:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-07 07:48:12.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-07 10:37:15.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 06:12:06.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-05-24 17:22:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-24 17:22:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-24 19:53:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f50e02efbaf2fafb634d568d520eba629e10ba76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 12:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c75f8db0bc74ab1a759d29dc663bb8268b1ecc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:40:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5654ad52b274ac7b6f4af7f54e97734 : FAILURE in 37m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee72acf9b71745c1ab7b0c6ebc1f35e0 : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bb2ab83959d3026c9d317845a32f95a89e81dda","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 14:50:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/1 : FAILURE in 1h 51m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fddedf13688a727ab046cd6afc72a529444507cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 14:51:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/60/790060/1/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/60/790060/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8c66096321cb15c933045542569985e7a4fa5460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:14:27.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/659738e974c449ebbc866459c3889bda : FAILURE in 46m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/29c39371311d48138e731462bedc8f96 : SUCCESS in 2h 05m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/04b4342df2d14f66a92817f27df39931 : SUCCESS in 2h 09m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/955a815e1c75491dae4b1594851366b5 : SUCCESS in 1h 27m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8353e5770d2443a0948f0cd58b4ee798 : SUCCESS in 2h 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7ccf601a814a33affd807024202746 : FAILURE in 36m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cea2bf3995747e5bb85cf6d400ab61e : FAILURE in 37m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d92630f8de4e8aab1df76f89f38cde : FAILURE in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1be30749acb4bcd83c5b40cfc8efb88 : FAILURE in 36m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/989a93a62ba84a229940567567d971e2 : FAILURE in 27m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9ee8312767a4f0b881d9d87cd07c65d : FAILURE in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a54c8df1819474a89085c81cf9c9a12 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c98c685039ed43978abda6abc3267255 : FAILURE in 48m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdd73463b13b4bc78729abe9a6715b74 : FAILURE in 1h 12m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e01b8322f97b40508405fcdb6480a645 : SUCCESS in 1h 30m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6caea84fa9584e8bbed3e6dd12fd8367 : SUCCESS in 1h 00m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61688ea9c9874826a467e608eeab4090 : SUCCESS in 1h 14m 44s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/890e90e3cd9048cfa118ffdf6292fad1 : FAILURE in 1h 12m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/406adcecfc124757bff831adb2673393 : FAILURE in 1h 11m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db72e53bef4b4d94ba47d6e6defc7d2e : SUCCESS in 1h 44m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/948dd008ceb04c698357cf753437ddb9 : FAILURE in 1h 02m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25f4b0fe466c48d991d2a62be437ef33 : FAILURE in 58m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8ea3c9150b3f4aefb0ac0733062268c8 : SUCCESS in 37m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0750ad2182644bbb518e082bc885f6e : SUCCESS in 1h 23m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abfc7f4c50fa4c89b7ec25eaa0582fc0 : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f17968c9d841d9a5a200a685edc8b8a80e385d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 16:53:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b96962f597b1acf017462d95adf9ee09f418528e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 17:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cca0e64673dd4445a556dcaa63549b4b : FAILURE in 32m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/368d43a612d247efb11f14bb65e50801 : FAILURE in 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7048c5bf6c877df81c587aed6ac4b2976178fd67","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 18:28:29.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/2 : FAILURE in 1h 33m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c960a393e295f124bfba8b89bdf246e8b6f0b744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 19:43:00.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05b4d6ff3d604f39af58057b2ed92325 : FAILURE in 46m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/120c154272f9491b95bf83051482fb4d : SUCCESS in 1h 36m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd71292eccd044e9b5ccb95b8848d34a : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fce710abb89d46cb8473549222ef4dd8 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f17faa22aa8b417ab211201ef12cd699 : SUCCESS in 1h 54m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e21dea14f3ce4b0e88adeb9c6056eb09 : FAILURE in 43m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e464195582e421ead57bd9cea0bc93c : FAILURE in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c7ee33dddb499fa9e4efdbdf5225e6 : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b13bc9b87a4fc6b95d273628e6541d : FAILURE in 24m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/338d6aae967d40b3b8c6068dfce188b1 : FAILURE in 25m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73eb5b862bf646eba6daed67f309054a : FAILURE in 23m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58998953c80a4be6b917abb0a929ca1a : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/632e6692ef004aa99e33056b244b926d : SUCCESS in 55m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/334eabc1823a48268b94128799584c47 : FAILURE in 1h 16m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ffe0faf03d014f109984807488969886 : SUCCESS in 1h 14m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aefb64b8131f49fcb3dae20a4e835682 : SUCCESS in 52m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c2bb8d16579f47338bb0f7a12dfbb498 : SUCCESS in 48m 45s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/228fe772046b44c9a9464462f061bafc : FAILURE in 53m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb372c08c9614df0be9c700dac139e3e : FAILURE in 1h 03m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c8700ea5706e492d89b3a7e5f9eae6c0 : SUCCESS in 1h 23m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/207b7adec0d2423492914ae34a4263df : FAILURE in 46m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c1018b1f101f4ecb9e1de24681a7571e : FAILURE in 1h 05m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/de1cc3933dbc4b0780bcd1b0936fd5f4 : SUCCESS in 34m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65ce3533cf8d4cac86e79e2950a9b9aa : SUCCESS in 1h 18m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/14785341f1fd454daec05e9860eee27b : SUCCESS in 1h 22m 11s","accounts_in_message":[],"_revision_number":2},{"id":"56ddce9bcc63fb0ca69ec4166162a10777b899b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-06 21:47:58.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3b32cff1575420a90bf739a32358c5e3d38c5d9a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 22:32:19.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/60/790060/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/60/790060/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"30d44a319e33f2e0e9ab01f49d47d05179302bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 09:07:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58cfa69b8b37c7879d179c2703df8cb3c0b5d165","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 09:21:26.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/3 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e0574d7042283702c7595a0295af89e66ed905ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 09:53:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/919fdfbafaa34697ad6e9cb95d995c97 : FAILURE in 32m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2ce436075b3449abd25d01cf1f1941f : FAILURE in 44m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ce79feef5e7e6042e36ef8c251f866f86684ead","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 10:37:15.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7e0459bd090c4850a061fe45d6f32cc81378b30d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 11:37:52.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d660b883113a406cba6bc21b1f1c1b85 : FAILURE in 42m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f6e6ea3491f44da2978c40970d6310e0 : SUCCESS in 2h 30m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/29b1967d1a5f4b8cb3975c68e8d233b7 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d6f3402accc4692af7f9940f1a71262 : SUCCESS in 1h 58m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/085c5cf21ce24b84816232c36ac1334e : SUCCESS in 2h 20m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6596a513362f4cc2b183646c69227e84 : FAILURE in 37m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce4e38339c0486fba802f9bd182079a : FAILURE in 42m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff432efe6f848528533aa0b3a908812 : FAILURE in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb9603d64a724a1a811711c63ac3e4ea : FAILURE in 39m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/699f0bc5986e4119ac3effc9e34015b7 : FAILURE in 26m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b70edb043784ba58d6889052f9b214e : FAILURE in 28m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d489978e96a949b7ac807cd8f9d22fb1 : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d469a914c146415db110777932f32291 : SUCCESS in 52m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74d4007b3c9443c1add4c84a754b737c : FAILURE in 1h 05m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/099115a1a1394be683e8c9d780cb9023 : SUCCESS in 1h 31m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/005b8e9cd27d46febf1b07f13c11be4b : SUCCESS in 1h 10m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2458991d42c24827b71a8e8adbd0cca0 : SUCCESS in 57m 11s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c112f31118684d63b4b80f5f70a3e43f : FAILURE in 1h 07m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7bb7ad80f78460ea34a84ac81a7b9d5 : FAILURE in 1h 06m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67224335da334ca0a821a12b79e6629d : SUCCESS in 1h 36m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e5531fa186742d7af6463abdc8697f1 : FAILURE in 51m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/efffef509a1341a49dbd4fc8659a67d4 : FAILURE in 57m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1d4c7e7e95ac4097a19b3707a228b89c : SUCCESS in 37m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f89e9f2d98554d50be262559d3d55d2a : SUCCESS in 1h 15m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2ee3083ea1b04c90b0743806e6112ee9 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":3},{"id":"fbbe44943d48dfdf6ce61bf65a11ec35c6e8b88a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 12:42:28.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/60/790060/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/60/790060/3/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9210d20df887ce9759ad017378bb1199403c8b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 15:45:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ff628fba6afe83a541556f8d22012f431256ee6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 15:59:59.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/4 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":4},{"id":"23a8b4a4333d629f802b306b1c1a7862476903de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 16:24:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2c40f1a9ed041d8bc1b7e7a2e545ba0 : SUCCESS in 33m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e531e4b2775f4b4ab3b6b1d4f5310bf8 : SUCCESS in 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"81ce6c761af150224d71eb76b3fa70ea291a0e27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:10:13.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83af6b1d784547b49b82eccb40c07696 : SUCCESS in 46m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/828cc655cb044ebba27a9d98381fc9fd : SUCCESS in 2h 15m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/82d2e4dd949d4829bbd482b8d09d4617 : SUCCESS in 2h 01m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86cc27eefbea4484a3530c9b11111cc3 : SUCCESS in 1h 24m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/37fbfebf94484c4f99751a613a55e1d2 : SUCCESS in 2h 18m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/587bb83d14124843b6e534c49697cb1e : SUCCESS in 31m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fcef48782674a85a5450311ba34b181 : SUCCESS in 28m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d083e844b2f7474ebbe9cc4e5b282491 : FAILURE in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cff9aea2f1446c89f272c9e1533859 : SUCCESS in 32m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6316c805824626848d50d6f1ed057b : SUCCESS in 30m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bceb1938cdf14f98be5a8c62ad62fcf7 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d674f0eb4b674f4fa320cdfe9cbc199b : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f76743190c4439cb676fabb3aa5c103 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1008363ce8804f34bd536bddc85e9d2a : FAILURE in 51m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fd4deed7b03448cae3db42598149d4d : SUCCESS in 1h 17m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6e9197cbf8904b4f81f29db0071dcac5 : SUCCESS in 1h 13m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8d8ac06308e4a9a99392067e8103431 : SUCCESS in 1h 20m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f90d930ced64aa4be1e4fa2257f1fa8 : SUCCESS in 50m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fc2f2b82652a42aebbc9a9d447e8b5ae : SUCCESS in 1h 39m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2329e8ee4d5403cb1a4bc3fb5437dbb : SUCCESS in 1h 53m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee56b8cc942a48f287432adf672bf50b : SUCCESS in 1h 29m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bb9599bf22f4261a46cbaf3fbc22b1b : SUCCESS in 53m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba37469b11f14fbd83d32efd59749706 : SUCCESS in 57m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f368ccb146d44b693e0133a71fb57b2 : SUCCESS in 30m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1009100ee1c478592d60e05d0cc38d0 : SUCCESS in 59m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ebff8301a1e34219b98f61b017d601ca : SUCCESS in 1h 55m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d0074e6bc2291e67a49b7f7eebb9bbd870f551b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 18:40:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3f4c7fe2ee9a2b17b648af1026fa273c5d608d8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-10 08:15:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75c8c421aa861b708897e3fc3e3b9698a881752e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 09:01:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cadbecfcc184bff8f98e99b1507f598 : SUCCESS in 43m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68893a510272456bae71ba8ef7507161 : SUCCESS in 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"276a36a452a72afa9e92cc51a1e2a5f356c6e2f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-10 09:36:28.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a6fdcad50da6097ae803db1b75e11694672d6f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:36:19.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d731483e371e44eeac8afb49f898b807 : SUCCESS in 42m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3aaea5f01ae747dbad21f23a0261b479 : FAILURE in 2h 19m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e2c8ca25ea64ea5b5d5f940719d43c5 : SUCCESS in 1h 45m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d4e68fc346e451e858323d7b68d13b1 : SUCCESS in 1h 24m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63a669d31ea64e5c874e2b7637df978d : SUCCESS in 1h 55m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3ef038eec39441abd979cff4190c1a6 : SUCCESS in 29m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce924e61f9484da2bdf9ce44af3fef9a : SUCCESS in 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2469d611d02429cb5172ea7d199028d : FAILURE in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d046b2577bc84b8db5f17a20c286adff : SUCCESS in 24m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e456676671e24da38aec5696f4af271b : SUCCESS in 22m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a98a1fa1fa48d8a296aba8955d1977 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd410a9e2c34d21a7cc5bcfc8cc7e77 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c441682a14254200b6c67839ac77339b : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1df6ac8ceea4454baae8246607edc971 : FAILURE in 48m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a70836137b134305898b15739c93c34c : SUCCESS in 1h 19m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d3f082d668a474b92676a191df3fdf7 : SUCCESS in 1h 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68cea5d6b2394ba9b7736f5eb917e2cb : SUCCESS in 1h 15m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb8645bb5e6041bfa3ca0122b5afa5da : SUCCESS in 58m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/382ad7c162f24b02ae56594410ffcbe1 : SUCCESS in 1h 18m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0df202a4e3c2495cbee19095784bf3f2 : SUCCESS in 1h 18m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/615c664fb6bf4ecc880e60c3626a3c13 : SUCCESS in 1h 37m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30748e0ebb414461aef8effab4ea267b : SUCCESS in 59m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/13e8dc86eebe466eba7056dd727e4e10 : SUCCESS in 1h 09m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8c5dec9025bc46b2ba59af328251a3b4 : FAILURE in 39m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2efb7e9412594758889451c59d3097a4 : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b174bfcadf354a959e941505e95f1b57 : SUCCESS in 1h 34m 10s","accounts_in_message":[],"_revision_number":5},{"id":"59062b0b7df43d3c1b33e837358475c732a6f7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-10 11:12:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2dd6f0fce22158e6cc7a24f7f79d739464973b1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:04:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0adc5f4392a24052b237dfeafe7a3745 : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b401f8202fd94f30ab000ba0b695a990 : SUCCESS in 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31033901a986adb28b210f00f6455cafb7af80d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-10 12:46:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/6 : SUCCESS in 1h 32m 26s","accounts_in_message":[],"_revision_number":6},{"id":"82442d86140c03633779fdf48b348b6e8bbaecf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:29:53.000000000","message":"Patch Set 6: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62b0520e0f944a728215718fbc846f79 : SUCCESS in 56m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8b174df7ccb6496c808820aeb70ffe74 : SUCCESS in 1h 33m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f0041b4fade41ae804cbc9e12ca84ac : SUCCESS in 1h 38m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb9bde3fb7b9491196b02bb3c0c75cc5 : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e5ed93973995456db558e20ddca785a8 : SUCCESS in 1h 53m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66e9c9f68ba045e883ce844c883a00c9 : SUCCESS in 30m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7599af4aee374ea0874afb45d5ce56a1 : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeeae2ca0488493e9408f9d021af16f9 : FAILURE in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8225e29e7b24e6899b62d0748920ec3 : SUCCESS in 23m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/307ba9ef36444ab7a9c85a48f6dfd47f : SUCCESS in 17m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7f128f0d94045f3b114d801d97b82f7 : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ff3985b30b49a6ba47ba5ae85855b4 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c76449e19c0b4a98a15b12ea50a05acb : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d73dd030a444208b0092c980907c796 : TIMED_OUT in 2h 15m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/87cd1701f6f449f191d6d80967a25671 : SUCCESS in 1h 10m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f2e2cb098b3d4923aeef63afcb167a42 : SUCCESS in 1h 16m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3094aec2ea14572b487ddf6971b1540 : SUCCESS in 57m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/388f8d04593a4924a94a305ced53ca0e : SUCCESS in 1h 03m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2713c6c755c54590939b88bb515c414d : SUCCESS in 1h 23m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8a7edf854a947f0bdfcc51b425d0d70 : FAILURE in 1h 08m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79a22c4a09c64ab993c60a07d4a083cf : SUCCESS in 1h 08m 43s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74b94e58a7b74cfb812daf7f8c9b0842 : SUCCESS in 51m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6508563115824629b91582d88caf9ffa : SUCCESS in 1h 00m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1809490c14774bab8d360bcd97d794cf : SUCCESS in 27m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/869fce51447a4b058cfda290ecaed311 : SUCCESS in 58m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8974d5b9c66a4a4487d9ffb1d554085b : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":6},{"id":"5c3e3fd17f20a0f90376c672c305bb49f71a5337","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-10 15:03:55.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"13e9c2addb09c8a708bbdf276bbc02591dd1d365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-11 16:14:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b75f8ff6c7e72eb6c8dcbe3820ab37172a1148d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:01:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/508d7316532c4b01b166bdc6509f0fa0 : FAILURE in 42m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ea8727db7854be3a48081360720b862 : FAILURE in 43m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b0dcba428443ce25a296190334cc027d3db08ca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-11 17:30:15.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"2e95aa0315c789944f4368294d59de147d96258a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-11 17:56:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/7 : SUCCESS in 1h 40m 38s","accounts_in_message":[],"_revision_number":7},{"id":"770465881d34efe0c6cbc06d7ce48a0c36a15c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:22:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7f9da0a0a3f24470b95c5eb90b4433ae : SUCCESS in 45m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e93c42c980f74ff89f057156cc2f9823 : SUCCESS in 1h 52m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d14f14284584111a42074b47c60057c : FAILURE in 1h 54m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/402b2c7413aa4abb9d2732fdd8ba7ca2 : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf62627e449c450b850a2095300f15c4 : SUCCESS in 2h 05m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4779cf0e291f4a52bf07f06e9b1e3847 : FAILURE in 40m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8242be69efcf48e586cbafc8a45d24f8 : FAILURE in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce88ad3ae1df45b1b2f9c9a682cd6e15 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9214f9c92d04b41848ecdd18903be6e : FAILURE in 33m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afb9fd01d1434249b167c84ae84233e1 : FAILURE in 30m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a24e2be6330e4469865703dce8f8a835 : FAILURE in 28m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1ce026da79405eb3875a52727339fa : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8094bfca4444d508e00f6ba213890d7 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00269adfb7214fbb970f2f6fdaea431e : FAILURE in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d66fc245606e42bc99cdc35e2110f92a : SUCCESS in 1h 15m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/775a5ee7132d402fb371fdd8838bf721 : SUCCESS in 1h 02m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b436c4435e46446386f270e790faa0c7 : SUCCESS in 52m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/321b3e67410b452e8d97a3e72fbcc337 : SUCCESS in 58m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/419fef17424645e9849cbca63d302b6f : FAILURE in 1h 48m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c40920857334729982a99f0e1bb65f5 : FAILURE in 1h 56m 43s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de97dff726d545d29f84704371b69dbd : SUCCESS in 1h 17m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/393ce971fc534defb66f9c6aa490e4a0 : SUCCESS in 1h 10m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9a6eb0860b9949fc8137b99e3cdf48e8 : SUCCESS in 1h 20m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c1c09e1f07542cd89cd2df9bedd96eb : SUCCESS in 31m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51e6f616274f4ffda3c55878ca669685 : SUCCESS in 1h 12m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a14a9055765745bf85ee734d81a3ba69 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":7},{"id":"94b114a51672a763c864a8520f94e97a9f459603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 11:30:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b073b007d1ca86017eb9a7df9612a5d579295409","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:25:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a9c0bb9c2994d86a95ae72ae7ca9d1a : FAILURE in 50m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a51f804c12e4c38977f2fb1642d85e6 : FAILURE in 52m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6878b0fd8d7958d9250919d2e70b59938d8a7792","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 13:21:39.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/8 : SUCCESS in 1h 50m 02s","accounts_in_message":[],"_revision_number":8},{"id":"d7cdcb121eaaa46f68cd14ec120fcf3f0fd9792d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 13:57:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ecfe9402b19f4be998315a1ec1aafbe4 : POST_FAILURE in 39m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bcb2b8d65ba54723ab300416ef1e4cca : SUCCESS in 2h 23m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cafb1a38d8434e64aabd427542533db9 : POST_FAILURE in 1h 56m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/278dc0bfafb84cc3bc6a81dee667d18b : POST_FAILURE in 1h 08m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c690c798439e41cfb9a22bc988385f83 : SUCCESS in 2h 22m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06337443866446b58908d4e6f8177d01 : FAILURE in 40m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f300efb651b84967809fc3214ef99cf4 : FAILURE in 47m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d475be4280594873b4ca8e5eba20ac7b : SUCCESS in 13m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16655a0e1b1d45079cd764f37eeb31a1 : FAILURE in 31m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/557aa463760a4026a017d2891cdf8c7a : POST_FAILURE in 31m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65bc5a11a4b9496b8a8737400cad63e8 : FAILURE in 30m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53470558d4d84db9a444a91c555be259 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4239a0ebccc94c5d8c2a71cd9835f834 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4c9f39f355c4ebdba6e24a05ff4af40 : FAILURE in 46m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23515e3f93eb46e0bf2c082f2b4a0d65 : POST_FAILURE in 1h 22m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ad0f22ad56de450c8708c315df9f7bdf : SUCCESS in 1h 24m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8b8b66b02a1425c83263a349c364834 : POST_FAILURE in 1h 10m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a565c9d44e54e38a9bde6cf11ae8743 : SUCCESS in 1h 00m 25s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c6ff96b43e494276a63d982d7d237a9e : SUCCESS in 1h 23m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ccec59876fb4776a5f851aa4c166113 : SUCCESS in 1h 19m 51s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6109a09b7c24449ad1d873cf9f804eb : SUCCESS in 1h 12m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e99e3621859240e38dbd3409b1bd7fa6 : POST_FAILURE in 44m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a41b201852eb49c382ca018d3840ec37 : SUCCESS in 51m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/204fbce189bd4f9e8dabb11c63fb0b39 : SUCCESS in 24m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b18efdc796549c582a69b638f520772 : SUCCESS in 1h 31m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/77043966a667428bb008ed73a0d4766b : SUCCESS in 1h 33m 22s","accounts_in_message":[],"_revision_number":8},{"id":"5c4d0a3007ed96fcf22ef750a7ab365c282aa480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 14:32:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5e1a1dfcbc660efc0775e189e16db8a6b83ce4cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:12:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a280f12d73ea4e3ca7abfb2e13e5cbba : FAILURE in 33m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75387c48e35341948b7e64192e773cb7 : FAILURE in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"490ffc0b4d5ba343874f81cb9d467496dbdc9b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 16:22:46.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5f72cb6c8144f10a65cec4e3ee02453 : SUCCESS in 41m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e484fe1f6f8c4b468b7b53439e7f206f : SUCCESS in 1h 48m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f41424f39f6d452397adc18a0684dfcd : SUCCESS in 1h 30m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca727b1aea1444599426cb039a1061de : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7fa9aeffa8584bd4a71d0b82ac7acfd7 : SUCCESS in 1h 42m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea5af53a09824f0581038adf5d2f06d8 : FAILURE in 29m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17cb977479b949b4aefb74df39b4c2ce : FAILURE in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26eecca7b961460daddfc0fab0f32eb1 : FAILURE in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18924c3a80624c1897765b06bbb71b15 : FAILURE in 25m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89db354c3bff4023b25c91a2f902e63a : FAILURE in 30m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53c5ac8386b48bcb89d46bd3b4ce5ca : FAILURE in 38m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df889947bf346f6971fd512026d8833 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88502b3804584e52b56babcce50191ca : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11f8d02d7ed546288ac533acbe98bed9 : FAILURE in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68344e56a55e4bbeae199c391246dea7 : SUCCESS in 1h 13m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/09ba14720e8346b38e161f5b8365f13e : SUCCESS in 1h 13m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60bd5e2102974bec8bc019c02e3dd0f0 : SUCCESS in 1h 00m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd348947556c42c7a2f553aaed4ba42e : SUCCESS in 57m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/42ac010ab6f1492998e781b7f698ace2 : SUCCESS in 1h 01m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a5630e6741514bd1818b90f4149dd15c : SUCCESS in 1h 16m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23fdbc643c9040e3b60f05eb3e78c361 : POST_FAILURE in 1h 21m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83f9fcbad3474b228664b2e12befddd1 : SUCCESS in 53m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc84241b157e4b609c8654c9c60262c8 : SUCCESS in 59m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/429ef2c81da9462c8287233c7209f571 : SUCCESS in 49m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c2ab5df3e834b299a968bd4947f7428 : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f288ea51a37643a5a591cf20f1265862 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":9},{"id":"89bf00ed909fe7a219819f87707bea794d8e12f0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 16:34:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/9 : SUCCESS in 2h 01m 31s","accounts_in_message":[],"_revision_number":9},{"id":"7d36c1f33b437f20e7945bcbcbc04d695c05e4af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 06:23:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9826a3e51eaf6279f3bc07089d3dc5245255123b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 07:18:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/401c12bee08040d3a74aac28bcbfcc02 : SUCCESS in 49m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5af790ad1135465d821e8b08e7f608c7 : SUCCESS in 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"53efa5cc3cf4ae180ceb6ee6e13c5683c8fc3055","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 07:27:30.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : FAILURE in 1h 02m 26s","accounts_in_message":[],"_revision_number":10},{"id":"d45e7e3c1749634544b57cff59da4b40274098ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 08:59:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d80e346de61647478158468cde813a9c : FAILURE in 38m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ba32de204ca9438883361fcda32572e5 : SUCCESS in 2h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6d234b4501de4a5eb49005cc4f5a346a : SUCCESS in 2h 04m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b0ae3f05cb94aad8282d45e31497052 : SUCCESS in 1h 12m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/29e5c0c16a194d3ea9d8e4dc4c6a09b5 : SUCCESS in 1h 45m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d7d82f553e648c380346dfb035d3957 : SUCCESS in 31m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24535c5886b44e33a245525d10bfd41f : SUCCESS in 27m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57b390b456c42afab2be3913d38c02f : SUCCESS in 11m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6778bb9ff3ea447389e3d1a429b4e879 : SUCCESS in 20m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fc7864727264a6e95b3b2b791618577 : SUCCESS in 30m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5862008cc9ef4d1ea8ae9c31392d9938 : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430f8973b3664c8aa52b05593c05eb7e : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ca17f3178a42ffa5099d03ff02a025 : SUCCESS in 12m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e53f5c0fc97422ebe2d4fc3e22ff4d6 : FAILURE in 42m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bef65930533a462abebff13ccb025a27 : SUCCESS in 1h 19m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b4ed075cb4134a5f93b5d7bc91cba230 : FAILURE in 1h 53m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82751a11392446c7948853f4a80f529d : SUCCESS in 1h 02m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/839b6a364b9e4928980c87844beb5665 : SUCCESS in 1h 04m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d677bd355554173986db758b6e7b460 : FAILURE in 1h 11m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c628f3d86f75439e96660199693f7674 : FAILURE in 24m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37696c43584a4252a94b4e91bcad497c : SUCCESS in 1h 34m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70d0256d1f5a4aa5961b1b86cf18bcd6 : SUCCESS in 59m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/deadc256e6484fef932d321fd5c7f087 : SUCCESS in 1h 00m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a99a2e29c2458f9b9b04e316432af2 : SUCCESS in 30m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b956b4200a3140178ab49a1ff9ad9fd6 : SUCCESS in 1h 15m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8f04166ae0574bbd98a775befd5670b8 : SUCCESS in 1h 29m 23s","accounts_in_message":[],"_revision_number":10},{"id":"2db1c402bd48d6e77b514c07f7dd33ae2fe8876b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 13:31:41.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"612e2f4c54798462a9aa07cf3bf0d801f9e26d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:11:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97885ade4b77477d8940afe8a250659f : SUCCESS in 35m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f1e49caa0604027804219f150a8e3c6 : SUCCESS in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5f375bdfcd17a2574209ba058084c86a463abe4f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 15:27:22.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : SUCCESS in 1h 54m 57s","accounts_in_message":[],"_revision_number":10},{"id":"bc2b4e8a766326125a65dffea3472e9ac97602f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:51:32.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a53d81f88e60403f93da4f8007594e6c : FAILURE in 50m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a3f7ac414854126bfbb8837526ccbc8 : SUCCESS in 2h 13m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ea9568d7f484456888a15aa787a9347 : SUCCESS in 1h 54m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4483b39acf9c481a97a142c818df4b7d : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60b879da44ce4ce8a806ed481b6d7257 : SUCCESS in 2h 16m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82d3750996340b894cbbb3aa83891e3 : SUCCESS in 45m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1a029492c3b4987afe07e0226c54dd5 : SUCCESS in 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6037f3664afc4fd391743739bc8a97db : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa182be9e25a4c538fb0d147980a0dac : SUCCESS in 24m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e2cf5258e724cc0a66b8cb0bb36bfe8 : SUCCESS in 22m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26ec63b40078461990b3b0d21e2fb342 : SUCCESS in 31m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf7ed1b8c554d8f911d5ad2e6c0ba81 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2d4d277eef74732a398e7c4be9981b9 : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5324262fb3ce486b8141cc3f600777a6 : SUCCESS in 48m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c84d07e3da9c4d68b0a499ebe25d59eb : FAILURE in 1h 06m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e00a858e97954e83928f117924050c4d : FAILURE in 2h 01m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/96f90f3d208e44fe99f08bc0fa8d012a : SUCCESS in 1h 06m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11283f0fb3d54881850145cc1f958901 : SUCCESS in 1h 06m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6428bd512f3843bbbf492ae1e54673bf : SUCCESS in 1h 36m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1fdb358ef87041788133218d8c16be80 : FAILURE in 19m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/166f1e80a4d24d71b24f9fcdd18f6aca : FAILURE in 42m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3177afce1ea545648bc6f600b3dbefae : SUCCESS in 59m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/677e153c24aa450689a494701e2ca003 : SUCCESS in 57m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2844f675390f45cea477875e6a940302 : SUCCESS in 24m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1350bf727a4e4e1cb7c73c50a2d85f0f : SUCCESS in 1h 33m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e69f088506c416a8bd3aca4402a362a : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":10},{"id":"bf5c5f4edd0338727181da19307342e50051b1f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 16:02:18.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"7e76b7f242ec8286fb0496e85380c75292ce2b1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 16:54:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4faac28b71140609c78b0430fd79cf6 : SUCCESS in 36m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c9430193b9e47d4b38db5a6fa597b40 : SUCCESS in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"053610e0f4da80b5e3bb125dcffa853989521ffc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 18:19:13.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : FAILURE in 2h 06m 06s","accounts_in_message":[],"_revision_number":10},{"id":"f41ddf551387375ecebdc3b33d41a5b6c2189d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 18:25:56.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f1474a9d3e44a98814c2d457a636dff : FAILURE in 40m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/651abd9b50d9490d8132240faf0b3ae8 : SUCCESS in 1h 43m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cf5f12412f404e319310b52d4be7a257 : SUCCESS in 2h 04m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bdd25df89a434a89ba634647c1d7d942 : SUCCESS in 1h 46m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/161169c6f551499f9e08c2eab82f1131 : SUCCESS in 2h 14m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a33c5d87be04ba6a657e51bd79acf22 : SUCCESS in 32m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/360a741eb81348718acdfce965c22611 : SUCCESS in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9029ba7e1ba54a3b96e3b5a1df7a79dd : SUCCESS in 15m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f6908bc00f04d13b6205ecb15553d7f : SUCCESS in 24m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761ab37f3c844d23b971c3fc43624404 : SUCCESS in 37m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a545f7f20c9040cabc541b34346aba3e : SUCCESS in 31m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5947a5a4a08648f98d18008bd878f7c9 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a644cef17f64affacfa162f797c43b9 : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/251677d8d85e4777ae6b1caa6770694a : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fda490e047a44b78c5cd1208d4bf0e9 : SUCCESS in 1h 25m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59bf90e296874540afd2d70d183be5ff : FAILURE in 1h 40m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47ec749173864c06ad88b3df224f0897 : SUCCESS in 51m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4880cff87754eed9e5956101eeea430 : SUCCESS in 1h 04m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/12300a6fcabb468b938886e69cdeba83 : SUCCESS in 1h 27m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4466dd4efb304637844a85070b62606c : FAILURE in 24m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9dde05fd23d4c798dca945f100a51ba : SUCCESS in 1h 42m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2103999c2ccd4a60aec293df5e738cb9 : SUCCESS in 1h 09m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f4b1c1c2c074540ac18d1adc406d338 : SUCCESS in 1h 09m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e7f30f500654078895d50d5ce056d3c : SUCCESS in 33m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b22c097ee18c4c629b360ec7819984e1 : SUCCESS in 1h 22m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/970818b83b5f4f059224a25e46b99100 : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":10},{"id":"645747b95042271cd6666a688a8009b5d9c03bf5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 18:33:27.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"68be3a093e22a04354d6cb05df6d167f05bd0868","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 19:23:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fe076d6e322471e80c17b89432fffa2 : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6450a6b8cb0941e18a67b4502dc5df7e : SUCCESS in 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1c067001e79c6a598489c47d11c1f5994c204dc0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 20:31:46.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : SUCCESS in 1h 57m 39s","accounts_in_message":[],"_revision_number":10},{"id":"9a3d199bd6983df9c7b043375729f41009b44eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 02:11:21.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/493a756ffe5445a598f02f9a2e99346a : FAILURE in 55m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f90118224ff0421b955d470bb082f05f : SUCCESS in 2h 00m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/71f39b14a44940bfab68bc5406ba7c42 : SUCCESS in 1h 39m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ebaf7de1c71479bb4356a118d8cdc2e : SUCCESS in 1h 59m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cca8913f9cf240c092c0dd1ae4278cdc : SUCCESS in 2h 13m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad3dc92b5f1f4788b87764820ee796c8 : SUCCESS in 41m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5988367ad4142ca99552c2cf2cef8ac : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b711144d9b5748058d4114a799b0df86 : SUCCESS in 15m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6fa7e4c98040f69d93b22b6c7743b6 : SUCCESS in 32m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e651f28224f446eab3c598f27e14e54 : SUCCESS in 27m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54128570f0f741478aa05bd6ee2d1cf7 : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66035fea6d5f4847ac2eca00e50b126a : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9eaa01351254b93a70eb9efa2e2e397 : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2db63dbfe0c42f0b3efa8e01296201d : FAILURE in 46m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/808b2b8d28e144038f424d34654bc3e5 : SUCCESS in 1h 21m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/deb0a372b4ca40e8bae26ab69ccace93 : FAILURE in 1h 32m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09fc0ffdb2e84c39aa2041ec609d132e : SUCCESS in 1h 11m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c0d526c18e34455dbe5e60f703f34ba2 : SUCCESS in 1h 05m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/09f9df09c440440e9e517c992b6f7981 : SUCCESS in 1h 13m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d33fe1bfc0084a55a22a70a7bdf60936 : FAILURE in 38m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6855db3ca28e4a23a1b025800c0007ea : SUCCESS in 1h 37m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6665341aba0456b8c1fc767e5f885a2 : SUCCESS in 1h 34m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e4236bec635b4a8c86d6a476dcb73ba6 : SUCCESS in 1h 07m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e415015676b04107a21d10174342f0be : SUCCESS in 28m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/919160080e44484799fafbb0d9301b6f : SUCCESS in 1h 26m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0b6ba625495048e6977022e2ae0e067e : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":10},{"id":"a451a67123d3bd4eb1f2aa99a5bfed67316b516d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-14 02:35:09.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"f1adf4cdc2d779b5ae87ed33e56e0a2f64c39b97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 03:24:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21ce157e2aca445fad9cfb6066860176 : SUCCESS in 46m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8195a3abb7094f3686714c1428112ae5 : SUCCESS in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6909c64c6f46f3621aa0151af75e6dc921bf1db4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 04:22:02.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : SUCCESS in 1h 46m 17s","accounts_in_message":[],"_revision_number":10},{"id":"96c32c2cb712ce8e5574e18aa03a1daf266b1eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:37:21.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cdfd3904a8c84f35aa8fcde540cf2126 : FAILURE in 59m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/92b0d70d4c254ff5aa45a5af7a0c8d08 : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72f0a7de9a924f759ca12fcb830524d9 : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a282db5ff3e4f1c942b3178c4c1d591 : SUCCESS in 1h 22m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ffb475154186416b841dffa80f9f15fb : SUCCESS in 1h 57m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cdf7472dcf44574b76c9a25ebac5d3f : SUCCESS in 46m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f932797b8e7a4935ac97a52103fee9ba : SUCCESS in 40m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10079f4470184ed4b0767ce67764115f : SUCCESS in 12m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0216909b70554fe2a01938fbc903ce49 : SUCCESS in 24m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e65994f61d9c4533b7cad5116fbaaa62 : SUCCESS in 26m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc5f332331d4b4db76b5425e90cf8ee : SUCCESS in 32m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c81fae512e4c158368cf921e993928 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6581e6becc241f3980fbfbd7e5aca91 : SUCCESS in 11m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9763c5cc78c54a0fb703d6645b7b44f4 : SUCCESS in 45m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcb29ee27b2840eba6e755946bfc9f7d : SUCCESS in 1h 17m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6462ca07607d49b3a3e1081d56e9c48a : FAILURE in 1h 27m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3be679f8bfe646a4b6d66aa031997d89 : SUCCESS in 1h 08m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb9cc6e6737d4e439a9e940639da9f9f : SUCCESS in 1h 01m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0636aba15aa40209ea255ffef430e08 : FAILURE in 1h 03m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6738f52290fb41d6bd311ee31ff7d4e9 : FAILURE in 24m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1177bfdb3f844b059b6723a28e59694a : SUCCESS in 1h 26m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5102bd1513744e9a9163f29863148d04 : SUCCESS in 1h 04m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8987f9956b0b4b47b29898345cd56f26 : SUCCESS in 1h 22m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/48e4a8fcca3e487993ece94c452acd02 : SUCCESS in 27m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fb8104b5b2f456080553aed69e55491 : SUCCESS in 1h 22m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/628bff1b89fe42978f56e94dd96a8f57 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":10},{"id":"a4252d5e86f165021a9f32ef3c59ade130b9fc09","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-14 05:05:44.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"d7f1b85a88a1d72f0d099a92dd21a698054bc4d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 05:57:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3574b659beb41b9838a6812c5c178a8 : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa722dea73f04f5bbaf2bf61b252a622 : SUCCESS in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"02dc068bf36448663e15e531e267318ee4ec028d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 06:59:39.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/10 : SUCCESS in 1h 53m 11s","accounts_in_message":[],"_revision_number":10},{"id":"38d203f76f22c5fa96bc55d6ac9eb07b2ea69e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:10:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d84cf9ff44f4d2a99625a3a8d913825 : SUCCESS in 42m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/700c289369884d28a9e1225eb48f01ab : SUCCESS in 1h 58m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/939ee7648bdf423a8df264427af47863 : SUCCESS in 1h 46m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a953bdea2b354b3ea7f4378c485244e6 : SUCCESS in 1h 35m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aea52af5d97e4d5b980bffdf5f051149 : SUCCESS in 2h 00m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30dcd94326ee4360947b84a2ad02d032 : SUCCESS in 40m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83c4130a1f3b4c49826ae1602191c7d8 : SUCCESS in 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859a6061876a4776bc12e7fb1ed3fce2 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd38dd9c47a4ced9f5f512403802ec4 : SUCCESS in 28m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f921bfa7714732a93e86c5fa356921 : SUCCESS in 25m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a411f484e2e5447294bdf01e3e1da1de : SUCCESS in 27m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d148755083b54280bb332fd0c5664530 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/172e8b35631c4e9391c59f023a981244 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/652270170cad4ebe924ec05bdc90ff65 : SUCCESS in 42m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a3687fbb0cf4c188996195ace5c00ab : SUCCESS in 1h 35m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9be78099d03c405188eb3aa087dec1da : SUCCESS in 1h 16m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09665cb8aafb4487bfb83af40977fba9 : SUCCESS in 47m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7999c2fb12f49f2b0d3bf4f58519973 : SUCCESS in 1h 05m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6467a8c8aaaf42078de02291fdb1f4bd : SUCCESS in 1h 13m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62ba894937e549fb808fba41a0ab7b42 : SUCCESS in 1h 20m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5f7608ea0a943db975958e80b5a0024 : SUCCESS in 1h 31m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/363451986ffb45cdbb8522ed1c7ee1c5 : SUCCESS in 1h 00m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b75b3183ca564d3e97dbd6d99dc9933d : SUCCESS in 1h 02m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/70ea30e14b314c70abce271a18c2c611 : SUCCESS in 31m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c1ed253f8594df784ae777888eb272f : SUCCESS in 1h 29m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5edee33b8d6147b699ada4fe29c4b83b : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":10},{"id":"cf888b4505bfe1d04ec971306a7213dd5642571d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-17 11:29:03.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4932c99c5f986737fb45ddb8b850964fdd371548","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-17 11:38:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6646c147ca2ad49ad31e84701ec940215b305b4b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 03:48:37.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"cf081187e9349de58da0d7895a0e453cb0be8eec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 05:25:18.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c8b5a422a36c203e9c315246b085a6e35b381952","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 07:25:14.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"77a37df93428916419322f4cfde0769f044e0dea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 11:03:00.000000000","message":"Patch Set 10:\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7b0dd2b2af66bd7d1989c5b908521ff0121fd27f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 11:11:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1053f6cb11308e78cf9f27d5d13a452152fc9609","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 11:15:18.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/60/790060/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"eb5169f8ad2369c9b2d38e2375954545975e3ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:59:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03dbc805230a45059ba5f53934d486bd : SUCCESS in 43m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23057c754a8743d9b613884c1d32dfa9 : SUCCESS in 46m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c56d4138e12bfd62bea1a962fa32d81234443855","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 13:04:18.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 52m 02s","accounts_in_message":[],"_revision_number":11},{"id":"af1e4674d6bb2b19037b34303f086b8a4d132472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:20:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05a756f1f76b4b5e95f2ef3e3c3a09ab : SUCCESS in 45m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/065faeefe4ab413a821d525b992accfc : SUCCESS in 2h 02m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/af9af41acbd046c68c0c95bd12cd057b : SUCCESS in 1h 38m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/012e997812cb472391ad0367ad6a02a7 : SUCCESS in 1h 28m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/feeca7524ee64b309014b3ab26af5e34 : SUCCESS in 1h 51m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ed99b046004426a978b7a7d2604b99 : SUCCESS in 48m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/090f664dc4a84289a4e908813e73804b : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90473d63499e4842b9bc8949cae236bc : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/608179f3e2774d1fb072f7d63c7807ef : SUCCESS in 33m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b8f03a6b08449b0972b96d0f4025ae2 : SUCCESS in 38m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/831cc93cddb2459f97c4f851c130dbbd : SUCCESS in 25m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954aa4cdab7d4da1b9d1ae6c549de618 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ae556b1453a4a7ab1ec3661bf644fca : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7bd5a79a428047ed8b83e8d1328e9dc6 : RETRY_LIMIT in 5m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/428a6d00f63f4540b6103fbdba6e36d0 : RETRY_LIMIT in 8m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8f9d08374d174e5298eebc40b765e788 : SUCCESS in 1h 26m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae6c580e559e407993da4b9f9f020e20 : SUCCESS in 54m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ebd818c58d754d99a7d433e6093db16c : SUCCESS in 1h 07m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/20cc398ca6cb4ac7bc985bacbc1a64f9 : SUCCESS in 1h 13m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eecfc73e50d6400ebd6fb15a0030037d : SUCCESS in 1h 20m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/820376def2bc4a309475f4eba2e8bae8 : SUCCESS in 1h 39m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa22652a689a402482fed2b3d3aae91d : SUCCESS in 1h 07m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5f5a0b0371554e5c94a487a4896a65bb : FAILURE in 2h 07m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9f02028cbf6c4b608462280374409dbf : SUCCESS in 32m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bb65eb4604234f348c7ee7854715aafd : SUCCESS in 1h 33m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2a8efdf6a6a1425ab730c9fb212ef575 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":11},{"id":"76e2297f85eaede97852f64fc0d6e32b011bf62d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 13:29:27.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"523d0656afb55feae5183d0eb8e400fa3bd22b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:13:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45e1fe36c4ed48088f3eaedae37c4e38 : SUCCESS in 41m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4b934a6702f4e3ea430e74ac12e2ae8 : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dbca9e4ba68c90f2cc266abced9c275a9f6a5bee","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 15:13:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":11},{"id":"ae1c027d428effb9e27902a9cbc13a5362eafb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 15:40:04.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/592ee75c2a034ac7950f03e0d41f9809 : SUCCESS in 36m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/15ef46d267e947619d07a8a14ed320f6 : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/deb7c237c1e64a3197232c8543dbd1cb : SUCCESS in 1h 31m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60fc4dfcf7f44a838aa14ea70260041d : SUCCESS in 2h 07m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/51b9711250854476aceee2303520dade : SUCCESS in 1h 46m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec228e05722145edb1d62eca90c17393 : SUCCESS in 42m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42c4a14a20dc4543bde62081f5752797 : SUCCESS in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85dd7df321034f019d73db980a250128 : SUCCESS in 15m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ba28077e3854a258ce9c2b0fececed8 : SUCCESS in 25m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f2d36349d434a96bc90564ef47f07f7 : SUCCESS in 24m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd25ed0f0fa4b1db942eba9913745e3 : SUCCESS in 27m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e7b09551db4dbda3f5ffbd18683fd6 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/271701a2fd3a45f78e7b1607a4b0269b : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/062c4bbf0d684347a89023d1cab42e2a : RETRY_LIMIT in 5m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e31f9f3caa2949ea89e4d4d68df37a70 : RETRY_LIMIT in 11m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/810b3b2248ea4451abd01412f17c813b : SUCCESS in 1h 14m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f6ab36d444c400d9545323a4e5ecc83 : SUCCESS in 58m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b9b5f0b74d164a4d81301aa535dec445 : SUCCESS in 1h 04m 53s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2191ca75a32c41fb95c7c43bbd4b935d : SUCCESS in 56m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8808714096343f9b7fb1034fbf1f0cd : SUCCESS in 1h 14m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/215b50bb60a04178bab27a566311d82b : SUCCESS in 1h 16m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4fa0cc82d684475a93274060dc843ea : SUCCESS in 50m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6afcc0f903a44b91b75d30bbac73d04b : SUCCESS in 1h 13m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f7a976e44b464a268ef20e6235a5c17b : SUCCESS in 36m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b99265219824408ba61394084c8b7fa : SUCCESS in 1h 14m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f520160a712745fe9d675839234874e8 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":11},{"id":"06e16096894da186d2cce2a7156f6da85e2c5a7b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 16:00:03.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"784916c50f3d24a66e84ab37887dbb7881eb787e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:47:59.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c4e46d68011452d927d47f7ecbdcacf : SUCCESS in 45m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5aa7099851444b88ec508583a9e2b5f : SUCCESS in 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7dfc97287567b2d6a68e3119ab22dce4c0856668","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 17:34:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":11},{"id":"41765abc74e620966d31b4ee3198fdc56eb8d11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 18:04:43.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/109e970dde8946e39585fd7c70d9f1b3 : SUCCESS in 35m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1416d26116344ad3a260d071711f484c : SUCCESS in 1h 20m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d8f0e508d10d4a4abf13aed815891a1a : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b14d5dd57ea45f0bd743cb94809b9db : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/81598e0ca8fa4193bfd3652c117f9d22 : SUCCESS in 2h 03m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0984a91d6d943fd995ea1cc5f284b62 : SUCCESS in 36m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ec330a4e9824b878b24fabe7dd1f188 : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66af839caa264b3b8827a69411b6d47c : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60acc489805f41479e92c06a224577af : SUCCESS in 32m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/499675a5387042e0bcc00058c26a01f5 : SUCCESS in 22m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41d62f2e5da64411ac2c8db4adadbaba : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89570f12156348c0b3dc823a32eb0894 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5447db3c336a4ea19e4a012ad4db5f73 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a71ef214cebb4664b92976c729fe52d6 : RETRY_LIMIT in 11m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1b130d1f9d34eea930f2818de2c1b30 : RETRY_LIMIT in 6m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/355087c95474429e9e0863ec79a0c055 : SUCCESS in 1h 12m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2feaa877960e451e857c066fb2785c02 : SUCCESS in 58m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f56b19ee0a9c4917bd686412c1f5ca1e : SUCCESS in 57m 26s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4297ad1880e74c448621a0e1bb0e8b77 : SUCCESS in 59m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61909c614d004d42934ae4a9cf75df58 : SUCCESS in 1h 18m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bf747893ac04a5185b4420b116b79af : FAILURE in 1h 27m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e409fad4ff54895b065208b283915cc : SUCCESS in 53m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/313588ace30d42a18a1e18df99737515 : SUCCESS in 1h 07m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9a32ade7a4cf421b87b11f963d671741 : SUCCESS in 31m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/277f9ef9b0644697bad071b0cee2a09e : SUCCESS in 1h 14m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/69cdeef979f84920b5c161780e205445 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":11},{"id":"7f45296b3602a2a1b7694618c2b0cc8a61bd7c64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 18:30:37.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a72d9a98c0aee6b333be713207d8937dc51c6132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 19:16:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c73e6b75f78340f084415e02afcfac39 : SUCCESS in 44m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f14d3fd22c444c87b9ac0850bbffe5fa : SUCCESS in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fa5a2c8517306ed7d6ffa45320046c7884145807","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 20:29:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 58m 33s","accounts_in_message":[],"_revision_number":11},{"id":"25a217750b848f1ed837e2b72bc7f92fa9af3af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 21:07:42.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6dc7bf7f1eb497c90c30da30613e94d : SUCCESS in 1h 01m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/35af5bcab3d84a79977110fbe9c27bca : FAILURE in 2h 20m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50208dd8fa1949b2ad636baa17c1880a : FAILURE in 2h 00m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15fe731e7db7406f83d1a6f353ce7557 : SUCCESS in 1h 49m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f2e6a697c93044808215db7fbcc7eda6 : SUCCESS in 2h 35m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1072d8e04f6e42868520328b7153ce4b : SUCCESS in 50m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9bfac589c6349aa917440f25eab173c : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c87b1c72504982a196e58f6da58721 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5471a03400c14d2ba8702a374006b45e : SUCCESS in 35m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/938976a8c90945d297ab28eddb52f099 : SUCCESS in 39m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05ad0b08d43a4292b59981fa2fd5b51f : SUCCESS in 33m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6985720a657b437eb28f8efd7303b761 : FAILURE in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd9770b41c3e4e95a44d984cd26a39fc : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef624dd368ca4c0b937d01b8325326c9 : SUCCESS in 41m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7064ccffc262446fb59b018fa920bf39 : TIMED_OUT in 2h 14m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c2bb77eab7524c0fb0933a6499c01888 : SUCCESS in 1h 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c89666bc4a9b4af5aea46b659390655b : SUCCESS in 1h 00m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/899f59bf5adc430f9adb8e007643269a : SUCCESS in 58m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f40b7f940d0248f384495fc0c5feba2b : SUCCESS in 1h 02m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86c463cfe9f741bd84c25a5e295aae1a : FAILURE in 1h 04m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2795c89e9d8f4fb8a02c929243a028ea : SUCCESS in 1h 06m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7808cbee6f964ce1b5c6f84525e599f2 : SUCCESS in 43m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6d1af2e0a1c54b539051019624336f5a : SUCCESS in 51m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/567e241dc8a54c5d906e54d4a1a749f5 : SUCCESS in 25m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c929e0f9e3a4459aa3d912628898b48 : SUCCESS in 1h 49m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8f6427d6d30446389eccdeea8d149842 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":11},{"id":"d038a1f9b5e757a1a91a20adcf1187d443bb2470","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 21:31:21.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ba75a636273fadf0eccb6bb31bbb035f49367511","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 22:15:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6aadf03904024c398e8d773c4ac25e45 : SUCCESS in 33m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1dcaf929aa143158e4f37c5d8b4cd3d : SUCCESS in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"54803646298cf45ad3a5dc63c0f6a81bf314aee3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 23:22:03.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":11},{"id":"72e725e490fe00cc279c481770d6681359242304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 23:40:27.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e156021e6b69460da4f5f3a49e5c022b : SUCCESS in 47m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/beb0c5b18b5146c58d4c632bb77e89e4 : SUCCESS in 1h 31m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e33c2d3b040c4d1f89f580342cdf5b31 : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/349f271e5f9e4df3a73662bde13646d7 : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/73a3fd8ff6814c6e8005857dccb869f9 : SUCCESS in 2h 08m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93eef05c57ae4e9c9c64533881cb6fff : SUCCESS in 30m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47e61a372c224941a4a8401075511168 : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e13ed7c043a4c2dbbd4d594a67db558 : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f3e6d4d30148e88f954a52243d317c : SUCCESS in 37m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e90b61e1ed6644189f82a2bca2df54ed : SUCCESS in 25m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20a9d4cdf8624ee4ad6fcbacb7a98c2b : SUCCESS in 35m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81789d380334a0aa6b4a19335aa0ba4 : FAILURE in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb11a97a19f4043ad40e398b0916747 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2582bf72f0a0414abb13670c807d8593 : SUCCESS in 36m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e551e21331ed4f3fb09b1d4fe2289dbd : SUCCESS in 1h 23m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6ead4633b739410f854a5a0b2d7357ad : SUCCESS in 1h 09m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6bd4b3976064f118639720186df24d4 : SUCCESS in 56m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19994938f6a1445a8930ee85ce031060 : SUCCESS in 53m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d4e9a628912f45afaf31ce75422d6588 : SUCCESS in 1h 06m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/decd4ff2a1884289acd651ea48176ad0 : SUCCESS in 1h 15m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9ff3322ad204f42b09d60956fcea933 : SUCCESS in 1h 36m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eec7b82089104071b2e26bbd5085f850 : SUCCESS in 52m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce2f1a0a865f41e3957790baf64124b0 : SUCCESS in 1h 11m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3d8dc6cb2d284250b52500711f39eafb : SUCCESS in 29m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b79acd29b20a4946811ee5dbc3e14343 : SUCCESS in 1h 13m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3a9aeaab2da14d2cb32f75b4d2510475 : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":11},{"id":"1200df6861a5d9318a0eec19351b2e1efeb00ec8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 00:01:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"13caef2cb71399ec4015c1e6e1afc1defc1cdc4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 00:48:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad9dcfd3c2c5498b9005652fd1764f2e : SUCCESS in 44m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc7d90629d4d4330af36930caaa0c968 : SUCCESS in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d49176956b73974fe056244e72da92138edf1f1c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 01:51:54.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 49m 23s","accounts_in_message":[],"_revision_number":11},{"id":"b3576e7d6708b0b8fda55b8cc2bf44d833341d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 02:50:31.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a14a2e7b80f460e8f4451048f16c1d2 : SUCCESS in 43m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1d2a4fd4685149829c6547eb3418e782 : SUCCESS in 1h 30m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bb0ec996e03d45379534a9fbe776d5cf : SUCCESS in 1h 48m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4ee807a52e341878788d105a7f653aa : SUCCESS in 1h 24m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2cb91d80888e4d12a4d4420156227651 : SUCCESS in 2h 47m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78f08664d4324a01bd027c8f0ddb3fa0 : SUCCESS in 33m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b9f1a3cd73d458283937543f9547ad6 : SUCCESS in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8ae510937d431b8e7c67e8ee0ae9ce : SUCCESS in 12m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c2f85dc5e444f6f9aa0260f86455d0b : SUCCESS in 26m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34d79706f61946b09da363900c98dd87 : SUCCESS in 28m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7af3330de44f444abd922a55ee4daea0 : SUCCESS in 25m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f55643199a54c1fb8ee092b3c4840cf : FAILURE in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/349e605ef6514ff8b68e6afc3841e16f : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa6e351e4be94c79b8824b8962314434 : SUCCESS in 43m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd89f1cc576b41938f6d1f10bc648d94 : SUCCESS in 1h 21m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9c5889e6d5cd4bb19c73378aaf0e2c3d : SUCCESS in 1h 12m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/603ce25f99714078811f0d7255e8b0a1 : SUCCESS in 51m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a8e6cce300e64f679417c3d37d886d7d : SUCCESS in 59m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a7d989d0d94948549f0f7a25ac8290f6 : SUCCESS in 1h 22m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6183caf0a02042e18a4ee31fff643704 : SUCCESS in 1h 18m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eeda730d38ed4940a7e32ec3a14ccb42 : SUCCESS in 1h 26m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7b426115ac244d4b52a5e5acb20846f : SUCCESS in 46m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/70e7d39a977d4d7c8bf295af60a85b44 : SUCCESS in 51m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a9811fe781e640c4af33cabc0cbd3f45 : SUCCESS in 40m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e187868f3ad24623916578a49d257c7f : SUCCESS in 1h 18m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/125b07e665e146769fd9e79fd95cdf66 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":11},{"id":"b26caaf5d160eb6dd4241aec243d44926b30d724","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 03:02:37.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"024603b440cbcc1fb05e3a33f3fbb19b9a2e40fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 03:49:37.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/844bdeae17844f0dacc1caf06228cb93 : SUCCESS in 44m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbf1e9ea1b2f46d5a5e109521bb6a5c0 : SUCCESS in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6795fe3be809f49e9c7cfde326fc81dd3071db0c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 04:34:31.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":11},{"id":"d85fc7cf410db204ed25cd84f127dd73e683354c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 05:20:00.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df9c0823b7bd4f389537a155c8c06099 : SUCCESS in 38m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b12d300306c446ab96b320cad5a70494 : SUCCESS in 1h 58m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/66eba58e4ed845249197d0111e216e79 : SUCCESS in 1h 35m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6538e810c2734c60822d0dc2134780bb : SUCCESS in 1h 16m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ba3eba93157848bb8a6690b461f03dc5 : SUCCESS in 2h 13m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9e3ad62478f4c9aad05a6b3a8042eab : SUCCESS in 43m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09dfc0f92cc04a328f41b1245a56c307 : SUCCESS in 29m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06de6fd97f2c4e72bbfd55c5b7b49ecc : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cffea757f2ec42ef9e8cffe8ee26073f : SUCCESS in 31m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73cfe79ae86d4ca4a6c839d3f7633798 : SUCCESS in 24m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2a35e73ec94f6cb1c903efbddffa31 : SUCCESS in 31m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1c950f7df94fcdbe112cf2974cc931 : FAILURE in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c81035d503c2482f9dc531a8362dc538 : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2c3788ce4834674a02ca750b774a8a2 : SUCCESS in 42m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/854163d1203248c9a2f27d8d364d8957 : SUCCESS in 1h 22m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9cfb1530a6ba42a49660b2b778e80e51 : SUCCESS in 1h 15m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84af893057004325bde836634f8b6310 : SUCCESS in 1h 07m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cddf1051ae9c450e9c2e847cfefba133 : SUCCESS in 50m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/040e96b6325e4854abd9a37217e59fb0 : SUCCESS in 1h 15m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cd2bfcd500d641dcbeb78d2a33d52834 : SUCCESS in 1h 25m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1dd549e903004b7fbb1dc7140c08f3d9 : SUCCESS in 1h 45m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/348543e1766e4dd58a66a275453b117d : SUCCESS in 51m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dae5b7e52d6b48919e3653de6f3e788a : SUCCESS in 56m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ad044631fba1463bab342f953cb309ca : SUCCESS in 33m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f3f6a65b89424127876c1768d6a2b030 : SUCCESS in 1h 28m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/535363095146487194a358d4cb28c7e4 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":11},{"id":"967fdd0400e942a1669a1a70c01657f2c13570c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 05:33:12.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"8e352da5d2eb0e1c132e3b8ffc18f03b4ef98441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 06:24:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38827cf870c5455384f7e3e718d7b4fe : SUCCESS in 49m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/152c288b93e04c57ba293bd9a825ece2 : SUCCESS in 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6bfadf4d4e717b98e947a594567f2c60297aa2cf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 07:09:02.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 35m 12s","accounts_in_message":[],"_revision_number":11},{"id":"121dfe9ad61cdcbf5e4f90da23fda277e693dcc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:49:53.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a3d6d4013d8d42c39ceb82edc465d388 : SUCCESS in 1h 15m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb5cc84f2df847bbaf4302f254387a49 : SUCCESS in 1h 35m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5adf3c7a225b4451bea0ec1779080fdc : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74489dd389474a1bab2f8781bbb4849a : SUCCESS in 1h 55m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/47f6a9abb0674136981220d763296b8a : SUCCESS in 2h 15m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7876f3db8690479d96ee23db3234dec4 : SUCCESS in 37m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ce082da03c846ec92edb2d467597071 : SUCCESS in 52m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0090a5743541a6a6a78d8787fcdf7e : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/195b6d40f5db4f4e840052ec1d9ace03 : SUCCESS in 51m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6a3660d44c54aeea2e143ea299f3ad0 : SUCCESS in 26m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/822b1e3647f140acb2b2a4c8587af430 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc3aa072c06460d8ee3dc2eb5e975ee : FAILURE in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4182c74dd35f4bb3adafdc57a82edd64 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd2a984181b04094a6ec1bb3e43282ec : SUCCESS in 54m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f013801c958f406d8fe0319d593a92d8 : SUCCESS in 1h 22m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d981c2b095ce4b1b83c4a8db6b3c6c7e : SUCCESS in 1h 18m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae425e194c4a4bbfbdc37e685753bef6 : SUCCESS in 1h 06m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d79e978ce454a4898fafa0411de302c : SUCCESS in 56m 07s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/520b646cd696498a8a0219a22ca9a4bd : SUCCESS in 1h 08m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b4cadbf5ff434d7f9d4744f42dbe41a1 : SUCCESS in 1h 23m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b85f852e12ab46f9985edede7fe84abc : SUCCESS in 1h 29m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a70bb30241374d9b8f2d64fa048cbfec : SUCCESS in 57m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40c023e5b00141a3a582e2d9d8a36c06 : SUCCESS in 1h 00m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c7ee7d0b6244901a2bb904ad1ca00f4 : SUCCESS in 27m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf7df79ee47440ec97e798e66cf11a7f : SUCCESS in 1h 25m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f79491814c7644278b3077ec567623f5 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":11},{"id":"12064cccf1cc3df07fe457a36fe745dd320644a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 08:03:48.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f08af8bff111151f3b98cff2ad2ddf208a7c698f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 09:07:04.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26d75a413e7048369b79fc4a73695c50 : TIMED_OUT in 1h 01m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2235aeb3b873472c9ab2146910c95b24 : TIMED_OUT in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dda250f9a92571df6ff6ddc9c767c263407e35a5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 09:39:25.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":11},{"id":"05d6d3306466cc11e1045144b7f1294b2e97e6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 10:29:45.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0f44ab5721a941d987341c584db9a04d : SUCCESS in 45m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d50287c64b694ec4bd4187df3a573895 : SUCCESS in 2h 21m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b00c1a9bccb490a90f896645132c7af : SUCCESS in 1h 53m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/499fe31c3f3f4295b53e4e594dcc6828 : SUCCESS in 1h 28m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3ea245b5b3bf450f84dcf77ce2ac6d25 : SUCCESS in 1h 47m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdbe1c9b94784915b03b44a112da8b3e : SUCCESS in 40m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00b21ace2536460181a1d1b81fe31a5d : SUCCESS in 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bace7517700849048d9bdcbb2b375618 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04951ee9502349da92a7a3cc3ed67ba0 : SUCCESS in 30m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058452f843e243f39d0caddbc94af79c : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edbfb05372ff4391b7c269c1d580fc40 : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79f52aab613410c8148e9225dc60bbd : FAILURE in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36a7a1195dc84848af44c23685f7fe9b : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e01facc31a274d74aa6e08f84339d07b : SUCCESS in 48m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/782c50547d5949ee9618c20c77e95448 : SUCCESS in 1h 23m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bfb6d38f743a4772bccc9a6997eb0166 : SUCCESS in 1h 18m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be9ae7b8a87a4b0e9e01b13f9b1e507b : SUCCESS in 59m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4623900e3062408f9dcec373dd26625f : SUCCESS in 1h 02m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ca9bb6b3a5a7470da92b41fe84e2c6fe : SUCCESS in 1h 15m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4bf92b7e59a14d69b3b8cf2832d7c2dc : FAILURE in 1h 10m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43722bebca544e67b287cfe46ef0ed6f : SUCCESS in 1h 36m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e747fb89773345ab992962303d2e4574 : SUCCESS in 59m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0397edf069e241d8baa8030652c3f040 : SUCCESS in 1h 09m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4d98259627f84e619cab02bedd698619 : SUCCESS in 39m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd34751cff3745bcb0facf1583318da8 : SUCCESS in 1h 28m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9ceb21e37ca84c118aeae828aab37b6f : SUCCESS in 1h 30m 46s","accounts_in_message":[],"_revision_number":11},{"id":"5850068499a530f71eb680b6e49b91144bbae16f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 10:34:24.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"6d4ab6ad46db01bf478b234940a9bfdfa6fb55ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 11:22:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8adc83e2d3d45c8a74c1f06235153a4 : SUCCESS in 46m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f1cdd2af0624b93b8e22dce6889941f : SUCCESS in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fab2590ae3b658ace7340a156680becdf7075429","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 12:11:22.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : FAILURE in 1h 36m 16s","accounts_in_message":[],"_revision_number":11},{"id":"e3a5bd38d5c63eca70ae04447e39b1003fcfb6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 13:00:10.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b0a01f4ca584487a8862ff436f7a1808 : SUCCESS in 44m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ed9f27db65b843e3897e28e8681ecd1a : SUCCESS in 2h 19m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/63c8e1f194844baa99ad2ed63a009dd6 : SUCCESS in 2h 20m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/608fb940c0f94250949397251a94124b : SUCCESS in 1h 29m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e9bcb9f6de8649baa9167dd30e00cfc8 : SUCCESS in 2h 13m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99bcf0a0c79646fcba2d808be69ee357 : SUCCESS in 31m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b863ea12b04945db86cc8a9e08788aef : SUCCESS in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927691df6e584b6b8839b42341b8559f : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20a90165fe234533bfb8a96ec7bd1a1b : SUCCESS in 19m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c89552e5ab154a8abb5bbd5b2fe3b14b : SUCCESS in 18m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad30c44d7e274450bc8ce69f8f0399a0 : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ffe07adadc42c1ac0d92a82aeacabc : FAILURE in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beea7aeaff454c4fb3b4116b934483b4 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad577f8fc8b344b897dff12fad40e11c : SUCCESS in 40m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6e517a9b6a64bd0a6c3333008ef5525 : SUCCESS in 1h 20m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/255bd2a64a434e2287631dfd67cba107 : SUCCESS in 1h 27m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3112bca15fc4b0883ab8f8ec16c41f5 : SUCCESS in 59m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f600f542f82d4980812547ec8faf51ec : SUCCESS in 49m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/118c70facb57484eb444f151ae526bf8 : SUCCESS in 1h 01m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a5b7a71150642a3aec63674c8c8a411 : SUCCESS in 1h 22m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9745fe20e76e4dd49291035dd2ba353b : FAILURE in 1h 18m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dcd35ff48264d6cada1543cf89933df : SUCCESS in 51m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d34f2f4d13db46faa1bec7b9806cdf97 : SUCCESS in 56m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc704de1297f4c14ba575ebc22e6286f : SUCCESS in 31m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74944aadfad54f03a4e821e67adb923b : SUCCESS in 56m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c1c464c7c9314e368faa8bb8b70d1fc9 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":11},{"id":"224f29ee551fe81f3053f0e8beac00e2a93e87e0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 13:04:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ce0f07e8a67b50cef01256c72a38bd43770eda6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 13:57:10.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6afc33d4d2c442f85e4a28fdc5d11bb : SUCCESS in 49m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5f1b1799405418e8a6bee90405dda21 : SUCCESS in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b420bf6ad56d259bee3a3d1cbb642e906dec3f8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 14:49:07.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":11},{"id":"bb3f1a2c73aa8648c2c03003ffb733c8fec24e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 15:28:27.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eaff62353498425abdb5ebc5a251e642 : SUCCESS in 42m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/88c0ab4c24494e6a998d129be65d4057 : SUCCESS in 2h 21m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/01f01a8862f64fcf9bdefe7803f4f1ee : SUCCESS in 2h 02m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82ce9ca3d1e14376ae50703af25c0088 : SUCCESS in 1h 41m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9b2de8d351ee42d699117193266f5b33 : SUCCESS in 1h 54m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47e01b6de8bf486c8b00c5283bf38abd : SUCCESS in 53m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8f44721e0c54706ac4803c9dccbccbe : TIMED_OUT in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed32676417b4a9bb709792a7951e6dc : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/282ed335525b4e5e8adb8430cbbc4d2e : SUCCESS in 29m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ed40857e4e942a897e12ae27543e2dd : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c47889fa91a43b8a2690f84b626f22c : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a1be667df1470e8bd912c7bd46200e : FAILURE in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfe16560a6ac4fb7b7f8ff80443d975b : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24422649bae74b4eb878a5b233a1e302 : FAILURE in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b59a5fc4eaab409f954af599d1033682 : SUCCESS in 1h 15m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f88130557f204587b715bbe9489d7d7c : SUCCESS in 1h 09m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ff88806e4ad47fc8d227fda10c2f032 : SUCCESS in 54m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45b60e5145c34ff5b4edee674f0e3ab3 : SUCCESS in 54m 08s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30285dadd4684147a07e5bef52083b0d : SUCCESS in 55m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eca248da8a5f4e7ba49b136181bf56aa : FAILURE in 59m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3869cb3ab784d53b474c26cea60ddd6 : SUCCESS in 1h 17m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d35cb175da2a42d2825f880633dd68c0 : SUCCESS in 57m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6ea6e9de5cb94a75b2f434ecf4b10b50 : SUCCESS in 1h 24m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0a584e58c0164e6c86e99c6b10793d45 : SUCCESS in 29m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8227896a21c04b5ab7154e7ffc176134 : SUCCESS in 1h 04m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8a8e28b9da5944a3aad42946c03508fa : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":11},{"id":"668da6cc72dfcde11bbb0a0ce5a29bff01357821","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 15:35:37.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"5a03788f566c811d8808e7a8f57717850c11052c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:21:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77504fda75aa4b8b86584999e6ac0372 : SUCCESS in 45m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb76895510be44ce8f5f0f7c53f7f76e : SUCCESS in 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"149a0f9a8fefc5363c1f788ae47074b9b6e48b67","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 17:21:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":11},{"id":"b94cbcfb38f1089a54fdacf724dbaedab2ab5b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:02:10.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/236344bd035049a39a364d6abd22f2c4 : SUCCESS in 42m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4b501f4292a945cdac1acf9f0f6d9273 : SUCCESS in 2h 07m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/54b9baf7f0a642399c6dc1137bda4fcf : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33639a7960ae4f60b607b76e8c6895e0 : SUCCESS in 1h 12m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3fd9abc48d834cf28463512a95b65818 : SUCCESS in 1h 40m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a99ac00775540df9e00cf826afc4fb3 : SUCCESS in 34m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3de65d9ec9224c79968224ed20525bf7 : SUCCESS in 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f1feb091a74ee790e6dbeab212b41d : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29668299a15c4067a37c37ff1d111ca2 : SUCCESS in 25m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8841e242720a49d9a6e48e2c1d1030b0 : SUCCESS in 25m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50862299d79449eb14e1dc782430f14 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82615eaec42e40fa849b1116cde29b34 : FAILURE in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3556fb8cfdf4059b08ea61d7badd15e : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcd9b1405d4747f5a3427321d42e5ffb : SUCCESS in 50m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/426c837acdff4cfdae36fed272350825 : SUCCESS in 1h 21m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9498b9148ad54a84a219a977bc118790 : SUCCESS in 1h 18m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ae5419f6d204b35b43cf545ee1726fa : SUCCESS in 59m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f2a01fe470234680b8476be06c44bf0d : SUCCESS in 52m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/450ad13e54be49f49f43a4da9376aa75 : SUCCESS in 1h 20m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fc1c16510e054f93b3716bc5fcbd66dd : SUCCESS in 1h 17m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bd1f28702754579811896beae93d7e6 : FAILURE in 2h 25m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a04c20e04b3c4f5cb436bae3aa9b57bd : SUCCESS in 56m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/65b2e5f252964e37a11cab63347d753f : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b03122184a041268ffadefd9db9ea1e : SUCCESS in 30m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1fc082ed3d59487da31aa37ec9057eff : TIMED_OUT in 2h 13m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5ca1404aff8a42d3bc3252258d41f0c4 : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":11},{"id":"82abfcecdbe87f76bce2f9871c7d3dc4806f7ce4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 18:06:14.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c7d115fbdb44d5b9c68224818c3693d899b07b84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:52:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85eb170093034e17bc638bed1877b8e0 : SUCCESS in 44m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dde7481b49d3450ca6a0c18ae651cfa6 : SUCCESS in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7b3d7343648e6405f189bb6f3875be9d3d120ad0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 19:47:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 40m 54s","accounts_in_message":[],"_revision_number":11},{"id":"9c1656f7e7587fe97996f03e14c3c224d09e4aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:02:18.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/715812f20f1f4f43a146c466d16814df : SUCCESS in 50m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7ae9cd5519d4563971831196c33bdbc : POST_FAILURE in 2h 54m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6853efdb3e2b407587638817cd708822 : POST_FAILURE in 2h 39m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/848943bda47b402c8e8a467d807811d1 : SUCCESS in 1h 50m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4a1f39b4452842a4af9f38d711512b47 : SUCCESS in 2h 02m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4375a30ab7447f4a878bdafd4cfbaf6 : SUCCESS in 38m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2565a69077094807abc7ced47032a8a7 : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ff87a94c154971a191ee5d65f25499 : SUCCESS in 13m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9e9eeb2ddc44d0182a147754f4fa4f4 : SUCCESS in 44m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a77183cf8774141a62dedfee05c2e97 : SUCCESS in 30m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fef41e4019c4a2cb5c7aa1a8ba8cfa7 : SUCCESS in 30m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c404c43921324164afea872c8444f6ef : FAILURE in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/755809823f92496bb30b90524b5ec9f9 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/692bfdc32ee54ec1bba94a4e62891e1e : SUCCESS in 52m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8203a27bcc644cc8dc8ad4ac7f527eb : SUCCESS in 1h 24m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5c279c1474904f83b48c33d4485d4da9 : SUCCESS in 1h 18m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb72ac1f622848838242888f1c5273fb : SUCCESS in 49m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e95cc6c48084d5caf7b97f8b3d4e01a : SUCCESS in 59m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/632dca98d91b454a82c9558de960bca3 : SUCCESS in 1h 11m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17d2326ef8014d658bc54769678a1fa2 : SUCCESS in 1h 19m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fd28c27e1a84d3ba2fe4bc33b242496 : SUCCESS in 1h 33m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49f2fad8e0a148648ab52a2d53add564 : SUCCESS in 1h 10m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec716ce756d24d92ab1776e804a3f03f : SUCCESS in 55m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a4629de7fb044902b540389b5dc2824b : SUCCESS in 26m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c0c73e3c0b74f459f5d17b5f841f847 : SUCCESS in 1h 42m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83931e212b3b4211800ae86416a3c79e : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":11},{"id":"a1966e25ad56d9ebe9bed3e3b361001dbb676ea7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 21:06:51.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"fca1b91099428423be19c3c3f5292bd8701a2b55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:51:09.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3e93d10246c4376a0db2970d75c239b : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ce2d3b7f7e44d5ea703be8d73fb686c : SUCCESS in 41m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c0ea9bce651e559d44ec66096818b772ed61274e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 22:57:17.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 49m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e9674666db79246c76259806691435464625c0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 23:22:58.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/41b67829ae38440499d4e74100879ce6 : SUCCESS in 37m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6893e1d66d8d48dd8c5b568a810e4809 : SUCCESS in 1h 48m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/38fff3ed651b49cca896c081f87aafc9 : SUCCESS in 1h 34m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb2e2e9ab2c7488a8e1201d584838af3 : SUCCESS in 1h 46m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fca748d0ed7345c7b74c3d4e30b75be5 : SUCCESS in 2h 08m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd06c22ce0e645c9913edb49f3611f98 : SUCCESS in 44m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/399e510c6181446bad5190c70f4c787a : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e47e048e0ca4ebda4533831c886ba9d : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0068392c6c42ccb793a8bacec7e697 : SUCCESS in 24m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8ed17582df14c31a9fb352dc39e84c6 : SUCCESS in 26m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c65edffd59a45e3a9d0a70eef5f8e1a : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a211eeb7c34ff6a36722248849211b : FAILURE in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3927e2688fd47139a55b24b2d6206c3 : SUCCESS in 8m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91ee946c3268405c87cbb564091b1c97 : SUCCESS in 44m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66176d4e5c8d4763a34e6bad1c4b9390 : TIMED_OUT in 2h 14m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a8ef902eb0ff44a3b658d7993bdfe8d0 : SUCCESS in 1h 11m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9288a25477c4315a405011c84ad6ced : SUCCESS in 56m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6f2262bce9974b33be9271b831bc4d3c : SUCCESS in 1h 00m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a2939015cf848f4adfd342befa9ca9a : POST_FAILURE in 1h 35m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/25bdd839a8d24e119786207b576a288f : SUCCESS in 1h 21m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07dc587334d246769d89e8df962805c8 : SUCCESS in 1h 14m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12aa74745db148179672d692dd479926 : SUCCESS in 1h 00m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c06a3156f8614c1aba65a7c5227df6f6 : SUCCESS in 47m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/04d80df761f7434abf0225ccf7ada721 : SUCCESS in 26m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fb29ff1bd3a4eb998a655fa6900c57c : SUCCESS in 1h 14m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/72d612b7fe034c4c889b354ae8449427 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":11},{"id":"9e456c7f7b3135b105c78feca83cd627bf2101a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 23:37:28.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"1f008c1a58604c5ca865784626809ae011cbba7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 00:23:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/699e37b20d77411482e66066c4142037 : SUCCESS in 43m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4466abf3b57e43b5b9fee841154ee099 : SUCCESS in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d05723d0482aec3abd1eb359251fb7be3097a2f3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 01:10:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 32m 20s","accounts_in_message":[],"_revision_number":11},{"id":"6df6adb4742304821f93f1520f04fe52b8e7ea67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:16:50.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db55b6010a794629ad4b9b0406b0433a : SUCCESS in 1h 02m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/86d47d87672043819b922432c4835d37 : SUCCESS in 1h 24m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ef9af8502e5c41408d20d683e11d5f08 : SUCCESS in 2h 25m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/804103224b3a4b208f4a45a0ef3b18cb : SUCCESS in 2h 01m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bc16131ab52744468309ecf8e8727d03 : SUCCESS in 2h 38m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d115f58c2e5342439bbcce4af272111f : SUCCESS in 1h 00m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21f8d583196240648172da25717bb2b2 : SUCCESS in 50m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad600004f894b2a9f1ee5ad349015d6 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32b90936559f4c55ae4357d6df3633cb : SUCCESS in 21m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca8f126304e4d65875016e44007717e : SUCCESS in 23m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5fde268ee5245cb817f921effb8a6d9 : SUCCESS in 30m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6a50bfe6154d868e4d24f754c49f48 : FAILURE in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7552422f8ba14d32bf232c92e813daad : SUCCESS in 6m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/298617571e264b298dd4d845ae078456 : SUCCESS in 1h 06m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68adc45ea484431688cafaafdc98abd6 : SUCCESS in 1h 39m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3811464115c14bf9863cab93839f01e0 : POST_FAILURE in 1h 20m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26bf519cd78c49b7a876b3f21b862490 : SUCCESS in 1h 05m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18b65e59fe5c4ee19ffa1c52c8e0a3f6 : SUCCESS in 50m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7f3ef46c8a0743b8b791eeab1e5aef42 : SUCCESS in 54m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dffc10fbe9b747dfbccf959934f07d4e : SUCCESS in 1h 25m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b28e53409ea24c6d95c9075f6807a68f : SUCCESS in 1h 46m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abff5782b6ab45ceb57e861e485d37ea : SUCCESS in 56m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e7e9981d0edb466ca9072af3214c93bd : SUCCESS in 1h 24m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7531a9b6b5f04f76a750dba46e3ad40b : SUCCESS in 25m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3393451d228e468d8e1bab98c0556878 : SUCCESS in 1h 00m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cb4b9e8b44ca4ab490693f9675466642 : SUCCESS in 1h 38m 10s","accounts_in_message":[],"_revision_number":11},{"id":"bc99a41053c6b5616e7f7e5a4b69305045f4ca07","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-20 02:38:11.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"87bdc03cba1314b1c877d21c739e5065aaf13b74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 03:22:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b7498518fab4b11917f2d7641e0bb10 : SUCCESS in 42m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8464fd95fa1643d283cfd8dfda1f04df : SUCCESS in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"55be9381444951ed0d9ab24ecfa06c04c80074f7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 04:33:06.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : FAILURE in 1h 54m 12s","accounts_in_message":[],"_revision_number":11},{"id":"135beedecffd238bfce72751d80cd9e7a49c1842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 04:53:30.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4edc6882a8d943c29926b45d7f66040c : SUCCESS in 1h 02m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/acc11f465f2a476c896889352ee4eae4 : SUCCESS in 2h 09m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/58277e9d789747de9eb9b780950e6f1e : SUCCESS in 1h 51m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec2afe7a0e364e53a9c8b6ec371ff67b : SUCCESS in 1h 38m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f7507c29dfeb43b786f85fde90042c8c : SUCCESS in 2h 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/114acde8351e44d0a7ec4d1c1ce261a0 : SUCCESS in 42m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cbf15f0793d468db21cdbbfb8109dd2 : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131d8cd83c524835bfc302cc3ccf578c : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7604f3ae5b4049dfa773afcac9f29c3b : SUCCESS in 22m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28e8099042bc4f3e9258c2b801868760 : SUCCESS in 32m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1dfda352f5543f294acf277b5267c1f : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9084f7a464e34298a0156fec450ad5fb : FAILURE in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a90719631024a3cb852b93f6b6de815 : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/507202bb1359443cb2b9538759bb0e66 : FAILURE in 46m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0355a7b1a40a40518f9491fbebf7e146 : SUCCESS in 1h 22m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/771900a781b3405dade8b5cc3e3fb29c : SUCCESS in 1h 08m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93b6efe7b2d04f399afa9c545ce5d47e : SUCCESS in 49m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53306ab93aa24d09889810c5b9b960c8 : SUCCESS in 51m 09s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a62e6a5305ee4ccb96ff8502f4a38821 : SUCCESS in 1h 39m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea9b97c0ea904e04bb82d32035fcd12a : SUCCESS in 1h 23m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b498433f9974f60a2ea133aa051fdbc : SUCCESS in 2h 06m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ebc90d5a3b4e958aaa6f3f963b8df2 : SUCCESS in 49m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c211c6f46ca64c8ea86625840290c996 : SUCCESS in 1h 05m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc15c05c05d34f77bd9e8b5c5f39ef1c : SUCCESS in 33m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b15ac6af4774b29965be3bea21d0955 : SUCCESS in 1h 38m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c04dcd9148014ae7bc386bf19788b804 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":11},{"id":"76f1f04b3e4becff98e76d337f1a0276ba01982e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-20 05:08:39.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"8fc76b6af6c875159059aaedb269f2246d860bae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 05:54:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/125f0d5d29654a6b9af53f3f4e32dcfe : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0edffe1dbac94f138c7cc9706f28733b : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c04ee291c3e61bcea0079d729d1ad5805d9a653f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 06:42:40.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/11 : SUCCESS in 1h 33m 15s","accounts_in_message":[],"_revision_number":11},{"id":"85227c343a0137aa73accb69f318244c30373c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:37:32.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f95e783d75664b4e9c1a98183dc4fed6 : SUCCESS in 38m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5d2d787d50ef4b10bc7b7c727717a002 : FAILURE in 2h 27m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/27466a795b4e4c1a97552beaa8494d69 : SUCCESS in 1h 56m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed755cc6064347c29be1451f9d132ead : SUCCESS in 1h 13m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a688300a3e624c49ad88793ba809cfe8 : SUCCESS in 2h 00m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/742532e225544ef5b8d70a6cfaa563ce : SUCCESS in 35m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2271eda55ac453ca17d441bbdea3b33 : SUCCESS in 57m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35218fbacf064be8be29c9f74ae768c6 : SUCCESS in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0385b0c5a5af4a4f80693ea207a2127f : SUCCESS in 42m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f0dbbb6f5fa4aab8f1d416cf0fe0579 : SUCCESS in 22m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e50303440b9040c3a9b6a5af971e288b : TIMED_OUT in 41m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6902a176fa9e4c72a903d03841b05cdb : FAILURE in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba45272f0bc48f38fb011177ca0605b : SUCCESS in 14m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a8996a674da446e990c4c581d31d5ed : SUCCESS in 41m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fafe1e5bebe64896808c0e937c7b8a33 : TIMED_OUT in 2h 15m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f984c2ac6474446ab0fd3a17259155c : SUCCESS in 1h 18m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de3b7e13cfdd4d20973e091da16b9e37 : FAILURE in 1h 19m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c408a590bb9b4a19a5ff197f64bd9dc9 : SUCCESS in 58m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4e46a335481946a39bfc43b2ed76f546 : SUCCESS in 1h 50m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/876c81015bf34056a3602077e43ee2fd : SUCCESS in 1h 24m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fbaf3f7dbf04be5a19e4b44b6eeea11 : SUCCESS in 1h 50m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcb88676f39a4ba681b02a1221d9db66 : SUCCESS in 53m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2c61bed93e7140a194648b0553ffa494 : SUCCESS in 53m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/582863a60a28405e84d07dd48fe8f944 : SUCCESS in 26m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8031a2bdcfa1405da6ab782dc634adba : SUCCESS in 1h 18m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/730eb9bba3e84543897888b66039480a : SUCCESS in 1h 15m 01s","accounts_in_message":[],"_revision_number":11},{"id":"ecf7d61331f8973e35b77fc62b22289cd458ba04","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-20 07:39:16.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"63ae44984023fd7657fb3faf254ee39e7f252fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-20 07:55:59.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"c45cfcbd83c70af87a91606ad7ac19049425bfdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:36:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82b8a4e5cde546bc8f500037639cac1a : SUCCESS in 32m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/244fb78cfc03431ebde5befdf3ff421e : SUCCESS in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cff7b95f78cd927a309dcf6adfe39b3a3b44e8e8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 09:35:08.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790060/12 : FAILURE in 1h 38m 11s","accounts_in_message":[],"_revision_number":12},{"id":"12fcc4f58d8e320a39e56aff258f1b7929b12961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:08:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a970f64478bd4b82be0257b9819a24f9 : SUCCESS in 41m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/daef228cc52543209cab1634320dcd84 : SUCCESS in 2h 09m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a5caf2d497de47bc8fa6db587f919e32 : SUCCESS in 1h 58m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dffa6f75e7f545d39ac2f81006030e0a : SUCCESS in 2h 09m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d22acbafb045453b95c332294e26c154 : SUCCESS in 1h 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a188b59faa914f74a0375a7d73e10fec : SUCCESS in 23m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c486f2e7f9444ca8335794b753c9fd0 : SUCCESS in 21m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd22d8f11bd7483aa23dc963e0b3383b : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23b6d10a33f24262a842f11c1646cfde : SUCCESS in 30m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4eade44a01450d94c19369691b6c19 : SUCCESS in 18m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a7971b2b6464c679e56827a70df2239 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da62a5321764d3e8d8fcc22100515c0 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/548c33cf2e604b1bbcf50d9d5fd1d81c : SUCCESS in 6m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbc5345244e4482da3e52486d14dbce9 : SUCCESS in 46m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/52cc2300ef9348c485c6295099ae14a8 : SUCCESS in 1h 23m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/89dcedf8a33c4bcaa44d814b5b6eb8fc : SUCCESS in 58m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/073ee869297f4841b9592e84b85bf17c : SUCCESS in 1h 00m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/44a94540f8d74c099d8d7291f2524ca2 : SUCCESS in 51m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ce50bf961a10467d8ad8f5c3757b568f : SUCCESS in 1h 09m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4086f411b2054f3298a377b6547d5026 : FAILURE in 1h 03m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9db4dfeb66f049c2bf9671e544b10e9e : SUCCESS in 1h 42m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/593b10d81ea24dd8838667e2ba5b45d4 : SUCCESS in 1h 04m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f7672a5f2494ea6a0d35f63ba8820a1 : SUCCESS in 1h 09m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/44b923b590f842589993fc50a9394b02 : SUCCESS in 24m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eac7b7723572403ab52da0095df96be8 : SUCCESS in 1h 13m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dcaa01fa84cf42d8aa3ababe8b17c0f9 : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":12},{"id":"d55c8193ac896c7cd3f2d20ccca4e59d2a59c20e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-20 10:57:37.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"44c48647f868397f5a38b568a815a7cc9310dea0","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-21 06:12:06.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"80d74e22e2a5c76814a6d7da948682377ba3602a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-24 19:53:41.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"53c9e98087a366a95a051294a941c042573156a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 19:54:03.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7d46536cc79cc80ea3a4e22b1fe3fed6b3f38a19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 22:48:19.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3867d99fff97435f954abe1263bfafea : SUCCESS in 38m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4bd104d6f44c4aa1936b8216a91cb45e : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c8f0c06ac2a4a35896dd38597db4cd4 : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9143cac3c61846c882f6c488adfc27f6 : SUCCESS in 2h 03m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/966034a8db434c83a206779d6a114447 : SUCCESS in 1h 23m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2497c3bc5be9430981fa5c7c9c210e5e : SUCCESS in 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2faac001b6a4b7bacced8b559ab955f : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f38064e5634b939b0cb58b5e25aceb : SUCCESS in 20m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f189c5592f4f3486c659ffff17c53a : SUCCESS in 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c956e5a9394e26890e5a1fbd2ca186 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0bb35df943749ddbda9f734fef27928 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a8591f25f7a4ffda178785ba8c03cda : FAILURE in 52m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82a8848e0256400fabeff9f3293c423a : SUCCESS in 1h 19m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90c03c6bbb5c4738b4b5ed9922f750c9 : SUCCESS in 57m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4e02e151b8e424fb2aaa7c1a878cda6 : SUCCESS in 57m 39s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07d748ce5d0342e39d0c001a71403687 : SUCCESS in 1h 26m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cb6193df1304baaa363d4a406a1d483 : SUCCESS in 1h 00m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e5b90ea51f964fc7ae680f86d7a4e371 : SUCCESS in 1h 10m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/28de4dcea51445eab4a0d3ea5453c9a3 : SUCCESS in 25m 50s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2767e896cf9a4e89830a20ee27d9ce19 : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":12},{"id":"0af30e82d4ac74dd78f315185be6979762fe6f74","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 15:44:12.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"564cf06d31ac92b0f7dc20a5d523bfe3dd6dae8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:52:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8af72191ffe34e07bb16fce22304a709 : SUCCESS in 50m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dca4a503bb954865a5de5623df2507bf : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3f53dde0082b7dff754f4b9c85de5f05b1564eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:01:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4cc5ba6b456a492e91cd940dd9dbbcc0 : SUCCESS in 43m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c885435f211d4e7c8339d14c26cb64ee : FAILURE in 1h 54m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/188e71f829cb4baa8c3c5fe97fbf4bda : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c30a5b19447e4608803af608bfd44170 : SUCCESS in 1h 29m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8fc91a674ec642a0b19de7c37987c835 : SUCCESS in 1h 59m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb9e254427944b4a3ed412f4fe7ddcd : SUCCESS in 41m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac8eb15cfc8a4abd8ec9c6579d93d2a5 : SUCCESS in 35m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857b26a0148b45da802fcc883cfef1e4 : SUCCESS in 16m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbbce9288bfe476f932e3a1339568009 : SUCCESS in 26m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98531504b09a47529ba216bc5560eb4a : SUCCESS in 23m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd901965ccb34450925d975759d6498b : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9151aa40068e4a98ad17cc25c8365fa6 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863bf1d40bc74f28aa39566087b6212b : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95eda50e2f854583906126612438182d : SUCCESS in 49m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/909b97a770e345f0bd57691be3d0f9f0 : SUCCESS in 1h 26m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/06a937e4038f4b6ab1898e09ccde900f : SUCCESS in 1h 21m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4266f6eb949f4115b9767860ea6290df : SUCCESS in 1h 02m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2475cc0acdbc4cd2ab802cf2869e1420 : SUCCESS in 1h 00m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58f63ad5eb824977853eae565da6d4ed : SUCCESS in 1h 23m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b47237fe11664d11addf67a8724ebc46 : FAILURE in 28m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/600e82febb0246a8a16e35de01c04cca : SUCCESS in 1h 26m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec43d3e1f5f4dc189e5330c66f111aa : SUCCESS in 1h 05m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5ada219b2b3f41b1b8a33d7a3e40fe34 : SUCCESS in 1h 10m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/196c7f7ee5bd428b997b4c89f442748c : SUCCESS in 52m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d438467684664bcd9a3959c60c5cfa31 : SUCCESS in 1h 27m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/848ebd15622b42979be749cbddc988e3 : SUCCESS in 1h 18m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5b2c53aa4f969a48e96ef08c3c792e39947e6f73","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 18:15:05.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"742c7d49ea17b08c2ab3c01b55e7b3f0e7c16113","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 18:42:24.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/12 : SUCCESS in 2h 01m 17s","accounts_in_message":[],"_revision_number":12},{"id":"2d19f7bd796816fe49733e891020b7425fb548d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:58:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1204ef37adad47208a7c93e304106092 : SUCCESS in 38m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8750b360dbe486782375d25c4bc5794 : SUCCESS in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"01d1087e40f35916dda28f59edecd019cd289064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 21:34:21.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/50751653c00f424da5a4587d4ab6f2db : SUCCESS in 47m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c1a44123ca3a4419bbe9a741180bd087 : SUCCESS in 2h 01m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de93303cca5c48b48cbf1f0666bf4cf9 : SUCCESS in 1h 38m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fffa1ce4123c459b92404a648560381f : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f340ca1d5a8c45d0bb604fd6bae6f3aa : SUCCESS in 1h 53m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/696a9cb9b46644398754534661631000 : SUCCESS in 43m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0a06c0bc30045d38ac5c0775d81357a : SUCCESS in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f92106d3b4441b8b322101c1bb4d788 : FAILURE in 16m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e66d820d36f94f328ce1df526b7cf8ea : SUCCESS in 32m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a9a1eddddae460abf63539c17089ac5 : SUCCESS in 22m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f048d8721978442d88198cd8b25f91f8 : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf853d8fbfe4a728f0c2d88d13f9d3e : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a30479443e47aeaafd89d4f76ec250 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17917a33ae3540f2b732167a705a4175 : SUCCESS in 48m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9657578031e45a6ac64f562127a102a : SUCCESS in 1h 23m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/da6a4a019303497495922a2d78458f4c : SUCCESS in 1h 25m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6fc0b5355b84cb9b4669acf98f1db13 : SUCCESS in 58m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/317d8c8d51464b778109328a2dabeb49 : SUCCESS in 1h 00m 50s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7a89b09385054004a635b1c9ece1e1a3 : SUCCESS in 1h 10m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b35e0fca4f2b41af84614ab32bb75e73 : FAILURE in 30m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/309d2964984e49a68901294a3bf92607 : SUCCESS in 1h 35m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fad08e5c2ff463cb40aaa2a9408c470 : SUCCESS in 1h 00m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/10625e0b98f74eb996ce1451eb7edc54 : SUCCESS in 1h 07m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/edc5979369a540a3a03a6dd7dfd87b46 : SUCCESS in 38m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0316cc63106b4b83acf5f1aa6835488c : SUCCESS in 1h 27m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f7afe9dd2f3f439ea2ada943b8f56aa2 : SUCCESS in 1h 23m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":12},{"id":"27e4b984b26ab11c77622e6af1d530796e1b29aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 21:05:41.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"98e39b4f96cc224fc16bea15937890a26cb3b649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:06:51.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aca6277bef404c42bfdf5f4b6c6cec0d : SUCCESS in 30m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a471cf2cea84feda0e27bc617e2ebf8 : SUCCESS in 25m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e1699e46c2401696815632b71178c4 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/410b8c85b1a5425d83d5e8da9eee19ce : SUCCESS in 23m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92719e3521cd4871b62a2b1612631a5f : SUCCESS in 22m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff78227e6aaf40f1b62224d600bbed06 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdaab8b950b842e5831a48e327c730a1 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eb1af13322e4dfeaaf254e8429db8da : SUCCESS in 9m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcf8111cf1374e72bb755ed6f58d9d8d : SUCCESS in 57m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a37990e8d00d49d6a10a247bc4d5dda5 : SUCCESS in 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"56799d525a9cc9428b0e4ff5df41217aa1eb8969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:07:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"dbd096e5de56501f8a33a5bc3b25a88aea796e77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:21:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/704b7aa848e84586b8970b8490550bca : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d94609b4408348b6b921081c8d7ad99a : SUCCESS in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5aecdc79cc0e4eca3088c7a2ec51af0c05a5f7e4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 22:47:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790060/12 : SUCCESS in 1h 40m 49s","accounts_in_message":[],"_revision_number":12},{"id":"ee6a2f6eb1d327fc2b58186a33c0cc3b111f3258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:19:36.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67523fd8eb784b9595822e4336222c7e : SUCCESS in 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92da9b1f8a064b309266a00a838e8706 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b8f4f8740d64368ad953c0a77cbed61 : SUCCESS in 32m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342be1efdd2d4b7f83c08df6e635b5a1 : SUCCESS in 29m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d1996a96a94b32850a37eccd902331 : SUCCESS in 16m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b600b1b631401989c12fc31327c2b9 : SUCCESS in 9m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f6eabff50144bef850025e2d30d346b : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":12},{"id":"740665ec89adf49ebd33515d6a90640dfe917ba0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:19:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"e095f3ba64bf8a8751bf5de9aef9103619aa6ddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:22:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ea560c1f2f470d9c9baa15791b22c6 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d46bc7b98ae4c768d773281a3f00ec2 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e135a8221dba3beac4047ca9351bdfe600fcf01a","revisions":{"0ab1c43380df46353ce2c925fb7e4f1868140147":{"kind":"REWORK","_number":1,"created":"2021-05-06 12:57:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/1"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:57:22.000000000","tz":0},"subject":"[WIP][POC] New Quota driver","message":"[WIP][POC] New Quota driver\n\nTODO: everything (testing, documentation, etc).\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ab1c43380df46353ce2c925fb7e4f1868140147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ab1c43380df46353ce2c925fb7e4f1868140147"}]},"branch":"refs/heads/master"},"66c39594916ca8f0925a895998c4b5dc5c2de58c":{"kind":"REWORK","_number":2,"created":"2021-05-06 16:53:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/2"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 16:53:36.000000000","tz":0},"subject":"[WIP][POC] New Quota driver","message":"[WIP][POC] New Quota driver\n\nTODO: everything (testing, documentation, etc).\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66c39594916ca8f0925a895998c4b5dc5c2de58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66c39594916ca8f0925a895998c4b5dc5c2de58c"}]},"branch":"refs/heads/master"},"c8946ef9766f83867450c8aee3254b6bf7604753":{"kind":"REWORK","_number":3,"created":"2021-05-07 09:07:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/3"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 09:06:42.000000000","tz":0},"subject":"[WIP][POC] New Quota driver","message":"[WIP][POC] New Quota driver\n\nTODO: everything (testing, documentation, etc).\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8946ef9766f83867450c8aee3254b6bf7604753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8946ef9766f83867450c8aee3254b6bf7604753"}]},"branch":"refs/heads/master"},"0eb60c94d39a1730613c07e8b1c53c50728c47da":{"kind":"REWORK","_number":4,"created":"2021-05-07 15:45:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/4"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 15:45:08.000000000","tz":0},"subject":"[WIP]New Quota driver ``DbQuotaNoLockDriver``","message":"[WIP]New Quota driver ``DbQuotaNoLockDriver``\n\nTODO: this commit is still not handling the stale Reservations\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eb60c94d39a1730613c07e8b1c53c50728c47da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eb60c94d39a1730613c07e8b1c53c50728c47da"}]},"branch":"refs/heads/master"},"b6d32c076407d591cfceab681ff8fbd596e4dc28":{"kind":"REWORK","_number":5,"created":"2021-05-10 08:15:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/5"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-10 08:14:47.000000000","tz":0},"subject":"[WIP]New Quota driver ``DbQuotaNoLockDriver``","message":"[WIP]New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6d32c076407d591cfceab681ff8fbd596e4dc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6d32c076407d591cfceab681ff8fbd596e4dc28"}]},"branch":"refs/heads/master"},"765a40803e6520a628441e9ba371ef3ade0594b1":{"kind":"REWORK","_number":6,"created":"2021-05-10 11:12:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/6"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-10 11:12:12.000000000","tz":0},"subject":"[WIP]New Quota driver ``DbQuotaNoLockDriver``","message":"[WIP]New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/765a40803e6520a628441e9ba371ef3ade0594b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/765a40803e6520a628441e9ba371ef3ade0594b1"}]},"branch":"refs/heads/master"},"78bd8b9394ad8f18b7e7edfb4bf5c8dcf8480e66":{"kind":"REWORK","_number":7,"created":"2021-05-11 16:14:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/7"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-11 16:14:01.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78bd8b9394ad8f18b7e7edfb4bf5c8dcf8480e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78bd8b9394ad8f18b7e7edfb4bf5c8dcf8480e66"}]},"branch":"refs/heads/master"},"f2b173f9034c7d744ecbd16685e06f97f234bd85":{"kind":"REWORK","_number":8,"created":"2021-05-12 11:30:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/8"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 11:30:43.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2b173f9034c7d744ecbd16685e06f97f234bd85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2b173f9034c7d744ecbd16685e06f97f234bd85"}]},"branch":"refs/heads/master"},"7fad6976853c5d41c1a9d407e12e6163a74a9b86":{"kind":"REWORK","_number":9,"created":"2021-05-12 14:32:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/9"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 14:32:01.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fad6976853c5d41c1a9d407e12e6163a74a9b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fad6976853c5d41c1a9d407e12e6163a74a9b86"}]},"branch":"refs/heads/master"},"202f667d59e9b89fb1ce984fb06a556d0f62cc52":{"kind":"REWORK","_number":10,"created":"2021-05-13 06:23:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/10"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-13 06:23:02.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/202f667d59e9b89fb1ce984fb06a556d0f62cc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/202f667d59e9b89fb1ce984fb06a556d0f62cc52"}]},"branch":"refs/heads/master"},"2892e093365a00df04247da7649fed0b509d7eba":{"kind":"REWORK","_number":11,"created":"2021-05-18 11:11:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/11"}}},"commit":{"parents":[{"commit":"9e9488be864619f4843a46972abcb101447dece5","subject":"Merge \"Implement multipath routing in route commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e9488be864619f4843a46972abcb101447dece5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 11:11:08.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2892e093365a00df04247da7649fed0b509d7eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2892e093365a00df04247da7649fed0b509d7eba"}]},"branch":"refs/heads/master"},"e135a8221dba3beac4047ca9351bdfe600fcf01a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-05-20 07:55:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/790060/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/790060/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/790060/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/790060/12"}}},"commit":{"parents":[{"commit":"9e6b7a2284a39ae74d2092083165e3fdf4cb8f33","subject":"Install \"pyroute2\" as a doc job depedency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e6b7a2284a39ae74d2092083165e3fdf4cb8f33"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 12:48:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-20 07:55:59.000000000","tz":0},"subject":"New Quota driver ``DbQuotaNoLockDriver``","message":"New Quota driver ``DbQuotaNoLockDriver``\n\nThis new quota driver, ``DbQuotaNoLockDriver``, does not create a lock\nper (resource, project_id) but retrieves the instant (resource,\nproject_id) usage and the current (resource, project_id) reservations.\nIf the requested number of resources fit the available quota, a new\n``Reservation`` register is created with the amount of units requested.\n\nAll those operations are done inside a DB transaction context. That\nmeans the amount of resources and reservations is guaranteed inside\nthis transaction (depending on the DB backend isolation level defined)\nand the new reservation created will not clash with other DB transation.\nThat will guarantee the number of resources and instant reservations\nnever exceed the quota limits defined for this (resource, project_id).\n\nNOTES:\n- This change tries to be as unobtrusive as possible. The new driver\n  uses the same ``DbQuotaDriver`` dabatase tables (except for\n  ``QuotaUsage``) and the same Quota engine API, located in\n  ``neutron.quota``. However, the Quota engine resources implements some\n  particular API actions like \"dirty\", that are not used in the new\n  driver.\n- The Pecan Quota enforcement hooks,\n  ``neutron.pecan_wgsi.hooks.quota_enforcement``, execute actions like\n  \"resync\", \"mark_resources_dirty\" or \"set_resources_dirty\", that has\n  no meaning in the new driver.\n- The isolation between the Quota engine and the Pecan hook, and the\n  driver itself is not clearly defined. A refactor of the Quota engine,\n  Quota service, Quota drivers and a common API between the driver and\n  the engine is needed.\n- If ``DbQuotaDriver`` is deprecated, ``CountableResource`` and\n  ``TrackedResource`` will be joined in a single class. This resource\n  class will have a count method (countable) or a hard dependency on a\n  database table (tracked resource). The only difference will be the\n  \"count\" method implementation.\n\nCloses-Bug: #1926787\n\nChange-Id: I4f98c6fcd781459fd7150aff426d19c7fdfa98c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e135a8221dba3beac4047ca9351bdfe600fcf01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e135a8221dba3beac4047ca9351bdfe600fcf01a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
