)]}'
{"id":"openstack%2Fneutron~790080","triplet_id":"openstack%2Fneutron~master~I6294098008fbebb2e64922b3aaa085c1361d48a2","project":"openstack/neutron","branch":"master","topic":"bug/1927494","hashtags":[],"change_id":"I6294098008fbebb2e64922b3aaa085c1361d48a2","subject":"Deprecate [designate] admin_* parameters","status":"MERGED","created":"2021-05-06 14:48:49.000000000","updated":"2021-05-11 08:41:21.000000000","submitted":"2021-05-11 08:38:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"790080-bug/1927494","meta_rev_id":"8556ee29bf56ca909109aebfefa711503bd7690d","_number":790080,"virtual_id_number":790080,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-05-11 04:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 08:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 06:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 06:21:53.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-11 06:21:53.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 15:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 16:02:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-05-06 16:54:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-06 18:05:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-06 18:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 21:05:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-07 00:33:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-07 01:40:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 08:08:33.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-07 12:48:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-10 21:18:16.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-11 00:13:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ce3d2ca6f0dbf390a5e8353235b84ca425e7ed08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 14:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45809fc94bf06de70aa57a77d94ef8bfed69e980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:02:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca3702155b524c3b925f4c16058bfa27 : FAILURE in 11m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0540bac96bf840cd9ff7ff93e657fed7 : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"939bdb07a81bd099f95895e28eeeaed29d8300c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 15:35:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48e95a340cae0cb121b0a6a1ba09625ad3359f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 15:36:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c87bb22ea11bc05e123b1771351f8adcd2febe5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-05-06 16:02:52.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nI like the idea, just some nits below.","accounts_in_message":[],"_revision_number":3},{"id":"dd7162aeb9bc78e6a23681f1426fb3180c44cc9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 16:12:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9d6c3286f63ca86fb5cb65e515aca8e9d0ec042","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 16:12:46.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dbf8c89100bd9d9fc4f634426f315875ecce1d4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 16:14:06.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6d1b1a8ec71371e4c6e56d1df17fe21bc00da3ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 16:54:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bff575b7a11429f856fe45c6581dd8c : SUCCESS in 36m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/952fc080680e43e9b75959a74bb0e3cb : SUCCESS in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b21e993aa08c9ef78fa104d8d2f8c33e63e2a9c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 16:58:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"09312be8b122541b37475f4c1107e8bceb53517f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 18:05:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790080/5 : SUCCESS in 1h 51m 13s","accounts_in_message":[],"_revision_number":5},{"id":"a9a82065234c1cd8405988b5642032c0b18f633e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 18:51:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b70be9e32724c07a4f15a7602d5e02a : SUCCESS in 37m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8f94e8f219414b2aa0b5424293f6b971 : SUCCESS in 2h 34m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8d6c366f58ac4eb99fb9dab52244db6a : SUCCESS in 1h 27m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5ec2610ed694874b8ac28de9d6e6eac : SUCCESS in 1h 14m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/72136b3165ef484394febd6014429269 : SUCCESS in 1h 26m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91bf0f9451fe4e248a22562de5a4185e : SUCCESS in 46m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334d3473a47248559409133eed4e4771 : SUCCESS in 26m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93db275cb814d78a2db9e0d575a6f9a : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60533d294d1b4e81b467ba80f5a7a4af : SUCCESS in 26m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec584eb28204b2ab88586e86b985a71 : SUCCESS in 24m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/240539b2b8f9459e86f8c31be435ad50 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538ab6f460124b88adb3f6b4692d1d12 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54f505c93e0a418c816b3f84c569ebb9 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/beed0d4619f549c9b391852f237d0fbf : SUCCESS in 48m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0baedbd1aea64c30a9311ccb5fc6f945 : SUCCESS in 1h 23m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a25b38dffae4d06af5a295cb18449c0 : SUCCESS in 1h 18m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8145524777e3425192ca718cd8c2e468 : SUCCESS in 56m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4fd1c0019ec446ca68f4fceb3116c04 : SUCCESS in 1h 01m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d07b1f79daad438faa804d0633437969 : FAILURE in 1h 06m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4da56d9baf014ce59330e34ded3f48fe : SUCCESS in 1h 34m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d083ff643124402486cef69f9ee8b586 : SUCCESS in 1h 32m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21897cafb3f24215bb37555dee2ec403 : SUCCESS in 56m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c56485d95a284c68822f9b5b10bf3509 : SUCCESS in 1h 00m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1e36a559648642919f2a8ef546c2bae7 : SUCCESS in 37m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb45a8c164b44bdab43eb77fa57cbf96 : SUCCESS in 1h 31m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8a936ae833b847ef9b75dcf19651c2da : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":5},{"id":"cdf16dc2fdb4a1a084c78d5b2aa2275e5737d7e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 21:05:19.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"1b09968b096cace43d2606204d3c92b675950140","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 00:33:21.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"79973fe91b340c8b804d5400c6a266eedabf1249","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 01:40:06.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"bd9faa70e44cca933b7283466694aa52fd53d274","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 01:47:10.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2c38101e79c0e1a38e02a01a28d19ab62ff24770","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 03:06:11.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"35f4dd429f270f0e5d0d77405d7b20d816c4c694","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 03:25:09.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f3c0d5edddc9dff9049e14f9e0ba411ff252f8b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-07 06:13:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"98e8c09e1ca3d059d4bfa0531c7a08dc1ccf1986","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-07 06:14:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3eefa1577ad764d131711207769fb68b49cf55e5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 06:27:57.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790080/6 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":6},{"id":"dc57d19879df63864a0ec7530e5829fb508c8e73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 06:28:35.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"730a06ed50e5defb44dd590d7dbadca11187861f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 07:09:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0717bbd12b2d4f419dda2cadf26f3b33 : SUCCESS in 39m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f22f59d40d9d498495b442f63faa5cc3 : SUCCESS in 46m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f06cdc8430bec55c30a9a716f58756c9717008ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 07:26:04.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"86ebe9e0ae01976467e19ba05f9596d2c730cf0c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 07:42:38.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ba8ceb1017364127e1a9d42117fbb97a73f9ce53","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-07 08:08:33.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9d133660ab9c7eb22bccb1f4db985a1b4a6b3004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 08:52:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/435744ae374f4524b1c627566f50e22b : SUCCESS in 45m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bba53dcb101e477b812e56961e516b5a : FAILURE in 1h 51m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9c3481f211294d0a97511923343d9a59 : SUCCESS in 1h 37m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22091b4396194e43a382db9bd5735ef5 : SUCCESS in 1h 34m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b6804460abfa418ebfe1ee5ade01d813 : SUCCESS in 2h 20m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499a57c0fb7444bba03fe9f0940ef386 : SUCCESS in 40m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/666d854503464895a1152f4b802565c7 : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e2f78bf12a45db88fddc722d2313da : SUCCESS in 15m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2704becfaf7d48b1947f73b26c151bf4 : SUCCESS in 34m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599dcf033a19442e9b95e5f5112ec150 : SUCCESS in 32m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72752dce62e24e0e87d30ad30e6eb961 : SUCCESS in 30m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ca4fc793274d42b6f532de682ab89f : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cbab4dd28a1457782b2b4d3670e4957 : SUCCESS in 12m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b5d0e029e8d4e38b2dabd65510aafb7 : SUCCESS in 46m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20a4c43bc57b49c8b0a74acd6a2a1de7 : SUCCESS in 1h 33m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0d19bf91ce24301a1a049e58a04783e : SUCCESS in 1h 33m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47163f58ebee437ea540f93761fe9478 : SUCCESS in 1h 10m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c06ed7d8a51488d9813cddc45a3a430 : SUCCESS in 1h 08m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/21cb2bfc8f704cc986c15eb867c29775 : SUCCESS in 1h 13m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/36edb9f19ec640e0b31e6172449365bb : SUCCESS in 1h 31m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b9925ae8965416f89ed79bcf676c340 : SUCCESS in 1h 37m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf16613c4d3d49eabe461011d64d6018 : SUCCESS in 58m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fad0e3a06964ac78b730d392e4fa368 : SUCCESS in 1h 05m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2c53355f6ed64f05964b290736df9007 : SUCCESS in 37m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00e51e97bb5b4e839d2736bc08622061 : SUCCESS in 1h 24m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a86978aef08a485f96b3025ac43e673b : SUCCESS in 1h 26m 47s","accounts_in_message":[],"_revision_number":6},{"id":"a78d82904e67742e85ebcd1f2b0ec379256b7e91","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-07 12:48:44.000000000","message":"Patch Set 6: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2ead191424bc2df9fe9187af6a060560e42fe1fa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 13:03:06.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790080/6 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":6},{"id":"00a2143b3e852e3b37ccf16982ae3b23598a51b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 13:33:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5716e7fe777c4e83b416fa5f21bf1ad8 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9dbffec843f4b389d99db49dfde8f9a : SUCCESS in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"60aa1176da9b1ebbd43c082f5a84febc16846d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 15:34:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca4611d0c62b41fc85574959be4d9bc7 : SUCCESS in 52m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/34874e40a64f4dc585d3650f21f8656f : SUCCESS in 1h 50m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11e2626052824fd59e2aa4bbfe2130ae : SUCCESS in 1h 59m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4008fe06bfad428788608577f06e529d : SUCCESS in 1h 30m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9681df885a5e40099a450a4c04978828 : SUCCESS in 2h 38m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a92555c0b04f949f701c384c68d6b8 : SUCCESS in 41m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d669b3890f0450a920815f4a7500d2e : SUCCESS in 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a3679a02584f24bde3246f6b71bad8 : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95468b2f69ee42a18903a8d63c78ceb0 : SUCCESS in 34m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/718838a314854d448288927960044df5 : SUCCESS in 28m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc47deac8014d05921fc23bbfe19305 : SUCCESS in 30m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2c4e89222e4ead86b4d38b78640918 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92279e04382c4f96bec5235124a90256 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b065e01ac76d47309d55aaf0fbfd1b06 : SUCCESS in 53m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4413bc41a5b94e2f933df18d7a1cddc9 : SUCCESS in 1h 25m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/33c682e1a7324ef6abed2f26a21d9773 : SUCCESS in 1h 28m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9033d25183541b3b1d3ab070ddc5a01 : SUCCESS in 1h 12m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4d29f885f32e4797b3886310616f49c6 : SUCCESS in 1h 10m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f1fd2fa7ec6e4df88f9aab3ed8e664fc : SUCCESS in 1h 28m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c848e4929644496b38031bf7a810625 : SUCCESS in 1h 30m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98b30a391e794acbb75a197749aac76c : SUCCESS in 1h 35m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74555552bd644846b67850348a45de7d : SUCCESS in 1h 06m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/83327ecb2e1d43dc907a628a481841cc : SUCCESS in 1h 08m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a76378d29433415b89c727697203d6fb : SUCCESS in 29m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c77301c6772a4e94af603c0f3fcab126 : FAILURE in 2h 04m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1ff1d4fc379e444ba0275622cc939d43 : SUCCESS in 1h 40m 51s","accounts_in_message":[],"_revision_number":6},{"id":"27ea96e0c26feea067302f1041e8c406962cde86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 17:54:49.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"6779bb834b7d084d709473f31d8a961c12efacdf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-10 21:18:16.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e439cf28592b3cd1c68a206ad042fd6eaadcfbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-10 23:29:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a13b562d52684e05bfbb527c4a1a7ec1bc85efa3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-10 23:30:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"12fe9774e2dd17a0064fe69266d1f8609ef4400b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 23:42:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8d7dc831e3d491eac37eafd641c9f9b : FAILURE in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46a63ddc3be54debaa715b82227d827f : FAILURE in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"90ef379d0781f3b3a27d80caacaa1ca93ac03dd1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-11 00:13:25.000000000","message":"Patch Set 7: Code-Review+1\n\nLGTM, I agree that these are legacy and should removed in favor of the standard parameters under the [designate] section.","accounts_in_message":[],"_revision_number":7},{"id":"15f978480ddeefbaefcb838d0877678ee983c1a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-11 00:47:53.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e94a976e00e514096810130681fd3b8c78aca9fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-11 01:07:14.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/7/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/7/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"a6745405b5e3e1820c1bc3dbfcb4308d272af85e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-11 01:21:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790080/7 : SUCCESS in 1h 51m 21s","accounts_in_message":[],"_revision_number":7},{"id":"d639751f390f408b21099f8e66dd8ebc1f223601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 02:06:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dd640c7e337543528cbe4225d53bb827 : SUCCESS in 46m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/25c6822858684559be02b949a4379f4d : SUCCESS in 1h 53m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8c1bad79b93f4ed1b4114143d83b8d6f : SUCCESS in 2h 36m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/417cc89eef9b4a948e28612244770765 : SUCCESS in 1h 38m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/18108bbd4041413f84f25c1da781266e : SUCCESS in 2h 15m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3370589dcf140e2b4c9cea5dafec666 : FAILURE in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bae4c0ea5b384285b43490cabc4a8cc9 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508ea3248f194144a9c804770c131084 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84381d7edfe1481ca9a2385210768607 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b7f1dcd146446d993c18087ca4e1c55 : FAILURE in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb4a37d16a294a62bf4b5378e72a2b02 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49c0102f9324e5a847dcefa9dd81c5a : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c73e10e12f254b7ba07b83a09ef219ee : SUCCESS in 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4842d459a279429684c593b9543905b1 : SUCCESS in 48m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddcd0c660ee749fc88f38481ab7cc1e3 : SUCCESS in 1h 18m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2b4fb0a66e51415d9afd7e8acca5535e : SUCCESS in 1h 26m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e4a9923abb047f9a7bc2a1589eff385 : SUCCESS in 1h 00m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ff86f5274584ea4b11dfc2a239f43d2 : SUCCESS in 1h 01m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6529e6cd76eb47d0adb64b7afa6c837e : SUCCESS in 1h 10m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/928f0c50bddd4acaa5cd4a76ff27a437 : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66d17f7c0ebd49708428fae8d0d1bdb9 : SUCCESS in 1h 31m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de19081f14f3466fbd3766dbe9a90eb2 : SUCCESS in 56m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2242b5606d74f218b2462dee04078c1 : SUCCESS in 1h 10m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c97b4f6aaf0b47a5a46f4002c740e9eb : SUCCESS in 32m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b68a2a7bf5a7470791a11122a332c2a0 : SUCCESS in 1h 22m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2e83746eca8f491ba3dea18808b3e809 : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":7},{"id":"1646c9d1511f367d85ae4360b3d36dfc3b9c215d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 03:09:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3a71b8e26d40571f79f51be98c00913d2a4c6138","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 03:09:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"66971cf2a3ac6f4db48ca818a8b946fbeb20bd6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 03:54:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0a41767fc0141cd9ac5adbe5b0a065f : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b526db89450e4893a4fe0596377afc81 : SUCCESS in 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fabce756f7cebb1bbe2fc06f5214bca926f732d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-11 04:11:31.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/790080/8/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/790080/8/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"af28c6b8c91973f1944a60da9f7a907c651ae7a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-11 04:27:23.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/80/790080/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"db221ef59f8af6b78ef6aa0d9fe852e004197173","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-11 05:00:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790080/8 : SUCCESS in 1h 50m 20s","accounts_in_message":[],"_revision_number":8},{"id":"5268755c6c084d2588271f2e97a3abf44e3820d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 05:26:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bb516f676952494d92bbf975e4fdafd5 : SUCCESS in 40m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6ee3e4dba2540f6a401237fe4621321 : SUCCESS in 1h 44m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/59adacb875464bde97c1c3a4878ff86d : SUCCESS in 1h 30m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ec622dd70034823b70e90c2db9e428f : SUCCESS in 1h 17m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fb7adce47cf24a95ab92025c779adf88 : SUCCESS in 2h 12m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8df84d4c129c486bb95bc9228b720948 : SUCCESS in 35m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90b8aeee6d4d43eb8f7651680302d84e : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a286f561cef4a0b8ae74383618fce12 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a9b09f6c6744f96a0919efe6599969c : SUCCESS in 27m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebd7b640e1b24e32bfb75ecba46da97a : SUCCESS in 23m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e4bf068d124f1288c74f178c06c1b9 : SUCCESS in 29m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cabd5c0da5e49bbbdf877f28bafdc7a : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/277464b0cfce4e26beb7f791c68ca6c4 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f4dd942a9b742adbcbd817e38cf8f04 : SUCCESS in 49m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b040a42255b40a6aea48663be9c5456 : SUCCESS in 1h 20m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c4020b56c4bd4d8fbf6e7ac8082f420d : SUCCESS in 1h 15m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/27d9ccb83e7f4805b41dba918859b371 : SUCCESS in 1h 01m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ebb37678f97c47ff925f11c16359f1eb : SUCCESS in 59m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3aa4874f91904320aea8a761eea6cc6e : SUCCESS in 1h 13m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad2ec7a9b5444ca682c8b252a7be6239 : SUCCESS in 1h 24m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a4f05794b994ff8ae3722c37e893710 : SUCCESS in 1h 29m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02f42dc83f734f03a88044e80299681f : SUCCESS in 1h 02m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8253c25b615b43b1af898346fd0ba401 : SUCCESS in 1h 07m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e11daff0588c4916aafb860b291c0af9 : SUCCESS in 34m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b57fdab9a1864b8cba4fe4f96e21bae7 : SUCCESS in 1h 33m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/952a629f7fd647779625831bc3d46f52 : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":8},{"id":"d066a81d8c69c87ab4c3defdbda64e2dcf294b8d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-11 06:17:56.000000000","message":"Patch Set 8: Code-Review+2\n\nThx","accounts_in_message":[],"_revision_number":8},{"id":"5fcbcef9e2f94ef68b344f7723ac3715e35f029a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-11 06:21:53.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"cb66e974ecc74226c82a8aa2f7d005b109ee90fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 06:22:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d99033a7ac5bcf5867f6b1580209b1276c451a98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:38:30.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1b5e3b4d6173437988ad84521cb65080 : SUCCESS in 53m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7ea02bc773864f3fa4aadb2308ff6ef2 : SUCCESS in 2h 10m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ed93bd4220f74628ae9ed6af7f385715 : SUCCESS in 1h 46m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4f9104ef83144cb28fb310a30a7dc435 : SUCCESS in 2h 06m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/669a3b48f6454a67aca60f75711c6fcc : SUCCESS in 1h 34m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c309c955140b441fa43406c51ef683cf : SUCCESS in 34m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a8da67713f44c148257f8fc672aa750 : SUCCESS in 18m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a1b4c0a8a7443cbabf770d5e014ebb : SUCCESS in 29m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e9acff88bd14fa69bfa1767cf2571dc : SUCCESS in 28m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b453bca60a44b489a6c665cdd95f163 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6fb6d32e2cf4daba956c29540cbea6a : SUCCESS in 12m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6d0dfe10464415cb21117d02907b465 : SUCCESS in 50m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/772c478c5ce240feac8991e09efa7bb2 : SUCCESS in 1h 17m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4fb4d02f2564dc0b03f8991cd3451ce : SUCCESS in 1h 31m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce7fedbf8660480a9def7c8b32bac4d3 : SUCCESS in 49m 44s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcc5da4f2d564a31a89e439c4deac04d : SUCCESS in 1h 34m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d2aed971ac241a19de6079c656b1d0f : SUCCESS in 48m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7e11a9f7f73a454d8ed1ef1c6f66c7ce : SUCCESS in 1h 23m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8f849cfd824e4799b8ba1b7ace5ecd7e : SUCCESS in 29m 59s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e3b62822ab84e70b9743bb774bf2a51 : SUCCESS in 1h 29m 46s","accounts_in_message":[],"_revision_number":8},{"id":"b638a7da6d3fa94a17d3306b8a157324f3b4396c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:38:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"8556ee29bf56ca909109aebfefa711503bd7690d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:41:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610c210495584c3cbd659ef5fb0a9b98 : SUCCESS in 2m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cfde0ec201540ddaf67ca832c5dc4e1 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"adfd853267ca529816f4c17a145d9e70e8abfac5","revisions":{"6215c07c622beda5ec66f76146e937fe6b1a229b":{"kind":"REWORK","_number":1,"created":"2021-05-06 14:48:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/1"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:47:56.000000000","tz":540},"subject":"Deprecate [designate] auth_* parameters","message":"Deprecate [designate] auth_* parameters\n\nThe auth_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these doesn\u0027t fully implement required\nparameters for Keystone v3 identity but are still based on Keystone v2\nidentity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6215c07c622beda5ec66f76146e937fe6b1a229b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6215c07c622beda5ec66f76146e937fe6b1a229b"}]},"branch":"refs/heads/master"},"5e8293100cbca96e54c252a173922237e2efbbd4":{"kind":"REWORK","_number":2,"created":"2021-05-06 15:35:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/2"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 15:35:33.000000000","tz":540},"subject":"Deprecate [designate] auth_* parameters","message":"Deprecate [designate] auth_* parameters\n\nThe auth_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these doesn\u0027t fully implement required\nparameters for Keystone v3 identity but are still based on Keystone v2\nidentity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e8293100cbca96e54c252a173922237e2efbbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e8293100cbca96e54c252a173922237e2efbbd4"}]},"branch":"refs/heads/master"},"d276cb5ec32242ea74d4eb7abe5ac3da04b23166":{"kind":"REWORK","_number":3,"created":"2021-05-06 15:36:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/3"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 15:36:40.000000000","tz":540},"subject":"Deprecate [designate] auth_* parameters","message":"Deprecate [designate] auth_* parameters\n\nThe auth_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these doesn\u0027t fully implement required\nparameters for Keystone v3 identity but are still based on Keystone v2\nidentity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d276cb5ec32242ea74d4eb7abe5ac3da04b23166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d276cb5ec32242ea74d4eb7abe5ac3da04b23166"}]},"branch":"refs/heads/master"},"b21ed012bb4ab79e4a71cde2e644def9bf6d5a84":{"kind":"REWORK","_number":4,"created":"2021-05-06 16:12:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/4"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 16:12:02.000000000","tz":540},"subject":"Deprecate [designate] admin_* parameters","message":"Deprecate [designate] admin_* parameters\n\nThe admin_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these doesn\u0027t fully implement required\nparameters for Keystone v3 identity but are still based on Keystone v2\nidentity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b21ed012bb4ab79e4a71cde2e644def9bf6d5a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b21ed012bb4ab79e4a71cde2e644def9bf6d5a84"}]},"branch":"refs/heads/master"},"70d1546c89b6ab0840319692fce9bcbdd5615162":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-05-06 16:14:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/5"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 16:12:59.000000000","tz":540},"subject":"Deprecate [designate] admin_* parameters","message":"Deprecate [designate] admin_* parameters\n\nThe admin_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these don\u0027t provide all parameters for\nKeystone v3 identity but are still based on Keystone v2 identity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70d1546c89b6ab0840319692fce9bcbdd5615162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70d1546c89b6ab0840319692fce9bcbdd5615162"}]},"branch":"refs/heads/master"},"5366217fadf2b15c86f2ad015a7f9a929e10d53e":{"kind":"REWORK","_number":6,"created":"2021-05-07 06:13:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/6"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-07 06:12:55.000000000","tz":540},"subject":"Deprecate [designate] admin_* parameters","message":"Deprecate [designate] admin_* parameters\n\nThe admin_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these don\u0027t provide all parameters for\nKeystone v3 identity but are still based on Keystone v2 identity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5366217fadf2b15c86f2ad015a7f9a929e10d53e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5366217fadf2b15c86f2ad015a7f9a929e10d53e"}]},"branch":"refs/heads/master"},"c06817284a6a5d470a219b7675a993e07576d3d1":{"kind":"REWORK","_number":7,"created":"2021-05-10 23:29:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/7"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-10 23:29:19.000000000","tz":540},"subject":"Deprecate [designate] admin_* parameters","message":"Deprecate [designate] admin_* parameters\n\nThe admin_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these don\u0027t provide all parameters for\nKeystone v3 identity but are still based on Keystone v2 identity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c06817284a6a5d470a219b7675a993e07576d3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c06817284a6a5d470a219b7675a993e07576d3d1"}]},"branch":"refs/heads/master"},"adfd853267ca529816f4c17a145d9e70e8abfac5":{"kind":"REWORK","_number":8,"created":"2021-05-11 03:09:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/790080/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/790080/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/790080/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/790080/8"}}},"commit":{"parents":[{"commit":"d6d68552347a70f4affb486758cdaf9ce8cf5950","subject":"Merge \"Physical NIC RP should be child of agent RP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6d68552347a70f4affb486758cdaf9ce8cf5950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:41:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 03:08:42.000000000","tz":540},"subject":"Deprecate [designate] admin_* parameters","message":"Deprecate [designate] admin_* parameters\n\nThe admin_* parameters are implementing the same functionality as\nkeystoneauth parameters alghouth these don\u0027t provide all parameters for\nKeystone v3 identity but are still based on Keystone v2 identity.\nThis change deprecates these parameters so that we can remove\nsuch redundant and outdated definitions in a future release.\n\nCloses-Bug: #1927494\nChange-Id: I6294098008fbebb2e64922b3aaa085c1361d48a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adfd853267ca529816f4c17a145d9e70e8abfac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adfd853267ca529816f4c17a145d9e70e8abfac5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
