)]}'
{"id":"openstack%2Fneutron~790168","triplet_id":"openstack%2Fneutron~stable%2Fstein~I07c70db027f2ae03ffb5a95072e019e8a5fdc411","project":"openstack/neutron","branch":"stable/stein","topic":"net_del_opt-stable/train-stable/stein","hashtags":[],"change_id":"I07c70db027f2ae03ffb5a95072e019e8a5fdc411","subject":"Improve Network delete performance","status":"ABANDONED","created":"2021-05-07 08:56:47.000000000","updated":"2021-06-14 10:33:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":789673,"cherry_pick_of_patch_set":3,"meta_rev_id":"51fd2ed6574fe006ccd32ae73360ae681171f0e4","_number":790168,"virtual_id_number":790168,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-07 10:44:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 08:56:47.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-07 10:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 10:33:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"845a260e09057882985fac789b283c516d89abaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-05-07 08:56:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"e34d26d79a32c27f1d337c338263546766121e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 10:44:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/bbbca20bda014cbb8f4cdeebf0131a29 : FAILURE in 44m 34s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/45a9d66475b147df8e2c7a5d7e0eb8b2 : FAILURE in 1h 42m 11s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/496e760a272a499e9cf54f2691dc0c9b : FAILURE in 1h 36m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/9cae4f9f05864f1b81bf7d0857aa8bc2 : FAILURE in 1h 19m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e4312c0ed574e7586f8ce2a687317cd : FAILURE in 23m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55bc47749d54ed4a6cbc2efdf4eed2c : FAILURE in 8m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2b0e7111b1e4dc2899f7de4dc7db164 : FAILURE in 26m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/52ec95b6a20242d49953773fe30b51b5 : FAILURE in 27m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc6d6fb75d0346f78af0f82ce403980b : FAILURE in 19m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f4e281176864d34be068b6e5c81ce47 : FAILURE in 24m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8a40ea1be2479786a28199ccf9357c : SUCCESS in 6m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/32b7ca3d16304fc388fb9298a2ef9f68 : FAILURE in 1h 00m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4dcc3374325b4983877f70ba8314cf76 : FAILURE in 46m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71234659162b401bb17c19f6717a6cbd : FAILURE in 1h 09m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e5ad4929529c4aa085b18cb16d218c62 : SUCCESS in 42m 26s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/de4e9391e67c40c68b6812c0f00e86be : SUCCESS in 36m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0dc5bca8f3c44eda9619ead9257e2878 : FAILURE in 1h 06m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e33d03cc0d4e4cd7a534744b12de858e : FAILURE in 1h 10m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4d2a7c14e01a46c992fc026e1be555c4 : FAILURE in 1h 11m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8f8a842f128f4f87b2f3153265122a4c : FAILURE in 1h 46m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/862860ce0f1f4a03827fb7e4eab66f10 : FAILURE in 52m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3c5b5ba6a064519afc914d1a1b30175 : FAILURE in 58m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bab1cc494eb34ef596cf7726616a72ab : FAILURE in 1h 21m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/790afdec408041f194cce1889f67943d : FAILURE in 1h 29m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/467a3ddfc0894393a95bd734925c11de : FAILURE in 1h 01m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec2b55d1045348c49929a01c0133937b : SUCCESS in 51m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85db1fd70e504d7aaacfd10cfee3fea1 : FAILURE in 1h 30m 49s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/56bf428db3574aa389869d8b8139c37d : FAILURE in 1h 08m 11s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0b546d36b8454bbc9882ee58ac7aba37 : FAILURE in 50m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1f6d3d6d42124fd2a3742b02dbd85fd3 : FAILURE in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51fd2ed6574fe006ccd32ae73360ae681171f0e4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-14 10:33:06.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4147779fa618fc6b75a4f6ca3abd02fb505529e","revisions":{"f4147779fa618fc6b75a4f6ca3abd02fb505529e":{"kind":"REWORK","_number":1,"created":"2021-05-07 08:56:47.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/68/790168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/790168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/790168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/790168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/790168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/790168/1"}}},"commit":{"parents":[{"commit":"27fbc5fc2ed6d334082acad1a00e488b2f6f17ea","subject":"Merge \"Call install_ingress_direct_goto_flows() when ovs restarts\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27fbc5fc2ed6d334082acad1a00e488b2f6f17ea"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 08:31:01.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2021-05-07 08:56:47.000000000","tz":0},"subject":"Improve Network delete performance","message":"Improve Network delete performance\n\n- pass existing network dict to PRECOMMIT_DELETE ml2 handler;\n- skip MTU and segment update handling in case network is\n  about to be deleted.\n\nAbove gives up to 50% net delete speed-up\n\nConflicts:\n\tneutron/services/segments/db.py\n\nChange-Id: I07c70db027f2ae03ffb5a95072e019e8a5fdc411\n(cherry picked from commit 80eddc40390e63c9c1102b827997054708f2618b)\n(cherry picked from commit 23e160131be16fc21eb789372b864489ef99c576)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4147779fa618fc6b75a4f6ca3abd02fb505529e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4147779fa618fc6b75a4f6ca3abd02fb505529e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
