)]}'
{"id":"openstack%2Fneutron~790169","triplet_id":"openstack%2Fneutron~stable%2Fstein~I0c10b824c0f919607bb6ea31c5533ce7bd65df42","project":"openstack/neutron","branch":"stable/stein","topic":"trivial_fix-stable/stein","hashtags":[],"change_id":"I0c10b824c0f919607bb6ea31c5533ce7bd65df42","subject":"Remove unneeded DB register retrieval and refresh in network update","status":"MERGED","created":"2021-05-07 08:57:31.000000000","updated":"2021-06-01 17:04:52.000000000","submitted":"2021-06-01 16:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790169-trivial_fix-stable/stein","cherry_pick_of_change":789772,"cherry_pick_of_patch_set":2,"meta_rev_id":"7ebf2d81594d6cf0017ef4accc0348d2beabf8fa","_number":790169,"virtual_id_number":790169,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 16:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-06-01 14:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-05-14 14:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-06-01 14:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 08:57:31.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-07 10:35:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-07 10:53:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 14:31:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-01 14:59:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"723180e2f911b71af1fabfa7e8845cb686ec77f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-05-07 08:57:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"26d3094745851ee6f0b70d17f40faa92f3d4d0bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-07 10:35:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/69/790169/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/69/790169/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0a7fe151ea9f8bc139dbdeda9c5d476115e12f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 10:53:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/629e95e3136c4d40a60161e233f621de : SUCCESS in 42m 58s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f2717a448d8044e890a4628e54649fad : SUCCESS in 1h 08m 04s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/b50fc58911054aa39d0a20c25b57aa69 : SUCCESS in 1h 16m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/daaf50cb970d4b77bcc70d9970cddc60 : SUCCESS in 1h 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44630d4135004a0fad0dacad63790eee : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bceed1855a734fa995e8a2a64c9a4eab : SUCCESS in 12m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6049b4bd10b34c089bb2177aebd7983e : SUCCESS in 17m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5c936de480434be9b87b74a0b67b5120 : SUCCESS in 27m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee6f01d71580455ead52bc49a7a908b4 : SUCCESS in 19m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48319454ffa749ac8e78520b265969bd : SUCCESS in 24m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f8c1a801b049808141f975487d19ca : SUCCESS in 5m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bac098f113b14dcaa34bb11d053ff937 : SUCCESS in 1h 33m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e94d8db03d14461290ddac5e0def61aa : SUCCESS in 1h 02m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c50e178ef80a4d8381c85e21bde8bfc5 : SUCCESS in 1h 20m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ff8397820cf4f798bd94f234df995c6 : SUCCESS in 39m 16s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/d37301cc985c4ca9a5a119de7d4b0642 : SUCCESS in 39m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/29261b270dae47dbb863529caf35771a : SUCCESS in 1h 19m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c59b64167fa44a67804eec509e7ce9fd : SUCCESS in 1h 22m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/feea367527d046a09021fbfe570c8165 : SUCCESS in 1h 26m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/396954d8e6da45eda62aabf14b1bc16a : SUCCESS in 1h 32m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fd7ab4c614a4310a2955e18d9436ae2 : SUCCESS in 1h 16m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f6675fd179a4a51bfedfaf4b72f6da9 : SUCCESS in 1h 07m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c35857832d774938b0b07d3c2c2d5083 : SUCCESS in 1h 48m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9bd79ac20da64c86ba877d44c662e507 : SUCCESS in 1h 50m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4474937a77294e20b5804a8d882ed404 : SUCCESS in 1h 24m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/44faf6f787664510ae8ac623fb4a2af5 : SUCCESS in 50m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c246750c03a4ad8ac97fbd6bde7d6cf : SUCCESS in 1h 38m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c5dcf45e420245f697bed23b4bcef9f0 : SUCCESS in 1h 15m 52s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79b0f58eedd147339a14a6cefdbb3955 : SUCCESS in 57m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/239a2931af7949deb54a245b1df8c987 : FAILURE in 44m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98088e736cfc592236f07f78154468ae6a87660","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:31:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bea3b50d2617c346b4a6e92e768ed7135ec45a96","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-01 14:59:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"add8866541d832208dcd158c85dec9cc3fb80ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:05:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25de715c5e75acb1af58e4ba3d3537d5faf1262b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:56:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/6ef50c1f154648ab9a3b6e6876bbe55e : SUCCESS in 53m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7571a62f2e7f45be8b9a75ba16081e53 : SUCCESS in 14m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f11a78a17da643edbc4be8f956c78391 : SUCCESS in 25m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/009fa1e04eb0415ab8d273af883629f5 : SUCCESS in 25m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b32c1b7064844c8832ca05290f67735 : SUCCESS in 22m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3eeb01627b14de09478f3f163d3706e : SUCCESS in 23m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f18a51d04f4717b159bd6b82e1dc19 : SUCCESS in 7m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf59323bde62432fb84313f500f28f62 : SUCCESS in 1h 41m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/67a6a854e25e4679ae8f192a6ff15ab4 : SUCCESS in 59m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9ba4f1ea4584f009d5e2631ca02963e : SUCCESS in 1h 32m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ad23c8a7944477d959b74dbd7017e27 : SUCCESS in 43m 07s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/34e065b401ce40b3ab60f4d3c3ccb53a : SUCCESS in 40m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2c3a2215005b4a4b8b9681fcb66b461c : SUCCESS in 1h 30m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/97db480264554a7cb86c34d86114e04e : SUCCESS in 1h 06m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4824704c1e374684aa6d7ad32d434291 : SUCCESS in 1h 11m 43s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31fe52141b4d4f5a91147f841336ae5e : SUCCESS in 1h 21m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cb3a198f36844c596815a5acfadbc2a : SUCCESS in 1h 06m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e843476eb6e4066a5369d96943b25c2 : SUCCESS in 1h 03m 58s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e02865ddcc4d438ca9b6957438b446a8 : SUCCESS in 1h 32m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73774669c9d5444488d8781131665502 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c839347460a7f33e6a15e3c67964fc4d6c255bd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:56:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ebf2d81594d6cf0017ef4accc0348d2beabf8fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:04:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12455b8273ac42898807152b17c31321 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42c099dd97acee5e77712470dcf9c45d3f5a1639","revisions":{"42c099dd97acee5e77712470dcf9c45d3f5a1639":{"kind":"REWORK","_number":1,"created":"2021-05-07 08:57:31.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/69/790169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/790169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/790169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/790169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/790169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/790169/1"}}},"commit":{"parents":[{"commit":"27fbc5fc2ed6d334082acad1a00e488b2f6f17ea","subject":"Merge \"Call install_ingress_direct_goto_flows() when ovs restarts\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27fbc5fc2ed6d334082acad1a00e488b2f6f17ea"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-08 15:42:53.000000000","tz":0},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2021-05-07 08:57:31.000000000","tz":0},"subject":"Remove unneeded DB register retrieval and refresh in network update","message":"Remove unneeded DB register retrieval and refresh in network update\n\nThe first DB operation done during a network update is a network DB\nregister retrieval. This information is stored in memory during\nthe DB transaction context and it\u0027s updated with any change done\nto the DB register and the relationships; e.g., the network DB model\nfield \"qos_policy_binding\" is a backref relationship with a\n\"qos_network_policy_bindings\" DB register. When a\n\"qos_network_policy_bindings\" register is created, updated or\ndeleted, the network DB object, holding the relationship, will be\nupdated too.\n\nThere is no need therefore to retrieve again the DB object and\nforce a session update. The DB object contains the updated\ninformation.\n\nTrivial-Fix\n\nConflicts:\n      neutron/plugins/ml2/plugin.py\n\nChange-Id: I0c10b824c0f919607bb6ea31c5533ce7bd65df42\n(cherry picked from commit 0e8ebf133d8fa9a6aee576e265cec64a140b7ded)\n(cherry picked from commit 1c55dbe82569c7ff7b9b1746358bc21e66790110)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42c099dd97acee5e77712470dcf9c45d3f5a1639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42c099dd97acee5e77712470dcf9c45d3f5a1639"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
