)]}'
{"id":"openstack%2Fneutron~790352","triplet_id":"openstack%2Fneutron~master~Id09de6c43c0fab4009336e253c88f54219398053","project":"openstack/neutron","branch":"master","topic":"bug/1927849","hashtags":[],"change_id":"Id09de6c43c0fab4009336e253c88f54219398053","subject":"HA-non-DVR router don\u0027t need manually add static route","status":"MERGED","created":"2021-05-09 06:28:58.000000000","updated":"2021-05-22 05:07:01.000000000","submitted":"2021-05-22 05:05:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790352-bug/1927849","meta_rev_id":"899ef3e05c6ab6060db7a57a53081d1734596939","_number":790352,"virtual_id_number":790352,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-22 05:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-17 20:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-18 07:21:50.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-18 07:21:55.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-09 06:43:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-09 07:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-09 07:43:11.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-09 08:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-09 08:24:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-10 11:44:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-11 14:21:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-14 11:29:58.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-17 20:32:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"cf57e9a77274cec15f39da869930e376f6061313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-09 06:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37cefacda070aa31149afe46d27620166b60b416","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-09 06:43:22.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790352/1 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ac4f2857b76ae426b303f88436cfb96d52f54302","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 07:16:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebc1e4e3c1754bd4a69f7a26690000c0 : SUCCESS in 44m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6653998740e04ef1afc77a7e5dc5d80a : SUCCESS in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbfc716f8848af86202988cd537abe2f576cfc8d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-09 07:43:11.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/52/790352/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d2bfda3683aac2e7fa4a5ad93c079b0adf498892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 08:21:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b0a5767b958470c983ab7458e6deec1 : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/776192411b4e40af804f927374a60b43 : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1620a09485646e4942a9bd5bd11dd9c : FAILURE in 1h 15m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bcda0fed1e64506aa7e316b09ef77b0 : SUCCESS in 38m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3f87057ecfd4b79b01e5105b53d84bd : SUCCESS in 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7179003622cf45308da33a8146cbf22b : SUCCESS in 15m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623ebd0d85544932a2239f6800903dc1 : SUCCESS in 19m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/123a9a65e7904d8d8131dc6964cb94bb : SUCCESS in 24m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/984c26670fb04273949d1ca077ef3e50 : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d37fab6e73413c9044f89a2e4336b8 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0154e2f95e74133b305824dcbfa0573 : FAILURE in 50m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dafc32c8b2a74b70bf77318d55fd36c7 : SUCCESS in 1h 34m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/44a399ecb88d405baccf51a7347920d6 : SUCCESS in 1h 23m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7f6a15380fe4d37ad3e0298f28cdcff : SUCCESS in 1h 04m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d7fa70730c2d4766afca40e6977ed74d : SUCCESS in 1h 06m 28s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/05aeda0fb0f24377a093726dd7d70075 : SUCCESS in 1h 04m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ceddffd1a7ef488baf011ac462df987f : SUCCESS in 1h 02m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52804978598e4efc8c90884a3094a412 : SUCCESS in 1h 17m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f67b425c36c247318011fded80f4cb51 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0eda05c0bf75808fb0bb2499a170e8ea2421b039","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-09 08:24:41.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/52/790352/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/52/790352/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d54a130f7df59e03436adbd56201719a6f407f69","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-10 11:44:05.000000000","message":"Patch Set 1: Code-Review-1\n\nAdd unit test please that cover this change for HA router.","accounts_in_message":[],"_revision_number":1},{"id":"244893e1a99feb5643def8d90c9db9aaed765be4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 10:33:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07cb42c080e177c749cbeb4acee42b566e64f1cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-14 07:12:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18b7af7f84dff317093e72b044c26dc80b4deb83","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-14 07:13:43.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Add unit test please that cover this change for HA router.\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"07cca7fbb35990dccbaea84f1b52abb8d1ec4758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:56:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f4e97873e874b75bb4db7fe40a6d22e : SUCCESS in 38m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23b5325e891542df8f8c2843b189fce9 : SUCCESS in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c4fd41c6dc53ea5b6ee91e05789e12d9d37c508","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 08:49:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/2 : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a920898f89f01020cbdfa15f786cd16877614315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:28:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/69bfc1b6b01141c5b6f4b3e26befd644 : SUCCESS in 2h 09m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6aa3ce6a30eb4954a2e3c201a28eb130 : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbf8358c199945f7937d4450123916bc : SUCCESS in 1h 31m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15cb4df6d955411cbd6cd497afd3c1fc : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a0142d9a4164ac29288cab56affc11e : SUCCESS in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248fd50361fe481ca6df07d96066366b : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f8b336349840a6980e4dccd05e1e4e : SUCCESS in 25m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c99c4d21ba41719d4775b913e45af6 : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b944a85cc59043679d3d70d986ced26e : SUCCESS in 32m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7151f98295f44ea88d9a134071aff65 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b39c42c5fe54693842be40af9e472d0 : FAILURE in 41m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e43d074c3d2c40af9f4f5b94711662b2 : TIMED_OUT in 2h 15m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a3f3540593f42fa9cc8a57bd3fce607 : SUCCESS in 1h 12m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8177b1f15c8647d590b8b54a8340e573 : SUCCESS in 55m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b4d77d73343f4e009100031a72c915fc : SUCCESS in 57m 10s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c99f47548a14ea68439fcb263aac13a : SUCCESS in 1h 05m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93103c8268904b589ca68a1257594c1a : FAILURE in 1h 11m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe355d81a8df4bedb9065618fda2931b : SUCCESS in 1h 32m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de2a03fa00cb44c6b02773d9f3c2735e : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ff8c8fdc09a842bb0d93b566a5b7a0900cd62aa0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-14 11:29:58.000000000","message":"Patch Set 2: Code-Review-1\n\nLooks like func tests failures are related","accounts_in_message":[],"_revision_number":2},{"id":"42b8ac1f31e9f4143093a150a9f0929162896817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-16 08:53:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc304c6404d65e65ea7c01afe3f3057e44bc0f7a","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-16 09:08:29.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Looks like func tests failures are related\n\nLooks like this is a keepalived\u0027s bug, but I am not certain. The patch is work fine in real environment, the keepalived correctly manage the extra routes. But, In functional test environment the extra routes can not be auto generated by keepalived. However, when I send a HUP signal to keepalived process, the missing routes appeared. So, I add a line code to send HUP signal the test case passed.","accounts_in_message":[],"_revision_number":3},{"id":"2bff9bb8351b27531bf1dd92292007ac05f51c3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-16 09:37:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a41aa49c02e84ce29b1252f58c00ae67 : SUCCESS in 35m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93d22d3f4133475da0475328aa11e344 : SUCCESS in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b403455d8d5e0bce0f0f104aff5533a33f4f86ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-16 10:46:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8e8877637e9c4f6f9fa17451c5f8ed2d : SUCCESS in 1h 45m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/590e76aa00e64e20b1af88256f4cbd87 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4bad233a50a54f7dbff0e86ecbafdaff : SUCCESS in 1h 51m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f727830bec39404a8c5e305fdac68e56 : SUCCESS in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6321601800cc4df7b5416ecfd6491181 : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7907b1baaa2442049e85bc44871fda00 : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fc08ff5ea824ad8b10bf5599b0f7b80 : SUCCESS in 28m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bd0b2d9523644e594092c39a1d06ef9 : SUCCESS in 22m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3615a73592344c1f8bdd291e40de5877 : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a56cb846044738bea84f6300ff8183 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bfb142983154b6583e7d67aec60a989 : SUCCESS in 49m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cacc85229eb4f1c89e9ffeb59aaa650 : SUCCESS in 1h 09m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/417dbc45445d44d69acac9e3715b3b8a : SUCCESS in 51m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/229788a7dfb94130a7407ba1b5aa2bbe : SUCCESS in 40m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/976b539d78f84943b712d718e7ba1505 : SUCCESS in 52m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ecc6d7a6b0d433aa25b47c8b88b7759 : SUCCESS in 1h 04m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21af0fa9117e46e1a4abff474f8b8ce6 : FAILURE in 1h 21m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8900e670ef5746b6a2aa4828c08cda19 : SUCCESS in 1h 35m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e6ef01d05a4e7ba4a4946336fe3e70 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3cd7562b00b9beafda9f5ab90f857fc872ce37f2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-16 10:49:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/3 : SUCCESS in 1h 54m 53s","accounts_in_message":[],"_revision_number":3},{"id":"73dcdf24e0aa4a3f0980a403f3259eb58d836f70","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-17 20:32:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dcdfbeed63428829160a30af78efc1a552d86fdf","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-18 07:21:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9fbbf4840c1414dc7707cd82a6aaa052a46a8c9b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-18 07:21:55.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ef96af4b655303759b1fb8ef2ae713ab905f6c37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 07:22:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6672f1e239b8c87ad243d7f43a6ca35e9d8c8906","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:01:25.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/52/790352/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"89cb57369b13649a4d92bc9e5c8477d2a6290539","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:10:53.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/52/790352/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"395575b128e6de266d965ab000e3e6b33cfa955b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:23:23.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/508ffcbd99b54c33a08c2feb46b75444 : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f4cf44d5ae1d434abe414104fc47d7b1 : SUCCESS in 2h 00m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06a1377b7bdf46fc8dc1633550591b10 : SUCCESS in 1h 25m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/988daa57abfe4e02be6a717c06d99dc9 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5435e00f90144be897e3585939acafb : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/461a04fe7fc741fdb2600bc4087e0947 : SUCCESS in 25m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ea7b6fa1a6b4510a26cecaf810f8721 : SUCCESS in 24m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78abe93ad834ab4ae8f9facc39e7c20 : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0523fa7767dd45e1a577301ceede358d : RETRY_LIMIT in 11m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35ea2965a27c46519e9af26b93ff83d1 : RETRY_LIMIT in 4m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9fa17e21fa4b4334889c1dfcbfa6c788 : SUCCESS in 1h 06m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccfb9d0f0ba44769bd88c6765580d091 : SUCCESS in 58m 29s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ca9003138914b21bc67791e7b988f8f : SUCCESS in 1h 25m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0d6768476d9438bb2f11e7403cadec6 : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":3},{"id":"bda300e701725432f322e49b2a5d747bcc15c52d","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-19 00:29:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"335c3fdc1c72d27c7fc34b2dfda2a79216d6d91e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 01:17:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a51b8eda3d54033b791a0d7e4c69feb : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c5d27ebcf824b4bb0c49dc05a8d2866 : SUCCESS in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"084822b7fd980a910dc2fd5d2e4854bf9b5a9bd8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 02:21:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/3 : SUCCESS in 1h 51m 48s","accounts_in_message":[],"_revision_number":3},{"id":"392e4e444bc92a22ceb5e8dab66f3cd7a2d7bfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 02:42:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fdd7286f9438488e9faf8dabf8811b51 : SUCCESS in 2h 11m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f3bc470b8fec48508d2180f14c9a005a : SUCCESS in 1h 46m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d7fcd433c2948f7a4b53228814621e7 : SUCCESS in 1h 05m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c17adc09db614b3b82f5c05ec65035eb : SUCCESS in 35m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a0de659cdfe424a9bbd7f683fb95fe5 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b727b6609b7643ad9e56e414a8d8fb67 : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee8046679fa44fb98f8f86ac9f9da98 : SUCCESS in 33m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16188fef281b4e3caf32a4126b373f4e : SUCCESS in 29m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ef871f480024f899c802906eeefe410 : SUCCESS in 23m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b452381b1b4fae8d31d29fb09e382a : FAILURE in 5m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70ca800b84eb4b11b4df0ed6e677b96e : SUCCESS in 41m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/621c5855b1c041f3a4d02b46c601c3d9 : SUCCESS in 1h 08m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9b40f25c10e5429180ef02a168da24ce : SUCCESS in 1h 03m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95ed83c998b84c91b10ff8212b315800 : SUCCESS in 44m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b3c7a23bb22a4db1bbcde3d208aec630 : SUCCESS in 1h 26m 26s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/020c4919ad474337945e4a4745f602aa : SUCCESS in 1h 10m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/075e864773ec4f33a0251773c4a08321 : SUCCESS in 1h 01m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86c63db58e3e4216b7e4802c23ffbeb5 : SUCCESS in 1h 32m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76aa12297ca14d68844a974432dce7f2 : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":3},{"id":"79b93def0cb9f3d9174878db9433addbc2a871ae","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-20 05:19:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f1a89c8e4eb7a3b27bf4d8e191d1ebf6b9ad5938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 06:08:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd39621b895248e3a228f504ecd28113 : SUCCESS in 47m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/020ed41bca374539a03fbed8cfc22560 : SUCCESS in 45m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"692bcbe1c2c260b51696eb16fa4b5bcc487a76a0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 06:51:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/3 : SUCCESS in 1h 31m 57s","accounts_in_message":[],"_revision_number":3},{"id":"6fe667ec9a46e93e771f90ba9251acd04abea4fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:37:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f9ae8d4720ab424088c037718b62b995 : SUCCESS in 2h 11m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1a1e8c87cdfc47ea8b5c4dbc79636e6c : SUCCESS in 2h 05m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e3f9486e81a443eadb3fc04f15ac44c : SUCCESS in 1h 28m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e0c6a8c06174b3b9bf13b81b162a81b : SUCCESS in 33m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50588e4b2726450abb745b5a0845f5af : SUCCESS in 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03968520a70e4731853486e99460e28f : SUCCESS in 12m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/877df3d17fba40cd8799d4f0812129cb : SUCCESS in 25m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8763ba56478b4be887c777340700837a : SUCCESS in 24m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ad45b600e94d5bad09a4e22110d6df : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c552e23cfa84388b0d1b469386e7d07 : FAILURE in 7m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b101c7caca3e4fb895e1b3c37c71597e : TIMED_OUT in 2h 16m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a775d3b9ded4822ae07d92700119fe6 : SUCCESS in 1h 43m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/374d64fd86194ec498cb355fae7a9ea3 : SUCCESS in 1h 23m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4adaebdc00d4b15bca9a97541aa8bad : SUCCESS in 1h 03m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11f083935e1a4ccf9f198deee3c38fba : SUCCESS in 53m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6fad7ead800a420a89562003f21108f1 : SUCCESS in 1h 15m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d31dec9994d4352973a8275ab9133d6 : FAILURE in 53m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/834b4852a57743bfac1e28f3463cd032 : SUCCESS in 1h 29m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cc06321330a470094ba38b6b6081a81 : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":3},{"id":"6474ab56e269ec42f11b1a1b3191a4ce1caf2841","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-21 05:16:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4199dfc12540e3ba76794e712fe5c66b1f617f2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:02:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a502b7be85549068bf50e59106657f1 : SUCCESS in 44m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e059e4aca9374f0780add170c5432d6b : SUCCESS in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b70413d210ddcc3c2e63a1d4cb48d9b38c5c45fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:51:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/929dc004aa02432184fab5c64d98d775 : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/51f19228247144f6ba21e9c7c3edbe37 : SUCCESS in 1h 33m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/39d64d585a554898b08a4a6dc1b89bb8 : FAILURE in 1h 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eae155d896e4d34bca17b1d59153ebf : SUCCESS in 37m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c02d65ed35f4a9bb8d909e46bbc5e20 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7f1c2e6ba44f5782b66471aa93dc82 : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd1fc9c6e524ec9b39beb29138d3b0b : SUCCESS in 35m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5ae4d99c4bc47588b445c65bf490ef8 : SUCCESS in 33m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b262f254ff4af3a409d7b18109057e : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc33531882b41aca3849355cac7b948 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2696f4dc4c864a5a85e76f23d94d174d : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81de5ebe83b0494abd158e8b140152f1 : FAILURE in 1h 02m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d4835a85cbc344faacfb4837836206c5 : SUCCESS in 1h 22m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61783edc9564458aadac1f04856b5d6e : SUCCESS in 54m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b1f8357ddd414bfdb1c8fc3ff43ce125 : SUCCESS in 1h 05m 25s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/267e52811d1b4cafbc9f0b612a7bb79c : SUCCESS in 1h 07m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/815564dfee4a4ae6bfea51e71421cac9 : FAILURE in 1h 11m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9a4cafccc9a4468a2d53d7b3f3c3fff : SUCCESS in 1h 30m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b174971aac47461a89b67927d76313c8 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":3},{"id":"57930167792b521037648e0e8c36fad421fa7490","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 06:52:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/3 : SUCCESS in 1h 36m 10s","accounts_in_message":[],"_revision_number":3},{"id":"62b462870828c4dbf4211951a554584acb04d7d1","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-22 01:26:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a67cbb3c24c7e8922e1c38de89424b21b47b3725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 02:04:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3782caaffea4413daec80d4d54b014bd : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23f59c2dbeb84e0a9fab401bb219bed6 : SUCCESS in 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0f636be52a1f38f22467004e2328cfb673498bc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-22 03:03:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790352/3 : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e7011b10c36b5d647711340c6bb38e524c13c239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 03:23:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/633eb55a4b2c41569fc5e534663f5d52 : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c8a568294c794b3584c30b39def25fb7 : SUCCESS in 1h 41m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70fcadeba3294fd6a9de02a3a47480d0 : SUCCESS in 1h 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b8271f2997248ea84af0db9a5a83e08 : SUCCESS in 30m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c525d8df9fc4023bb3edfdab4e1847e : SUCCESS in 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9214a97fd247a2b97229320b155f10 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64bc5cbc857f4ee783e6536fd2f3da5d : SUCCESS in 29m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/066fb22fd578416c91a30eaaa84f5a0c : SUCCESS in 23m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92875f58bf064ed1a4a15c8f2a2f6f5a : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfc0653f2fc4c79876e3a4c4aac14b4 : SUCCESS in 8m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6da3858563d54f98aa4f20ce801e7c30 : SUCCESS in 43m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f8420c455444c699e1e63f7984a0561 : SUCCESS in 1h 16m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2663e69541264fef8dac6a5908ea20e0 : SUCCESS in 1h 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/45c8e444c3ea4bd9b0122479a9a69f22 : SUCCESS in 54m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/618c123a51764221bb40dbb05a88a15a : SUCCESS in 1h 15m 16s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8c6e2b3845b54592967bc3fae5a4d429 : SUCCESS in 1h 07m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/544e29b059cb4664a97002a9079bc6dc : FAILURE in 55m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/496494a979a9443a81cd2af510748a4a : SUCCESS in 1h 21m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3a557f2f6fc47d7965d698ea7b72831 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":3},{"id":"4a14d6b53a757b2ea6d054f00387d6f128cf1dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 03:23:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"82b839882894989d28fe5931ce30be7fe5104cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 05:05:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/54b6ceeb87a949d8a74ce1f9c14bdc64 : SUCCESS in 1h 28m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a4d98f4147e34da4a247d164741093f8 : SUCCESS in 1h 40m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53afef0025c2415689125a8a230771b4 : SUCCESS in 1h 00m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b3d3eb63004c1fb294b6b9b468d4ca : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878cbc447e7a41fab337a1ac3c518849 : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9b3654994143b8bca0bc1b64a48fe7 : SUCCESS in 25m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c838cbf8435c4844a5aad38f778fd4d7 : SUCCESS in 23m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd4e16811da487198a7fe980bf0ec09 : SUCCESS in 7m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9dbb76799604d62ae9752eb2e4826d6 : SUCCESS in 37m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30333de48c7448dabbf5fd5fb9c8c8ca : SUCCESS in 1h 13m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/835bddf8478845b2acff810731bf1e17 : SUCCESS in 1h 30m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55b7b7c7c31a4d3d80060c857f2eb730 : SUCCESS in 55m 21s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35ca348b96df40a3821c7e50cb40ab0c : SUCCESS in 1h 31m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d9e574f18474bdaab75582cf46faca4 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"68594f7f567542c0f76ec28bf876df2487569953","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 05:05:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"899ef3e05c6ab6060db7a57a53081d1734596939","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 05:07:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53efea5e70544eaa898bc17e5496114d : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f192153b449877e00c57c2b2a1b8ce89c9c733da","revisions":{"3235c64e2e1b53156f19892da0efc97e9af320eb":{"kind":"REWORK","_number":1,"created":"2021-05-09 06:28:58.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/52/790352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/790352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/790352/1"}}},"commit":{"parents":[{"commit":"9241c76b04e6745cc648ee42037cfe6ddad3600a","subject":"Merge \"Add enable_dhcp filter in get_network_info rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9241c76b04e6745cc648ee42037cfe6ddad3600a"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 06:28:06.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 06:28:06.000000000","tz":0},"subject":"[WIP] Remove add static route codes from HaRouter class","message":"[WIP] Remove add static route codes from HaRouter class\n\nWhen a router set as HA mode, The keepalived process will take over\nthe route entry\u0027s generation. So, the codes that add static route\nis redundant.\n\nCloses-Bug: #1927849\nChange-Id: Id09de6c43c0fab4009336e253c88f54219398053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3235c64e2e1b53156f19892da0efc97e9af320eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3235c64e2e1b53156f19892da0efc97e9af320eb"}]},"branch":"refs/heads/master"},"ca7a5fd23140a2e7bccfc6170665e2ecfeefed22":{"kind":"REWORK","_number":2,"created":"2021-05-14 07:12:05.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/52/790352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/790352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/790352/2"}}},"commit":{"parents":[{"commit":"2c192b78a2017da761e351321b996231ed641893","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 06:28:06.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-14 07:09:13.000000000","tz":-240},"subject":"HA-non-DVR router don\u0027t need manually add static route","message":"HA-non-DVR router don\u0027t need manually add static route\n\nWhen a router set as HA mode, The keepalived process will take over\nthe route entry\u0027s generation. So, the codes that add static route\nis redundant.\n\nBut, for DVR-HA router, in dvr_snat node the keepalived process run\nin snat-namespace and don\u0027t take over qrouter-namespace, so the\nmanually add static route codes still need be called.\n\nCloses-Bug: #1927849\nChange-Id: Id09de6c43c0fab4009336e253c88f54219398053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca7a5fd23140a2e7bccfc6170665e2ecfeefed22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca7a5fd23140a2e7bccfc6170665e2ecfeefed22"}]},"branch":"refs/heads/master"},"f192153b449877e00c57c2b2a1b8ce89c9c733da":{"kind":"REWORK","_number":3,"created":"2021-05-16 08:53:35.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/52/790352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/790352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/790352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/790352/3"}}},"commit":{"parents":[{"commit":"2c192b78a2017da761e351321b996231ed641893","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 06:28:06.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-16 08:49:01.000000000","tz":-240},"subject":"HA-non-DVR router don\u0027t need manually add static route","message":"HA-non-DVR router don\u0027t need manually add static route\n\nWhen a router set as HA mode, The keepalived process will take over\nthe route entry\u0027s generation. So, the codes that add static route\nis redundant.\n\nBut, for DVR-HA router, in dvr_snat node the keepalived process run\nin snat-namespace and don\u0027t take over qrouter-namespace, so the\nmanually add static route codes still need be called.\n\nCloses-Bug: #1927849\nChange-Id: Id09de6c43c0fab4009336e253c88f54219398053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f192153b449877e00c57c2b2a1b8ce89c9c733da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f192153b449877e00c57c2b2a1b8ce89c9c733da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
