)]}'
{"id":"openstack%2Fneutron~790428","triplet_id":"openstack%2Fneutron~master~I3eeecbc18cbd3b715b9070d1cb667f93c01bc35c","project":"openstack/neutron","branch":"master","topic":"fix-bug-#1927926","hashtags":[],"change_id":"I3eeecbc18cbd3b715b9070d1cb667f93c01bc35c","subject":"Fix the increase of fd caused by HA port not recycled","status":"ABANDONED","created":"2021-05-10 12:21:54.000000000","updated":"2021-05-11 12:11:18.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"532446d800da509e88ed63dde9b3b16c188cd78b","_number":790428,"virtual_id_number":790428,"owner":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-10 17:45:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-10 15:06:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2021-05-10 15:20:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-05-11 01:29:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 12:26:03.000000000","updated_by":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 14:56:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-10 15:20:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-10 16:24:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 01:29:03.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"}],"messages":[{"id":"46cee90ba4bf009de0c3bc20b70b52b3b6478a7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"date":"2021-05-10 12:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"220ff40136d56968e3d37ad8d5d60af0f5261773","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:34:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f0415a01d4d4de283bc69175a9eb108 : FAILURE in 10m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d56cf18b5f1a4191ab1f549f43de3754 : FAILURE in 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b0da214a020488aa2c0ec455c33f3ab43f43669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"date":"2021-05-10 12:46:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"edf1404c33a9c02a099ecd776dcba4f14a3cfc6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:32:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c4f6e98fa3e41a7a3500f0aaf42bf56 : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f65b37254e5f4e70b3e51e9cce00413a : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e663c882df92439be1b5a2f1ab074331d1f61154","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-10 14:56:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790428/2 : SUCCESS in 2h 09m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4c62d0b9c4e379b6b2b39144141576817487ca4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:06:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9aa3a22947424f1bbfeb89e945157be8 : SUCCESS in 1h 35m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b19064a46d340fea5e0eef25dcd1be8 : SUCCESS in 2h 04m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc2a693df2434bb4973b412bc2eb226f : SUCCESS in 1h 23m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/736306e3ffbf4ad68dfe0847b15fc403 : SUCCESS in 44m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65ace896f50e479b89b9f8b6db2eae53 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9985b420ed2f4030af645fccbd9d8a84 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cf098c727bb43138db5f2ca80c8967d : SUCCESS in 36m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29c42f61faf14671be03826eaf7cadec : SUCCESS in 33m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c053bf325e674ccdbd311f93c80cd967 : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf4ef5d40d142bc8ea423b13f13f752 : SUCCESS in 11m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59a88419c92a4d2b8c325ca2b1b5445a : SUCCESS in 54m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8151011879254a48afc8984e3b2fb652 : SUCCESS in 1h 25m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2a70b1ee235949009f673d942bc78385 : SUCCESS in 1h 18m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e47f13b7be14dc2a50c69403a09c6f4 : SUCCESS in 1h 00m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94c9b8d14c3449a2a32e8940a29681da : SUCCESS in 56m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5857ea62cf894d84aa455b2a1e376213 : SUCCESS in 55m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bec1e81ee6d84ea788676c6aae9e56e2 : FAILURE in 1h 27m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5427d4c7302443b59f100ac5d4649a91 : SUCCESS in 1h 24m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/607af8f2d00242adaf242d4d8806cc8a : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1c1ca8b99959080a8d3e8dc44928355abe71a190","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-10 15:20:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81bc0e8b1bee18677c4c9937a901fcaea0270958","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-10 16:24:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/28/790428/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"bf1c028acc2ee24f67f3c943f0a8a4923f4aad9a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-10 17:45:48.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/28/790428/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"65494ab6c4a338d838cd184666f46456c3ff425e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-11 01:29:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"532446d800da509e88ed63dde9b3b16c188cd78b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"date":"2021-05-11 12:11:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00762cbc3394f82c60e79d0f4046a73ae405f52d","revisions":{"df33fe47970a33b81755e84599b59d00d325bd18":{"kind":"REWORK","_number":1,"created":"2021-05-10 12:21:54.000000000","uploader":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"ref":"refs/changes/28/790428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/790428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/790428/1"}}},"commit":{"parents":[{"commit":"9241c76b04e6745cc648ee42037cfe6ddad3600a","subject":"Merge \"Add enable_dhcp filter in get_network_info rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9241c76b04e6745cc648ee42037cfe6ddad3600a"}]}],"author":{"name":"fanxiujian","email":"xiujian.fan@easystack.cn","date":"2021-05-10 12:20:00.000000000","tz":-240},"committer":{"name":"fanxiujian","email":"xiujian.fan@easystack.cn","date":"2021-05-10 12:20:00.000000000","tz":-240},"subject":"Fix the increase of fd caused by HA port not recycled","message":"Fix the increase of fd caused by HA port not recycled\n\nAdding ha port cleanup causes the ha port\nin the bridge device to be removed normally\n\nCloses-Bug: #1927926\nChange-Id: I3eeecbc18cbd3b715b9070d1cb667f93c01bc35c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df33fe47970a33b81755e84599b59d00d325bd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df33fe47970a33b81755e84599b59d00d325bd18"}]},"branch":"refs/heads/master"},"00762cbc3394f82c60e79d0f4046a73ae405f52d":{"kind":"REWORK","_number":2,"created":"2021-05-10 12:46:07.000000000","uploader":{"_account_id":33423,"name":"fanxiujian","email":"xiujian.fan@easystack.cn","username":"fanxiujian"},"ref":"refs/changes/28/790428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/790428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/790428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/790428/2"}}},"commit":{"parents":[{"commit":"9241c76b04e6745cc648ee42037cfe6ddad3600a","subject":"Merge \"Add enable_dhcp filter in get_network_info rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9241c76b04e6745cc648ee42037cfe6ddad3600a"}]}],"author":{"name":"fanxiujian","email":"xiujian.fan@easystack.cn","date":"2021-05-10 12:20:00.000000000","tz":-240},"committer":{"name":"fanxiujian","email":"xiujian.fan@easystack.cn","date":"2021-05-10 12:42:57.000000000","tz":-240},"subject":"Fix the increase of fd caused by HA port not recycled","message":"Fix the increase of fd caused by HA port not recycled\n\nAdding ha port cleanup causes the ha port\nin the bridge device to be removed normally\n\nCloses-Bug: #1927926\nChange-Id: I3eeecbc18cbd3b715b9070d1cb667f93c01bc35c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00762cbc3394f82c60e79d0f4046a73ae405f52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00762cbc3394f82c60e79d0f4046a73ae405f52d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
