)]}'
{"id":"openstack%2Fneutron~790912","triplet_id":"openstack%2Fneutron~master~Ib40eabd6a8e9d59896e0e20383d8061eb4b5c710","project":"openstack/neutron","branch":"master","topic":"bug/1922934","hashtags":[],"change_id":"Ib40eabd6a8e9d59896e0e20383d8061eb4b5c710","subject":"Set the default ``Backend.lookup`` timeout to 2 seconds","status":"MERGED","created":"2021-05-12 08:01:29.000000000","updated":"2021-06-05 08:00:03.000000000","submitted":"2021-06-04 23:20:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790912-bug/1922934","meta_rev_id":"be0edcc6f1aa97a0083fe1a8d87b634c686d4d52","_number":790912,"virtual_id_number":790912,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 23:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-17 20:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-26 07:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-26 07:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 08:06:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 08:15:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-12 08:15:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-05-12 08:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 09:44:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-12 10:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 20:29:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-18 08:44:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-26 07:40:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"91415fe370f3005fa9c6e6dd14fce664a00cd069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 08:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8415bcca3f4a80e99cd0d53c01f732a145f29ddb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-12 08:06:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/790912/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/790912/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1af66598e06e9cffc31cb194bf8f84f93cc9d2e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:48:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d630e5380c7f45bcaaa2a7ffffaba658 : FAILURE in 44m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00e195a25c0a4979a82ab6c5af6bbc57 : FAILURE in 44m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"856a747a17e749953e3be621d3775fadfa03b675","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 09:44:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790912/1 : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d954120f199326b5aef149240f60b5f13dab2bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:16:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/803418ec9a3b435988152498911b20c3 : SUCCESS in 41m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/78f74a95825540a5a8340e89cf87313c : SUCCESS in 2h 03m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36bb4aa62790473db4fdd745d55c2ddc : SUCCESS in 1h 48m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b651850c81d84a65aee7990eda743621 : SUCCESS in 1h 19m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0fc489f6cd2b408fb45e42b92569b89b : FAILURE in 25m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15db09dabe4843babc5cc3564623a003 : FAILURE in 39m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b25f3f607847da9ecc937be20a1863 : FAILURE in 35m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1a81516ff84eedb061f794693a2482 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92b57734e1bb4c1b923cbf803f24e313 : FAILURE in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36183f3a6e0345cd940b33d0ba65689c : FAILURE in 20m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c180a3deabd4f8bb6deb5390d5edf1a : FAILURE in 24m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7478123d6d489389968cba3361b9f1 : SUCCESS in 8m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/584f0c24b49a4c9daab160916f239dba : SUCCESS in 3m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60849b86c1f74cccb372cf2e4c464a28 : FAILURE in 40m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4468e4fbf3614ecdb4ad52a672d4fadf : SUCCESS in 1h 11m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6a8ca37784a34932ae17bb6907119668 : SUCCESS in 1h 14m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae074847f836473c855a1480b6f2bc4e : SUCCESS in 47m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f1ff68d112fa436198934cfbc387a961 : SUCCESS in 1h 09m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b59b5c669e2a43c782bb116854e01792 : SUCCESS in 1h 00m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec83de8f49f746f9be27a616c40fd604 : SUCCESS in 1h 06m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b8f3a49e6344a3da804193707c0f28c : SUCCESS in 1h 22m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86de524c009e49f4bda4a19e6fa7b961 : SUCCESS in 55m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd90337639c24909a44c1ce31b678a31 : FAILURE in 23m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38ee6c5e584843fda6669a46b7976f13 : FAILURE in 22m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e4aedd2b2654d1598666f55f3714320 : FAILURE in 24m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f793e3cdda148338b44e22c14a28d9c : FAILURE in 31m 13s","accounts_in_message":[],"_revision_number":1},{"id":"44c40fd7f3d844b142d456a9b9aa1c3dea4cb928","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-13 08:01:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1521f1a49b7bda958f739a7869e0e48b71a251c8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-13 12:15:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c788177cde6f286bf782d27904442e06757cee5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 13:20:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a4341be8aaf553228eee6a9f4ff03846162afa31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 13:24:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25d79d903c2ad6483c22f0f8f3f323fe54808af4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-13 13:29:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/790912/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/790912/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c6301d205396918da414e677ee5d3bcc6cd0a238","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:13:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d2f1ac9446a47ffa6d33322fd4d83e8 : SUCCESS in 47m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e117ddd4d072436798d75f12959e25c7 : SUCCESS in 47m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58f6be7608c579b7b631c60442779cc2837f182b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 15:22:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790912/2 : SUCCESS in 1h 57m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a90b085482b1162fab8dcf47bbe90fa8f0c5aa4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:59:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3a12200933054b04acaed1879e0749b0 : FAILURE in 45m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/22c12a6253604a16b8d8116329bdcb8d : SUCCESS in 1h 51m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6c0c9f464dc54d99a74f01da8859eb0b : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34211924c70342f7b0e221e020804c00 : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9c34c2c23f041889d3d87bb7d5b0e48 : SUCCESS in 2h 31m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a3ebf6cb0d246c89120cb97db06d5c5 : SUCCESS in 35m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b39b5256f44fe3bb865a33c0e6db1c : TIMED_OUT in 1h 02m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8931fbd8a622442cb7909e87fd6ce4a8 : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248a9ad069bd4a1ab2fdd565e1e27803 : SUCCESS in 28m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/719a3df45838411bb5b7acd211792736 : SUCCESS in 27m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a87dd1d8671840e2a132f902dd452604 : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da8e12426c44866abceb7eff8e9b52e : SUCCESS in 10m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23c4b3c19aaf4d2786e315a5492c231b : SUCCESS in 3m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/36735034bc3b4a1691ee3be21b9a4ed6 : TIMED_OUT in 2h 16m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b777b25ca5e847028051cee61062c29c : SUCCESS in 1h 32m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ff31b59eca954674a5bc1489cd591bad : FAILURE in 1h 39m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e284a616d4149e9b04a236c59e6c929 : SUCCESS in 1h 17m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f6e4049624ca4476b4b88b1e5cff7ced : FAILURE in 1h 24m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ef02c34349d14aa093ba8cf01c72871e : SUCCESS in 1h 18m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ce5ec4443e5a4ab18b8a2784d08701ac : FAILURE in 21m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a0d526efb3e42ba9e1cb4717bad8e6e : SUCCESS in 1h 20m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81fa45c832a64cefbd94a11c81101813 : SUCCESS in 56m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e36e0e79641d4d79b540bbb46d539d87 : SUCCESS in 1h 07m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/83729d38ef8446009a7bafe7ea464d98 : SUCCESS in 26m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2cf7236bc95149c7a619a78ce4e7c009 : SUCCESS in 1h 30m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e2b2d57914d4834807f5a279008df2f : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f429cc0d0775d75b842baceee0594a997c42a60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-17 07:46:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"75bb807775cfeb5d9e659c7225571fde299e1ae7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-17 07:50:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b974e9eca1949f3d038ddc255988bc13bc8d4da7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:40:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b31800fd6d3a47bd9f1337561fbfee9e : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/539dc689266f48e7944a5a1b401d2259 : SUCCESS in 47m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65e7c59ddc61238f8efb185fe332d831b80e5bba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-17 09:36:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790912/3 : SUCCESS in 1h 50m 05s","accounts_in_message":[],"_revision_number":3},{"id":"623db25da723e50794dc9a0da253c1c57dac3680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:00:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6fa6cee3ca3b4d9da6bbee3df9dcd01c : SUCCESS in 45m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/01cdfdedab3a44a3b8beac1d7e587e0b : SUCCESS in 1h 38m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3d9e3a15271c4817ad86375f5aedcecc : SUCCESS in 2h 08m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1eaf1a8df1634fab9e4d559f53a5433e : SUCCESS in 1h 39m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5ed7d7020df48ab8a72ee7010c16bae : SUCCESS in 2h 08m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b55ba805b2465385e19024e4958ae0 : SUCCESS in 34m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/703240dd103d42c69dea11618b6ce54a : SUCCESS in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad25e2b88104a67bf83b7f7e7a3ada3 : SUCCESS in 14m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51d9ddc1daa249318aa2a3a40c081208 : SUCCESS in 26m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc48d855a79349f08da5a504f576f4a2 : SUCCESS in 30m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b06bfded32024923a43889896e9712ba : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b1cc2831c744249ec498619b7d3128 : SUCCESS in 9m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8270db554c504c8f96361c3a43ba061a : SUCCESS in 4m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9b2523a1c7643fd950cc671d00d89e5 : SUCCESS in 53m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa81a789750b49f0af16c516b9a12392 : SUCCESS in 1h 25m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ca3bdd75ea4841e58230f2aadc78c544 : SUCCESS in 1h 15m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d934a01518b94d8a97020964d21464c6 : SUCCESS in 57m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58d5355e5b6c465da5801196667ab962 : SUCCESS in 1h 07m 45s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b7c9f087b2d46f3923a0ab64f19e58a : SUCCESS in 1h 13m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/00ec6bf930244ecd92db4df1c2bac799 : SUCCESS in 1h 19m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f069354d44749d4966693ec42d9b485 : SUCCESS in 1h 28m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb43c1b2a964e418a914e22f02002f9 : SUCCESS in 50m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d5a633289c4f4121b18401c9f1c03415 : SUCCESS in 1h 02m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7396d03232e64ebc92308d1f172c50df : SUCCESS in 30m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d27eaefeb4841abb91e2a15ab344411 : SUCCESS in 1h 18m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ab825b6ac09e46738deb9d4c513755df : SUCCESS in 1h 18m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7118d9c1c6de2e71a860995315fb27c41730afd6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-17 20:29:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c0ab8335590ca6c0fdaea5f4268d758d6ba9f68c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:44:18.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/12/790912/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"80899f503b3634cf9f1f7160eb564a628dce2005","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:16:18.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/12/790912/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"5c20c616786d4898b0332b079af8b79be559c84d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 07:40:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9acce7da33154956e169be5d25aaf0fb08453099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 07:40:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"322aea2c62d9b94dafb41f5c78136c4242ceb70e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 10:00:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3fb7b703f8d24572a4823d4cb1bb6d61 : FAILURE in 52m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/faa87f4db38d409487176a56fdef2734 : SUCCESS in 2h 03m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6d9eb066d6b845ee808ebb04aea84d0a : SUCCESS in 1h 38m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d07d5d188f03479fb253cd5f59971b8e : SUCCESS in 2h 00m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9edc2b6e37b341c99aedb00e38ec9c0f : SUCCESS in 1h 04m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e3828f643b454f9415f53f6c1b72cf : SUCCESS in 35m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7610c1a0a4184d8c88085bf44f1c7800 : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f60e59db86df49adbf9f498dac4f459d : SUCCESS in 38m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7feb461f9545259a6f2cc93c5b8848 : SUCCESS in 29m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a552a2d20e4e07bf02c8179f0f0b01 : SUCCESS in 10m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef71db41eb6848928a29a1a53670afb2 : SUCCESS in 4m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3689628f0fa94fc4adc81aadaa32e8a4 : TIMED_OUT in 2h 17m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0eff2952260242ac83b488f00b6fee5a : SUCCESS in 1h 25m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/38303ba1adc344d5adca8d4d1e65cde0 : SUCCESS in 1h 34m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7df7a4723a0c40b3a963526fa2bd60d9 : SUCCESS in 1h 10m 03s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a98a59a138bf4a6db4353e28289f3d41 : SUCCESS in 1h 31m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b59af027fa3c44b0909aeff98d7bb401 : SUCCESS in 1h 07m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ff77378959154b80aaa4f1c1e9050eee : SUCCESS in 1h 07m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/291eef11ef544bc3ad2b6c2291f12cfb : SUCCESS in 26m 48s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1be1a8dd0fce44308ae2a1c8b65fdd25 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":3},{"id":"1b90b35f3a4ddb8643d70aa8cdf21778637e6556","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 15:59:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"781eb389cb251086f4d86fe61f2c2accec6c1d03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:05:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5b06ad89adb4e7bb5342587d1551b16 : SUCCESS in 47m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ddb643ea8a445d49e482b88d831edb4 : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72174a70226d6b1fed71b54802e476b32715c1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:23:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3902e978ce7f40f8bd1563693904c0ed : SUCCESS in 46m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7737739f55fb459488d00ac26e9834eb : SUCCESS in 2h 02m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50747135477745fe897e51528b90d2e8 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f99db7b100c4b48a073f93736a4069b : SUCCESS in 1h 33m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a86f547829be411fb6ab5fa997a1ea2a : SUCCESS in 1h 55m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3df4f2e9bd74465e82737c0c5d46edad : SUCCESS in 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2cf81425c2a4e03bb78aa636ed6bb99 : SUCCESS in 36m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3946394ca1e3438f8d53301af768ece3 : FAILURE in 17m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc6eb401b25d488bbc28ce984cd5f98a : SUCCESS in 32m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7cddd9aa194eecbb858580447f74fc : SUCCESS in 28m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2416712f2a44e51a9f4489f8c144bf5 : SUCCESS in 26m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636a25d923444c8fa2bb426083cf0e9f : SUCCESS in 10m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e62a629d187428080a4da5283ebd33a : SUCCESS in 5m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/741fbd3a5b904be68f59c41daf4c15a2 : SUCCESS in 49m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4db5928c2ed4c6f8f0832c4c1cd2545 : SUCCESS in 2h 14m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/42b378113a73498791eb6522788b1d82 : SUCCESS in 1h 14m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ecf99f284cba4c3597a78620da9065bd : SUCCESS in 55m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fedff6775128458d83cc740e842a86d4 : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a036345dd242425aacc33f666f2621e4 : SUCCESS in 1h 24m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c47173830004adf851c9b637d8c5c08 : FAILURE in 25m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dba4f97b40354bfe8e622bb5268d7b04 : SUCCESS in 1h 53m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97fdb6880def48bdbe1c1d2430c2cec6 : SUCCESS in 1h 03m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b29c0608abde478595c2d9ca978fe1ec : SUCCESS in 1h 04m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d60a6849c2f0423893dc054d35100112 : SUCCESS in 33m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/01c7f145b934459ca665453c11d482d4 : SUCCESS in 1h 28m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/969d42a292c5402b81d89595340b019f : FAILURE in 1h 34m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":3},{"id":"89dd30c508478e117b716aab5f8cb1d8dfdb8975","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-03 20:11:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a4ccb6bb6a9b82c0d32c59ce6ff8e228e2cb62ba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 20:30:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790912/3 : SUCCESS in 2h 08m 54s","accounts_in_message":[],"_revision_number":3},{"id":"25b166fda97a896b7f2bc1040dffd2285cb64733","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 21:05:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"328b72e10b257e2a650e360cd3fccff48cc2996e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:09:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a22421d362a43999a97e70b19afedbf : SUCCESS in 29m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5eda261465043e2a85aaa87af2c1a42 : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e717308508494fcea45eb6eb4849de57 : SUCCESS in 11m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac8e3c6c28e41a59bbee0d91d5f36af : SUCCESS in 24m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd207b7ecdfd454faf6767286611f17b : SUCCESS in 22m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d5fd07e967f4ddf91b31279f1663311 : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4552e965658e411297d3a66b073bb737 : SUCCESS in 9m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d019b07995c54b31a5b6209f45129d98 : SUCCESS in 4m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b962cb6a1cc4301a4414583b42967e6 : SUCCESS in 58m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/658fe612c3d147f986be6e686bd33b8a : SUCCESS in 1h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"839a7564acc75e5fbdd78e73e42033828b700cff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:10:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"82d11b8fe94392fbfd69863e93b33e69bbd6db7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:28:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b77556a86e248ed9c4ccae4b53a1075 : TIMED_OUT in 1h 01m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49b5d76116f5446e81b98c9e26337dcf : SUCCESS in 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3dc770cf43212a093e8e72c8c9b7feaf6c7b38de","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 22:48:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/790912/3 : FAILURE in 1h 39m 59s","accounts_in_message":[],"_revision_number":3},{"id":"a8400a5ac9c3b2d9807e6a029a71ea95b4c322e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:19:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48bd7ade5f75440ab49a6cfef36c1c66 : SUCCESS in 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9afa2b7e9e547908a4ce6c56d8fdbcf : SUCCESS in 15m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8f84fd0b81c475eabe379f50e2da840 : SUCCESS in 24m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6631d3eb4719435ba364efd8160c443c : SUCCESS in 28m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd61bfd23b446648416bf7dda65ae4f : SUCCESS in 9m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e18c7f9a3e3463cb68f11736d95d10c : SUCCESS in 4m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd8879e77a644d7a8597bd7e12a30598 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b1e87eedf9d34382a1f1fba47034f200407b7a64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:20:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8138e5580b8378c5d9abb7d6a1de5781cd8b75c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 23:23:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5455611dcdcd49b0924ee393f519bbc2 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"be0edcc6f1aa97a0083fe1a8d87b634c686d4d52","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-05 08:00:03.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/790912/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9bda4b1e439b5e761b445ce6686166eb41a0e45","revisions":{"e3a1f7ad0c3729964a91bd40d80cd384e268c2bf":{"kind":"REWORK","_number":1,"created":"2021-05-12 08:01:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/790912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/790912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/790912/1"}}},"commit":{"parents":[{"commit":"ba2dc74c9acf788c78d2d26a64256bc4e39d7af7","subject":"Merge \"Adding placement auth options to oslo.config entry_points\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba2dc74c9acf788c78d2d26a64256bc4e39d7af7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 07:55:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 07:59:32.000000000","tz":0},"subject":"Set the default ``Backend.lookup`` timeout to 2 seconds","message":"Set the default ``Backend.lookup`` timeout to 2 seconds\n\nIn order to broadly cover the issue described in the referred bug,\nthis patch sets a default timeout of 2 seconds in the\n``ovs_idl.Backend.lookup`` method.\n\nThis timeout should cover most of the situations where the IDL local\ncache update is delayed. This patch does not change the default\nbehavior, except it will not fail if the DB cache is synchronized\nwithin 2 seconds.\n\nIf we don\u0027t pass the notify handler or the backend does not implement\none, in case of not finding the requested register, the method will\nraise like before.\n\novsdbapp library is bumped to version 1.10.0 to receive the change\nthat introduces the active wait in ``Backend.lookup`` and adds the\ntimeout parameter to the method signature.\n\nChange-Id: Ib40eabd6a8e9d59896e0e20383d8061eb4b5c710\nRelated-Bug: #1922934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3a1f7ad0c3729964a91bd40d80cd384e268c2bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3a1f7ad0c3729964a91bd40d80cd384e268c2bf"}]},"branch":"refs/heads/master"},"311a5565c0e7315400e1c85107dcd9c5d62614f6":{"kind":"REWORK","_number":2,"created":"2021-05-13 13:24:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/790912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/790912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/790912/2"}}},"commit":{"parents":[{"commit":"ba2dc74c9acf788c78d2d26a64256bc4e39d7af7","subject":"Merge \"Adding placement auth options to oslo.config entry_points\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba2dc74c9acf788c78d2d26a64256bc4e39d7af7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 07:55:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-13 13:24:19.000000000","tz":0},"subject":"Set the default ``Backend.lookup`` timeout to 2 seconds","message":"Set the default ``Backend.lookup`` timeout to 2 seconds\n\nIn order to broadly cover the issue described in the referred bug,\nthis patch sets a default timeout of 2 seconds in the\n``ovs_idl.Backend.lookup`` method.\n\nThis timeout should cover most of the situations where the IDL local\ncache update is delayed. This patch does not change the default\nbehavior, except it will not fail if the DB cache is synchronized\nwithin 2 seconds.\n\nIf we don\u0027t pass the notify handler or the backend does not implement\none, in case of not finding the requested register, the method will\nraise like before.\n\novsdbapp library is bumped to version 1.10.0 to receive the change\nthat introduces the active wait in ``Backend.lookup`` and adds the\ntimeout parameter to the method signature.\n\nChange-Id: Ib40eabd6a8e9d59896e0e20383d8061eb4b5c710\nRelated-Bug: #1922934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/311a5565c0e7315400e1c85107dcd9c5d62614f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/311a5565c0e7315400e1c85107dcd9c5d62614f6"}]},"branch":"refs/heads/master"},"f9bda4b1e439b5e761b445ce6686166eb41a0e45":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-17 07:46:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/790912/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/790912/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/790912/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/790912/3"}}},"commit":{"parents":[{"commit":"3b3398b8c5c0d597e0995ab7fdccef6d64d456cf","subject":"Merge \"use payloads for NETWORK callback events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b3398b8c5c0d597e0995ab7fdccef6d64d456cf"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 07:55:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-17 07:46:02.000000000","tz":0},"subject":"Set the default ``Backend.lookup`` timeout to 2 seconds","message":"Set the default ``Backend.lookup`` timeout to 2 seconds\n\nIn order to broadly cover the issue described in the referred bug,\nthis patch sets a default timeout of 2 seconds in the\n``ovs_idl.Backend.lookup`` method.\n\nThis timeout should cover most of the situations where the IDL local\ncache update is delayed. This patch does not change the default\nbehavior, except it will not fail if the DB cache is synchronized\nwithin 2 seconds.\n\nIf we don\u0027t pass the notify handler or the backend does not implement\none, in case of not finding the requested register, the method will\nraise like before.\n\novsdbapp library is bumped to version 1.10.0 to receive the change\nthat introduces the active wait in ``Backend.lookup`` and adds the\ntimeout parameter to the method signature.\n\nChange-Id: Ib40eabd6a8e9d59896e0e20383d8061eb4b5c710\nRelated-Bug: #1922934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9bda4b1e439b5e761b445ce6686166eb41a0e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9bda4b1e439b5e761b445ce6686166eb41a0e45"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
