)]}'
{"id":"openstack%2Fneutron~790916","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ie36cb025540c158fbbed4126be5d654888715300","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1914754","hashtags":[],"change_id":"Ie36cb025540c158fbbed4126be5d654888715300","subject":"add_fake_chassis() may need to create a Chassis_Private","status":"MERGED","created":"2021-05-12 08:18:09.000000000","updated":"2021-05-14 04:23:05.000000000","submitted":"2021-05-14 04:21:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790916-bug/1914754","meta_rev_id":"0a9960876d1ab9bcff49818ae8498860afb2864b","_number":790916,"virtual_id_number":790916,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 04:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-12 19:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-14 01:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-14 01:40:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 08:18:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-12 10:25:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-12 11:18:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 19:57:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-14 01:40:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"1ebb76384a174158008eda14654cfd2b8e2fedfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 08:18:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b0cd6427f483ad552b5f23024a79021c5a53c65","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 10:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790916/1 : SUCCESS in 2h 06m 24s","accounts_in_message":[],"_revision_number":1},{"id":"67271f095b3a9ed8a1e025b441cc6dc887d346e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:18:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/8be1d0022b82405daff3b7c8e8a9ba9c : SUCCESS in 49m 55s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/48407cd215324572a80f36789aa3b67f : SUCCESS in 1h 57m 04s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/71f137aea4754db980a2c8c383ec34c2 : SUCCESS in 2h 22m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/0ea198948f1e4a829d3434e53a09fd33 : SUCCESS in 1h 30m 13s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/8cc228e26b744703a58d5a2c713f7907 : SUCCESS in 2h 06m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2a40dc0b71840978c138d49a17e78a0 : SUCCESS in 41m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4921281cb87439c9fed5db1ddab2029 : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c3d67d3ff2746d0800545993dcc93da : SUCCESS in 28m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52891f76a93d48479a42e86c5edeecc9 : SUCCESS in 27m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3efd39c6e61846e5bc77cc30bb6bc2d2 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66986b0ca3fa40558e5f2c167f0559f8 : SUCCESS in 10m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b801e044ab1d40b880a91e44159a2344 : SUCCESS in 52m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3a3f4bf9a9da4d1387d3935ca5ef500c : SUCCESS in 1h 18m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/70fa46eedd3b4eb8acfe1a26a0ae4718 : SUCCESS in 1h 20m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/418b8c3a7ec4457592ffdbd5cfbc3013 : SUCCESS in 1h 35m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35e36f12f5db44e1a7f1437ad88ff407 : SUCCESS in 1h 21m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/816366433ecf4b0cba3fb8cc1cda4999 : SUCCESS in 1h 18m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/098c69728c02458095ef8d713fe5c89c : SUCCESS in 1h 12m 19s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e6a49d51fcc40b38414ea0f5a441394 : SUCCESS in 1h 08m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a97ae45547e4a39967845cfc8b2846c : SUCCESS in 1h 11m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a521f9ab4a334a439d88cd1da25cfc4b : SUCCESS in 1h 34m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98970a4cfa104a1b80ced86fbb4f2c66 : FAILURE in 1h 31m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7df11eb86f564891947e47d96df3685b : SUCCESS in 1h 10m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/78a25f2c1778495abff2eee7e1d99d63 : SUCCESS in 57m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/034b11ce885e4f24940d29d134b10233 : SUCCESS in 1h 49m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d461da33289d4871a77d0e2b2a729a4f : SUCCESS in 55m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f82d25d81cd94ef1af0730e0e8a7c8fc : SUCCESS in 1h 17m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2b1d0f18748140a59e7e3cfeb98f798f : SUCCESS in 50m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b51ad25270bd419cb6af77fc5afbb7d3 : SUCCESS in 1h 12m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04eef38cfd8246d78f304cf96eafc4d3 : TIMED_OUT in 2h 15m 16s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/239dee9551f942c081d664bd3d71d033 : SUCCESS in 2h 55m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a404be51273f419abd625d2c1b1ba1d7 : SUCCESS in 1h 01m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/865eb58eb3e64ee09690e25aee222c2e : SUCCESS in 1h 45m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/394849d0cf7a498b98244304f5b89e09 : SUCCESS in 1h 46m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d79b4a01aca4a778bd4fe818bba7de8 : FAILURE in 1h 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6a8c8fc09dd1b5679d101f1df61f82ee1238482","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-12 19:57:37.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9acc59787b47a96ce4a7943e31fb5ea02cfd033","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-12 19:58:30.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e recheck\n\nSorry, I\u0027m doing too many rechecks recently. That one wasn\u0027t needed :/","accounts_in_message":[],"_revision_number":1},{"id":"f7734edf521ac1a68e92652ef21425d000164e42","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 21:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790916/1 : SUCCESS in 1h 59m 16s","accounts_in_message":[],"_revision_number":1},{"id":"194861a78501016ea5cc4d841b3cca846e239539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:53:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/ee3936d6334245ab8d09c5b2fe72db29 : SUCCESS in 39m 11s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/33ec55d068714fe2b3903d964ebf487c : SUCCESS in 1h 12m 29s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/25cafa3bf0f74a95b19776491ad3d91a : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/45c56e44c07044bb8009e70bdbdc9c01 : SUCCESS in 1h 09m 17s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/61c641cb9f49492e9b014b8fc3a3912e : SUCCESS in 2h 21m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16c7f72862fa4a6b8bac06474a06435c : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd42c2b4de04277bda9fac9ddbb69fb : SUCCESS in 10m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549b693b0704409da479d834667e1551 : SUCCESS in 18m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bd5a16c88704428a86257cab85dcf10 : SUCCESS in 29m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d415e3c7c43c48e6ace90e91bd871ac0 : SUCCESS in 17m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b4cc037c714e52ab0e6cb45fc99658 : SUCCESS in 12m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/252bf61d81064e2e8649f02e38bfba0b : SUCCESS in 49m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/43cc5b975ee24657833344d926b0a98a : SUCCESS in 1h 27m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34e5be89198c4dc995b0bdee01ece2d7 : FAILURE in 1h 20m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e012f45f9bb046f390a614e5c78fd2c9 : SUCCESS in 1h 44m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c27a209c305a4efdb7fb4197fc8e32f7 : SUCCESS in 1h 00m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4eca5b64a7404c83a99327a5866f35f8 : SUCCESS in 1h 29m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/60553a2524e84324b98a9daa300d74d0 : SUCCESS in 1h 07m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf95effac66f4a2896182d46d137b1aa : SUCCESS in 1h 15m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/007fa290ee174bf68cadf96e21a210b3 : SUCCESS in 1h 15m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/601fe554ab3243928fbe862cb9af7d5d : SUCCESS in 1h 22m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b3c942755264d07b8dfd9a1165c46bc : FAILURE in 1h 18m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac5550690d0a47f2b2d2a672fe597ede : SUCCESS in 1h 10m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fde2eeb4d3384ec0938b858cfe18e042 : SUCCESS in 1h 02m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff2421eb8cdc455ea324d9a0ebfb94bd : SUCCESS in 1h 50m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f80a56279c334797b5ac19f5411aa0a5 : SUCCESS in 1h 03m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d80ed3c87add430cb21aa6467eea79c0 : SUCCESS in 1h 26m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4dbc35cf40394585bbd1c2c3ef670a94 : SUCCESS in 51m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5ee45fca3d9248d9bae41e4b43b91448 : SUCCESS in 1h 20m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/34670d43cc874b81927cce6047eead6b : SUCCESS in 1h 03m 49s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d62600e0331444508f7d05499e970a0d : SUCCESS in 2h 52m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e489ee01634f46a184e281891aea6bc7 : SUCCESS in 56m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1906e691145b493da06e70ab6d7d824a : SUCCESS in 1h 38m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e23ccdec399545e5acf9c95691b11952 : SUCCESS in 1h 30m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e37f669a85584776ae53f53da93cc16a : FAILURE in 1h 49m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d64c91e048b032a846d4c823735b1a75479a4f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 07:52:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a8005c1a030dc0e54d98ac9ae3bca8482570fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 11:41:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/280c5357558246dcbd8ce9415eecc23a : SUCCESS in 35m 21s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/bf9e5f14f34a437ba583c6d70a6b47cb : SUCCESS in 1h 37m 28s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/6c8f954393344f0fa0d8cb9da6ea4a19 : SUCCESS in 1h 34m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/b4ed25d2ec8d4fb1893234b885741862 : SUCCESS in 1h 15m 33s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/280da7b422a646a1a81545c44a1a0b10 : SUCCESS in 1h 36m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5995fee0ebfd43e5b0314d396a2ec8ce : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e062848f5b49049b6805f47a225c96 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f7f85be269402c9cc0caa984cb241b : SUCCESS in 27m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1615360a40b4f98a15047d48a626413 : SUCCESS in 26m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bda6f3dfdf746fc9d8fd0ecd3c68fca : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105d8add8a0348e79db1a75a8fcf2871 : SUCCESS in 9m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27897f861fe44d30a09e2466657577ce : SUCCESS in 44m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/16ef4388cec540f7b9e7ce112088504f : SUCCESS in 1h 16m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a8a8ed321e841829f35c0827d14b0e9 : SUCCESS in 1h 08m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/680ea1b8a92546a8890f021ebc773da6 : SUCCESS in 1h 33m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab0086111797400eb931d13851a71e43 : SUCCESS in 1h 05m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/602b3bb06d334c5a9dc0089df3519176 : SUCCESS in 1h 10m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/08274542fc2245289f13f61b364644ba : SUCCESS in 1h 07m 02s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b7eb98cf1964f26b8cdc25c79a691fe : SUCCESS in 1h 19m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2df1efd90b1f4151bb4ff799fa903a7d : SUCCESS in 1h 04m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d87bf38ce2f0425bad076c254c58fb26 : SUCCESS in 1h 30m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bfaf2023fa0047d6a8167a3f35743bd1 : FAILURE in 1h 33m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1d203b9235244e5a2f89672a616e7f7 : SUCCESS in 1h 10m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f82ad230d79347fe9e1ec371030d1eda : FAILURE in 1h 00m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbe4902c9eb342ce84c437bae96df030 : SUCCESS in 1h 57m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cefd4ffe342843c8a9e7e29fea08faf0 : SUCCESS in 58m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c8ee83271c28495385acf2077c88475b : SUCCESS in 1h 17m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e468cb6bbf24b2a8e770de64a57705e : SUCCESS in 47m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/59d0a838c2a14485a06805ecd775c630 : SUCCESS in 1h 09m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5c4f3235b30406db19d178d3d6ed467 : SUCCESS in 51m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cb6c8137d9fb4f878cf68a8c9320b11a : SUCCESS in 3h 42m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d2c4341f591475280316d459d10460e : SUCCESS in 1h 21m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d49ba16610df4dc48afe8de5894cc40e : SUCCESS in 1h 41m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/07c84984ce044d1cac3fb180c5d6a957 : SUCCESS in 2h 05m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/af070335067a4abd9f2e909081bd2a39 : FAILURE in 2h 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec09f888856ce68a426c00fd436be188d14de26d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-14 01:40:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb4bed0b9868dc63aca503c998135d71ad4af5ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 01:40:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51d23cba72e92bd31dad279556807673b6c7fabf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:21:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/92295eef7dfe43278497f34661c17a27 : SUCCESS in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b196a598ac74694872f4882ab7601c0 : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8176476698c64f12b1e3aa30256cd3b7 : SUCCESS in 26m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8753a99c68c44089865670893f0a0ccf : SUCCESS in 27m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfe2105b74b46b3aa7d772b3767c53e : SUCCESS in 9m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/611cfe55125749759d1f4ac8fc14a70e : SUCCESS in 46m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/026d53f4312d497580dd6ee352dc7095 : SUCCESS in 1h 22m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ca16717726d4159a6ddcf250b1b5a2b : SUCCESS in 55m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e1febbd1d83148afa9734f8603b3aff7 : SUCCESS in 56m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c042a4c09a1f4adb8962eb941b2fa1c7 : SUCCESS in 1h 01m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/96b88c2112c44a00af3733ba4a8d75e1 : SUCCESS in 1h 00m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0479b4945bc04df19b607fc325483fe8 : SUCCESS in 1h 09m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bacda666aee049f4ad711f041824c00d : SUCCESS in 1h 41m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71358fd077d8445c82b6348a45f8b094 : SUCCESS in 47m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/87fce537a558454fa2c27b1cf9c1391c : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"45409837c6a53626450e5c690cb1924170cb791c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:21:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a9960876d1ab9bcff49818ae8498860afb2864b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:23:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0742f1be1a43568f7d2dd5240edd9e : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92b3b628404dfcfb77fb3b7bd64e6d459c8077d8","revisions":{"92b3b628404dfcfb77fb3b7bd64e6d459c8077d8":{"kind":"REWORK","_number":1,"created":"2021-05-12 08:18:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/16/790916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/790916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/790916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/790916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/790916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/790916/1"}}},"commit":{"parents":[{"commit":"6619e9881aa5eaccadee9e2f52bd22e8351af7ae","subject":"Merge \"Improve Subnet update performance\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6619e9881aa5eaccadee9e2f52bd22e8351af7ae"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-05 19:07:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-12 08:17:53.000000000","tz":0},"subject":"add_fake_chassis() may need to create a Chassis_Private","message":"add_fake_chassis() may need to create a Chassis_Private\n\nIf the Chassis_Private table exists in the schema, add_fake_chassis\nwill need to create it since we aren\u0027t running ovn-controller.\n\nCloses-Bug: #1914754\nChange-Id: Ie36cb025540c158fbbed4126be5d654888715300\n(cherry picked from commit d452e7868756935b074cc2026262e588e9523413)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92b3b628404dfcfb77fb3b7bd64e6d459c8077d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92b3b628404dfcfb77fb3b7bd64e6d459c8077d8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
