)]}'
{"id":"openstack%2Fneutron~790935","triplet_id":"openstack%2Fneutron~master~I68b247f1abc93df91fd2ece6b7a14c633e27b279","project":"openstack/neutron","branch":"master","topic":"setup-cfg","hashtags":[],"change_id":"I68b247f1abc93df91fd2ece6b7a14c633e27b279","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-12 10:08:32.000000000","updated":"2021-05-12 14:18:57.000000000","submitted":"2021-05-12 14:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790935-setup-cfg","meta_rev_id":"f455d414096db4683ea2a7e2edc5e059ee3eba37","_number":790935,"virtual_id_number":790935,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 14:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-12 11:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-05-12 13:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-12 13:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 11:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 11:21:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-12 12:18:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-12 13:35:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"032cf4f03d0917319fc5d96e4bb49d904990371a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 10:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3172c4b6c5c1d0a153663bbba8482ab3c976bf68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 10:27:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2e75d0111fa2b62244f96030286b61b71bc79040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6bfdcf7d00a4004880faed0a6ad4b74 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3b85b223dfd4ed4b6efb75f029fedcb : SUCCESS in 41m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/236286500a544a46afc06e50f7193d13 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73315a6cf4bc4bfbbe5be592a7f17206 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30aecd5453e04b2a9ef5017b4cc55807 : SUCCESS in 35m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/337c327c3bd04fe79afcd60e89ab8210 : SUCCESS in 26m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b27794939624f1f86f703f2cdbe6816 : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef81a0818fa44c58ddf50e6cf968717 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9b23adb2573cb3d03a4c7d44a445ed349e38220f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:15:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdb67ed5727d4cb0a6ad6d057b50010b : SUCCESS in 42m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13c1928f06134db4969e629d5d51681f : SUCCESS in 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4d6c3a86d3a3860247095ec4f6f2abe62db7378","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 11:21:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"edda58f6c2b8c00ccedd55497484bf6e186a865d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-12 12:18:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/790935/2 : SUCCESS in 1h 49m 43s","accounts_in_message":[],"_revision_number":2},{"id":"95243bfcf4666b8afb3316f26b60db3adf9c01b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-12 13:35:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac30b5c63bec2b1a11d449661e27f7ac247fb673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 13:35:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b87261fb2f73adbd8e47e9c22d82fe791e966dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:16:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad5857d48bb747579275da629a70420f : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f955fd6e4624eb68a690c27e76cd1a4 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc0d3685e8249a79f2fe2e65cf2e899 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/290fad5ac796435ebe2e825ebb2dbe88 : SUCCESS in 40m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03bc0b1f34224701b11610fbc7c62275 : SUCCESS in 32m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470ddc74ad364ef5b3a3332bf6398851 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":2},{"id":"575076a5e7537e7db77506dbdba3e50fbd36e4c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:16:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f455d414096db4683ea2a7e2edc5e059ee3eba37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:18:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242b88335c1345c28384a197baaef544 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c192b78a2017da761e351321b996231ed641893","revisions":{"9037760d8f58d4a3ebc526b6d984032b69fb14f4":{"kind":"REWORK","_number":1,"created":"2021-05-12 10:08:32.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/35/790935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/790935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/790935/1"}}},"commit":{"parents":[{"commit":"63e4932496dbdd2bece835954b05d3e0ab947891","subject":"Merge \"Add physnet name case sensitivity limitation to the qos doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63e4932496dbdd2bece835954b05d3e0ab947891"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 09:56:27.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 09:57:16.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of\ndash-separated options in \u0027setup.cfg\u0027 will not be supported\nin a future version [1].\nGet ahead of the issue by replacing the dashes with underscores.\nWithout this, we see \u0027UserWarning\u0027 messages\nlike the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I68b247f1abc93df91fd2ece6b7a14c633e27b279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9037760d8f58d4a3ebc526b6d984032b69fb14f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9037760d8f58d4a3ebc526b6d984032b69fb14f4"}]},"branch":"refs/heads/master"},"2c192b78a2017da761e351321b996231ed641893":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-12 10:27:44.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/35/790935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/790935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/790935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/790935/2"}}},"commit":{"parents":[{"commit":"ba2dc74c9acf788c78d2d26a64256bc4e39d7af7","subject":"Merge \"Adding placement auth options to oslo.config entry_points\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba2dc74c9acf788c78d2d26a64256bc4e39d7af7"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 09:56:27.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 10:23:20.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of\ndash-separated options in \u0027setup.cfg\u0027 will not be supported\nin a future version [1].\nGet ahead of the issue by replacing the dashes with underscores.\nWithout this, we see \u0027UserWarning\u0027 messages\nlike the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I68b247f1abc93df91fd2ece6b7a14c633e27b279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
