)]}'
{"id":"openstack%2Fneutron~791113","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ib468c04779af7aaf4dedf84ba885bce71078248b","project":"openstack/neutron","branch":"stable/victoria","topic":"ovn-migration-stable/victoria","hashtags":[],"change_id":"Ib468c04779af7aaf4dedf84ba885bce71078248b","subject":"ovn-migration: Use DHCP nodes to configure dnsmasq","status":"MERGED","created":"2021-05-14 09:43:59.000000000","updated":"2021-05-14 20:22:19.000000000","submitted":"2021-05-14 20:20:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791113-ovn-migration-stable/victoria","cherry_pick_of_change":781627,"cherry_pick_of_patch_set":3,"meta_rev_id":"9469693d49e74d02d7120b9b6a301cc0355a0e0c","_number":791113,"virtual_id_number":791113,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 20:20:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-14 14:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-14 19:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-14 19:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 11:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 12:11:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-14 14:29:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-14 19:01:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"dc524b03416ad3eaf48ab22786757d48da20730b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-14 09:43:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fde3755c72c5859fce5d4dbf05a6335a8f63a7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:11:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9ed38d7f1548668e730df2691e99e2 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6b93e2e39a410ba7c41e094cccec3e : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c87c33525eac42138349c50c5e5706a3 : SUCCESS in 26m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42699b90aafc4627b4d0c655a0eef896 : SUCCESS in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62a74dfbfdf43a18bf25fffe7e21d49 : SUCCESS in 13m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c5f8a59adda430c9b85c514739bf1e0 : SUCCESS in 1h 12m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3350697f6628465ebeb3a1723cde25c4 : SUCCESS in 1h 19m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2cc933128fca4290802d49468e1a0f80 : SUCCESS in 1h 17m 19s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3c00475a27754a8994f3010b39735d01 : SUCCESS in 1h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"119c4ea7be511a176f0552e1bc23fb4df161f195","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 12:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791113/1 : SUCCESS in 2h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"884fd2ff5debbce468ac4ba96395b0fec16e4b5c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:29:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57785921cafaba29acf275e007064ea69cf4a797","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-14 19:01:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab763f6080b44bdf0a76b0751a490ff43c26b837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 19:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7bbe08b95865f464ce80a8b8868483735f02b4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 20:20:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e0d283d951447aa4ab6c5b5feea709 : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9499561d59af43b399dc4f9fb7f4ba5b : SUCCESS in 22m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/017a27617676448b83276643687208ff : SUCCESS in 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400d7a23b0b64529b8119937bcaabea4 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41c145a4f2a74ad68abcc8d1814fc11b : SUCCESS in 58m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce462b218c14c8e97378063b25fccab : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"821a46befbe56a8416f359d68c6774cd5dde6c85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 20:20:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9469693d49e74d02d7120b9b6a301cc0355a0e0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 20:22:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c300d1c287f49ff8e23f48dd00ae8d0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"208fea3bf6835ce460dc515d24cd3871c2420276","revisions":{"208fea3bf6835ce460dc515d24cd3871c2420276":{"kind":"REWORK","_number":1,"created":"2021-05-14 09:43:59.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/13/791113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/791113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/791113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/791113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/791113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/791113/1"}}},"commit":{"parents":[{"commit":"3e7f78c62a57f0556e1c6db10523b4a5a2a68457","subject":"add_fake_chassis() may need to create a Chassis_Private","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e7f78c62a57f0556e1c6db10523b4a5a2a68457"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-03-19 11:07:12.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-05-14 09:43:59.000000000","tz":0},"subject":"ovn-migration: Use DHCP nodes to configure dnsmasq","message":"ovn-migration: Use DHCP nodes to configure dnsmasq\n\nPreviously, the migration script used controller nodes to configure\ndnsmasq when configuring new MTU. Controller nodes may not run DHCP\nagents. The patch detects DHCP agents instead and uses those to\nconfigure dnsmasq.\n\nCloses-bug: #1928450\n\nChange-Id: Ib468c04779af7aaf4dedf84ba885bce71078248b\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 569aa4452a42d884f365fc7800164a9f7f984687)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/208fea3bf6835ce460dc515d24cd3871c2420276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/208fea3bf6835ce460dc515d24cd3871c2420276"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
