)]}'
{"id":"openstack%2Fneutron~791469","triplet_id":"openstack%2Fneutron~stable%2Frocky~Iffb6643359562487414460f5a7e19a7fae9f935c","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1902917-stable/stein-stable/rocky","hashtags":[],"change_id":"Iffb6643359562487414460f5a7e19a7fae9f935c","subject":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses","status":"MERGED","created":"2021-05-14 14:47:31.000000000","updated":"2021-06-04 09:30:48.000000000","submitted":"2021-06-04 09:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791469-bug/1902917-stable/stein-stable/rocky","cherry_pick_of_change":791500,"cherry_pick_of_patch_set":1,"meta_rev_id":"34e670b3ea8822c77069b1a0893bdc085c0e42af","_number":791469,"virtual_id_number":791469,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 09:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-04 07:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-05-25 09:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-04 07:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 17:01:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 12:18:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-05-25 09:56:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-06-04 07:45:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"2c23f4d51a26a908ea6da58413587b2788ff4bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:47:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"a3513aab7aacd2d04de5e3499c1b505057ba652b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 17:01:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/1c03b8a46e8b4bc4ab7fd584437a9cca : SUCCESS in 57m 55s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/5dba4fdcb24e4dc88360c574c15bb93c : FAILURE in 18m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/2b97c1cff0934826a6b45ad0bf1f6f53 : FAILURE in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070bcb89f4c444a69c98718d5147c2e2 : FAILURE in 8m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/692832429bac4ade8e54aab4c03445a0 : FAILURE in 5m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3eea7666b458439383f2e06cf2909e51 : FAILURE in 24m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bc9dd07434c64d00aadab3c6a4421bbb : SUCCESS in 7m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4422743d47b941f1a41592b1beb6b613 : FAILURE in 30m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cbce8db70be44040a1b038cb14c18221 : FAILURE in 16m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2fb51fd4f6d42ca852cdcf8ca890eb1 : TIMED_OUT in 2h 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/771808b07b1947c8bacc2e7f8cf6b221 : FAILURE in 20m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/411b3439e4ca48afb558fb82eb835a3a : FAILURE in 34m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6bf0c17309024760b0da5125aa2e7dea : SUCCESS in 1h 46m 41s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f971ca79caf64089bf0dc1aa8205214b : FAILURE in 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac21278dee5420fa344a8e40b47b0e1 : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6d2fbd86c4ba4984d52354852157191321ea3efc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-19 09:38:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8dabaa874ab8310a3de78d54a2883b51f32a9de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 12:13:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e9b86a4c2e4748abb5ba1196c6dedf74 : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/e8aa18a8a90e4775ae93e70d9c3a982a : FAILURE in 2h 03m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/fd0e3ee66c7f48168272de1b7d255cd4 : POST_FAILURE in 1h 24m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a38e635ecb4b6abfba657e9c1f44ae : FAILURE in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8cbf3beaa6b4507a1e298d6c144a490 : FAILURE in 25m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5f4e4667c6a40eea2af0da9e6676ef4 : FAILURE in 22m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9cb775b9e1f44ffc8ce0c7de4935fa86 : SUCCESS in 7m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/de297d6bcdcb400f9e476c8c187d684e : TIMED_OUT in 2h 09m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ccb4e61fc8704e558e58a660e4f9e740 : FAILURE in 9m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36599980a458465eb6f059b31488f75b : TIMED_OUT in 2h 10m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ef4e913b9ec24a61ba9955513eab10a1 : FAILURE in 39m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/15fd79cf050e46109e1694640d8ac2f7 : FAILURE in 2h 33m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1502d254a02e4bfa8f92412e1c01adb2 : SUCCESS in 1h 13m 38s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e5efc7bd2d74def9793b292462c4fc8 : SUCCESS in 1h 25m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c86e96ddc53647d2887820ffb87734ea : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a4e96b937758b6108e39428f9a23e5d95e66a51a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-20 12:18:52.000000000","message":"Patch Set 2:\n\nrecheck rocky should be fixed now","accounts_in_message":[],"_revision_number":2},{"id":"9bce91be0a5a7ffca478f4d128638ebff7dd8471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 15:11:41.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/516d4b2e08d34405be42511113fae7dd : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/9f6ff43698c24a1d937ddc3653e64c6d : FAILURE in 2h 14m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/ad7ba50522074b279eb66085273cb21f : SUCCESS in 1h 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcec08a808546429450e268ef8eb7e2 : FAILURE in 9m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48cab8965a4141fea82dfabc6c91792e : FAILURE in 27m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bedf3125a3f64517a282b525881fa09a : FAILURE in 24m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1bffc1b5b88e4b7ba3f9036afd1e0145 : SUCCESS in 8m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f703c91bafa4d139b38eab6d627521b : TIMED_OUT in 2h 09m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/94d88392845f4d4caff3043b888b4f82 : FAILURE in 15m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfd475ebe2be45b8b675d575abdc8fe4 : TIMED_OUT in 2h 11m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58f7b1fc9dbe4c78bb6312544446bca3 : FAILURE in 43m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1458d07949f14425b805fd1dfc6f84bd : FAILURE in 2h 48m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/23ef281962904be1be7e7fcba310fd01 : SUCCESS in 1h 19m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1095bad775de4c1eb62062b0e269f38f : SUCCESS in 1h 24m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9a7f648fb0a45078fc2942acfe3b930 : FAILURE in 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0ddf4535f21ec1982267c2a75266259f8be47cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-24 11:03:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"138ec02311724fc5b9ec88c9e512b4283d9d6591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:47:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/cb1779a3346b4d098761c2062c6bbc4b : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/e3bbcd08d4bd4573b1bfa16b286f60f0 : FAILURE in 1h 26m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/019cee17263046ec95e6af9b89969c69 : FAILURE in 1h 07m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7adc359df446659c45923980bbadad : SUCCESS in 17m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6706525003544dc38f59d1f81d4d87b7 : SUCCESS in 26m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8fb9da576bac4e1fbbcb7a8bd236a8ad : SUCCESS in 29m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7d3d379a0a0440d1b3a573ac1c251c4a : SUCCESS in 7m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd93b91cee1946db9446d6ac67589fb9 : SUCCESS in 1h 09m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2858935144624ff7bca5a75349464a43 : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a4fda29b32549edaae3f34becad7a5f : SUCCESS in 1h 05m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/82a46b55b1074675aa15c0bb7a4526fa : SUCCESS in 36m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/15163ad1a232491080bd1a61a735cba6 : SUCCESS in 1h 07m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a1743c44ec684c0088383afefd61ab82 : SUCCESS in 1h 10m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c3b2c2513644f88857ea56369e0fce0 : SUCCESS in 1h 08m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d847129cbf4a01904095bc7e46b853 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":3},{"id":"630304e8e4e275ba41fd403a3357ec16cf676d32","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-25 08:16:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"acae3387bcfc67c552d42ad15ec01dadb3132ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 09:52:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/4d63bd0f3bca426b81f83e8d245a50be : SUCCESS in 1h 07m 59s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/dc1c3119e30148e5951432d775968120 : SUCCESS in 1h 10m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/8e217cf1edc049eea672171d94714965 : FAILURE in 1h 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ef14f8a04242c8ad5da3c9a5d2889c : SUCCESS in 11m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bad440d3e7d4bb9a202f0729d67b1c9 : SUCCESS in 21m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0f1f4609ef474359beed4035b5acbfd8 : SUCCESS in 23m 27s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2cfd854568fd48a099756f6745979a21 : SUCCESS in 8m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5883beb843df462cac5fc636a710c73f : SUCCESS in 1h 16m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ef9c238d0b54dccbbbce0035711a801 : SUCCESS in 50m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c379311b9a04a879d09386f28c905ab : SUCCESS in 1h 02m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/52fd0d7ca04e4ed292b1ff8322386223 : SUCCESS in 39m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b0a9d83520cd428698903a58561ead4e : SUCCESS in 1h 33m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/553d8ba5175a4f5f804fdb7c1957b2ff : SUCCESS in 1h 25m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/016367536ed94dbd9db075ca6cdec1a5 : SUCCESS in 1h 27m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/496ffa40095a45bcb5cb9beb6d9442d4 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3fe4ccb0ab6f41b24f06682324b74d54f687b42a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-25 09:56:28.000000000","message":"Patch Set 3: Code-Review+2\n\nHmm test_floating_ip_update in neutron-tempest-plugin-scenario-openvswitch-rocky and test_qos in neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky look quite unstable (SSHTimeout). Unrelated to this patch though","accounts_in_message":[],"_revision_number":3},{"id":"059184599a75dfe85983ec06dca1bee86ab02531","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-31 12:45:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"55207b961b2774102fbdc9f4f8bf4c5161aeb14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:46:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/50c8ae73cabf4bd2961c32dd33f468a6 : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/09955220d8aa4e338cc87db2db13713d : SUCCESS in 1h 39m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e5277fc5b26c4d3bb4f7f6c5ef8a5742 : FAILURE in 1h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991b274c88c746b490d5f0c85a503def : SUCCESS in 12m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e15adeef3a5443e78c1448747c7fcf64 : SUCCESS in 25m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a9d395d534f24f0cb5cbb1e383db1f6b : SUCCESS in 29m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bb5bdc82b813433d91716e21e1801a81 : SUCCESS in 7m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa8fbbec29aa4c5e9a0cacfcaa7229d0 : SUCCESS in 1h 15m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b58d5a8e0fc5489f899d21b8606965ee : SUCCESS in 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b07f95619e549e48b569d1e4c94b1bb : SUCCESS in 1h 25m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14c3074fc374409fa1b12d37a00ed4cd : SUCCESS in 41m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/440d1e71a6fd40ecb4b041f94bd11e21 : SUCCESS in 1h 47m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3c6390a3177d4ffeb8b948cdb135e56f : SUCCESS in 1h 26m 41s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f343367627ec423a8b323f27c3c50ac1 : SUCCESS in 1h 19m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cae7191bc0114996a02e57ce807d45a4 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f89e9d414e40d65458cbe6773ecfcc66bf6c9c93","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-01 07:52:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce52efdf56089586819175d6a9bc73797125b591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:08:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/ffc937c92ed744dfa9aa3880349cb719 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/55955b5ec21b478790ca2ff74f999ac1 : FAILURE in 1h 36m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/0f1f46482d9b4cd0b508ee7c17b9262c : SUCCESS in 1h 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f8af3ebf5d4a719a878d4d643c99df : SUCCESS in 13m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0914d1c455848fea97eca70a3498f7c : SUCCESS in 25m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/188e3cfdb23e4fa3ad03d42a90465124 : SUCCESS in 23m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e10cc0ba2eb9455dbf6c1fed815d7632 : SUCCESS in 7m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0ffbf37df8a7465481af6ae34ba5690f : TIMED_OUT in 2h 10m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d7787504013149d7a893e9492d54a21e : SUCCESS in 44m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f24897b212c84109b59cf71399984dea : SUCCESS in 1h 14m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2a84462e15f34748ae85ca6528cd7410 : SUCCESS in 44m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d1d8387e3cbb4df0a1ef6cc99b05330a : SUCCESS in 1h 28m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/944ea4a6344e4815b3c4b55ffe2eca07 : SUCCESS in 1h 16m 05s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b46d00088494b19a0cd20b241079382 : SUCCESS in 1h 19m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f45dd79789e843ca8e3ca8be4cdf251d : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d61c7ab0c3c2a9172f787c22e65729384ed67c24","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-01 14:46:28.000000000","message":"Patch Set 3:\n\nrecheck TIMED_OUT","accounts_in_message":[],"_revision_number":3},{"id":"4fbbe6642804863c1c6921d44dc6effd1eadad31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:44:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/cb75f6d7ddf442fc9408cc7d34487314 : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/ee267192dcbf46ee9dd6aa53364a8df0 : FAILURE in 1h 08m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/6ca28eb90bd94204a63fbc22f56c53c9 : SUCCESS in 1h 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2db8554d5e46adbbdf53f504a85f22 : SUCCESS in 12m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e3783639b774f11b1c5aedfc14a1817 : SUCCESS in 28m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/059d0ac0fcc44719852e4795fc6bf44f : SUCCESS in 29m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/133e89b96e524f8ba5013bb266fe3408 : SUCCESS in 8m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f362e22eb9ee41d98aabcd5b20e1889b : SUCCESS in 1h 45m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c31561637aed49f7852720ad5071f532 : SUCCESS in 50m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a914dcc1ba084776a5a0dd175dd00ae0 : SUCCESS in 1h 21m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d5e2b5e249254fa4a44d6ec43f296fe3 : SUCCESS in 49m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8db177d6042f42c0a34aa7fc63352278 : SUCCESS in 1h 35m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a87811e58a80437382f6653572298457 : SUCCESS in 1h 18m 18s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e5e1219b9444963bed34f51fe0cb6c7 : SUCCESS in 1h 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4311dfcabbfe4ef78002337b2fefdbf0 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f53270f88ce342ee8893b795cc324555ab76e17e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 07:39:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9e28fd61b8c7ef309325c77629796c06a182ca7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:23:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/451ccf9ecbda439c9e96ead80a3e95ab : SUCCESS in 1h 05m 14s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/e284624853f64f3bbea3250f7bbd755e : FAILURE in 1h 07m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/4821ca40df41436297b314ef09f87bc9 : SUCCESS in 1h 08m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ce6e21c44c4108a8a635233ee1f4ed : SUCCESS in 11m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a01eb67816ec495ca6c5aa98eb1603dd : SUCCESS in 21m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6087d0a7c2394304bdd0c7881fc5da18 : SUCCESS in 23m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3bac50ccfd144b82b8c993ef86cf08fa : SUCCESS in 7m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/60288d739bce456d93865cb01732a78b : SUCCESS in 1h 23m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df97c4774ceb45868364613237dadc7b : SUCCESS in 1h 02m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/540e1566f3894dd19e57ce50ac08f3a3 : SUCCESS in 1h 15m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0aac35f640f942d9bc36965b56d5c275 : SUCCESS in 41m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8cb20ba65bcb40a2a1ccae2b95feb4ef : SUCCESS in 1h 38m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5d265e705eaf4e4e823b381626cfde88 : SUCCESS in 1h 19m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05b12904925f4a37aae36df48e0ad803 : SUCCESS in 1h 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/592507a451cc4402bfa3008566f0cad6 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":3},{"id":"20266c272a66387a37c3b620f6770687cf525e68","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-02 14:20:32.000000000","message":"Patch Set 3:\n\nrecheck cirros ssh timeout","accounts_in_message":[],"_revision_number":3},{"id":"5bdff25059c2f2f85682e93ccf3f791b806b0978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:29:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/cf58d3efc19a4c18b26c1e018a947f41 : SUCCESS in 1h 37m 36s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/976125a23096494399510580ed754d6d : SUCCESS in 1h 15m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/648722aaae444e468343b6337d253a54 : SUCCESS in 1h 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa6b7fb9f0249ef9d92bb2048ccddf5 : SUCCESS in 13m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6755312cdce34f3780c14e1040ff2beb : SUCCESS in 25m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47faaa61ed314567b550a19fbdece910 : SUCCESS in 28m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6655b309b3d540cbbb214d25ad262762 : SUCCESS in 7m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/49c925fdb56c4f778ae25810d4a0ef1b : SUCCESS in 1h 14m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc933ea4af164f759196b79ac6b4e29b : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6011819449174a119315841a13a0216e : SUCCESS in 1h 27m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8711e50d50004f1282e79673bd5d79d4 : SUCCESS in 46m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4ac5c3c703f24637a42a53d776e19a56 : SUCCESS in 1h 32m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/13ece9dd01cf4c48899afe8fb94e3e88 : SUCCESS in 1h 38m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2c01ec332cb48ba89fdf49e8ae282f4 : SUCCESS in 1h 46m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/976f327fca9d4153873e0864bc77d7ab : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":3},{"id":"64c943c3167c2c0fbec94ae59a06a3167aa1098d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 07:45:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2bb1b2f76c06651af4996e96d1ea64d184f56755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 07:46:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c693ef09dd85d8bcb6eafbe92334d09c690e490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:30:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975ff80f0a084c11aa0f3072aa3620bb : SUCCESS in 13m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a5989f9ca0f4499ba6a96951ec6dbbe : SUCCESS in 21m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40abc64dbf304d4cbfe2c7ca2e6a9fd8 : SUCCESS in 23m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a1b39851d25641cc8684d8cd204e1031 : SUCCESS in 9m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a9b3cffed02c40f69e02b6a47f364281 : SUCCESS in 1h 38m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aa8a3215cd1a49e295dd2b5ca40957d7 : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd13afe47bb74f30bf48f1249e15d488 : SUCCESS in 1h 16m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2fbe46fe955f4733be76bcfb70e817ce : SUCCESS in 46m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7fff693ef6eb47919afb117c870bb774 : SUCCESS in 1h 15m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a19e507bbda14990867e4d7966b00f91 : SUCCESS in 1h 27m 49s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2cf6478ef6b6432aad9ff7e0662e58aa : SUCCESS in 1h 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e73ee9f736e448bd9ebe4b2331df71d9 : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":3},{"id":"34e670b3ea8822c77069b1a0893bdc085c0e42af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f4a9f8f86720567a8ed5e7c5c4430c115ad3c11","revisions":{"7e0128829565397873941451e3d47416f5622e4b":{"kind":"REWORK","_number":1,"created":"2021-05-14 14:47:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/791469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/791469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/791469/1"}}},"commit":{"parents":[{"commit":"308de8290a8fd66c89e81968775c51825a0c4bfa","subject":"Merge \"Fix OVS conjunctive IP flows cleanup\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308de8290a8fd66c89e81968775c51825a0c4bfa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-29 20:21:15.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-14 14:47:31.000000000","tz":0},"subject":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses","message":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses\n\nNeighbor Advertisments are used to inform other machines of the MAC\naddress to use to reach an IPv6. This commits prevents VMs from\npretending they are assigned IPv6 they should not use.\n\nIt also prevents sending UDP packets with spoofed IP or MAC even using\nDHCP(v6) request ports.\n\nCo-authored-by: David Sinquin \u003cdavid.sinquin@gandi.net\u003e\n\nCloses-bug: #1902917\n\nConflicts:\n    neutron/agent/linux/openvswitch_firewall/firewall.py\n\nChange-Id: Iffb6643359562487414460f5a7e19a7fae9f935c\n(cherry picked from commit ca7822e2108c151bda992ef8a6d454ec2c6d890e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0128829565397873941451e3d47416f5622e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0128829565397873941451e3d47416f5622e4b"}]},"branch":"refs/heads/stable/rocky"},"487de481d913d77d9c60ce5fa19de33231c74477":{"kind":"REWORK","_number":2,"created":"2021-05-19 09:38:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/791469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/791469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/791469/2"}}},"commit":{"parents":[{"commit":"308de8290a8fd66c89e81968775c51825a0c4bfa","subject":"Merge \"Fix OVS conjunctive IP flows cleanup\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308de8290a8fd66c89e81968775c51825a0c4bfa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-29 20:21:15.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-19 09:38:00.000000000","tz":0},"subject":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses","message":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses\n\nNeighbor Advertisments are used to inform other machines of the MAC\naddress to use to reach an IPv6. This commits prevents VMs from\npretending they are assigned IPv6 they should not use.\n\nIt also prevents sending UDP packets with spoofed IP or MAC even using\nDHCP(v6) request ports.\n\nCo-authored-by: David Sinquin \u003cdavid.sinquin@gandi.net\u003e\n\nCloses-bug: #1902917\n\nConflicts:\n    neutron/agent/linux/openvswitch_firewall/firewall.py\n\nChange-Id: Iffb6643359562487414460f5a7e19a7fae9f935c\n(cherry picked from commit ca7822e2108c151bda992ef8a6d454ec2c6d890e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/487de481d913d77d9c60ce5fa19de33231c74477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/487de481d913d77d9c60ce5fa19de33231c74477"}]},"branch":"refs/heads/stable/rocky"},"9f4a9f8f86720567a8ed5e7c5c4430c115ad3c11":{"kind":"REWORK","_number":3,"created":"2021-05-24 11:03:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/791469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/791469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/791469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/791469/3"}}},"commit":{"parents":[{"commit":"308de8290a8fd66c89e81968775c51825a0c4bfa","subject":"Merge \"Fix OVS conjunctive IP flows cleanup\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308de8290a8fd66c89e81968775c51825a0c4bfa"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-29 20:21:15.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-24 11:03:42.000000000","tz":0},"subject":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses","message":"[ovs fw] Restrict IPv6 NA and DHCP(v6) IP and MAC source addresses\n\nNeighbor Advertisments are used to inform other machines of the MAC\naddress to use to reach an IPv6. This commits prevents VMs from\npretending they are assigned IPv6 they should not use.\n\nIt also prevents sending UDP packets with spoofed IP or MAC even using\nDHCP(v6) request ports.\n\nCo-authored-by: David Sinquin \u003cdavid.sinquin@gandi.net\u003e\n\nCloses-bug: #1902917\n\nConflicts:\n    neutron/agent/linux/openvswitch_firewall/firewall.py\n\nChange-Id: Iffb6643359562487414460f5a7e19a7fae9f935c\n(cherry picked from commit ca7822e2108c151bda992ef8a6d454ec2c6d890e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f4a9f8f86720567a8ed5e7c5c4430c115ad3c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f4a9f8f86720567a8ed5e7c5c4430c115ad3c11"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
