)]}'
{"id":"openstack%2Fneutron~791997","triplet_id":"openstack%2Fneutron~master~Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956","project":"openstack/neutron","branch":"master","topic":"bug/1929578","hashtags":[],"change_id":"Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956","subject":"[ovn][metadata] Remove metadata readiness mechanism","status":"MERGED","created":"2021-05-18 14:57:13.000000000","updated":"2021-06-09 09:22:54.000000000","submitted":"2021-06-09 09:20:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"791997-bug/1929578","meta_rev_id":"c19989ec86046317ae3adfcb8820da2c4fd9e17c","_number":791997,"virtual_id_number":791997,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 09:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-09 09:22:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 08:38:13.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-06-09 07:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-06-09 07:16:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 15:00:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-18 15:18:08.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2021-05-18 15:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-18 16:03:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-05-18 17:54:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-18 18:15:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 08:54:19.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2021-05-24 10:05:55.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-05-27 11:49:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-28 22:43:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 08:38:13.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-06-09 07:16:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"98b414d89ea994d39f1b3f5c1aac04769de11df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-18 14:57:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"513a373de053ffe2ee64c9ab00c04e047849544c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 15:00:28.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/97/791997/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7067d6123b93c3d510b99f812d88011f4b3d5130","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-18 15:18:08.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAll in all, I\u0027m all for this change. I don\u0027t think the eventual 15 seconds timeout buys us anything given that cloud-init retries 20 times before bailing out.","accounts_in_message":[],"_revision_number":1},{"id":"3a15fcc579dd06a84ddcefc7336d503df4927fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 15:53:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78da32d3acbb4050aa9ec718c3181304 : FAILURE in 38m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18a27d870a7449be9756448aef8fafc5 : FAILURE in 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff07f9379f9f88642719437838458c66b1d63265","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 16:03:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c547354f48fed668cfa0ffe3a28cfae11ff9b993","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-18 16:13:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7133d7dc62caa9646df937ce33bb728da935c42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-18 16:16:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdd46f76b6b0a43ba4f9037520a2b0f1533a6478","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 16:19:28.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/97/791997/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"13f36898ec7304efb3700a2a86225293a52f00d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:56:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/462b69dd075a440fa112dfe5b6574310 : FAILURE in 34m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b674ce04f594be49a08e6af060ea033 : FAILURE in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50e38f7ff79e75ce37f6ff951988dc0efa9d2689","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 17:54:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/2 : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":2},{"id":"099bec5cda690876c9d3b18edc67ef2781410898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 18:15:21.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd1974c39e6749ec9caaaaef27e0d655 : SUCCESS in 39m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/340d0fd1a32947159a8c7211d79f32b7 : SUCCESS in 1h 56m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b856994d5e046179cfe7091ee7fbe93 : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d407cc66be84ae494784641aa1dff39 : SUCCESS in 1h 18m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7fbd25d1c05e4d2782fd8a4a76d7bc47 : SUCCESS in 1h 25m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/376c8e70548a4c8bb055c250856ec8f2 : FAILURE in 39m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f017b806d5f548099aadf792f5e27e87 : FAILURE in 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d1d5424d2c40cba2b871c9d844c5ef : FAILURE in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370968abb34a4a1795c61cd565324ebb : FAILURE in 32m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4a16b5ed10e45eab1642e2ce33e0728 : FAILURE in 28m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99ec8672ab724de9b435c3f98b1eb710 : FAILURE in 34m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fcaf84d14441ca8d024a9062fcd747 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d733799043f4f039b1d5467f1c14b26 : SUCCESS in 52m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/40b37638bccc4bb1b675776542192fa7 : SUCCESS in 1h 30m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/10d6b53c9aa546f7b623ba07d038d461 : SUCCESS in 1h 13m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f9b7244527948008951edb904a050c5 : SUCCESS in 1h 02m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4d815b93dccf43649a87aa7815fec48f : SUCCESS in 57m 07s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b807a4769e9e4519a8a97ab2ac003608 : SUCCESS in 1h 11m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41073656ea4b435f872b52eafc4c5794 : SUCCESS in 1h 34m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c03595173b84a0ba5015bc91a458f83 : SUCCESS in 1h 38m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcd02b8fc21a4041af9af07676601775 : SUCCESS in 1h 15m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a8dda21e0c8486ebccd0bbfde0a85e7 : SUCCESS in 58m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/045b069f7e7c4eadbbd142d0ea10df75 : SUCCESS in 33m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7497e8d953b4397820c3afd53749070 : SUCCESS in 1h 31m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b8a400ad940b4550bd53002da866a918 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a5e7e9bab330e78a3513381507931b6a0c2772b0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-19 08:54:19.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nQuick question","accounts_in_message":[],"_revision_number":2},{"id":"bad94acd51be9675b217747950ef8db42402a283","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-19 09:46:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"359f66897ade8856f4442c7f959d7558ce058e73","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-19 09:55:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc2aa002a063c97a40c947673632106bb22f5c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-19 09:57:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"469377e25c55785931e8819a23417ee3e6df0b21","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-19 09:58:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e98a3c4d27ff9321d0412fa5b442748c7ac8b3f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 10:51:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76e76903c19242769786cd31293bb903 : FAILURE in 51m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ae2bdc5eaf843d7961fd2b57d6e2498 : FAILURE in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a091f0ff582e758a5d2c27a466c85cae87610a1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 11:40:02.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/791997/3 : FAILURE in 1h 41m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4eeba90f7cfc896ad318a39d64859742deb63db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 11:45:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/09ba896a552349a791f34be1459004c1 : SUCCESS in 1h 08m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e1d085e38b7c46f7afb58c4777c9bee2 : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/87d961ea0d7949489e91f90c35ec86f3 : SUCCESS in 1h 31m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4525c26e29e74c68818fc64b4410d40e : SUCCESS in 56m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ecbe15632d9340a089c59865223d688d : SUCCESS in 1h 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44eb4a0391a549378e72e84aa0d73127 : FAILURE in 26m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58206703adce44e292ff1fe914f72f32 : FAILURE in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6764787c194d91b0731a6af43eb4a8 : SUCCESS in 11m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f33560fdd734d088ffcc1a7a508e2d0 : FAILURE in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88f25470913446bb86c366aa4ba779fd : FAILURE in 28m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4346135f40324d37b8aecb394fcd8d46 : FAILURE in 28m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3177b35949de434594e70713221cd959 : FAILURE in 6m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d315a96cb744d6bbb86ef8a6e742419 : SUCCESS in 44m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5d8446cf72a463188a2292f1d575f88 : SUCCESS in 1h 18m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2636628c95ef4c13bc347d97164c1566 : SUCCESS in 1h 15m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/20ac4b1b9b524ee698c997cff0e418a8 : SUCCESS in 1h 06m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4041f98301cc45fa96a58a452ac4c875 : SUCCESS in 1h 24m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5059850a81c34501b267b1105a96d995 : SUCCESS in 1h 11m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ba169fff99fe4da1bf0f5107c02dbc4b : SUCCESS in 1h 27m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ce5cc1d7c0d4b8dbff32c63cad5fd73 : SUCCESS in 1h 33m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35f8cf33c4dd4380911d459c8e8316f0 : SUCCESS in 56m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba4fd377294e4918b1590c4413e2326f : SUCCESS in 1h 04m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d94169aeea184fd997b3b2691fc492b4 : SUCCESS in 23m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b52339057105479ba1e447b9ee6ebfe0 : SUCCESS in 1h 15m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/432102232fdd401aa4600c04b40186f6 : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":3},{"id":"210617bc53f15516513b76f28b39fca7fb3c2a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-24 09:47:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5da2e759978e24045ca3a958f76bcc5c1dab1a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-24 10:05:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0d475e6bf2c9afefa217e7aa604a52e4c0b8c762","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:47:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1fc6aa159104ac1aee044e0243ca07a : SUCCESS in 35m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f265a054f8c24ba0a2ca0f01c15b810b : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d6aa050dbaef633cad911a02355e39049f49cf72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:14:13.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ff0f8efdb25240eabf047960ba6e2244 : SUCCESS in 41m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7ef2e6007b041df90543457bee283a9 : FAILURE in 1h 31m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c907dae4b0c94aebb808164c3173dd59 : SUCCESS in 2h 06m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a41d9936fff4e6185eee65671fc10f0 : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be70e664849b4b8cb95a260889c7cc68 : SUCCESS in 1h 30m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5848575d884a409aa187f443f6e1c98d : SUCCESS in 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a0982034d34e91a8aaf7dbd1f4cf3d : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41471847b6ab484caad0ad08221e6bf4 : FAILURE in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2164be8fc74329a0effb19e3e129fd : SUCCESS in 25m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/969ddc07e55f43d6a8927e062ecfba42 : SUCCESS in 22m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12157cbc9d61483cb9c1373421de08eb : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ec88ed33d84fec9d5128deb4487442 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06f24d7676114f0289d568040f36fa32 : SUCCESS in 45m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98d14dbdd6dc4550b8b4919d9f971aee : SUCCESS in 1h 16m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c2ce10137dbc4d3ca870d01fa1099f2d : SUCCESS in 1h 12m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65591e5a54b1496cbd74eadf1f82e4e7 : SUCCESS in 57m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b155c87a8a01436996a3f965424c73f7 : SUCCESS in 55m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3a9316a5bfc14704864e1a269d16250d : SUCCESS in 1h 15m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5535ce7988d486b8862f49cc3469816 : SUCCESS in 1h 33m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56756ae872fc46179516bd797590af85 : SUCCESS in 1h 36m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aedb9908eb949808995d3be262d3f74 : SUCCESS in 58m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4dbbb1968d3f4a418dd4f3d3a51b7c35 : SUCCESS in 57m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0547835811cb48eda7ac40c77815e46e : SUCCESS in 23m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b415e8033c3c478a82e6e3410b7ecbf4 : SUCCESS in 58m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c8a234f26bb49f790ab8b560e6fed90 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":5},{"id":"6b0d24080260e91e48823ac21a1b3f37cbf44bb0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 12:14:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/5 : SUCCESS in 1h 36m 59s","accounts_in_message":[],"_revision_number":5},{"id":"49affb927b1ed60fa48e42ae5f078b3ac7ae2699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-24 13:15:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"61c2776342d785c0556cd79e55e6d765ae7fdd59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:00:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64f63444c6454f9d98194edb67328198 : FAILURE in 42m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12a1faeb47304d5c90a71d43f72887ce : FAILURE in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a46a4d35aac7890d8f24043c9b18f2da10daf61c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-24 14:19:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cb235a8379ff53ac8e5781d9be8a07c595b81f88","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-24 14:51:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f5262dfb2a8a3d851b288c98419ec7c53ffca56e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 14:53:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/6 : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":6},{"id":"4395b2c9c1cdfb28d478dc7aa74e1b7276bf6341","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-24 15:15:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"629b97652c198687f88e93e8ad40aefe3c0ef5c4","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-24 15:22:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e04ad077d59c720e208b366d2ecb52ee739e1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:40:51.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8b5e9e9dcf124189a8cdcde6aadbdaf0 : SUCCESS in 49m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a74b35a277fa4ed58fff78991aceaab1 : SUCCESS in 1h 48m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/edf912a699a546d4bf7c28130c5f2499 : SUCCESS in 1h 33m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41c2376e970f4dcd83730adb07951ae0 : SUCCESS in 1h 24m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1cbef09c24bb42c4a7c1335227a73773 : SUCCESS in 2h 23m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/224fb903b4484366930fc399b917576d : FAILURE in 43m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49c7d5452c0c4cbb9b1640105509f965 : FAILURE in 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac71e54eccc49dbb485c9dfe9d737c2 : FAILURE in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6683c5cb02784c44b973dbabfb0b7668 : FAILURE in 28m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1312289d541e4fdeb0a3b723119657d7 : FAILURE in 25m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30d13cfa164c4ff1824e6c90d96c0b90 : FAILURE in 35m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf72c64a941d43378427cf4e6ddedc4a : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/857baa89a0f64dd091781f93ca973046 : SUCCESS in 44m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc4fe2ec8d474b8084b7e8992d9326ae : SUCCESS in 1h 16m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4ef045e6ad034615ad705c5a4d5445a4 : SUCCESS in 1h 17m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8717e49a62e0464c9ae9dc0d6f3802b4 : SUCCESS in 54m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4d7a975ca2f4ffa800174f2d4e8baa0 : SUCCESS in 59m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/850e4d2d8faf4054b56f680abab2ce22 : SUCCESS in 59m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c71ebc3939c3411da29827069c19c78e : SUCCESS in 1h 06m 43s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5e53c7ce5eb466194b01ea48c646076 : FAILURE in 1h 19m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/676b79b25c51481aa805f4e77657f913 : SUCCESS in 46m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dd274e37096341f6bb119556bcc6fb8b : SUCCESS in 50m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4597c460fbbd4f35bf2014bee6cff1cb : SUCCESS in 21m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7685daee570949c89b644fb5738b2618 : SUCCESS in 1h 23m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10bd7ae1ff074392a9fe76f648aa3638 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":6},{"id":"e0f9297f44ae6ab78cd3eab3ddde2baf31a0c924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-05-25 07:29:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b8565035b51a8898537d154c3c021fbcd83919f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 08:17:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb5b3435028740018657f220b756f773 : SUCCESS in 45m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aff8570f36ab47308a75bca6cfcd2923 : SUCCESS in 45m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aea9450e5e6d99d52b952a222b499211f6945181","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 09:05:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/7 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":7},{"id":"442c144a4f810c8e4aa2a484bd0a4ed19c141188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 09:45:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af180b2cb3e2434fa21397ef3af86cd4 : FAILURE in 44m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0dfd03804f94497aa17c240520740f2d : SUCCESS in 1h 51m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2836dd392f424a9dae99219d4dfbb323 : SUCCESS in 1h 49m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a29f9e721c14eba836d0d80774471d0 : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f7ffa904bf0e44c1b097ae012baf100c : SUCCESS in 2h 09m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b25e6fe621c4e19ae6e68779bf06711 : SUCCESS in 45m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03fa7400a81f4a62927882edaad2c2a7 : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da5ea0c079c48859057dd59c21a7386 : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c78be770027c413e8abcf16fab0ff619 : SUCCESS in 23m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e59f9ad95a4b87bb63b205ddfff8b9 : SUCCESS in 22m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff592a9dcff446578577acc3cae369fd : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147797d0265042d2aba0cf338886b4f4 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f091fbe3ecf649c6a762ab05abd20709 : FAILURE in 45m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68ee13c417694176ab050c85262a6d81 : SUCCESS in 1h 13m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3a967dc230c14a6faf800991a216ecd3 : FAILURE in 1h 22m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b28cb18427743e09ef30884aba4a343 : SUCCESS in 52m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66e214a0e0e5434f8002e28c62af7880 : SUCCESS in 59m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a3e0e48b281543f1ba8c3031c5255398 : SUCCESS in 1h 15m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d51c3f268fdf451488014d64dc859c99 : FAILURE in 27m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/223a60c3864d4842ae4a279e8687ada9 : SUCCESS in 1h 44m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d27da6ea21f4c08bb090e79a9c7e858 : SUCCESS in 1h 03m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/078df04876e44ff582be9e7418b81c98 : SUCCESS in 52m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cdf01ae152ec4fd1b6d3f37caa536dae : SUCCESS in 34m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8babfe04513244149fc130e0119ebe92 : SUCCESS in 1h 14m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/441a118c45c94302a1c46e4ffe218f58 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":7},{"id":"c94208a6e237ac35cf0448beb2218b74fd5560ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-25 15:57:57.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"df1262508e86b2eb8226e5418681e75df84d39ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:41:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ba5d1dd2b0d4f76b4454309a8d7b14d : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31d5cf2b0fcd4cf1bda0742e022127f7 : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"143437d2b774fa319384ac02ec398fff7a57779e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 17:57:02.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/423a3b01f5224de9af002e5aa3adf823 : FAILURE in 48m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6fd72942b6047d9b439f68fd04a4ccf : SUCCESS in 1h 58m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eb02d4f0edf046308658e05a924e5a1a : SUCCESS in 1h 48m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86537f2f815d4dd2ae37d61e1c691fec : SUCCESS in 1h 45m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5bb83dfc8f97423aa378ac31db413c49 : SUCCESS in 1h 55m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83401a11e744163890f4ff181eb09ec : SUCCESS in 40m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59643c8e61c4d22ab8a0ca3e82ec33c : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4e5514768f483890ab5c77bae26cef : SUCCESS in 15m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c80a08aa3fb14f73ae7dfbecf03eee87 : SUCCESS in 36m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d100faa783094586b54bb6243ff1e41f : SUCCESS in 29m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2423240b66f1473c8a04ec242f3eda06 : SUCCESS in 30m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb9d6a455fb4433b6f507f2cf410885 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b30769a130fb4742b6ae30168605fb04 : FAILURE in 48m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff7221890fa942eebddddb56edf71514 : SUCCESS in 1h 22m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/564d0092766945c6976a3a423a2ac7b1 : FAILURE in 1h 35m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57c12299c03d4af2ba61ffe0004a1282 : SUCCESS in 1h 05m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c637361c16144e29894edbefab6502af : SUCCESS in 1h 13m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/85ac2ad6b53446f5a2394a1e5566924f : SUCCESS in 1h 15m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7122823bfad54f0bbae96d2d26a7d6a1 : FAILURE in 30m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/842729a807c94104ab65ebbdb2fe26a4 : SUCCESS in 1h 35m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/069c4398999c4107b1c21339cd985950 : SUCCESS in 1h 16m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/452cfdd673f94587ac8d48ba005b647e : SUCCESS in 1h 01m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4433aa30b7cd4f22ab0542204a5d3f6b : SUCCESS in 36m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7e40c0fa6394819b5ddb4de68130c57 : SUCCESS in 1h 31m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d6852428d2c141a091812a693159be54 : POST_FAILURE in 1h 19m 23s","accounts_in_message":[],"_revision_number":8},{"id":"737848f3e5030b487814d2f867a86f0de3c35a30","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 17:57:41.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/8 : SUCCESS in 1h 58m 51s","accounts_in_message":[],"_revision_number":8},{"id":"fbc5a368bbca6a9b1ecc33628247221b97409824","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-27 11:49:25.000000000","message":"Patch Set 8: Code-Review+1\n\nI\u0027m ok with this patch writing a release note, maybe in \"other\" section, saying that we are leaving a harmless leftover in the \"chassis\" registers.","accounts_in_message":[],"_revision_number":8},{"id":"befcbe11f99145cec6330c9114d61deff5b6260e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-27 12:50:02.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"4671577e43748fd0f7d47df040014c2f786706a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:42:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a88c667982e84aba9f93ee554cc06239 : SUCCESS in 33m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f615762fb7d149129a33609e0c585d09 : FAILURE in 48m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7657df29207b0b562e2a63456f003f450a485ab7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-27 13:57:03.000000000","message":"Patch Set 8:\n\nNeutron CI is still broken, we need https://review.opendev.org/c/openstack/neutron/+/791365/ merged. Still waiting for the CI.","accounts_in_message":[],"_revision_number":8},{"id":"8bda33dfbc6daa40737dcb478b2f21766635189c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-27 13:58:20.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"bfa8ec43bbfa2e704ab471b62fce1b707400a7fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:47:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e469e8b30d9d4d869f016254a327848d : SUCCESS in 46m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d0abd0d14314fad9e687d0183b6f263 : FAILURE in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8c2014e09b4b2343c854e703e3dfe1935e28e271","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-27 15:36:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/9 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":9},{"id":"cf314e616f3f574b1e8fdca789a97f88fc00105b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:15:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cf0f1086c95f429a932ed21dc4e16dac : SUCCESS in 47m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e284c2390816467982faae2594c9f42e : SUCCESS in 1h 53m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b331f4eec98d4b53a1ce82007f362764 : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ce0efd249b342c9bd5e4197f885cec9 : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/86151d1f39444a17be101f17fd4a51ef : SUCCESS in 2h 07m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/167b669af37441a7ba0f404f5cae53c1 : SUCCESS in 39m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcc0e98902684e97b6a2ff19bfc0bb14 : SUCCESS in 38m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ec7c61040f499f8001e160ee21555b : SUCCESS in 16m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbccc65548194c34a4d72b1c51cbfa7c : SUCCESS in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/600fc478bcee4f8ba17f3e57d08a8c3e : SUCCESS in 29m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba0af259be3d4e6f87fab622e75a664d : FAILURE in 29m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316ac101f6f245fb863026b5d28f37d5 : SUCCESS in 13m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aee0c729727e42e983837d0a79bc7432 : SUCCESS in 41m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb92dab219cc4fdbbfc1e5dd8939b5ec : SUCCESS in 1h 26m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7f96e6ab8d1b45b1bf57c20a96942867 : FAILURE in 1h 35m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98c5dc47ece0435ab87baf4a16a7cb44 : SUCCESS in 47m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25b69d141ae14941a3fd4aacb12ce55a : SUCCESS in 55m 10s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7d90023e6711403b8828936cb5a420ce : SUCCESS in 1h 20m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db5368dc267c4e71ae3005a0c92c51e5 : FAILURE in 28m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0ca3bf3fd0349a5970986804b52c9a0 : SUCCESS in 1h 31m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/578fe74132014c42b07a20ecb38dae88 : SUCCESS in 57m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/384e8b2fd1cb4d97935028a0dae70fae : SUCCESS in 1h 03m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/50e364567e1b48108770f705cc75ee27 : SUCCESS in 32m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76736dbcb5a0469a934ca130ad3d32cd : SUCCESS in 1h 23m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3325fb6ee161400ab5bec9675eabd863 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":9},{"id":"20e3b17350d69eb9e93acd2f2fae56cc36ec6d76","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-28 22:43:32.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/97/791997/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d7d3bbdbb749e307d64cff52e65345b358938a65","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-06-01 10:29:44.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8d3291f17d6a5cb1e2ad9113e6c97e1d924cf4c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 11:13:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51a66960df4f40898acb5ee989ae4787 : SUCCESS in 42m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02de8dd59a10474e8c1cec9410ab50bf : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f7144c8b1d2917892cfe5abf379d4baf79f3f6b5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-01 12:19:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/9 : SUCCESS in 1h 49m 17s","accounts_in_message":[],"_revision_number":9},{"id":"42810b1e3cb2d943f295abc1485b8351c699a4af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:25:01.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/10b9ab9226964e5fa3762329dc7ca4ac : SUCCESS in 1h 02m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4741b47e99a42fc8f931ec1fa602ae1 : TIMED_OUT in 2h 52m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fb22d7c7979b4ea0b64103025555d05b : FAILURE in 2h 32m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4389539a54164e4793b73f36a65b8369 : SUCCESS in 1h 40m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b8b95fd0d7764e95a79e9794b198a23f : SUCCESS in 1h 57m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aa2f6ec3cae42cb990ce6b92dd5e55d : SUCCESS in 34m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf7a3336c32f400cacefdacf7477fa31 : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ee1c7d99fa4b88824f5473b2526dcd : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8eda3c3ca264411ae0f7991c214bf83 : SUCCESS in 34m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44c95baa899a456682bc39df16a988d0 : SUCCESS in 31m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb39ff21d50347ad81ece7f418bb1b9c : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73674bf2a3a4febbc7eea16d6c189f1 : SUCCESS in 9m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94caeaf17ab841c3a74cd257451575d2 : SUCCESS in 50m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fdd08d3c37b463198f66317c2c07ca0 : SUCCESS in 1h 21m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9a28068452514a50b74e2940c03beec4 : FAILURE in 1h 33m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/174f682df0aa4978b713e2b3c75a4601 : SUCCESS in 52m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbbac9f2b22f4927971fc4296ec91bb8 : SUCCESS in 1h 01m 52s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/211f75aa328f4c5eb3d0df876ccb59de : SUCCESS in 1h 16m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/594536880cd649e2aa697dbab7b019a9 : FAILURE in 27m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c6dfe53376643bb8d6d7794ea8067d6 : SUCCESS in 1h 37m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2397e4950f0e4a748c0c5b71c17e99c7 : SUCCESS in 1h 01m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82f50ac1a0fa4e49bd3ccf6a73a97dd9 : SUCCESS in 1h 02m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/307a3405875842f18ab0b90164c86929 : SUCCESS in 33m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eca5ab31347241e4988ea6ed1fcac194 : SUCCESS in 1h 28m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ca3a27e54fee431093ad7cf381c29776 : FAILURE in 1h 53m 15s","accounts_in_message":[],"_revision_number":9},{"id":"2b10e19d3a3c51279fda4a61a272ff8332fc5ba8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-01 16:09:29.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/97/791997/4 ]\\n\\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"eaba4a5342c0c63729f4e042f624d4eddf18ff4c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-01 21:31:06.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/97/791997/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"f609b74c9857c1d6877b9f440dbd797b620fbd0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 16:09:31.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"be138cf99d123b35a9707f762a3d9bf07a9e0761","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-03 17:23:02.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/97/791997/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e356be5a4022b82f09bc91d986986fd99fb78b87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:49:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f71494d55c044b17afb5d7ff0eba7a09 : SUCCESS in 49m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c77f429fe05d4c4999a893cd11279f6c : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"567645e26e926ada0166c840e7a4f8390e515dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 20:06:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d7b8f90ac9494720ad3597738b26d645 : SUCCESS in 48m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/51e4204120e844529d9971e8ce67b563 : SUCCESS in 1h 43m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f322218c69a44b4682d2ff0afbab9b91 : SUCCESS in 1h 48m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3d5b235bcef4056837eebb06e285578 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3f03006ae2624a96b741965da406edd3 : SUCCESS in 1h 35m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e486365189c4518a12c160ad7de868c : SUCCESS in 31m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d264192acdea40aab22a88fb0e5aaba9 : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413bb35a432d476081b8cbb876de1841 : FAILURE in 16m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4615ac051bb94db2b0be3d133bcf3b63 : SUCCESS in 28m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03864ce805834d44a671a18431ed7281 : SUCCESS in 28m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83a17657cdaa435583ddb9f011fbe1ee : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd50a7f0e45446884e2b1db7d9683a6 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd8df8d950384f02ae11d2e42a2edf1d : SUCCESS in 45m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/51a8a946fbb64e7e97af01aca7ac808c : SUCCESS in 1h 21m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0dd16d5b57a44f797fde4345a9e886b : SUCCESS in 1h 15m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3e95b076b6a478ab5f9f48611a013b7 : SUCCESS in 1h 01m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfc8cc8a0ef4487194211f9f2c15aae3 : SUCCESS in 1h 04m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9579f8d1777246b984b53bbc7a236ffe : SUCCESS in 1h 23m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d8b973d716340c7ae0d9ba8f1e6e300 : FAILURE in 26m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be4a5d30352c45759cc09eb3f963c5c8 : SUCCESS in 1h 32m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55f9f5db9bbb489dad5ceb60915bdc37 : SUCCESS in 51m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a47a5176de5148bb9513472d71e7cb45 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2d15375fa5e74f43b7cd0932c4ca3e8c : SUCCESS in 27m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2dd2dee7cfd34bd4bfbd4f97beeb020c : SUCCESS in 1h 22m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bbe5d8ba79da4303867211897e63a257 : SUCCESS in 1h 08m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":10},{"id":"1f603ee90df7f534828dc0dd896d373e63b2cdd1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 22:22:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/10 : SUCCESS in 2h 02m 49s","accounts_in_message":[],"_revision_number":10},{"id":"e6b19df45fd506ee79a54d458c0c509dc8d9b0eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 07:50:47.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/97/791997/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"154594e5762a60d71c72810de2b6a862d303b835","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 16:38:52.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/97/791997/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"74e4fcb210ba0ae0b498affa1547b85bddfd2df8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:33:56.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"9629fc7e85bb410445c0bb2343ba1a39b5435e12","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-06-08 08:38:13.000000000","message":"Patch Set 10: Code-Review+2\n\nrecheck\n\nThanks Daniel! Let\u0027s give this a whirl!","accounts_in_message":[],"_revision_number":10},{"id":"a6176d4ee8ac2406e60ed8ef986ab775617ccb33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:27:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081cf99e9f3f49538f02186ee8d80183 : SUCCESS in 47m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ca22dbf14e7478ca19b687649bb065c : SUCCESS in 43m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a52139a4f1f91a362fbd7f20525a9b34b69deb1a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 10:37:54.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/791997/10 : SUCCESS in 1h 59m 01s","accounts_in_message":[],"_revision_number":10},{"id":"5d285acfe50bd99d27558b724da70b18e29ea176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 11:10:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b176171da37440a085f93164ce9e575f : SUCCESS in 43m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/03235331095241368519a2a49d3b14e1 : SUCCESS in 2h 19m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9beb1dabc05746ad84c6208abba00f41 : SUCCESS in 1h 57m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd3af2f287704a33839201ddc93a28bc : SUCCESS in 1h 36m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/40bfc3bed54c47b19bdaaff56d510a52 : SUCCESS in 1h 39m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1002464bf284526991f97eb1530d07e : SUCCESS in 37m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8e269d20ca24d08b5eb4912e039bc77 : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a77bac3cdcd4ec79df9586fcbb07fba : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0dafad57759473a81cca60cd1efcbe5 : SUCCESS in 32m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3e304e82f714880896f9f5f0e6746e9 : SUCCESS in 28m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c22dc43538b0429995a3479beb6de126 : SUCCESS in 29m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefcc30e8f1d434c986f36cb7d162c7e : SUCCESS in 11m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18fded1c5c9341e5bb520230ccda4ed0 : SUCCESS in 56m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26ab134ef7194126ab7a54149d739191 : SUCCESS in 1h 24m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/75793b3f5d344030a93571338243e999 : SUCCESS in 1h 07m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e14aff13c654831afd224aa1f76827a : SUCCESS in 56m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aca8e60afeac40fb885b423af8ddfca4 : SUCCESS in 1h 02m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4e408f2337a7428fbe307348a3fdb1da : SUCCESS in 1h 19m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee02f0b5924b443d9f92b38445c94f29 : FAILURE in 31m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/14e61191510b41d0a10805416327a20e : SUCCESS in 1h 40m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc36049694e14436a6156a9c322e9f06 : SUCCESS in 50m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe967c14504b42bc906842bf084459ea : SUCCESS in 56m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ed62320b138644499b8b2d5a5ad43a9e : SUCCESS in 32m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a6f565a68f154040ba70a48847ea33ba : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/20346dfa2ca14632a30f5a0b4ece7fd5 : SUCCESS in 1h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"87d479279d058d85ed4d27655f46de6fe7d70602","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-09 07:16:28.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"387d5751d841d1ede7667d90a49045bf3066e181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 07:17:03.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0a5acea67444b293fe0cd983501b71545982407f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:19:50.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99ffa8c79a1a4233aaab99eb07336d08 : SUCCESS in 46m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2394364c986e464392db55047277d51c : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f5525ab7317c4555a80874b3167cd71f : SUCCESS in 2h 01m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf30af45bc834d5a9dc830e97575ebbb : SUCCESS in 1h 28m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9494cc8a9ab4afe8dea5af3822146a8 : SUCCESS in 1h 19m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ae0355b66ca4b319c177151c80784f8 : SUCCESS in 26m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e75ea84fc634098acf352f0dc0d7735 : SUCCESS in 15m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44114d418f52483c91b6ba2e9b9dc9b6 : SUCCESS in 25m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c1bb00883944ae1b5e6ef538af238a7 : SUCCESS in 28m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75491927317445efb7539d6aab226171 : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca31e78ae8b341f580f7ef29c895ecc8 : SUCCESS in 54m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63b3022da8e546ba9c59aa5abba85e86 : SUCCESS in 1h 26m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d47564fd978d435fb74b5a36c0041b4c : SUCCESS in 1h 14m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21cc1c646d514aae8dbf3277adbacb75 : SUCCESS in 1h 04m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bcfde67899f48328d71991a653aa517 : SUCCESS in 54m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b48c7bc0bcb541f5b31fd984b8b64654 : SUCCESS in 1h 02m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/db3db44213d14899bca82eaa5e173bac : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":10},{"id":"e870de9f5078a26670fd363e3efe2d6b52373667","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:20:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"2bd29cc01fab4aad62ae051103db6c3648bb738a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:22:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6e37005d6a4334a24774a3ec657854 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"36ba1cc3196148850263c8a8c8058e6f7ced3958","revisions":{"dd33687e1e494dae449ab6d3a9dc17463d29279c":{"kind":"REWORK","_number":1,"created":"2021-05-18 14:57:13.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/1"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n2) We can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed\n3) There\u0027s a chance that the VM boots very fast and requests.\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd33687e1e494dae449ab6d3a9dc17463d29279c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd33687e1e494dae449ab6d3a9dc17463d29279c"}]},"branch":"refs/heads/master"},"bbe69a0281bc45694a0fe376a5a256b893462c9d":{"kind":"REWORK","_number":2,"created":"2021-05-18 16:16:49.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/2"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 16:16:07.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) We can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbe69a0281bc45694a0fe376a5a256b893462c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbe69a0281bc45694a0fe376a5a256b893462c9d"}]},"branch":"refs/heads/master"},"802fce3713e0397b5a378924bfce0f785359c447":{"kind":"REWORK","_number":3,"created":"2021-05-19 09:57:50.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/3"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-19 09:56:54.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/802fce3713e0397b5a378924bfce0f785359c447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/802fce3713e0397b5a378924bfce0f785359c447"}]},"branch":"refs/heads/master"},"a857ab5b44d84eac47152ea96a437ecf444366dc":{"kind":"REWORK","_number":4,"created":"2021-05-24 09:47:35.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/4"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-24 09:47:04.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a857ab5b44d84eac47152ea96a437ecf444366dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a857ab5b44d84eac47152ea96a437ecf444366dc"}]},"branch":"refs/heads/master"},"4ecf9689d12cdb2dcd4df442a26c682a9108f43b":{"kind":"REWORK","_number":5,"created":"2021-05-24 10:05:55.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/97/791997/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/5"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-05-24 10:05:52.000000000","tz":0},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ecf9689d12cdb2dcd4df442a26c682a9108f43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ecf9689d12cdb2dcd4df442a26c682a9108f43b"}]},"branch":"refs/heads/master"},"6d2faad27ff1e8c056afd1425159ee7f05d07a26":{"kind":"REWORK","_number":6,"created":"2021-05-24 13:15:48.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/6"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-24 13:15:27.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d2faad27ff1e8c056afd1425159ee7f05d07a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d2faad27ff1e8c056afd1425159ee7f05d07a26"}]},"branch":"refs/heads/master"},"9bd8d8eb4256d6407c6aa9998ed1d8a6d3f61a35":{"kind":"REWORK","_number":7,"created":"2021-05-25 07:29:24.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/97/791997/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/7"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-25 07:28:49.000000000","tz":120},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bd8d8eb4256d6407c6aa9998ed1d8a6d3f61a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bd8d8eb4256d6407c6aa9998ed1d8a6d3f61a35"}]},"branch":"refs/heads/master"},"9648bcb742651dda4d3d29db676c809ee1a3412f":{"kind":"NO_CHANGE","_number":8,"created":"2021-05-25 15:57:57.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/97/791997/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/8"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-05-25 15:51:02.000000000","tz":0},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9648bcb742651dda4d3d29db676c809ee1a3412f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9648bcb742651dda4d3d29db676c809ee1a3412f"}]},"branch":"refs/heads/master"},"a4db798ed4182b0b342f7ecdbf32a65586260b72":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-05-27 13:58:20.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/97/791997/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/9"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-05-27 13:43:39.000000000","tz":0},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4db798ed4182b0b342f7ecdbf32a65586260b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4db798ed4182b0b342f7ecdbf32a65586260b72"}]},"branch":"refs/heads/master"},"36ba1cc3196148850263c8a8c8058e6f7ced3958":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-06-03 16:09:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/791997/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/791997/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/791997/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/791997/10"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-05-18 14:50:12.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-06-03 16:09:31.000000000","tz":0},"subject":"[ovn][metadata] Remove metadata readiness mechanism","message":"[ovn][metadata] Remove metadata readiness mechanism\n\nPrior to this patch, the metadata agent was writing into SB\ndatabase when a network had been provisioned with metadata\non a particular chassis.\n\nThen, neutron-server would wait for that event to happen with\na 15s timeout before sending the vif-plugged event to Nova.\n\nBy removing this mechanism:\n\n1) We\u0027ll save writes to OVN SB database which, in highly loaded\nsystems and at scale reduces significantly the load on ovsdb-server.\n\n2) Ignoring healthchecks (that still requires write to the SB DB),\nwe can make OVN metadata agent to connect to slave instances when\nusing active-backup OVN databases since writes are not needed.\n\n3) There\u0027s a chance that the VM boots very fast and requests\nmetadata before the service is ready but since the timeout was\n15 seconds, we can safely rely on the the cloud-init retries.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ia6cd7a9a3b9662a9a8ce106e01a93c357c255956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36ba1cc3196148850263c8a8c8058e6f7ced3958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36ba1cc3196148850263c8a8c8058e6f7ced3958"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
