)]}'
{"id":"openstack%2Fneutron~792272","triplet_id":"openstack%2Fneutron~master~I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb","subject":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled","status":"ABANDONED","created":"2021-05-19 21:30:32.000000000","updated":"2021-05-20 13:18:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64adf3c948ef8fce7a434feb9c02a1e05429df25","_number":792272,"virtual_id_number":792272,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 21:35:34.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 22:18:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 23:05:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-19 23:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 01:44:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-20 01:44:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2021-05-20 01:44:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-20 01:44:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-05-20 01:44:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-05-20 08:34:57.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"c34083f43b06141d180545e7e4379e86fba18517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-19 21:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767aeccd527fe79c77cdfb1ad2e1a890aa166c58","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-19 21:31:28.000000000","message":"Patch Set 1: Workflow-1\n\nWIP: Missing tests.","accounts_in_message":[],"_revision_number":1},{"id":"ef6eeebbbfa441026a1cff7707f2306952877dd6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-19 21:35:34.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/72/792272/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/72/792272/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b4afc9baca77b90a16f1fd9aca01bbeb0359caa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 22:18:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a4b183cdae34cb2ae1e8ebc17ea4e62 : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36d4b3bb10c74662805c9cd73de8654a : SUCCESS in 45m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4d73ac19ac6f6b2996d44517019d1d3f9ccbf7f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 23:05:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/792272/1 : SUCCESS in 1h 34m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c09d8fcf971e99ce83d50f9fe44a5f11455ab3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 23:28:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5947da0c1284e7d80d00666a498c492 : SUCCESS in 39m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/34b8b41410a34b73b76d3d737948584a : SUCCESS in 1h 53m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6644eb671359435f932feae50f877a68 : SUCCESS in 1h 08m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5483d06b9577408680c14441eac4e174 : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd64aee126e4120bbb21b7890fe40be : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82cd2cc2e0c2472bb15cc59716e27e7c : SUCCESS in 36m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0d238631a84378ac1fcde0e4144d2e : SUCCESS in 23m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11c4a6cf74454d93819ac3ba1e77e648 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daac87ec23c845abab059a62de96829d : FAILURE in 6m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4334eefae4544f7ab2c352fd517d52c2 : SUCCESS in 46m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4eaa12aea1d140669efdfbc90129accc : SUCCESS in 1h 15m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7226ad31890c4ffa8e4c6bd3c206607d : SUCCESS in 27m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6af8e703fa1147a18534f4f6645775ea : SUCCESS in 1h 55m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/df3676ee218b4edb8ec4028b100493a2 : POST_FAILURE in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5907bc961adc8443a9d3b1152be361175f188c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 01:42:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b2e7dd8410c893bb3efeebd4e927293ede7bb7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 01:43:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b1b64be63aa732405bb383105edbcc090810a25b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:26:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efc729604c344f7082b2f7667107374a : SUCCESS in 32m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/158cdbd2d36a4a64986a250aafc69099 : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1e999e1d340dfb1f51b4a7f360e396816f9104f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 03:16:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/792272/3 : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ab5462a9e09719ae2b2cc9da12ac64ce17f1d204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 03:48:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/81a4a9a81d474dfc8df1cd0089ba3af9 : SUCCESS in 42m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/38b3e196a42649db9a03cc7d0b82c620 : SUCCESS in 2h 04m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fee0f48bb524c5b9292d558e0d63d98 : SUCCESS in 47m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b109f67166c430ba48506550fe419d9 : SUCCESS in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e129f4df88e4cabbc1365918e48957a : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45fb3cfabfdb4ce3acd2763cebb1b32a : SUCCESS in 25m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0388a4fec704ebe90f9746bb9282428 : SUCCESS in 36m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bc18005ac0748afb4f2dd22a7ae8854 : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5cd55aac02464db853d7ba1113c091 : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9cde85c5192454198a00a3382637a57 : SUCCESS in 43m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/089f66d9a5164258ade56ce52e4a7264 : SUCCESS in 52m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c83d24a867491d8220f0cb9b55a488 : SUCCESS in 40m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54e3f43f1f274a408f3dc9872d3a9897 : SUCCESS in 1h 02m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/18c0be23a06343a5ad0f627b766e1c0f : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":3},{"id":"505705f07fe44fd61c35426245bbc02f775782f7","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-05-20 08:29:55.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":3},{"id":"f7c737176fcf003bf4b8b43f61b4158d4d35ce04","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-20 08:34:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7a827469623e468385f2d8e56b95380ac52323e3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 12:47:10.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"828f8c807b07c73d7434409629eb9d0340db13d0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 12:47:42.000000000","message":"Patch Set 3:\n\nSorry about my dup.... I should not have checked in late in pm. :P","accounts_in_message":[],"_revision_number":3},{"id":"502694ddd1b78a6af704dfd4ada69fd64fd6165c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 13:00:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8c0d875f482058f38a625d64d1052d4c9205240c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 13:02:57.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nOleg and Elvira TY for looking at this and please give it one more pass. There is less to look at now. ;)","accounts_in_message":[],"_revision_number":4},{"id":"dd8629ff8104ddcd3dcfc0c8e705b9e1055ec02a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-20 13:09:43.000000000","message":"Patch Set 4:\n\nThanks Flavio, this is duplicated with https://review.opendev.org/c/openstack/neutron/+/788485","accounts_in_message":[],"_revision_number":4},{"id":"64adf3c948ef8fce7a434feb9c02a1e05429df25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-20 13:18:45.000000000","message":"Abandoned\n\nAlready fixed in https://review.opendev.org/c/openstack/neutron/+/788485  closing this as duplicate.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f230c5002900e3583cc199412437f93a95ce5699","revisions":{"5c9dd415eedf9972f698f5a42104169369cdc90f":{"kind":"REWORK","_number":1,"created":"2021-05-19 21:30:32.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/792272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/792272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/792272/1"}}},"commit":{"parents":[{"commit":"2a4074918c248adf84b0afa404d30c18710da4b2","subject":"Remove call to \"fixup_ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a4074918c248adf84b0afa404d30c18710da4b2"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-19 21:25:01.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-19 21:28:12.000000000","tz":-240},"subject":"WIP: [OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled","message":"WIP: [OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled\n\nFix logic in OVN router update where SNAT is changing.\n\nCloses-bug: #1928988\nChange-Id: I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9dd415eedf9972f698f5a42104169369cdc90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9dd415eedf9972f698f5a42104169369cdc90f"}]},"branch":"refs/heads/master"},"e661d11cde8f65b16ad057143afc49092737955a":{"kind":"REWORK","_number":2,"created":"2021-05-20 01:42:36.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/792272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/792272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/792272/2"}}},"commit":{"parents":[{"commit":"2a4074918c248adf84b0afa404d30c18710da4b2","subject":"Remove call to \"fixup_ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a4074918c248adf84b0afa404d30c18710da4b2"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-19 21:25:01.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-20 01:31:55.000000000","tz":-240},"subject":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled","message":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled\n\nFix logic in OVN router update where SNAT is changing.\n\nCloses-bug: #1928988\nChange-Id: I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e661d11cde8f65b16ad057143afc49092737955a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e661d11cde8f65b16ad057143afc49092737955a"}]},"branch":"refs/heads/master"},"7c7d32bccea31f47b47f6f9343ba0e6815e5be65":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-20 01:43:10.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/792272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/792272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/792272/3"}}},"commit":{"parents":[{"commit":"b881c5a032cccbe0f78444af6571907ad4bc98c5","subject":"Add mitogen to the docs requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b881c5a032cccbe0f78444af6571907ad4bc98c5"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-19 21:25:01.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-05-20 01:43:10.000000000","tz":0},"subject":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled","message":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled\n\nFix logic in OVN router update where SNAT is changing.\n\nCloses-bug: #1928988\nChange-Id: I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7d32bccea31f47b47f6f9343ba0e6815e5be65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7d32bccea31f47b47f6f9343ba0e6815e5be65"}]},"branch":"refs/heads/master","description":"Rebase"},"f230c5002900e3583cc199412437f93a95ce5699":{"kind":"REWORK","_number":4,"created":"2021-05-20 13:00:50.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/792272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/792272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/792272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/792272/4"}}},"commit":{"parents":[{"commit":"a75fc891bc17fd49b9194d2aed4678d623ec8caf","subject":"Add mitogen to the docs requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a75fc891bc17fd49b9194d2aed4678d623ec8caf"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-19 21:25:01.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-20 13:00:40.000000000","tz":-240},"subject":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled","message":"[OVN] Disabling SNAT doesn\u0027t take effect if the router has already snat enabled\n\nFix logic in OVN router update where SNAT is changing.\n\nCloses-bug: #1928988\nChange-Id: I5e3037ccbdf7e86b1ea7cb868f88ce0179611afb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f230c5002900e3583cc199412437f93a95ce5699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f230c5002900e3583cc199412437f93a95ce5699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
