)]}'
{"id":"openstack%2Fneutron~792874","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Id09de6c43c0fab4009336e253c88f54219398053","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1927849-stable/wallaby","hashtags":[],"change_id":"Id09de6c43c0fab4009336e253c88f54219398053","subject":"HA-non-DVR router don\u0027t need manually add static route","status":"MERGED","created":"2021-05-25 00:56:27.000000000","updated":"2021-05-31 09:52:45.000000000","submitted":"2021-05-31 09:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792874-bug/1927849-stable/wallaby","cherry_pick_of_change":790352,"cherry_pick_of_patch_set":3,"meta_rev_id":"9e719ae285f9d1666bc658245d11bc508ee639b9","_number":792874,"virtual_id_number":792874,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-31 09:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-05-26 08:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-05-31 08:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-25 07:55:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-05-31 08:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 01:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 02:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 02:58:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-25 07:55:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-26 08:04:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-05-31 08:07:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"183ce804a7423c1513baeb89cb8bd6fb13f60063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-25 00:56:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"72fe84521390552ec29b373b5c5c0ba424909fde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 01:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffd7f34879d749d3a8bc57cb2acc44e0 : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72576f95e79c4dd4a80a4202c15f6906 : SUCCESS in 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fd5de6f1804d0a55e6336a83b1fd062a1111d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 02:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/248b56e8115c4d64bb32f543dbffcac0 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/549b37cc4e13473c92542e1e9bb471d3 : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/7e610b93229e48eab826bdc57862ba59 : SUCCESS in 1h 16m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55608b2f25f644c3aa53ab147c41d847 : SUCCESS in 44m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee6904758324d3f9f0242f6806b6591 : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f19ae94c654c398ecfd29d739d9104 : SUCCESS in 15m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d339c220df34b25a03844cc7cb36e55 : SUCCESS in 22m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/559edd57056c4ba081883944f5ede377 : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa0f0d39057e4424b3a11ffeaa9c0cb9 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19cabaea578a4afdaf228aab9a8e7f3e : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fe98005a3854e588f7a6ee4802d1887 : SUCCESS in 42m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b69a96f3a0649328ab4479e3a5b16ef : SUCCESS in 1h 13m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5dc1c50dbef46e3b8cc35ba66050474 : SUCCESS in 1h 15m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29ff273a8083439080b869118a211caf : SUCCESS in 51m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e193297828724b60a9c785b2d52393e4 : SUCCESS in 1h 02m 49s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3e6d4bacdcd341a6b64a865e5833ac8b : SUCCESS in 1h 14m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8409a6a728e2483c9f3681c6c2df47d1 : SUCCESS in 1h 22m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83321a4da1b7424c9ef8186974f55242 : SUCCESS in 1h 29m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6469f467fc084adca9e6d1976202ce95 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"76b1cf46c7d84b2bbe014933004a213c4559671b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 02:58:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/792874/1 : SUCCESS in 2h 01m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b597144a0b9e12cbb20d60fb770d1e51145a8075","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-25 07:55:39.000000000","message":"Patch Set 1: Code-Review+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"73782af2a0f3a728b3f4a8329241943bb813ad3c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-26 08:04:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b3eea792b4392419203d6cd8ad596c916348b98","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-31 08:07:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9be1682f0d48f39e8aa507569fcb4504d0b99ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 08:07:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fb6b914645fb2417307d519ca376226a4292a4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 09:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d117fb519e446929289a723a9c35b1e : SUCCESS in 35m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a003d5d34d6c4204b515761305fa4dd4 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a8548331aa47799b593e713de75431 : SUCCESS in 31m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31c9c99f13264673a7be54ee6371046e : SUCCESS in 29m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f67b9f201c42e0a85ea0a41ea9ee6d : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e47c2efffcd5483ca848bf361f1f05fc : SUCCESS in 52m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa8edfcf29c44a3e93107c8c231dcd54 : SUCCESS in 1h 20m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/48cbdc203abc463196462bb5ee79d4a8 : SUCCESS in 1h 18m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e313071889954713b385dd02186d2015 : SUCCESS in 1h 08m 23s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/861033f53bd34c4ebac468e3e4e04c0b : SUCCESS in 1h 39m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ae978b78e0453bb6d29a3a18487779 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b1ed6a64c783ee4d317fcf68925418efd59957e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 09:51:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e719ae285f9d1666bc658245d11bc508ee639b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 09:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c3de1d06a64326890aa8695799a316 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f29641a7f7d1cca32c7d9f011981e076b5906fed","revisions":{"f29641a7f7d1cca32c7d9f011981e076b5906fed":{"kind":"REWORK","_number":1,"created":"2021-05-25 00:56:27.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/74/792874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/792874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/792874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/792874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/792874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/792874/1"}}},"commit":{"parents":[{"commit":"8c6f7745041dca8eb4568c255e2033bf54ee9fe5","subject":"[OVN] Fix: Disabling snat after it was enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c6f7745041dca8eb4568c255e2033bf54ee9fe5"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 06:28:06.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-25 00:56:27.000000000","tz":0},"subject":"HA-non-DVR router don\u0027t need manually add static route","message":"HA-non-DVR router don\u0027t need manually add static route\n\nWhen a router set as HA mode, The keepalived process will take over\nthe route entry\u0027s generation. So, the codes that add static route\nis redundant.\n\nBut, for DVR-HA router, in dvr_snat node the keepalived process run\nin snat-namespace and don\u0027t take over qrouter-namespace, so the\nmanually add static route codes still need be called.\n\nCloses-Bug: #1927849\nChange-Id: Id09de6c43c0fab4009336e253c88f54219398053\n(cherry picked from commit f192153b449877e00c57c2b2a1b8ce89c9c733da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f29641a7f7d1cca32c7d9f011981e076b5906fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f29641a7f7d1cca32c7d9f011981e076b5906fed"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
