)]}'
{"id":"openstack%2Fneutron~793473","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I5093747ed9bde03d27e32e3206df91e4c27a2632","project":"openstack/neutron","branch":"stable/train","topic":"bug/1843165","hashtags":[],"change_id":"I5093747ed9bde03d27e32e3206df91e4c27a2632","subject":"Add support for direct ports with QoS in OVS","status":"ABANDONED","created":"2021-05-27 16:50:12.000000000","updated":"2021-06-09 17:01:10.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"91707e06e469a11bb4535bcde3b075a5078f78fb","_number":793473,"virtual_id_number":793473,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-03 17:38:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2021-06-09 07:40:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 18:57:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-27 19:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-29 19:13:58.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2021-06-03 13:46:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-06-09 07:40:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f054f0b91a92e9fe07fdc9116284fd3607e019a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-27 16:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fbb18701692b14289b7013447e9082ea899e427","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-27 18:57:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/793473/1 : FAILURE in 2h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4da550370fe672f5003d49e049a41b336d1181ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 19:53:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/09616ce6ed7346229dac58805b417346 : SUCCESS in 58m 07s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/801f45d30d91495db1e5897d6bb7f175 : FAILURE in 2h 34m 44s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/7471387e62c4468b9cd6571c486dc90d : SUCCESS in 1h 57m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/46ff430dec9748ccb9a4b58ef0f7be97 : SUCCESS in 1h 25m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5da0b63170e4091aaa4bfbb68afa60d : SUCCESS in 30m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9abbb4cd177e4ee588ae9a55b7b469e2 : SUCCESS in 13m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b57657c63fee4d868919df4eba221e0d : SUCCESS in 20m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d0fb83fb86468085ea2968c5426223 : SUCCESS in 20m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a959f53a7f60407ca66c049be191bf35 : SUCCESS in 26m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec60ffc765f24040b79c3a46266181fc : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/db25d405664c4ad4a09b9ffb3a3884db : SUCCESS in 1h 10m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/510dd85218ee42acbaae2cce19ebe7f2 : SUCCESS in 1h 17m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a5985593737430989d1994c3e6d4e6e : SUCCESS in 10m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7168d154ed734dd9bca53759477104a7 : SUCCESS in 37m 37s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/ae26bccd5ef247a19fe27b97c1c4c28f : SUCCESS in 41m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1e6a95e71a614bb3a72ba673c2d1d7f6 : FAILURE in 1h 24m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2535c00503384d2cae9130cd566fa560 : SUCCESS in 1h 20m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df06d16f75a949da849d351c85b5da44 : SUCCESS in 1h 15m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/74847a0cbbcb4fb8b812e979c0ca5764 : SUCCESS in 1h 03m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3238f432bf2047b98715645bdd7e5cdc : SUCCESS in 46m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f99c465bbe42423e954df7bb541c3963 : SUCCESS in 1h 07m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d93b99cab8d643a7aeb11f0d95612232 : SUCCESS in 1h 28m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c61c1cd31968470faab0ff0789cd2f87 : SUCCESS in 1h 58m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f99a00bb55544829b0c820d0e1074987 : SUCCESS in 1h 15m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a170f8837c864263b07160e5adfe334c : SUCCESS in 58m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f9d5146070ee45db8fb8b663a864b552 : SUCCESS in 53m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36f8b6f713b947ecbf583b0aeaa91c67 : SUCCESS in 1h 34m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/49ab656373d84804b8b81f5de98e73e8 : SUCCESS in 2h 14m 50s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/70775406ab36438fb05036e93a2e92a7 : FAILURE in 15m 33s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ae78b1dff3434ef28fdc15179431b8e5 : SUCCESS in 53m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/608d3b8f62b1417f8bd8825fbc7e9a05 : SUCCESS in 1h 00m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/00c94cc8dc6a4cd09358842bb9e40f68 : SUCCESS in 1h 19m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24403d2c376d4b4ca0ae741fc244c94f : SUCCESS in 47m 59s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2b1b0dc3103490a8bb2d03514a227ae : SUCCESS in 1h 13m 30s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/76d3708b55ac411292e65c76cd643f1c : SUCCESS in 1h 07m 07s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/133e8a7946b14c2ca762780a459a97cc : SUCCESS in 2h 53m 45s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/90f2840abb4a4c1099cc1761c8c5fe82 : FAILURE in 1h 03m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e4aad598622c5b12d8ae39843d3928f5f8422dd","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-03 13:46:31.000000000","message":"Patch Set 1:\n\nWhile the change itself is minimal (extending vnic_types) and it mostly allows to use an external feature (in ovs), I wonder about existing deployments not matching the reqs.\n\nneutron train works with ovs from 2.8, so failing sanity checks could happen on minor updates. Also what would happen on existing deployments?","accounts_in_message":[],"_revision_number":1},{"id":"5d422f3e017dfc75568663061f55df7a6d53c0b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 14:57:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e259194033973ab7cee8b83fb503e5289065abdc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 17:12:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793473/1 : SUCCESS in 2h 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c88f3abf62adb186cc8aa9d06054f7668514d22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 17:38:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7dc457738f7d40448f6501425cd4b7b4 : SUCCESS in 1h 44m 32s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/da068cd9d4614dc082b1b09d328feb1f : SUCCESS in 2h 00m 39s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/61f8ed4f5cf84bfdaef0849d02858ecb : SUCCESS in 2h 34m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/02389f19def44b12ba64a5ef807dc90e : SUCCESS in 1h 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f6c795077f749498a39b4e86c542266 : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a67cd396664361bec314e8f7718571 : SUCCESS in 14m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beb49a3e652e4dac8a4048a244e6e6a8 : SUCCESS in 20m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/001bc3d064f74a4ca9e54d56d8c36de0 : SUCCESS in 20m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22d8a0286d6c484d83e4304715a4fd3e : SUCCESS in 22m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67b1c450c924df0b65f5d29daed9722 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/21deff1d3ac64393a921517b0d18abe4 : SUCCESS in 52m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a546bdfa68cc48fca694ee8f3e139d12 : SUCCESS in 1h 07m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d7d781b47649b69bd1b83e451f2c51 : SUCCESS in 10m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2b753ab1c029420abd50da0ec1cef3df : SUCCESS in 46m 37s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/57bc40b0865e4a4eaddfad021902b9b1 : SUCCESS in 45m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ba0d80bbb01d4802880943d3d1b31a67 : SUCCESS in 1h 11m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1db26e36583e43d5997110d0b16f5192 : SUCCESS in 1h 39m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fb1a575df36c4a8587c2187ee9f6b041 : SUCCESS in 1h 09m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d77626fc71864bf2999ad364ca250b2c : SUCCESS in 1h 13m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b1c58b026c1419e8d9ee529c9ed5a79 : SUCCESS in 1h 06m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1dd51eacbb654ef295bc36023f9b1701 : SUCCESS in 1h 03m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a03495683fb24ed58a5978ebe4f48493 : SUCCESS in 1h 31m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/252ebb10aa7943ab8095a43a338ccc7e : SUCCESS in 1h 45m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a968acad8d474d48a5334211ab14e556 : SUCCESS in 1h 07m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b430c9b9d29e45dbb226875439e731a2 : SUCCESS in 54m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/50afba9e249948ccb339618db71e2bd2 : SUCCESS in 1h 08m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7618741289e54a3daffd6f21f296932f : SUCCESS in 1h 33m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/390226258b8949feabbebc5b247ac2de : SUCCESS in 2h 10m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/29f33cc767a04ca2b52c073b3e42c572 : FAILURE in 17m 41s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b527ab0b5bdc44598b4e27381c5d958e : SUCCESS in 1h 01m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6842169c54f949a6b685b4387c257d46 : SUCCESS in 1h 07m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d54949c0b1214698b91cd1c11093ee2b : SUCCESS in 1h 21m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/524de27e38d4429e9f21fe2e6ebfaa81 : SUCCESS in 44m 08s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c648bb4201b84b2dbff39d4c8d9f8b68 : SUCCESS in 1h 10m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6f88d3840244d64b0dead93dcec950b : SUCCESS in 1h 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d096e75960fa4f1af4506ffa1ce05df27d087ac0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 23:10:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"760547c41487d9f76a06fa753e4639671ccb195c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 02:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/63f59dc8439943bcb7edf87a0358fb2b : SUCCESS in 50m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/1a7cc17bcb75448a8d7cd04eb1923fca : SUCCESS in 2h 18m 35s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ef65991e7c2a42e8947d839421befaaf : SUCCESS in 2h 06m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/0f45f0763c3e430da9ac065653b72881 : SUCCESS in 1h 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4937bbcb36614fc59390701b3c65a9af : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2a20e419de41f1b3f6be0a1a28cbb3 : SUCCESS in 13m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82bba4be4c164dc8864b9a2df4af3a40 : SUCCESS in 22m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/456e436fe4c842ef8dae72dbe740cd28 : SUCCESS in 23m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4bf8212139a4001be9cd0f0026b5b96 : SUCCESS in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d84eee044545ef98b25e282d3e53a3 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/500f62431a0f48c398c9a99d76fe7d71 : SUCCESS in 1h 02m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16ed98b2e617460e94c1c2a8c5897a68 : SUCCESS in 1h 21m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbb6fd5ac91b4ab68ec1b33a502f2d74 : SUCCESS in 10m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31e2b6ebe20e46279cc8cf9f3ec753cb : SUCCESS in 51m 48s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/fb2d6c553ce647d5b0c2fd7fb1499e61 : SUCCESS in 47m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4ac588d9096a4663ba90f0c3786ef1a3 : SUCCESS in 1h 12m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f49c6399204c470b95574c8b6f89a38d : SUCCESS in 1h 44m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c8f1c5ad3ea84a5faa691ed2aaa68ce8 : SUCCESS in 1h 16m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d14abab5fb674a338d299629fd71b4eb : SUCCESS in 1h 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/beeea82d962e41f5a1885f731539c8dd : SUCCESS in 1h 07m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efc83e6782834b19bfe5d2349e6d02f8 : SUCCESS in 1h 07m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/632668ebc1ad487d9362d414039dd1d8 : SUCCESS in 1h 51m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39a567939d534855860f1f83b2d4b3aa : SUCCESS in 1h 43m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29abbc2a606841e99708192f18c52f9e : SUCCESS in 1h 12m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/85c40b0df27b4266b50c408221e4c8da : SUCCESS in 1h 01m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6c5434de44064d6c9187815b1cb2ece3 : SUCCESS in 1h 17m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3e7a07113b3460fbcd613e9b0891dc6 : SUCCESS in 1h 36m 10s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ca14e190ee2d47d1951749413294b3cf : SUCCESS in 2h 48m 30s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/7088b58d8b40402587769e781c98d7b6 : FAILURE in 22m 27s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bee64e9132e54d779bde961bc87d8c59 : SUCCESS in 1h 00m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86ee7489dcc94144a23487f4dfc2069d : SUCCESS in 59m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d53cc4d263de492680dc66c734d5bef6 : SUCCESS in 1h 22m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4939b536aa5743ac96e87b751de0c8f1 : SUCCESS in 46m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8dc9149a75442019fe5f7ae6fc958bd : SUCCESS in 1h 10m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/08c110191cd54dabaf5a87af42adf551 : SUCCESS in 1h 06m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26ec4c0cd2aa442a39e7e7686217d2e2a9b75722","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 04:18:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793473/1 : SUCCESS in 2h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"be6ae79e8b03a6378d3fd8c3b37bc084e0eab946","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-09 07:40:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nDo we really want to backport it to Train? If says that this require kernel 5.4.0 and ovs 2.11.0 at least. But officially supported distro for Train is e.g. Ubuntu/Bionic where there is kernel 4.15 and ovs 2.9.8 and Centos 7 where kernel is even older.","accounts_in_message":[],"_revision_number":1},{"id":"a5d8e3a606b0fc3ce93cc12382945be0889f786b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-09 17:00:24.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Do we really want to backport it to Train? If says that this require kernel 5.4.0 and ovs 2.11.0 at least. But officially supported distro for Train is e.g. Ubuntu/Bionic where there is kernel 4.15 and ovs 2.9.8 and Centos 7 where kernel is even older.\n\nYou are right and makes sense, we should constrict the features to the max versions in the support matrix.","accounts_in_message":[],"_revision_number":1},{"id":"91707e06e469a11bb4535bcde3b075a5078f78fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-09 17:01:10.000000000","message":"Abandoned\n\nMaximum versions in the support matrix do not comply with the feature requirements.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd04c06769547e72cad15bcc93a02da2dbf5bce8","revisions":{"fd04c06769547e72cad15bcc93a02da2dbf5bce8":{"kind":"REWORK","_number":1,"created":"2021-05-27 16:50:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/793473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/793473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/793473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/793473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/793473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/793473/1"}}},"commit":{"parents":[{"commit":"e00f444211bf8dd349d9a159387e7ee23058a923","subject":"Install \"pyroute2\" as a doc job depedency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e00f444211bf8dd349d9a159387e7ee23058a923"}]}],"author":{"name":"waleed mousa","email":"waleedm@mellanox.com","date":"2018-10-18 13:41:38.000000000","tz":-240},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-27 16:49:43.000000000","tz":0},"subject":"Add support for direct ports with QoS in OVS","message":"Add support for direct ports with QoS in OVS\n\nToday OVS mechanism driver can bind Direct port see [1] for OVS\nhardware offloads.\n\nOVS was extended with tc-offload to support rate limit see [2].\nThe OVS QoS driver [3] is limited to work only with Normal Ports, so we\ncan\u0027t put QoS rules on direct port.\nThis patch proposes to add support in OVS QoS driver for direct ports.\nThe mechanism to enforce such policies is the same with normal and\nhardware offloaded direct ports.\n\n[1]https://github.com/openvswitch/ovs/commit/e7f6ba220e10c0b560da097185514b6e33e2dc71\n[2]https://github.com/openstack/neutron/blob/3b074128ca91ba335bd8ef9db8dfe00fda51a63d/neutron/plugins/ml2/drivers/openvswitch/mech_driver/mech_openvswitch.py#L83\n[3]https://github.com/openstack/neutron/blob/3b074128ca91ba335bd8ef9db8dfe00fda51a63d/neutron/services/qos/drivers/openvswitch/driver.py#L56\n\nConflicts:\n    neutron/cmd/sanity/checks.py\n\nCloses-Bug: #1843165\n(cherry picked from commit 12089a526e25cfe5ccb8b4da86b653970d75f14a)\n\nChange-Id: I5093747ed9bde03d27e32e3206df91e4c27a2632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd04c06769547e72cad15bcc93a02da2dbf5bce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd04c06769547e72cad15bcc93a02da2dbf5bce8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
