)]}'
{"id":"openstack%2Fneutron~793500","triplet_id":"openstack%2Fneutron~master~I9c54a268fc4ef9705637556ecba161bd6523a047","project":"openstack/neutron","branch":"master","topic":"bug/1929998","hashtags":[],"change_id":"I9c54a268fc4ef9705637556ecba161bd6523a047","subject":"Use local and ip address to create vxlan interface","status":"MERGED","created":"2021-05-27 20:14:01.000000000","updated":"2021-06-08 01:37:29.000000000","submitted":"2021-06-02 10:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"793500-bug/1929998","meta_rev_id":"1308e95c2add722a6e2703462f3889274eb28596","_number":793500,"virtual_id_number":793500,"owner":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-06-08 01:37:29.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 10:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-02 07:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-06-01 15:59:38.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-02 07:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 20:15:33.000000000","updated_by":{"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"},"reviewer":{"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"},"state":"REVIEWER"},{"updated":"2021-05-27 20:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 22:19:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-28 00:48:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-28 01:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 06:07:52.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-31 13:34:15.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2021-06-01 07:18:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-01 09:00:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-06-06 20:06:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9040fe10e2d0fe114ff051e4966f378decf2519e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-27 20:14:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01f8e105498729d17ba5efc29c166cb1a67cd80a","author":{"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"},"date":"2021-05-27 20:15:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7b591dd09ec6752ec1d9bda0fa2b19ec63531ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 20:58:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d727cf589fd4cfc8db02c0b46f2265d : FAILURE in 41m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/553294a527e54754af9a16f6ef9749ed : FAILURE in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9650413959277cc12a61c34c7beba362eab5f325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-27 22:00:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89e2ae97f32a3504bccee9d6b55bc5d8f5bd3bba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-27 22:19:53.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/00/793500/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/793500/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"babd41d513a9da5f11d7a419bda47fdbaf8dd642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:46:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f2dcb87a21c41b7ac6d3b5326470fb4 : FAILURE in 44m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/821b396d28ab44cfb9f976c9648c5425 : FAILURE in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84d283661466643a4277af4711761655a136d59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-27 22:54:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a551e36038ce06c9bd7ce064e724b75e0204b88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 23:40:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a7c04aacd144f23b7f760e3bbbcf074 : FAILURE in 44m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed0fe5ebe0ef4a24a5681406c7ee5695 : FAILURE in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"979812dc00ea3d18339318139913b57c1c9ac150","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-28 00:48:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/3 : SUCCESS in 1h 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"334727f5d393058115c85507a5baecdbcd269c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 01:09:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33384771fb584ec7bfbb53fe0fc2f786 : SUCCESS in 45m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5e115ba0e26448b7aab5b0a8a0aa2727 : SUCCESS in 2h 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c6f98be8cb41dd98544888898c310c : FAILURE in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef7b9f44766c4f78a6e5ffdc5a42a2fb : FAILURE in 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc481cd9b936424bb5193f33644de816 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f5d618ddb01406896e1f306a47878e3 : FAILURE in 27m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d616011093d43ecb656507d69e31cda : FAILURE in 30m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99c4529317dc43598f5db16bee422fd1 : FAILURE in 26m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e0d39f632044388b0dd62a46980076 : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bda563960a4e4e81b2bf77da40cc891e : SUCCESS in 44m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a6707985c584836afd6bf5c101ce727 : TIMED_OUT in 2h 13m 48s","accounts_in_message":[],"_revision_number":3},{"id":"31dc63477e0f427bf17fa6f3d72d36b37df85550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-28 02:15:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dee1906416a64dacfe6fa0a20d9179a0a5bbfbf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 03:04:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/996613788f964314a5b53f1e99107b5e : SUCCESS in 47m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/696d92440d46447ba88f805e00b8ebe2 : FAILURE in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e7483c04402294e07ea2d1a3a15d6980c52a7fb7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-28 03:53:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/4 : SUCCESS in 1h 37m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f627edc7d6795b6caf4a8e5ac67b9a210ff928b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 04:14:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfa258c60de6481ab0447cf2f1f9ef12 : SUCCESS in 38m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4a3937d03293475cbb2b440c9f4c7bcd : SUCCESS in 1h 58m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2775d8be3d594c67acf145401f3277f0 : SUCCESS in 30m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00d59735155641bd85086eeb4b1c1b34 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d52af9673e4c7e8ae3ce12a7469166 : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0a0aacdea824e55815eca92649664ff : SUCCESS in 23m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314812a287394fe4851f9613e7d4299b : SUCCESS in 25m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124db4828bfc4602a07d07d5847e6b4e : FAILURE in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1adcccbe6db14024a6d90fb7b05d218a : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/017fb50bbbd3479680763d5ae4edd8c2 : SUCCESS in 45m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/943545865bbe44bb8c3b4edc41451ccf : SUCCESS in 1h 18m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9cb682af3796bf25135068c44c18a6c8c3ce0d42","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-28 06:07:52.000000000","message":"Patch Set 4: Code-Review-1\n\nFix LGTM. Please open a LP bug for the described issue, it might be used to track backports","accounts_in_message":[],"_revision_number":4},{"id":"6abec9532c539a008ebe8f7784995bc19513a777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-28 12:51:39.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c8c1b2d4ad04c285387dbe17b123da92e440b667","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-28 12:53:42.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Fix LGTM. Please open a LP bug for the described issue, it might be used to track backports\n\nLP bug 1929998 opened and linked.","accounts_in_message":[],"_revision_number":5},{"id":"212e056ee88e665de45c199321e81c219e4cb8c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 14:25:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9af6cdc543224dd5b1a8003ad7b99c36 : SUCCESS in 45m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce8dee0090ca43729c3bc4b13aa15c93 : FAILURE in 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3a5d8df7f6adcfbb3fa1fa7fc02cc6af0805243e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 14:44:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37a0d0b976494af094137f207e16b46c : SUCCESS in 43m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6ef18d4166124c2cae32f7a0c78bef0f : SUCCESS in 1h 46m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b5800c69d7b4e63a616848609a5c77a : SUCCESS in 39m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ac5469d7fbc41c5aa5b8a26921b891f : SUCCESS in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77561db4b24648eba2892f2bd92dc891 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e9acc4932d4b738e575d8432e4cd0b : SUCCESS in 35m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e6d32c106e42229efb3f37d29f75ee : SUCCESS in 29m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68c3af973c34df3bdacd0484b9993c4 : FAILURE in 28m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579fa08a9fdd4f54aaae6efd3e41838f : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecf109db65cd449cbe71bf289ac56e68 : SUCCESS in 49m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e90659fb1ecf4a90bc19675d51775c79 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":5},{"id":"8c868f2ce5e999eda8a9116e5c77062e66e4bd8e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-28 15:23:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/5 : SUCCESS in 2h 09m 56s","accounts_in_message":[],"_revision_number":5},{"id":"09b808aa5df188e3a0c7dc317d210bf2b3969d2a","author":{"_account_id":20366,"name":"Neal Gompa","display_name":"Conan Kudo (ニール・ゴンパ)","email":"ngompa13@gmail.com","username":"ngompa"},"date":"2021-05-28 16:29:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cfee63d6ea95f3a52d418fdc182bd6a56297e1ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-28 17:10:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4f3badbbb3ba1bcd32c0d066d36be0334811d4e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-28 17:17:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"945306d8f33b807db9f3450b589039923dc30a77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 18:01:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80c4eb229c8c4888a1624c81ffd3b049 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6ff5c2dba7d4742b682c1c3f2987d8b : FAILURE in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"04e24f625b799cce768258913812e90c9a2e412e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-28 18:31:17.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/00/793500/7/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/793500/7/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"4d36a34a741c2f251c18f147fc44c48bc942fe92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 18:52:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e657bfa79194b56ab4ba0f40e818eb0 : SUCCESS in 40m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/094a6828a145424986e1e9673dc059f4 : SUCCESS in 1h 33m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/101b0d9310284641b6e2af707e4f40e3 : SUCCESS in 39m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53f4d7b50f7c432eb629ee9ac5bb70e7 : SUCCESS in 33m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39500423b41445caf2a533ebd023905 : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1323037506f4e33aaa27f844c10e987 : SUCCESS in 24m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c16b94ece43c40fa9296c46791094f51 : SUCCESS in 25m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9fa4ff67a6d4f288f78ab444058a1fa : FAILURE in 18m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2a02f3741347949c787667a066c8b9 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9b48329562143afb268623675c83128 : SUCCESS in 40m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5afca2de56184edc92552b6116a66582 : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":7},{"id":"19455248b3eb8ca2d79100a59152b12f8c55c5c1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-28 19:30:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/7 : SUCCESS in 1h 52m 05s","accounts_in_message":[],"_revision_number":7},{"id":"93b22c1ed7b6fa2a3dc4c06f10603016894183be","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-31 07:16:53.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\nwhy unit test has gone?","accounts_in_message":[],"_revision_number":7},{"id":"e5d255b2541d9f9a3c92b33af322005806f88d27","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-31 13:34:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f2543b243374aadd8bef78064de0d44547b08f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-31 21:23:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7aee794a020a0e1c7ca4090909ad50502443ed8f","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-05-31 21:33:52.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e why unit test has gone?\n\nI\u0027ve updated the logic here. We do not want to set a local ip address when using multicast. Even though the kernel allows for this, it\u0027s unnecessary and may have unintended consequences. We only need to set the local argument when using L2population. Therefore, we can insert this step as part of the L2population check, where the proxy argument is currently being set. I\u0027ve updated the relevant unit test section where L2popution is set to true. This section now asserts that add_vxlan is also called with the local argument set to the local_ip.","accounts_in_message":[],"_revision_number":8},{"id":"9afab75dc56cb88c3eb243c6e49d3516a003d041","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 22:07:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0d84554a78948f1816dc68c5ac755c9 : SUCCESS in 41m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2087a4a9e2c4ae183d2468f7a00fa60 : FAILURE in 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7bda9033d14d8b504460c3aedabc2da4019b0747","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-31 23:15:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/8 : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":8},{"id":"8671f2bb2abf75bc060563d6ab7a5f39e9966220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 23:42:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e648261d18b547b5ba10942b2aeaa2e1 : SUCCESS in 48m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1568a68a73fe4b1f800ec52c2b64c58f : SUCCESS in 2h 00m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78100fa0e6744c08894ad39096aba204 : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfe1d7a3260a4c859b7bbcf6169cc57f : SUCCESS in 35m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f476370c243d4e6a9e607b2c57b2f259 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a59a27b2f6c944e5867cee1bae15b1a5 : SUCCESS in 30m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e9f342b81742a7a19dec9e8ada8868 : SUCCESS in 29m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/568bd7c8fef74d25a9f2b2168149bd8e : FAILURE in 27m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7d6997a74a4812a816e4ca1b0df600 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3cb2f9cb18f4d448cc9d059e249f8c7 : SUCCESS in 50m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c7fea97011b4c879972b463f7b25543 : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":8},{"id":"4ae9c3eed0e8d6705be83450e0f1d531b0615b2a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-01 07:18:25.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2988d39ee91485388a79eb0a917296779ed3fe96","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-01 07:46:55.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"95fc5439248f989d4f4646d736327a148b392156","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-01 09:00:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"17dba5c57961055219f976c891df217ad99b412c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-06-01 13:13:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5cd1e73dc7fd0004378aa3febb94315a6b1eaf00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:56:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/108d2a31b06d45d0b57dd7d7b6d73597 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef4976ce316849c095cc7cfd4dffa60f : SUCCESS in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c191e94455aa3556c12d2e8e8dec81e2640d7720","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-01 14:02:14.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"20b0fbca837d22b72de4784a1140745f33bdd394","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-01 14:47:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/9 : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":9},{"id":"d6d331db6b8213f92b971f67c07d178b72857da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:53:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd6b9c04e6c04b55b0e12c2205181511 : SUCCESS in 42m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7bf9e4d75cc472aa79b9cc5dc1b776d : SUCCESS in 1h 37m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e138674011a444ce9666b8b1464d281d : SUCCESS in 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/804962a1b8934f20964abf1d4f93a93d : SUCCESS in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d9852f3c27405ab3327ffce9c7b346 : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63e28dfd8f224992a4e6500ec667478a : SUCCESS in 24m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02503878273c4a53a0e379c790035b13 : SUCCESS in 23m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fe4e8638ce24a8b87997813f947e458 : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3ed1b7272f42e8babc6e44bbfee864 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef7e953f766c40099d0d41f63d63f44b : SUCCESS in 53m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98a56bcad39c4764a5475d6887d93082 : FAILURE in 1h 06m 37s","accounts_in_message":[],"_revision_number":9},{"id":"34fed3f6e0992a9e15c10aae6f2e8c4f78d34a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-06-01 14:59:43.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"4c971e7dec6e15526352fdec821363a9bf078da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"date":"2021-06-01 15:02:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"eddf44996aec27f6102bf8c182df507d74c67446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:50:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02edae4da055410cbea0f2f66350c059 : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebf893e7369243a1b1898c5756283e67 : SUCCESS in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"814a2ab24e27ac9a3000248e5660295b79c62087","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-01 15:59:38.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"4696bc24041ae45fca033ad6ef2cf127bd74dc63","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-01 16:37:54.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793500/11 : SUCCESS in 1h 34m 40s","accounts_in_message":[],"_revision_number":11},{"id":"e1ddb0a6c636475f10d497cd4e6b2525f57d8f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:14:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ed4ce5f9281e43a38aa4be7236e690e1 : SUCCESS in 45m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c7d81b720e914fc3addb6df7b38d079c : SUCCESS in 1h 51m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a01d58f4a8547fb9b699bbdb1081a70 : SUCCESS in 42m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7df63e86f56a40abba2bb7bf079f6181 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086ee36199d244f3b5db0d37f54295dc : SUCCESS in 14m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875fdf305c934648acbab86d14eda515 : SUCCESS in 30m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4cf5f661cd54719a43f763d85b18485 : SUCCESS in 31m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1837e632f4024fe680f782cbc62f3638 : SUCCESS in 33m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9e625297f34d348e8f44919c0a086c : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7317b43d706140289cba6565cf07263d : SUCCESS in 47m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/afb5665f58f84a2c998e503237ccb091 : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":11},{"id":"f157d5942c38e34c63440f91467c3138a1c95c8b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 07:16:30.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f81036321825d019bed86831377839b58377b6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 07:17:19.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"76a37f0a22424b4f779179f5125cfab3a4443dcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 10:38:56.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe2480565d6b406ea2bfe51e4caccbb3 : SUCCESS in 46m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/de8bd9a644c541bbad4f82bde03458bd : SUCCESS in 1h 38m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58ef7b4682fd4ed9a1ae40eed540ec9c : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91cb645eb1f42d6bcc85fe724e70c74 : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e95aa345332440cda80d723935dc835e : SUCCESS in 32m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c0d1aff42bd4532b6132f4014147b26 : SUCCESS in 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793bbcd6c4bb484bb9a1d48c8a7d44d8 : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d6b8110128548f38de00d0d35ed1900 : SUCCESS in 41m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/929ca61f5e7b4ba4bc96ce22b96f081f : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":11},{"id":"ea4e9ef50a5319a5b2380a3f89e90c402df5d517","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 10:38:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"adcbec4637aed7f698347034886403c1f9047c55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 10:40:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1440f6102d2f4a0ebf86a0bed575cbe5 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":11},{"id":"2b67864448e04019d77d6e4979c20223f9b58d0a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 20:06:30.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9e4a7edef094ae2c47e467323de535ae3abfabab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 21:24:10.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c1dea5aa51e8dd72e37c99a6df2fc4c509a2de42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 22:55:26.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"1a5df5582f9bbda06af37957220c16ae610e5f7c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 02:54:31.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2aad90c111da23dc8bd97b50fbfeadb84f70bf95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 05:15:24.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"9bc87f118ef969ad1c88679903cee8d97234f8ec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 06:52:13.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"bcb43817061bdca0d87ee36138c5dcc36ad26f45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 07:58:56.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"7acab504ecb5d4ad480fd82d4b594a6f8284fa88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 19:21:43.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"3e1804518afb4939e5c7655d1bcb27ca6f75f8ec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 23:16:53.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c4957eccb0e2e79dd8b1b89cf22c2470ddbb832b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 01:13:55.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"1308e95c2add722a6e2703462f3889274eb28596","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 01:37:29.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/793500/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4cd11f4dee494c5851318410c523dfcfbce4c824","revisions":{"dd751a7d850f76f6e9b3c16212299785ec7e7839":{"kind":"REWORK","_number":1,"created":"2021-05-27 20:14:01.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/1"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:13:20.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nLinux kernel does not allow a loopback interface to be an endpoint\ndev interface for VXLAN terminations. In order to allow loopback\ninterfaces to be VXLAN termination endpoints, we need to change how\nthe vxlan is created: don\u0027t use dev, use local and an ip address.\nUsing local is also, perhaps, a more flexible solution, since it\nallows for picking the specific IP on the interface, as opposed to\nwhatever the OS decides to use.\n\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd751a7d850f76f6e9b3c16212299785ec7e7839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd751a7d850f76f6e9b3c16212299785ec7e7839"}]},"branch":"refs/heads/master"},"3bd2d3c52e9ad2a45d4003808807421c8c7033a0":{"kind":"REWORK","_number":2,"created":"2021-05-27 22:00:15.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/2"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 21:30:28.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. By setting and\nincluding the local argument during vxlan interface creation,\nwe can ensure that the intended ip address is used.\n\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3bd2d3c52e9ad2a45d4003808807421c8c7033a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3bd2d3c52e9ad2a45d4003808807421c8c7033a0"}]},"branch":"refs/heads/master"},"92cc74aa9746b3ce8b6252048246c18e33260766":{"kind":"REWORK","_number":3,"created":"2021-05-27 22:54:16.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/3"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 22:53:58.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. By setting and\nincluding the local argument during vxlan interface creation,\nwe can ensure that the intended ip address is used.\n\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92cc74aa9746b3ce8b6252048246c18e33260766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92cc74aa9746b3ce8b6252048246c18e33260766"}]},"branch":"refs/heads/master"},"941db00b53749ca6cf54015b4496e179f2ec56e0":{"kind":"REWORK","_number":4,"created":"2021-05-28 02:15:17.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/4"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-28 02:15:05.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. By setting and\nincluding the local argument during vxlan interface creation,\nwe can ensure that the intended ip address is used.\n\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/941db00b53749ca6cf54015b4496e179f2ec56e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/941db00b53749ca6cf54015b4496e179f2ec56e0"}]},"branch":"refs/heads/master"},"603b5ac7a58c341693f45765f05c4c54505cbc66":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-05-28 12:51:39.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/5"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-28 12:50:44.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. By setting and\nincluding the local argument during vxlan interface creation,\nwe can ensure that the intended ip address is used.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/603b5ac7a58c341693f45765f05c4c54505cbc66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/603b5ac7a58c341693f45765f05c4c54505cbc66"}]},"branch":"refs/heads/master"},"d81a6fa366e21b2f48a3dfb68dec58ade88b52dd":{"kind":"REWORK","_number":6,"created":"2021-05-28 17:10:46.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/6"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-28 17:07:28.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when a vxlan interface is unicast, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is used.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d81a6fa366e21b2f48a3dfb68dec58ade88b52dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d81a6fa366e21b2f48a3dfb68dec58ade88b52dd"}]},"branch":"refs/heads/master"},"892bfea7dd51db9984d0144e32ac9b2a52334c7c":{"kind":"REWORK","_number":7,"created":"2021-05-28 17:17:44.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/7"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-28 17:17:30.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when a vxlan interface is unicast, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is used.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/892bfea7dd51db9984d0144e32ac9b2a52334c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/892bfea7dd51db9984d0144e32ac9b2a52334c7c"}]},"branch":"refs/heads/master"},"fddb1972010447542934d855d6cdb7806b41326b":{"kind":"REWORK","_number":8,"created":"2021-05-31 21:23:19.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/8"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-31 21:21:02.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when L2population is used, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is configured.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fddb1972010447542934d855d6cdb7806b41326b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fddb1972010447542934d855d6cdb7806b41326b"}]},"branch":"refs/heads/master"},"419bae7ffdd6ebf9e15296226ad15c44f451251f":{"kind":"REWORK","_number":9,"created":"2021-06-01 13:13:43.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/9"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-06-01 13:13:31.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when L2population is used, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is configured.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/419bae7ffdd6ebf9e15296226ad15c44f451251f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/419bae7ffdd6ebf9e15296226ad15c44f451251f"}]},"branch":"refs/heads/master"},"ff8dcc1a263ed0a0ace5c68ed0cbdcae6d33dad9":{"kind":"NO_CHANGE","_number":10,"created":"2021-06-01 14:59:43.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/10"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-06-01 14:58:35.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when L2population is used, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is configured.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff8dcc1a263ed0a0ace5c68ed0cbdcae6d33dad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff8dcc1a263ed0a0ace5c68ed0cbdcae6d33dad9"}]},"branch":"refs/heads/master"},"4cd11f4dee494c5851318410c523dfcfbce4c824":{"kind":"REWORK","_number":11,"created":"2021-06-01 15:02:31.000000000","uploader":{"_account_id":33525,"name":"Anthony","email":"atimmins@datto.com","username":"atimmins"},"ref":"refs/changes/00/793500/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/793500/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/793500/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/793500/11"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-05-27 20:11:04.000000000","tz":-240},"committer":{"name":"Anthony Timmins","email":"atimmins@datto.com","date":"2021-06-01 15:02:21.000000000","tz":-240},"subject":"Use local and ip address to create vxlan interface","message":"Use local and ip address to create vxlan interface\n\nAttempting to terminate a vxlan on the lo interface with\nmultiple ip addresses fails. This seems to be because only\nthe first ip address on the interface is used. If this address\nis invalid for vxlan creation (ie. 127.0.0.1), the vxlan\ninterface will be created, but will not have a source ip\naddress, and will therefore be non-functional. To remedy this\nissue, when L2population is used, we can set the local\nargument to the local_ip, thus ensuring the intended ip\naddress is configured.\n\nCloses-Bug: 1929998\nChange-Id: I9c54a268fc4ef9705637556ecba161bd6523a047\nSigned-off-by: Anthony Timmins \u003catimmins@datto.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cd11f4dee494c5851318410c523dfcfbce4c824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cd11f4dee494c5851318410c523dfcfbce4c824"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
