)]}'
{"id":"openstack%2Fneutron~793762","triplet_id":"openstack%2Fneutron~master~I432fd4a7ee07863d116a0435783958fd918723fb","project":"openstack/neutron","branch":"master","topic":"dynamic-logging-extension","hashtags":[],"change_id":"I432fd4a7ee07863d116a0435783958fd918723fb","subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","status":"ABANDONED","created":"2021-05-31 10:51:00.000000000","updated":"2021-08-15 08:30:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2d709c5fb078342462f9654a1509651b6def4364","_number":793762,"virtual_id_number":793762,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-31 17:02:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 11:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 12:03:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 15:01:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-31 17:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 11:10:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-15 08:30:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"00b19e0e5108fa9204c1c473407cc359e2028c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-31 10:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46bdf463229ba21bde9b01e21a715c36a5e9b020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:41:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9131dbb85b7a40cab7dee9239d623e24 : FAILURE in 48m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62726fc7905d4dd18e5517e25d34041b : FAILURE in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b863711a2519637d6f80679c8c7ceee85391b2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 12:03:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/793762/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/793762/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9ed68379229999593bedeae39c0f4db4f69cf4a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-31 12:25:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17af8a6a962aefbfa61443fa683fc85a34ef5dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-31 12:57:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5dede45227359f34c07b3daa24ce571f1123d94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 12:58:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/793762/2/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/793762/2/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"ab37068bc49adb7f300e9e4006bee3bb2fceab7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 13:08:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22ea349468bb45b992c64260e8ccb8b4 : FAILURE in 8m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f1a41babbb6403ea7beabf67e848294 : FAILURE in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"725c851372729e4d1bd9ca89dadc7c531c1e5260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-31 13:17:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"005f749c9c5a0e2afbd469da9197800ec902618c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 13:18:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/793762/3/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/793762/3/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"c14a876328a47d9434b96f8092b5e4ec1e7a6def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-31 13:27:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b1786691634187066a68489a27633413020edd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:15:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c75495f56d14917a45a44670fbe6e68 : SUCCESS in 44m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85dac36b993343388ab9613d4bcbebc7 : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b6219b572bf3a281595c606a8227e93eff7b52ca","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-31 15:01:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793762/5 : SUCCESS in 1h 33m 53s","accounts_in_message":[],"_revision_number":5},{"id":"f9575d5271081c0a5a68e76d8dff9744e98d8fb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 15:34:26.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/793762/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/793762/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"ab50b3d1a51d32c9624b01183d0c88ff8ad5edec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 17:02:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/688c0270c2994286b4c86cbe4a6cff27 : SUCCESS in 46m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a62390f3b10455781c99c93a2561c6c : TIMED_OUT in 3h 34m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8f7421bfcdfc41cfb2081535d3b17f75 : SUCCESS in 1h 57m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f14e930032c44c81bf30c72e3eaad460 : FAILURE in 51m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fde1e5e6afac43a0987f59df4441356d : TIMED_OUT in 2h 50m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41fcb66260aa46e0a9717f3fdb8e565b : SUCCESS in 1h 04m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7484a71b50f4323baa1206610da34bf : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f16ae44ac228437ca4b4efd8fed31ab2 : SUCCESS in 17m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e62997639c4fb8b265a8fb080f9e69 : FAILURE in 55m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc34b99dd5be4a5591b326821218c740 : SUCCESS in 29m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c0a6e8cdccd4b3c91e5f5c864ee333b : TIMED_OUT in 41m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635f5c17a20f49d6a2848ee96cd64ebd : SUCCESS in 17m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a301c84c459848a78bc99075952fb4c7 : FAILURE in 1h 16m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddc09950d7cb43be96e715cf08c63780 : SUCCESS in 1h 22m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b5a0e815f2824b92817a880d3b354fc6 : FAILURE in 1h 38m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97bf6ce7c1264d93a53fdd7412b35efa : SUCCESS in 1h 06m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/282149da47ce4d999b468c1a5814e081 : SUCCESS in 1h 04m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/02ae7ea8730c49e2a9faae05b3d22fb2 : SUCCESS in 1h 24m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/414b56d2a1694158b2604a3751e55fdd : FAILURE in 25m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52506f941d464c8aa13895ade3acd978 : SUCCESS in 1h 27m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c8013ae1b47476fabdb47c4ca37fa06 : SUCCESS in 53m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de39a6f2a5db4b02b11950a809400b3f : SUCCESS in 58m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/14babef29ad845309b1c628bfdea4301 : SUCCESS in 21m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/98a2de1d4289443bb099cdc90008d420 : SUCCESS in 1h 28m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ec3ccba9e1849fa980ff1a10dfe806a : FAILURE in 1h 40m 51s","accounts_in_message":[],"_revision_number":5},{"id":"7c95a299a39b23150944e2cc8e40e9e6d36449e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 11:10:16.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/62/793762/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4b85112d0dc21c22cbc129e1a27dbae0895c507d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 13:39:22.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/62/793762/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"916a09c88a9b9d0ceccb3fd2a20717a5504dcd13","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 13:59:19.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/62/793762/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"43b18e2217bf89cbd59bd2464750f143388325fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 14:17:08.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/62/793762/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"8e0e9f4013809316b1846470792e41e0ffe8b4a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 14:49:02.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/62/793762/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2d709c5fb078342462f9654a1509651b6def4364","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-15 08:30:56.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aeeade69df7422fab9eb268dc46d2e0884427300","revisions":{"e6da49cd6e620ee4d1d6f661d153a8ea5bbe9849":{"kind":"REWORK","_number":1,"created":"2021-05-31 10:51:00.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/793762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/793762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/793762/1"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","message":"DO NOT REVIEW: Enable OVN logging extension dynamically\n\nChange-Id: I432fd4a7ee07863d116a0435783958fd918723fb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6da49cd6e620ee4d1d6f661d153a8ea5bbe9849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6da49cd6e620ee4d1d6f661d153a8ea5bbe9849"}]},"branch":"refs/heads/master"},"4e0bc05b6fb3ebbaf81e028ad8a0b6d9a255f96b":{"kind":"REWORK","_number":2,"created":"2021-05-31 12:25:57.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/793762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/793762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/793762/2"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 12:25:45.000000000","tz":60},"subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","message":"DO NOT REVIEW: Enable OVN logging extension dynamically\n\nChange-Id: I432fd4a7ee07863d116a0435783958fd918723fb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e0bc05b6fb3ebbaf81e028ad8a0b6d9a255f96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e0bc05b6fb3ebbaf81e028ad8a0b6d9a255f96b"}]},"branch":"refs/heads/master"},"8734f76d1c125e508416de3ac023c82653161e6f":{"kind":"REWORK","_number":3,"created":"2021-05-31 12:57:29.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/793762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/793762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/793762/3"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 12:57:13.000000000","tz":60},"subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","message":"DO NOT REVIEW: Enable OVN logging extension dynamically\n\nChange-Id: I432fd4a7ee07863d116a0435783958fd918723fb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8734f76d1c125e508416de3ac023c82653161e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8734f76d1c125e508416de3ac023c82653161e6f"}]},"branch":"refs/heads/master"},"3ab813114de7e7c161061e1c27956740eab92e19":{"kind":"REWORK","_number":4,"created":"2021-05-31 13:17:46.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/793762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/793762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/793762/4"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 13:17:38.000000000","tz":60},"subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","message":"DO NOT REVIEW: Enable OVN logging extension dynamically\n\nChange-Id: I432fd4a7ee07863d116a0435783958fd918723fb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab813114de7e7c161061e1c27956740eab92e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab813114de7e7c161061e1c27956740eab92e19"}]},"branch":"refs/heads/master"},"aeeade69df7422fab9eb268dc46d2e0884427300":{"kind":"REWORK","_number":5,"created":"2021-05-31 13:27:12.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/793762/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/793762/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/793762/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/793762/5"}}},"commit":{"parents":[{"commit":"5a6f61af4a438433f669e2c57bb152fc4ae8f3a9","subject":"Merge \"Add tests for metering API\u0027s new policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6f61af4a438433f669e2c57bb152fc4ae8f3a9"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 10:50:34.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-05-31 13:27:02.000000000","tz":60},"subject":"DO NOT REVIEW: Enable OVN logging extension dynamically","message":"DO NOT REVIEW: Enable OVN logging extension dynamically\n\nChange-Id: I432fd4a7ee07863d116a0435783958fd918723fb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeeade69df7422fab9eb268dc46d2e0884427300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeeade69df7422fab9eb268dc46d2e0884427300"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
