)]}'
{"id":"openstack%2Fneutron~793801","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I1fb91734260b24d71ad6bd661fae71fc71bcbbc9","project":"openstack/neutron","branch":"stable/ussuri","hashtags":[],"change_id":"I1fb91734260b24d71ad6bd661fae71fc71bcbbc9","subject":"[Stable only] Set irrelevant-files in some additional jobs","status":"MERGED","created":"2021-05-31 14:28:35.000000000","updated":"2021-06-26 16:01:33.000000000","submitted":"2021-06-26 15:59:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793801","meta_rev_id":"4448aef9432e9ee11858aa56fa89922397f743e1","_number":793801,"virtual_id_number":793801,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-26 15:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-26 14:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-02 07:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-26 14:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 16:20:34.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-05-31 17:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 07:31:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-26 14:23:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3c34b649a08086f74174ebf9fbd0bf88ca0d1d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-31 14:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76039636cd61255f773877eef6cd2898ce3913be","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-31 16:20:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793801/1 : SUCCESS in 1h 51m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2f5f72895ef07739c438314a027a9673aff426e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 17:41:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/2c93b72dda9343f99148bd5c4bbd2570 : SUCCESS in 44m 17s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/3eb52076c5f044f78bb8711820d78227 : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/813e69ca30384fd79a14138a5db5a7c9 : SUCCESS in 1h 31m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/d24ab0fa0ce44d429b9a3269a82389ee : SUCCESS in 1h 22m 31s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/b363357e5818410e93b3a08bac0b2c8e : SUCCESS in 2h 17m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a33e05c661a04769b4cf0a9a1c75905f : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2683c14bc814b70ad1db64f19f1143c : SUCCESS in 16m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64c17d79cdcf49e0815845adb673ff1b : SUCCESS in 27m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2fe675c3fd544d39ba5cb1539b74d6a : SUCCESS in 28m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc63f0f438a9487da4662e991f301532 : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da961d03ac44339a8bd16a9dafbcf74 : SUCCESS in 10m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bad153fce8c94a24a3806932f35cbaa3 : SUCCESS in 46m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/74e64da6cf02480a9b10a97fd0b08b8a : SUCCESS in 1h 18m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1d8bf9037a641c5b30935b7c133f0c5 : FAILURE in 1h 17m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e88bb463cfe04c5391d2c472520a4a51 : SUCCESS in 1h 36m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e10f921651944bed94c83aaf29397f4b : SUCCESS in 1h 20m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e6c82effca654781a7bf21d66ba4ff7b : SUCCESS in 1h 14m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/90616a2d8af34af1b4976b9ae2b9a106 : SUCCESS in 1h 08m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/86a191d55f27493ebd338058d316e068 : SUCCESS in 1h 01m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d7c4a2f118ce44dba7bb3a788302b925 : SUCCESS in 1h 09m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/46f532b074204c23af57af83eeac00c2 : SUCCESS in 1h 51m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d9684ebf45341189167990a780d3ff4 : SUCCESS in 1h 34m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb9870dfa45b4b5db5846e355a0243b3 : SUCCESS in 1h 18m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c4515afc8b264ad1b856e7d4dbde7249 : SUCCESS in 1h 00m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22ff4f2521d44115a2df5b4a8174bae9 : SUCCESS in 2h 02m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feab1b9330bd40f2a1d8f42deadfca98 : SUCCESS in 39m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34a9ac937fcf4496a210fd691ca9c30d : SUCCESS in 1h 18m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c3991f98493d495499adc0dad86a53a2 : SUCCESS in 36m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/20876af05e5741d48e1d885ded8da847 : SUCCESS in 1h 18m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7d5f35ef7a440b5b86debd06ddceffd : SUCCESS in 50m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d866fc4c34c548f9bcbc7f24da9f79de : SUCCESS in 3h 10m 46s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e0c2e002eb347399383178c0f4fc123 : SUCCESS in 1h 08m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2966415c371a49ebb159a154ae2c7206 : SUCCESS in 1h 33m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c61c9f341f9a46cdb315d8df2ed710c0 : SUCCESS in 1h 52m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/56b16af0f7fe4db8bc6e63ff1a79f316 : FAILURE in 1h 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f8f7d9be35b1ec5678d2940e9760aba61a265ae","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-01 07:22:49.000000000","message":"Patch Set 1:\n\nrecheck unrelated fullstack","accounts_in_message":[],"_revision_number":1},{"id":"138ae365b43eefb034ad6716f06e99a6e2ad87a5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-01 09:28:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/793801/1 : SUCCESS in 2h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e2a2baa62a7d392a485ef9fab776e3da729aa167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 11:25:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/cfae24edf98f498597bcab9b832040e5 : SUCCESS in 48m 21s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/144b848b3ea04fa89f797780bb33e3ee : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/232f11f967fc4f72946db1b09bef1c77 : SUCCESS in 2h 03m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/0417cfe0789a4c6ea4ed7f9b446ac8d6 : SUCCESS in 1h 34m 28s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/56fef84c6c124ecf8a9a83a32c2c7982 : SUCCESS in 2h 14m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b012855b71140818f26459000debde8 : SUCCESS in 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db7b148f91e464694fabf96b3dacc08 : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a309a8e15a1f44d9a44b4bb768dbac49 : SUCCESS in 27m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcf3f5ddbceb40aea895ade13f903e67 : SUCCESS in 27m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3173a043b8543deae83ff3b697fedd3 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120623cd96c7493297efeffd66a105df : SUCCESS in 10m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4bdf409510b54d8c8229c58df9eaa58b : SUCCESS in 54m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a362485d5d154aec818e3df124d0bb04 : SUCCESS in 1h 21m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/159a27e1652b4c0aba8ce0d14097134a : SUCCESS in 1h 21m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a4bc147f2c11422493df63ab8451eedc : SUCCESS in 1h 34m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5257ebb6c1e945d0832fd3d86f14466f : SUCCESS in 1h 19m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/044446af1dc0485694b89d4167cd1709 : SUCCESS in 1h 16m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e7cbc69101754546bbaae006ec3d4aa9 : SUCCESS in 1h 16m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/65c300733d054206a639ca8e56adbc3e : SUCCESS in 1h 09m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4146696e48204bef9e4932657a9585ab : SUCCESS in 1h 11m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/488f94a58ba74c88b75504b2f95513c6 : SUCCESS in 1h 44m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c6d1cd4ae264c0f8bba8d6aa8ee1ce6 : SUCCESS in 1h 43m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9929171e1c0467a8ffb36bd0bb6a5eb : SUCCESS in 1h 14m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7918f60f37c844b8b6d53cd669a71311 : SUCCESS in 1h 03m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7df4c195fab24486a0c8aa1d9821b57c : SUCCESS in 2h 02m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/392ddbe6c9aa4da5a518d92c2061face : SUCCESS in 1h 05m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/788fbf7bd2ec41dc860c793361167a76 : SUCCESS in 1h 27m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38a83d9e918c4f28967126a061e8246b : SUCCESS in 42m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/56d8cba37841482eb82b744cc86a4488 : SUCCESS in 1h 16m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15b86cbefb30429ebd59db041f2306c4 : SUCCESS in 53m 41s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a82001f1e9f74676b216f02208b3d652 : SUCCESS in 3h 52m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e69ce5f1b43542f7bcc06eb986ef1332 : SUCCESS in 1h 24m 40s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/454ee893017b48cca94524f123790a45 : SUCCESS in 1h 47m 33s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/240076827e6946c393dfd87191c7381e : SUCCESS in 2h 12m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d868ae8f315d42e8a2b081509c84018e : FAILURE in 1h 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"345082d229662cb6dc3ab74523ed3d291bbd79f4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 07:31:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1945d217fb342340fc6f41b077c603fccf89c88","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-26 14:23:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2aa2374da21b53f8b255dd4ee7ff117d8b14e1a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 14:24:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07b7e1fe4e9993c48a60275ca427d2993856b64d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 15:59:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/9f5e61bdd0d7454986719d60be2689df : SUCCESS in 48m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f17ee331c043f6a6b9965ecf02d6b2 : SUCCESS in 16m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf7a72f0b9e4f1196f64bc8cda70e54 : SUCCESS in 27m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17970b8a71d946fd92a9cc24ee8d4c27 : SUCCESS in 35m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d2b3039b9446ccaf5d7b0664e3817d : SUCCESS in 12m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5035c2670d4342f1a67569df3e213fdf : SUCCESS in 56m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fa42df3465744e86b4d78602a0a8c475 : SUCCESS in 1h 22m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3680a903f4fb408c93575133976c8506 : SUCCESS in 1h 08m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/36782293b41546be8bbf9ada99982333 : SUCCESS in 1h 16m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc1a3a41ef8449b4aa3a5f34fbda14db : SUCCESS in 1h 09m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3573324d7c1f481d81b81c51390981fa : SUCCESS in 1h 16m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/578a197b13cc45a5bca62bb06a7d4dc2 : SUCCESS in 58m 42s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1351103d685d466aa3ce2d66b5f81acd : SUCCESS in 1h 33m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/752db569f42f4aedb2062d536a30aa2c : SUCCESS in 1h 09m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4eeb292237214455b329a8a8daed6bf2 : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ca98b61c3469f4a8238aef314a7bcc9c3f2bcc95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 15:59:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4448aef9432e9ee11858aa56fa89922397f743e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 16:01:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e0c142429c47949b49fd850bd12e00 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f7061a41f915899d81f307d16582bb5d6abeab0","revisions":{"0f7061a41f915899d81f307d16582bb5d6abeab0":{"kind":"REWORK","_number":1,"created":"2021-05-31 14:28:35.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/01/793801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/793801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/793801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/793801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/793801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/793801/1"}}},"commit":{"parents":[{"commit":"7e1df72c16f2e1736b068387e76f5b9070d6d5bf","subject":"[OVN] Fix: Disabling snat after it was enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e1df72c16f2e1736b068387e76f5b9070d6d5bf"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-05-31 14:13:22.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-05-31 14:28:23.000000000","tz":120},"subject":"[Stable only] Set irrelevant-files in some additional jobs","message":"[Stable only] Set irrelevant-files in some additional jobs\n\ntripleo-ci-centos-8-content-provider runs in check queue in victoria and\nussuri, and does not have irrelevant-files set while other centos-8 jobs\ndo. This can trigger \"Unable to freeze job graph\" error when changing\nonly such files, for example in [1]\n\n[1] https://review.opendev.org/c/openstack/neutron/+/793417\n\nChange-Id: I1fb91734260b24d71ad6bd661fae71fc71bcbbc9\n(cherry picked from commit a8d4063bf1ccab9c68967cb085e886cae46069ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f7061a41f915899d81f307d16582bb5d6abeab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f7061a41f915899d81f307d16582bb5d6abeab0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
